)]}'
{"id":"openstack%2Fkolla~679277","triplet_id":"openstack%2Fkolla~master~Ie26623183a9e5e5d54456b73b79667ba9cf61f54","project":"openstack/kolla","branch":"master","topic":"bug/1841907","hashtags":[],"change_id":"Ie26623183a9e5e5d54456b73b79667ba9cf61f54","subject":"DNM: Update ubuntu image to 19.04","status":"ABANDONED","created":"2019-08-29 14:43:53.000000000","updated":"2019-08-31 06:47:57.000000000","total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"meta_rev_id":"8ec4552711caf4a0e9b4d4cfed13110f3054086c","_number":679277,"virtual_id_number":679277,"owner":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},"actions":{},"labels":{"Verified":{"disliked":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":-1,"date":"2019-08-29 16:19:26.000000000","permitted_voting_range":{"min":-2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"date":"2019-08-29 15:23:54.000000000","_account_id":23181,"name":"RDO Third Party CI","email":"dmsimard+rdothirdparty@redhat.com","username":"rdothirdparty","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","value":-1,"default_value":0,"optional":true},"Code-Review":{"all":[{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":23181,"name":"RDO Third Party CI","email":"dmsimard+rdothirdparty@redhat.com","username":"rdothirdparty","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"all":[{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":23181,"name":"RDO Third Party CI","email":"dmsimard+rdothirdparty@redhat.com","username":"rdothirdparty","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Backport-Candidate":{"all":[{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":23181,"name":"RDO Third Party CI","email":"dmsimard+rdothirdparty@redhat.com","username":"rdothirdparty","tags":["SERVICE_USER"]}],"values":{"-1":"Do Not Backport"," 0":"Backport Review Needed","+1":"Should Backport"},"description":"","default_value":0,"optional":true},"Review-Priority":{"all":[{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":23181,"name":"RDO Third Party CI","email":"dmsimard+rdothirdparty@redhat.com","username":"rdothirdparty","tags":["SERVICE_USER"]}],"values":{"-1":"Branch Freeze"," 0":"No Priority","+1":"Important Change","+2":"Gate Blocker Fix / Urgent Change"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":23181,"name":"RDO Third Party CI","email":"dmsimard+rdothirdparty@redhat.com","username":"rdothirdparty","tags":["SERVICE_USER"]}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2019-08-29 15:23:54.000000000","updated_by":{"_account_id":23181,"name":"RDO Third Party CI","email":"dmsimard+rdothirdparty@redhat.com","username":"rdothirdparty","tags":["SERVICE_USER"]},"reviewer":{"_account_id":23181,"name":"RDO Third Party CI","email":"dmsimard+rdothirdparty@redhat.com","username":"rdothirdparty","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2019-08-29 16:19:26.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"}],"messages":[{"id":"4feabcea6ffb2f1e1c655438b537668a25deebe1","author":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},"date":"2019-08-29 14:43:53.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"b8ba25ed6dc8b382f8074f7d02367dce81cbc445","author":{"_account_id":23181,"name":"RDO Third Party CI","email":"dmsimard+rdothirdparty@redhat.com","username":"rdothirdparty","tags":["SERVICE_USER"]},"date":"2019-08-29 15:23:54.000000000","message":"Patch Set 1:\n\nBuild succeeded (check pipeline).\n\n- tripleo-build-containers-rhel-8-master http://logs.rdoproject.org/77/679277/1/openstack-check/tripleo-build-containers-rhel-8-master/2b67aa3/ : SUCCESS in 38m 29s","accounts_in_message":[],"_revision_number":1},{"id":"bb89365a807ca532ad7d7d3f9813da44b3bdce7b","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-08-29 16:19:26.000000000","message":"Patch Set 1: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttp://docs.openstack.org/infra/manual/developers.html#automated-testing\n\n\n- kolla-build-centos-binary https://zuul.opendev.org/t/openstack/build/73f69701f5b24eae80836b42f0915485 : SUCCESS in 45m 46s (non-voting)\n- kolla-build-centos-source https://zuul.opendev.org/t/openstack/build/797adf78c00a4afe8458a859c26d1eb7 : SUCCESS in 54m 30s\n- kolla-ansible-centos-source https://zuul.opendev.org/t/openstack/build/fd1717cb84da44e6864526351eb216c3 : SUCCESS in 1h 22m 38s\n- kolla-ansible-centos-binary https://zuul.opendev.org/t/openstack/build/dd265510b9cb44fdaed194a8038447e4 : SUCCESS in 1h 01m 02s (non-voting)\n- kolla-ansible-centos-source-upgrade https://zuul.opendev.org/t/openstack/build/58bbe5b9e5bd48b8a3d928cfdac6eba0 : SUCCESS in 1h 20m 10s\n- tripleo-build-containers-centos-7 https://zuul.opendev.org/t/openstack/build/da028b91095643b3926cc274032870ad : SUCCESS in 50m 30s (non-voting)\n- kolla-build-debian-source https://zuul.opendev.org/t/openstack/build/8af5011a780548f0bea6794d88f38ea8 : SUCCESS in 21m 38s (non-voting)\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/525f49846d894fc59ee107208c71415b : SUCCESS in 4m 53s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/85ae6924658e4862b7f0d79f98f5860b : SUCCESS in 5m 23s\n- openstack-tox-py27 https://zuul.opendev.org/t/openstack/build/0dae076617ea4b42b75f19dbb9153cc0 : SUCCESS in 3m 53s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/a60f7878ba0042449341f71db34def42 : SUCCESS in 4m 04s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/7fca8ca553964381b3398a773f646a92 : SUCCESS in 3m 50s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/d8c4f04c6258445699a4ff16af7ae055 : SUCCESS in 3m 41s\n- kolla-build-ubuntu-binary https://zuul.opendev.org/t/openstack/build/5adf97c7899e4a8ea5c1c65fd556ad9c : FAILURE in 4m 32s (non-voting)\n- kolla-build-ubuntu-source https://zuul.opendev.org/t/openstack/build/9f3593cfd4384f6e80a9d06b9e5b5bb8 : FAILURE in 4m 03s\n- kolla-ansible-ubuntu-source https://zuul.opendev.org/t/openstack/build/3842a543b2d948a28797848e06393e61 : FAILURE in 7m 58s\n- kolla-ansible-ubuntu-binary https://zuul.opendev.org/t/openstack/build/0cc63f29e96540debb991d41f52dab8c : FAILURE in 10m 59s (non-voting)\n- kolla-ansible-ubuntu-source-upgrade https://zuul.opendev.org/t/openstack/build/d1ad6c7ad4fc4d91bd819e07ed65256b : FAILURE in 10m 35s","accounts_in_message":[],"_revision_number":1},{"id":"6c3ce0cd94a5180c68fe22898ef2bb17c6970263","author":{"_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"},"date":"2019-08-31 06:47:57.000000000","message":"Abandoned\n\nnon-lts needs changing more logic","accounts_in_message":[],"_revision_number":1}],"current_revision_number":1,"current_revision":"8a72e1f5fa9cc5700e68313d6bdbce44ba37a138","revisions":{"8a72e1f5fa9cc5700e68313d6bdbce44ba37a138":{"kind":"REWORK","_number":1,"created":"2019-08-29 14:43:53.000000000","uploader":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},"ref":"refs/changes/77/679277/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/kolla","ref":"refs/changes/77/679277/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/kolla refs/changes/77/679277/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/kolla refs/changes/77/679277/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/kolla refs/changes/77/679277/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/kolla refs/changes/77/679277/1"}}},"commit":{"parents":[{"commit":"857cf3effcda2dcd178fd74c70d8384f0c22ca40","subject":"Merge \"Add fluentd_binary and fluentd_version labels\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/857cf3effcda2dcd178fd74c70d8384f0c22ca40"}]}],"author":{"name":"Mark Goddard","email":"mark@stackhpc.com","date":"2019-08-29 14:41:09.000000000","tz":60},"committer":{"name":"Mark Goddard","email":"mark@stackhpc.com","date":"2019-08-29 14:41:09.000000000","tz":60},"subject":"DNM: Update ubuntu image to 19.04","message":"DNM: Update ubuntu image to 19.04\n\nUbuntu 19.04 includes mariadb 10.3, which should avoid the current issue\nwith neutron.\n\nChange-Id: Ie26623183a9e5e5d54456b73b79667ba9cf61f54\nRelated-Bug: #1841907\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/8a72e1f5fa9cc5700e68313d6bdbce44ba37a138"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/8a72e1f5fa9cc5700e68313d6bdbce44ba37a138"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[],"submit_requirements":[]}
