)]}'
{"id":"openstack%2Fkolla~679442","triplet_id":"openstack%2Fkolla~stable%2Fqueens~I6314c0ccd8c49b41a7393731d2e0a9536cca832e","project":"openstack/kolla","branch":"stable/queens","topic":"bump-versions-queens","hashtags":[],"change_id":"I6314c0ccd8c49b41a7393731d2e0a9536cca832e","subject":"Bump service versions for Queens","status":"MERGED","created":"2019-08-30 13:13:52.000000000","updated":"2019-09-03 18:18:12.000000000","submitted":"2019-09-03 18:18:12.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"679442-1567534692564-8928f7a0","meta_rev_id":"7a69186aba8d12d015083b9b54ddb1ed9f88b68f","_number":679442,"virtual_id_number":679442,"owner":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":24072,"name":"Marcin Juszkiewicz","email":"mjuszkiewicz@redhat.com","username":"hrw"},{"value":0,"date":"2019-09-03 16:02:42.000000000","_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},{"value":0,"_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"},{"value":2,"date":"2019-09-03 18:18:12.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":24072,"name":"Marcin Juszkiewicz","email":"mjuszkiewicz@redhat.com","username":"hrw"},"all":[{"value":2,"date":"2019-09-03 08:18:39.000000000","_account_id":24072,"name":"Marcin Juszkiewicz","email":"mjuszkiewicz@redhat.com","username":"hrw"},{"value":0,"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},{"value":2,"date":"2019-08-31 09:08:55.000000000","_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":24072,"name":"Marcin Juszkiewicz","email":"mjuszkiewicz@redhat.com","username":"hrw"},"all":[{"value":1,"date":"2019-09-03 08:18:39.000000000","_account_id":24072,"name":"Marcin Juszkiewicz","email":"mjuszkiewicz@redhat.com","username":"hrw"},{"value":0,"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},{"value":0,"_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Backport-Candidate":{"all":[{"value":0,"_account_id":24072,"name":"Marcin Juszkiewicz","email":"mjuszkiewicz@redhat.com","username":"hrw"},{"value":0,"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},{"value":0,"_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Do Not Backport"," 0":"Backport Review Needed","+1":"Should Backport"},"description":"","default_value":0,"optional":true},"Review-Priority":{"recommended":{"_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"},"all":[{"value":0,"_account_id":24072,"name":"Marcin Juszkiewicz","email":"mjuszkiewicz@redhat.com","username":"hrw"},{"value":0,"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},{"value":1,"date":"2019-08-31 09:08:55.000000000","_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Branch Freeze"," 0":"No Priority","+1":"Important Change","+2":"Gate Blocker Fix / Urgent Change"},"description":"","value":1,"default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":24072,"name":"Marcin Juszkiewicz","email":"mjuszkiewicz@redhat.com","username":"hrw"},{"_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2019-08-31 09:08:55.000000000","updated_by":{"_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"},"reviewer":{"_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"},"state":"REVIEWER"},{"updated":"2019-09-03 08:18:39.000000000","updated_by":{"_account_id":24072,"name":"Marcin Juszkiewicz","email":"mjuszkiewicz@redhat.com","username":"hrw"},"reviewer":{"_account_id":24072,"name":"Marcin Juszkiewicz","email":"mjuszkiewicz@redhat.com","username":"hrw"},"state":"REVIEWER"},{"updated":"2019-09-03 18:18:12.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"}],"messages":[{"id":"1f552ab9cf893906415feaa87483b2cdb616eeeb","author":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},"date":"2019-08-30 13:13:52.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"f46d37c0d420e4ebbae45a5bf139aca339b2dde5","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-08-30 14:58:21.000000000","message":"Patch Set 1: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttp://docs.openstack.org/infra/manual/developers.html#automated-testing\n\n\n- kolla-build-centos-binary https://zuul.opendev.org/t/openstack/build/7b2880eae2bc49e89e1d80c69a6acb49 : SUCCESS in 41m 21s (non-voting)\n- kolla-build-centos-source https://zuul.opendev.org/t/openstack/build/8350a3dc279642408c437b023e486c3d : FAILURE in 48m 47s\n- kolla-ansible-centos-source https://zuul.opendev.org/t/openstack/build/0747635aacc84aaf98350bd4a5ee8920 : SUCCESS in 59m 40s (non-voting)\n- kolla-ansible-centos-binary https://zuul.opendev.org/t/openstack/build/827164ad40cc464d8b36938c08450eaf : SUCCESS in 1h 02m 07s (non-voting)\n- kolla-build-oraclelinux-binary https://zuul.opendev.org/t/openstack/build/4c469562554041a5800959071ddcd64e : SUCCESS in 46m 52s (non-voting)\n- kolla-build-oraclelinux-source https://zuul.opendev.org/t/openstack/build/6ee77b36738f4909bef4eec431c6ca48 : FAILURE in 28m 13s\n- kolla-ansible-oraclelinux-source https://zuul.opendev.org/t/openstack/build/a33bc097566b4b07b6ca24017aaa483e : SUCCESS in 1h 00m 23s (non-voting)\n- kolla-ansible-oraclelinux-binary https://zuul.opendev.org/t/openstack/build/7236bdc73b5f4356aeb75e6fe0243a84 : SUCCESS in 1h 05m 16s (non-voting)\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/33d0f5abaaad4415ae443ad59254624e : SUCCESS in 4m 52s\n- openstack-tox-py27 https://zuul.opendev.org/t/openstack/build/aa41edf294f44c26a5764324f50cd95f : SUCCESS in 3m 59s\n- openstack-tox-py35 https://zuul.opendev.org/t/openstack/build/c17816cc84ee4a1daff551bbba71d077 : SUCCESS in 8m 22s\n- build-openstack-sphinx-docs https://zuul.opendev.org/t/openstack/build/9550bd28e63b46d0846af83c333f9066 : SUCCESS in 6m 34s\n- kolla-build-ubuntu-binary https://zuul.opendev.org/t/openstack/build/d1f9f173551640bc9f64fad2039bdd84 : FAILURE in 23m 35s (non-voting)\n- kolla-build-ubuntu-source https://zuul.opendev.org/t/openstack/build/80ad878393d34801876f9f6a4fadd0cb : FAILURE in 39m 58s\n- kolla-ansible-ubuntu-source https://zuul.opendev.org/t/openstack/build/747a3162f9f647bcbe7f65be7680df2d : FAILURE in 21m 54s (non-voting)\n- kolla-ansible-ubuntu-binary https://zuul.opendev.org/t/openstack/build/154d12430b94431ca459ae947a362a49 : FAILURE in 32m 52s (non-voting)","accounts_in_message":[],"_revision_number":1},{"id":"04e074ecf7cbf907f71b326cd212e36be7400fb9","author":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},"date":"2019-08-30 15:49:05.000000000","message":"Patch Set 2: Published edit on patch set 1.","accounts_in_message":[],"_revision_number":2},{"id":"ccd0aeb4d4d6a44defa42251e215b5a2c7845f62","author":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},"date":"2019-08-30 15:49:53.000000000","message":"Patch Set 2:\n\nHeat 10.0.3 not available on tarballs.o.o.","accounts_in_message":[],"_revision_number":2},{"id":"0ba1f00e0ade746b9c8818761252223f8d251872","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-08-30 17:04:58.000000000","message":"Patch Set 2: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttp://docs.openstack.org/infra/manual/developers.html#automated-testing\n\n\n- kolla-build-centos-binary https://zuul.opendev.org/t/openstack/build/ba57c1d2c6fe4a98987828c11c89c5e8 : SUCCESS in 34m 30s (non-voting)\n- kolla-build-centos-source https://zuul.opendev.org/t/openstack/build/b56ff8ae7ed846c5947275ea6cb01b80 : SUCCESS in 48m 30s\n- kolla-ansible-centos-source https://zuul.opendev.org/t/openstack/build/2b18713953a94918bc6885787148385c : SUCCESS in 1h 00m 11s (non-voting)\n- kolla-ansible-centos-binary https://zuul.opendev.org/t/openstack/build/bd81b8c68bf745e5849766e97f96dbba : SUCCESS in 1h 09m 35s (non-voting)\n- kolla-build-oraclelinux-binary https://zuul.opendev.org/t/openstack/build/7f823dbf422744f89dc8e1e024a43be1 : SUCCESS in 38m 15s (non-voting)\n- kolla-build-oraclelinux-source https://zuul.opendev.org/t/openstack/build/264300b3a6ce4595893ee4a30f53240e : SUCCESS in 53m 23s\n- kolla-ansible-oraclelinux-source https://zuul.opendev.org/t/openstack/build/0f8397831b6c4b77be1521cd383a7a6c : FAILURE in 45m 44s (non-voting)\n- kolla-ansible-oraclelinux-binary https://zuul.opendev.org/t/openstack/build/bc72f71a7a314eb5bc1e026aae3d7a7a : SUCCESS in 1h 04m 00s (non-voting)\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/c6d9243627914eb3945082e3c394e353 : SUCCESS in 4m 48s\n- openstack-tox-py27 https://zuul.opendev.org/t/openstack/build/4cb1723c70b84cd0a08043e736a8be46 : SUCCESS in 4m 28s\n- openstack-tox-py35 https://zuul.opendev.org/t/openstack/build/5cfaa6bda4d34309bc085b31ea987785 : SUCCESS in 5m 07s\n- build-openstack-sphinx-docs https://zuul.opendev.org/t/openstack/build/0b7c2af955c24bf39d06243f09a9d6b8 : SUCCESS in 5m 34s\n- kolla-build-ubuntu-binary https://zuul.opendev.org/t/openstack/build/7471352361f24ddda29b45df11bfe5fc : FAILURE in 34m 06s (non-voting)\n- kolla-build-ubuntu-source https://zuul.opendev.org/t/openstack/build/e16ba8fb848142189eb775026af72437 : FAILURE in 31m 22s\n- kolla-ansible-ubuntu-source https://zuul.opendev.org/t/openstack/build/bb0be5999a824d97817859b548ac6249 : FAILURE in 26m 29s (non-voting)\n- kolla-ansible-ubuntu-binary https://zuul.opendev.org/t/openstack/build/0eaad900a40f4672b98cf4c553af6d49 : FAILURE in 33m 15s (non-voting)","accounts_in_message":[],"_revision_number":2},{"id":"70c1979a2b5aecd76dd61c36dd942e1350de82ef","author":{"_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"},"date":"2019-08-31 06:45:33.000000000","message":"Patch Set 3: Patch Set 2 was rebased","accounts_in_message":[],"_revision_number":3},{"id":"2b0a7ec5d5819f0160066b3e3612f7174d0b42e8","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-08-31 08:01:10.000000000","message":"Patch Set 3: Verified+1\n\nBuild succeeded (check pipeline).\n\n- kolla-build-centos-binary https://zuul.opendev.org/t/openstack/build/d492c7a9a7f44360bbbe610d9f8e1aed : SUCCESS in 37m 47s (non-voting)\n- kolla-build-centos-source https://zuul.opendev.org/t/openstack/build/d329171a47724bee9f90958c6147511f : SUCCESS in 36m 27s\n- kolla-ansible-centos-source https://zuul.opendev.org/t/openstack/build/5171790abbe3498e879efdce4375b70a : SUCCESS in 1h 05m 24s (non-voting)\n- kolla-ansible-centos-binary https://zuul.opendev.org/t/openstack/build/982045610557406fa2b3902cde9c6d1b : SUCCESS in 1h 11m 21s (non-voting)\n- kolla-build-oraclelinux-binary https://zuul.opendev.org/t/openstack/build/ce1c32ce4a11436e9ca5ee1acc353487 : SUCCESS in 41m 06s (non-voting)\n- kolla-build-oraclelinux-source https://zuul.opendev.org/t/openstack/build/c27174fc455b497aa5de61d92e23604f : SUCCESS in 48m 14s\n- kolla-ansible-oraclelinux-source https://zuul.opendev.org/t/openstack/build/c42ba6cb63ab4ac08e3cdd2784d365ba : SUCCESS in 59m 31s (non-voting)\n- kolla-ansible-oraclelinux-binary https://zuul.opendev.org/t/openstack/build/88a0b43d90724f7795302aa15181b023 : SUCCESS in 1h 13m 50s (non-voting)\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/a16146921e854bdb921864f4c15dc3fe : SUCCESS in 5m 16s\n- openstack-tox-py27 https://zuul.opendev.org/t/openstack/build/500cecabd57148a5bbaf385cdb3ee01b : SUCCESS in 4m 08s\n- openstack-tox-py35 https://zuul.opendev.org/t/openstack/build/8a5943ee2a234b948d8850f81bcca37e : SUCCESS in 4m 52s\n- build-openstack-sphinx-docs https://zuul.opendev.org/t/openstack/build/58f96e251d914c17bdffc3a27b24a654 : SUCCESS in 5m 57s\n- kolla-build-ubuntu-binary https://zuul.opendev.org/t/openstack/build/fd7b01853efe4e7781634ed0d8e00754 : SUCCESS in 25m 11s (non-voting)\n- kolla-build-ubuntu-source https://zuul.opendev.org/t/openstack/build/f84dbc389ac942e09e7c14b39362d3e7 : SUCCESS in 30m 29s\n- kolla-ansible-ubuntu-source https://zuul.opendev.org/t/openstack/build/fea8d5ff73ba4ba9b2db8340ac31e1bd : SUCCESS in 1h 03m 44s (non-voting)\n- kolla-ansible-ubuntu-binary https://zuul.opendev.org/t/openstack/build/7054b205670742a191f278408ac3d938 : SUCCESS in 56m 28s (non-voting)","accounts_in_message":[],"_revision_number":3},{"id":"95af6b4630483a2c57b255fe75575bac7000ce91","author":{"_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"},"date":"2019-08-31 09:08:55.000000000","message":"Patch Set 3: Code-Review+2 Review-Priority+1","accounts_in_message":[],"_revision_number":3},{"id":"081f7a311b3a506e3a94f290acf1c8209631e682","author":{"_account_id":24072,"name":"Marcin Juszkiewicz","email":"mjuszkiewicz@redhat.com","username":"hrw"},"date":"2019-09-03 08:18:39.000000000","message":"Patch Set 3: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":3},{"id":"28f361101304d37522d2fcd8e95600354b009853","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-09-03 08:18:51.000000000","message":"Patch Set 3: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":3},{"id":"7c9364de090fba1db55841bd10cd204b943653c3","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-09-03 09:33:20.000000000","message":"Patch Set 3: Verified-2\n\nBuild failed (gate pipeline).  For information on how to proceed, see\nhttp://docs.openstack.org/infra/manual/developers.html#automated-testing\n\n\n- kolla-build-centos-source https://zuul.opendev.org/t/openstack/build/e75911e1792145d1bf52081c3a91bf8d : FAILURE in 38m 18s\n- kolla-build-oraclelinux-source https://zuul.opendev.org/t/openstack/build/8a6da8574b6c40668a94026e609c83bb : SUCCESS in 28m 04s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/749c27e3c5864a13acd40a0971a8681c : SUCCESS in 4m 54s\n- openstack-tox-py27 https://zuul.opendev.org/t/openstack/build/e9fa6f2d3f3b48d99b8289909ab61b68 : SUCCESS in 4m 25s\n- openstack-tox-py35 https://zuul.opendev.org/t/openstack/build/4fb4f275d40d47e586e7b2d69b4d86c3 : SUCCESS in 4m 14s\n- build-openstack-sphinx-docs https://zuul.opendev.org/t/openstack/build/45fbf4a9a9384c3ab384178125575377 : SUCCESS in 4m 54s\n- kolla-build-ubuntu-source https://zuul.opendev.org/t/openstack/build/b6e3f7b1a5f24e41bd99200c8568ba20 : SUCCESS in 49m 24s","accounts_in_message":[],"_revision_number":3},{"id":"15ec0808bdece96e48b7d474a69e23cb146841a1","author":{"_account_id":24072,"name":"Marcin Juszkiewicz","email":"mjuszkiewicz@redhat.com","username":"hrw"},"date":"2019-09-03 11:15:39.000000000","message":"Patch Set 3:\n\nrecheck","accounts_in_message":[],"_revision_number":3},{"id":"8bc9736036084a71ea13f955a0510a9a2a64294e","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-09-03 12:26:54.000000000","message":"Patch Set 3: Verified+1\n\nBuild succeeded (check pipeline).\n\n- kolla-build-centos-binary https://zuul.opendev.org/t/openstack/build/58f8ecd71af14efc9be231ef428a19ea : SUCCESS in 39m 49s (non-voting)\n- kolla-build-centos-source https://zuul.opendev.org/t/openstack/build/eb9199313c8d4e46bf06aae24c0b52a4 : SUCCESS in 44m 15s\n- kolla-ansible-centos-source https://zuul.opendev.org/t/openstack/build/d68d445e4f3246d0aa03ff23d8baa509 : SUCCESS in 1h 01m 55s (non-voting)\n- kolla-ansible-centos-binary https://zuul.opendev.org/t/openstack/build/4e3eacce857546a2bb673dac1dd8fe5a : SUCCESS in 1h 01m 38s (non-voting)\n- kolla-build-oraclelinux-binary https://zuul.opendev.org/t/openstack/build/ce0f264bdc7148d88dfe229224cad283 : SUCCESS in 39m 46s (non-voting)\n- kolla-build-oraclelinux-source https://zuul.opendev.org/t/openstack/build/4ba2cbad86e54177b2be8ac3cb92b8c6 : SUCCESS in 49m 11s\n- kolla-ansible-oraclelinux-source https://zuul.opendev.org/t/openstack/build/2c7d78b9c3c542d58e08e8081f0706a4 : SUCCESS in 1h 00m 24s (non-voting)\n- kolla-ansible-oraclelinux-binary https://zuul.opendev.org/t/openstack/build/d2c597fa07ab41a1a01785eb6058ef71 : SUCCESS in 1h 02m 40s (non-voting)\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/8424d0db8a324a4795b23dfade55c535 : SUCCESS in 5m 29s\n- openstack-tox-py27 https://zuul.opendev.org/t/openstack/build/32e4c69f62a043d49a591c80f8d076ab : SUCCESS in 4m 41s\n- openstack-tox-py35 https://zuul.opendev.org/t/openstack/build/af81f0c6ca2a4fe98cf2e078f8b1c16c : SUCCESS in 5m 18s\n- build-openstack-sphinx-docs https://zuul.opendev.org/t/openstack/build/d774ad8f9dec41629112e8219276584c : SUCCESS in 4m 15s\n- kolla-build-ubuntu-binary https://zuul.opendev.org/t/openstack/build/05b0fd8362204b908d7f53c05e57524a : SUCCESS in 35m 13s (non-voting)\n- kolla-build-ubuntu-source https://zuul.opendev.org/t/openstack/build/3e0e40f833e94356bf3374be2c0a238e : SUCCESS in 32m 15s\n- kolla-ansible-ubuntu-source https://zuul.opendev.org/t/openstack/build/d18e0294e91c451f8125edf37002256f : SUCCESS in 55m 01s (non-voting)\n- kolla-ansible-ubuntu-binary https://zuul.opendev.org/t/openstack/build/c0e3796b56d048e3b4d547869578c2e3 : SUCCESS in 58m 18s (non-voting)","accounts_in_message":[],"_revision_number":3},{"id":"b23af20d6db0790c3c55e1a0ab15e4735b42b2d8","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-09-03 12:27:09.000000000","message":"Patch Set 3: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":3},{"id":"b71879f6333bc624671be8b8ae3db9ffdbb454f5","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-09-03 13:30:26.000000000","message":"Patch Set 3: Verified-2\n\nBuild failed (gate pipeline).  For information on how to proceed, see\nhttp://docs.openstack.org/infra/manual/developers.html#automated-testing\n\n\n- kolla-build-centos-source https://zuul.opendev.org/t/openstack/build/0df51c0ef7154cecb4094843881e0995 : SUCCESS in 32m 25s\n- kolla-build-oraclelinux-source https://zuul.opendev.org/t/openstack/build/e74e4ee2697945b788f6c05861594850 : SUCCESS in 58m 30s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/9523715734f8447db485dba5013a42ca : SUCCESS in 4m 45s\n- openstack-tox-py27 https://zuul.opendev.org/t/openstack/build/52037c22fbbd4ad4977cb302372bbbec : SUCCESS in 4m 26s\n- openstack-tox-py35 https://zuul.opendev.org/t/openstack/build/19af0691a2d54ec99beaf52fd3728316 : SUCCESS in 6m 02s\n- build-openstack-sphinx-docs https://zuul.opendev.org/t/openstack/build/a01c4cdabfee40ceb6e4e4b9361d8aba : SUCCESS in 5m 23s\n- kolla-build-ubuntu-source https://zuul.opendev.org/t/openstack/build/82257f4c3b764a2393152cecf21969e3 : FAILURE in 27m 43s","accounts_in_message":[],"_revision_number":3},{"id":"3c76160e3f9ca55322d4b745a667792110c5cd53","author":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},"date":"2019-09-03 13:50:05.000000000","message":"Patch Set 3:\n\nrecheck","accounts_in_message":[],"_revision_number":3},{"id":"4d4c6572332f241e8233966d40808ce140796b9a","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-09-03 15:55:16.000000000","message":"Patch Set 3: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttp://docs.openstack.org/infra/manual/developers.html#automated-testing\n\n\n- kolla-build-centos-binary https://zuul.opendev.org/t/openstack/build/0823d24571c7427eac7d685ac504b51e : SUCCESS in 33m 50s (non-voting)\n- kolla-build-centos-source https://zuul.opendev.org/t/openstack/build/f775c5fd83d84defa03eb47aa51d43a7 : SUCCESS in 34m 28s\n- kolla-ansible-centos-source https://zuul.opendev.org/t/openstack/build/7901807899464fb592d24b3b02def37c : TIMED_OUT in 2h 02m 14s (non-voting)\n- kolla-ansible-centos-binary https://zuul.opendev.org/t/openstack/build/5e19bf3c539e45cd9732d94166131b2c : SUCCESS in 1h 07m 06s (non-voting)\n- kolla-build-oraclelinux-binary https://zuul.opendev.org/t/openstack/build/45d2b5593ec94648b2291755c5af257a : SUCCESS in 46m 58s (non-voting)\n- kolla-build-oraclelinux-source https://zuul.opendev.org/t/openstack/build/7aedf0cc811244169a2b84534587f3fe : RETRY_LIMIT in 2m 44s\n- kolla-ansible-oraclelinux-source https://zuul.opendev.org/t/openstack/build/42f67415a980402bb6f83e06e21b0917 : SUCCESS in 1h 08m 33s (non-voting)\n- kolla-ansible-oraclelinux-binary https://zuul.opendev.org/t/openstack/build/61596eb5252545e58a04a2216a812669 : SUCCESS in 1h 13m 35s (non-voting)\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/54dfa0e193364464b0f66fde3ab5498a : SUCCESS in 4m 39s\n- openstack-tox-py27 https://zuul.opendev.org/t/openstack/build/3a61cc97424041d6af8d5de9f223ec08 : SUCCESS in 6m 05s\n- openstack-tox-py35 https://zuul.opendev.org/t/openstack/build/d8df9bfb4fad42b28466d70ac4087cc6 : SUCCESS in 4m 26s\n- build-openstack-sphinx-docs https://zuul.opendev.org/t/openstack/build/b754c1cb2ec9407ca083d155671bea09 : SUCCESS in 5m 28s\n- kolla-build-ubuntu-binary https://zuul.opendev.org/t/openstack/build/cf3f75afd17a4c6aab1191df19cc6dd3 : SUCCESS in 32m 15s (non-voting)\n- kolla-build-ubuntu-source https://zuul.opendev.org/t/openstack/build/b7d94f1f8fb542fd9fe7770c3804e750 : SUCCESS in 33m 49s\n- kolla-ansible-ubuntu-source https://zuul.opendev.org/t/openstack/build/8f64ba7c08a34ab6a736bf8f4d674f23 : SUCCESS in 58m 11s (non-voting)\n- kolla-ansible-ubuntu-binary https://zuul.opendev.org/t/openstack/build/c99a0c0443524e54875fc631f15b0cd0 : SUCCESS in 1h 10m 07s (non-voting)","accounts_in_message":[],"_revision_number":3},{"id":"e4ea69ee48fdafae39af36e3749a367f0e12aa37","author":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},"date":"2019-09-03 16:02:42.000000000","message":"Patch Set 3:\n\nrecheck","accounts_in_message":[],"_revision_number":3},{"id":"080505bc4ffceea3b6f75a1711f63c095f3c10fd","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-09-03 17:17:24.000000000","message":"Patch Set 3: Verified+1\n\nBuild succeeded (check pipeline).\n\n- kolla-build-centos-binary https://zuul.opendev.org/t/openstack/build/dddb823e39ee4ff3ad8c5346e701b359 : SUCCESS in 34m 57s (non-voting)\n- kolla-build-centos-source https://zuul.opendev.org/t/openstack/build/05859da77b07448689bb02c09696d758 : SUCCESS in 41m 34s\n- kolla-ansible-centos-source https://zuul.opendev.org/t/openstack/build/a87bacac587545fbbefe5fb71d452adc : SUCCESS in 1h 11m 18s (non-voting)\n- kolla-ansible-centos-binary https://zuul.opendev.org/t/openstack/build/63c795b28e214dd894ac7d313d9aa22b : SUCCESS in 59m 45s (non-voting)\n- kolla-build-oraclelinux-binary https://zuul.opendev.org/t/openstack/build/76d1ce3b073c43498e2e4192f3fe5035 : SUCCESS in 48m 23s (non-voting)\n- kolla-build-oraclelinux-source https://zuul.opendev.org/t/openstack/build/aa82063922634a66a30c15c74048171c : SUCCESS in 45m 27s\n- kolla-ansible-oraclelinux-source https://zuul.opendev.org/t/openstack/build/cce556091dbe41f382b11488cdb34664 : SUCCESS in 1h 02m 50s (non-voting)\n- kolla-ansible-oraclelinux-binary https://zuul.opendev.org/t/openstack/build/7bc319ad1442486c889933cdb67aa4a9 : SUCCESS in 1h 04m 58s (non-voting)\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/4d2c3a15ef3a49959fb3af0a2ef48f49 : SUCCESS in 4m 49s\n- openstack-tox-py27 https://zuul.opendev.org/t/openstack/build/bb005f837d814af4873e49f5cb7b7f4f : SUCCESS in 4m 41s\n- openstack-tox-py35 https://zuul.opendev.org/t/openstack/build/984b253d61854b5cbc17b220dfb1efc0 : SUCCESS in 6m 04s\n- build-openstack-sphinx-docs https://zuul.opendev.org/t/openstack/build/143475c2548d490d9f7d976b09147328 : SUCCESS in 6m 43s\n- kolla-build-ubuntu-binary https://zuul.opendev.org/t/openstack/build/71ddcf3e18b847f580b80244019e855c : SUCCESS in 1h 02m 04s (non-voting)\n- kolla-build-ubuntu-source https://zuul.opendev.org/t/openstack/build/91e42116845c4dfdae2378b27a2f61a9 : SUCCESS in 40m 27s\n- kolla-ansible-ubuntu-source https://zuul.opendev.org/t/openstack/build/7eaf21e53cfd44ac9728ed2ccfe27295 : SUCCESS in 58m 28s (non-voting)\n- kolla-ansible-ubuntu-binary https://zuul.opendev.org/t/openstack/build/bc2ccc4630014743863c6894a3bb03ad : SUCCESS in 1h 03m 04s (non-voting)","accounts_in_message":[],"_revision_number":3},{"id":"a392bd2d8901b40d0d2a0b1f0b790468976bfaf5","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-09-03 17:17:36.000000000","message":"Patch Set 3: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":3},{"id":"52ae652912c49af7cb3656d6df2ac18ea6b0844c","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-09-03 18:18:12.000000000","message":"Patch Set 3: Verified+2\n\nBuild succeeded (gate pipeline).\n\n- kolla-build-centos-source https://zuul.opendev.org/t/openstack/build/c731a3e57ed04f0f8faef6e64bb04c8b : SUCCESS in 39m 29s\n- kolla-build-oraclelinux-source https://zuul.opendev.org/t/openstack/build/549f734d8a32482aa276839c02ccdf1b : SUCCESS in 54m 16s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/ac69c09242f04edab14b9580593b211a : SUCCESS in 4m 44s\n- openstack-tox-py27 https://zuul.opendev.org/t/openstack/build/d3223edb198540cfb68317f6f997ee18 : SUCCESS in 4m 24s\n- openstack-tox-py35 https://zuul.opendev.org/t/openstack/build/deecac2e80c940648e8f48d6bb84cd37 : SUCCESS in 4m 52s\n- build-openstack-sphinx-docs https://zuul.opendev.org/t/openstack/build/7bfa2016ef7a4c3cab0c4f41d7c19984 : SUCCESS in 5m 14s\n- kolla-build-ubuntu-source https://zuul.opendev.org/t/openstack/build/9d21b007c1c24412b329fe90dfcea051 : SUCCESS in 35m 24s","accounts_in_message":[],"_revision_number":3},{"id":"7dcd3331e4f92d46963a0c927c587a700788e0ab","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-09-03 18:18:12.000000000","message":"Change has been successfully merged by Zuul","accounts_in_message":[],"_revision_number":3}],"current_revision_number":3,"current_revision":"c717c5872db2e73c8243bc2a3b8974c91fe11464","revisions":{"a5fff80cf43ed2eb42043d5266770ad5ec355cc9":{"kind":"REWORK","_number":1,"created":"2019-08-30 13:13:52.000000000","uploader":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},"ref":"refs/changes/42/679442/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/kolla","ref":"refs/changes/42/679442/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/kolla refs/changes/42/679442/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/kolla refs/changes/42/679442/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/kolla refs/changes/42/679442/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/kolla refs/changes/42/679442/1"}}},"commit":{"parents":[{"commit":"13ad3974560dfb9cc89b0f966a61f7a592efee30","subject":"Fix ovs-appctl by adding --pidfile to ovsdb command.","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/13ad3974560dfb9cc89b0f966a61f7a592efee30"}]}],"author":{"name":"Mark Goddard","email":"mark@stackhpc.com","date":"2019-08-30 13:12:56.000000000","tz":60},"committer":{"name":"Mark Goddard","email":"mark@stackhpc.com","date":"2019-08-30 13:13:20.000000000","tz":60},"subject":"Bump service versions for Queens","message":"Bump service versions for Queens\n\nChange-Id: I6314c0ccd8c49b41a7393731d2e0a9536cca832e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/a5fff80cf43ed2eb42043d5266770ad5ec355cc9"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/a5fff80cf43ed2eb42043d5266770ad5ec355cc9"}]},"branch":"refs/heads/stable/queens"},"f0896bae418515ce834fe78e2bba05742cae86a5":{"kind":"REWORK","_number":2,"created":"2019-08-30 15:49:05.000000000","uploader":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},"ref":"refs/changes/42/679442/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/kolla","ref":"refs/changes/42/679442/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/kolla refs/changes/42/679442/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/kolla refs/changes/42/679442/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/kolla refs/changes/42/679442/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/kolla refs/changes/42/679442/2"}}},"commit":{"parents":[{"commit":"13ad3974560dfb9cc89b0f966a61f7a592efee30","subject":"Fix ovs-appctl by adding --pidfile to ovsdb command.","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/13ad3974560dfb9cc89b0f966a61f7a592efee30"}]}],"author":{"name":"Mark Goddard","email":"mark@stackhpc.com","date":"2019-08-30 13:12:56.000000000","tz":60},"committer":{"name":"Mark Goddard","email":"mark@stackhpc.com","date":"2019-08-30 15:49:02.000000000","tz":0},"subject":"Bump service versions for Queens","message":"Bump service versions for Queens\n\nChange-Id: I6314c0ccd8c49b41a7393731d2e0a9536cca832e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/f0896bae418515ce834fe78e2bba05742cae86a5"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/f0896bae418515ce834fe78e2bba05742cae86a5"}]},"branch":"refs/heads/stable/queens"},"c717c5872db2e73c8243bc2a3b8974c91fe11464":{"kind":"TRIVIAL_REBASE","_number":3,"created":"2019-08-31 06:45:33.000000000","uploader":{"_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"},"ref":"refs/changes/42/679442/3","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/kolla","ref":"refs/changes/42/679442/3","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/kolla refs/changes/42/679442/3 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/kolla refs/changes/42/679442/3 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/kolla refs/changes/42/679442/3 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/kolla refs/changes/42/679442/3"}}},"commit":{"parents":[{"commit":"1e2ff199d4d639f45e7c476bb555d12316a6bfe3","subject":"Unbreak CI: fluentd","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/1e2ff199d4d639f45e7c476bb555d12316a6bfe3"}]}],"author":{"name":"Mark Goddard","email":"mark@stackhpc.com","date":"2019-08-30 13:12:56.000000000","tz":60},"committer":{"name":"Radosław Piliszek","email":"radoslaw.piliszek@gmail.com","date":"2019-08-31 06:45:33.000000000","tz":0},"subject":"Bump service versions for Queens","message":"Bump service versions for Queens\n\nChange-Id: I6314c0ccd8c49b41a7393731d2e0a9536cca832e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/c717c5872db2e73c8243bc2a3b8974c91fe11464"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/c717c5872db2e73c8243bc2a3b8974c91fe11464"}]},"branch":"refs/heads/stable/queens"}},"requirements":[],"submit_records":[],"submit_requirements":[]}
