)]}'
{"id":"openstack%2Fkolla~679617","triplet_id":"openstack%2Fkolla~stable%2Frocky~Ib31e733eba12deebdf469c306dfba4191a1b89f5","project":"openstack/kolla","branch":"stable/rocky","topic":"bug/1832306-stable/stein-stable/rocky","hashtags":[],"change_id":"Ib31e733eba12deebdf469c306dfba4191a1b89f5","subject":"Delete default gnocchi \u0026 aodh wsgi conf in Ubuntu binary","status":"MERGED","created":"2019-09-02 09:42:34.000000000","updated":"2019-09-03 09:25:32.000000000","submitted":"2019-09-03 09:25:32.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"679617-1567502732459-3374a85a","meta_rev_id":"1505b518fee837619de4a46152d5fd3165754b28","_number":679617,"virtual_id_number":679617,"owner":{"_account_id":30687,"name":"Eddie Yen","email":"missile0407@gmail.com","username":"eddieak"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},{"value":0,"_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"},{"value":2,"date":"2019-09-03 09:25:32.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},"all":[{"value":2,"date":"2019-09-02 15:15:43.000000000","_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},{"value":2,"date":"2019-09-03 07:58:18.000000000","_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"},"all":[{"value":0,"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},{"value":1,"date":"2019-09-03 07:58:18.000000000","_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Backport-Candidate":{"all":[{"value":0,"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},{"value":0,"_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Do Not Backport"," 0":"Backport Review Needed","+1":"Should Backport"},"description":"","default_value":0,"optional":true},"Review-Priority":{"all":[{"value":0,"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},{"value":0,"_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Branch Freeze"," 0":"No Priority","+1":"Important Change","+2":"Gate Blocker Fix / Urgent Change"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2019-09-02 15:15:43.000000000","updated_by":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},"reviewer":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},"state":"REVIEWER"},{"updated":"2019-09-03 07:58:18.000000000","updated_by":{"_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"},"reviewer":{"_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"},"state":"REVIEWER"},{"updated":"2019-09-03 09:25:32.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"}],"messages":[{"id":"093dad45535ba9139bbf7ef6bd9586850051c1cc","author":{"_account_id":30687,"name":"Eddie Yen","email":"missile0407@gmail.com","username":"eddieak"},"date":"2019-09-02 09:42:34.000000000","message":"Patch Set 1: Cherry Picked from branch stable/stein.","accounts_in_message":[],"_revision_number":1},{"id":"98e3008456c3ad0c6de6ce503c26b498367f571d","author":{"_account_id":30687,"name":"Eddie Yen","email":"missile0407@gmail.com","username":"eddieak"},"date":"2019-09-02 09:44:31.000000000","message":"Patch Set 2: Published edit on patch set 1.","accounts_in_message":[],"_revision_number":2},{"id":"08aada469359357c5c5da2a6c49a918692f2ffc3","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-09-02 11:55:46.000000000","message":"Patch Set 2: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttp://docs.openstack.org/infra/manual/developers.html#automated-testing\n\n\n- kolla-build-centos-binary https://zuul.opendev.org/t/openstack/build/55839cc866294b2182a11c435e6fb2e0 : SUCCESS in 42m 07s (non-voting)\n- kolla-build-centos-source https://zuul.opendev.org/t/openstack/build/398eebbb695c40babac91f4b690cc9a9 : SUCCESS in 49m 11s\n- kolla-ansible-centos-source https://zuul.opendev.org/t/openstack/build/48395614dd8c4a98aa39eca3b34363e5 : SUCCESS in 1h 08m 56s\n- kolla-ansible-centos-binary https://zuul.opendev.org/t/openstack/build/a3197370aae043bf95615808be2a100f : SUCCESS in 1h 11m 45s (non-voting)\n- kolla-build-oraclelinux-binary https://zuul.opendev.org/t/openstack/build/5ded6bec808842f58ae4d252a21a0326 : SUCCESS in 47m 17s (non-voting)\n- kolla-build-oraclelinux-source https://zuul.opendev.org/t/openstack/build/fe99a51b6dc7480d9ed34aa8237a2891 : SUCCESS in 45m 01s\n- kolla-ansible-oraclelinux-source https://zuul.opendev.org/t/openstack/build/7d74f1decdc14000bdee1b8567622a6a : SUCCESS in 1h 23m 49s (non-voting)\n- kolla-ansible-oraclelinux-binary https://zuul.opendev.org/t/openstack/build/15e4ecbb0cca41f0bbe9fea87edb3f81 : SUCCESS in 1h 08m 03s (non-voting)\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/f59f9b274a714ad3b7b15f9b11b37431 : SUCCESS in 3m 45s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/2b270fea3f5c429b969662301296f576 : SUCCESS in 4m 24s\n- openstack-tox-py27 https://zuul.opendev.org/t/openstack/build/efd5c0db882a4822a4a6adb67b64a36f : SUCCESS in 3m 50s\n- openstack-tox-py35 https://zuul.opendev.org/t/openstack/build/c019e0f7cba94758a7494b2b5be798e9 : SUCCESS in 3m 41s\n- build-openstack-sphinx-docs https://zuul.opendev.org/t/openstack/build/4c751e4ef5c14b02af2c540d88fda14c : SUCCESS in 4m 41s\n- kolla-build-ubuntu-binary https://zuul.opendev.org/t/openstack/build/1c3ce41ef37e44489d1e01390dfd9fe4 : SUCCESS in 1h 27m 21s (non-voting)\n- kolla-build-ubuntu-source https://zuul.opendev.org/t/openstack/build/09c1d9d2472149c19f8a3295d4666457 : FAILURE in 2h 05m 34s\n- kolla-ansible-ubuntu-source https://zuul.opendev.org/t/openstack/build/157842e785cb437293154ffd2cd5b2e7 : SUCCESS in 1h 08m 24s\n- kolla-ansible-ubuntu-binary https://zuul.opendev.org/t/openstack/build/26479c2a5dcb4f68adecf567058360eb : SUCCESS in 1h 32m 04s (non-voting)","accounts_in_message":[],"_revision_number":2},{"id":"a13047995658163ee0c961ab38424d25ebb236f2","author":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},"date":"2019-09-02 12:38:29.000000000","message":"Patch Set 2:\n\nrecheck","accounts_in_message":[],"_revision_number":2},{"id":"fbab1d986b757b47d384f81f9f83689396ed6bb8","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-09-02 14:48:42.000000000","message":"Patch Set 2: Verified+1\n\nBuild succeeded (check pipeline).\n\n- kolla-build-centos-binary https://zuul.opendev.org/t/openstack/build/9797ab4c82984b59b5ba5f45e8d35fe8 : SUCCESS in 23m 51s (non-voting)\n- kolla-build-centos-source https://zuul.opendev.org/t/openstack/build/279bfbbbf41c4e19885eb20596133fc3 : SUCCESS in 51m 06s\n- kolla-ansible-centos-source https://zuul.opendev.org/t/openstack/build/43798e0b2cbc4d40a1bd548e71bf1bed : SUCCESS in 1h 14m 32s\n- kolla-ansible-centos-binary https://zuul.opendev.org/t/openstack/build/5960613da259437697957b713bf88707 : SUCCESS in 1h 04m 38s (non-voting)\n- kolla-build-oraclelinux-binary https://zuul.opendev.org/t/openstack/build/78a6c1d71753482f89f8ed62cfd20792 : SUCCESS in 43m 38s (non-voting)\n- kolla-build-oraclelinux-source https://zuul.opendev.org/t/openstack/build/ee4393244ffc4421ae72edeac07bc69f : SUCCESS in 34m 54s\n- kolla-ansible-oraclelinux-source https://zuul.opendev.org/t/openstack/build/1b83395563e14fdaa7600b80a522f24b : SUCCESS in 1h 14m 02s (non-voting)\n- kolla-ansible-oraclelinux-binary https://zuul.opendev.org/t/openstack/build/2d14adb3bba84e3ab50c312886472c22 : SUCCESS in 1h 11m 27s (non-voting)\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/d8570e3ad7434d26a9ccf6026d6724e5 : SUCCESS in 6m 53s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/55e9e60bc8494179ada80277ec3c3056 : SUCCESS in 4m 04s\n- openstack-tox-py27 https://zuul.opendev.org/t/openstack/build/485b41b082f44af3a8e6eb48d48760f5 : SUCCESS in 4m 14s\n- openstack-tox-py35 https://zuul.opendev.org/t/openstack/build/027a917c0cd248ab8a83314cc1f88189 : SUCCESS in 6m 39s\n- build-openstack-sphinx-docs https://zuul.opendev.org/t/openstack/build/14886b827a3242439758ba46685d89a4 : SUCCESS in 4m 32s\n- kolla-build-ubuntu-binary https://zuul.opendev.org/t/openstack/build/07a93c848c5047d89300628dcc3a3939 : SUCCESS in 1h 29m 51s (non-voting)\n- kolla-build-ubuntu-source https://zuul.opendev.org/t/openstack/build/48398c7602b74342b5e1ab0f9c844bd6 : SUCCESS in 44m 29s\n- kolla-ansible-ubuntu-source https://zuul.opendev.org/t/openstack/build/9f41f462a7824a219e3d647adb011195 : SUCCESS in 1h 05m 58s\n- kolla-ansible-ubuntu-binary https://zuul.opendev.org/t/openstack/build/6afcb147114148faa7b26837873f144b : TIMED_OUT in 2h 04m 38s (non-voting)","accounts_in_message":[],"_revision_number":2},{"id":"6a77abc9eceab5af1207efa6c6877fdd28cb6c60","author":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},"date":"2019-09-02 15:15:43.000000000","message":"Patch Set 2: Code-Review+2","accounts_in_message":[],"_revision_number":2},{"id":"fc9f764a837c3c2665e05aac3ec5d72c3ef9db95","author":{"_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"},"date":"2019-09-03 07:58:18.000000000","message":"Patch Set 2: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":2},{"id":"22f27c9597838f769a5bf4ac3024bd49ee585d5b","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-09-03 07:58:31.000000000","message":"Patch Set 2: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":2},{"id":"05ef66fb2aa2b2cffb780d48f3f9d33896204485","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-09-03 09:25:32.000000000","message":"Patch Set 2: Verified+2\n\nBuild succeeded (gate pipeline).\n\n- kolla-build-centos-source https://zuul.opendev.org/t/openstack/build/3727c078f41d4248a06b35e7a9518bb7 : SUCCESS in 50m 07s\n- kolla-build-oraclelinux-source https://zuul.opendev.org/t/openstack/build/8b5d568721c64a3c8efd69d218067f16 : SUCCESS in 40m 12s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/96cc2a620a584bac8780289eec643e18 : SUCCESS in 5m 53s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/f42822a9b55e43adbdce1c20be33ba69 : SUCCESS in 4m 41s\n- openstack-tox-py27 https://zuul.opendev.org/t/openstack/build/9608c0702bb64d10b9024f171d62df45 : SUCCESS in 4m 41s\n- openstack-tox-py35 https://zuul.opendev.org/t/openstack/build/8aae27a57fdb4e33b7c065d19a8abfea : SUCCESS in 4m 23s\n- build-openstack-sphinx-docs https://zuul.opendev.org/t/openstack/build/982a1fbdcb17406cb2eaeffb0039d73a : SUCCESS in 5m 08s\n- kolla-build-ubuntu-source https://zuul.opendev.org/t/openstack/build/83fe765a79144459885f23ac1ea99a2f : SUCCESS in 55m 20s","accounts_in_message":[],"_revision_number":2},{"id":"572e844a187b1e400147f83c83be07c94977f92e","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-09-03 09:25:32.000000000","message":"Change has been successfully merged by Zuul","accounts_in_message":[],"_revision_number":2}],"current_revision_number":2,"current_revision":"69e46f78aff21ecad5f7793fb9aa79b56da87747","revisions":{"ad76782797be61e737b71618ef9aa6171851a17a":{"kind":"REWORK","_number":1,"created":"2019-09-02 09:42:34.000000000","uploader":{"_account_id":30687,"name":"Eddie Yen","email":"missile0407@gmail.com","username":"eddieak"},"ref":"refs/changes/17/679617/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/kolla","ref":"refs/changes/17/679617/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/kolla refs/changes/17/679617/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/kolla refs/changes/17/679617/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/kolla refs/changes/17/679617/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/kolla refs/changes/17/679617/1"}}},"commit":{"parents":[{"commit":"634ea03038d95af71964dd3cebb5cde7c0d27d9c","subject":"Unbreak CI: fluentd","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/634ea03038d95af71964dd3cebb5cde7c0d27d9c"}]}],"author":{"name":"Eddie Yen","email":"missile0407@gmail.com","date":"2019-07-15 13:20:39.000000000","tz":0},"committer":{"name":"Eddie Yen","email":"missile0407@gmail.com","date":"2019-09-02 09:42:34.000000000","tz":0},"subject":"Delete default gnocchi \u0026 aodh wsgi conf in Ubuntu binary","message":"Delete default gnocchi \u0026 aodh wsgi conf in Ubuntu binary\n\nUbuntu distro package will create default wsgi conf in apache\nduring gnocchi-api and aodh-api installation. We need to delete\nthem since we only use the conf that pushed by kolla-ansible.\n\nChange-Id: Ib31e733eba12deebdf469c306dfba4191a1b89f5\nRelated-Bug: 1832306\nRelated-Bug: 1836520\n(cherry picked from commit 5eb780d0b0d12186c57b828320c3421b98b97840)\n(cherry picked from commit cc6694dbc8907bad5085f9129fccd5b682e72b66)\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/ad76782797be61e737b71618ef9aa6171851a17a"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/ad76782797be61e737b71618ef9aa6171851a17a"}]},"branch":"refs/heads/stable/rocky"},"69e46f78aff21ecad5f7793fb9aa79b56da87747":{"kind":"REWORK","_number":2,"created":"2019-09-02 09:44:31.000000000","uploader":{"_account_id":30687,"name":"Eddie Yen","email":"missile0407@gmail.com","username":"eddieak"},"ref":"refs/changes/17/679617/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/kolla","ref":"refs/changes/17/679617/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/kolla refs/changes/17/679617/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/kolla refs/changes/17/679617/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/kolla refs/changes/17/679617/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/kolla refs/changes/17/679617/2"}}},"commit":{"parents":[{"commit":"634ea03038d95af71964dd3cebb5cde7c0d27d9c","subject":"Unbreak CI: fluentd","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/634ea03038d95af71964dd3cebb5cde7c0d27d9c"}]}],"author":{"name":"Eddie Yen","email":"missile0407@gmail.com","date":"2019-07-15 13:20:39.000000000","tz":0},"committer":{"name":"Eddie Yen","email":"missile0407@gmail.com","date":"2019-09-02 09:44:18.000000000","tz":0},"subject":"Delete default gnocchi \u0026 aodh wsgi conf in Ubuntu binary","message":"Delete default gnocchi \u0026 aodh wsgi conf in Ubuntu binary\n\nUbuntu distro package will create default wsgi conf in apache\nduring gnocchi-api and aodh-api installation. We need to delete\nthem since we only use the conf that pushed by kolla-ansible.\n\nChange-Id: Ib31e733eba12deebdf469c306dfba4191a1b89f5\nRelated-Bug: 1832306\nRelated-Bug: 1836520\n(cherry picked from commit 5eb780d0b0d12186c57b828320c3421b98b97840)\n(cherry picked from commit cc6694dbc8907bad5085f9129fccd5b682e72b66)\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/69e46f78aff21ecad5f7793fb9aa79b56da87747"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/69e46f78aff21ecad5f7793fb9aa79b56da87747"}]},"branch":"refs/heads/stable/rocky"}},"requirements":[],"submit_records":[],"submit_requirements":[]}
