)]}'
{"id":"openstack%2Fkolla~680123","triplet_id":"openstack%2Fkolla~stable%2Frocky~I53ad946ead4f0b485d8c7e8a91ac2ef5d7735c5a","project":"openstack/kolla","branch":"stable/rocky","topic":"bz:1747445","hashtags":[],"change_id":"I53ad946ead4f0b485d8c7e8a91ac2ef5d7735c5a","subject":"base: update to dumb-init 1.2.2","status":"ABANDONED","created":"2019-09-04 18:35:23.000000000","updated":"2019-09-04 20:50:33.000000000","total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"meta_rev_id":"f8e37853ea92b8ba53699bb56d61daf39e12f0ca","_number":680123,"virtual_id_number":680123,"owner":{"_account_id":7353,"name":"Kevin Carter","email":"kevin@cloudnull.com","username":"cloudnull"},"actions":{},"labels":{"Verified":{"disliked":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"_account_id":24072,"name":"Marcin Juszkiewicz","email":"mjuszkiewicz@redhat.com","username":"hrw"},{"value":-1,"date":"2019-09-04 20:35:43.000000000","permitted_voting_range":{"min":-2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","value":-1,"default_value":0,"optional":true},"Code-Review":{"all":[{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":24072,"name":"Marcin Juszkiewicz","email":"mjuszkiewicz@redhat.com","username":"hrw"},{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"all":[{"_account_id":24072,"name":"Marcin Juszkiewicz","email":"mjuszkiewicz@redhat.com","username":"hrw"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Backport-Candidate":{"all":[{"_account_id":24072,"name":"Marcin Juszkiewicz","email":"mjuszkiewicz@redhat.com","username":"hrw"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Do Not Backport"," 0":"Backport Review Needed","+1":"Should Backport"},"description":"","default_value":0,"optional":true},"Review-Priority":{"all":[{"date":"2019-09-04 18:35:23.000000000","_account_id":24072,"name":"Marcin Juszkiewicz","email":"mjuszkiewicz@redhat.com","username":"hrw"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Branch Freeze"," 0":"No Priority","+1":"Important Change","+2":"Gate Blocker Fix / Urgent Change"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":24072,"name":"Marcin Juszkiewicz","email":"mjuszkiewicz@redhat.com","username":"hrw"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2019-09-04 18:35:23.000000000","updated_by":{"_account_id":24072,"name":"Marcin Juszkiewicz","email":"mjuszkiewicz@redhat.com","username":"hrw"},"reviewer":{"_account_id":24072,"name":"Marcin Juszkiewicz","email":"mjuszkiewicz@redhat.com","username":"hrw"},"state":"REVIEWER"},{"updated":"2019-09-04 20:35:43.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"}],"messages":[{"id":"5563b49f5e6d991c8a86d7c9a3b7e2ee08fe7d0e","author":{"_account_id":7353,"name":"Kevin Carter","email":"kevin@cloudnull.com","username":"cloudnull"},"date":"2019-09-04 18:35:23.000000000","message":"Patch Set 1: Cherry Picked from branch master.","accounts_in_message":[],"_revision_number":1},{"id":"2bb95ec7a940f309498f1bd496f2e7ba12e42dfd","author":{"_account_id":7353,"name":"Kevin Carter","email":"kevin@cloudnull.com","username":"cloudnull"},"date":"2019-09-04 18:36:17.000000000","message":"Topic set to bz:1747442","accounts_in_message":[],"_revision_number":1},{"id":"5f0cc548026891d5eb85502b3712d21a6246b1cd","author":{"_account_id":7353,"name":"Kevin Carter","email":"kevin@cloudnull.com","username":"cloudnull"},"date":"2019-09-04 18:36:56.000000000","message":"Topic changed from bz:1747442 to bz:1747445","accounts_in_message":[],"_revision_number":1},{"id":"3522da690141d67c3616469a489f95c8e3c9e49d","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-09-04 20:35:43.000000000","message":"Patch Set 1: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttp://docs.openstack.org/infra/manual/developers.html#automated-testing\n\n\n- kolla-build-centos-binary https://zuul.opendev.org/t/openstack/build/de4968237f7240e3b16a1b66ae6a4583 : SUCCESS in 40m 26s (non-voting)\n- kolla-build-centos-source https://zuul.opendev.org/t/openstack/build/9f62f684a8bd49239040faed3f7d6852 : SUCCESS in 45m 51s\n- kolla-ansible-centos-source https://zuul.opendev.org/t/openstack/build/2c2add201cca43fc9d7968ce098e4678 : FAILURE in 40m 02s\n- kolla-ansible-centos-binary https://zuul.opendev.org/t/openstack/build/31f55c0a8c0540a590d818174ceca9b4 : FAILURE in 33m 56s (non-voting)\n- kolla-build-oraclelinux-binary https://zuul.opendev.org/t/openstack/build/f13b3f7ba23b4050bcdbfd428b9896ec : SUCCESS in 41m 50s (non-voting)\n- kolla-build-oraclelinux-source https://zuul.opendev.org/t/openstack/build/0ee58fdbc3e44b9fbbcaa63fe1611b0b : SUCCESS in 49m 42s\n- kolla-ansible-oraclelinux-source https://zuul.opendev.org/t/openstack/build/24b28d5b54b84cad9207c9732d63801a : FAILURE in 31m 29s (non-voting)\n- kolla-ansible-oraclelinux-binary https://zuul.opendev.org/t/openstack/build/4ad2d9e675fa41c39e4b6e7daa465601 : FAILURE in 33m 06s (non-voting)\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/8fd0079af68c4af1834c9f7ab9d92f82 : SUCCESS in 5m 24s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/41190ffd86234077b7638bddfe6e5e3c : SUCCESS in 4m 16s\n- openstack-tox-py27 https://zuul.opendev.org/t/openstack/build/1c83b8cc9ff049918b3588281d19db9b : SUCCESS in 4m 45s\n- openstack-tox-py35 https://zuul.opendev.org/t/openstack/build/0d1bef6c0b784855b0bc6a45b33c5f41 : SUCCESS in 4m 36s\n- build-openstack-sphinx-docs https://zuul.opendev.org/t/openstack/build/9fba6463235e48e5bfe9e1cd9220df97 : SUCCESS in 5m 49s\n- kolla-build-ubuntu-binary https://zuul.opendev.org/t/openstack/build/b682fb7fc1e747a199e098617d56292a : SUCCESS in 1h 42m 19s (non-voting)\n- kolla-build-ubuntu-source https://zuul.opendev.org/t/openstack/build/10f7cc7676a84911999e4382a2ba89e0 : SUCCESS in 1h 03m 46s\n- kolla-ansible-ubuntu-source https://zuul.opendev.org/t/openstack/build/9a83c5afc97a4442ae3dab74d7b43d83 : FAILURE in 43m 36s\n- kolla-ansible-ubuntu-binary https://zuul.opendev.org/t/openstack/build/025231e37e724285bc4aa51d77f37b9b : FAILURE in 1h 52m 50s (non-voting)","accounts_in_message":[],"_revision_number":1},{"id":"f8e37853ea92b8ba53699bb56d61daf39e12f0ca","author":{"_account_id":7353,"name":"Kevin Carter","email":"kevin@cloudnull.com","username":"cloudnull"},"date":"2019-09-04 20:50:33.000000000","message":"Abandoned","accounts_in_message":[],"_revision_number":1}],"current_revision_number":1,"current_revision":"919f16c9339e8c7af7a5e943a4689e1b10791fd2","revisions":{"919f16c9339e8c7af7a5e943a4689e1b10791fd2":{"kind":"REWORK","_number":1,"created":"2019-09-04 18:35:23.000000000","uploader":{"_account_id":7353,"name":"Kevin Carter","email":"kevin@cloudnull.com","username":"cloudnull"},"ref":"refs/changes/23/680123/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/kolla","ref":"refs/changes/23/680123/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/kolla refs/changes/23/680123/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/kolla refs/changes/23/680123/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/kolla refs/changes/23/680123/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/kolla refs/changes/23/680123/1"}}},"commit":{"parents":[{"commit":"4e63ecc9abec139f429c003ff8f0c72c10399caf","subject":"Merge \"Bump service versions for Rocky\" into stable/rocky","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/4e63ecc9abec139f429c003ff8f0c72c10399caf"}]}],"author":{"name":"Marcin Juszkiewicz","email":"marcin.juszkiewicz@linaro.org","date":"2019-02-25 09:20:29.000000000","tz":60},"committer":{"name":"Kevin Carter (cloudnull)","email":"kevin@cloudnull.com","date":"2019-09-04 18:35:23.000000000","tz":0},"subject":"base: update to dumb-init 1.2.2","message":"base: update to dumb-init 1.2.2\n\nThis version allows us to use Debian binaries for non-x86 architectures\ninstead of building \u0027dumb-init\u0027.\n\n\u0027base_compiler_packages\u0027 are installed as they will be needed anyway to\nbuild other Python packages (no binary wheels for !x86). We also need\n\u0027binutils\u0027 to be installed for \u0027ar\u0027 command.\n\nChange-Id: I53ad946ead4f0b485d8c7e8a91ac2ef5d7735c5a\n(cherry picked from commit f8fd56056114af7e2c8b8185cd258c6fd3993c8c)\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/919f16c9339e8c7af7a5e943a4689e1b10791fd2"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/919f16c9339e8c7af7a5e943a4689e1b10791fd2"}]},"branch":"refs/heads/stable/rocky"}},"requirements":[],"submit_records":[],"submit_requirements":[]}
