)]}'
{"id":"openstack%2Fkolla~680129","triplet_id":"openstack%2Fkolla~stable%2Fstein~I220d8cfc81c25471e635b2f5790759f21e788e64","project":"openstack/kolla","branch":"stable/stein","topic":"bug/1680770-stein","hashtags":[],"change_id":"I220d8cfc81c25471e635b2f5790759f21e788e64","subject":"Move shebang customization into jinja2 block","status":"ABANDONED","created":"2019-09-04 18:58:18.000000000","updated":"2019-09-04 20:49:20.000000000","total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"meta_rev_id":"d2c85e31346dff15bd0a583c40b4c19c9d86b25c","_number":680129,"virtual_id_number":680129,"owner":{"_account_id":7353,"name":"Kevin Carter","email":"kevin@cloudnull.com","username":"cloudnull"},"actions":{},"labels":{"Verified":{"recommended":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":1,"date":"2019-09-04 20:42:43.000000000","permitted_voting_range":{"min":-2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":13039,"name":"Martin André","email":"m.andre@redhat.com","username":"mandre"}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","value":1,"default_value":0,"optional":true},"Code-Review":{"all":[{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":13039,"name":"Martin André","email":"m.andre@redhat.com","username":"mandre"}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"all":[{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":13039,"name":"Martin André","email":"m.andre@redhat.com","username":"mandre"}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Backport-Candidate":{"all":[{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":13039,"name":"Martin André","email":"m.andre@redhat.com","username":"mandre"}],"values":{"-1":"Do Not Backport"," 0":"Backport Review Needed","+1":"Should Backport"},"description":"","default_value":0,"optional":true},"Review-Priority":{"all":[{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":13039,"name":"Martin André","email":"m.andre@redhat.com","username":"mandre"}],"values":{"-1":"Branch Freeze"," 0":"No Priority","+1":"Important Change","+2":"Gate Blocker Fix / Urgent Change"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":13039,"name":"Martin André","email":"m.andre@redhat.com","username":"mandre"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2019-09-04 18:58:18.000000000","updated_by":{"_account_id":13039,"name":"Martin André","email":"m.andre@redhat.com","username":"mandre"},"reviewer":{"_account_id":13039,"name":"Martin André","email":"m.andre@redhat.com","username":"mandre"},"state":"REVIEWER"},{"updated":"2019-09-04 20:42:43.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"}],"messages":[{"id":"06ea1e917be47c1d730377d72da037111d4027c4","author":{"_account_id":7353,"name":"Kevin Carter","email":"kevin@cloudnull.com","username":"cloudnull"},"date":"2019-09-04 18:58:18.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"cb0a18f42a463b89a0b6a21beb3bd9fdb459ab1b","author":{"_account_id":7353,"name":"Kevin Carter","email":"kevin@cloudnull.com","username":"cloudnull"},"date":"2019-09-04 18:58:59.000000000","message":"Patch Set 2: Published edit on patch set 1.","accounts_in_message":[],"_revision_number":2},{"id":"7e433b4019f484acb26b8287d07180cfd5e47dbe","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-09-04 20:42:43.000000000","message":"Patch Set 2: Verified+1\n\nBuild succeeded (check pipeline).\n\n- kolla-build-centos-binary https://zuul.opendev.org/t/openstack/build/edb6fd6a9e9a4521b18eff1eeb36dd52 : SUCCESS in 40m 18s (non-voting)\n- kolla-build-centos-source https://zuul.opendev.org/t/openstack/build/bb3f30459807411fb569292d9e9fc28b : SUCCESS in 1h 12m 05s\n- kolla-ansible-centos-source https://zuul.opendev.org/t/openstack/build/a68a036295b7443db21978763a8b3a24 : SUCCESS in 1h 15m 37s\n- kolla-ansible-centos-binary https://zuul.opendev.org/t/openstack/build/54f85d4637a842f5bd7081c40bda95a2 : SUCCESS in 1h 35m 07s (non-voting)\n- kolla-ansible-centos-source-upgrade https://zuul.opendev.org/t/openstack/build/0ad5b758dabf491a92747092b1ae4ad1 : SUCCESS in 1h 31m 57s\n- tripleo-build-containers-centos-7 https://zuul.opendev.org/t/openstack/build/cdf658d8ad114049b789d100c536021a : SUCCESS in 1h 13m 08s (non-voting)\n- kolla-build-oraclelinux-binary https://zuul.opendev.org/t/openstack/build/51d07aa85a214fde955f9e32004b99af : SUCCESS in 1h 12m 58s (non-voting)\n- kolla-build-oraclelinux-source https://zuul.opendev.org/t/openstack/build/a4f4707de0bd4bbf908df521be5ced2b : SUCCESS in 55m 04s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/e33d8beb9415476fa0a86e5ed647e3d7 : SUCCESS in 3m 23s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/ca85b3984b064db287c51e01b2c68a0a : SUCCESS in 3m 49s\n- openstack-tox-py27 https://zuul.opendev.org/t/openstack/build/3be171c5b13048c688dcdfe976aee25c : SUCCESS in 5m 43s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/9bbee9a9095b45388cfb8477bb7747d2 : SUCCESS in 4m 00s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/cd253ffd6d864e0f8521d5415bc47697 : SUCCESS in 4m 40s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/b3c83aa70ade465f8ff4e53955f9ac10 : SUCCESS in 5m 26s\n- kolla-build-ubuntu-binary https://zuul.opendev.org/t/openstack/build/8e60db97f7914276a2327d7ca33940a0 : SUCCESS in 28m 04s (non-voting)\n- kolla-build-ubuntu-source https://zuul.opendev.org/t/openstack/build/cfc3ff9f80c0412bb3e418bdeec9cc08 : SUCCESS in 34m 34s\n- kolla-ansible-ubuntu-source https://zuul.opendev.org/t/openstack/build/1ba92894570b407794ec0475a9a3ec6b : SUCCESS in 1h 03m 51s\n- kolla-ansible-ubuntu-binary https://zuul.opendev.org/t/openstack/build/88b8c63791fd49cfb7847b3fe7a186dc : SUCCESS in 51m 03s (non-voting)\n- kolla-ansible-ubuntu-source-upgrade https://zuul.opendev.org/t/openstack/build/5272d2e0125d4aea91a397b91ea71711 : SUCCESS in 1h 17m 40s","accounts_in_message":[],"_revision_number":2},{"id":"546e1112712c93e2a3b80f4915753697e103ff9c","author":{"_account_id":7353,"name":"Kevin Carter","email":"kevin@cloudnull.com","username":"cloudnull"},"date":"2019-09-04 20:49:20.000000000","message":"Abandoned","accounts_in_message":[],"_revision_number":2}],"current_revision_number":2,"current_revision":"283fffdfa563548d20fd78ab038a07b5d71528bf","revisions":{"e17caa0375e3552fb6e7882ffd5874dcf1c12878":{"kind":"REWORK","_number":1,"created":"2019-09-04 18:58:18.000000000","uploader":{"_account_id":7353,"name":"Kevin Carter","email":"kevin@cloudnull.com","username":"cloudnull"},"ref":"refs/changes/29/680129/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/kolla","ref":"refs/changes/29/680129/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/kolla refs/changes/29/680129/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/kolla refs/changes/29/680129/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/kolla refs/changes/29/680129/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/kolla refs/changes/29/680129/1"}}},"commit":{"parents":[{"commit":"a6329f721ee95a858b8552fc74d64de383a7b59d","subject":"Bump service versions for Stein","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/a6329f721ee95a858b8552fc74d64de383a7b59d"}]}],"author":{"name":"Martin André","email":"m.andre@redhat.com","date":"2017-04-07 09:35:16.000000000","tz":120},"committer":{"name":"Kevin Carter","email":"kecarter@redhat.com","date":"2019-09-04 18:58:01.000000000","tz":-300},"subject":"Move shebang customization into jinja2 block","message":"Move shebang customization into jinja2 block\n\nBy setting up the shebang inside a jinja2 block, this allows the\noperator to use a different init system of simply disable it. This is\nespecially useful with docker 1.13 that comes with its own stripped\ndown init system (a build of tini) that would cause the containers to\nrun double init systems if used with the \u0027--init\u0027 command line\nargument.\n\nChange-Id: I220d8cfc81c25471e635b2f5790759f21e788e64\nCloses-Bug: #1680770\n(cherry picked from commit f9e634ca6b5e6aa33369858e73433409fe10dd2f)\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/e17caa0375e3552fb6e7882ffd5874dcf1c12878"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/e17caa0375e3552fb6e7882ffd5874dcf1c12878"}]},"branch":"refs/heads/stable/stein"},"283fffdfa563548d20fd78ab038a07b5d71528bf":{"kind":"REWORK","_number":2,"created":"2019-09-04 18:58:59.000000000","uploader":{"_account_id":7353,"name":"Kevin Carter","email":"kevin@cloudnull.com","username":"cloudnull"},"ref":"refs/changes/29/680129/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/kolla","ref":"refs/changes/29/680129/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/kolla refs/changes/29/680129/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/kolla refs/changes/29/680129/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/kolla refs/changes/29/680129/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/kolla refs/changes/29/680129/2"}}},"commit":{"parents":[{"commit":"a6329f721ee95a858b8552fc74d64de383a7b59d","subject":"Bump service versions for Stein","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/a6329f721ee95a858b8552fc74d64de383a7b59d"}]}],"author":{"name":"Martin André","email":"m.andre@redhat.com","date":"2017-04-07 09:35:16.000000000","tz":120},"committer":{"name":"Kevin Carter (cloudnull)","email":"kevin@cloudnull.com","date":"2019-09-04 18:58:56.000000000","tz":0},"subject":"Move shebang customization into jinja2 block","message":"Move shebang customization into jinja2 block\n\nBy setting up the shebang inside a jinja2 block, this allows the\noperator to use a different init system of simply disable it. This is\nespecially useful with docker 1.13 that comes with its own stripped\ndown init system (a build of tini) that would cause the containers to\nrun double init systems if used with the \u0027--init\u0027 command line\nargument.\n\nChange-Id: I220d8cfc81c25471e635b2f5790759f21e788e64\nCloses-Bug: #1680770\n(cherry picked from commit f9e634ca6b5e6aa33369858e73433409fe10dd2f)\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/283fffdfa563548d20fd78ab038a07b5d71528bf"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/283fffdfa563548d20fd78ab038a07b5d71528bf"}]},"branch":"refs/heads/stable/stein"}},"requirements":[],"submit_records":[],"submit_requirements":[]}
