)]}'
{"id":"openstack%2Fkolla~681405","triplet_id":"openstack%2Fkolla~stable%2Fpike~I203f70176af7f2b53398ddeb5c21e6234a45e268","project":"openstack/kolla","branch":"stable/pike","hashtags":[],"change_id":"I203f70176af7f2b53398ddeb5c21e6234a45e268","subject":"Unbreak CI: fluentd","status":"MERGED","created":"2019-09-11 08:38:48.000000000","updated":"2019-09-11 11:30:57.000000000","submitted":"2019-09-11 11:30:57.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"681405-1568201457889-9806e538","meta_rev_id":"5002536002d23b2f1c0dd0d6952de46ce5e67b39","_number":681405,"virtual_id_number":681405,"owner":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},{"value":0,"_account_id":24072,"name":"Marcin Juszkiewicz","email":"mjuszkiewicz@redhat.com","username":"hrw"},{"value":0,"_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"},{"value":2,"date":"2019-09-11 11:30:57.000000000","post_submit":true,"permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"all":[{"value":2,"date":"2019-09-11 08:47:41.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},{"value":2,"date":"2019-09-11 08:49:33.000000000","_account_id":24072,"name":"Marcin Juszkiewicz","email":"mjuszkiewicz@redhat.com","username":"hrw"},{"value":0,"_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":24072,"name":"Marcin Juszkiewicz","email":"mjuszkiewicz@redhat.com","username":"hrw"},"all":[{"value":0,"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},{"value":1,"date":"2019-09-11 10:15:58.000000000","_account_id":24072,"name":"Marcin Juszkiewicz","email":"mjuszkiewicz@redhat.com","username":"hrw"},{"value":0,"_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Backport-Candidate":{"all":[{"value":0,"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},{"value":0,"_account_id":24072,"name":"Marcin Juszkiewicz","email":"mjuszkiewicz@redhat.com","username":"hrw"},{"value":0,"_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Do Not Backport"," 0":"Backport Review Needed","+1":"Should Backport"},"description":"","default_value":0,"optional":true},"Review-Priority":{"all":[{"value":0,"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},{"value":0,"_account_id":24072,"name":"Marcin Juszkiewicz","email":"mjuszkiewicz@redhat.com","username":"hrw"},{"value":0,"date":"2019-09-11 08:38:48.000000000","_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Branch Freeze"," 0":"No Priority","+1":"Important Change","+2":"Gate Blocker Fix / Urgent Change"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},{"_account_id":24072,"name":"Marcin Juszkiewicz","email":"mjuszkiewicz@redhat.com","username":"hrw"},{"_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2019-09-11 08:38:48.000000000","updated_by":{"_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"},"reviewer":{"_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"},"state":"REVIEWER"},{"updated":"2019-09-11 08:47:41.000000000","updated_by":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"reviewer":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"state":"REVIEWER"},{"updated":"2019-09-11 10:15:58.000000000","updated_by":{"_account_id":24072,"name":"Marcin Juszkiewicz","email":"mjuszkiewicz@redhat.com","username":"hrw"},"reviewer":{"_account_id":24072,"name":"Marcin Juszkiewicz","email":"mjuszkiewicz@redhat.com","username":"hrw"},"state":"REVIEWER"},{"updated":"2019-09-11 11:30:57.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"}],"messages":[{"id":"78cd4bd5961d3ea6ba80ca38f51426225faf1801","author":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},"date":"2019-09-11 08:38:48.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"23c1fc163f340ee115d80079c7980738d1dbb48e","author":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"date":"2019-09-11 08:47:41.000000000","message":"Patch Set 1: Code-Review+2","accounts_in_message":[],"_revision_number":1},{"id":"654e6b1a3663151c2f354828838be87fc64eda16","author":{"_account_id":24072,"name":"Marcin Juszkiewicz","email":"mjuszkiewicz@redhat.com","username":"hrw"},"date":"2019-09-11 08:49:33.000000000","message":"Patch Set 1: Code-Review+2","accounts_in_message":[],"_revision_number":1},{"id":"464081e08635cde77281eb882c905d3bee16e261","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-09-11 09:48:16.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/3cf458312f514db5808a9f041e93670a : SUCCESS in 3m 55s\n- openstack-tox-py27 https://zuul.opendev.org/t/openstack/build/827bf2cc22e44baaafb1a4832430b8f0 : SUCCESS in 3m 33s\n- openstack-tox-py35 https://zuul.opendev.org/t/openstack/build/72150669534f48f48748bac7519191d5 : SUCCESS in 5m 09s\n- build-openstack-sphinx-docs https://zuul.opendev.org/t/openstack/build/a204faa8371d451f9d2fb957e1aa81c4 : SUCCESS in 4m 13s\n- kolla-build-centos-binary https://zuul.opendev.org/t/openstack/build/2bc0e1513c074bb6ba8c74bce2e933af : SUCCESS in 33m 07s (non-voting)\n- kolla-build-centos-source https://zuul.opendev.org/t/openstack/build/22bcc5ab218f4375b8b161871ded77bb : SUCCESS in 50m 14s\n- kolla-build-ubuntu-binary https://zuul.opendev.org/t/openstack/build/341e384402ec4b2ead9f0cc774183c70 : SUCCESS in 29m 28s (non-voting)\n- kolla-build-ubuntu-source https://zuul.opendev.org/t/openstack/build/a7f95b6b112a4ae6977769791cd3e216 : SUCCESS in 27m 35s\n- kolla-build-oraclelinux-binary https://zuul.opendev.org/t/openstack/build/160a64fad5db49a9b4f346ace3680ccf : SUCCESS in 45m 52s (non-voting)\n- kolla-build-oraclelinux-source https://zuul.opendev.org/t/openstack/build/d0f80d5dde314fe5b6b0ccee98b68993 : SUCCESS in 52m 56s\n- kolla-ansible-centos-source https://zuul.opendev.org/t/openstack/build/1a68687443c04fafa59df927a79af6b3 : SUCCESS in 57m 33s (non-voting)\n- kolla-ansible-centos-binary https://zuul.opendev.org/t/openstack/build/439da4d190f448cea4ef8711db6b81c0 : SUCCESS in 54m 32s (non-voting)\n- kolla-ansible-ubuntu-source https://zuul.opendev.org/t/openstack/build/0f5ea8386d7a40dd967b6411b3d39815 : SUCCESS in 49m 05s (non-voting)\n- kolla-ansible-ubuntu-binary https://zuul.opendev.org/t/openstack/build/0bb4026190974a92a91d67df262ff037 : SUCCESS in 51m 26s (non-voting)\n- kolla-ansible-oraclelinux-source https://zuul.opendev.org/t/openstack/build/af064834f12244e2b1f9c45c7afbe84c : SUCCESS in 1h 01m 25s (non-voting)\n- kolla-ansible-oraclelinux-binary https://zuul.opendev.org/t/openstack/build/e41bcc901470475abd6d3106c6a43647 : FAILURE in 18m 36s (non-voting)","accounts_in_message":[],"_revision_number":1},{"id":"167cb9f4a2bc07d8f727bc3e23e8f8589c05e487","author":{"_account_id":24072,"name":"Marcin Juszkiewicz","email":"mjuszkiewicz@redhat.com","username":"hrw"},"date":"2019-09-11 10:15:58.000000000","message":"Patch Set 1: Workflow+1","accounts_in_message":[],"_revision_number":1},{"id":"54b7924487c1ad1d4a8e0b3929f674e8e1f31eaa","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-09-11 10:16:12.000000000","message":"Patch Set 1: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":1},{"id":"d8667bb71d94cec70e441c62495071c1d287db6b","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-09-11 11:30:57.000000000","message":"Change has been successfully merged by Zuul","accounts_in_message":[],"_revision_number":1},{"id":"5002536002d23b2f1c0dd0d6952de46ce5e67b39","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-09-11 11:30:57.000000000","message":"Patch Set 1: Verified+2\n\nBuild succeeded (gate pipeline).\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/136e1366280a4804897a768164028815 : SUCCESS in 3m 21s\n- openstack-tox-py27 https://zuul.opendev.org/t/openstack/build/f75dc3dc7bab406889fac4ef6c92b2f2 : SUCCESS in 3m 44s\n- openstack-tox-py35 https://zuul.opendev.org/t/openstack/build/1d83d962bcc342858dd675e00f443408 : SUCCESS in 4m 30s\n- build-openstack-sphinx-docs https://zuul.opendev.org/t/openstack/build/e284ed7a60134c6e83c7e517706e525e : SUCCESS in 5m 10s\n- kolla-build-centos-source https://zuul.opendev.org/t/openstack/build/4681d58e3f0b48b8ae276ed78aa4dd5d : SUCCESS in 48m 11s\n- kolla-build-ubuntu-source https://zuul.opendev.org/t/openstack/build/a130e41df9bb4393b18c6a52901153bd : SUCCESS in 26m 05s\n- kolla-build-oraclelinux-source https://zuul.opendev.org/t/openstack/build/14c23a3c434e4fc1bb30d5dbfe46c8e0 : SUCCESS in 1h 07m 30s","accounts_in_message":[],"_revision_number":1}],"current_revision_number":1,"current_revision":"f634eee6ef1fa8275dc3277ece6e13baca109019","revisions":{"f634eee6ef1fa8275dc3277ece6e13baca109019":{"kind":"REWORK","_number":1,"created":"2019-09-11 08:38:48.000000000","uploader":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},"ref":"refs/changes/05/681405/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/kolla","ref":"refs/changes/05/681405/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/kolla refs/changes/05/681405/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/kolla refs/changes/05/681405/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/kolla refs/changes/05/681405/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/kolla refs/changes/05/681405/1"}}},"commit":{"parents":[{"commit":"430f2587125c1d5f305f05653ee3ea06d064f1ff","subject":"Move pike jobs to use validated RDO Trunk","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/430f2587125c1d5f305f05653ee3ea06d064f1ff"}]}],"author":{"name":"Radosław Piliszek","email":"radoslaw.piliszek@gmail.com","date":"2019-08-19 08:30:58.000000000","tz":120},"committer":{"name":"Mark Goddard","email":"mark@stackhpc.com","date":"2019-09-11 08:38:37.000000000","tz":60},"subject":"Unbreak CI: fluentd","message":"Unbreak CI: fluentd\n\nfluentd - we are not deploying kubernetes, no need for plugins\n\nIn the original commit we also dropped Tripleo from the gate, but in\nRocky the Tripleo job did not exist.\n\nChange-Id: I203f70176af7f2b53398ddeb5c21e6234a45e268\nSigned-off-by: Radosław Piliszek \u003cradoslaw.piliszek@gmail.com\u003e\n(cherry picked from commit 07b55048e63f27c1045735624c845d1af95a11fd)\n(cherry picked from commit b5a6da0dd7735e272ff2e2d1b6d931036601f308)\n(cherry picked from commit 634ea03038d95af71964dd3cebb5cde7c0d27d9c)\n(cherry picked from commit 1e2ff199d4d639f45e7c476bb555d12316a6bfe3)\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/f634eee6ef1fa8275dc3277ece6e13baca109019"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/f634eee6ef1fa8275dc3277ece6e13baca109019"}]},"branch":"refs/heads/stable/pike"}},"requirements":[],"submit_records":[],"submit_requirements":[]}
