)]}'
{"id":"openstack%2Fkolla~681627","triplet_id":"openstack%2Fkolla~master~Ic76ddeec52dc7ca4264457ad2c85ceeafeb450ae","project":"openstack/kolla","branch":"master","topic":"bug/1843356","hashtags":[],"change_id":"Ic76ddeec52dc7ca4264457ad2c85ceeafeb450ae","subject":"Shift kolla_keystone classes from shade to openstack sdk","status":"ABANDONED","created":"2019-09-11 23:19:28.000000000","updated":"2019-09-18 16:22:31.000000000","total_comment_count":3,"unresolved_comment_count":0,"has_review_started":true,"meta_rev_id":"7b61ce23f3116221e879466676d1b0508ed77789","_number":681627,"virtual_id_number":681627,"owner":{"_account_id":30810,"name":"James Kirsch","email":"generalfuzz@gmail.com","username":"generalfuzz"},"actions":{},"labels":{"Verified":{"disliked":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"date":"2019-09-12 09:28:30.000000000","_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},{"date":"2019-09-12 03:07:47.000000000","_account_id":30810,"name":"James Kirsch","email":"generalfuzz@gmail.com","username":"generalfuzz"},{"value":-1,"date":"2019-09-12 00:39:29.000000000","permitted_voting_range":{"min":-2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"date":"2019-09-12 00:01:22.000000000","_account_id":23181,"name":"RDO Third Party CI","email":"dmsimard+rdothirdparty@redhat.com","username":"rdothirdparty","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","value":-1,"default_value":0,"optional":true},"Code-Review":{"all":[{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":30810,"name":"James Kirsch","email":"generalfuzz@gmail.com","username":"generalfuzz"},{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":23181,"name":"RDO Third Party CI","email":"dmsimard+rdothirdparty@redhat.com","username":"rdothirdparty","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"all":[{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},{"value":0,"permitted_voting_range":{"min":-1,"max":0},"_account_id":30810,"name":"James Kirsch","email":"generalfuzz@gmail.com","username":"generalfuzz"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":23181,"name":"RDO Third Party CI","email":"dmsimard+rdothirdparty@redhat.com","username":"rdothirdparty","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Backport-Candidate":{"all":[{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},{"_account_id":30810,"name":"James Kirsch","email":"generalfuzz@gmail.com","username":"generalfuzz"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":23181,"name":"RDO Third Party CI","email":"dmsimard+rdothirdparty@redhat.com","username":"rdothirdparty","tags":["SERVICE_USER"]}],"values":{"-1":"Do Not Backport"," 0":"Backport Review Needed","+1":"Should Backport"},"description":"","default_value":0,"optional":true},"Review-Priority":{"all":[{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},{"_account_id":30810,"name":"James Kirsch","email":"generalfuzz@gmail.com","username":"generalfuzz"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":23181,"name":"RDO Third Party CI","email":"dmsimard+rdothirdparty@redhat.com","username":"rdothirdparty","tags":["SERVICE_USER"]}],"values":{"-1":"Branch Freeze"," 0":"No Priority","+1":"Important Change","+2":"Gate Blocker Fix / Urgent Change"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":23181,"name":"RDO Third Party CI","email":"dmsimard+rdothirdparty@redhat.com","username":"rdothirdparty","tags":["SERVICE_USER"]},{"_account_id":30810,"name":"James Kirsch","email":"generalfuzz@gmail.com","username":"generalfuzz"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2019-09-12 00:01:22.000000000","updated_by":{"_account_id":23181,"name":"RDO Third Party CI","email":"dmsimard+rdothirdparty@redhat.com","username":"rdothirdparty","tags":["SERVICE_USER"]},"reviewer":{"_account_id":23181,"name":"RDO Third Party CI","email":"dmsimard+rdothirdparty@redhat.com","username":"rdothirdparty","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2019-09-12 00:39:29.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2019-09-12 09:28:30.000000000","updated_by":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},"reviewer":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},"state":"REVIEWER"}],"messages":[{"id":"6176247df6b23c188a3512d501bcb5c4fda1cd67","author":{"_account_id":30810,"name":"James Kirsch","email":"generalfuzz@gmail.com","username":"generalfuzz"},"date":"2019-09-11 23:19:28.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"317ff35bc55868bdccf0159bbbabb530ff092e2a","author":{"_account_id":23181,"name":"RDO Third Party CI","email":"dmsimard+rdothirdparty@redhat.com","username":"rdothirdparty","tags":["SERVICE_USER"]},"date":"2019-09-12 00:01:22.000000000","message":"Patch Set 1:\n\nBuild succeeded (check pipeline).\n\n- tripleo-build-containers-rhel-8-master http://logs.rdoproject.org/27/681627/1/openstack-check/tripleo-build-containers-rhel-8-master/1cf6743/ : SUCCESS in 40m 56s","accounts_in_message":[],"_revision_number":1},{"id":"1273fe63658d2e3b33ecdd00fb3ad47db8ebe320","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-09-12 00:39:29.000000000","message":"Patch Set 1: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttp://docs.openstack.org/infra/manual/developers.html#automated-testing\n\n\n- kolla-build-centos-binary https://zuul.opendev.org/t/openstack/build/d2f1b0b3d7554f498fbfbeb2c01e8453 : SUCCESS in 44m 32s (non-voting)\n- kolla-build-centos-source https://zuul.opendev.org/t/openstack/build/5f6b87e84d3b4b96acac1b06113d8b76 : FAILURE in 25m 12s\n- kolla-ansible-centos-source https://zuul.opendev.org/t/openstack/build/4a4d35ed130d45398fd8c47514ec13d4 : FAILURE in 20m 12s\n- kolla-ansible-centos-binary https://zuul.opendev.org/t/openstack/build/84d63f1199f4408fb80eea052be906a3 : SUCCESS in 1h 13m 44s (non-voting)\n- kolla-ansible-centos-source-upgrade https://zuul.opendev.org/t/openstack/build/89d8407e606f449cadd8672a1a1a2b5c : FAILURE in 20m 59s\n- tripleo-build-containers-centos-7 https://zuul.opendev.org/t/openstack/build/883b99b592e84c1b917da5df3cffcbed : SUCCESS in 53m 41s\n- kolla-build-debian-source https://zuul.opendev.org/t/openstack/build/6665e53cf0a04ea5b8a2405f5e4ea278 : SUCCESS in 30m 00s (non-voting)\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/be7f37bbcbd74e5d9570fd545354a099 : SUCCESS in 3m 16s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/f128ff91590f410e9ea0aed3afcfe183 : FAILURE in 3m 28s\n- openstack-tox-py27 https://zuul.opendev.org/t/openstack/build/fdd037ae59dc4bb599512526a3d9203d : SUCCESS in 3m 29s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/4037cee4eb5a4ade86890ac7156989c4 : SUCCESS in 3m 30s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/977a559cd4254e778c0ecb4880611c31 : SUCCESS in 3m 42s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/c2bc51c71253408ba67abae96cd0b3f3 : SUCCESS in 3m 42s\n- kolla-build-ubuntu-binary https://zuul.opendev.org/t/openstack/build/e713a0970b544dd38acba45d91b7cc77 : FAILURE in 35m 27s (non-voting)\n- kolla-build-ubuntu-source https://zuul.opendev.org/t/openstack/build/0c79350812b64bb3a797e0781970885c : FAILURE in 39m 54s\n- kolla-ansible-ubuntu-source https://zuul.opendev.org/t/openstack/build/730aa3e8589145d39a172be88adf0914 : SUCCESS in 54m 48s\n- kolla-ansible-ubuntu-binary https://zuul.opendev.org/t/openstack/build/845b58928e5e43509b186d8097a7f72f : SUCCESS in 53m 11s (non-voting)\n- kolla-ansible-ubuntu-source-upgrade https://zuul.opendev.org/t/openstack/build/3ba5d7d71be04a908ef73d8049e09e81 : SUCCESS in 1h 00m 23s","accounts_in_message":[],"_revision_number":1},{"id":"216510d78f447506b1f44e6ffc7293b781bff7d8","author":{"_account_id":30810,"name":"James Kirsch","email":"generalfuzz@gmail.com","username":"generalfuzz"},"date":"2019-09-12 03:07:47.000000000","message":"Patch Set 1:\n\nAny help understanding the zuul build failure would be appreciated.","accounts_in_message":[],"_revision_number":1},{"id":"ed2b7e46d824da0d3a4c12a8b172410127cfcddd","author":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},"date":"2019-09-12 09:28:30.000000000","message":"Patch Set 1:\n\n(3 comments)\n\nAs came up on IRC this morning, there are licensing issues if we copy Ansible module code directly, since it\u0027s GPL. :(","accounts_in_message":[],"_revision_number":1},{"id":"ab7433cc4e4d4823ebc6f3697844ca724a31ff02","author":{"_account_id":30810,"name":"James Kirsch","email":"generalfuzz@gmail.com","username":"generalfuzz"},"date":"2019-09-18 16:22:31.000000000","message":"Abandoned\n\nMoved away in KA from relying on Kolla classes for role creation","accounts_in_message":[],"_revision_number":1}],"current_revision_number":1,"current_revision":"bdca7ec7c7d7818ba9a76b4ec1d6ea50ef065044","revisions":{"bdca7ec7c7d7818ba9a76b4ec1d6ea50ef065044":{"kind":"REWORK","_number":1,"created":"2019-09-11 23:19:28.000000000","uploader":{"_account_id":30810,"name":"James Kirsch","email":"generalfuzz@gmail.com","username":"generalfuzz"},"ref":"refs/changes/27/681627/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/kolla","ref":"refs/changes/27/681627/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/kolla refs/changes/27/681627/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/kolla refs/changes/27/681627/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/kolla refs/changes/27/681627/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/kolla refs/changes/27/681627/1"}}},"commit":{"parents":[{"commit":"532af2efc8893cfd23f8d2beadd32a4e3463ae85","subject":"Do not fan out DB sync to all cells","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/532af2efc8893cfd23f8d2beadd32a4e3463ae85"}]}],"author":{"name":"James Kirsch","email":"generalfuzz@gmail.com","date":"2019-09-11 23:16:19.000000000","tz":-420},"committer":{"name":"James Kirsch","email":"generalfuzz@gmail.com","date":"2019-09-11 23:19:27.000000000","tz":-420},"subject":"Shift kolla_keystone classes from shade to openstack sdk","message":"Shift kolla_keystone classes from shade to openstack sdk\n\nCurrently kolla-toolbox/kolla_keystone_service.py and kolla-toolbox/kolla_keystone_user.py rely on shade to execute Keystone libraries instead of using open stack SDK. This limits our ability to leverage validate_certs parameter to allow self signed certificates when testing TLS enabled openstack deployment. This change removes shade from  kolla_keystone_service.py, and relies on the openstack sdk (like os_keystone_endpoint.py class in the ansible openstack library).\n\nChange-Id: Ic76ddeec52dc7ca4264457ad2c85ceeafeb450ae\nImplements: Shift to using openstack sdk in keystone modules\nCloses-Bug: #1843356\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/bdca7ec7c7d7818ba9a76b4ec1d6ea50ef065044"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/bdca7ec7c7d7818ba9a76b4ec1d6ea50ef065044"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[],"submit_requirements":[]}
