)]}'
{"id":"openstack%2Fkolla~681694","triplet_id":"openstack%2Fkolla~master~I9cf62c8e03a02574d85e5349df1d23a51d4c99be","project":"openstack/kolla","branch":"master","hashtags":[],"change_id":"I9cf62c8e03a02574d85e5349df1d23a51d4c99be","subject":"Drop Percona XtraBackup in favour of MariaDB Backup","status":"MERGED","created":"2019-09-12 09:49:04.000000000","updated":"2019-09-16 13:16:03.000000000","submitted":"2019-09-16 13:14:01.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":13,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"681694-1568639641857-4cbec2ef","meta_rev_id":"334c76e41e5a41609f363ccd5ef5dacc787d979b","_number":681694,"virtual_id_number":681694,"owner":{"_account_id":24072,"name":"Marcin Juszkiewicz","email":"mjuszkiewicz@redhat.com","username":"hrw"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":7488,"name":"Jeffrey Zhang","email":"zhang.lei.fly@gmail.com","username":"Jeffrey4l"},{"value":0,"_account_id":19316,"name":"Eduardo Gonzalez","email":"dabarren@gmail.com","username":"egonzalez90"},{"value":0,"_account_id":23717,"name":"Surya Prakash Singh","email":"singh.surya64mnnit@gmail.com","username":"SuryaPrakash"},{"value":0,"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},{"value":0,"_account_id":22165,"name":"caoyuan","email":"cao.yingjunz@gmail.com","username":"caoyuan"},{"value":0,"_account_id":167,"name":"Christian Berendt","email":"berendt@osism.tech","username":"berendt"},{"value":0,"date":"2019-09-16 08:33:34.000000000","_account_id":24072,"name":"Marcin Juszkiewicz","email":"mjuszkiewicz@redhat.com","username":"hrw"},{"value":0,"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},{"value":0,"_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"},{"value":2,"date":"2019-09-16 13:14:01.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"date":"2019-09-16 09:10:34.000000000","_account_id":23181,"name":"RDO Third Party CI","email":"dmsimard+rdothirdparty@redhat.com","username":"rdothirdparty","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},"all":[{"value":0,"_account_id":7488,"name":"Jeffrey Zhang","email":"zhang.lei.fly@gmail.com","username":"Jeffrey4l"},{"value":0,"_account_id":19316,"name":"Eduardo Gonzalez","email":"dabarren@gmail.com","username":"egonzalez90"},{"value":0,"_account_id":23717,"name":"Surya Prakash Singh","email":"singh.surya64mnnit@gmail.com","username":"SuryaPrakash"},{"value":0,"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},{"value":0,"_account_id":22165,"name":"caoyuan","email":"cao.yingjunz@gmail.com","username":"caoyuan"},{"value":0,"_account_id":167,"name":"Christian Berendt","email":"berendt@osism.tech","username":"berendt"},{"value":0,"_account_id":24072,"name":"Marcin Juszkiewicz","email":"mjuszkiewicz@redhat.com","username":"hrw"},{"value":2,"date":"2019-09-13 08:40:42.000000000","_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},{"value":0,"_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":23181,"name":"RDO Third Party CI","email":"dmsimard+rdothirdparty@redhat.com","username":"rdothirdparty","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},"all":[{"value":0,"_account_id":7488,"name":"Jeffrey Zhang","email":"zhang.lei.fly@gmail.com","username":"Jeffrey4l"},{"value":0,"_account_id":19316,"name":"Eduardo Gonzalez","email":"dabarren@gmail.com","username":"egonzalez90"},{"value":0,"_account_id":23717,"name":"Surya Prakash Singh","email":"singh.surya64mnnit@gmail.com","username":"SuryaPrakash"},{"value":0,"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},{"value":0,"_account_id":22165,"name":"caoyuan","email":"cao.yingjunz@gmail.com","username":"caoyuan"},{"value":0,"_account_id":167,"name":"Christian Berendt","email":"berendt@osism.tech","username":"berendt"},{"value":0,"_account_id":24072,"name":"Marcin Juszkiewicz","email":"mjuszkiewicz@redhat.com","username":"hrw"},{"value":1,"date":"2019-09-13 12:27:16.000000000","_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},{"value":0,"_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":23181,"name":"RDO Third Party CI","email":"dmsimard+rdothirdparty@redhat.com","username":"rdothirdparty","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Backport-Candidate":{"all":[{"value":0,"_account_id":7488,"name":"Jeffrey Zhang","email":"zhang.lei.fly@gmail.com","username":"Jeffrey4l"},{"value":0,"_account_id":19316,"name":"Eduardo Gonzalez","email":"dabarren@gmail.com","username":"egonzalez90"},{"value":0,"_account_id":23717,"name":"Surya Prakash Singh","email":"singh.surya64mnnit@gmail.com","username":"SuryaPrakash"},{"value":0,"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},{"value":0,"_account_id":22165,"name":"caoyuan","email":"cao.yingjunz@gmail.com","username":"caoyuan"},{"value":0,"_account_id":167,"name":"Christian Berendt","email":"berendt@osism.tech","username":"berendt"},{"value":0,"_account_id":24072,"name":"Marcin Juszkiewicz","email":"mjuszkiewicz@redhat.com","username":"hrw"},{"value":0,"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},{"value":0,"_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":23181,"name":"RDO Third Party CI","email":"dmsimard+rdothirdparty@redhat.com","username":"rdothirdparty","tags":["SERVICE_USER"]}],"values":{"-1":"Do Not Backport"," 0":"Backport Review Needed","+1":"Should Backport"},"description":"","default_value":0,"optional":true},"Review-Priority":{"approved":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},"all":[{"value":0,"date":"2019-09-13 10:20:01.000000000","_account_id":7488,"name":"Jeffrey Zhang","email":"zhang.lei.fly@gmail.com","username":"Jeffrey4l"},{"value":0,"date":"2019-09-13 10:20:01.000000000","_account_id":19316,"name":"Eduardo Gonzalez","email":"dabarren@gmail.com","username":"egonzalez90"},{"value":0,"date":"2019-09-13 10:20:01.000000000","_account_id":23717,"name":"Surya Prakash Singh","email":"singh.surya64mnnit@gmail.com","username":"SuryaPrakash"},{"value":0,"date":"2019-09-13 10:20:01.000000000","permitted_voting_range":{"min":0,"max":2},"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},{"value":0,"date":"2019-09-13 10:20:01.000000000","_account_id":22165,"name":"caoyuan","email":"cao.yingjunz@gmail.com","username":"caoyuan"},{"value":0,"date":"2019-09-13 10:20:01.000000000","_account_id":167,"name":"Christian Berendt","email":"berendt@osism.tech","username":"berendt"},{"value":0,"_account_id":24072,"name":"Marcin Juszkiewicz","email":"mjuszkiewicz@redhat.com","username":"hrw"},{"value":2,"date":"2019-09-13 10:19:50.000000000","_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},{"value":0,"date":"2019-09-16 08:31:56.000000000","_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":23181,"name":"RDO Third Party CI","email":"dmsimard+rdothirdparty@redhat.com","username":"rdothirdparty","tags":["SERVICE_USER"]}],"values":{"-1":"Branch Freeze"," 0":"No Priority","+1":"Important Change","+2":"Gate Blocker Fix / Urgent Change"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":167,"name":"Christian Berendt","email":"berendt@osism.tech","username":"berendt"},{"_account_id":7488,"name":"Jeffrey Zhang","email":"zhang.lei.fly@gmail.com","username":"Jeffrey4l"},{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},{"_account_id":19316,"name":"Eduardo Gonzalez","email":"dabarren@gmail.com","username":"egonzalez90"},{"_account_id":22165,"name":"caoyuan","email":"cao.yingjunz@gmail.com","username":"caoyuan"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},{"_account_id":23181,"name":"RDO Third Party CI","email":"dmsimard+rdothirdparty@redhat.com","username":"rdothirdparty","tags":["SERVICE_USER"]},{"_account_id":23717,"name":"Surya Prakash Singh","email":"singh.surya64mnnit@gmail.com","username":"SuryaPrakash"},{"_account_id":24072,"name":"Marcin Juszkiewicz","email":"mjuszkiewicz@redhat.com","username":"hrw"},{"_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2019-09-13 10:20:01.000000000","updated_by":{"_account_id":167,"name":"Christian Berendt","email":"berendt@osism.tech","username":"berendt"},"reviewer":{"_account_id":167,"name":"Christian Berendt","email":"berendt@osism.tech","username":"berendt"},"state":"REVIEWER"},{"updated":"2019-09-13 10:20:01.000000000","updated_by":{"_account_id":7488,"name":"Jeffrey Zhang","email":"zhang.lei.fly@gmail.com","username":"Jeffrey4l"},"reviewer":{"_account_id":7488,"name":"Jeffrey Zhang","email":"zhang.lei.fly@gmail.com","username":"Jeffrey4l"},"state":"REVIEWER"},{"updated":"2019-09-13 10:20:01.000000000","updated_by":{"_account_id":19316,"name":"Eduardo Gonzalez","email":"dabarren@gmail.com","username":"egonzalez90"},"reviewer":{"_account_id":19316,"name":"Eduardo Gonzalez","email":"dabarren@gmail.com","username":"egonzalez90"},"state":"REVIEWER"},{"updated":"2019-09-13 10:20:01.000000000","updated_by":{"_account_id":22165,"name":"caoyuan","email":"cao.yingjunz@gmail.com","username":"caoyuan"},"reviewer":{"_account_id":22165,"name":"caoyuan","email":"cao.yingjunz@gmail.com","username":"caoyuan"},"state":"REVIEWER"},{"updated":"2019-09-13 10:20:01.000000000","updated_by":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"reviewer":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"state":"REVIEWER"},{"updated":"2019-09-13 10:20:01.000000000","updated_by":{"_account_id":23717,"name":"Surya Prakash Singh","email":"singh.surya64mnnit@gmail.com","username":"SuryaPrakash"},"reviewer":{"_account_id":23717,"name":"Surya Prakash Singh","email":"singh.surya64mnnit@gmail.com","username":"SuryaPrakash"},"state":"REVIEWER"},{"updated":"2019-09-13 12:27:16.000000000","updated_by":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},"reviewer":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},"state":"REVIEWER"},{"updated":"2019-09-16 08:31:56.000000000","updated_by":{"_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"},"reviewer":{"_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"},"state":"REVIEWER"},{"updated":"2019-09-16 09:10:34.000000000","updated_by":{"_account_id":23181,"name":"RDO Third Party CI","email":"dmsimard+rdothirdparty@redhat.com","username":"rdothirdparty","tags":["SERVICE_USER"]},"reviewer":{"_account_id":23181,"name":"RDO Third Party CI","email":"dmsimard+rdothirdparty@redhat.com","username":"rdothirdparty","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2019-09-16 13:14:01.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"}],"messages":[{"id":"aa124dc61bd29bb7b664af10a8d81a6d0b27fa04","author":{"_account_id":24072,"name":"Marcin Juszkiewicz","email":"mjuszkiewicz@redhat.com","username":"hrw"},"date":"2019-09-12 09:49:04.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"1768fa9fd82ebd0c34d8292fe6f5c19d41ee4fa8","author":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},"date":"2019-09-12 10:25:23.000000000","message":"Patch Set 1:\n\n(4 comments)\n\nThanks for starting this.","accounts_in_message":[],"_revision_number":1},{"id":"7a7f83bdb33433760efa15d2861cc8d475122b3e","author":{"_account_id":23181,"name":"RDO Third Party CI","email":"dmsimard+rdothirdparty@redhat.com","username":"rdothirdparty","tags":["SERVICE_USER"]},"date":"2019-09-12 10:27:07.000000000","message":"Patch Set 1:\n\nBuild succeeded (check pipeline).\n\n- tripleo-build-containers-rhel-8-master http://logs.rdoproject.org/94/681694/1/openstack-check/tripleo-build-containers-rhel-8-master/01b0b84/ : SUCCESS in 36m 48s","accounts_in_message":[],"_revision_number":1},{"id":"62e2355427d420db1db5c33554046c7aed815039","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-09-12 11:52:44.000000000","message":"Patch Set 1: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttp://docs.openstack.org/infra/manual/developers.html#automated-testing\n\n\n- test-release-openstack https://zuul.opendev.org/t/openstack/build/4014d0e461a340698ca77b0e45387e7a : SUCCESS in 2m 09s\n- kolla-build-centos-binary https://zuul.opendev.org/t/openstack/build/8eaad202ebf447c7af40e6da216ace34 : SUCCESS in 40m 32s (non-voting)\n- kolla-build-centos-source https://zuul.opendev.org/t/openstack/build/fb68c137324c4db9a782bb756685c1a1 : SUCCESS in 49m 43s\n- kolla-ansible-centos-source https://zuul.opendev.org/t/openstack/build/d8d73cf8321544adb1984b5ffc424e9f : SUCCESS in 1h 13m 23s\n- kolla-ansible-centos-binary https://zuul.opendev.org/t/openstack/build/4b89fbdfd2c9437789e88e232c27f04e : SUCCESS in 1h 26m 33s (non-voting)\n- kolla-ansible-centos-source-upgrade https://zuul.opendev.org/t/openstack/build/483bf371532c4d54aa81e1e10d60aa52 : SUCCESS in 1h 59m 45s\n- tripleo-build-containers-centos-7 https://zuul.opendev.org/t/openstack/build/d0beccea9aa941f0bcba7e861a969a26 : SUCCESS in 50m 36s\n- kolla-ansible-bifrost-centos-source https://zuul.opendev.org/t/openstack/build/1affdba0acaf41b399e7e2aa0bb63f73 : SUCCESS in 48m 06s (non-voting)\n- kolla-ansible-centos-source-ceph https://zuul.opendev.org/t/openstack/build/1b00fa8fcece456bb74450b2d86cd9fb : SUCCESS in 1h 56m 56s (non-voting)\n- kolla-ansible-centos-source-upgrade-ceph https://zuul.opendev.org/t/openstack/build/ebae94730283463cbd72f2f35fdb04d1 : SUCCESS in 1h 49m 34s (non-voting)\n- kolla-ansible-centos-source-cinder-lvm https://zuul.opendev.org/t/openstack/build/b0e76f5143134b5d8f5042b35019a66f : SUCCESS in 1h 38m 51s (non-voting)\n- kolla-ansible-centos-source-scenario-nfv https://zuul.opendev.org/t/openstack/build/f815d2ef21c74c6a84cc47db4e1d30b3 : SUCCESS in 1h 39m 01s (non-voting)\n- kolla-ansible-centos-source-ironic https://zuul.opendev.org/t/openstack/build/614fe2bc45c44d4788813517baa03970 : SUCCESS in 1h 24m 41s (non-voting)\n- kolla-build-debian-source https://zuul.opendev.org/t/openstack/build/75f8fee652d04453a379e94053870483 : FAILURE in 28m 35s (non-voting)\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/aeb331974e8f4485af9d602ea260c33e : SUCCESS in 4m 12s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/91e67086e068468780132a3410e7c7f1 : SUCCESS in 5m 00s\n- openstack-tox-py27 https://zuul.opendev.org/t/openstack/build/4e1e8bcf9ac24816a3ac0934008b9205 : SUCCESS in 4m 12s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/6cb413eb8807405a9379861acd090276 : SUCCESS in 3m 43s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/eadaeee7d60b41cb9261e92aaa6afe20 : SUCCESS in 4m 03s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/f7105c4916a14b718e92f72b14aee30b : SUCCESS in 3m 57s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/abd4b7fe87e944b7bd92d30f3b61fcd1 : SUCCESS in 3m 56s\n- kolla-build-ubuntu-binary https://zuul.opendev.org/t/openstack/build/646ab61fd85b4e3ba1daa96c95717313 : FAILURE in 27m 46s (non-voting)\n- kolla-build-ubuntu-source https://zuul.opendev.org/t/openstack/build/9b30cdcf75774d5fa88bdb282598f37e : RETRY_LIMIT in 9m 10s\n- kolla-ansible-ubuntu-source https://zuul.opendev.org/t/openstack/build/09a53e3210df4a61b9f4802f72b92f54 : FAILURE in 1h 51m 14s\n- kolla-ansible-ubuntu-binary https://zuul.opendev.org/t/openstack/build/c898e432a8874551ba2d28ee723881ce : FAILURE in 1h 49m 35s (non-voting)\n- kolla-ansible-ubuntu-source-upgrade https://zuul.opendev.org/t/openstack/build/5f62a0748f0f4408a4751780eeb3ec80 : FAILURE in 1h 50m 19s\n- kolla-ansible-ubuntu-source-zun https://zuul.opendev.org/t/openstack/build/45ca6d4fd6db4f1aa6d5e99345ed41f7 : FAILURE in 24m 56s (non-voting)\n- kolla-ansible-ubuntu-source-ceph https://zuul.opendev.org/t/openstack/build/5538dea20e4a44b4a09482c724ea757d : FAILURE in 23m 01s (non-voting)\n- kolla-ansible-ubuntu-source-upgrade-ceph https://zuul.opendev.org/t/openstack/build/f4a11a8761a34435806b43df3d9befbb : FAILURE in 23m 54s (non-voting)\n- kolla-ansible-ubuntu-source-cinder-lvm https://zuul.opendev.org/t/openstack/build/cb4b3f728e6c4329a98be75b8468b2ea : FAILURE in 20m 39s (non-voting)\n- kolla-ansible-ubuntu-source-ironic https://zuul.opendev.org/t/openstack/build/f351b9344707458aa21ed05aaafb87b0 : FAILURE in 33m 56s (non-voting)","accounts_in_message":[],"_revision_number":1},{"id":"0a81073e1f0a61a26d83e34476bff4ecaf0d671d","author":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},"date":"2019-09-12 12:10:52.000000000","message":"Patch Set 1:\n\n(1 comment)","accounts_in_message":[],"_revision_number":1},{"id":"870e1e8b9bf526a9a1bcc7bdf34e6d901700ac3a","author":{"_account_id":24072,"name":"Marcin Juszkiewicz","email":"mjuszkiewicz@redhat.com","username":"hrw"},"date":"2019-09-12 12:11:47.000000000","message":"Patch Set 1:\n\n(3 comments)","accounts_in_message":[],"_revision_number":1},{"id":"17def6975983312ee7d018862af15dee7f3af4d8","author":{"_account_id":24072,"name":"Marcin Juszkiewicz","email":"mjuszkiewicz@redhat.com","username":"hrw"},"date":"2019-09-12 12:12:00.000000000","message":"Patch Set 1:\n\n(1 comment)","accounts_in_message":[],"_revision_number":1},{"id":"501a2aa83f0286a3edac067379f7c10443606aaf","author":{"_account_id":24072,"name":"Marcin Juszkiewicz","email":"mjuszkiewicz@redhat.com","username":"hrw"},"date":"2019-09-12 12:14:39.000000000","message":"Patch Set 1:\n\n(1 comment)","accounts_in_message":[],"_revision_number":1},{"id":"0509d275005db8ba700f8c0926d0413c966b9050","author":{"_account_id":24072,"name":"Marcin Juszkiewicz","email":"mjuszkiewicz@redhat.com","username":"hrw"},"date":"2019-09-12 12:14:48.000000000","message":"Uploaded patch set 2.","accounts_in_message":[],"_revision_number":2},{"id":"d506cfdfdfab39bf1ef7cda74fd270c7d205efb4","author":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},"date":"2019-09-12 12:37:08.000000000","message":"Patch Set 2:\n\n(1 comment)","accounts_in_message":[],"_revision_number":2},{"id":"6c5b5e8e98029354512c821d14ef103ab426b74e","author":{"_account_id":23181,"name":"RDO Third Party CI","email":"dmsimard+rdothirdparty@redhat.com","username":"rdothirdparty","tags":["SERVICE_USER"]},"date":"2019-09-12 12:54:24.000000000","message":"Patch Set 2:\n\nBuild succeeded (check pipeline).\n\n- tripleo-build-containers-rhel-8-master http://logs.rdoproject.org/94/681694/2/openstack-check/tripleo-build-containers-rhel-8-master/3f00a27/ : SUCCESS in 38m 38s","accounts_in_message":[],"_revision_number":2},{"id":"aa04bfb2a4f7e9bccb90b4445b095f5f62f89e43","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-09-12 14:21:38.000000000","message":"Patch Set 2: Verified+1\n\nBuild succeeded (check pipeline).\n\n- test-release-openstack https://zuul.opendev.org/t/openstack/build/0af057f3fba84c208d37b3da127d7875 : SUCCESS in 1m 53s\n- kolla-build-centos-binary https://zuul.opendev.org/t/openstack/build/f2fac779758548a3ac2884a561f8ed09 : POST_FAILURE in 48m 59s (non-voting)\n- kolla-build-centos-source https://zuul.opendev.org/t/openstack/build/1dbd364553534b9a8bbe9d117918c25a : SUCCESS in 49m 19s\n- kolla-ansible-centos-source https://zuul.opendev.org/t/openstack/build/3775d446e4f1457f884011863aeb5922 : SUCCESS in 1h 24m 17s\n- kolla-ansible-centos-binary https://zuul.opendev.org/t/openstack/build/dd529a3e6e724e32b9b5cb200ce0131d : SUCCESS in 1h 17m 06s (non-voting)\n- kolla-ansible-centos-source-upgrade https://zuul.opendev.org/t/openstack/build/63614fab00774b87b4e55312c76bcabf : SUCCESS in 1h 23m 48s\n- tripleo-build-containers-centos-7 https://zuul.opendev.org/t/openstack/build/ab03cdeba9bc47a48474e42b26e5b326 : SUCCESS in 47m 11s\n- kolla-ansible-bifrost-centos-source https://zuul.opendev.org/t/openstack/build/a33a8fd20931401b977980a294114d0b : SUCCESS in 56m 57s (non-voting)\n- kolla-ansible-centos-source-ceph https://zuul.opendev.org/t/openstack/build/edd089d209624f8887ee27f2e4e02d23 : POST_FAILURE in 1h 45m 00s (non-voting)\n- kolla-ansible-centos-source-upgrade-ceph https://zuul.opendev.org/t/openstack/build/4b4acf1ffca84862b03f8f0e453ed0ed : SUCCESS in 1h 54m 29s (non-voting)\n- kolla-ansible-centos-source-cinder-lvm https://zuul.opendev.org/t/openstack/build/ac73d150f5a94b46b597745b0193baba : SUCCESS in 1h 27m 01s (non-voting)\n- kolla-ansible-centos-source-scenario-nfv https://zuul.opendev.org/t/openstack/build/89e0b14968fa4bc79d92568f74e22427 : SUCCESS in 1h 47m 34s (non-voting)\n- kolla-ansible-centos-source-ironic https://zuul.opendev.org/t/openstack/build/260fe973826f4052b6400e63a41ae7e7 : SUCCESS in 1h 25m 45s (non-voting)\n- kolla-build-debian-source https://zuul.opendev.org/t/openstack/build/6196ee3db9c54cec8009374824f7c46a : SUCCESS in 22m 22s (non-voting)\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/30942fe7ac0b4bb897deff40d57cd164 : SUCCESS in 4m 03s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/876564aa67694e0095d2ded02dbd95d7 : SUCCESS in 5m 35s\n- openstack-tox-py27 https://zuul.opendev.org/t/openstack/build/cf26e63a2e1b4a76bebcba7309e3b543 : SUCCESS in 3m 09s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/80c6b208f306418d85140ace305b987a : SUCCESS in 4m 23s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/401b9c7220274e46a7367afbbbaa6d66 : SUCCESS in 4m 43s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/d77370d78e9e46f494902de10274221c : SUCCESS in 3m 31s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/7270325d34914fb7b0768e53e7291280 : SUCCESS in 5m 31s\n- kolla-build-ubuntu-binary https://zuul.opendev.org/t/openstack/build/4bf7f650d62f4422915b05a76b23a3e9 : SUCCESS in 26m 20s (non-voting)\n- kolla-build-ubuntu-source https://zuul.opendev.org/t/openstack/build/0c0c2e2b217949bb865381f7e002f6f4 : SUCCESS in 31m 10s\n- kolla-ansible-ubuntu-source https://zuul.opendev.org/t/openstack/build/065045cf28594aefa14c3322bb54c491 : SUCCESS in 51m 04s\n- kolla-ansible-ubuntu-binary https://zuul.opendev.org/t/openstack/build/4aa5d025acb1479ebdf3e5444c921029 : SUCCESS in 48m 13s (non-voting)\n- kolla-ansible-ubuntu-source-upgrade https://zuul.opendev.org/t/openstack/build/f8ff622c3f574e608821796d177e647d : SUCCESS in 58m 37s\n- kolla-ansible-ubuntu-source-zun https://zuul.opendev.org/t/openstack/build/1b314ce1a5df4184989c9c9d86eff30d : SUCCESS in 1h 03m 17s (non-voting)\n- kolla-ansible-ubuntu-source-ceph https://zuul.opendev.org/t/openstack/build/28e7d387cf394964adf73b46bbcea3ed : SUCCESS in 1h 24m 17s (non-voting)\n- kolla-ansible-ubuntu-source-upgrade-ceph https://zuul.opendev.org/t/openstack/build/19878556973042a799c80310fc009167 : SUCCESS in 1h 36m 31s (non-voting)\n- kolla-ansible-ubuntu-source-cinder-lvm https://zuul.opendev.org/t/openstack/build/45484dad3fba4b9cab4500582957de14 : SUCCESS in 1h 04m 26s (non-voting)\n- kolla-ansible-ubuntu-source-ironic https://zuul.opendev.org/t/openstack/build/1952e256402942c4b3a2cfb65d1bf392 : POST_FAILURE in 1h 01m 25s (non-voting)","accounts_in_message":[],"_revision_number":2},{"id":"6df4d1d8f6f92c99822c26742d2eee3531d00e46","author":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},"date":"2019-09-12 15:04:19.000000000","message":"Patch Set 2: Code-Review+1\n\nrecheck","accounts_in_message":[],"_revision_number":2},{"id":"963b0ba9a2b503c45bd35d2273b1b6afcbf052ad","author":{"_account_id":23181,"name":"RDO Third Party CI","email":"dmsimard+rdothirdparty@redhat.com","username":"rdothirdparty","tags":["SERVICE_USER"]},"date":"2019-09-12 16:32:39.000000000","message":"Patch Set 2:\n\nBuild succeeded (check pipeline).\n\n- tripleo-build-containers-rhel-8-master http://logs.rdoproject.org/94/681694/2/openstack-check/tripleo-build-containers-rhel-8-master/6347d27/ : SUCCESS in 1h 27m 18s","accounts_in_message":[],"_revision_number":2},{"id":"337ae8c44be2f64b31faa746a511ec177713e4cc","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-09-12 17:38:24.000000000","message":"Patch Set 2:\n\nBuild succeeded (check pipeline).\n\n- test-release-openstack https://zuul.opendev.org/t/openstack/build/15d7e60496cb4edeb180f60892dd6499 : SUCCESS in 2m 51s\n- kolla-build-centos-binary https://zuul.opendev.org/t/openstack/build/c9adee23d3e1445191854d19a5fb06b5 : SUCCESS in 35m 22s (non-voting)\n- kolla-build-centos-source https://zuul.opendev.org/t/openstack/build/6bad088d95f54a99b39a72246fca7439 : SUCCESS in 49m 52s\n- kolla-ansible-centos-source https://zuul.opendev.org/t/openstack/build/ffd10b4220414c7a875946a3ef5f1a85 : SUCCESS in 1h 15m 43s\n- kolla-ansible-centos-binary https://zuul.opendev.org/t/openstack/build/37022e8fe2c14a4eb4f97ad38d9c22ef : SUCCESS in 1h 10m 40s (non-voting)\n- kolla-ansible-centos-source-upgrade https://zuul.opendev.org/t/openstack/build/20da8304a31846d0a3c6cf0500af7577 : SUCCESS in 1h 29m 46s\n- tripleo-build-containers-centos-7 https://zuul.opendev.org/t/openstack/build/ed2bf70371084794a6055b5f87d19439 : SUCCESS in 45m 12s\n- kolla-ansible-bifrost-centos-source https://zuul.opendev.org/t/openstack/build/1959e2f6e6d0432fb9bec74fd0eca196 : SUCCESS in 46m 00s (non-voting)\n- kolla-ansible-centos-source-ceph https://zuul.opendev.org/t/openstack/build/af3afa45ac8445b6b1a8737afd61f632 : SUCCESS in 1h 43m 40s (non-voting)\n- kolla-ansible-centos-source-upgrade-ceph https://zuul.opendev.org/t/openstack/build/b39fe6f286f34dfba39ee92a5ff37d1b : SUCCESS in 2h 29m 13s (non-voting)\n- kolla-ansible-centos-source-cinder-lvm https://zuul.opendev.org/t/openstack/build/f64bb94034c44222b4a171eb8ecb434f : SUCCESS in 1h 26m 38s (non-voting)\n- kolla-ansible-centos-source-scenario-nfv https://zuul.opendev.org/t/openstack/build/87c4bdcca0e54420882694cd338968f3 : FAILURE in 1h 32m 47s (non-voting)\n- kolla-ansible-centos-source-ironic https://zuul.opendev.org/t/openstack/build/4ae785890e214da6bbb651c5df35ab5d : SUCCESS in 1h 21m 39s (non-voting)\n- kolla-build-debian-source https://zuul.opendev.org/t/openstack/build/30d94691b64046298982ae6f56ce2b42 : SUCCESS in 29m 14s (non-voting)\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/0f76ccbaa5de48728f8cb52639a50e58 : SUCCESS in 4m 42s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/975ed8b82d8842baae71a5a6b667ce66 : SUCCESS in 4m 58s\n- openstack-tox-py27 https://zuul.opendev.org/t/openstack/build/6fdcb038acee4915beed986f2ee689ff : SUCCESS in 2m 55s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/18faa8fcacfc442e9d8b07d368748be0 : SUCCESS in 3m 20s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/b12757998d7e435e9c317a4cb0e3603c : SUCCESS in 4m 52s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/a3691d32aad547bb8a9acb155cf9a9c3 : SUCCESS in 3m 52s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/56eac6f8da774ab7aaaa4e3de76bda29 : SUCCESS in 4m 30s\n- kolla-build-ubuntu-binary https://zuul.opendev.org/t/openstack/build/3f826c76a3e748da87116d8a8b2e5607 : SUCCESS in 27m 37s (non-voting)\n- kolla-build-ubuntu-source https://zuul.opendev.org/t/openstack/build/db7f026dffbf49f2a1a78361050b63d5 : SUCCESS in 38m 44s\n- kolla-ansible-ubuntu-source https://zuul.opendev.org/t/openstack/build/7f437ae5773c40568a80eab1eaeeaf89 : SUCCESS in 1h 03m 20s\n- kolla-ansible-ubuntu-binary https://zuul.opendev.org/t/openstack/build/9f964553f9ce492192dfaf35fa03eb23 : SUCCESS in 48m 49s (non-voting)\n- kolla-ansible-ubuntu-source-upgrade https://zuul.opendev.org/t/openstack/build/a74a6de2cddc4bdd8153476197f98fd3 : SUCCESS in 1h 03m 44s\n- kolla-ansible-ubuntu-source-zun https://zuul.opendev.org/t/openstack/build/21e53f446bfb43faa08e8d182f111967 : SUCCESS in 1h 11m 18s (non-voting)\n- kolla-ansible-ubuntu-source-ceph https://zuul.opendev.org/t/openstack/build/ad85978921864357a8846e232b25c450 : SUCCESS in 1h 17m 33s (non-voting)\n- kolla-ansible-ubuntu-source-upgrade-ceph https://zuul.opendev.org/t/openstack/build/6655535fb442472bb84911abd63b96ad : SUCCESS in 1h 22m 37s (non-voting)\n- kolla-ansible-ubuntu-source-cinder-lvm https://zuul.opendev.org/t/openstack/build/e579807b1e4c467dae4ba89a06332da0 : SUCCESS in 1h 02m 35s (non-voting)\n- kolla-ansible-ubuntu-source-ironic https://zuul.opendev.org/t/openstack/build/08bfb34bd39645bf933823e3c88e72af : SUCCESS in 1h 00m 33s (non-voting)","accounts_in_message":[],"_revision_number":2},{"id":"af0731acf45b237af4c373596d54638da151b7ce","author":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},"date":"2019-09-13 08:40:42.000000000","message":"Patch Set 2: Code-Review+2","accounts_in_message":[],"_revision_number":2},{"id":"b215aae48d578f5b862dbb94a268e6ac46945dd2","author":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},"date":"2019-09-13 10:19:50.000000000","message":"Patch Set 2: Review-Priority+2","accounts_in_message":[],"_revision_number":2},{"id":"da128dc0bfe63ebda39d544b037362128781214d","author":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},"date":"2019-09-13 12:27:16.000000000","message":"Patch Set 2: Workflow+1\n\nApproving to unblock CI. Please leave comments and changes can be made in a follow up patch if necessary.","accounts_in_message":[],"_revision_number":2},{"id":"8e5bff8ffb465e6609cfcf41b4bbbddcba5b20ca","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-09-13 12:27:29.000000000","message":"Patch Set 2: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":2},{"id":"385cd83bc1aef60faaf4a63b7c7244833c9fb491","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-09-13 13:56:52.000000000","message":"Patch Set 2: Verified-2\n\nBuild failed (gate pipeline).  For information on how to proceed, see\nhttp://docs.openstack.org/infra/manual/developers.html#automated-testing\n\n\n- test-release-openstack https://zuul.opendev.org/t/openstack/build/5743d3b574b147e5a23ede279ecc1b6b : SUCCESS in 2m 09s\n- kolla-build-centos-source https://zuul.opendev.org/t/openstack/build/6770b2f8d84b43468a4596dff11fa6ad : SUCCESS in 48m 50s\n- kolla-ansible-centos-source https://zuul.opendev.org/t/openstack/build/060e0eb287224f7baf7af46066a7acb7 : SUCCESS in 1h 10m 39s\n- kolla-ansible-centos-source-upgrade https://zuul.opendev.org/t/openstack/build/72b6097dff6a427bb09fea27636e5915 : SUCCESS in 1h 23m 40s\n- tripleo-build-containers-centos-7 https://zuul.opendev.org/t/openstack/build/5d05c0d03f7743fca8f5cd21f880545b : SUCCESS in 44m 39s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/a28bd827afed4960945ed3c37f206bb2 : SUCCESS in 3m 16s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/fc96886838e04d939ac47356cd5c4928 : SUCCESS in 4m 30s\n- openstack-tox-py27 https://zuul.opendev.org/t/openstack/build/86aa8ed3e4ce4440856269c21ab033e5 : SUCCESS in 3m 10s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/fa7e3f97ea824745a9536acc4ea62858 : SUCCESS in 3m 09s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/627ae5ae20e94492897f4d462cfc6535 : SUCCESS in 3m 42s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/df1dc149ff1e4567aa1a4ea1b189793f : SUCCESS in 4m 24s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/27ec9035326d425a815c2f11ac4b6265 : SUCCESS in 5m 35s\n- kolla-build-ubuntu-source https://zuul.opendev.org/t/openstack/build/11bc9f1ff06842899894e8045250d872 : FAILURE in 23m 34s\n- kolla-ansible-ubuntu-source https://zuul.opendev.org/t/openstack/build/64a8bafe9d5045c79b309f8b3a6f99c7 : FAILURE in 13m 56s\n- kolla-ansible-ubuntu-source-upgrade https://zuul.opendev.org/t/openstack/build/5fab06cf49fe402ba7ce05d7a7c02dc4 : FAILURE in 12m 47s","accounts_in_message":[],"_revision_number":2},{"id":"42e3e781a04cd84c9b5045f3e49dd2af9712d345","author":{"_account_id":24072,"name":"Marcin Juszkiewicz","email":"mjuszkiewicz@redhat.com","username":"hrw"},"date":"2019-09-13 14:19:15.000000000","message":"Patch Set 2:\n\nrecheck\n\nand hope it hit other server for ubuntu","accounts_in_message":[],"_revision_number":2},{"id":"caef52978df27f4a931f006095d33bbffe8461dc","author":{"_account_id":24072,"name":"Marcin Juszkiewicz","email":"mjuszkiewicz@redhat.com","username":"hrw"},"date":"2019-09-13 15:07:15.000000000","message":"Patch Set 2:\n\nrecheck\n\nubuntu mirrors in infra got refreshed","accounts_in_message":[],"_revision_number":2},{"id":"cb643937a4a672fc3420cad3f7286ef4da932deb","author":{"_account_id":23181,"name":"RDO Third Party CI","email":"dmsimard+rdothirdparty@redhat.com","username":"rdothirdparty","tags":["SERVICE_USER"]},"date":"2019-09-13 15:48:21.000000000","message":"Patch Set 2:\n\nBuild succeeded (check pipeline).\n\n- tripleo-build-containers-rhel-8-master http://logs.rdoproject.org/94/681694/2/openstack-check/tripleo-build-containers-rhel-8-master/884c520/ : SUCCESS in 40m 31s","accounts_in_message":[],"_revision_number":2},{"id":"4f2986791d5602e37828617fc78a04b8be0a8877","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-09-13 16:34:36.000000000","message":"Patch Set 2: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttp://docs.openstack.org/infra/manual/developers.html#automated-testing\n\n\n- test-release-openstack https://zuul.opendev.org/t/openstack/build/a81c3acd775445f4b884c68aefca1951 : SUCCESS in 2m 13s\n- kolla-build-centos-binary https://zuul.opendev.org/t/openstack/build/1a53254fef6340eeaa383ca4056391f2 : SUCCESS in 45m 30s (non-voting)\n- kolla-build-centos-source https://zuul.opendev.org/t/openstack/build/631a880175d6403a8ae771414a858f94 : SUCCESS in 55m 27s\n- kolla-ansible-centos-source https://zuul.opendev.org/t/openstack/build/dbb098ccebe841718953d0c36c973d92 : SUCCESS in 1h 15m 52s\n- kolla-ansible-centos-binary https://zuul.opendev.org/t/openstack/build/cd1abd34ae29470091cc49e6031c3a0c : SUCCESS in 1h 13m 46s (non-voting)\n- kolla-ansible-centos-source-upgrade https://zuul.opendev.org/t/openstack/build/dcafb1cbe4c542e58970e011a9b37f26 : SUCCESS in 1h 22m 24s\n- tripleo-build-containers-centos-7 https://zuul.opendev.org/t/openstack/build/88eaf9dfdc694473aa3736f4326b56ba : SUCCESS in 50m 14s\n- kolla-ansible-bifrost-centos-source https://zuul.opendev.org/t/openstack/build/2b4d5acb7b124da1bbc207e608619785 : SUCCESS in 45m 36s (non-voting)\n- kolla-ansible-centos-source-ceph https://zuul.opendev.org/t/openstack/build/a74230bc65034c21928b5b4bc1821c88 : SUCCESS in 1h 42m 28s (non-voting)\n- kolla-ansible-centos-source-upgrade-ceph https://zuul.opendev.org/t/openstack/build/70da97487dbb4523a00766fb780ad9e9 : SUCCESS in 2h 07m 50s (non-voting)\n- kolla-ansible-centos-source-cinder-lvm https://zuul.opendev.org/t/openstack/build/c04ccdb2f5ad4ad4857014aaf5616b77 : SUCCESS in 1h 41m 42s (non-voting)\n- kolla-ansible-centos-source-scenario-nfv https://zuul.opendev.org/t/openstack/build/45b1c33620114a1e91914f0972cedd9e : SUCCESS in 1h 40m 01s (non-voting)\n- kolla-ansible-centos-source-ironic https://zuul.opendev.org/t/openstack/build/85b81c2175a843b19d4f08b115eaa254 : SUCCESS in 1h 42m 12s (non-voting)\n- kolla-build-debian-source https://zuul.opendev.org/t/openstack/build/97bd6473de94475b816effe2aa4ea5be : SUCCESS in 23m 22s (non-voting)\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/4aa37d022c7d402dbd1b3fce318967a2 : SUCCESS in 3m 34s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/2720c00416824593807dc2b067ccc741 : SUCCESS in 4m 14s\n- openstack-tox-py27 https://zuul.opendev.org/t/openstack/build/b35a2c745ca04b7cbb4e9f1dee842455 : SUCCESS in 4m 19s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/ab0c6b0fbdf642b295ac206a0abd2e01 : SUCCESS in 3m 18s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/b73116ffda88417bba3e76dbf77658d8 : SUCCESS in 3m 36s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/744f6aa5f01e4a66bb7bdb979480b999 : SUCCESS in 2m 51s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/de078306affc4f2e99c31b4b78384ba4 : SUCCESS in 5m 46s\n- kolla-build-ubuntu-binary https://zuul.opendev.org/t/openstack/build/2b34baf69b7d4da18cd60abe7bc786b4 : FAILURE in 25m 35s (non-voting)\n- kolla-build-ubuntu-source https://zuul.opendev.org/t/openstack/build/beaf046f7e67408b946dfa753f5437a2 : FAILURE in 23m 26s\n- kolla-ansible-ubuntu-source https://zuul.opendev.org/t/openstack/build/97c8038844e24eaa8ecb5cf0ed79e8cc : FAILURE in 14m 10s\n- kolla-ansible-ubuntu-binary https://zuul.opendev.org/t/openstack/build/07e2f3a284424aa384a5d92380245fca : FAILURE in 19m 11s (non-voting)\n- kolla-ansible-ubuntu-source-upgrade https://zuul.opendev.org/t/openstack/build/2d304cbdeec54bbcb05cbb2b481c5a91 : FAILURE in 13m 54s\n- kolla-ansible-ubuntu-source-zun https://zuul.opendev.org/t/openstack/build/f6b3a0cf059540cbbd80e9f8b7018f74 : FAILURE in 15m 07s (non-voting)\n- kolla-ansible-ubuntu-source-ceph https://zuul.opendev.org/t/openstack/build/e843a976b18d422ba1e56b6bccaf77d7 : FAILURE in 13m 17s (non-voting)\n- kolla-ansible-ubuntu-source-upgrade-ceph https://zuul.opendev.org/t/openstack/build/21b72c5b96c1491eae7e8d331c439d2c : FAILURE in 23m 22s (non-voting)\n- kolla-ansible-ubuntu-source-cinder-lvm https://zuul.opendev.org/t/openstack/build/6f66783b93e64159ada4b9742dc8c3f3 : FAILURE in 15m 36s (non-voting)\n- kolla-ansible-ubuntu-source-ironic https://zuul.opendev.org/t/openstack/build/3c83286bc42942658ee587adfa3bc3c1 : FAILURE in 11m 26s (non-voting)","accounts_in_message":[],"_revision_number":2},{"id":"0eb1e6104e5547081b706baa474dfc1625e61ae1","author":{"_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"},"date":"2019-09-13 17:07:23.000000000","message":"Patch Set 2:\n\nrecheck","accounts_in_message":[],"_revision_number":2},{"id":"a9a99f628422b5f88ee0c6126a69466db457d98a","author":{"_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"},"date":"2019-09-13 17:14:09.000000000","message":"Patch Set 2:\n\n(1 comment)","accounts_in_message":[],"_revision_number":2},{"id":"61c62e66b5393c82c7ea72fdb1f5754b732f0d58","author":{"_account_id":23181,"name":"RDO Third Party CI","email":"dmsimard+rdothirdparty@redhat.com","username":"rdothirdparty","tags":["SERVICE_USER"]},"date":"2019-09-13 19:18:58.000000000","message":"Patch Set 2:\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttp://docs.openstack.org/infra/manual/developers.html#automated-testing\n\n\n- tripleo-build-containers-rhel-8-master http://logs.rdoproject.org/94/681694/2/openstack-check/tripleo-build-containers-rhel-8-master/5c2baa9/ : TIMED_OUT in 2h 10m 34s","accounts_in_message":[],"_revision_number":2},{"id":"cce024aff915e4d48553f948df46c144a4b11f59","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-09-13 20:01:01.000000000","message":"Patch Set 2:\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttp://docs.openstack.org/infra/manual/developers.html#automated-testing\n\n\n- test-release-openstack https://zuul.opendev.org/t/openstack/build/5b278f1dd23d4d13b7358aaf89ad6965 : SUCCESS in 1m 57s\n- kolla-build-centos-binary https://zuul.opendev.org/t/openstack/build/68dad735eed6410b816436148a3caa02 : SUCCESS in 42m 30s (non-voting)\n- kolla-build-centos-source https://zuul.opendev.org/t/openstack/build/2be2bf02fdf041919e86ca164d0c6966 : FAILURE in 27m 11s\n- kolla-ansible-centos-source https://zuul.opendev.org/t/openstack/build/1dcade51f81643089b01bdbc6961d280 : SUCCESS in 1h 27m 25s\n- kolla-ansible-centos-binary https://zuul.opendev.org/t/openstack/build/7bc04399f91f486f92bf1af8fedf446c : SUCCESS in 1h 11m 06s (non-voting)\n- kolla-ansible-centos-source-upgrade https://zuul.opendev.org/t/openstack/build/522abdbd03844c3cb1d3bd9e4b57bef2 : SUCCESS in 1h 41m 24s\n- tripleo-build-containers-centos-7 https://zuul.opendev.org/t/openstack/build/8aadbff1fd9e4988a2eabc19988bd4b3 : SUCCESS in 48m 25s\n- kolla-ansible-bifrost-centos-source https://zuul.opendev.org/t/openstack/build/e97dae2df94c4bb1babde2fa8df943d4 : SUCCESS in 59m 47s (non-voting)\n- kolla-ansible-centos-source-ceph https://zuul.opendev.org/t/openstack/build/277d4a7ab94a47ceb2d00c20bcaf18a9 : SUCCESS in 1h 50m 14s (non-voting)\n- kolla-ansible-centos-source-upgrade-ceph https://zuul.opendev.org/t/openstack/build/23e5d4772f5b47b8bc80f6d483cc6a9e : SUCCESS in 2h 00m 38s (non-voting)\n- kolla-ansible-centos-source-cinder-lvm https://zuul.opendev.org/t/openstack/build/734aa9e9ef794cc29ab6888b741c8660 : SUCCESS in 1h 36m 51s (non-voting)\n- kolla-ansible-centos-source-scenario-nfv https://zuul.opendev.org/t/openstack/build/67c5b864df814c65bf5037e2c50b41f0 : SUCCESS in 1h 37m 50s (non-voting)\n- kolla-ansible-centos-source-ironic https://zuul.opendev.org/t/openstack/build/8d0dbfb88f1140efb739ea12585bbe91 : SUCCESS in 1h 24m 42s (non-voting)\n- kolla-build-debian-source https://zuul.opendev.org/t/openstack/build/c0e5f21f3fe04ef8a39b2007fa774122 : SUCCESS in 27m 45s (non-voting)\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/a686e34f0064458db0100b2dd856c5c0 : SUCCESS in 3m 54s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/08d923e801de4c6aa3408e952a336151 : SUCCESS in 4m 48s\n- openstack-tox-py27 https://zuul.opendev.org/t/openstack/build/61865438b01044d6b89639fb663923ac : SUCCESS in 4m 11s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/6f0c8c01a6704c918e521d00a980aed5 : SUCCESS in 4m 17s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/765cf3fdefc140799e416e09912b8e0e : SUCCESS in 4m 58s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/7c0ddf120528478c8e0bbfc513f23adf : SUCCESS in 3m 19s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/10789b3b6b4044c6a586a94832d2146b : SUCCESS in 4m 27s\n- kolla-build-ubuntu-binary https://zuul.opendev.org/t/openstack/build/76f48dc477ae42f7bd91d2fe24d88d19 : SUCCESS in 28m 04s (non-voting)\n- kolla-build-ubuntu-source https://zuul.opendev.org/t/openstack/build/620e23a2fecc4ff3913474a459c798ac : SUCCESS in 30m 29s\n- kolla-ansible-ubuntu-source https://zuul.opendev.org/t/openstack/build/cc2f3fe6d00d466d91e420a8f3dca383 : SUCCESS in 50m 14s\n- kolla-ansible-ubuntu-binary https://zuul.opendev.org/t/openstack/build/bc11986945ef409797b18c11e7aa5cf5 : SUCCESS in 52m 16s (non-voting)\n- kolla-ansible-ubuntu-source-upgrade https://zuul.opendev.org/t/openstack/build/a025820b0d084536844d5dcf01c06edd : SUCCESS in 1h 04m 34s\n- kolla-ansible-ubuntu-source-zun https://zuul.opendev.org/t/openstack/build/2d578d568d494bcfb974f3f6b4dda928 : SUCCESS in 1h 05m 57s (non-voting)\n- kolla-ansible-ubuntu-source-ceph https://zuul.opendev.org/t/openstack/build/ad20d1b708264054b35d793b30160024 : SUCCESS in 1h 19m 33s (non-voting)\n- kolla-ansible-ubuntu-source-upgrade-ceph https://zuul.opendev.org/t/openstack/build/91dbfefbd0624ff0b69711c8bee9b726 : TIMED_OUT in 2h 34m 30s (non-voting)\n- kolla-ansible-ubuntu-source-cinder-lvm https://zuul.opendev.org/t/openstack/build/8389cd9686934b74b1aa7da5a7e00bf1 : SUCCESS in 1h 05m 09s (non-voting)\n- kolla-ansible-ubuntu-source-ironic https://zuul.opendev.org/t/openstack/build/981ea9ffc9fa4f7b85eb6480146eea08 : SUCCESS in 1h 55m 00s (non-voting)","accounts_in_message":[],"_revision_number":2},{"id":"bb14b7445b2ae89408054f0c7c2dabd7bd492dbe","author":{"_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"},"date":"2019-09-16 08:31:56.000000000","message":"Patch Set 2:\n\nrecheck","accounts_in_message":[],"_revision_number":2},{"id":"1fa54bc9c54ba475d502f9dd0af205078d316dc4","author":{"_account_id":24072,"name":"Marcin Juszkiewicz","email":"mjuszkiewicz@redhat.com","username":"hrw"},"date":"2019-09-16 08:33:34.000000000","message":"Patch Set 2:\n\n(1 comment)","accounts_in_message":[],"_revision_number":2},{"id":"e982da8ca829a4afe64171eeb02ef19d429f5cf1","author":{"_account_id":23181,"name":"RDO Third Party CI","email":"dmsimard+rdothirdparty@redhat.com","username":"rdothirdparty","tags":["SERVICE_USER"]},"date":"2019-09-16 09:10:34.000000000","message":"Patch Set 2:\n\nBuild succeeded (check pipeline).\n\n- tripleo-build-containers-rhel-8-master http://logs.rdoproject.org/94/681694/2/openstack-check/tripleo-build-containers-rhel-8-master/756547e/ : SUCCESS in 37m 36s","accounts_in_message":[],"_revision_number":2},{"id":"f29852f8f119f2bd9bd73df2336ec4b38276046d","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-09-16 11:31:29.000000000","message":"Patch Set 2: Verified+1\n\nBuild succeeded (check pipeline).\n\n- test-release-openstack https://zuul.opendev.org/t/openstack/build/39bc1dbc847d498f811fc84b3bac38a4 : SUCCESS in 3m 00s\n- kolla-build-centos-binary https://zuul.opendev.org/t/openstack/build/2cdc2070865a427b874a3a4af6b1d24f : SUCCESS in 40m 18s (non-voting)\n- kolla-build-centos-source https://zuul.opendev.org/t/openstack/build/89b2d6cb35f84461b323c36ae5d1511c : SUCCESS in 55m 28s\n- kolla-ansible-centos-source https://zuul.opendev.org/t/openstack/build/515142b6169e4e2da065e34e4552b58b : SUCCESS in 1h 15m 25s\n- kolla-ansible-centos-binary https://zuul.opendev.org/t/openstack/build/896640ef60fa4e4d950fed56e54a8917 : SUCCESS in 1h 14m 51s (non-voting)\n- kolla-ansible-centos-source-upgrade https://zuul.opendev.org/t/openstack/build/9c73a896ffb347b99dc52408f91dfc3f : SUCCESS in 1h 22m 22s\n- tripleo-build-containers-centos-7 https://zuul.opendev.org/t/openstack/build/5099d9d6d37943f5905841974c212fad : SUCCESS in 42m 24s\n- kolla-ansible-bifrost-centos-source https://zuul.opendev.org/t/openstack/build/f0f546d6ba5641a7ba16d39d532e2bb8 : SUCCESS in 43m 18s (non-voting)\n- kolla-ansible-centos-source-ceph https://zuul.opendev.org/t/openstack/build/53736cc260b846cf9945ee9e28674019 : SUCCESS in 1h 41m 31s (non-voting)\n- kolla-ansible-centos-source-upgrade-ceph https://zuul.opendev.org/t/openstack/build/2d24da191c1342eb9a24daa1e369b294 : SUCCESS in 2h 28m 55s (non-voting)\n- kolla-ansible-centos-source-cinder-lvm https://zuul.opendev.org/t/openstack/build/a27138f0cde34ca183a5786bfe9b5b7f : SUCCESS in 1h 30m 31s (non-voting)\n- kolla-ansible-centos-source-scenario-nfv https://zuul.opendev.org/t/openstack/build/dc92d52d7a5144c6ac2ac6a8b734d6e6 : SUCCESS in 1h 46m 49s (non-voting)\n- kolla-ansible-centos-source-ironic https://zuul.opendev.org/t/openstack/build/65fc638f840a4ea6a19f09bb16495a2e : SUCCESS in 1h 16m 17s (non-voting)\n- kolla-build-debian-source https://zuul.opendev.org/t/openstack/build/b5e7927f2bd442f2ad87bd49043024be : SUCCESS in 39m 46s (non-voting)\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/5cb1dee3e98245438de04d6d3db64c83 : SUCCESS in 3m 36s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/50932fca58a642b9a33b1dd5172e1a8c : SUCCESS in 4m 53s\n- openstack-tox-py27 https://zuul.opendev.org/t/openstack/build/6d3fbaac980241fab9799505f3a0e4fc : SUCCESS in 4m 03s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/924234470965461eb8b43168b1d2b98f : SUCCESS in 4m 48s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/cb37a7b5efe04b3ba55656e939b00d48 : SUCCESS in 3m 58s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/2642dac54519438bba01b368832d07ed : SUCCESS in 3m 51s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/fb0c9d5a189949289099e376b9e61c6d : SUCCESS in 5m 09s\n- kolla-build-ubuntu-binary https://zuul.opendev.org/t/openstack/build/8fb197deab7141b6bbc8dc47a8cb78af : SUCCESS in 30m 29s (non-voting)\n- kolla-build-ubuntu-source https://zuul.opendev.org/t/openstack/build/823987e06ba244ba8110619048b3bafd : SUCCESS in 35m 53s\n- kolla-ansible-ubuntu-source https://zuul.opendev.org/t/openstack/build/4db416837a1f4d2aba2ccb5d3eccdcff : SUCCESS in 50m 46s\n- kolla-ansible-ubuntu-binary https://zuul.opendev.org/t/openstack/build/993b9d8b26354cc4a7ff1ef2ed98b066 : SUCCESS in 53m 31s (non-voting)\n- kolla-ansible-ubuntu-source-upgrade https://zuul.opendev.org/t/openstack/build/db1b1736b8df411ca491930cf1c70d7a : SUCCESS in 54m 43s\n- kolla-ansible-ubuntu-source-zun https://zuul.opendev.org/t/openstack/build/93ec67d45b27436ca6838022ded7b1c3 : SUCCESS in 1h 06m 15s (non-voting)\n- kolla-ansible-ubuntu-source-ceph https://zuul.opendev.org/t/openstack/build/8649f1faddf94538bc939bd66c89c39d : SUCCESS in 1h 21m 06s (non-voting)\n- kolla-ansible-ubuntu-source-upgrade-ceph https://zuul.opendev.org/t/openstack/build/8e9947364db5490aaa26f3b6fdcfc639 : SUCCESS in 1h 41m 35s (non-voting)\n- kolla-ansible-ubuntu-source-cinder-lvm https://zuul.opendev.org/t/openstack/build/0a6c0ef5ac79477ea5c2624d9eb39196 : SUCCESS in 1h 13m 18s (non-voting)\n- kolla-ansible-ubuntu-source-ironic https://zuul.opendev.org/t/openstack/build/5b8c660321cf46569c5e04f9234bea56 : SUCCESS in 1h 06m 56s (non-voting)","accounts_in_message":[],"_revision_number":2},{"id":"d1c953f4519ebb755c148d93e3781cc135125eb1","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-09-16 11:31:43.000000000","message":"Patch Set 2: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":2},{"id":"aefb40e07e208eb86c496015cde1fef99308e78a","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-09-16 13:14:01.000000000","message":"Change has been successfully merged by Zuul","accounts_in_message":[],"_revision_number":2},{"id":"6aa8456d5873f96525c2e2b0df293e6730ec5e36","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-09-16 13:14:01.000000000","message":"Patch Set 2: Verified+2\n\nBuild succeeded (gate pipeline).\n\n- test-release-openstack https://zuul.opendev.org/t/openstack/build/e39807c50c5c4782bbc9a34985cec4e2 : SUCCESS in 2m 30s\n- kolla-build-centos-source https://zuul.opendev.org/t/openstack/build/d00b639afc3645999682006cd64a117c : SUCCESS in 57m 12s\n- kolla-ansible-centos-source https://zuul.opendev.org/t/openstack/build/eeb0c15125c3434c91d57eace25ef537 : SUCCESS in 1h 10m 14s\n- kolla-ansible-centos-source-upgrade https://zuul.opendev.org/t/openstack/build/a47abb731cd447968fffc5048ec48d70 : SUCCESS in 1h 38m 59s\n- tripleo-build-containers-centos-7 https://zuul.opendev.org/t/openstack/build/05333a4fae4b4b2888502a884bb84fa2 : SUCCESS in 55m 21s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/43845463b6f84b41b3776e9f556fce3a : SUCCESS in 3m 21s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/35d4e8c2cf94493c9c1a3674564b41b5 : SUCCESS in 4m 58s\n- openstack-tox-py27 https://zuul.opendev.org/t/openstack/build/84a15e8ffe364c24aa1521216dd9dfae : SUCCESS in 4m 11s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/ce291c36807248ac9081958c6668b25d : SUCCESS in 3m 49s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/0136356141734fdf82eff48b276bd886 : SUCCESS in 3m 34s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/e41f592b96b04e46ac7fec0274740931 : SUCCESS in 3m 22s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/b41449b7a3a44a4aa66c798dac91403e : SUCCESS in 4m 20s\n- kolla-build-ubuntu-source https://zuul.opendev.org/t/openstack/build/33ea33310c924fd887efd822ff1423fb : SUCCESS in 31m 34s\n- kolla-ansible-ubuntu-source https://zuul.opendev.org/t/openstack/build/cf6628cf733b469fa7c978183000d3af : SUCCESS in 57m 43s\n- kolla-ansible-ubuntu-source-upgrade https://zuul.opendev.org/t/openstack/build/52cfea85f0144581bdc20779d0513974 : SUCCESS in 1h 02m 24s","accounts_in_message":[],"_revision_number":2},{"id":"334c76e41e5a41609f363ccd5ef5dacc787d979b","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-09-16 13:16:03.000000000","message":"Patch Set 2:\n\nBuild succeeded (promote pipeline).\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/25e7b34543d84441aa9c17714bd4250a : SUCCESS in 1m 25s\n- promote-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/24bc28ac6e914e0e84a1e1eccd53e955 : SUCCESS in 1m 12s","accounts_in_message":[],"_revision_number":2}],"current_revision_number":2,"current_revision":"016c0f577c535def40dd32ddf41e6cb145a2f9b8","revisions":{"2eb9eb9a454a62bc326df642e029f10957fe5e5e":{"kind":"REWORK","_number":1,"created":"2019-09-12 09:49:04.000000000","uploader":{"_account_id":24072,"name":"Marcin Juszkiewicz","email":"mjuszkiewicz@redhat.com","username":"hrw"},"ref":"refs/changes/94/681694/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/kolla","ref":"refs/changes/94/681694/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/kolla refs/changes/94/681694/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/kolla refs/changes/94/681694/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/kolla refs/changes/94/681694/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/kolla refs/changes/94/681694/1"}}},"commit":{"parents":[{"commit":"7cda7ac96834c72ca818a3af449b2af6132bad9b","subject":"Merge \"base: drop not needed Linaro OBS key\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/7cda7ac96834c72ca818a3af449b2af6132bad9b"}]}],"author":{"name":"Marcin Juszkiewicz","email":"marcin.juszkiewicz@linaro.org","date":"2019-09-12 09:48:07.000000000","tz":120},"committer":{"name":"Marcin Juszkiewicz","email":"marcin.juszkiewicz@linaro.org","date":"2019-09-12 09:48:07.000000000","tz":120},"subject":"Drop Percona XtraBackup in favour of MariaDB Backup","message":"Drop Percona XtraBackup in favour of MariaDB Backup\n\nXtraBackup started failing to install on Ubuntu and does not support\nMariaDB 10.3 version.\n\nChange-Id: I9cf62c8e03a02574d85e5349df1d23a51d4c99be\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/2eb9eb9a454a62bc326df642e029f10957fe5e5e"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/2eb9eb9a454a62bc326df642e029f10957fe5e5e"}]},"branch":"refs/heads/master"},"016c0f577c535def40dd32ddf41e6cb145a2f9b8":{"kind":"REWORK","_number":2,"created":"2019-09-12 12:14:48.000000000","uploader":{"_account_id":24072,"name":"Marcin Juszkiewicz","email":"mjuszkiewicz@redhat.com","username":"hrw"},"ref":"refs/changes/94/681694/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/kolla","ref":"refs/changes/94/681694/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/kolla refs/changes/94/681694/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/kolla refs/changes/94/681694/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/kolla refs/changes/94/681694/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/kolla refs/changes/94/681694/2"}}},"commit":{"parents":[{"commit":"532af2efc8893cfd23f8d2beadd32a4e3463ae85","subject":"Do not fan out DB sync to all cells","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/532af2efc8893cfd23f8d2beadd32a4e3463ae85"}]}],"author":{"name":"Marcin Juszkiewicz","email":"marcin.juszkiewicz@linaro.org","date":"2019-09-12 09:48:07.000000000","tz":120},"committer":{"name":"Marcin Juszkiewicz","email":"marcin.juszkiewicz@linaro.org","date":"2019-09-12 12:14:08.000000000","tz":120},"subject":"Drop Percona XtraBackup in favour of MariaDB Backup","message":"Drop Percona XtraBackup in favour of MariaDB Backup\n\nXtraBackup started failing to install on Ubuntu and does not support\nMariaDB 10.3 version.\n\nRelated-Bug: #1843043\n\nChange-Id: I9cf62c8e03a02574d85e5349df1d23a51d4c99be\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/016c0f577c535def40dd32ddf41e6cb145a2f9b8"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/016c0f577c535def40dd32ddf41e6cb145a2f9b8"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[],"submit_requirements":[]}
