)]}'
{"id":"openstack%2Fkolla~681768","triplet_id":"openstack%2Fkolla~master~Id9db205c1af7607fde12dd8be9e0e25139597ccd","project":"openstack/kolla","branch":"master","topic":"bump_up_skydive","hashtags":[],"change_id":"Id9db205c1af7607fde12dd8be9e0e25139597ccd","subject":"Update skydive to v0.25.0","status":"MERGED","created":"2019-09-12 14:40:21.000000000","updated":"2019-09-18 11:44:57.000000000","submitted":"2019-09-18 11:43:20.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"681768-1568807000473-ab427080","meta_rev_id":"0908eff82b3848079063f0c732e2740b923476a5","_number":681768,"virtual_id_number":681768,"owner":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":19316,"name":"Eduardo Gonzalez","email":"dabarren@gmail.com","username":"egonzalez90"},{"value":0,"date":"2019-09-18 07:59:42.000000000","_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},{"value":0,"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},{"value":2,"date":"2019-09-18 11:43:20.000000000","post_submit":true,"permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"date":"2019-09-18 08:39:30.000000000","_account_id":23181,"name":"RDO Third Party CI","email":"dmsimard+rdothirdparty@redhat.com","username":"rdothirdparty","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":19316,"name":"Eduardo Gonzalez","email":"dabarren@gmail.com","username":"egonzalez90"},"all":[{"value":2,"date":"2019-09-17 06:57:04.000000000","_account_id":19316,"name":"Eduardo Gonzalez","email":"dabarren@gmail.com","username":"egonzalez90"},{"value":0,"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},{"value":2,"date":"2019-09-17 08:52:10.000000000","_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":23181,"name":"RDO Third Party CI","email":"dmsimard+rdothirdparty@redhat.com","username":"rdothirdparty","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},"all":[{"value":0,"_account_id":19316,"name":"Eduardo Gonzalez","email":"dabarren@gmail.com","username":"egonzalez90"},{"value":0,"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},{"value":1,"date":"2019-09-17 08:52:10.000000000","_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":23181,"name":"RDO Third Party CI","email":"dmsimard+rdothirdparty@redhat.com","username":"rdothirdparty","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Backport-Candidate":{"all":[{"value":0,"_account_id":19316,"name":"Eduardo Gonzalez","email":"dabarren@gmail.com","username":"egonzalez90"},{"value":0,"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},{"value":0,"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":23181,"name":"RDO Third Party CI","email":"dmsimard+rdothirdparty@redhat.com","username":"rdothirdparty","tags":["SERVICE_USER"]}],"values":{"-1":"Do Not Backport"," 0":"Backport Review Needed","+1":"Should Backport"},"description":"","default_value":0,"optional":true},"Review-Priority":{"all":[{"value":0,"_account_id":19316,"name":"Eduardo Gonzalez","email":"dabarren@gmail.com","username":"egonzalez90"},{"value":0,"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},{"value":0,"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":23181,"name":"RDO Third Party CI","email":"dmsimard+rdothirdparty@redhat.com","username":"rdothirdparty","tags":["SERVICE_USER"]}],"values":{"-1":"Branch Freeze"," 0":"No Priority","+1":"Important Change","+2":"Gate Blocker Fix / Urgent Change"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},{"_account_id":19316,"name":"Eduardo Gonzalez","email":"dabarren@gmail.com","username":"egonzalez90"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},{"_account_id":23181,"name":"RDO Third Party CI","email":"dmsimard+rdothirdparty@redhat.com","username":"rdothirdparty","tags":["SERVICE_USER"]}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2019-09-17 06:57:04.000000000","updated_by":{"_account_id":19316,"name":"Eduardo Gonzalez","email":"dabarren@gmail.com","username":"egonzalez90"},"reviewer":{"_account_id":19316,"name":"Eduardo Gonzalez","email":"dabarren@gmail.com","username":"egonzalez90"},"state":"REVIEWER"},{"updated":"2019-09-17 08:52:10.000000000","updated_by":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},"reviewer":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},"state":"REVIEWER"},{"updated":"2019-09-18 08:39:30.000000000","updated_by":{"_account_id":23181,"name":"RDO Third Party CI","email":"dmsimard+rdothirdparty@redhat.com","username":"rdothirdparty","tags":["SERVICE_USER"]},"reviewer":{"_account_id":23181,"name":"RDO Third Party CI","email":"dmsimard+rdothirdparty@redhat.com","username":"rdothirdparty","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2019-09-18 11:43:20.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"}],"messages":[{"id":"ce87f8e1dc539b820cf8a3cbeb79fc4740bab724","author":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"date":"2019-09-12 14:40:21.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"66e4bde3470c2890eff71c44f3bcbb5f4c2d71af","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-09-12 16:17:38.000000000","message":"Patch Set 1: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttp://docs.openstack.org/infra/manual/developers.html#automated-testing\n\n\n- kolla-build-centos-binary https://zuul.opendev.org/t/openstack/build/c9c2e437c53140439c4359b757c2e2e3 : SUCCESS in 39m 11s (non-voting)\n- kolla-build-centos-source https://zuul.opendev.org/t/openstack/build/a0770bd58b6842df98dc5a46a69fcd07 : SUCCESS in 45m 00s\n- kolla-ansible-centos-source https://zuul.opendev.org/t/openstack/build/74811da416a640f896400e9608787806 : SUCCESS in 1h 20m 26s\n- kolla-ansible-centos-binary https://zuul.opendev.org/t/openstack/build/3a816a1871ac47f09c063ed3bac4c310 : SUCCESS in 1h 12m 58s (non-voting)\n- kolla-ansible-centos-source-upgrade https://zuul.opendev.org/t/openstack/build/8caa5e8e48ea4c05ad9b145c888368c1 : SUCCESS in 1h 31m 25s\n- tripleo-build-containers-centos-7 https://zuul.opendev.org/t/openstack/build/71bbd26ab3894668b92646d7b89e7ec2 : SUCCESS in 43m 31s\n- kolla-build-debian-source https://zuul.opendev.org/t/openstack/build/1c3b6774fab24574872b124b80b9238e : SUCCESS in 51m 56s (non-voting)\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/2dbf247829654d418d583a6c48848d2d : SUCCESS in 4m 01s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/0d47c2d6cb694a30ada1a2407b342f07 : SUCCESS in 5m 03s\n- openstack-tox-py27 https://zuul.opendev.org/t/openstack/build/40e2125a1eed434a87f324708bc55997 : SUCCESS in 2m 54s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/bf9cddcb7e754a009f6361c610dda652 : SUCCESS in 4m 43s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/ef20009ab0f6472898b9b55cc6530dbb : SUCCESS in 5m 09s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/14037f7040c3466daced71eb75f3a1c2 : SUCCESS in 3m 45s\n- kolla-build-ubuntu-binary https://zuul.opendev.org/t/openstack/build/f2eb9e98935043d988f4f3fa5a20799a : FAILURE in 24m 22s (non-voting)\n- kolla-build-ubuntu-source https://zuul.opendev.org/t/openstack/build/10ff6f490f484360bdd678f2e43077fa : FAILURE in 29m 49s\n- kolla-ansible-ubuntu-source https://zuul.opendev.org/t/openstack/build/9691357f8fa84a1b98aa7ca5ae2b18ea : SUCCESS in 55m 35s\n- kolla-ansible-ubuntu-binary https://zuul.opendev.org/t/openstack/build/871fda0a6f074f008fca6b443b6e6eff : SUCCESS in 1h 01m 20s (non-voting)\n- kolla-ansible-ubuntu-source-upgrade https://zuul.opendev.org/t/openstack/build/6f60085890dc491baf87197d75189bb3 : SUCCESS in 1h 13m 06s","accounts_in_message":[],"_revision_number":1},{"id":"af304c5fbd6d1a4e2a4e73e1067bf534a9df4ba6","author":{"_account_id":23181,"name":"RDO Third Party CI","email":"dmsimard+rdothirdparty@redhat.com","username":"rdothirdparty","tags":["SERVICE_USER"]},"date":"2019-09-12 16:21:14.000000000","message":"Patch Set 1:\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttp://docs.openstack.org/infra/manual/developers.html#automated-testing\n\n\n- tripleo-build-containers-rhel-8-master finger://ze02.softwarefactory-project.io:7979/2884b22295b64e8694fff96eeb356ede : POST_FAILURE in 1h 38m 56s","accounts_in_message":[],"_revision_number":1},{"id":"4e71de8cb189d2a9dc848e8f2bde0a172f63b001","author":{"_account_id":19316,"name":"Eduardo Gonzalez","email":"dabarren@gmail.com","username":"egonzalez90"},"date":"2019-09-17 06:56:42.000000000","message":"Patch Set 1:\n\nrecheck","accounts_in_message":[],"_revision_number":1},{"id":"dc29e7460c53def7984af91e2785d9372597044b","author":{"_account_id":19316,"name":"Eduardo Gonzalez","email":"dabarren@gmail.com","username":"egonzalez90"},"date":"2019-09-17 06:57:04.000000000","message":"Patch Set 1: Code-Review+2","accounts_in_message":[],"_revision_number":1},{"id":"79232f7a65376841bc9d1915e6ce3cd6a02da8fb","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-09-17 08:49:20.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\n\n- kolla-build-centos-binary https://zuul.opendev.org/t/openstack/build/1232cb12a17649d9b4b5c1aff80569a5 : SUCCESS in 47m 52s (non-voting)\n- kolla-build-centos-source https://zuul.opendev.org/t/openstack/build/925cf122ff9040c496701de5094f3227 : SUCCESS in 57m 27s\n- kolla-ansible-centos-source https://zuul.opendev.org/t/openstack/build/9a8cada5e38a4aa19e4ff6bde6000f86 : SUCCESS in 1h 18m 52s\n- kolla-ansible-centos-binary https://zuul.opendev.org/t/openstack/build/33c81a15448b48849c18a7d618f9af1e : SUCCESS in 1h 14m 18s (non-voting)\n- kolla-ansible-centos-source-upgrade https://zuul.opendev.org/t/openstack/build/a7fb37f9427e421d8f87ea90eefb5945 : SUCCESS in 1h 25m 19s\n- tripleo-build-containers-centos-7 https://zuul.opendev.org/t/openstack/build/acc00fd887a24c72bc6a012e1ce82e6c : SUCCESS in 50m 34s\n- kolla-build-debian-source https://zuul.opendev.org/t/openstack/build/524cf480fd9f4d42930316cd3360a1cf : SUCCESS in 28m 40s (non-voting)\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/0fb1dbbb80f240e796aa91e3c201cc0f : SUCCESS in 3m 53s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/f63086f23f444989a287c59ec14211e6 : SUCCESS in 5m 33s\n- openstack-tox-py27 https://zuul.opendev.org/t/openstack/build/00125d0b784e427c8304d2448e27ff59 : SUCCESS in 4m 08s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/82e7da08d8b440ee9d477a4bb03c8ea4 : SUCCESS in 4m 11s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/dca9c69c4fe94f88994c6e9f007a0a37 : SUCCESS in 3m 44s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/bb66d3da3cff4a5eac91652b20a470b2 : SUCCESS in 4m 19s\n- kolla-build-ubuntu-binary https://zuul.opendev.org/t/openstack/build/8770fc25a73a42faab0c0bfd433d2bb7 : SUCCESS in 23m 22s (non-voting)\n- kolla-build-ubuntu-source https://zuul.opendev.org/t/openstack/build/7bd6e537594d4e98889ac4be8244f456 : SUCCESS in 28m 45s\n- kolla-ansible-ubuntu-source https://zuul.opendev.org/t/openstack/build/a8a3d75187474735b6289b19c5b8f63c : SUCCESS in 49m 06s\n- kolla-ansible-ubuntu-binary https://zuul.opendev.org/t/openstack/build/2d1ac8a7d2904218a9397288dbfcfb43 : SUCCESS in 51m 16s (non-voting)\n- kolla-ansible-ubuntu-source-upgrade https://zuul.opendev.org/t/openstack/build/f27d4fa3ef7444e49286362cdc91ec7e : SUCCESS in 1h 00m 25s","accounts_in_message":[],"_revision_number":1},{"id":"e88b86cf9e5a3aa93d54c7677e4a0f131eb079f2","author":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},"date":"2019-09-17 08:52:10.000000000","message":"Patch Set 1: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":1},{"id":"3198536698f333bde5906268bfe708aee49f9d7a","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-09-17 08:52:25.000000000","message":"Patch Set 1: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":1},{"id":"f6f9fabd9fb64c9df6548437b17b83b248181f50","author":{"_account_id":23181,"name":"RDO Third Party CI","email":"dmsimard+rdothirdparty@redhat.com","username":"rdothirdparty","tags":["SERVICE_USER"]},"date":"2019-09-17 09:02:54.000000000","message":"Patch Set 1:\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttp://docs.openstack.org/infra/manual/developers.html#automated-testing\n\n\n- tripleo-build-containers-rhel-8-master http://logs.rdoproject.org/68/681768/1/openstack-check/tripleo-build-containers-rhel-8-master/ef48464/ : TIMED_OUT in 2h 05m 21s","accounts_in_message":[],"_revision_number":1},{"id":"bbec7c5f398294c32364f8972f475d8172a44c8e","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-09-17 10:26:41.000000000","message":"Patch Set 1: Verified-2\n\nBuild failed (gate pipeline).  For information on how to proceed, see\nhttp://docs.openstack.org/infra/manual/developers.html#automated-testing\n\n\n- kolla-build-centos-source https://zuul.opendev.org/t/openstack/build/d9f813cd944d4334b4aac3a84386be03 : SUCCESS in 59m 50s\n- kolla-ansible-centos-source https://zuul.opendev.org/t/openstack/build/7145ebe6c4f148709bb4cacf2ce1c15a : SUCCESS in 1h 28m 39s\n- kolla-ansible-centos-source-upgrade https://zuul.opendev.org/t/openstack/build/b9c645152b7644a88269307edccd5aab : FAILURE in 53m 03s\n- tripleo-build-containers-centos-7 https://zuul.opendev.org/t/openstack/build/78713c01bc564d2ab0bb54ccc3415e50 : SUCCESS in 48m 11s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/ddcf95bf012d4a1f8cf677efe8e51e2e : SUCCESS in 4m 09s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/ab7b97cc12a84dfdaae3318ff3f529d4 : SUCCESS in 3m 37s\n- openstack-tox-py27 https://zuul.opendev.org/t/openstack/build/b5a688a8e2b74befb0ea653bec3ef5b3 : SUCCESS in 3m 40s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/9789e0a976de414e8939e24869efbb24 : SUCCESS in 3m 18s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/0d78a55b6abf4327a6e85166f2f54d92 : SUCCESS in 3m 16s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/a126397b33ca47c690363ade768eb327 : SUCCESS in 2m 54s\n- kolla-build-ubuntu-source https://zuul.opendev.org/t/openstack/build/e4d103e5400b449896635f309303ad8a : RETRY_LIMIT in 2m 17s\n- kolla-ansible-ubuntu-source https://zuul.opendev.org/t/openstack/build/0f378f01d57f435e918d25b67536eecc : RETRY_LIMIT in 1m 36s\n- kolla-ansible-ubuntu-source-upgrade https://zuul.opendev.org/t/openstack/build/8df3a3ed89d54da6a29d7a35a8dbde48 : SUCCESS in 1h 10m 49s","accounts_in_message":[],"_revision_number":1},{"id":"900c38c605f945b282bf2426b5869c39a5547b12","author":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"date":"2019-09-18 07:59:42.000000000","message":"Patch Set 1:\n\nrecheck","accounts_in_message":[],"_revision_number":1},{"id":"acfec17a3e904cc152d19f936db1250fc5a01dc6","author":{"_account_id":23181,"name":"RDO Third Party CI","email":"dmsimard+rdothirdparty@redhat.com","username":"rdothirdparty","tags":["SERVICE_USER"]},"date":"2019-09-18 08:39:30.000000000","message":"Patch Set 1:\n\nBuild succeeded (check pipeline).\n\n- tripleo-build-containers-rhel-8-master http://logs.rdoproject.org/68/681768/1/openstack-check/tripleo-build-containers-rhel-8-master/f9d0b56/ : SUCCESS in 39m 09s","accounts_in_message":[],"_revision_number":1},{"id":"1c09e23f5c1015ba612d333977eec44104808a78","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-09-18 10:00:36.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\n\n- kolla-build-centos-binary https://zuul.opendev.org/t/openstack/build/ca2e3474458a4f8b92929955099f50ad : SUCCESS in 52m 42s (non-voting)\n- kolla-build-centos-source https://zuul.opendev.org/t/openstack/build/25b2e2660803454a86f155c334afb355 : SUCCESS in 58m 17s\n- kolla-ansible-centos-source https://zuul.opendev.org/t/openstack/build/6f6c7303af924ec7b139c124a0c48258 : SUCCESS in 1h 30m 51s\n- kolla-ansible-centos-binary https://zuul.opendev.org/t/openstack/build/972cea1de7cd4fe5a2ca457ddc5e3207 : SUCCESS in 1h 22m 22s (non-voting)\n- kolla-ansible-centos-source-upgrade https://zuul.opendev.org/t/openstack/build/278fa89cfdd245b28f733afd59401327 : SUCCESS in 1h 47m 09s\n- tripleo-build-containers-centos-7 https://zuul.opendev.org/t/openstack/build/ca2bd7b0872f4ef0aaab03830f3b2625 : SUCCESS in 59m 37s\n- kolla-build-debian-source https://zuul.opendev.org/t/openstack/build/6886e50c61e3446399415b88938ba7ca : SUCCESS in 31m 47s (non-voting)\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/45f77271fd6f4af480b56253a1fa258a : SUCCESS in 4m 23s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/744275dedb81484f978c707a54b873b1 : SUCCESS in 6m 30s\n- openstack-tox-py27 https://zuul.opendev.org/t/openstack/build/67800608fc074ea591fbd81303dcfb07 : SUCCESS in 4m 07s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/f1a1565514724a2f82552bc09dfe04d9 : SUCCESS in 5m 15s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/30bec992610348ccbc6a251418311d45 : SUCCESS in 3m 43s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/62beaba520814422abf191e98c29a2e8 : SUCCESS in 3m 49s\n- kolla-build-ubuntu-binary https://zuul.opendev.org/t/openstack/build/ea7e44feb67c42739d5efb2cbc2249f2 : SUCCESS in 26m 06s (non-voting)\n- kolla-build-ubuntu-source https://zuul.opendev.org/t/openstack/build/944d324015b141d6b09f5cfe0ac1d753 : SUCCESS in 34m 10s\n- kolla-ansible-ubuntu-source https://zuul.opendev.org/t/openstack/build/e4d759c1273a4e5683fea0b4e79ebbfd : SUCCESS in 50m 58s\n- kolla-ansible-ubuntu-binary https://zuul.opendev.org/t/openstack/build/a4db45cf111c4dd9937cc8bd627b053e : SUCCESS in 59m 22s (non-voting)\n- kolla-ansible-ubuntu-source-upgrade https://zuul.opendev.org/t/openstack/build/3a927271520c4715a5de3cfbaaf0b2fe : SUCCESS in 1h 07m 32s","accounts_in_message":[],"_revision_number":1},{"id":"17e3c8ba7bcdea16e7436e594b744bc53b8eda43","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-09-18 10:01:01.000000000","message":"Patch Set 1: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":1},{"id":"46018ccd15032ad1f8e92671c8bc7bb97f0d4735","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-09-18 11:43:20.000000000","message":"Change has been successfully merged by Zuul","accounts_in_message":[],"_revision_number":1},{"id":"fb370b2a6ffcc973115e1d727c3863f3fe2a0df8","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-09-18 11:43:20.000000000","message":"Patch Set 1: Verified+2\n\nBuild succeeded (gate pipeline).\n\n- kolla-build-centos-source https://zuul.opendev.org/t/openstack/build/3dfb17159a3f498d81e83306e7325077 : SUCCESS in 1h 06m 15s\n- kolla-ansible-centos-source https://zuul.opendev.org/t/openstack/build/335ceab6790443caac772420ea5354a3 : SUCCESS in 1h 31m 30s\n- kolla-ansible-centos-source-upgrade https://zuul.opendev.org/t/openstack/build/a3ab117b5fe544c1a15b4ec1b15ba00e : SUCCESS in 1h 30m 11s\n- tripleo-build-containers-centos-7 https://zuul.opendev.org/t/openstack/build/238ebefdee9240a48c7a376c52f6d23e : SUCCESS in 1h 00m 03s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/4795fc3eea9d41659c66dad5294085c0 : SUCCESS in 4m 14s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/f004f201ec16498d980c99e72fab8d62 : SUCCESS in 4m 40s\n- openstack-tox-py27 https://zuul.opendev.org/t/openstack/build/1c725881bf8243b39a75e21b0c5044b7 : SUCCESS in 4m 29s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/6ae052ede40e41a1bbec503ff4287231 : SUCCESS in 4m 23s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/ce0419c4b60244c7b21883b959973d95 : SUCCESS in 4m 25s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/ec8cc6366f7b48d0be71042325473a9e : SUCCESS in 3m 15s\n- kolla-build-ubuntu-source https://zuul.opendev.org/t/openstack/build/ffe22d3b392b41d49e72da9e0e82fd6b : SUCCESS in 29m 28s\n- kolla-ansible-ubuntu-source https://zuul.opendev.org/t/openstack/build/37d9f49e8751410aac4e2662d59f3c99 : SUCCESS in 58m 11s\n- kolla-ansible-ubuntu-source-upgrade https://zuul.opendev.org/t/openstack/build/8a89c7ec3c624249a3b1d32d8393cec0 : SUCCESS in 1h 11m 56s","accounts_in_message":[],"_revision_number":1},{"id":"ffd8c6e35e164a937a4a03ae60c08a363019ad56","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-09-18 11:44:57.000000000","message":"Patch Set 1:\n\nBuild succeeded (promote pipeline).\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/c42bc0f23d3e4d0295756ce570f33f80 : SUCCESS in 1m 12s","accounts_in_message":[],"_revision_number":1}],"current_revision_number":1,"current_revision":"2da7f76ee0e43f0965b577fab499d60f874776df","revisions":{"2da7f76ee0e43f0965b577fab499d60f874776df":{"kind":"REWORK","_number":1,"created":"2019-09-12 14:40:21.000000000","uploader":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"ref":"refs/changes/68/681768/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/kolla","ref":"refs/changes/68/681768/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/kolla refs/changes/68/681768/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/kolla refs/changes/68/681768/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/kolla refs/changes/68/681768/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/kolla refs/changes/68/681768/1"}}},"commit":{"parents":[{"commit":"532af2efc8893cfd23f8d2beadd32a4e3463ae85","subject":"Do not fan out DB sync to all cells","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/532af2efc8893cfd23f8d2beadd32a4e3463ae85"}]}],"author":{"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","date":"2019-09-12 14:39:45.000000000","tz":0},"committer":{"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","date":"2019-09-12 14:40:20.000000000","tz":0},"subject":"Update skydive to v0.25.0","message":"Update skydive to v0.25.0\n\nChange-Id: Id9db205c1af7607fde12dd8be9e0e25139597ccd\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/2da7f76ee0e43f0965b577fab499d60f874776df"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/2da7f76ee0e43f0965b577fab499d60f874776df"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[],"submit_requirements":[]}
