)]}'
{"id":"openstack%2Fkolla~682068","triplet_id":"openstack%2Fkolla~master~I6b8f52a1f2227ecb74e3ae59ef1e67d5b5ce1886","project":"openstack/kolla","branch":"master","hashtags":[],"change_id":"I6b8f52a1f2227ecb74e3ae59ef1e67d5b5ce1886","subject":"ppc64le: add UNBUILDABLE_IMAGES entries","status":"MERGED","created":"2019-09-13 14:34:19.000000000","updated":"2019-09-19 21:06:46.000000000","submitted":"2019-09-19 21:04:47.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":8,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"682068-1568927087889-d2cb535d","meta_rev_id":"d2c7a93a578113ff8329b31aa0b41035e01c3c45","_number":682068,"virtual_id_number":682068,"owner":{"_account_id":24072,"name":"Marcin Juszkiewicz","email":"mjuszkiewicz@redhat.com","username":"hrw"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"date":"2019-09-19 17:42:08.000000000","_account_id":24072,"name":"Marcin Juszkiewicz","email":"mjuszkiewicz@redhat.com","username":"hrw"},{"value":0,"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},{"value":0,"_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"},{"value":2,"date":"2019-09-19 21:04:47.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"date":"2019-09-19 17:57:09.000000000","_account_id":23181,"name":"RDO Third Party CI","email":"dmsimard+rdothirdparty@redhat.com","username":"rdothirdparty","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},"all":[{"value":0,"_account_id":24072,"name":"Marcin Juszkiewicz","email":"mjuszkiewicz@redhat.com","username":"hrw"},{"value":2,"date":"2019-09-19 09:30:12.000000000","_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},{"value":2,"date":"2019-09-19 07:31:53.000000000","_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":23181,"name":"RDO Third Party CI","email":"dmsimard+rdothirdparty@redhat.com","username":"rdothirdparty","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},"all":[{"value":0,"_account_id":24072,"name":"Marcin Juszkiewicz","email":"mjuszkiewicz@redhat.com","username":"hrw"},{"value":1,"date":"2019-09-19 09:30:12.000000000","_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},{"value":0,"_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":23181,"name":"RDO Third Party CI","email":"dmsimard+rdothirdparty@redhat.com","username":"rdothirdparty","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Backport-Candidate":{"all":[{"value":0,"_account_id":24072,"name":"Marcin Juszkiewicz","email":"mjuszkiewicz@redhat.com","username":"hrw"},{"value":0,"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},{"value":0,"_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":23181,"name":"RDO Third Party CI","email":"dmsimard+rdothirdparty@redhat.com","username":"rdothirdparty","tags":["SERVICE_USER"]}],"values":{"-1":"Do Not Backport"," 0":"Backport Review Needed","+1":"Should Backport"},"description":"","default_value":0,"optional":true},"Review-Priority":{"all":[{"value":0,"_account_id":24072,"name":"Marcin Juszkiewicz","email":"mjuszkiewicz@redhat.com","username":"hrw"},{"value":0,"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},{"value":0,"_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":23181,"name":"RDO Third Party CI","email":"dmsimard+rdothirdparty@redhat.com","username":"rdothirdparty","tags":["SERVICE_USER"]}],"values":{"-1":"Branch Freeze"," 0":"No Priority","+1":"Important Change","+2":"Gate Blocker Fix / Urgent Change"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":23181,"name":"RDO Third Party CI","email":"dmsimard+rdothirdparty@redhat.com","username":"rdothirdparty","tags":["SERVICE_USER"]},{"_account_id":24072,"name":"Marcin Juszkiewicz","email":"mjuszkiewicz@redhat.com","username":"hrw"},{"_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2019-09-19 07:31:53.000000000","updated_by":{"_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"},"reviewer":{"_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"},"state":"REVIEWER"},{"updated":"2019-09-19 09:30:12.000000000","updated_by":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},"reviewer":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},"state":"REVIEWER"},{"updated":"2019-09-19 17:57:09.000000000","updated_by":{"_account_id":23181,"name":"RDO Third Party CI","email":"dmsimard+rdothirdparty@redhat.com","username":"rdothirdparty","tags":["SERVICE_USER"]},"reviewer":{"_account_id":23181,"name":"RDO Third Party CI","email":"dmsimard+rdothirdparty@redhat.com","username":"rdothirdparty","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2019-09-19 21:04:47.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"}],"messages":[{"id":"02b8d19b8ebccd47edbb49d3fcaedd7917fb5ccc","author":{"_account_id":24072,"name":"Marcin Juszkiewicz","email":"mjuszkiewicz@redhat.com","username":"hrw"},"date":"2019-09-13 14:34:19.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"5a9eda0ea08a9c286cdbab0fd34ad4a531672303","author":{"_account_id":23181,"name":"RDO Third Party CI","email":"dmsimard+rdothirdparty@redhat.com","username":"rdothirdparty","tags":["SERVICE_USER"]},"date":"2019-09-13 16:00:16.000000000","message":"Patch Set 1:\n\nBuild succeeded (check pipeline).\n\n- tripleo-build-containers-rhel-8-master http://logs.rdoproject.org/68/682068/1/openstack-check/tripleo-build-containers-rhel-8-master/1e711e8/ : SUCCESS in 38m 08s","accounts_in_message":[],"_revision_number":1},{"id":"716f1ae21195dca7a8ae4ff60ca520f41b54faaf","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-09-13 16:12:35.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\n\n- kolla-build-centos-binary https://zuul.opendev.org/t/openstack/build/2df6265d590d4982bc69ce4c7fd42140 : SUCCESS in 43m 06s (non-voting)\n- kolla-build-centos-source https://zuul.opendev.org/t/openstack/build/49789196711c42959f750d145171b2c9 : SUCCESS in 54m 28s\n- kolla-ansible-centos-source https://zuul.opendev.org/t/openstack/build/561b0fcf925440b1a193f59529e2f5d8 : SUCCESS in 1h 22m 12s\n- kolla-ansible-centos-binary https://zuul.opendev.org/t/openstack/build/999984cc766f4e60ac445ec5063073cf : SUCCESS in 1h 11m 57s (non-voting)\n- kolla-ansible-centos-source-upgrade https://zuul.opendev.org/t/openstack/build/89f7ad9e3c784acabda5c85bd8a5fc30 : SUCCESS in 1h 27m 33s\n- tripleo-build-containers-centos-7 https://zuul.opendev.org/t/openstack/build/64944856f2ac40ecb239524ccbee762a : SUCCESS in 52m 45s\n- kolla-build-debian-source https://zuul.opendev.org/t/openstack/build/203aa4faf7d5449da44a5cb9c8519b94 : RETRY_LIMIT in 10m 00s (non-voting)\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/af023fd57d9546ebabf024fccd5a406e : SUCCESS in 4m 10s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/5bcbab8c718f4463bfaae39393187ef5 : SUCCESS in 4m 48s\n- openstack-tox-py27 https://zuul.opendev.org/t/openstack/build/e6ce1d4c0aa448d79662c557e954ad2a : SUCCESS in 3m 53s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/3e4c77ac216e4e12b147a64811acbbb3 : SUCCESS in 3m 27s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/54e3534b19de4f29be94872837e04819 : SUCCESS in 3m 41s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/3444a448fc8549e7935f5c67a6c3156c : SUCCESS in 4m 12s\n- kolla-build-ubuntu-binary https://zuul.opendev.org/t/openstack/build/1d410dd14880410e8b6ed7ab92ebd4cc : RETRY_LIMIT in 1m 26s (non-voting)\n- kolla-build-ubuntu-source https://zuul.opendev.org/t/openstack/build/0b5ee00faa2d4081a31cd9320386f414 : SUCCESS in 32m 19s\n- kolla-ansible-ubuntu-source https://zuul.opendev.org/t/openstack/build/15458b41d556410ba6d00f4a493cf2e7 : SUCCESS in 54m 27s\n- kolla-ansible-ubuntu-binary https://zuul.opendev.org/t/openstack/build/4877ef5964274e42b2b1e57e2394141e : SUCCESS in 52m 21s (non-voting)\n- kolla-ansible-ubuntu-source-upgrade https://zuul.opendev.org/t/openstack/build/f6d9826b747d4edc8398cac572972330 : SUCCESS in 57m 30s","accounts_in_message":[],"_revision_number":1},{"id":"95734f9f92cea055fc26f97cd3cdb75acca76a74","author":{"_account_id":24072,"name":"Marcin Juszkiewicz","email":"mjuszkiewicz@redhat.com","username":"hrw"},"date":"2019-09-14 16:03:59.000000000","message":"Uploaded patch set 2.","accounts_in_message":[],"_revision_number":2},{"id":"e6be41626fb467dedf2b00313975861cf2efa359","author":{"_account_id":23181,"name":"RDO Third Party CI","email":"dmsimard+rdothirdparty@redhat.com","username":"rdothirdparty","tags":["SERVICE_USER"]},"date":"2019-09-14 16:43:00.000000000","message":"Patch Set 2:\n\nBuild succeeded (check pipeline).\n\n- tripleo-build-containers-rhel-8-master http://logs.rdoproject.org/68/682068/2/openstack-check/tripleo-build-containers-rhel-8-master/c33e5e2/ : SUCCESS in 38m 14s","accounts_in_message":[],"_revision_number":2},{"id":"b012524a1aa53d9104921c5adabf7a9cbd607d48","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-09-14 17:23:17.000000000","message":"Patch Set 2: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttp://docs.openstack.org/infra/manual/developers.html#automated-testing\n\n\n- kolla-build-centos-binary https://zuul.opendev.org/t/openstack/build/6f93ab7c847248b890a476caa9c64577 : FAILURE in 27m 18s (non-voting)\n- kolla-build-centos-source https://zuul.opendev.org/t/openstack/build/10ece34ab0b84185a56fb63a7957b8d4 : SUCCESS in 48m 04s\n- kolla-ansible-centos-source https://zuul.opendev.org/t/openstack/build/423201d495774213a7fdaf898a2534d6 : FAILURE in 31m 39s\n- kolla-ansible-centos-binary https://zuul.opendev.org/t/openstack/build/15f78dec13404d80bec3f6b630689447 : FAILURE in 31m 10s (non-voting)\n- kolla-ansible-centos-source-upgrade https://zuul.opendev.org/t/openstack/build/3e683d61a0b44319b8c247d177a91d5b : FAILURE in 31m 34s\n- tripleo-build-containers-centos-7 https://zuul.opendev.org/t/openstack/build/cf77762f32b143618f32d5951f2d8875 : FAILURE in 7m 39s\n- kolla-build-debian-source https://zuul.opendev.org/t/openstack/build/5e61276fc15d4b3c9ded0f6a8f431b51 : SUCCESS in 26m 59s (non-voting)\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/ebcc6f7aa2834645bfa60650488fad80 : SUCCESS in 3m 30s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/41d768c3478341fca41d348a02ae5a46 : SUCCESS in 3m 27s\n- openstack-tox-py27 https://zuul.opendev.org/t/openstack/build/fcf5f43e4cf448188432a9f7f2d8d715 : SUCCESS in 3m 22s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/71cb7a639d4c4408a9bcc0e9e49f2e9b : SUCCESS in 2m 50s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/97c3bc5136f6488aa307beb664a82b6c : SUCCESS in 2m 56s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/e60c9a927ad54316bb4ba331fc4cada1 : SUCCESS in 2m 42s\n- kolla-build-ubuntu-binary https://zuul.opendev.org/t/openstack/build/0fab4f23037a4bb18291448d3e783d27 : SUCCESS in 28m 17s (non-voting)\n- kolla-build-ubuntu-source https://zuul.opendev.org/t/openstack/build/97b26ba3633b424289a719c031edb831 : SUCCESS in 34m 35s\n- kolla-ansible-ubuntu-source https://zuul.opendev.org/t/openstack/build/57cbc3e60d544bd8811002b64e3a67b2 : SUCCESS in 55m 08s\n- kolla-ansible-ubuntu-binary https://zuul.opendev.org/t/openstack/build/c930a4debb794712be50d0fe10683a45 : SUCCESS in 56m 27s (non-voting)\n- kolla-ansible-ubuntu-source-upgrade https://zuul.opendev.org/t/openstack/build/01374a8fa95e4f07a2bc003ad30fcc72 : SUCCESS in 1h 14m 31s","accounts_in_message":[],"_revision_number":2},{"id":"6aafc7f0e1880eabf7ac7659109db2856b07b16a","author":{"_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"},"date":"2019-09-16 08:25:47.000000000","message":"Patch Set 2:\n\n(3 comments)","accounts_in_message":[],"_revision_number":2},{"id":"e68a6ac44c7e47fb307bb085189e9ffef51779ca","author":{"_account_id":24072,"name":"Marcin Juszkiewicz","email":"mjuszkiewicz@redhat.com","username":"hrw"},"date":"2019-09-16 08:37:58.000000000","message":"Patch Set 2:\n\n(2 comments)","accounts_in_message":[],"_revision_number":2},{"id":"ac90d73a044b206a24ea6b278eedfdfef894c8e7","author":{"_account_id":24072,"name":"Marcin Juszkiewicz","email":"mjuszkiewicz@redhat.com","username":"hrw"},"date":"2019-09-16 08:41:25.000000000","message":"Patch Set 2:\n\nrecheck","accounts_in_message":[],"_revision_number":2},{"id":"c6bd7aa39576b9be607a24f2da8005c655857daa","author":{"_account_id":23181,"name":"RDO Third Party CI","email":"dmsimard+rdothirdparty@redhat.com","username":"rdothirdparty","tags":["SERVICE_USER"]},"date":"2019-09-16 09:18:27.000000000","message":"Patch Set 2:\n\nBuild succeeded (check pipeline).\n\n- tripleo-build-containers-rhel-8-master http://logs.rdoproject.org/68/682068/2/openstack-check/tripleo-build-containers-rhel-8-master/3231a9c/ : SUCCESS in 36m 20s","accounts_in_message":[],"_revision_number":2},{"id":"7cc2220b1dd49a7572fca100715a49172db9a1f1","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-09-16 10:38:59.000000000","message":"Patch Set 2:\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttp://docs.openstack.org/infra/manual/developers.html#automated-testing\n\n\n- kolla-build-centos-binary https://zuul.opendev.org/t/openstack/build/902ac169264442a19c32ea64710ac22e : SUCCESS in 40m 50s (non-voting)\n- kolla-build-centos-source https://zuul.opendev.org/t/openstack/build/7532356aa0334dcda5fe08bf299d6802 : SUCCESS in 48m 22s\n- kolla-ansible-centos-source https://zuul.opendev.org/t/openstack/build/c968aaa897234037b0a0eb3cbc1bec2a : FAILURE in 31m 30s\n- kolla-ansible-centos-binary https://zuul.opendev.org/t/openstack/build/51fd8b6c23754019aaaaa366b31f3a0f : SUCCESS in 1h 19m 36s (non-voting)\n- kolla-ansible-centos-source-upgrade https://zuul.opendev.org/t/openstack/build/3ef248ddb5254084b348f5f2c002bb5b : SUCCESS in 1h 14m 17s\n- tripleo-build-containers-centos-7 https://zuul.opendev.org/t/openstack/build/97ba5166f1024562b8351a9281d474cf : SUCCESS in 53m 32s\n- kolla-build-debian-source https://zuul.opendev.org/t/openstack/build/07711883820b497bb4a02d5b2f2da541 : SUCCESS in 24m 29s (non-voting)\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/edabf731f5cb4927b69bd445dbbfe047 : SUCCESS in 3m 17s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/5a9399663e194bd1b40ba23ea3437543 : SUCCESS in 4m 09s\n- openstack-tox-py27 https://zuul.opendev.org/t/openstack/build/4fdb5c1d19924eed9b0a08cec0299e56 : SUCCESS in 4m 22s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/3b839cf66e764f17b5a087d0a9a48a7d : SUCCESS in 3m 07s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/4113a32222d0439995e26a227d360508 : SUCCESS in 3m 27s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/0715c9cf1cea44748032cfc016120931 : SUCCESS in 4m 29s\n- kolla-build-ubuntu-binary https://zuul.opendev.org/t/openstack/build/a7139d708a164cb9b9c0c3a495dddd65 : SUCCESS in 29m 18s (non-voting)\n- kolla-build-ubuntu-source https://zuul.opendev.org/t/openstack/build/21e03a66c40c4c7aa287c140b16ee2b1 : SUCCESS in 32m 39s\n- kolla-ansible-ubuntu-source https://zuul.opendev.org/t/openstack/build/97cb8ea11e4248a896900c4ba5917eb1 : SUCCESS in 55m 37s\n- kolla-ansible-ubuntu-binary https://zuul.opendev.org/t/openstack/build/c8473598c8d0496987a55238d214ba55 : SUCCESS in 48m 55s (non-voting)\n- kolla-ansible-ubuntu-source-upgrade https://zuul.opendev.org/t/openstack/build/4885e1f13c9e41e28557e85f31efe95b : SUCCESS in 59m 22s","accounts_in_message":[],"_revision_number":2},{"id":"b0545b735f27711757e93b0bc49303977ed5119c","author":{"_account_id":24072,"name":"Marcin Juszkiewicz","email":"mjuszkiewicz@redhat.com","username":"hrw"},"date":"2019-09-17 14:27:04.000000000","message":"Patch Set 2:\n\nrecheck","accounts_in_message":[],"_revision_number":2},{"id":"f6f068712809ebc3314357b0def3f3209790c98b","author":{"_account_id":23181,"name":"RDO Third Party CI","email":"dmsimard+rdothirdparty@redhat.com","username":"rdothirdparty","tags":["SERVICE_USER"]},"date":"2019-09-17 15:04:37.000000000","message":"Patch Set 2:\n\nBuild succeeded (check pipeline).\n\n- tripleo-build-containers-rhel-8-master http://logs.rdoproject.org/68/682068/2/openstack-check/tripleo-build-containers-rhel-8-master/762a277/ : SUCCESS in 36m 00s","accounts_in_message":[],"_revision_number":2},{"id":"496093727dc7725a96a53bf14b4670631464a32f","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-09-17 15:59:04.000000000","message":"Patch Set 2: Verified+1\n\nBuild succeeded (check pipeline).\n\n- kolla-build-centos-binary https://zuul.opendev.org/t/openstack/build/aaeda724df8a4f3d8f40536b0ec52dbc : SUCCESS in 49m 22s (non-voting)\n- kolla-build-centos-source https://zuul.opendev.org/t/openstack/build/c17f1aa031cd4475a584512801f3dbf7 : SUCCESS in 54m 15s\n- kolla-ansible-centos-source https://zuul.opendev.org/t/openstack/build/b876dceb7a9243b58f2634e27404b528 : SUCCESS in 1h 25m 33s\n- kolla-ansible-centos-binary https://zuul.opendev.org/t/openstack/build/cfcfb53f6ab642339d3c7b952966e9dc : SUCCESS in 1h 17m 15s (non-voting)\n- kolla-ansible-centos-source-upgrade https://zuul.opendev.org/t/openstack/build/b713d31f036f4dbb9186b92fc950d537 : SUCCESS in 1h 22m 59s\n- tripleo-build-containers-centos-7 https://zuul.opendev.org/t/openstack/build/e05560fc1a1244bcaf6baf238401a0ef : SUCCESS in 59m 27s\n- kolla-build-debian-source https://zuul.opendev.org/t/openstack/build/37dac613952b420cbad35d1b94bdeb16 : SUCCESS in 21m 46s (non-voting)\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/6d136a26769342369768cb4e42f8c1e2 : SUCCESS in 3m 40s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/62a18832ab5b4283b67ef2811870b345 : SUCCESS in 5m 25s\n- openstack-tox-py27 https://zuul.opendev.org/t/openstack/build/3fba97b860794627bbed3af766f306fc : SUCCESS in 4m 18s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/27b3c59fdd5c48949681164d89819772 : SUCCESS in 3m 53s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/4245eaae964a418cb7327fabf6be02d3 : SUCCESS in 3m 55s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/495fa2dfab174f1ea950a84c98c158b4 : SUCCESS in 2m 35s\n- kolla-build-ubuntu-binary https://zuul.opendev.org/t/openstack/build/607bd95bf8ae43e985b115ef7b7c9559 : SUCCESS in 33m 42s (non-voting)\n- kolla-build-ubuntu-source https://zuul.opendev.org/t/openstack/build/e13a281e610b49fca697b6556b65d5fa : SUCCESS in 34m 11s\n- kolla-ansible-ubuntu-source https://zuul.opendev.org/t/openstack/build/209861189b2a41f18b50e702126ee691 : SUCCESS in 52m 49s\n- kolla-ansible-ubuntu-binary https://zuul.opendev.org/t/openstack/build/723a94f33e4f4fd299710df6d456a7c7 : SUCCESS in 51m 55s (non-voting)\n- kolla-ansible-ubuntu-source-upgrade https://zuul.opendev.org/t/openstack/build/ebbaae7e8425407b94032c6e7302c2c1 : SUCCESS in 1h 00m 55s","accounts_in_message":[],"_revision_number":2},{"id":"dcd76b62f86c2d973ffc82fbf5328f391b016c65","author":{"_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"},"date":"2019-09-19 07:12:44.000000000","message":"Patch Set 2: Code-Review-1\n\n(3 comments)","accounts_in_message":[],"_revision_number":2},{"id":"9634c9a7153fe9d02722e8c698fb71fd711235f1","author":{"_account_id":24072,"name":"Marcin Juszkiewicz","email":"mjuszkiewicz@redhat.com","username":"hrw"},"date":"2019-09-19 07:22:39.000000000","message":"Patch Set 3: Published edit on patch set 2.","accounts_in_message":[],"_revision_number":3},{"id":"e4414802d0d7f9441c7b9aedba29e50d8fa96dd0","author":{"_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"},"date":"2019-09-19 07:31:53.000000000","message":"Patch Set 3: Code-Review+2","accounts_in_message":[],"_revision_number":3},{"id":"b862eaeb8f10c47e6e6324f51408b30949f80973","author":{"_account_id":23181,"name":"RDO Third Party CI","email":"dmsimard+rdothirdparty@redhat.com","username":"rdothirdparty","tags":["SERVICE_USER"]},"date":"2019-09-19 08:03:51.000000000","message":"Patch Set 3:\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttp://docs.openstack.org/infra/manual/developers.html#automated-testing\n\n\n- tripleo-build-containers-rhel-8-master http://logs.rdoproject.org/68/682068/3/openstack-check/tripleo-build-containers-rhel-8-master/a097d9e/ : FAILURE in 10m 07s","accounts_in_message":[],"_revision_number":3},{"id":"d3762497e4768d483d8e2f13365539c0de728980","author":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},"date":"2019-09-19 09:30:12.000000000","message":"Patch Set 3: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":3},{"id":"14cd8f1dbb38d2a92c79902ee3db46ac2f919b86","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-09-19 10:03:28.000000000","message":"Patch Set 3: Verified+1\n\nBuild succeeded (check pipeline).\n\n- kolla-build-centos-binary https://zuul.opendev.org/t/openstack/build/4d1fa9938aa14cf59a179834516cf06f : SUCCESS in 54m 28s (non-voting)\n- kolla-build-centos-source https://zuul.opendev.org/t/openstack/build/7c7629ffa18e4feab1e3b092b7b76941 : SUCCESS in 51m 43s\n- kolla-ansible-centos-source https://zuul.opendev.org/t/openstack/build/7132797a50404d1ea8702ff0ba9f6849 : SUCCESS in 1h 26m 04s\n- kolla-ansible-centos-binary https://zuul.opendev.org/t/openstack/build/494b6c6db0a0468386b1c2c4d3af2eb8 : SUCCESS in 1h 21m 40s (non-voting)\n- kolla-ansible-centos-source-upgrade https://zuul.opendev.org/t/openstack/build/8764c2e77d3b43fd8d376b3e69af3f5b : SUCCESS in 1h 21m 17s\n- tripleo-build-containers-centos-7 https://zuul.opendev.org/t/openstack/build/8a8f1ee4e1b0428495e4d9a9d98c04f9 : SUCCESS in 59m 44s\n- kolla-build-debian-source https://zuul.opendev.org/t/openstack/build/914c517f0c80472d973a690ba78c3de6 : SUCCESS in 29m 15s (non-voting)\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/bd38b04c830f486a95e147bd25d43272 : SUCCESS in 3m 48s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/16bd35819beb433c9543dce8bb9173b6 : SUCCESS in 4m 49s\n- openstack-tox-py27 https://zuul.opendev.org/t/openstack/build/b4b33f578bdb4b49bd447252a7cd5d02 : SUCCESS in 3m 56s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/b9d2b4f56bfd4aeb9fb6d822639729a4 : SUCCESS in 3m 45s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/06ecb01f19ea4c93be4074c17dd3abc5 : SUCCESS in 4m 18s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/5b424406efec4a4084f46f9a14d20a35 : SUCCESS in 4m 30s\n- kolla-build-ubuntu-binary https://zuul.opendev.org/t/openstack/build/d99d24597b944421af1ebebe43b68466 : SUCCESS in 26m 30s (non-voting)\n- kolla-build-ubuntu-source https://zuul.opendev.org/t/openstack/build/936f0d2e798044998d0da08a19559a71 : SUCCESS in 36m 07s\n- kolla-ansible-ubuntu-source https://zuul.opendev.org/t/openstack/build/ddfa16de03594397beae50beac5360d7 : SUCCESS in 57m 15s\n- kolla-ansible-ubuntu-binary https://zuul.opendev.org/t/openstack/build/2306443faef8493d86550fe82c4edf5a : SUCCESS in 55m 43s (non-voting)\n- kolla-ansible-ubuntu-source-upgrade https://zuul.opendev.org/t/openstack/build/1d8b36a9e917466386d74414b00cf094 : SUCCESS in 57m 26s","accounts_in_message":[],"_revision_number":3},{"id":"72071c9aedc8bb40accca122d773a049f098e97a","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-09-19 10:03:44.000000000","message":"Patch Set 3: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":3},{"id":"928a02765bd851394f8c19a844fe5aa91a7b6c97","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-09-19 11:44:18.000000000","message":"Patch Set 3: Verified-2\n\nBuild failed (gate pipeline).  For information on how to proceed, see\nhttp://docs.openstack.org/infra/manual/developers.html#automated-testing\n\n\n- kolla-build-centos-source https://zuul.opendev.org/t/openstack/build/2a2263dc967c40998eeff26aa61b4f20 : FAILURE in 47m 26s\n- kolla-ansible-centos-source https://zuul.opendev.org/t/openstack/build/91006948b32640e6a0947629fe40f547 : SUCCESS in 1h 24m 22s\n- kolla-ansible-centos-source-upgrade https://zuul.opendev.org/t/openstack/build/b3ce6e804f5b405ab05e0b005af4f938 : SUCCESS in 1h 38m 00s\n- tripleo-build-containers-centos-7 https://zuul.opendev.org/t/openstack/build/c199d671541142fc968fa1ab356e74c2 : SUCCESS in 54m 43s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/f6588436c01e47589eca2bbe96d6f98f : SUCCESS in 5m 07s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/67a678911b2b40b5823e540395bf1a6d : SUCCESS in 4m 59s\n- openstack-tox-py27 https://zuul.opendev.org/t/openstack/build/de87d67e1b124a1897c62b098e1235a3 : SUCCESS in 3m 22s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/aa5b79e33d6643aa8cf63a1a7ee42644 : SUCCESS in 3m 54s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/e1a2fa5c23ed4010ae1f92c7c1ee425b : SUCCESS in 4m 20s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/d65d6cc132254725af1965c71e929dbc : SUCCESS in 7m 57s\n- kolla-build-ubuntu-source https://zuul.opendev.org/t/openstack/build/f80c97d042cf4340b16be58d1d7ef056 : SUCCESS in 42m 18s\n- kolla-ansible-ubuntu-source https://zuul.opendev.org/t/openstack/build/fddc504255db4a3bad1392149bbfa3a1 : SUCCESS in 56m 46s\n- kolla-ansible-ubuntu-source-upgrade https://zuul.opendev.org/t/openstack/build/f32b9a2edacf475492463b0fdd331f6a : SUCCESS in 59m 59s","accounts_in_message":[],"_revision_number":3},{"id":"bc238700724a3c2df223b71d5400b6f2bc7b4f21","author":{"_account_id":24072,"name":"Marcin Juszkiewicz","email":"mjuszkiewicz@redhat.com","username":"hrw"},"date":"2019-09-19 17:42:08.000000000","message":"Patch Set 3:\n\nrecheck","accounts_in_message":[],"_revision_number":3},{"id":"01fe689de544d064c13ddab12ea3417e98fd4873","author":{"_account_id":23181,"name":"RDO Third Party CI","email":"dmsimard+rdothirdparty@redhat.com","username":"rdothirdparty","tags":["SERVICE_USER"]},"date":"2019-09-19 17:57:09.000000000","message":"Patch Set 3:\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttp://docs.openstack.org/infra/manual/developers.html#automated-testing\n\n\n- tripleo-build-containers-rhel-8-master http://logs.rdoproject.org/68/682068/3/openstack-check/tripleo-build-containers-rhel-8-master/1958ed7/ : FAILURE in 10m 16s","accounts_in_message":[],"_revision_number":3},{"id":"a0e41ca71af7eb767e1391d98f06e39c3a0b0e7a","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-09-19 19:27:04.000000000","message":"Patch Set 3: Verified+1\n\nBuild succeeded (check pipeline).\n\n- kolla-build-centos-binary https://zuul.opendev.org/t/openstack/build/84645160cc0c43fda29870a1fdb8a95b : SUCCESS in 41m 20s (non-voting)\n- kolla-build-centos-source https://zuul.opendev.org/t/openstack/build/8bb5b45549cf4c609e6b2d7aaaaa9c71 : SUCCESS in 50m 09s\n- kolla-ansible-centos-source https://zuul.opendev.org/t/openstack/build/ed5215dca6a94041a84c1bf541f03394 : SUCCESS in 1h 36m 09s\n- kolla-ansible-centos-binary https://zuul.opendev.org/t/openstack/build/d97a9072fba24e0ea07f4c6017faca56 : SUCCESS in 1h 25m 44s (non-voting)\n- kolla-ansible-centos-source-upgrade https://zuul.opendev.org/t/openstack/build/65f2454566f34c9e82261d782e73ddec : SUCCESS in 1h 39m 05s\n- tripleo-build-containers-centos-7 https://zuul.opendev.org/t/openstack/build/1f10382faff14c4e9161cbff9c9f18ef : SUCCESS in 57m 37s\n- kolla-build-debian-source https://zuul.opendev.org/t/openstack/build/374173aba69845ba96452b11d544aa1a : SUCCESS in 31m 27s (non-voting)\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/b0869bada34c43f1bb9949adc1b9a67c : SUCCESS in 3m 55s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/e71dff71c15a4850b544d23fbda4ad7a : SUCCESS in 5m 24s\n- openstack-tox-py27 https://zuul.opendev.org/t/openstack/build/321e5903e171448e8e05e92102c492fb : SUCCESS in 3m 27s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/11bb8be91b0f442a9f393eb45eb75e4c : SUCCESS in 3m 34s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/4d65aebd25494a07a25aa7b6c8e372e3 : SUCCESS in 3m 31s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/24f3c1dde6754f0587131392bad27679 : SUCCESS in 5m 16s\n- kolla-build-ubuntu-binary https://zuul.opendev.org/t/openstack/build/6bbe307f82924c72a4db304814a01a0c : SUCCESS in 29m 57s (non-voting)\n- kolla-build-ubuntu-source https://zuul.opendev.org/t/openstack/build/697ee84e9d55479dbebd1b832f14d14d : SUCCESS in 31m 39s\n- kolla-ansible-ubuntu-source https://zuul.opendev.org/t/openstack/build/2c8a27a422044ab1adc4fe8d65803b21 : SUCCESS in 1h 00m 29s\n- kolla-ansible-ubuntu-binary https://zuul.opendev.org/t/openstack/build/39c540f9a66a48b79d4188f2f8df7a3e : SUCCESS in 54m 37s (non-voting)\n- kolla-ansible-ubuntu-source-upgrade https://zuul.opendev.org/t/openstack/build/eb7970df1a4b41ea972b61bf2cbddfe7 : SUCCESS in 1h 09m 13s","accounts_in_message":[],"_revision_number":3},{"id":"31477b35b4b9d24b98cf592312f8b651ec811ef8","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-09-19 19:27:19.000000000","message":"Patch Set 3: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":3},{"id":"960e941e85fa343b8a8e4b14adad4ed437a1a661","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-09-19 21:04:47.000000000","message":"Patch Set 3: Verified+2\n\nBuild succeeded (gate pipeline).\n\n- kolla-build-centos-source https://zuul.opendev.org/t/openstack/build/c40babe132a047d69ee7602af6dc98de : SUCCESS in 55m 06s\n- kolla-ansible-centos-source https://zuul.opendev.org/t/openstack/build/614205f13bf44221a2322ba0bb16440d : SUCCESS in 1h 32m 33s\n- kolla-ansible-centos-source-upgrade https://zuul.opendev.org/t/openstack/build/e3829d10507f433b824828a3fcc44a22 : SUCCESS in 1h 34m 38s\n- tripleo-build-containers-centos-7 https://zuul.opendev.org/t/openstack/build/90afa11071f343e0a331b0ebc3c4fb22 : SUCCESS in 1h 00m 17s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/a29d5dd5577a4e109d9584fdab94e931 : SUCCESS in 4m 19s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/015b326826f34dfcb1f81b18e9a4b5a7 : SUCCESS in 5m 02s\n- openstack-tox-py27 https://zuul.opendev.org/t/openstack/build/87fab96a07cc41d1b14f57acddeffacd : SUCCESS in 5m 02s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/4fcea8bb278e4e8a80133a65ac6ad183 : SUCCESS in 3m 11s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/311d3d0b668f4e8d80647c810e482dae : SUCCESS in 5m 01s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/17b42654107e440a8463d13c6570ae10 : SUCCESS in 5m 44s\n- kolla-build-ubuntu-source https://zuul.opendev.org/t/openstack/build/2058b8d6053e44dbba0a261d47099f19 : SUCCESS in 29m 38s\n- kolla-ansible-ubuntu-source https://zuul.opendev.org/t/openstack/build/82183a12f75942919220e9da537b2314 : SUCCESS in 57m 29s\n- kolla-ansible-ubuntu-source-upgrade https://zuul.opendev.org/t/openstack/build/e10420d240b0419190887b05c08d803b : SUCCESS in 58m 10s","accounts_in_message":[],"_revision_number":3},{"id":"906278f9e6b39e04336bf3ce1b3284fbea5619cb","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-09-19 21:04:47.000000000","message":"Change has been successfully merged by Zuul","accounts_in_message":[],"_revision_number":3},{"id":"d2c7a93a578113ff8329b31aa0b41035e01c3c45","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-09-19 21:06:46.000000000","message":"Patch Set 3:\n\nBuild succeeded (promote pipeline).\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/6ffd5864d7384edea951567ed63ed747 : SUCCESS in 1m 20s","accounts_in_message":[],"_revision_number":3}],"current_revision_number":3,"current_revision":"bf3d051c733bfbe8e83331eb7076d13fffefed8d","revisions":{"a71bc0149e5a4d039c0400025321d846f0c17fe3":{"kind":"REWORK","_number":1,"created":"2019-09-13 14:34:19.000000000","uploader":{"_account_id":24072,"name":"Marcin Juszkiewicz","email":"mjuszkiewicz@redhat.com","username":"hrw"},"ref":"refs/changes/68/682068/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/kolla","ref":"refs/changes/68/682068/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/kolla refs/changes/68/682068/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/kolla refs/changes/68/682068/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/kolla refs/changes/68/682068/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/kolla refs/changes/68/682068/1"}}},"commit":{"parents":[{"commit":"532af2efc8893cfd23f8d2beadd32a4e3463ae85","subject":"Do not fan out DB sync to all cells","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/532af2efc8893cfd23f8d2beadd32a4e3463ae85"}]}],"author":{"name":"Marcin Juszkiewicz","email":"marcin.juszkiewicz@linaro.org","date":"2019-09-13 14:34:09.000000000","tz":120},"committer":{"name":"Marcin Juszkiewicz","email":"marcin.juszkiewicz@linaro.org","date":"2019-09-13 14:34:09.000000000","tz":120},"subject":"ppc64le: add UNBUILDABLE_IMAGES entries","message":"ppc64le: add UNBUILDABLE_IMAGES entries\n\nChange-Id: I6b8f52a1f2227ecb74e3ae59ef1e67d5b5ce1886\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/a71bc0149e5a4d039c0400025321d846f0c17fe3"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/a71bc0149e5a4d039c0400025321d846f0c17fe3"}]},"branch":"refs/heads/master"},"1dd7e80a5ab0b41aba149e7053313af37b8b47d1":{"kind":"REWORK","_number":2,"created":"2019-09-14 16:03:59.000000000","uploader":{"_account_id":24072,"name":"Marcin Juszkiewicz","email":"mjuszkiewicz@redhat.com","username":"hrw"},"ref":"refs/changes/68/682068/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/kolla","ref":"refs/changes/68/682068/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/kolla refs/changes/68/682068/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/kolla refs/changes/68/682068/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/kolla refs/changes/68/682068/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/kolla refs/changes/68/682068/2"}}},"commit":{"parents":[{"commit":"532af2efc8893cfd23f8d2beadd32a4e3463ae85","subject":"Do not fan out DB sync to all cells","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/532af2efc8893cfd23f8d2beadd32a4e3463ae85"}]}],"author":{"name":"Marcin Juszkiewicz","email":"marcin.juszkiewicz@linaro.org","date":"2019-09-13 14:34:09.000000000","tz":120},"committer":{"name":"Marcin Juszkiewicz","email":"marcin.juszkiewicz@linaro.org","date":"2019-09-14 16:03:49.000000000","tz":120},"subject":"ppc64le: add UNBUILDABLE_IMAGES entries","message":"ppc64le: add UNBUILDABLE_IMAGES entries\n\nChange-Id: I6b8f52a1f2227ecb74e3ae59ef1e67d5b5ce1886\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/1dd7e80a5ab0b41aba149e7053313af37b8b47d1"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/1dd7e80a5ab0b41aba149e7053313af37b8b47d1"}]},"branch":"refs/heads/master"},"bf3d051c733bfbe8e83331eb7076d13fffefed8d":{"kind":"REWORK","_number":3,"created":"2019-09-19 07:22:39.000000000","uploader":{"_account_id":24072,"name":"Marcin Juszkiewicz","email":"mjuszkiewicz@redhat.com","username":"hrw"},"ref":"refs/changes/68/682068/3","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/kolla","ref":"refs/changes/68/682068/3","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/kolla refs/changes/68/682068/3 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/kolla refs/changes/68/682068/3 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/kolla refs/changes/68/682068/3 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/kolla refs/changes/68/682068/3"}}},"commit":{"parents":[{"commit":"532af2efc8893cfd23f8d2beadd32a4e3463ae85","subject":"Do not fan out DB sync to all cells","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/532af2efc8893cfd23f8d2beadd32a4e3463ae85"}]}],"author":{"name":"Marcin Juszkiewicz","email":"marcin.juszkiewicz@linaro.org","date":"2019-09-13 14:34:09.000000000","tz":120},"committer":{"name":"Marcin Juszkiewicz","email":"marcin.juszkiewicz@linaro.org","date":"2019-09-19 07:22:28.000000000","tz":0},"subject":"ppc64le: add UNBUILDABLE_IMAGES entries","message":"ppc64le: add UNBUILDABLE_IMAGES entries\n\nChange-Id: I6b8f52a1f2227ecb74e3ae59ef1e67d5b5ce1886\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/bf3d051c733bfbe8e83331eb7076d13fffefed8d"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/bf3d051c733bfbe8e83331eb7076d13fffefed8d"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[],"submit_requirements":[]}
