)]}'
{"id":"openstack%2Fkolla~682280","triplet_id":"openstack%2Fkolla~master~I1d5cd3d9af98444acac5bedd7daeaa6c6673dcd6","project":"openstack/kolla","branch":"master","topic":"bug/1814552","hashtags":[],"change_id":"I1d5cd3d9af98444acac5bedd7daeaa6c6673dcd6","subject":"nova-libvirt: add UEFI packages to support UEFI instances","status":"MERGED","created":"2019-09-16 07:02:32.000000000","updated":"2020-01-31 19:28:59.000000000","submitted":"2020-01-31 17:57:29.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":10,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"682280-1580493449208-f323dc7b","meta_rev_id":"1f4367d1478b19f32eae6de8c15c681edb55e137","_number":682280,"virtual_id_number":682280,"owner":{"_account_id":19779,"name":"Chason Chan","display_name":"chenxing","email":"chason.chan@foxmail.com","username":"chenxing"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},{"value":0,"_account_id":24072,"name":"Marcin Juszkiewicz","email":"mjuszkiewicz@redhat.com","username":"hrw"},{"value":0,"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},{"value":0,"date":"2020-01-31 15:13:30.000000000","_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2020-01-31 17:57:28.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"date":"2020-01-31 11:44:21.000000000","_account_id":23181,"name":"RDO Third Party CI","email":"dmsimard+rdothirdparty@redhat.com","username":"rdothirdparty","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"all":[{"value":2,"date":"2020-01-31 15:57:19.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},{"value":0,"_account_id":24072,"name":"Marcin Juszkiewicz","email":"mjuszkiewicz@redhat.com","username":"hrw"},{"value":2,"date":"2020-01-31 15:23:48.000000000","_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},{"value":0,"_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":23181,"name":"RDO Third Party CI","email":"dmsimard+rdothirdparty@redhat.com","username":"rdothirdparty","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"all":[{"value":1,"date":"2020-01-31 15:57:19.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},{"value":0,"_account_id":24072,"name":"Marcin Juszkiewicz","email":"mjuszkiewicz@redhat.com","username":"hrw"},{"value":0,"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},{"value":0,"_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":23181,"name":"RDO Third Party CI","email":"dmsimard+rdothirdparty@redhat.com","username":"rdothirdparty","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Backport-Candidate":{"all":[{"value":0,"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},{"value":0,"_account_id":24072,"name":"Marcin Juszkiewicz","email":"mjuszkiewicz@redhat.com","username":"hrw"},{"value":0,"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},{"value":0,"_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":23181,"name":"RDO Third Party CI","email":"dmsimard+rdothirdparty@redhat.com","username":"rdothirdparty","tags":["SERVICE_USER"]}],"values":{"-1":"Do Not Backport"," 0":"Backport Review Needed","+1":"Should Backport"},"description":"","default_value":0,"optional":true},"Review-Priority":{"recommended":{"_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"},"all":[{"value":0,"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},{"value":0,"_account_id":24072,"name":"Marcin Juszkiewicz","email":"mjuszkiewicz@redhat.com","username":"hrw"},{"value":0,"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},{"value":1,"date":"2020-01-31 19:28:59.000000000","post_submit":true,"_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":23181,"name":"RDO Third Party CI","email":"dmsimard+rdothirdparty@redhat.com","username":"rdothirdparty","tags":["SERVICE_USER"]}],"values":{"-1":"Branch Freeze"," 0":"No Priority","+1":"Important Change","+2":"Gate Blocker Fix / Urgent Change"},"description":"","value":1,"default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},{"_account_id":23181,"name":"RDO Third Party CI","email":"dmsimard+rdothirdparty@redhat.com","username":"rdothirdparty","tags":["SERVICE_USER"]},{"_account_id":24072,"name":"Marcin Juszkiewicz","email":"mjuszkiewicz@redhat.com","username":"hrw"},{"_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2020-01-30 21:02:39.000000000","updated_by":{"_account_id":24072,"name":"Marcin Juszkiewicz","email":"mjuszkiewicz@redhat.com","username":"hrw"},"reviewer":{"_account_id":24072,"name":"Marcin Juszkiewicz","email":"mjuszkiewicz@redhat.com","username":"hrw"},"state":"REVIEWER"},{"updated":"2020-01-31 11:44:21.000000000","updated_by":{"_account_id":23181,"name":"RDO Third Party CI","email":"dmsimard+rdothirdparty@redhat.com","username":"rdothirdparty","tags":["SERVICE_USER"]},"reviewer":{"_account_id":23181,"name":"RDO Third Party CI","email":"dmsimard+rdothirdparty@redhat.com","username":"rdothirdparty","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2020-01-31 15:13:30.000000000","updated_by":{"_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"},"reviewer":{"_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"},"state":"REVIEWER"},{"updated":"2020-01-31 15:23:48.000000000","updated_by":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},"reviewer":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},"state":"REVIEWER"},{"updated":"2020-01-31 15:57:19.000000000","updated_by":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"reviewer":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"state":"REVIEWER"},{"updated":"2020-01-31 17:57:28.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"}],"messages":[{"id":"46d86fdf3e16769625ee8e07b08724557be4412d","author":{"_account_id":19779,"name":"Chason Chan","display_name":"chenxing","email":"chason.chan@foxmail.com","username":"chenxing"},"date":"2019-09-16 07:02:32.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"a1b7fb31f786ac966991c1ffa518a82937d3027f","author":{"_account_id":23181,"name":"RDO Third Party CI","email":"dmsimard+rdothirdparty@redhat.com","username":"rdothirdparty","tags":["SERVICE_USER"]},"date":"2019-09-16 07:50:37.000000000","message":"Patch Set 1:\n\nBuild succeeded (check pipeline).\n\n- tripleo-build-containers-rhel-8-master http://logs.rdoproject.org/80/682280/1/openstack-check/tripleo-build-containers-rhel-8-master/b07ad26/ : SUCCESS in 47m 16s","accounts_in_message":[],"_revision_number":1},{"id":"1ea976c30554fd6c98f5ec9ebc728f7985849173","author":{"_account_id":24072,"name":"Marcin Juszkiewicz","email":"mjuszkiewicz@redhat.com","username":"hrw"},"date":"2019-09-16 08:01:55.000000000","message":"Patch Set 1: Code-Review-1\n\n(2 comments)","accounts_in_message":[],"_revision_number":1},{"id":"10224f617978f23a9655f0c6f60ff97ce4fc5185","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-09-16 08:44:13.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\n\n- kolla-build-centos-binary https://zuul.opendev.org/t/openstack/build/da72da59efdd45f08bba6e59726ef478 : SUCCESS in 50m 37s (non-voting)\n- kolla-build-centos-source https://zuul.opendev.org/t/openstack/build/d66a50dc2ea347ea949828d463e94088 : SUCCESS in 42m 41s\n- kolla-ansible-centos-source https://zuul.opendev.org/t/openstack/build/ca0fbe74c74b456d8d389e7737fef5b0 : SUCCESS in 1h 13m 38s\n- kolla-ansible-centos-binary https://zuul.opendev.org/t/openstack/build/23f6271ad6f449e3a69fcd3c3acfa016 : SUCCESS in 1h 15m 23s (non-voting)\n- kolla-ansible-centos-source-upgrade https://zuul.opendev.org/t/openstack/build/3ad35891756e48518784fe8493365557 : SUCCESS in 1h 25m 50s\n- tripleo-build-containers-centos-7 https://zuul.opendev.org/t/openstack/build/b5286a6ea9314ce0b4cf0f1315318282 : SUCCESS in 53m 00s\n- kolla-build-debian-source https://zuul.opendev.org/t/openstack/build/171bb00dc1aa4f1eaaa5485edbaa12a9 : SUCCESS in 28m 45s (non-voting)\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/3ca5c5eee8bc482e97b20d66d1751d48 : SUCCESS in 3m 41s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/b06f7fabaed840029d7670bf26fde366 : SUCCESS in 4m 18s\n- openstack-tox-py27 https://zuul.opendev.org/t/openstack/build/b4fb2dfaae1f4f038d01b92ef17aaea5 : SUCCESS in 5m 16s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/b393839acb4b4d45a8dfe66ca7538ac1 : SUCCESS in 4m 32s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/a8d0bb3c86594625b5a58c48eb74f3b6 : SUCCESS in 5m 17s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/f0bb24af78464fd9a4a7687415c99142 : SUCCESS in 4m 50s\n- kolla-build-ubuntu-binary https://zuul.opendev.org/t/openstack/build/9e826f3336474c0db5218d3be9086665 : SUCCESS in 29m 56s (non-voting)\n- kolla-build-ubuntu-source https://zuul.opendev.org/t/openstack/build/40e5dba4276f4c1f9f21fd632fc94e38 : SUCCESS in 30m 57s\n- kolla-ansible-ubuntu-source https://zuul.opendev.org/t/openstack/build/222f1582d85f4629b0148c753747c522 : SUCCESS in 52m 36s\n- kolla-ansible-ubuntu-binary https://zuul.opendev.org/t/openstack/build/176583b03983458ab09285e5430684a0 : SUCCESS in 1h 00m 14s (non-voting)\n- kolla-ansible-ubuntu-source-upgrade https://zuul.opendev.org/t/openstack/build/d6c3b7c77a8a44789ed1bcdcaef28c2f : SUCCESS in 1h 14m 12s","accounts_in_message":[],"_revision_number":1},{"id":"7795d9a4e3bf7bc43b919d2eee7f90c496481248","author":{"_account_id":24072,"name":"Marcin Juszkiewicz","email":"mjuszkiewicz@redhat.com","username":"hrw"},"date":"2019-09-16 09:25:46.000000000","message":"Patch Set 1:\n\nlook at https://review.opendev.org/#/c/608579/","accounts_in_message":[],"_revision_number":1},{"id":"476273dee2f987c002d7b11d40eaaed7df02e490","author":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},"date":"2019-09-17 16:48:37.000000000","message":"Patch Set 1:\n\n(2 comments)","accounts_in_message":[],"_revision_number":1},{"id":"30ceb9e862bc7843f6a97f19b19e4835805ef2c6","author":{"_account_id":24072,"name":"Marcin Juszkiewicz","email":"mjuszkiewicz@redhat.com","username":"hrw"},"date":"2020-01-30 15:28:40.000000000","message":"Uploaded patch set 2.","accounts_in_message":[],"_revision_number":2},{"id":"8b926e898638dd780d9ad62ebfea63f753a46643","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-01-30 15:57:49.000000000","message":"Patch Set 2:\n\nBuild succeeded (ARM64 pipeline).\n\n- kolla-build-debian-source-aarch64 https://zuul.opendev.org/t/openstack/build/None : NODE_FAILURE in 0s (non-voting)","accounts_in_message":[],"_revision_number":2},{"id":"55ecf9a7420f342bca1875f3c410fbcd42ff972a","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-01-30 17:13:52.000000000","message":"Patch Set 2: Verified+1\n\nBuild succeeded (check pipeline).\n\n- kolla-build-centos-binary https://zuul.opendev.org/t/openstack/build/fe37e8b6e79245faaf0b9fba62cee179 : SUCCESS in 34m 43s (non-voting)\n- kolla-build-centos-source https://zuul.opendev.org/t/openstack/build/9fa6852fb0804ccabecfa4019d4d0e79 : SUCCESS in 41m 07s\n- kolla-build-centos8-binary https://zuul.opendev.org/t/openstack/build/c1c0617224344f7583b6ced7ebcbc765 : SUCCESS in 13m 19s (non-voting)\n- kolla-build-centos8-source https://zuul.opendev.org/t/openstack/build/15d8b84c520d46be9a43009a9f797d34 : SUCCESS in 11m 42s\n- kolla-ansible-centos-source https://zuul.opendev.org/t/openstack/build/ea1651c7e5b64812befea95f9a57066f : SUCCESS in 1h 17m 05s\n- kolla-ansible-centos-binary https://zuul.opendev.org/t/openstack/build/d3bc2f9ab663426a9ef4e8ab5b26a155 : SUCCESS in 1h 15m 34s (non-voting)\n- kolla-ansible-centos-source-upgrade https://zuul.opendev.org/t/openstack/build/466a5b81ffc04b56ba496cf5ccc3b357 : SUCCESS in 1h 40m 30s\n- tripleo-build-containers-centos-7 https://zuul.opendev.org/t/openstack/build/07a0b1d657fc496f8126860a76156d97 : SUCCESS in 1h 03m 50s\n- kolla-build-debian-source https://zuul.opendev.org/t/openstack/build/f249e01fd5354a0ba51c8834304b9fc1 : SUCCESS in 26m 49s (non-voting)\n- kolla-build-debian-binary https://zuul.opendev.org/t/openstack/build/08b1c26c64f14f3191b512e914d190f9 : SUCCESS in 23m 31s (non-voting)\n- kolla-ansible-debian-source https://zuul.opendev.org/t/openstack/build/a92a63f4b9d74fc9898ea4ad3be265ee : SUCCESS in 1h 21m 10s (non-voting)\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/5a0634975cb543f8aae58ef67b7273b3 : SUCCESS in 5m 22s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/4d5abd8a2758423da416e378155be3bd : SUCCESS in 6m 48s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/d08f081a062d4d62bb80238f97f68405 : SUCCESS in 4m 02s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/ec5c50098f6a4a8a9b19f78947c4820b : SUCCESS in 5m 28s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/df8e657a18b448f59367bfa75f373cbb : SUCCESS in 4m 10s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/9aab91154d054c508ae66059240b0255 : SUCCESS in 8m 04s\n- kolla-build-ubuntu-binary https://zuul.opendev.org/t/openstack/build/fbb9051419d743a5aa381a06a3e4b6a4 : SUCCESS in 25m 14s (non-voting)\n- kolla-build-ubuntu-source https://zuul.opendev.org/t/openstack/build/bf02eeb3cccf43bfb319af6ee5487479 : SUCCESS in 27m 51s\n- kolla-ansible-ubuntu-source https://zuul.opendev.org/t/openstack/build/31afb15ff63c40b8b11440588b690943 : SUCCESS in 55m 50s\n- kolla-ansible-ubuntu-binary https://zuul.opendev.org/t/openstack/build/395a68ef0ac4495b98a6eacb74515070 : SUCCESS in 51m 45s (non-voting)\n- kolla-ansible-ubuntu-source-upgrade https://zuul.opendev.org/t/openstack/build/885e5218186240c38d1f571503bbdf38 : SUCCESS in 1h 03m 15s","accounts_in_message":[],"_revision_number":2},{"id":"ba7f93f2ab126fac6e12798b40f6d90f27987bee","author":{"_account_id":23181,"name":"RDO Third Party CI","email":"dmsimard+rdothirdparty@redhat.com","username":"rdothirdparty","tags":["SERVICE_USER"]},"date":"2020-01-30 17:26:06.000000000","message":"Patch Set 2:\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttp://docs.openstack.org/infra/manual/developers.html#automated-testing\n\n\n- tripleo-build-containers-rhel-8-master https://review.rdoproject.org/zuul/build/466896b72b174884b4fd272e76b80eb4 : POST_FAILURE in 31m 35s","accounts_in_message":[],"_revision_number":2},{"id":"feae8004d10559d419e212bdc013d937861c8009","author":{"_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"},"date":"2020-01-30 18:20:25.000000000","message":"Patch Set 2: Code-Review-1\n\n(2 comments)\n\nthis needs aligning with nova-compute","accounts_in_message":[],"_revision_number":2},{"id":"e049ff3fef179e58f40868a413ff522325e9bc0f","author":{"_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"},"date":"2020-01-30 18:23:57.000000000","message":"Patch Set 2:\n\n(see both changes for comment deps)","accounts_in_message":[],"_revision_number":2},{"id":"ce7a06f5ca3d80bf7c109822c3fd777592ba5bdb","author":{"_account_id":24072,"name":"Marcin Juszkiewicz","email":"mjuszkiewicz@redhat.com","username":"hrw"},"date":"2020-01-30 21:02:39.000000000","message":"Patch Set 2:\n\n\u003e this needs aligning with nova-compute\n\nnova-compute base on nova-base which is when I copied from - see my other patch.","accounts_in_message":[],"_revision_number":2},{"id":"9f8592518553dfbd4b2b7529841f753d77604671","author":{"_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"},"date":"2020-01-31 09:19:54.000000000","message":"Uploaded patch set 3.","accounts_in_message":[],"_revision_number":3},{"id":"b11fd994aa2471aa6ccd3161a8c5e29d3dbad429","author":{"_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"},"date":"2020-01-31 09:20:55.000000000","message":"Uploaded patch set 4: Commit message was updated.","accounts_in_message":[],"_revision_number":4},{"id":"77b010c2a7979b5dfa0582445e11902acaa89039","author":{"_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"},"date":"2020-01-31 09:21:01.000000000","message":"Patch Set 3: Review-Priority+1","accounts_in_message":[],"_revision_number":3},{"id":"c8b46beca23065fb155b867d7934e7961c13e96e","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-01-31 09:25:22.000000000","message":"Patch Set 4:\n\nBuild succeeded (ARM64 pipeline).\n\n- kolla-build-debian-source-aarch64 https://zuul.opendev.org/t/openstack/build/None : NODE_FAILURE in 0s (non-voting)","accounts_in_message":[],"_revision_number":4},{"id":"274eab29695e6841f81b651ec7df6ee899f123ed","author":{"_account_id":23181,"name":"RDO Third Party CI","email":"dmsimard+rdothirdparty@redhat.com","username":"rdothirdparty","tags":["SERVICE_USER"]},"date":"2020-01-31 09:55:36.000000000","message":"Patch Set 4:\n\nBuild succeeded (check pipeline).\n\n- tripleo-build-containers-rhel-8-master https://review.rdoproject.org/zuul/build/c22e93df06b7458c9b7631780f4202a8 : SUCCESS in 32m 27s","accounts_in_message":[],"_revision_number":4},{"id":"486c829348cd66f3598e04053731c527597b4b7b","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-01-31 10:50:30.000000000","message":"Patch Set 4: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttp://docs.openstack.org/infra/manual/developers.html#automated-testing\n\n\n- kolla-build-centos-binary https://zuul.opendev.org/t/openstack/build/1782eff39a2b4b00a691ecd3e6b4a167 : SUCCESS in 34m 23s (non-voting)\n- kolla-build-centos-source https://zuul.opendev.org/t/openstack/build/34dfdbb0b6b3429f891ba21876c8a1eb : SUCCESS in 37m 53s\n- kolla-build-centos8-binary https://zuul.opendev.org/t/openstack/build/04058ab8da2e46caa0d015460ae0565d : SUCCESS in 24m 43s (non-voting)\n- kolla-build-centos8-source https://zuul.opendev.org/t/openstack/build/382141651491417c970b7408dcd1bfea : SUCCESS in 28m 39s\n- kolla-ansible-centos-source https://zuul.opendev.org/t/openstack/build/d072294108d043068c14c31c62c0e4b5 : SUCCESS in 1h 22m 57s\n- kolla-ansible-centos-binary https://zuul.opendev.org/t/openstack/build/d13e5584b4d74daba63f22104fab095a : SUCCESS in 1h 22m 40s (non-voting)\n- kolla-ansible-centos-source-upgrade https://zuul.opendev.org/t/openstack/build/001b1af2ab684bcc8d2090b2fe150d44 : FAILURE in 59m 28s\n- tripleo-build-containers-centos-7 https://zuul.opendev.org/t/openstack/build/0c1c343fde8346e79f186be7913ea57a : SUCCESS in 51m 41s\n- kolla-build-debian-source https://zuul.opendev.org/t/openstack/build/fd4e6090804848328dfd966b1d08767c : SUCCESS in 44m 40s (non-voting)\n- kolla-build-debian-binary https://zuul.opendev.org/t/openstack/build/a23bf4c5a1ef40e3b997788ea60be450 : SUCCESS in 43m 17s (non-voting)\n- kolla-ansible-debian-source https://zuul.opendev.org/t/openstack/build/159a88c4cbea45ad96e4f5d7d0963270 : SUCCESS in 54m 55s (non-voting)\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/c3e5b7855da54f9ba10899630a93c665 : SUCCESS in 4m 39s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/a3919951e04d405e99b60cf5df41764d : SUCCESS in 5m 42s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/4279066f65994b78be8cc8e0134653f6 : SUCCESS in 4m 15s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/fd32cbef3c6b4f46929d586841fbac5f : SUCCESS in 5m 51s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/6c10dcb6b62c458ab03282fc0f612d98 : SUCCESS in 3m 58s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/1b43637eb82b4a79b28a1591025bea29 : SUCCESS in 6m 07s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/a5a0a99ff3f24fcfb1feee905b541af6 : SUCCESS in 5m 35s\n- kolla-build-ubuntu-binary https://zuul.opendev.org/t/openstack/build/4e01b187a0aa43dea83243eef1449924 : SUCCESS in 26m 20s (non-voting)\n- kolla-build-ubuntu-source https://zuul.opendev.org/t/openstack/build/d7141ad08c244407ac83a5f2b15c613f : SUCCESS in 27m 29s\n- kolla-ansible-ubuntu-source https://zuul.opendev.org/t/openstack/build/d3ce576599bc4b74b400d86e722de38b : SUCCESS in 54m 31s\n- kolla-ansible-ubuntu-binary https://zuul.opendev.org/t/openstack/build/3ef780a6af3144daa7954683531c37ed : SUCCESS in 1h 06m 16s (non-voting)\n- kolla-ansible-ubuntu-source-upgrade https://zuul.opendev.org/t/openstack/build/0f529cac0e984c578c7cb0c051c442ba : SUCCESS in 1h 19m 34s","accounts_in_message":[],"_revision_number":4},{"id":"1efb8bca67c1501de0553401540c77ea14fa8d54","author":{"_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"},"date":"2020-01-31 10:56:13.000000000","message":"Patch Set 4:\n\nrecheck","accounts_in_message":[],"_revision_number":4},{"id":"4c3d0e309db8e8632c11ccd73abe84ec8afc3208","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-01-31 10:59:42.000000000","message":"Patch Set 4:\n\nBuild succeeded (ARM64 pipeline).\n\n- kolla-build-debian-source-aarch64 https://zuul.opendev.org/t/openstack/build/None : NODE_FAILURE in 0s (non-voting)","accounts_in_message":[],"_revision_number":4},{"id":"0bce2878610a23eb45e74e5b0ad7ea1ad28c585f","author":{"_account_id":23181,"name":"RDO Third Party CI","email":"dmsimard+rdothirdparty@redhat.com","username":"rdothirdparty","tags":["SERVICE_USER"]},"date":"2020-01-31 11:44:21.000000000","message":"Patch Set 4:\n\nBuild succeeded (check pipeline).\n\n- tripleo-build-containers-rhel-8-master https://review.rdoproject.org/zuul/build/fcb55e5e0072439aa206e1b3085104ec : SUCCESS in 27m 53s","accounts_in_message":[],"_revision_number":4},{"id":"d08ae11f5a275d2f4032e36eae0718d39465ec76","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-01-31 12:57:52.000000000","message":"Patch Set 4: Verified+1\n\nBuild succeeded (check pipeline).\n\n- kolla-build-centos-binary https://zuul.opendev.org/t/openstack/build/576737a5b95145e5a42b226c9278437d : SUCCESS in 45m 44s (non-voting)\n- kolla-build-centos-source https://zuul.opendev.org/t/openstack/build/240fb502587c4256bb3506c61d224930 : SUCCESS in 53m 08s\n- kolla-build-centos8-binary https://zuul.opendev.org/t/openstack/build/f607ab3da57444df882a1acefba8b95a : SUCCESS in 25m 29s (non-voting)\n- kolla-build-centos8-source https://zuul.opendev.org/t/openstack/build/e8150736d2e446458889505753241c51 : SUCCESS in 37m 50s\n- kolla-ansible-centos-source https://zuul.opendev.org/t/openstack/build/03fc5752d3f245218964509a90a62015 : SUCCESS in 1h 30m 58s\n- kolla-ansible-centos-binary https://zuul.opendev.org/t/openstack/build/5df925a6a79942db9cf56bc12a6e0600 : SUCCESS in 1h 52m 45s (non-voting)\n- kolla-ansible-centos-source-upgrade https://zuul.opendev.org/t/openstack/build/b478aae16d6c400d99adb13272e14c12 : SUCCESS in 1h 30m 02s\n- tripleo-build-containers-centos-7 https://zuul.opendev.org/t/openstack/build/6e7eb2f87bd642a9a3f85d1d8c8473b0 : SUCCESS in 55m 19s\n- kolla-build-debian-source https://zuul.opendev.org/t/openstack/build/6d6bc4eb83db43f1a2aa1505bcacfb8d : SUCCESS in 20m 13s (non-voting)\n- kolla-build-debian-binary https://zuul.opendev.org/t/openstack/build/cc0da04829a6401aaf36d0379b5cc836 : SUCCESS in 24m 01s (non-voting)\n- kolla-ansible-debian-source https://zuul.opendev.org/t/openstack/build/619fd84af98a40569f5a5252afdec1ef : SUCCESS in 56m 48s (non-voting)\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/bc1c6b27476b4acc9a8e9f0f851548b2 : SUCCESS in 4m 11s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/c9cd457746b8436281a9ebe0d6b35602 : SUCCESS in 5m 45s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/1b9ca440129e4788b83ce77c25f0f4a9 : SUCCESS in 7m 48s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/2e4370dd868249c887ec44b4a59cd0da : SUCCESS in 4m 09s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/b4dbeab20a424e419832313cbf2c200a : SUCCESS in 4m 40s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/14bca569a81d4ddda61187173e1611f7 : SUCCESS in 6m 35s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/3ca81b6bf7974a08800312158b5ffd02 : SUCCESS in 5m 45s\n- kolla-build-ubuntu-binary https://zuul.opendev.org/t/openstack/build/9053a31bff4041f4b1e6d3cf2ebbbf6a : SUCCESS in 36m 08s (non-voting)\n- kolla-build-ubuntu-source https://zuul.opendev.org/t/openstack/build/100a919308674bd7818222a6b413895d : SUCCESS in 38m 48s\n- kolla-ansible-ubuntu-source https://zuul.opendev.org/t/openstack/build/2ea4cb1ba11645fea952f6081ee2eb22 : SUCCESS in 56m 46s\n- kolla-ansible-ubuntu-binary https://zuul.opendev.org/t/openstack/build/24c5136a94ab41f0a4cc3870066ccb0d : SUCCESS in 56m 33s (non-voting)\n- kolla-ansible-ubuntu-source-upgrade https://zuul.opendev.org/t/openstack/build/21ab61f3e59641efb88b468fb39c86e8 : SUCCESS in 1h 11m 38s","accounts_in_message":[],"_revision_number":4},{"id":"cab9ba114adf77cc9cf3f1b4768e760da41048d9","author":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},"date":"2020-01-31 13:16:04.000000000","message":"Patch Set 4:\n\n(2 comments)","accounts_in_message":[],"_revision_number":4},{"id":"07f13e881c5608656079cbd9f860c5b5f6f235d0","author":{"_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"},"date":"2020-01-31 15:13:30.000000000","message":"Patch Set 4:\n\n(1 comment)","accounts_in_message":[],"_revision_number":4},{"id":"d877c779634ffe0898e8a099a82649b1e011e6ef","author":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},"date":"2020-01-31 15:23:48.000000000","message":"Patch Set 4: Code-Review+2\n\n(1 comment)","accounts_in_message":[],"_revision_number":4},{"id":"a82a94e4a4d95094980e1e5a4b6042974a732533","author":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"date":"2020-01-31 15:57:19.000000000","message":"Patch Set 4: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":4},{"id":"34e6e3d907563308b3d95908728b4f735354344f","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-01-31 15:57:32.000000000","message":"Patch Set 4: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":4},{"id":"fecbacefe29c01830effb38936e2c02008f23676","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-01-31 17:57:28.000000000","message":"Patch Set 4: Verified+2\n\nBuild succeeded (gate pipeline).\n\n- kolla-build-centos-source https://zuul.opendev.org/t/openstack/build/1d92d3bcde7143e78f66a762d7a4377d : SUCCESS in 1h 11m 13s\n- kolla-build-centos8-source https://zuul.opendev.org/t/openstack/build/5c2253ec9c544016b58cfe149ca829a5 : SUCCESS in 56m 47s\n- kolla-ansible-centos-source https://zuul.opendev.org/t/openstack/build/12be71bd5fa445308cc31bcb109e05d0 : SUCCESS in 1h 38m 45s\n- kolla-ansible-centos-source-upgrade https://zuul.opendev.org/t/openstack/build/9e355c13be1c4864b59c5b62efab1a05 : SUCCESS in 1h 46m 29s\n- tripleo-build-containers-centos-7 https://zuul.opendev.org/t/openstack/build/b689afc902294432a0d73ee633f6b373 : SUCCESS in 52m 25s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/cf935a5c77ea431a9b95cca155c77393 : SUCCESS in 5m 39s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/349161a3288048d69c0c0a327b031f04 : SUCCESS in 6m 00s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/7f8638f92f4d4b1ca851e74bc72a2e8c : SUCCESS in 5m 36s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/8af5c2994dbc459391bc8851f24d8dae : SUCCESS in 5m 32s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/7beafd27bd6f4844a05bf4dd43a2cc52 : SUCCESS in 5m 24s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/00b9a7083ab14907abfb0ac65ec13904 : SUCCESS in 7m 14s\n- kolla-build-ubuntu-source https://zuul.opendev.org/t/openstack/build/81823ee89cf94a20b28474b5102e4477 : SUCCESS in 27m 39s\n- kolla-ansible-ubuntu-source https://zuul.opendev.org/t/openstack/build/cbbf8e8520194bb385aca4656b67a682 : SUCCESS in 1h 01m 24s\n- kolla-ansible-ubuntu-source-upgrade https://zuul.opendev.org/t/openstack/build/f5fbf075d37b400d9a15230ef9729791 : SUCCESS in 1h 14m 28s","accounts_in_message":[],"_revision_number":4},{"id":"638b7b2eda6e078d9871f4a73894f14c4ae3861d","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-01-31 17:57:29.000000000","message":"Change has been successfully merged by Zuul","accounts_in_message":[],"_revision_number":4},{"id":"3a1a45cb4809c6cbf65510a572cc2d6028b9796b","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-01-31 18:00:00.000000000","message":"Patch Set 4:\n\nBuild succeeded (promote pipeline).\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/3941b3a2fe914c91860a94a03527ddbb : SUCCESS in 1m 53s\n- promote-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/ed8a3aa9ac8e4a43b3318f69854c2416 : SUCCESS in 1m 50s","accounts_in_message":[],"_revision_number":4},{"id":"7b3dacd8bfb7eeb471c4dbda3e0ae684bd0a11b0","author":{"_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"},"date":"2020-01-31 19:28:59.000000000","message":"Patch Set 4: Cherry Picked\n\nThis patchset was cherry picked to branch stable/train as commit ffaf585d0f05ccd9123459bb0cdf345af5b09573","accounts_in_message":[],"_revision_number":4}],"current_revision_number":4,"current_revision":"15b68c15c1dd0c2df1cc49b550f2fa03a4c65e8e","revisions":{"acf33407f62177074a9069e3f5d1560daa59f717":{"kind":"REWORK","_number":1,"created":"2019-09-16 07:02:32.000000000","uploader":{"_account_id":19779,"name":"Chason Chan","display_name":"chenxing","email":"chason.chan@foxmail.com","username":"chenxing"},"ref":"refs/changes/80/682280/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/kolla","ref":"refs/changes/80/682280/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/kolla refs/changes/80/682280/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/kolla refs/changes/80/682280/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/kolla refs/changes/80/682280/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/kolla refs/changes/80/682280/1"}}},"commit":{"parents":[{"commit":"6c74c73a2415229c136f9854102fc99f9a8bc7c2","subject":"Merge \"base/deb: bump \u0027system\u0027 users limit beyond Kolla ones\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/6c74c73a2415229c136f9854102fc99f9a8bc7c2"}]}],"author":{"name":"chenxing","email":"chason.chan@foxmail.com","date":"2019-09-16 06:46:41.000000000","tz":480},"committer":{"name":"chenxing","email":"chason.chan@foxmail.com","date":"2019-09-16 07:01:33.000000000","tz":480},"subject":"Add OVMF package to support UEFI instance","message":"Add OVMF package to support UEFI instance\n\nChange-Id: I1d5cd3d9af98444acac5bedd7daeaa6c6673dcd6\nCloses-Bug: #1814552\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/acf33407f62177074a9069e3f5d1560daa59f717"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/acf33407f62177074a9069e3f5d1560daa59f717"}]},"branch":"refs/heads/master"},"2b37092af4597ec801224b8e430ef87a9b763d71":{"kind":"REWORK","_number":2,"created":"2020-01-30 15:28:40.000000000","uploader":{"_account_id":24072,"name":"Marcin Juszkiewicz","email":"mjuszkiewicz@redhat.com","username":"hrw"},"ref":"refs/changes/80/682280/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/kolla","ref":"refs/changes/80/682280/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/kolla refs/changes/80/682280/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/kolla refs/changes/80/682280/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/kolla refs/changes/80/682280/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/kolla refs/changes/80/682280/2"}}},"commit":{"parents":[{"commit":"a68223412e6e3289173adca1af760157b25121c2","subject":"Deprecate OpenDaylight","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/a68223412e6e3289173adca1af760157b25121c2"}]}],"author":{"name":"chenxing","email":"chason.chan@foxmail.com","date":"2019-09-16 06:46:41.000000000","tz":480},"committer":{"name":"Marcin Juszkiewicz","email":"marcin.juszkiewicz@linaro.org","date":"2020-01-30 15:28:27.000000000","tz":60},"subject":"nova-libvirt: add uefi package to support UEFI instances","message":"nova-libvirt: add uefi package to support UEFI instances\n\nCloses-Bug: #1814552\n\nChange-Id: I1d5cd3d9af98444acac5bedd7daeaa6c6673dcd6\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/2b37092af4597ec801224b8e430ef87a9b763d71"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/2b37092af4597ec801224b8e430ef87a9b763d71"}]},"branch":"refs/heads/master"},"7f94a3cf31ef2d619305b6080a81ad511eb91dda":{"kind":"REWORK","_number":3,"created":"2020-01-31 09:19:54.000000000","uploader":{"_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"},"ref":"refs/changes/80/682280/3","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/kolla","ref":"refs/changes/80/682280/3","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/kolla refs/changes/80/682280/3 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/kolla refs/changes/80/682280/3 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/kolla refs/changes/80/682280/3 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/kolla refs/changes/80/682280/3"}}},"commit":{"parents":[{"commit":"a68223412e6e3289173adca1af760157b25121c2","subject":"Deprecate OpenDaylight","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/a68223412e6e3289173adca1af760157b25121c2"}]}],"author":{"name":"chenxing","email":"chason.chan@foxmail.com","date":"2019-09-16 06:46:41.000000000","tz":480},"committer":{"name":"Radosław Piliszek","email":"radoslaw.piliszek@gmail.com","date":"2020-01-31 09:18:32.000000000","tz":60},"subject":"nova-libvirt: add UEFI packages to support UEFI instances","message":"nova-libvirt: add UEFI packages to support UEFI instances\n\nFix inability to run UEFI-based images/instances by installing UEFI\npackages also in nova-libvirt image which is not based on nova-base.\n\nIncludes support for C8.\nBackport below Train w/o C8.\n\nCloses-Bug: #1814552\n\nChange-Id: I1d5cd3d9af98444acac5bedd7daeaa6c6673dcd6\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/7f94a3cf31ef2d619305b6080a81ad511eb91dda"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/7f94a3cf31ef2d619305b6080a81ad511eb91dda"}]},"branch":"refs/heads/master"},"15b68c15c1dd0c2df1cc49b550f2fa03a4c65e8e":{"kind":"NO_CODE_CHANGE","_number":4,"created":"2020-01-31 09:20:55.000000000","uploader":{"_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"},"ref":"refs/changes/80/682280/4","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/kolla","ref":"refs/changes/80/682280/4","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/kolla refs/changes/80/682280/4 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/kolla refs/changes/80/682280/4 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/kolla refs/changes/80/682280/4 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/kolla refs/changes/80/682280/4"}}},"commit":{"parents":[{"commit":"a68223412e6e3289173adca1af760157b25121c2","subject":"Deprecate OpenDaylight","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/a68223412e6e3289173adca1af760157b25121c2"}]}],"author":{"name":"chenxing","email":"chason.chan@foxmail.com","date":"2019-09-16 06:46:41.000000000","tz":480},"committer":{"name":"Radosław Piliszek","email":"radoslaw.piliszek@gmail.com","date":"2020-01-31 09:19:55.000000000","tz":60},"subject":"nova-libvirt: add UEFI packages to support UEFI instances","message":"nova-libvirt: add UEFI packages to support UEFI instances\n\nFix inability to run UEFI-based images/instances by installing UEFI\npackages also in nova-libvirt image which is not based on nova-base.\n\nIncludes support for C8.\nBackport below Train w/o C8.\n\nCloses-Bug: #1814552\nCo-authored-by: Marcin Juszkiewicz \u003cmarcin.juszkiewicz@linaro.org\u003e\nCo-authored-by: Radosław Piliszek \u003cradoslaw.piliszek@gmail.com\u003e\nChange-Id: I1d5cd3d9af98444acac5bedd7daeaa6c6673dcd6\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/15b68c15c1dd0c2df1cc49b550f2fa03a4c65e8e"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/15b68c15c1dd0c2df1cc49b550f2fa03a4c65e8e"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[],"submit_requirements":[]}
