)]}'
{"id":"openstack%2Fkolla~682350","triplet_id":"openstack%2Fkolla~master~Iaa9bd337ad200340110c50c73c6f0b997204bfe1","project":"openstack/kolla","branch":"master","topic":"remove-systemd-sysv-init-bin-from-os-base","hashtags":[],"change_id":"Iaa9bd337ad200340110c50c73c6f0b997204bfe1","subject":"Remove systemd-sysv package from openstack-base for RPM-based","status":"ABANDONED","created":"2019-09-16 12:02:11.000000000","updated":"2020-02-21 11:27:23.000000000","total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"meta_rev_id":"2a456896f3175a998c660184f7ba2b761b1e40aa","_number":682350,"virtual_id_number":682350,"owner":{"_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"},"actions":{},"labels":{"Verified":{"recommended":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"},{"value":1,"date":"2019-09-16 19:01:29.000000000","permitted_voting_range":{"min":-2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"date":"2019-09-16 17:26:47.000000000","_account_id":23181,"name":"RDO Third Party CI","email":"dmsimard+rdothirdparty@redhat.com","username":"rdothirdparty","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","value":1,"default_value":0,"optional":true},"Code-Review":{"all":[{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"},{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":23181,"name":"RDO Third Party CI","email":"dmsimard+rdothirdparty@redhat.com","username":"rdothirdparty","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"rejected":{"_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"},"all":[{"value":-1,"date":"2019-09-16 12:25:07.000000000","permitted_voting_range":{"min":-1,"max":0},"_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":23181,"name":"RDO Third Party CI","email":"dmsimard+rdothirdparty@redhat.com","username":"rdothirdparty","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Backport-Candidate":{"all":[{"_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":23181,"name":"RDO Third Party CI","email":"dmsimard+rdothirdparty@redhat.com","username":"rdothirdparty","tags":["SERVICE_USER"]}],"values":{"-1":"Do Not Backport"," 0":"Backport Review Needed","+1":"Should Backport"},"description":"","default_value":0,"optional":true},"Review-Priority":{"all":[{"_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":23181,"name":"RDO Third Party CI","email":"dmsimard+rdothirdparty@redhat.com","username":"rdothirdparty","tags":["SERVICE_USER"]}],"values":{"-1":"Branch Freeze"," 0":"No Priority","+1":"Important Change","+2":"Gate Blocker Fix / Urgent Change"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":23181,"name":"RDO Third Party CI","email":"dmsimard+rdothirdparty@redhat.com","username":"rdothirdparty","tags":["SERVICE_USER"]},{"_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2019-09-16 17:26:47.000000000","updated_by":{"_account_id":23181,"name":"RDO Third Party CI","email":"dmsimard+rdothirdparty@redhat.com","username":"rdothirdparty","tags":["SERVICE_USER"]},"reviewer":{"_account_id":23181,"name":"RDO Third Party CI","email":"dmsimard+rdothirdparty@redhat.com","username":"rdothirdparty","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2019-09-16 19:01:29.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"}],"messages":[{"id":"32ca9f7ab18b69b7aa012b2c72395759a46b2730","author":{"_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"},"date":"2019-09-16 12:02:11.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"ed74d47dc6e5bfecbeba8b111ad391f1a076a035","author":{"_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"},"date":"2019-09-16 12:02:56.000000000","message":"Patch Set 1:\n\nlet\u0027s see the impact now\n\nwonder if bifrost should not just be built from base rather than openstack-base anyway","accounts_in_message":[],"_revision_number":1},{"id":"6ed3f8c439b1c38b81b3d683335c0df53df03897","author":{"_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"},"date":"2019-09-16 12:24:59.000000000","message":"Patch Set 1:\n\nTODO: check if it\u0027s downloaded anyway in subimages","accounts_in_message":[],"_revision_number":1},{"id":"a6fc1ff55af7b0121fb9292368cd39868d33fdb5","author":{"_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"},"date":"2019-09-16 12:25:07.000000000","message":"Patch Set 1: Workflow-1","accounts_in_message":[],"_revision_number":1},{"id":"52a16d80a145fcae285667c1a325669c29d25d51","author":{"_account_id":23181,"name":"RDO Third Party CI","email":"dmsimard+rdothirdparty@redhat.com","username":"rdothirdparty","tags":["SERVICE_USER"]},"date":"2019-09-16 12:48:30.000000000","message":"Patch Set 1:\n\nBuild succeeded (check pipeline).\n\n- tripleo-build-containers-rhel-8-master http://logs.rdoproject.org/50/682350/1/openstack-check/tripleo-build-containers-rhel-8-master/292f8fd/ : SUCCESS in 42m 15s","accounts_in_message":[],"_revision_number":1},{"id":"ef3e066c209a805767fbff55e1236041b45ae0f8","author":{"_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"},"date":"2019-09-16 16:40:17.000000000","message":"Patch Set 1:\n\nrecheck\n\nZuul dropped the jobs","accounts_in_message":[],"_revision_number":1},{"id":"2b3ce45c92b435ceff5934718b50b099563ce4e6","author":{"_account_id":23181,"name":"RDO Third Party CI","email":"dmsimard+rdothirdparty@redhat.com","username":"rdothirdparty","tags":["SERVICE_USER"]},"date":"2019-09-16 17:26:47.000000000","message":"Patch Set 1:\n\nBuild succeeded (check pipeline).\n\n- tripleo-build-containers-rhel-8-master http://logs.rdoproject.org/50/682350/1/openstack-check/tripleo-build-containers-rhel-8-master/3d7fcc4/ : SUCCESS in 45m 22s","accounts_in_message":[],"_revision_number":1},{"id":"08eef1542cd41f9434f145a2cbcd73f6ee8d3f81","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-09-16 19:01:29.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\n\n- kolla-build-centos-binary https://zuul.opendev.org/t/openstack/build/a3334af188c041cba3c032ab73dd3e58 : SUCCESS in 34m 49s (non-voting)\n- kolla-build-centos-source https://zuul.opendev.org/t/openstack/build/75ea7bcf4593405caa3152b467fe6f10 : SUCCESS in 50m 34s\n- kolla-ansible-centos-source https://zuul.opendev.org/t/openstack/build/c357f3b3a8954575b022a8891a697333 : SUCCESS in 1h 13m 44s\n- kolla-ansible-centos-binary https://zuul.opendev.org/t/openstack/build/c12d23b5ae984b4e82eb0969246b4cfc : SUCCESS in 1h 21m 40s (non-voting)\n- kolla-ansible-centos-source-upgrade https://zuul.opendev.org/t/openstack/build/1c7dcabc61564624afb14b58205576c2 : SUCCESS in 1h 32m 38s\n- tripleo-build-containers-centos-7 https://zuul.opendev.org/t/openstack/build/bc142af679794deb8c36008f535b2b66 : SUCCESS in 49m 23s\n- kolla-ansible-bifrost-centos-source https://zuul.opendev.org/t/openstack/build/5aa41a7aed6a46bb9cac854a425594e3 : FAILURE in 31m 04s (non-voting)\n- kolla-ansible-centos-source-ceph https://zuul.opendev.org/t/openstack/build/4d5446b9fe454761bc5c56009ad52ced : SUCCESS in 1h 52m 47s (non-voting)\n- kolla-ansible-centos-source-upgrade-ceph https://zuul.opendev.org/t/openstack/build/9b09cf738f4044fb9ee3587a3c95f099 : SUCCESS in 2h 12m 54s (non-voting)\n- kolla-ansible-centos-source-cinder-lvm https://zuul.opendev.org/t/openstack/build/518cc4340eed49c09b8487925a1a9ae6 : SUCCESS in 1h 33m 11s (non-voting)\n- kolla-ansible-centos-source-scenario-nfv https://zuul.opendev.org/t/openstack/build/c984b1c550ba4e9f9ffebca1178fa65d : SUCCESS in 1h 36m 14s (non-voting)\n- kolla-ansible-centos-source-ironic https://zuul.opendev.org/t/openstack/build/5f8ec349801c4d14b3d737b2f66033dc : SUCCESS in 1h 27m 39s (non-voting)\n- kolla-build-debian-source https://zuul.opendev.org/t/openstack/build/78b57a6fbc2940d586a7d34d0104050a : SUCCESS in 29m 31s (non-voting)\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/377ed0773d6e494e86197f9dc74a828d : SUCCESS in 4m 55s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/6f65187211ef45949a60939e44dd36ea : SUCCESS in 5m 13s\n- openstack-tox-py27 https://zuul.opendev.org/t/openstack/build/0c4832e907fc48468b066d72274be8d9 : SUCCESS in 5m 01s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/dafcd9bca2724dca8521151abcb9484b : SUCCESS in 2m 51s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/66e4659aac1c49969bfe6a918ca57b12 : SUCCESS in 3m 16s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/812ef854aac4403ca4bff6be3f2decab : SUCCESS in 4m 00s\n- kolla-build-ubuntu-binary https://zuul.opendev.org/t/openstack/build/6493e6392ce44c7a8881b830af9b55a0 : SUCCESS in 23m 46s (non-voting)\n- kolla-build-ubuntu-source https://zuul.opendev.org/t/openstack/build/53d19aa4973a40d39f221bb66b96598f : SUCCESS in 28m 29s\n- kolla-ansible-ubuntu-source https://zuul.opendev.org/t/openstack/build/9e5a22b3df1f4d228ddf185e7d707bb7 : SUCCESS in 48m 56s\n- kolla-ansible-ubuntu-binary https://zuul.opendev.org/t/openstack/build/1969046912254db5bf74474efb0ab845 : SUCCESS in 52m 14s (non-voting)\n- kolla-ansible-ubuntu-source-upgrade https://zuul.opendev.org/t/openstack/build/1e50cd62ec89485f922ec475a6821c4a : SUCCESS in 58m 12s\n- kolla-ansible-ubuntu-source-zun https://zuul.opendev.org/t/openstack/build/c806653024a74550b3d6163379d789c2 : SUCCESS in 1h 03m 38s (non-voting)\n- kolla-ansible-ubuntu-source-ceph https://zuul.opendev.org/t/openstack/build/6c20f63614b24486898037aeabdd4090 : SUCCESS in 1h 38m 19s (non-voting)\n- kolla-ansible-ubuntu-source-upgrade-ceph https://zuul.opendev.org/t/openstack/build/3dcaf512c1054516ad78c23c076e9436 : SUCCESS in 1h 41m 38s (non-voting)\n- kolla-ansible-ubuntu-source-cinder-lvm https://zuul.opendev.org/t/openstack/build/02009484ffcb40e6a45e42099dc902ed : SUCCESS in 1h 09m 45s (non-voting)\n- kolla-ansible-ubuntu-source-ironic https://zuul.opendev.org/t/openstack/build/27f6ffe9b0984e848506b5c93e669c89 : SUCCESS in 1h 11m 22s (non-voting)","accounts_in_message":[],"_revision_number":1},{"id":"af77d8f36c8168f1d82f4d8d8d46c62c189c11a5","author":{"_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"},"date":"2020-02-21 11:27:23.000000000","message":"Abandoned\n\nobscure change","accounts_in_message":[],"_revision_number":1}],"current_revision_number":1,"current_revision":"64e3069400683a99ff814b864aa2cbb2cd721f9d","revisions":{"64e3069400683a99ff814b864aa2cbb2cd721f9d":{"kind":"REWORK","_number":1,"created":"2019-09-16 12:02:11.000000000","uploader":{"_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"},"ref":"refs/changes/50/682350/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/kolla","ref":"refs/changes/50/682350/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/kolla refs/changes/50/682350/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/kolla refs/changes/50/682350/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/kolla refs/changes/50/682350/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/kolla refs/changes/50/682350/1"}}},"commit":{"parents":[{"commit":"b98182d9e9146144dfa711f6ee91bcca5d2708b1","subject":"Merge \"Remove unneeded Zuul branch matcher\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/b98182d9e9146144dfa711f6ee91bcca5d2708b1"}]}],"author":{"name":"Radosław Piliszek","email":"radoslaw.piliszek@gmail.com","date":"2019-09-16 12:01:32.000000000","tz":120},"committer":{"name":"Radosław Piliszek","email":"radoslaw.piliszek@gmail.com","date":"2019-09-16 12:01:32.000000000","tz":120},"subject":"Remove systemd-sysv package from openstack-base for RPM-based","message":"Remove systemd-sysv package from openstack-base for RPM-based\n\nChange-Id: Iaa9bd337ad200340110c50c73c6f0b997204bfe1\nSigned-off-by: Radosław Piliszek \u003cradoslaw.piliszek@gmail.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/64e3069400683a99ff814b864aa2cbb2cd721f9d"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/64e3069400683a99ff814b864aa2cbb2cd721f9d"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[],"submit_requirements":[]}
