)]}'
{"id":"openstack%2Fkolla~682436","triplet_id":"openstack%2Fkolla~stable%2Frocky~Iebc1f2c05c1f57c2b6f98ca9772f1ae9d9a420fa","project":"openstack/kolla","branch":"stable/rocky","hashtags":[],"change_id":"Iebc1f2c05c1f57c2b6f98ca9772f1ae9d9a420fa","subject":"Add pciutils package to zun-compute","status":"MERGED","created":"2019-09-16 17:29:09.000000000","updated":"2019-09-17 08:55:11.000000000","submitted":"2019-09-17 08:55:11.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"682436-1568710511665-06954540","meta_rev_id":"2b2d6b2543fa0f3985588fa44a0ec2d8764d9f28","_number":682436,"virtual_id_number":682436,"owner":{"_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":19316,"name":"Eduardo Gonzalez","email":"dabarren@gmail.com","username":"egonzalez90"},{"value":0,"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},{"value":0,"_account_id":24072,"name":"Marcin Juszkiewicz","email":"mjuszkiewicz@redhat.com","username":"hrw"},{"value":0,"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},{"value":0,"_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"},{"value":2,"date":"2019-09-17 08:55:11.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":19316,"name":"Eduardo Gonzalez","email":"dabarren@gmail.com","username":"egonzalez90"},"all":[{"value":2,"date":"2019-09-17 06:46:52.000000000","_account_id":19316,"name":"Eduardo Gonzalez","email":"dabarren@gmail.com","username":"egonzalez90"},{"value":0,"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},{"value":0,"_account_id":24072,"name":"Marcin Juszkiewicz","email":"mjuszkiewicz@redhat.com","username":"hrw"},{"value":2,"date":"2019-09-17 08:10:09.000000000","_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},{"value":0,"_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},"all":[{"value":0,"_account_id":19316,"name":"Eduardo Gonzalez","email":"dabarren@gmail.com","username":"egonzalez90"},{"value":0,"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},{"value":0,"_account_id":24072,"name":"Marcin Juszkiewicz","email":"mjuszkiewicz@redhat.com","username":"hrw"},{"value":1,"date":"2019-09-17 08:10:09.000000000","_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},{"value":0,"_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Backport-Candidate":{"all":[{"value":0,"_account_id":19316,"name":"Eduardo Gonzalez","email":"dabarren@gmail.com","username":"egonzalez90"},{"value":0,"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},{"value":0,"_account_id":24072,"name":"Marcin Juszkiewicz","email":"mjuszkiewicz@redhat.com","username":"hrw"},{"value":0,"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},{"value":0,"_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Do Not Backport"," 0":"Backport Review Needed","+1":"Should Backport"},"description":"","default_value":0,"optional":true},"Review-Priority":{"recommended":{"_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"},"all":[{"value":0,"_account_id":19316,"name":"Eduardo Gonzalez","email":"dabarren@gmail.com","username":"egonzalez90"},{"value":0,"date":"2019-09-16 17:52:56.000000000","permitted_voting_range":{"min":0,"max":2},"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},{"value":0,"date":"2019-09-16 17:52:59.000000000","_account_id":24072,"name":"Marcin Juszkiewicz","email":"mjuszkiewicz@redhat.com","username":"hrw"},{"value":0,"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},{"value":1,"date":"2019-09-16 17:52:37.000000000","_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Branch Freeze"," 0":"No Priority","+1":"Important Change","+2":"Gate Blocker Fix / Urgent Change"},"description":"","value":1,"default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},{"_account_id":19316,"name":"Eduardo Gonzalez","email":"dabarren@gmail.com","username":"egonzalez90"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},{"_account_id":24072,"name":"Marcin Juszkiewicz","email":"mjuszkiewicz@redhat.com","username":"hrw"},{"_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2019-09-16 17:52:56.000000000","updated_by":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"reviewer":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"state":"REVIEWER"},{"updated":"2019-09-16 17:52:59.000000000","updated_by":{"_account_id":24072,"name":"Marcin Juszkiewicz","email":"mjuszkiewicz@redhat.com","username":"hrw"},"reviewer":{"_account_id":24072,"name":"Marcin Juszkiewicz","email":"mjuszkiewicz@redhat.com","username":"hrw"},"state":"REVIEWER"},{"updated":"2019-09-17 06:46:52.000000000","updated_by":{"_account_id":19316,"name":"Eduardo Gonzalez","email":"dabarren@gmail.com","username":"egonzalez90"},"reviewer":{"_account_id":19316,"name":"Eduardo Gonzalez","email":"dabarren@gmail.com","username":"egonzalez90"},"state":"REVIEWER"},{"updated":"2019-09-17 08:10:09.000000000","updated_by":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},"reviewer":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},"state":"REVIEWER"},{"updated":"2019-09-17 08:55:11.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"}],"messages":[{"id":"59d218b961459ae4c3f0ea41abeee9413321b005","author":{"_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"},"date":"2019-09-16 17:29:09.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"308082a2cddd5c3f61c386db18a4cc62a0061bfe","author":{"_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"},"date":"2019-09-16 17:52:37.000000000","message":"Patch Set 1: Review-Priority+1","accounts_in_message":[],"_revision_number":1},{"id":"c12db391ad96f00eecf48ee6b2094a7c487437d4","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-09-16 19:37:21.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\n\n- kolla-build-centos-binary https://zuul.opendev.org/t/openstack/build/bbbe1026d057497184487eee9f79b80f : SUCCESS in 42m 00s (non-voting)\n- kolla-build-centos-source https://zuul.opendev.org/t/openstack/build/6c46624c9aa04e13bad025f5f4ff166a : SUCCESS in 36m 59s\n- kolla-ansible-centos-source https://zuul.opendev.org/t/openstack/build/866ea6e2aa284cc7b89de78682e58ee0 : SUCCESS in 1h 11m 29s\n- kolla-ansible-centos-binary https://zuul.opendev.org/t/openstack/build/88081fd594534dd49a67b910008f56cf : SUCCESS in 1h 02m 02s (non-voting)\n- kolla-build-oraclelinux-binary https://zuul.opendev.org/t/openstack/build/d624a02194d64fc4b366d704102f9117 : SUCCESS in 51m 12s (non-voting)\n- kolla-build-oraclelinux-source https://zuul.opendev.org/t/openstack/build/3427592128a649fb809da72e0bffbcac : SUCCESS in 47m 01s\n- kolla-ansible-oraclelinux-source https://zuul.opendev.org/t/openstack/build/292c3e14850643018871146abbbe260c : SUCCESS in 1h 16m 31s (non-voting)\n- kolla-ansible-oraclelinux-binary https://zuul.opendev.org/t/openstack/build/d05fb3c523624f028b00bf25f33ef5d2 : SUCCESS in 1h 14m 01s (non-voting)\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/04f448cfdd3242bd9677b391d4ef66b7 : SUCCESS in 3m 43s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/841f4e59403646cbbce209554d1a2e5a : SUCCESS in 5m 02s\n- openstack-tox-py27 https://zuul.opendev.org/t/openstack/build/db1a6d494e534047b4787f303f798fa4 : SUCCESS in 3m 23s\n- openstack-tox-py35 https://zuul.opendev.org/t/openstack/build/15beb40901044906854057e17360f07f : SUCCESS in 3m 37s\n- build-openstack-sphinx-docs https://zuul.opendev.org/t/openstack/build/63ec445f80154b21851ccec7a51ef607 : SUCCESS in 5m 24s\n- kolla-build-ubuntu-binary https://zuul.opendev.org/t/openstack/build/0074a37b5a954d1c8fef17bf5a99b845 : SUCCESS in 1h 22m 30s (non-voting)\n- kolla-build-ubuntu-source https://zuul.opendev.org/t/openstack/build/e2158a2727ff4f058a37ff071de17b8e : SUCCESS in 37m 28s\n- kolla-ansible-ubuntu-source https://zuul.opendev.org/t/openstack/build/1f87dbfaa0e24f2c8945bfe4df869dac : SUCCESS in 1h 05m 43s\n- kolla-ansible-ubuntu-binary https://zuul.opendev.org/t/openstack/build/01b4028034ac467ea936e8114b07ce27 : SUCCESS in 1h 59m 32s (non-voting)","accounts_in_message":[],"_revision_number":1},{"id":"5e1aae2dd5340b09338c90b770cd7a1348807206","author":{"_account_id":19316,"name":"Eduardo Gonzalez","email":"dabarren@gmail.com","username":"egonzalez90"},"date":"2019-09-17 06:46:52.000000000","message":"Patch Set 1: Code-Review+2","accounts_in_message":[],"_revision_number":1},{"id":"d70a9587e2b65cb7b28f1fe61cdcd65b7f74300c","author":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},"date":"2019-09-17 08:10:09.000000000","message":"Patch Set 1: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":1},{"id":"584fde828e9a3881decd661911289d832f05de94","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-09-17 08:10:23.000000000","message":"Patch Set 1: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":1},{"id":"99e40c41ab516f36705ff66907e6a9b741f1d33f","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-09-17 08:55:11.000000000","message":"Change has been successfully merged by Zuul","accounts_in_message":[],"_revision_number":1},{"id":"2b2d6b2543fa0f3985588fa44a0ec2d8764d9f28","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-09-17 08:55:11.000000000","message":"Patch Set 1: Verified+2\n\nBuild succeeded (gate pipeline).\n\n- kolla-build-centos-source https://zuul.opendev.org/t/openstack/build/f10df2d77f7145f0b2a3d3b3c9e9e2f2 : SUCCESS in 39m 51s\n- kolla-build-oraclelinux-source https://zuul.opendev.org/t/openstack/build/01708ea59f7f40d9a07b9d8a092cd4d8 : SUCCESS in 42m 37s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/1391cbdbc71f4d92a3471d183ab10ecf : SUCCESS in 5m 08s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/13040f219f1446cdbd21bd910f67a7a7 : SUCCESS in 3m 55s\n- openstack-tox-py27 https://zuul.opendev.org/t/openstack/build/3252e1a5f8f54acebd29eac44cbf24a2 : SUCCESS in 3m 27s\n- openstack-tox-py35 https://zuul.opendev.org/t/openstack/build/d6f0a69af055497e983b97bdff99ceaa : SUCCESS in 4m 00s\n- build-openstack-sphinx-docs https://zuul.opendev.org/t/openstack/build/8b5050244f254812aabb955677680bea : SUCCESS in 3m 42s\n- kolla-build-ubuntu-source https://zuul.opendev.org/t/openstack/build/6659c44c4e534677a1c335064b46170b : SUCCESS in 41m 09s","accounts_in_message":[],"_revision_number":1}],"current_revision_number":1,"current_revision":"8b6a7845f331a39e6eaef899b0d2a3ccc7336379","revisions":{"8b6a7845f331a39e6eaef899b0d2a3ccc7336379":{"kind":"REWORK","_number":1,"created":"2019-09-16 17:29:09.000000000","uploader":{"_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"},"ref":"refs/changes/36/682436/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/kolla","ref":"refs/changes/36/682436/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/kolla refs/changes/36/682436/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/kolla refs/changes/36/682436/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/kolla refs/changes/36/682436/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/kolla refs/changes/36/682436/1"}}},"commit":{"parents":[{"commit":"a3419f060e1aa8c3c9f59fbd9b71595841a8287d","subject":"Add config option use_dumb_init to add it to container images","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/a3419f060e1aa8c3c9f59fbd9b71595841a8287d"}]}],"author":{"name":"Radosław Piliszek","email":"radoslaw.piliszek@gmail.com","date":"2019-08-16 13:13:03.000000000","tz":120},"committer":{"name":"Radosław Piliszek","email":"radoslaw.piliszek@gmail.com","date":"2019-09-16 17:26:03.000000000","tz":120},"subject":"Add pciutils package to zun-compute","message":"Add pciutils package to zun-compute\n\nIt runs lspci and has this requirement in its bindep.\n\nCentOS images seem to have it installed but Ubuntu not.\nEnsure pciutils package is always installed.\n\nzun-compute seemed to work nonetheless but generated errors\nabout being unable to take host inventory.\n(now we know it would fail for SRIOV)\n\nChange adapted for Rocky by ignoring cyborg (not present).\n\nRelated-bug: #1797448\nChange-Id: Iebc1f2c05c1f57c2b6f98ca9772f1ae9d9a420fa\nSigned-off-by: Radosław Piliszek \u003cradoslaw.piliszek@gmail.com\u003e\n(cherry picked from commit c323c5be7fd4ebcc71ca4e10a5c15b9ab1f29ba6)\n(cherry picked from commit 904f1fcbd7b1bf133547980eed7807b06a51ef3b)\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/8b6a7845f331a39e6eaef899b0d2a3ccc7336379"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/8b6a7845f331a39e6eaef899b0d2a3ccc7336379"}]},"branch":"refs/heads/stable/rocky"}},"requirements":[],"submit_records":[],"submit_requirements":[]}
