)]}'
{"id":"openstack%2Fkolla~682600","triplet_id":"openstack%2Fkolla~master~Ic7494eff4c740b6e528da4508dc88a64e697bc04","project":"openstack/kolla","branch":"master","topic":"ci-override-mirrors","hashtags":[],"change_id":"Ic7494eff4c740b6e528da4508dc88a64e697bc04","subject":"CI: CentOS - override mirrors ASAP + crmsh mirror override","status":"MERGED","created":"2019-09-17 10:17:34.000000000","updated":"2019-09-18 09:34:07.000000000","submitted":"2019-09-18 09:30:59.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":3,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"682600-1568799060238-ab427080","meta_rev_id":"11a02205593bc9bd666dfd1f2077639f38925b88","_number":682600,"virtual_id_number":682600,"owner":{"_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},{"value":0,"_account_id":24072,"name":"Marcin Juszkiewicz","email":"mjuszkiewicz@redhat.com","username":"hrw"},{"value":0,"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},{"value":0,"_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"},{"value":2,"date":"2019-09-18 09:30:59.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"date":"2019-09-17 18:20:40.000000000","_account_id":23181,"name":"RDO Third Party CI","email":"dmsimard+rdothirdparty@redhat.com","username":"rdothirdparty","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"all":[{"value":2,"date":"2019-09-18 07:48:50.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},{"value":2,"date":"2019-09-18 07:57:51.000000000","_account_id":24072,"name":"Marcin Juszkiewicz","email":"mjuszkiewicz@redhat.com","username":"hrw"},{"value":0,"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},{"value":0,"_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":23181,"name":"RDO Third Party CI","email":"dmsimard+rdothirdparty@redhat.com","username":"rdothirdparty","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":24072,"name":"Marcin Juszkiewicz","email":"mjuszkiewicz@redhat.com","username":"hrw"},"all":[{"value":0,"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},{"value":1,"date":"2019-09-18 07:57:51.000000000","_account_id":24072,"name":"Marcin Juszkiewicz","email":"mjuszkiewicz@redhat.com","username":"hrw"},{"value":0,"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},{"value":0,"_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":23181,"name":"RDO Third Party CI","email":"dmsimard+rdothirdparty@redhat.com","username":"rdothirdparty","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Backport-Candidate":{"all":[{"value":0,"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},{"value":0,"_account_id":24072,"name":"Marcin Juszkiewicz","email":"mjuszkiewicz@redhat.com","username":"hrw"},{"value":0,"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},{"value":0,"_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":23181,"name":"RDO Third Party CI","email":"dmsimard+rdothirdparty@redhat.com","username":"rdothirdparty","tags":["SERVICE_USER"]}],"values":{"-1":"Do Not Backport"," 0":"Backport Review Needed","+1":"Should Backport"},"description":"","default_value":0,"optional":true},"Review-Priority":{"approved":{"_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"},"all":[{"value":0,"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},{"value":0,"_account_id":24072,"name":"Marcin Juszkiewicz","email":"mjuszkiewicz@redhat.com","username":"hrw"},{"value":0,"date":"2019-09-18 06:40:46.000000000","_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},{"value":2,"date":"2019-09-18 06:40:37.000000000","_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":23181,"name":"RDO Third Party CI","email":"dmsimard+rdothirdparty@redhat.com","username":"rdothirdparty","tags":["SERVICE_USER"]}],"values":{"-1":"Branch Freeze"," 0":"No Priority","+1":"Important Change","+2":"Gate Blocker Fix / Urgent Change"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},{"_account_id":23181,"name":"RDO Third Party CI","email":"dmsimard+rdothirdparty@redhat.com","username":"rdothirdparty","tags":["SERVICE_USER"]},{"_account_id":24072,"name":"Marcin Juszkiewicz","email":"mjuszkiewicz@redhat.com","username":"hrw"},{"_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2019-09-17 18:20:40.000000000","updated_by":{"_account_id":23181,"name":"RDO Third Party CI","email":"dmsimard+rdothirdparty@redhat.com","username":"rdothirdparty","tags":["SERVICE_USER"]},"reviewer":{"_account_id":23181,"name":"RDO Third Party CI","email":"dmsimard+rdothirdparty@redhat.com","username":"rdothirdparty","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2019-09-18 06:40:46.000000000","updated_by":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},"reviewer":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},"state":"REVIEWER"},{"updated":"2019-09-18 07:48:50.000000000","updated_by":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"reviewer":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"state":"REVIEWER"},{"updated":"2019-09-18 07:57:51.000000000","updated_by":{"_account_id":24072,"name":"Marcin Juszkiewicz","email":"mjuszkiewicz@redhat.com","username":"hrw"},"reviewer":{"_account_id":24072,"name":"Marcin Juszkiewicz","email":"mjuszkiewicz@redhat.com","username":"hrw"},"state":"REVIEWER"},{"updated":"2019-09-18 09:30:59.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"}],"messages":[{"id":"adafefc036fe9779c15e3eef766d33f76dabface","author":{"_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"},"date":"2019-09-17 10:17:34.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"a23591e7e0d1d5ae731fe5e96c7ec2c0c9f9eec8","author":{"_account_id":24072,"name":"Marcin Juszkiewicz","email":"mjuszkiewicz@redhat.com","username":"hrw"},"date":"2019-09-17 10:33:42.000000000","message":"Patch Set 1: Code-Review+1\n\n(2 comments)","accounts_in_message":[],"_revision_number":1},{"id":"c0e2b4408053df7eabb888f0e77593f9d65ecf6a","author":{"_account_id":23181,"name":"RDO Third Party CI","email":"dmsimard+rdothirdparty@redhat.com","username":"rdothirdparty","tags":["SERVICE_USER"]},"date":"2019-09-17 10:56:55.000000000","message":"Patch Set 1:\n\nBuild succeeded (check pipeline).\n\n- tripleo-build-containers-rhel-8-master http://logs.rdoproject.org/00/682600/1/openstack-check/tripleo-build-containers-rhel-8-master/eb1930b/ : SUCCESS in 38m 18s","accounts_in_message":[],"_revision_number":1},{"id":"e5cbf0418ff4ce6fd07b18784c7daf8f3d10e7fa","author":{"_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"},"date":"2019-09-17 11:05:02.000000000","message":"Patch Set 1:\n\n(1 comment)","accounts_in_message":[],"_revision_number":1},{"id":"372d71777ef2535ebf50f11017276606afc3878d","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-09-17 12:21:04.000000000","message":"Patch Set 1: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttp://docs.openstack.org/infra/manual/developers.html#automated-testing\n\n\n- kolla-build-centos-binary https://zuul.opendev.org/t/openstack/build/5c98e13da064440dbfd5cc3a1cca2aa5 : FAILURE in 5m 53s (non-voting)\n- kolla-build-centos-source https://zuul.opendev.org/t/openstack/build/34ca7d2ff36249eb906969e0de46bd32 : FAILURE in 6m 49s\n- kolla-ansible-centos-source https://zuul.opendev.org/t/openstack/build/6125c990e76a459ebded730384ef4b5e : FAILURE in 13m 42s\n- kolla-ansible-centos-binary https://zuul.opendev.org/t/openstack/build/67f1c43839544bc3aebf3cf0f54d9ca0 : FAILURE in 11m 54s (non-voting)\n- kolla-ansible-centos-source-upgrade https://zuul.opendev.org/t/openstack/build/d56a91e64e114d648e16c63aac1f4895 : FAILURE in 14m 00s\n- tripleo-build-containers-centos-7 https://zuul.opendev.org/t/openstack/build/b1fdeb11fc8445debbb81ee80c9f7ccf : SUCCESS in 39m 44s\n- kolla-ansible-bifrost-centos-source https://zuul.opendev.org/t/openstack/build/86d5536c468c437498fb200602a7cf9e : FAILURE in 10m 10s (non-voting)\n- kolla-ansible-centos-source-ceph https://zuul.opendev.org/t/openstack/build/7606c9fe24324c9f970661a123675d71 : FAILURE in 12m 46s (non-voting)\n- kolla-ansible-centos-source-upgrade-ceph https://zuul.opendev.org/t/openstack/build/f70a8a38409e4e15ba486bd9b0c621fc : FAILURE in 18m 38s (non-voting)\n- kolla-ansible-centos-source-cinder-lvm https://zuul.opendev.org/t/openstack/build/c9e5e3b37930446d87f433ef6ce519ef : FAILURE in 12m 08s (non-voting)\n- kolla-ansible-centos-source-scenario-nfv https://zuul.opendev.org/t/openstack/build/ef75e987b75a43c28ff6b254dc828d2a : FAILURE in 10m 36s (non-voting)\n- kolla-ansible-centos-source-ironic https://zuul.opendev.org/t/openstack/build/bb551291516646ecac852deb53d71697 : FAILURE in 9m 50s (non-voting)\n- kolla-build-debian-source https://zuul.opendev.org/t/openstack/build/bf0581fe6ef9445b89565136a1d491be : SUCCESS in 20m 53s (non-voting)\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/05a961b2fad24f98bdaf841e80b84332 : RETRY_LIMIT in 2m 07s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/d83ec0bdee0148b2bb5f9d0bb55db953 : RETRY_LIMIT in 1m 35s\n- openstack-tox-py27 https://zuul.opendev.org/t/openstack/build/7ed08f91f6f1439d96affe6c0a8b23be : SUCCESS in 4m 24s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/1f4b73d98ebc4913a91465cc8883f731 : RETRY_LIMIT in 2m 06s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/9203ce687d694d868ad1a1beedcd29e7 : RETRY_LIMIT in 2m 54s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/598b08408fa54b008530f2d7592e8448 : SUCCESS in 4m 05s\n- kolla-build-ubuntu-binary https://zuul.opendev.org/t/openstack/build/c1b0a5cd28a644d48b2db4778424318b : SUCCESS in 26m 37s (non-voting)\n- kolla-build-ubuntu-source https://zuul.opendev.org/t/openstack/build/db6b18ddb7744b2c8d9124973f2481a0 : SUCCESS in 29m 18s\n- kolla-ansible-ubuntu-source https://zuul.opendev.org/t/openstack/build/d3ecff927a15459bb2163260472f6069 : SUCCESS in 48m 48s\n- kolla-ansible-ubuntu-binary https://zuul.opendev.org/t/openstack/build/330978e9286f4fe78ef9568bb2f21b0c : SUCCESS in 54m 34s (non-voting)\n- kolla-ansible-ubuntu-source-upgrade https://zuul.opendev.org/t/openstack/build/b5514f0f68e24f559ac27de06a4ec1bd : SUCCESS in 58m 31s\n- kolla-ansible-ubuntu-source-zun https://zuul.opendev.org/t/openstack/build/85d7664b8c6c4826b50da8ff4bd29387 : SUCCESS in 1h 03m 49s (non-voting)\n- kolla-ansible-ubuntu-source-ceph https://zuul.opendev.org/t/openstack/build/a9fad7a7dbb24d5fa5133c0ddb0ffc72 : RETRY_LIMIT in 3m 10s (non-voting)\n- kolla-ansible-ubuntu-source-upgrade-ceph https://zuul.opendev.org/t/openstack/build/a13c6a021d9e43a79431539e3e50e70b : RETRY_LIMIT in 2m 25s (non-voting)\n- kolla-ansible-ubuntu-source-cinder-lvm https://zuul.opendev.org/t/openstack/build/f8d3dd5eff254e90996b41c57a8c1905 : SUCCESS in 1h 05m 10s (non-voting)\n- kolla-ansible-ubuntu-source-ironic https://zuul.opendev.org/t/openstack/build/81ba1d97e4d14cd8a11928e5cd4d4079 : SUCCESS in 1h 58m 58s (non-voting)","accounts_in_message":[],"_revision_number":1},{"id":"b750bc5af8a796332ac7712bb8e6c0b96eda3d36","author":{"_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"},"date":"2019-09-17 17:42:59.000000000","message":"Patch Set 2: Published edit on patch set 1.","accounts_in_message":[],"_revision_number":2},{"id":"10967561b0957de07eed49c7cff31a8cb1871633","author":{"_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"},"date":"2019-09-17 17:43:41.000000000","message":"Patch Set 2:\n\nraws added for completeness","accounts_in_message":[],"_revision_number":2},{"id":"1ef2c2d279f79666f981c58cf98f8104018dd3b2","author":{"_account_id":23181,"name":"RDO Third Party CI","email":"dmsimard+rdothirdparty@redhat.com","username":"rdothirdparty","tags":["SERVICE_USER"]},"date":"2019-09-17 18:20:40.000000000","message":"Patch Set 2:\n\nBuild succeeded (check pipeline).\n\n- tripleo-build-containers-rhel-8-master http://logs.rdoproject.org/00/682600/2/openstack-check/tripleo-build-containers-rhel-8-master/ddbcf22/ : SUCCESS in 36m 50s","accounts_in_message":[],"_revision_number":2},{"id":"d49d79e850dbbff6b941faea1da130f67c8771d9","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-09-17 22:06:05.000000000","message":"Patch Set 2: Verified+1\n\nBuild succeeded (check pipeline).\n\n- kolla-build-centos-binary https://zuul.opendev.org/t/openstack/build/831c64d6919a4a28b10e14043c19619a : SUCCESS in 48m 04s (non-voting)\n- kolla-build-centos-source https://zuul.opendev.org/t/openstack/build/e5e4dc5d7a384a1982476a9c6e919bcb : SUCCESS in 40m 07s\n- kolla-ansible-centos-source https://zuul.opendev.org/t/openstack/build/b2e130877003480db3327302aee33e60 : SUCCESS in 1h 29m 23s\n- kolla-ansible-centos-binary https://zuul.opendev.org/t/openstack/build/d09933857dda434391a830da4b2284cc : SUCCESS in 1h 23m 12s (non-voting)\n- kolla-ansible-centos-source-upgrade https://zuul.opendev.org/t/openstack/build/de2dbfd1093443e0b80a6eeb73129d55 : SUCCESS in 1h 26m 38s\n- tripleo-build-containers-centos-7 https://zuul.opendev.org/t/openstack/build/c927240187dc4d79a215accd6f826d6f : SUCCESS in 56m 22s\n- kolla-ansible-bifrost-centos-source https://zuul.opendev.org/t/openstack/build/e4304c0fd9654af38e9d27dfcd2d2025 : SUCCESS in 46m 55s (non-voting)\n- kolla-ansible-centos-source-ceph https://zuul.opendev.org/t/openstack/build/d9f6efe44b7c46319f0546cfaac188b0 : SUCCESS in 1h 31m 59s (non-voting)\n- kolla-ansible-centos-source-upgrade-ceph https://zuul.opendev.org/t/openstack/build/e82f845451964665b092224b950a9bbb : SUCCESS in 2h 12m 05s (non-voting)\n- kolla-ansible-centos-source-cinder-lvm https://zuul.opendev.org/t/openstack/build/139480a7aa344af1861ec3c1af8a88d6 : SUCCESS in 1h 42m 26s (non-voting)\n- kolla-ansible-centos-source-scenario-nfv https://zuul.opendev.org/t/openstack/build/5c8757e3c2354b8b93050efb122c5cf2 : SUCCESS in 1h 48m 13s (non-voting)\n- kolla-ansible-centos-source-ironic https://zuul.opendev.org/t/openstack/build/f0af67e9a7264df38958d168e663ebb5 : SUCCESS in 1h 24m 34s (non-voting)\n- kolla-build-debian-source https://zuul.opendev.org/t/openstack/build/5948772586244015ad7d9b9228bab26f : SUCCESS in 26m 01s (non-voting)\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/437f9025a8de424b968c181c42bdf25c : SUCCESS in 4m 45s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/af77ef0aedac4088bf8feac404a3ab5e : SUCCESS in 4m 58s\n- openstack-tox-py27 https://zuul.opendev.org/t/openstack/build/a3b3f7b1b14940598bac0bfab46f5d3b : SUCCESS in 4m 38s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/b4cd13a7d4af4f19a868fdbe9a1df062 : SUCCESS in 3m 41s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/9ea1a0f80a4249ffb9ff86a596b4e2cf : SUCCESS in 4m 07s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/5b6eb315ad7b4837932bd54768406a51 : SUCCESS in 3m 46s\n- kolla-build-ubuntu-binary https://zuul.opendev.org/t/openstack/build/ca216858bf1241159b4d2e3bba70b71b : SUCCESS in 30m 59s (non-voting)\n- kolla-build-ubuntu-source https://zuul.opendev.org/t/openstack/build/89738458a3714a3fb3b70aac409dd43a : SUCCESS in 32m 33s\n- kolla-ansible-ubuntu-source https://zuul.opendev.org/t/openstack/build/ef7c2529fc67423d8ce9d65996f3cf0b : SUCCESS in 50m 21s\n- kolla-ansible-ubuntu-binary https://zuul.opendev.org/t/openstack/build/f6cca2263bc6445d93f3be5e61ab1be6 : SUCCESS in 48m 39s (non-voting)\n- kolla-ansible-ubuntu-source-upgrade https://zuul.opendev.org/t/openstack/build/5eb27df689fa4f5d9242a4b37688b1b6 : SUCCESS in 57m 34s\n- kolla-ansible-ubuntu-source-zun https://zuul.opendev.org/t/openstack/build/7a94de4ec316474d97fc15a4f3d721d4 : SUCCESS in 1h 10m 07s (non-voting)\n- kolla-ansible-ubuntu-source-ceph https://zuul.opendev.org/t/openstack/build/ea9715b369684128a538b89f84f2a532 : SUCCESS in 1h 23m 00s (non-voting)\n- kolla-ansible-ubuntu-source-upgrade-ceph https://zuul.opendev.org/t/openstack/build/eb542b0ee5984d15a5b9fcfffcc0633c : SUCCESS in 1h 45m 55s (non-voting)\n- kolla-ansible-ubuntu-source-cinder-lvm https://zuul.opendev.org/t/openstack/build/7b24583d49364006b0227b3b49d88cf4 : SUCCESS in 1h 20m 32s (non-voting)\n- kolla-ansible-ubuntu-source-ironic https://zuul.opendev.org/t/openstack/build/55730141203149548f5c665523946ad0 : SUCCESS in 58m 11s (non-voting)","accounts_in_message":[],"_revision_number":2},{"id":"73d7bafcedc0ea33a5d3f62ff50f806e823845e1","author":{"_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"},"date":"2019-09-18 06:39:46.000000000","message":"Patch Set 2:\n\nI think we should rethink our approach to testing. 4h+ of testing does not look too good. ;/","accounts_in_message":[],"_revision_number":2},{"id":"2a322b1a139ee4279dbe66e970f387c2a7d5c586","author":{"_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"},"date":"2019-09-18 06:40:37.000000000","message":"Patch Set 2: Review-Priority+2","accounts_in_message":[],"_revision_number":2},{"id":"0e0f85363897dcaed72863823879a7b55e66a36d","author":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"date":"2019-09-18 07:48:50.000000000","message":"Patch Set 2: Code-Review+2\n\nWell - you have touched base, that\u0027s why we test everything :)","accounts_in_message":[],"_revision_number":2},{"id":"f9af1c8f89d74885d5737bf2f9636924a6f67b7c","author":{"_account_id":24072,"name":"Marcin Juszkiewicz","email":"mjuszkiewicz@redhat.com","username":"hrw"},"date":"2019-09-18 07:57:51.000000000","message":"Patch Set 2: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":2},{"id":"00d7066e5e3445f5d1c3b02413b987750e18b6b7","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-09-18 07:58:18.000000000","message":"Patch Set 2: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":2},{"id":"ef0c564aff9947ffdbdff8720894b0ba2c30948a","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-09-18 09:30:59.000000000","message":"Patch Set 2: Verified+2\n\nBuild succeeded (gate pipeline).\n\n- kolla-build-centos-source https://zuul.opendev.org/t/openstack/build/387f63c14b10463a9acc66a59a1a5db1 : SUCCESS in 59m 31s\n- kolla-ansible-centos-source https://zuul.opendev.org/t/openstack/build/cf3eb3b7a9924145bbaa0ab1f4c04aa3 : SUCCESS in 1h 25m 48s\n- kolla-ansible-centos-source-upgrade https://zuul.opendev.org/t/openstack/build/504d88fb62f5468a80c10434213add41 : SUCCESS in 1h 31m 00s\n- tripleo-build-containers-centos-7 https://zuul.opendev.org/t/openstack/build/00aeb24f943b4ca8af01af2bce2843d7 : SUCCESS in 54m 22s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/c08304f25d9e4f8ca953463558f9965c : SUCCESS in 3m 50s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/1c23e6918f6e44b49d93fcd3b8453ab8 : SUCCESS in 4m 58s\n- openstack-tox-py27 https://zuul.opendev.org/t/openstack/build/6959732130a546798b51bc34c095547e : SUCCESS in 3m 45s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/7c97138464b848a99b66f248096753a8 : SUCCESS in 3m 06s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/7407b7d0a98d4ecbb5d20439e56d8ce0 : SUCCESS in 3m 54s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/6b45e96d8c4c461aad18172dde4edc1b : SUCCESS in 4m 27s\n- kolla-build-ubuntu-source https://zuul.opendev.org/t/openstack/build/2cfbfece0cb94cb2b3f6ea78b0c3ef9a : SUCCESS in 28m 52s\n- kolla-ansible-ubuntu-source https://zuul.opendev.org/t/openstack/build/207a937047004edf98e3bde7b21b6366 : SUCCESS in 51m 24s\n- kolla-ansible-ubuntu-source-upgrade https://zuul.opendev.org/t/openstack/build/11699a87cd8d47bfbea28225d3cd3c36 : SUCCESS in 58m 10s","accounts_in_message":[],"_revision_number":2},{"id":"42771a4edc84607295afe520d3b7974d29a0976b","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-09-18 09:31:00.000000000","message":"Change has been successfully merged by Zuul","accounts_in_message":[],"_revision_number":2},{"id":"f451087351dcde4488cba8da3842508e5a39751d","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-09-18 09:34:07.000000000","message":"Patch Set 2:\n\nBuild succeeded (promote pipeline).\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/9c29c75435f84f1fa21c9e0d402b4b0d : SUCCESS in 1m 54s","accounts_in_message":[],"_revision_number":2}],"current_revision_number":2,"current_revision":"6d853dc04e7e11687a8a0f57d49b276f156cc2c7","revisions":{"63b5ed14cd5bbcda5e4604d7595adb1fde807d35":{"kind":"REWORK","_number":1,"created":"2019-09-17 10:17:34.000000000","uploader":{"_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"},"ref":"refs/changes/00/682600/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/kolla","ref":"refs/changes/00/682600/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/kolla refs/changes/00/682600/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/kolla refs/changes/00/682600/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/kolla refs/changes/00/682600/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/kolla refs/changes/00/682600/1"}}},"commit":{"parents":[{"commit":"b8950dbfae16e4959e7e502b584018fc8e58eaaf","subject":"Merge \"Drop Percona XtraBackup in favour of MariaDB Backup\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/b8950dbfae16e4959e7e502b584018fc8e58eaaf"}]}],"author":{"name":"Radosław Piliszek","email":"radoslaw.piliszek@gmail.com","date":"2019-09-17 10:14:22.000000000","tz":120},"committer":{"name":"Radosław Piliszek","email":"radoslaw.piliszek@gmail.com","date":"2019-09-17 10:14:22.000000000","tz":120},"subject":"CI: CentOS - override mirrors ASAP + crmsh mirror override","message":"CI: CentOS - override mirrors ASAP + crmsh mirror override\n\nThis makes mirrored repos used as soon as they are installed\nin the image, instead of hitting upstreams.\n\nThis patch also adds crmsh override.\nUsable only after infra change merges.\n\nDepends-on: https://review.opendev.org/682576\nChange-Id: Ic7494eff4c740b6e528da4508dc88a64e697bc04\nSigned-off-by: Radosław Piliszek \u003cradoslaw.piliszek@gmail.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/63b5ed14cd5bbcda5e4604d7595adb1fde807d35"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/63b5ed14cd5bbcda5e4604d7595adb1fde807d35"}]},"branch":"refs/heads/master"},"6d853dc04e7e11687a8a0f57d49b276f156cc2c7":{"kind":"REWORK","_number":2,"created":"2019-09-17 17:42:59.000000000","uploader":{"_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"},"ref":"refs/changes/00/682600/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/kolla","ref":"refs/changes/00/682600/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/kolla refs/changes/00/682600/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/kolla refs/changes/00/682600/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/kolla refs/changes/00/682600/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/kolla refs/changes/00/682600/2"}}},"commit":{"parents":[{"commit":"b8950dbfae16e4959e7e502b584018fc8e58eaaf","subject":"Merge \"Drop Percona XtraBackup in favour of MariaDB Backup\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/b8950dbfae16e4959e7e502b584018fc8e58eaaf"}]}],"author":{"name":"Radosław Piliszek","email":"radoslaw.piliszek@gmail.com","date":"2019-09-17 10:14:22.000000000","tz":120},"committer":{"name":"Radosław Piliszek","email":"radoslaw.piliszek@gmail.com","date":"2019-09-17 17:42:56.000000000","tz":0},"subject":"CI: CentOS - override mirrors ASAP + crmsh mirror override","message":"CI: CentOS - override mirrors ASAP + crmsh mirror override\n\nThis makes mirrored repos used as soon as they are installed\nin the image, instead of hitting upstreams.\n\nThis patch also adds crmsh override.\nUsable only after infra change merges.\n\nDepends-on: https://review.opendev.org/682576\nChange-Id: Ic7494eff4c740b6e528da4508dc88a64e697bc04\nSigned-off-by: Radosław Piliszek \u003cradoslaw.piliszek@gmail.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/6d853dc04e7e11687a8a0f57d49b276f156cc2c7"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/6d853dc04e7e11687a8a0f57d49b276f156cc2c7"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[],"submit_requirements":[]}
