)]}'
{"id":"openstack%2Fkolla~682610","triplet_id":"openstack%2Fkolla~master~Ia0f8b73bcbda6295319f5c8de1aa02a40844207b","project":"openstack/kolla","branch":"master","topic":"build-pdf-docs","hashtags":[],"change_id":"Ia0f8b73bcbda6295319f5c8de1aa02a40844207b","subject":"Build pdf doc","status":"MERGED","created":"2019-09-17 10:51:44.000000000","updated":"2019-09-19 08:47:55.000000000","submitted":"2019-09-19 08:44:47.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"682610-1568882687845-5c82e828","meta_rev_id":"1ef2424731bd655f7d6806b4392c4f9acc9ed719","_number":682610,"virtual_id_number":682610,"owner":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":24072,"name":"Marcin Juszkiewicz","email":"mjuszkiewicz@redhat.com","username":"hrw"},{"value":0,"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},{"value":0,"_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"},{"value":2,"date":"2019-09-19 08:44:47.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"date":"2019-09-18 10:23:35.000000000","_account_id":23181,"name":"RDO Third Party CI","email":"dmsimard+rdothirdparty@redhat.com","username":"rdothirdparty","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":24072,"name":"Marcin Juszkiewicz","email":"mjuszkiewicz@redhat.com","username":"hrw"},"all":[{"value":2,"date":"2019-09-18 15:34:23.000000000","_account_id":24072,"name":"Marcin Juszkiewicz","email":"mjuszkiewicz@redhat.com","username":"hrw"},{"value":0,"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},{"value":2,"date":"2019-09-18 15:36:31.000000000","_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":23181,"name":"RDO Third Party CI","email":"dmsimard+rdothirdparty@redhat.com","username":"rdothirdparty","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"},"all":[{"value":0,"_account_id":24072,"name":"Marcin Juszkiewicz","email":"mjuszkiewicz@redhat.com","username":"hrw"},{"value":0,"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},{"value":1,"date":"2019-09-18 15:36:31.000000000","_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":23181,"name":"RDO Third Party CI","email":"dmsimard+rdothirdparty@redhat.com","username":"rdothirdparty","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Backport-Candidate":{"all":[{"value":0,"_account_id":24072,"name":"Marcin Juszkiewicz","email":"mjuszkiewicz@redhat.com","username":"hrw"},{"value":0,"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},{"value":0,"_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":23181,"name":"RDO Third Party CI","email":"dmsimard+rdothirdparty@redhat.com","username":"rdothirdparty","tags":["SERVICE_USER"]}],"values":{"-1":"Do Not Backport"," 0":"Backport Review Needed","+1":"Should Backport"},"description":"","default_value":0,"optional":true},"Review-Priority":{"recommended":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},"all":[{"value":0,"_account_id":24072,"name":"Marcin Juszkiewicz","email":"mjuszkiewicz@redhat.com","username":"hrw"},{"value":1,"date":"2019-09-18 15:32:50.000000000","_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},{"value":0,"_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":23181,"name":"RDO Third Party CI","email":"dmsimard+rdothirdparty@redhat.com","username":"rdothirdparty","tags":["SERVICE_USER"]}],"values":{"-1":"Branch Freeze"," 0":"No Priority","+1":"Important Change","+2":"Gate Blocker Fix / Urgent Change"},"description":"","value":1,"default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":23181,"name":"RDO Third Party CI","email":"dmsimard+rdothirdparty@redhat.com","username":"rdothirdparty","tags":["SERVICE_USER"]},{"_account_id":24072,"name":"Marcin Juszkiewicz","email":"mjuszkiewicz@redhat.com","username":"hrw"},{"_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2019-09-18 10:23:35.000000000","updated_by":{"_account_id":23181,"name":"RDO Third Party CI","email":"dmsimard+rdothirdparty@redhat.com","username":"rdothirdparty","tags":["SERVICE_USER"]},"reviewer":{"_account_id":23181,"name":"RDO Third Party CI","email":"dmsimard+rdothirdparty@redhat.com","username":"rdothirdparty","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2019-09-18 15:34:23.000000000","updated_by":{"_account_id":24072,"name":"Marcin Juszkiewicz","email":"mjuszkiewicz@redhat.com","username":"hrw"},"reviewer":{"_account_id":24072,"name":"Marcin Juszkiewicz","email":"mjuszkiewicz@redhat.com","username":"hrw"},"state":"REVIEWER"},{"updated":"2019-09-18 15:36:31.000000000","updated_by":{"_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"},"reviewer":{"_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"},"state":"REVIEWER"},{"updated":"2019-09-19 08:44:47.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"}],"messages":[{"id":"67aed67e43dab5d92eda46e197e18bbc7f546b27","author":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},"date":"2019-09-17 10:51:44.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"c067ffafd8828e066f1182a3223071ab9824b0aa","author":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},"date":"2019-09-17 10:52:11.000000000","message":"Uploaded patch set 2.","accounts_in_message":[],"_revision_number":2},{"id":"2b142483e874ae04dcb4c4eaf1ed843e57f0aebc","author":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},"date":"2019-09-17 10:58:13.000000000","message":"Uploaded patch set 3.","accounts_in_message":[],"_revision_number":3},{"id":"02b9b69ad4de409930fbb7acd1070436941edbb3","author":{"_account_id":23181,"name":"RDO Third Party CI","email":"dmsimard+rdothirdparty@redhat.com","username":"rdothirdparty","tags":["SERVICE_USER"]},"date":"2019-09-17 11:40:00.000000000","message":"Patch Set 3:\n\nBuild succeeded (check pipeline).\n\n- tripleo-build-containers-rhel-8-master http://logs.rdoproject.org/10/682610/3/openstack-check/tripleo-build-containers-rhel-8-master/64f4f4b/ : SUCCESS in 40m 42s","accounts_in_message":[],"_revision_number":3},{"id":"3c64bbe43b1f47b07ef0bf01d3d35445bcdc55e6","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-09-17 12:17:07.000000000","message":"Patch Set 3: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttp://docs.openstack.org/infra/manual/developers.html#automated-testing\n\n\n- kolla-build-centos-binary https://zuul.opendev.org/t/openstack/build/858f7db451b34055a387589f4ff3db93 : SUCCESS in 1h 14m 19s (non-voting)\n- kolla-build-centos-source https://zuul.opendev.org/t/openstack/build/d225bba6c482439bb3378b149d134dd5 : FAILURE in 27m 12s\n- kolla-ansible-centos-source https://zuul.opendev.org/t/openstack/build/4250b5cee44d469e9105feb08e3ba28f : FAILURE in 29m 59s\n- kolla-ansible-centos-binary https://zuul.opendev.org/t/openstack/build/80689c3ce4444e1490aa5d5c69664381 : FAILURE in 29m 37s (non-voting)\n- kolla-ansible-centos-source-upgrade https://zuul.opendev.org/t/openstack/build/3d763ebb19494305a7eb884bab41bd7e : FAILURE in 32m 42s\n- kolla-build-debian-source https://zuul.opendev.org/t/openstack/build/06a6b06063044de49592db6c70317214 : SUCCESS in 29m 27s (non-voting)\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/7be0ecf52990484ba55fd0b70a2d391b : SUCCESS in 3m 11s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/addc70f268e1425cbe6819247798c26e : SUCCESS in 4m 13s\n- openstack-tox-py27 https://zuul.opendev.org/t/openstack/build/e7f775f868774f35a869a7dd2d2e6285 : SUCCESS in 3m 30s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/225ff179b9d14784a30f7e3b42bd4448 : SUCCESS in 2m 57s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/9c217a8ced7f496b8ecc21b21cb907db : SUCCESS in 3m 15s\n- requirements-check https://zuul.opendev.org/t/openstack/build/468ef356a22f42458b69b3612a657976 : FAILURE in 1m 52s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/2674e76d39b045e789cb81c66c82b36b : SUCCESS in 4m 21s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/b10913c3447c4c769fe2a5078cc761d9 : SUCCESS in 4m 04s\n- kolla-build-ubuntu-binary https://zuul.opendev.org/t/openstack/build/1c1970a72a0a4a299ba48ca7b5a0e02a : SUCCESS in 24m 58s (non-voting)\n- kolla-build-ubuntu-source https://zuul.opendev.org/t/openstack/build/51d617103da04d1781b7df9cd3bf1d56 : SUCCESS in 35m 16s\n- kolla-ansible-ubuntu-source https://zuul.opendev.org/t/openstack/build/34b8a5ef9fbc4fcfb0304b9ac841c3a4 : SUCCESS in 52m 43s\n- kolla-ansible-ubuntu-binary https://zuul.opendev.org/t/openstack/build/d619dde5d90d4ecc89b69fa0f5aa360f : SUCCESS in 52m 12s (non-voting)\n- kolla-ansible-ubuntu-source-upgrade https://zuul.opendev.org/t/openstack/build/945b1350e12f4f7f9871af1e50d3fe46 : SUCCESS in 1h 04m 41s","accounts_in_message":[],"_revision_number":3},{"id":"3f33f0e6c3714daefad1e4f176b960295f296fef","author":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},"date":"2019-09-17 13:25:47.000000000","message":"Uploaded patch set 4: Patch Set 3 was rebased.","accounts_in_message":[],"_revision_number":4},{"id":"1137b5c436aa97c611727d83f2f320af6fbadcf3","author":{"_account_id":23181,"name":"RDO Third Party CI","email":"dmsimard+rdothirdparty@redhat.com","username":"rdothirdparty","tags":["SERVICE_USER"]},"date":"2019-09-17 14:07:39.000000000","message":"Patch Set 4:\n\nBuild succeeded (check pipeline).\n\n- tripleo-build-containers-rhel-8-master http://logs.rdoproject.org/10/682610/4/openstack-check/tripleo-build-containers-rhel-8-master/645044f/ : SUCCESS in 40m 57s","accounts_in_message":[],"_revision_number":4},{"id":"d6db1a230424c81d203b2b2d9ddac2dc9aa29416","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-09-17 14:58:55.000000000","message":"Patch Set 4: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttp://docs.openstack.org/infra/manual/developers.html#automated-testing\n\n\n- kolla-build-centos-binary https://zuul.opendev.org/t/openstack/build/8a90c3bc73ab4d749b4e49c9383e5951 : SUCCESS in 30m 49s (non-voting)\n- kolla-build-centos-source https://zuul.opendev.org/t/openstack/build/fb653330ffc04ac985a41368e2b18b3c : SUCCESS in 54m 31s\n- kolla-ansible-centos-source https://zuul.opendev.org/t/openstack/build/f9e250dcfcac4a6d9ff39621ac21e4cd : SUCCESS in 1h 26m 48s\n- kolla-ansible-centos-binary https://zuul.opendev.org/t/openstack/build/b7edfc993a064ac38b6026e2787b8277 : SUCCESS in 1h 02m 19s (non-voting)\n- kolla-ansible-centos-source-upgrade https://zuul.opendev.org/t/openstack/build/879b703096af4740a48364328338bfe0 : FAILURE in 32m 16s\n- kolla-build-debian-source https://zuul.opendev.org/t/openstack/build/10b7e14db17543ebb9ac090d49d61530 : SUCCESS in 33m 00s (non-voting)\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/07197859f02b4bb49ed6f3ec57dafc37 : SUCCESS in 5m 26s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/59a9279829024473ac5e557ed235989f : SUCCESS in 5m 36s\n- openstack-tox-py27 https://zuul.opendev.org/t/openstack/build/d0a649dba8af40b493c9c7df2c80ecf2 : SUCCESS in 3m 54s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/5f917907a1fa4d1abe722eb2d9fe0e88 : SUCCESS in 2m 53s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/5fdacaf477ba4be7927153c53d9b44c5 : SUCCESS in 4m 10s\n- requirements-check https://zuul.opendev.org/t/openstack/build/10d915866c8e4b59ad65e04d131e6008 : SUCCESS in 2m 57s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/16a70eb6e1184d2986b7682ce460a9a2 : SUCCESS in 6m 33s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/1cb1a2eb300c45d3bee20f71be9da385 : SUCCESS in 6m 27s\n- kolla-build-ubuntu-binary https://zuul.opendev.org/t/openstack/build/f540f0a709124b70bda5afd046252d13 : SUCCESS in 35m 53s (non-voting)\n- kolla-build-ubuntu-source https://zuul.opendev.org/t/openstack/build/84e970dbfe614f67955192ea1e9d265a : SUCCESS in 28m 26s\n- kolla-ansible-ubuntu-source https://zuul.opendev.org/t/openstack/build/345c7a43fb9b42feb2a58a62f5f4786e : SUCCESS in 53m 48s\n- kolla-ansible-ubuntu-binary https://zuul.opendev.org/t/openstack/build/ae15ae57345341f8afb3a62356a72ccb : SUCCESS in 50m 16s (non-voting)\n- kolla-ansible-ubuntu-source-upgrade https://zuul.opendev.org/t/openstack/build/2c629bc0e116465ebc54cdb51f9b3274 : SUCCESS in 1h 03m 33s","accounts_in_message":[],"_revision_number":4},{"id":"4a46cf69ecf6f4e9424fbe240e78b780a9f9482e","author":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},"date":"2019-09-17 15:51:17.000000000","message":"Patch Set 4:\n\nrecheck","accounts_in_message":[],"_revision_number":4},{"id":"394d39774de9fdfd12e0fa5227e45ade1431b725","author":{"_account_id":23181,"name":"RDO Third Party CI","email":"dmsimard+rdothirdparty@redhat.com","username":"rdothirdparty","tags":["SERVICE_USER"]},"date":"2019-09-17 16:29:36.000000000","message":"Patch Set 4:\n\nBuild succeeded (check pipeline).\n\n- tripleo-build-containers-rhel-8-master http://logs.rdoproject.org/10/682610/4/openstack-check/tripleo-build-containers-rhel-8-master/74445d7/ : SUCCESS in 37m 25s","accounts_in_message":[],"_revision_number":4},{"id":"efea76adaa5f291387accb13e2ce833e029e409d","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-09-17 18:26:49.000000000","message":"Patch Set 4:\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttp://docs.openstack.org/infra/manual/developers.html#automated-testing\n\n\n- kolla-build-centos-binary https://zuul.opendev.org/t/openstack/build/5b3ad4319dbf493fbf5b0ca36e27ebb8 : SUCCESS in 41m 50s (non-voting)\n- kolla-build-centos-source https://zuul.opendev.org/t/openstack/build/a21bb06168e349809815a56a898d766b : SUCCESS in 53m 07s\n- kolla-ansible-centos-source https://zuul.opendev.org/t/openstack/build/7610d50fa89a43ccb0f6e40d06bdd926 : SUCCESS in 1h 39m 23s\n- kolla-ansible-centos-binary https://zuul.opendev.org/t/openstack/build/db519a4b0be743b5b960a8c23df2f1c6 : SUCCESS in 1h 12m 30s (non-voting)\n- kolla-ansible-centos-source-upgrade https://zuul.opendev.org/t/openstack/build/32e3216a626e41e5b15b4cb16c3b5026 : TIMED_OUT in 2h 01m 55s\n- kolla-build-debian-source https://zuul.opendev.org/t/openstack/build/3bf0f44eae634670b25dee037aeda040 : SUCCESS in 21m 48s (non-voting)\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/088d2840c46c4534878b928a8c81438a : SUCCESS in 4m 55s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/ed42648b03dc48318484c8a3146ee50b : SUCCESS in 4m 20s\n- openstack-tox-py27 https://zuul.opendev.org/t/openstack/build/4092fb75d2434c5da7ebd69e18d1dbb9 : SUCCESS in 3m 21s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/8a58c231124e4529a1c4ad220320bde1 : SUCCESS in 3m 55s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/7308ac2c2d7c4b1cad6e3e5769436edb : SUCCESS in 5m 20s\n- requirements-check https://zuul.opendev.org/t/openstack/build/def6897bc2d04b28840f3f029b25d83b : SUCCESS in 3m 12s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/e9cc4bf5ebe3416f95b4b4a05675080e : SUCCESS in 6m 49s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/aecd785eae2f4e12905f3222d5b3b8ed : SUCCESS in 4m 19s\n- kolla-build-ubuntu-binary https://zuul.opendev.org/t/openstack/build/a243a30d977c4176a0aedce8626cfce5 : SUCCESS in 27m 05s (non-voting)\n- kolla-build-ubuntu-source https://zuul.opendev.org/t/openstack/build/8c076e6a29dd47d59ff68c5c5e799e7d : SUCCESS in 33m 22s\n- kolla-ansible-ubuntu-source https://zuul.opendev.org/t/openstack/build/b6ca16d6771448e4b4b320c9a399df02 : SUCCESS in 50m 51s\n- kolla-ansible-ubuntu-binary https://zuul.opendev.org/t/openstack/build/531bb4e94a44473fa7c897823e4d4ea4 : SUCCESS in 51m 01s (non-voting)\n- kolla-ansible-ubuntu-source-upgrade https://zuul.opendev.org/t/openstack/build/118ae259c1b04919a35274dee3ed8eb2 : SUCCESS in 1h 17m 23s","accounts_in_message":[],"_revision_number":4},{"id":"d3e07107037ccd5d43e8e2fa3424bef54e6d7dc5","author":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},"date":"2019-09-18 09:42:23.000000000","message":"Patch Set 4:\n\nrecheck","accounts_in_message":[],"_revision_number":4},{"id":"41439609d831275db9e72e1e48f6fd689fa70bf1","author":{"_account_id":23181,"name":"RDO Third Party CI","email":"dmsimard+rdothirdparty@redhat.com","username":"rdothirdparty","tags":["SERVICE_USER"]},"date":"2019-09-18 10:23:35.000000000","message":"Patch Set 4:\n\nBuild succeeded (check pipeline).\n\n- tripleo-build-containers-rhel-8-master http://logs.rdoproject.org/10/682610/4/openstack-check/tripleo-build-containers-rhel-8-master/6f518dc/ : SUCCESS in 39m 41s","accounts_in_message":[],"_revision_number":4},{"id":"845a3951a27a2bcef5340ea6a926ed2926ccd408","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-09-18 11:13:23.000000000","message":"Patch Set 4: Verified+1\n\nBuild succeeded (check pipeline).\n\n- kolla-build-centos-binary https://zuul.opendev.org/t/openstack/build/2e82e305a15a41afb31476b03ae4b7a9 : SUCCESS in 47m 25s (non-voting)\n- kolla-build-centos-source https://zuul.opendev.org/t/openstack/build/1f26366ba9dd45829ecfeb6a4433b598 : SUCCESS in 51m 47s\n- kolla-ansible-centos-source https://zuul.opendev.org/t/openstack/build/b2dc9770c4ee44f18fd759e12ca18a02 : SUCCESS in 1h 22m 04s\n- kolla-ansible-centos-binary https://zuul.opendev.org/t/openstack/build/676e2867c0c84093b542f2119f3f356f : SUCCESS in 1h 19m 55s (non-voting)\n- kolla-ansible-centos-source-upgrade https://zuul.opendev.org/t/openstack/build/25ed888aa4e14cd2a07621afb92ae43b : SUCCESS in 1h 25m 50s\n- kolla-build-debian-source https://zuul.opendev.org/t/openstack/build/8e7674e24efe435abdfbd23eb4270b8e : SUCCESS in 24m 45s (non-voting)\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/04519cfcba2a451c811d79df4898ed08 : SUCCESS in 3m 17s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/4bd561b3d85c47819257d8105734c8b3 : SUCCESS in 4m 17s\n- openstack-tox-py27 https://zuul.opendev.org/t/openstack/build/72666fd91a9a4139ac9ad84a3f45f03e : SUCCESS in 3m 23s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/8ada149bae6146e4bdd2083b70ef7115 : SUCCESS in 3m 33s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/26b2d469339f42bcb72a91b027de8052 : SUCCESS in 3m 32s\n- requirements-check https://zuul.opendev.org/t/openstack/build/59b3f1580a774246802fc525b8954e23 : SUCCESS in 4m 11s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/d132bd42acd44889b75e10138c71a20e : SUCCESS in 4m 40s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/32ec1b2e1a1d4b28aae243034e44bc3a : SUCCESS in 5m 21s\n- kolla-build-ubuntu-binary https://zuul.opendev.org/t/openstack/build/219277acd3ac42668c6c8adfb54bf085 : SUCCESS in 20m 32s (non-voting)\n- kolla-build-ubuntu-source https://zuul.opendev.org/t/openstack/build/48704c16beeb418788cfd016cffb66ee : SUCCESS in 29m 22s\n- kolla-ansible-ubuntu-source https://zuul.opendev.org/t/openstack/build/55c13b00cc3f4425bdeb2038f40d9031 : SUCCESS in 59m 34s\n- kolla-ansible-ubuntu-binary https://zuul.opendev.org/t/openstack/build/c2f2926ad7bc4ab2bbdfd3fe8f144576 : SUCCESS in 1h 06m 23s (non-voting)\n- kolla-ansible-ubuntu-source-upgrade https://zuul.opendev.org/t/openstack/build/5872dec5337a457ba9969401bfcdbc68 : SUCCESS in 1h 00m 20s","accounts_in_message":[],"_revision_number":4},{"id":"4302bb50c1d942a76b376f6f5a520990039c3e8d","author":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},"date":"2019-09-18 15:32:50.000000000","message":"Patch Set 4: Review-Priority+1","accounts_in_message":[],"_revision_number":4},{"id":"2f7850f916db34a60112525e416b5c920da3861b","author":{"_account_id":24072,"name":"Marcin Juszkiewicz","email":"mjuszkiewicz@redhat.com","username":"hrw"},"date":"2019-09-18 15:34:23.000000000","message":"Patch Set 4: Code-Review+2","accounts_in_message":[],"_revision_number":4},{"id":"fb58fd24d85dcaed6d8d8807e8715d2a73beab9a","author":{"_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"},"date":"2019-09-18 15:36:31.000000000","message":"Patch Set 4: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":4},{"id":"51998e197c4f0c02cb21ad11391c952aa70517a2","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-09-19 07:10:15.000000000","message":"Patch Set 4: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":4},{"id":"1863e25994226a74e272d94db4ba8d2ff12c5cdd","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-09-19 08:44:47.000000000","message":"Patch Set 4: Verified+2\n\nBuild succeeded (gate pipeline).\n\n- kolla-build-centos-source https://zuul.opendev.org/t/openstack/build/a4801b9ea4a94fd09c18f31082df9a85 : SUCCESS in 55m 18s\n- kolla-ansible-centos-source https://zuul.opendev.org/t/openstack/build/04cac409388b437389072a406872ee30 : SUCCESS in 1h 21m 46s\n- kolla-ansible-centos-source-upgrade https://zuul.opendev.org/t/openstack/build/4431eae2cbc84a79807eacfe5bb490b7 : SUCCESS in 1h 26m 44s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/08129339196a42ad9e44232d6bc6acbb : SUCCESS in 4m 12s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/115bad31175f48928a8405ee890babad : SUCCESS in 3m 59s\n- openstack-tox-py27 https://zuul.opendev.org/t/openstack/build/4b8b285aa2e643a2ad9e3abedf4e770b : SUCCESS in 3m 58s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/0e08da96e04a4653946b539dccd10393 : SUCCESS in 3m 26s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/68a9ed9ca9ec40d7a0f18f992a4a3797 : SUCCESS in 3m 34s\n- requirements-check https://zuul.opendev.org/t/openstack/build/68d387efd2a9447b82ea490f7b838880 : SUCCESS in 2m 20s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/c7de1bbd6ed143a19487fd7812c314e4 : SUCCESS in 7m 05s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/be7fdcad88724fb587bd4099e3483cd8 : SUCCESS in 4m 59s\n- kolla-build-ubuntu-source https://zuul.opendev.org/t/openstack/build/59857c53eaa2413a8fb8f1618f1fc4cb : SUCCESS in 27m 59s\n- kolla-ansible-ubuntu-source https://zuul.opendev.org/t/openstack/build/925ca336b1cd43cf870dbd6e1e670696 : SUCCESS in 59m 39s\n- kolla-ansible-ubuntu-source-upgrade https://zuul.opendev.org/t/openstack/build/2986bd75dacd49d09ed332ea343947d5 : SUCCESS in 1h 03m 47s","accounts_in_message":[],"_revision_number":4},{"id":"7c9b75b05104e818467f5089836bf605140ca28c","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-09-19 08:44:47.000000000","message":"Change has been successfully merged by Zuul","accounts_in_message":[],"_revision_number":4},{"id":"4a27f24334bcf7b43f4e06344a12071f58d105e3","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-09-19 08:47:55.000000000","message":"Patch Set 4:\n\nBuild succeeded (promote pipeline).\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/4acbc1713ccc4c5dbf034286c3090d03 : SUCCESS in 1m 18s\n- promote-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/06380b95a23442edba7c5ebd8f9f4cbf : SUCCESS in 54s","accounts_in_message":[],"_revision_number":4}],"current_revision_number":4,"current_revision":"75e7bd530d31c7b8c8544952534122db5bf53c61","revisions":{"e22fe82bedde9c243b9a14bbcce707398adbf57b":{"kind":"REWORK","_number":1,"created":"2019-09-17 10:51:44.000000000","uploader":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},"ref":"refs/changes/10/682610/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/kolla","ref":"refs/changes/10/682610/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/kolla refs/changes/10/682610/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/kolla refs/changes/10/682610/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/kolla refs/changes/10/682610/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/kolla refs/changes/10/682610/1"}}},"commit":{"parents":[{"commit":"532af2efc8893cfd23f8d2beadd32a4e3463ae85","subject":"Do not fan out DB sync to all cells","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/532af2efc8893cfd23f8d2beadd32a4e3463ae85"}]}],"author":{"name":"Mark Goddard","email":"mark@stackhpc.com","date":"2019-09-17 10:50:41.000000000","tz":60},"committer":{"name":"Mark Goddard","email":"mark@stackhpc.com","date":"2019-09-17 10:50:41.000000000","tz":60},"subject":"Build pdf doc","message":"Build pdf doc\n\nThe is one of community goals that each project should produce a\nsingle PDF file. The pdf should be in the output of openstack-tox-docs\njob.\n\nTeX packages are required to build PDF locally, following is\nrecommended:\n\n* inkscape\n* texlive-latex-base\n* texlive-latex-extra\n* texlive-fonts-recommended\n\nMore about the goal:\nhttps://governance.openstack.org/tc/goals/train/pdf-doc-generation.html\nhttps://etherpad.openstack.org/p/train-pdf-support-goal\nhttps://etherpad.openstack.org/p/pdf-goal-train-common-problems\n\nChange-Id: Ia0f8b73bcbda6295319f5c8de1aa02a40844207b\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/e22fe82bedde9c243b9a14bbcce707398adbf57b"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/e22fe82bedde9c243b9a14bbcce707398adbf57b"}]},"branch":"refs/heads/master"},"7583a381b19b172567ebf3cdede31dfcbabb4f7d":{"kind":"REWORK","_number":2,"created":"2019-09-17 10:52:11.000000000","uploader":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},"ref":"refs/changes/10/682610/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/kolla","ref":"refs/changes/10/682610/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/kolla refs/changes/10/682610/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/kolla refs/changes/10/682610/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/kolla refs/changes/10/682610/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/kolla refs/changes/10/682610/2"}}},"commit":{"parents":[{"commit":"532af2efc8893cfd23f8d2beadd32a4e3463ae85","subject":"Do not fan out DB sync to all cells","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/532af2efc8893cfd23f8d2beadd32a4e3463ae85"}]}],"author":{"name":"Mark Goddard","email":"mark@stackhpc.com","date":"2019-09-17 10:50:41.000000000","tz":60},"committer":{"name":"Mark Goddard","email":"mark@stackhpc.com","date":"2019-09-17 10:52:02.000000000","tz":60},"subject":"Build pdf doc","message":"Build pdf doc\n\nThe is one of community goals that each project should produce a\nsingle PDF file. The pdf should be in the output of openstack-tox-docs\njob.\n\nTeX packages are required to build PDF locally, following is\nrecommended:\n\n* inkscape\n* texlive-latex-base\n* texlive-latex-extra\n* texlive-fonts-recommended\n\nMore about the goal:\nhttps://governance.openstack.org/tc/goals/train/pdf-doc-generation.html\nhttps://etherpad.openstack.org/p/train-pdf-support-goal\nhttps://etherpad.openstack.org/p/pdf-goal-train-common-problems\n\nChange-Id: Ia0f8b73bcbda6295319f5c8de1aa02a40844207b\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/7583a381b19b172567ebf3cdede31dfcbabb4f7d"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/7583a381b19b172567ebf3cdede31dfcbabb4f7d"}]},"branch":"refs/heads/master"},"5c07c8ac92354b0789e4b6290fa8acf021ccc061":{"kind":"REWORK","_number":3,"created":"2019-09-17 10:58:13.000000000","uploader":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},"ref":"refs/changes/10/682610/3","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/kolla","ref":"refs/changes/10/682610/3","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/kolla refs/changes/10/682610/3 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/kolla refs/changes/10/682610/3 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/kolla refs/changes/10/682610/3 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/kolla refs/changes/10/682610/3"}}},"commit":{"parents":[{"commit":"532af2efc8893cfd23f8d2beadd32a4e3463ae85","subject":"Do not fan out DB sync to all cells","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/532af2efc8893cfd23f8d2beadd32a4e3463ae85"}]}],"author":{"name":"Mark Goddard","email":"mark@stackhpc.com","date":"2019-09-17 10:50:41.000000000","tz":60},"committer":{"name":"Mark Goddard","email":"mark@stackhpc.com","date":"2019-09-17 10:57:58.000000000","tz":60},"subject":"Build pdf doc","message":"Build pdf doc\n\nThe is one of community goals that each project should produce a\nsingle PDF file. The pdf should be in the output of openstack-tox-docs\njob.\n\nTeX packages are required to build PDF locally, following is\nrecommended:\n\n* inkscape\n* texlive-latex-base\n* texlive-latex-extra\n* texlive-fonts-recommended\n\nMore about the goal:\nhttps://governance.openstack.org/tc/goals/train/pdf-doc-generation.html\nhttps://etherpad.openstack.org/p/train-pdf-support-goal\nhttps://etherpad.openstack.org/p/pdf-goal-train-common-problems\n\nChange-Id: Ia0f8b73bcbda6295319f5c8de1aa02a40844207b\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/5c07c8ac92354b0789e4b6290fa8acf021ccc061"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/5c07c8ac92354b0789e4b6290fa8acf021ccc061"}]},"branch":"refs/heads/master"},"75e7bd530d31c7b8c8544952534122db5bf53c61":{"kind":"TRIVIAL_REBASE","_number":4,"created":"2019-09-17 13:25:47.000000000","uploader":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},"ref":"refs/changes/10/682610/4","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/kolla","ref":"refs/changes/10/682610/4","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/kolla refs/changes/10/682610/4 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/kolla refs/changes/10/682610/4 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/kolla refs/changes/10/682610/4 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/kolla refs/changes/10/682610/4"}}},"commit":{"parents":[{"commit":"20a5957a41ac63ed0313aac8216df9fefb0ab0aa","subject":"Fix GitPython entry in requirements.txt","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/20a5957a41ac63ed0313aac8216df9fefb0ab0aa"}]}],"author":{"name":"Mark Goddard","email":"mark@stackhpc.com","date":"2019-09-17 10:50:41.000000000","tz":60},"committer":{"name":"Mark Goddard","email":"mark@stackhpc.com","date":"2019-09-17 13:25:32.000000000","tz":60},"subject":"Build pdf doc","message":"Build pdf doc\n\nThe is one of community goals that each project should produce a\nsingle PDF file. The pdf should be in the output of openstack-tox-docs\njob.\n\nTeX packages are required to build PDF locally, following is\nrecommended:\n\n* inkscape\n* texlive-latex-base\n* texlive-latex-extra\n* texlive-fonts-recommended\n\nMore about the goal:\nhttps://governance.openstack.org/tc/goals/train/pdf-doc-generation.html\nhttps://etherpad.openstack.org/p/train-pdf-support-goal\nhttps://etherpad.openstack.org/p/pdf-goal-train-common-problems\n\nChange-Id: Ia0f8b73bcbda6295319f5c8de1aa02a40844207b\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/75e7bd530d31c7b8c8544952534122db5bf53c61"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/75e7bd530d31c7b8c8544952534122db5bf53c61"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[],"submit_requirements":[]}
