)]}'
{"id":"openstack%2Fkolla~682646","triplet_id":"openstack%2Fkolla~master~Ie7073e034641276f47c113de945971e8ed4b87ab","project":"openstack/kolla","branch":"master","hashtags":[],"change_id":"Ie7073e034641276f47c113de945971e8ed4b87ab","subject":"Fix GitPython entry in requirements.txt","status":"MERGED","created":"2019-09-17 13:25:47.000000000","updated":"2019-09-19 08:46:13.000000000","submitted":"2019-09-19 08:44:45.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"682646-1568882686134-4b665cae","meta_rev_id":"ad385b5061a2714d6407203371faa0d63ab1b1bb","_number":682646,"virtual_id_number":682646,"owner":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":24072,"name":"Marcin Juszkiewicz","email":"mjuszkiewicz@redhat.com","username":"hrw"},{"value":0,"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},{"value":0,"_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"},{"value":2,"date":"2019-09-19 08:44:45.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"date":"2019-09-17 14:08:16.000000000","_account_id":23181,"name":"RDO Third Party CI","email":"dmsimard+rdothirdparty@redhat.com","username":"rdothirdparty","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":24072,"name":"Marcin Juszkiewicz","email":"mjuszkiewicz@redhat.com","username":"hrw"},"all":[{"value":2,"date":"2019-09-19 07:09:58.000000000","_account_id":24072,"name":"Marcin Juszkiewicz","email":"mjuszkiewicz@redhat.com","username":"hrw"},{"value":0,"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},{"value":2,"date":"2019-09-18 15:35:33.000000000","_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":23181,"name":"RDO Third Party CI","email":"dmsimard+rdothirdparty@redhat.com","username":"rdothirdparty","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":24072,"name":"Marcin Juszkiewicz","email":"mjuszkiewicz@redhat.com","username":"hrw"},"all":[{"value":1,"date":"2019-09-19 07:09:58.000000000","_account_id":24072,"name":"Marcin Juszkiewicz","email":"mjuszkiewicz@redhat.com","username":"hrw"},{"value":0,"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},{"value":0,"_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":23181,"name":"RDO Third Party CI","email":"dmsimard+rdothirdparty@redhat.com","username":"rdothirdparty","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Backport-Candidate":{"all":[{"value":0,"_account_id":24072,"name":"Marcin Juszkiewicz","email":"mjuszkiewicz@redhat.com","username":"hrw"},{"value":0,"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},{"value":0,"_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":23181,"name":"RDO Third Party CI","email":"dmsimard+rdothirdparty@redhat.com","username":"rdothirdparty","tags":["SERVICE_USER"]}],"values":{"-1":"Do Not Backport"," 0":"Backport Review Needed","+1":"Should Backport"},"description":"","default_value":0,"optional":true},"Review-Priority":{"approved":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},"all":[{"value":0,"_account_id":24072,"name":"Marcin Juszkiewicz","email":"mjuszkiewicz@redhat.com","username":"hrw"},{"value":2,"date":"2019-09-18 15:33:13.000000000","_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},{"value":0,"_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":23181,"name":"RDO Third Party CI","email":"dmsimard+rdothirdparty@redhat.com","username":"rdothirdparty","tags":["SERVICE_USER"]}],"values":{"-1":"Branch Freeze"," 0":"No Priority","+1":"Important Change","+2":"Gate Blocker Fix / Urgent Change"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":23181,"name":"RDO Third Party CI","email":"dmsimard+rdothirdparty@redhat.com","username":"rdothirdparty","tags":["SERVICE_USER"]},{"_account_id":24072,"name":"Marcin Juszkiewicz","email":"mjuszkiewicz@redhat.com","username":"hrw"},{"_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2019-09-17 14:08:16.000000000","updated_by":{"_account_id":23181,"name":"RDO Third Party CI","email":"dmsimard+rdothirdparty@redhat.com","username":"rdothirdparty","tags":["SERVICE_USER"]},"reviewer":{"_account_id":23181,"name":"RDO Third Party CI","email":"dmsimard+rdothirdparty@redhat.com","username":"rdothirdparty","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2019-09-18 15:35:33.000000000","updated_by":{"_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"},"reviewer":{"_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"},"state":"REVIEWER"},{"updated":"2019-09-19 07:09:58.000000000","updated_by":{"_account_id":24072,"name":"Marcin Juszkiewicz","email":"mjuszkiewicz@redhat.com","username":"hrw"},"reviewer":{"_account_id":24072,"name":"Marcin Juszkiewicz","email":"mjuszkiewicz@redhat.com","username":"hrw"},"state":"REVIEWER"},{"updated":"2019-09-19 08:44:45.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"}],"messages":[{"id":"63425e96531432bb7c39855ed22fa4befb63fa0f","author":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},"date":"2019-09-17 13:25:47.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"ddca47334cfc284e1c8bc34641ff714675013cbf","author":{"_account_id":23181,"name":"RDO Third Party CI","email":"dmsimard+rdothirdparty@redhat.com","username":"rdothirdparty","tags":["SERVICE_USER"]},"date":"2019-09-17 14:08:16.000000000","message":"Patch Set 1:\n\nBuild succeeded (check pipeline).\n\n- tripleo-build-containers-rhel-8-master http://logs.rdoproject.org/46/682646/1/openstack-check/tripleo-build-containers-rhel-8-master/c222d51/ : SUCCESS in 41m 35s","accounts_in_message":[],"_revision_number":1},{"id":"3329939ff3122e4492ac0a98aaf87c74fe1e4ba0","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-09-17 15:06:55.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\n\n- kolla-build-centos-binary https://zuul.opendev.org/t/openstack/build/1163eb7640554aa7bde3305515fb4777 : SUCCESS in 45m 09s (non-voting)\n- kolla-build-centos-source https://zuul.opendev.org/t/openstack/build/2e488f2d84ee42a9a3ab97d28df86a40 : SUCCESS in 41m 53s\n- kolla-ansible-centos-source https://zuul.opendev.org/t/openstack/build/2b889d800fe24c2585f8cd3d31f7ab81 : SUCCESS in 1h 28m 55s\n- kolla-ansible-centos-binary https://zuul.opendev.org/t/openstack/build/193a7fba9f5746e1a5d536321a70ca41 : SUCCESS in 1h 16m 45s (non-voting)\n- kolla-ansible-centos-source-upgrade https://zuul.opendev.org/t/openstack/build/b78239732afc4b6090faf4ddb9c52789 : SUCCESS in 1h 33m 10s\n- tripleo-build-containers-centos-7 https://zuul.opendev.org/t/openstack/build/90bd906abbce4341beed47d86acb712e : SUCCESS in 51m 37s\n- kolla-build-debian-source https://zuul.opendev.org/t/openstack/build/5b33da58b42b435383dfa0a346d4c8a1 : SUCCESS in 29m 40s (non-voting)\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/09c3fe192fbb4baa885df9783e0324fa : SUCCESS in 3m 54s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/ae3463eb0b414abc965d88bb696dde74 : SUCCESS in 4m 58s\n- openstack-tox-py27 https://zuul.opendev.org/t/openstack/build/6ea5a2fc6d23491aaa774ac4ac87f0bf : SUCCESS in 5m 48s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/dea822838a754ceca14bcc4cbe552d29 : SUCCESS in 4m 06s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/20dd0dbbf4ef4ddeae3b667c1bd97ae0 : SUCCESS in 5m 20s\n- requirements-check https://zuul.opendev.org/t/openstack/build/c027b35ade23460bbf10bc4b70d74e0f : SUCCESS in 2m 21s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/dd4ec7d31f164a6099324db833576baf : SUCCESS in 2m 58s\n- kolla-build-ubuntu-binary https://zuul.opendev.org/t/openstack/build/9132b091ad584146b2ae030ba5268798 : SUCCESS in 23m 54s (non-voting)\n- kolla-build-ubuntu-source https://zuul.opendev.org/t/openstack/build/8f65cd9ef33a47cfa70766c858e43fef : SUCCESS in 32m 04s\n- kolla-ansible-ubuntu-source https://zuul.opendev.org/t/openstack/build/b8fc928d930c404e894b3c9d457a8e25 : SUCCESS in 55m 05s\n- kolla-ansible-ubuntu-binary https://zuul.opendev.org/t/openstack/build/be1cdbab3b254afa9df05e419b5236d6 : SUCCESS in 52m 32s (non-voting)\n- kolla-ansible-ubuntu-source-upgrade https://zuul.opendev.org/t/openstack/build/5378f102faa14f258d05aa55428b5863 : SUCCESS in 1h 01m 26s","accounts_in_message":[],"_revision_number":1},{"id":"876856f2718309cb458be74d5c1ddfa53fe20182","author":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},"date":"2019-09-18 15:33:13.000000000","message":"Patch Set 1: Review-Priority+2\n\nBlocks changes to requirements.txt","accounts_in_message":[],"_revision_number":1},{"id":"a3fa8de46554a0e5331a4eb22a20e844b37a38e6","author":{"_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"},"date":"2019-09-18 15:35:33.000000000","message":"Patch Set 1: Code-Review+2","accounts_in_message":[],"_revision_number":1},{"id":"6e39dc7ec480f09635407de29e46cc7242404ae1","author":{"_account_id":24072,"name":"Marcin Juszkiewicz","email":"mjuszkiewicz@redhat.com","username":"hrw"},"date":"2019-09-19 07:09:58.000000000","message":"Patch Set 1: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":1},{"id":"d6327dba3a4bed29a35864abee2ca68d4ea29878","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-09-19 07:10:14.000000000","message":"Patch Set 1: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":1},{"id":"38bdf149831662d7f6218107be133567bc01272d","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-09-19 08:44:45.000000000","message":"Patch Set 1: Verified+2\n\nBuild succeeded (gate pipeline).\n\n- kolla-build-centos-source https://zuul.opendev.org/t/openstack/build/ebcf8634625d484da22056cbf6e36bd0 : SUCCESS in 51m 28s\n- kolla-ansible-centos-source https://zuul.opendev.org/t/openstack/build/a23546ff9b5a46288805b2d38976d661 : SUCCESS in 1h 28m 48s\n- kolla-ansible-centos-source-upgrade https://zuul.opendev.org/t/openstack/build/f2e705f6a0dc47409c469852cdcf3a98 : SUCCESS in 1h 27m 05s\n- tripleo-build-containers-centos-7 https://zuul.opendev.org/t/openstack/build/df67c541da234278b1c5f2db026c1920 : SUCCESS in 59m 23s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/f6d62fef6b1e470186a1f5606acd7aaf : SUCCESS in 3m 54s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/f44134c3a9cb40439522bd81c7a27ecc : SUCCESS in 4m 42s\n- openstack-tox-py27 https://zuul.opendev.org/t/openstack/build/761a5d1d7ab147a68553c2b7b6c2b795 : SUCCESS in 7m 31s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/726539d7639d4786bced277d7536c4a4 : SUCCESS in 6m 42s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/e951c276935b439eaa42fa9dfe312289 : SUCCESS in 8m 54s\n- requirements-check https://zuul.opendev.org/t/openstack/build/9799282e7e294e3db28b81bf5cdc1e3c : SUCCESS in 1m 53s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/ec3517ace9974f3681153ec0e984236e : SUCCESS in 2m 54s\n- kolla-build-ubuntu-source https://zuul.opendev.org/t/openstack/build/1141f993216f4ffcb2be7f93ca1f7ed8 : SUCCESS in 32m 15s\n- kolla-ansible-ubuntu-source https://zuul.opendev.org/t/openstack/build/74001857e9a34eee945797b6ef2afcf3 : SUCCESS in 1h 05m 10s\n- kolla-ansible-ubuntu-source-upgrade https://zuul.opendev.org/t/openstack/build/3285127fd56c4bf89dd41894405e461f : SUCCESS in 1h 01m 00s","accounts_in_message":[],"_revision_number":1},{"id":"f289632185d48a0e012c0b568a383515827f54b8","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-09-19 08:44:46.000000000","message":"Change has been successfully merged by Zuul","accounts_in_message":[],"_revision_number":1},{"id":"ad385b5061a2714d6407203371faa0d63ab1b1bb","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-09-19 08:46:13.000000000","message":"Patch Set 1:\n\nBuild succeeded (promote pipeline).\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/80a09fd53b5f446b88c5a0969b63989f : SUCCESS in 1m 01s","accounts_in_message":[],"_revision_number":1}],"current_revision_number":1,"current_revision":"20a5957a41ac63ed0313aac8216df9fefb0ab0aa","revisions":{"20a5957a41ac63ed0313aac8216df9fefb0ab0aa":{"kind":"REWORK","_number":1,"created":"2019-09-17 13:25:47.000000000","uploader":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},"ref":"refs/changes/46/682646/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/kolla","ref":"refs/changes/46/682646/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/kolla refs/changes/46/682646/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/kolla refs/changes/46/682646/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/kolla refs/changes/46/682646/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/kolla refs/changes/46/682646/1"}}},"commit":{"parents":[{"commit":"532af2efc8893cfd23f8d2beadd32a4e3463ae85","subject":"Do not fan out DB sync to all cells","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/532af2efc8893cfd23f8d2beadd32a4e3463ae85"}]}],"author":{"name":"Mark Goddard","email":"mark@stackhpc.com","date":"2019-09-17 13:24:26.000000000","tz":60},"committer":{"name":"Mark Goddard","email":"mark@stackhpc.com","date":"2019-09-17 13:25:32.000000000","tz":60},"subject":"Fix GitPython entry in requirements.txt","message":"Fix GitPython entry in requirements.txt\n\nThis now has different requirements in global-requirements for py2 and\npy3.\n\nTrivialFix\n\nChange-Id: Ie7073e034641276f47c113de945971e8ed4b87ab\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/20a5957a41ac63ed0313aac8216df9fefb0ab0aa"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/20a5957a41ac63ed0313aac8216df9fefb0ab0aa"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[],"submit_requirements":[]}
