)]}'
{"id":"openstack%2Fkolla~682673","triplet_id":"openstack%2Fkolla~master~Ifaeed6917636e2bcb5ea3bda32ae0393ebb08a4f","project":"openstack/kolla","branch":"master","hashtags":[],"change_id":"Ifaeed6917636e2bcb5ea3bda32ae0393ebb08a4f","subject":"build.py: fix pep8 warnings","status":"MERGED","created":"2019-09-17 15:11:48.000000000","updated":"2019-09-19 08:48:56.000000000","submitted":"2019-09-19 08:46:10.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"682673-1568882770546-de761889","meta_rev_id":"1492e84d7fbb48b4ca269ccd3227939739574607","_number":682673,"virtual_id_number":682673,"owner":{"_account_id":24072,"name":"Marcin Juszkiewicz","email":"mjuszkiewicz@redhat.com","username":"hrw"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"date":"2019-09-18 12:28:26.000000000","_account_id":24072,"name":"Marcin Juszkiewicz","email":"mjuszkiewicz@redhat.com","username":"hrw"},{"value":0,"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},{"value":0,"_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"},{"value":2,"date":"2019-09-19 08:46:10.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"date":"2019-09-18 13:17:56.000000000","_account_id":23181,"name":"RDO Third Party CI","email":"dmsimard+rdothirdparty@redhat.com","username":"rdothirdparty","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},"all":[{"value":0,"_account_id":24072,"name":"Marcin Juszkiewicz","email":"mjuszkiewicz@redhat.com","username":"hrw"},{"value":2,"date":"2019-09-17 16:54:47.000000000","_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},{"value":2,"date":"2019-09-19 07:11:34.000000000","_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":23181,"name":"RDO Third Party CI","email":"dmsimard+rdothirdparty@redhat.com","username":"rdothirdparty","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"},"all":[{"value":0,"_account_id":24072,"name":"Marcin Juszkiewicz","email":"mjuszkiewicz@redhat.com","username":"hrw"},{"value":0,"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},{"value":1,"date":"2019-09-19 07:11:34.000000000","_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":23181,"name":"RDO Third Party CI","email":"dmsimard+rdothirdparty@redhat.com","username":"rdothirdparty","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Backport-Candidate":{"all":[{"value":0,"_account_id":24072,"name":"Marcin Juszkiewicz","email":"mjuszkiewicz@redhat.com","username":"hrw"},{"value":0,"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},{"value":0,"_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":23181,"name":"RDO Third Party CI","email":"dmsimard+rdothirdparty@redhat.com","username":"rdothirdparty","tags":["SERVICE_USER"]}],"values":{"-1":"Do Not Backport"," 0":"Backport Review Needed","+1":"Should Backport"},"description":"","default_value":0,"optional":true},"Review-Priority":{"all":[{"value":0,"_account_id":24072,"name":"Marcin Juszkiewicz","email":"mjuszkiewicz@redhat.com","username":"hrw"},{"value":0,"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},{"value":0,"_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":23181,"name":"RDO Third Party CI","email":"dmsimard+rdothirdparty@redhat.com","username":"rdothirdparty","tags":["SERVICE_USER"]}],"values":{"-1":"Branch Freeze"," 0":"No Priority","+1":"Important Change","+2":"Gate Blocker Fix / Urgent Change"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":23181,"name":"RDO Third Party CI","email":"dmsimard+rdothirdparty@redhat.com","username":"rdothirdparty","tags":["SERVICE_USER"]},{"_account_id":24072,"name":"Marcin Juszkiewicz","email":"mjuszkiewicz@redhat.com","username":"hrw"},{"_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2019-09-17 16:54:47.000000000","updated_by":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},"reviewer":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},"state":"REVIEWER"},{"updated":"2019-09-18 13:17:56.000000000","updated_by":{"_account_id":23181,"name":"RDO Third Party CI","email":"dmsimard+rdothirdparty@redhat.com","username":"rdothirdparty","tags":["SERVICE_USER"]},"reviewer":{"_account_id":23181,"name":"RDO Third Party CI","email":"dmsimard+rdothirdparty@redhat.com","username":"rdothirdparty","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2019-09-19 07:11:34.000000000","updated_by":{"_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"},"reviewer":{"_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"},"state":"REVIEWER"},{"updated":"2019-09-19 08:46:10.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"}],"messages":[{"id":"7ad62a036d27c2bf63145e9ab424abcbc80e066e","author":{"_account_id":24072,"name":"Marcin Juszkiewicz","email":"mjuszkiewicz@redhat.com","username":"hrw"},"date":"2019-09-17 15:11:48.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"46de6c41ac2c01c60cb9393d0b9dc9ab2d4b7310","author":{"_account_id":23181,"name":"RDO Third Party CI","email":"dmsimard+rdothirdparty@redhat.com","username":"rdothirdparty","tags":["SERVICE_USER"]},"date":"2019-09-17 15:51:55.000000000","message":"Patch Set 1:\n\nBuild succeeded (check pipeline).\n\n- tripleo-build-containers-rhel-8-master http://logs.rdoproject.org/73/682673/1/openstack-check/tripleo-build-containers-rhel-8-master/57de560/ : SUCCESS in 39m 16s","accounts_in_message":[],"_revision_number":1},{"id":"965ac41d592487f7ac373c5cf16bcf5313dba532","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-09-17 16:52:05.000000000","message":"Patch Set 1: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttp://docs.openstack.org/infra/manual/developers.html#automated-testing\n\n\n- kolla-build-centos-binary https://zuul.opendev.org/t/openstack/build/17bd81ffe79d4919a5509e4fb9dc8c92 : FAILURE in 27m 22s (non-voting)\n- kolla-build-centos-source https://zuul.opendev.org/t/openstack/build/d9a70cf0fad946cca07fa92a2372aa9d : FAILURE in 27m 19s\n- kolla-ansible-centos-source https://zuul.opendev.org/t/openstack/build/eb79fa5f917d46c4866f92af66b02161 : SUCCESS in 1h 19m 33s\n- kolla-ansible-centos-binary https://zuul.opendev.org/t/openstack/build/989b89b07268438485b6ca516d2c2ec8 : SUCCESS in 1h 14m 13s (non-voting)\n- kolla-ansible-centos-source-upgrade https://zuul.opendev.org/t/openstack/build/77156ad3500146888227e42e9eef2f7c : SUCCESS in 1h 20m 15s\n- tripleo-build-containers-centos-7 https://zuul.opendev.org/t/openstack/build/4728896731ac4b47b21d00d95aeb9a5c : SUCCESS in 55m 33s\n- kolla-build-debian-source https://zuul.opendev.org/t/openstack/build/dbf016d96acb4522b14ebdc5460219be : SUCCESS in 24m 02s (non-voting)\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/86d6bdc1b4fa4a48b261845c06b94ccf : SUCCESS in 4m 52s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/e008240541644817aff0b70139e23734 : SUCCESS in 4m 08s\n- openstack-tox-py27 https://zuul.opendev.org/t/openstack/build/2f7bf0f44e6d4ac8b4006068b0297e7a : SUCCESS in 3m 30s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/ee26329f12d34780abf13403b0845a47 : SUCCESS in 4m 49s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/ab315bc18b8b4be581a894a230bf7c31 : SUCCESS in 4m 55s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/129452bab4ce4494a1343c5aec3dd5f9 : SUCCESS in 4m 03s\n- kolla-build-ubuntu-binary https://zuul.opendev.org/t/openstack/build/949ef12ead864f538f613ed7d5427250 : SUCCESS in 29m 10s (non-voting)\n- kolla-build-ubuntu-source https://zuul.opendev.org/t/openstack/build/f7018e9b14894767be10370a1eca7401 : SUCCESS in 32m 32s\n- kolla-ansible-ubuntu-source https://zuul.opendev.org/t/openstack/build/5e44662d598a4b189b6e5f62749547f8 : SUCCESS in 1h 00m 19s\n- kolla-ansible-ubuntu-binary https://zuul.opendev.org/t/openstack/build/7c0fdde717734dc5ac2ff578556b3548 : SUCCESS in 50m 29s (non-voting)\n- kolla-ansible-ubuntu-source-upgrade https://zuul.opendev.org/t/openstack/build/a3d49d4374dc438c8be7fcd8e8c9c22e : SUCCESS in 1h 16m 45s","accounts_in_message":[],"_revision_number":1},{"id":"a3fdf89752f000ca40f2a39dbc35ddd3721fdfca","author":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},"date":"2019-09-17 16:54:47.000000000","message":"Patch Set 1: Code-Review+2","accounts_in_message":[],"_revision_number":1},{"id":"e17e6eeb544402ec57b0d1a44dc60341a6f75f27","author":{"_account_id":24072,"name":"Marcin Juszkiewicz","email":"mjuszkiewicz@redhat.com","username":"hrw"},"date":"2019-09-18 12:28:26.000000000","message":"Patch Set 1:\n\nrecheck","accounts_in_message":[],"_revision_number":1},{"id":"904afe0a19d47a8c5228a54ef207657a4a3d9528","author":{"_account_id":23181,"name":"RDO Third Party CI","email":"dmsimard+rdothirdparty@redhat.com","username":"rdothirdparty","tags":["SERVICE_USER"]},"date":"2019-09-18 13:17:56.000000000","message":"Patch Set 1:\n\nBuild succeeded (check pipeline).\n\n- tripleo-build-containers-rhel-8-master http://logs.rdoproject.org/73/682673/1/openstack-check/tripleo-build-containers-rhel-8-master/048ef5c/ : SUCCESS in 48m 44s","accounts_in_message":[],"_revision_number":1},{"id":"d133f07116065a6b10fb99c9ac85c0fb3607e3e8","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-09-18 14:05:11.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\n\n- kolla-build-centos-binary https://zuul.opendev.org/t/openstack/build/357b137a71634703b9eb1ccb5e5b3ff2 : SUCCESS in 44m 28s (non-voting)\n- kolla-build-centos-source https://zuul.opendev.org/t/openstack/build/44b7c7560e164f09a718803db6a40ec2 : SUCCESS in 58m 18s\n- kolla-ansible-centos-source https://zuul.opendev.org/t/openstack/build/e57cb284ee1b4b7ea6620726599b5111 : SUCCESS in 1h 26m 32s\n- kolla-ansible-centos-binary https://zuul.opendev.org/t/openstack/build/183f82e36e704fd08c7f04459c34f205 : SUCCESS in 1h 23m 56s (non-voting)\n- kolla-ansible-centos-source-upgrade https://zuul.opendev.org/t/openstack/build/bf10063589c84ae685b14ac660b72d82 : SUCCESS in 1h 17m 45s\n- tripleo-build-containers-centos-7 https://zuul.opendev.org/t/openstack/build/3ffd83ce4260497b88a9d601b20a19c0 : SUCCESS in 56m 37s\n- kolla-build-debian-source https://zuul.opendev.org/t/openstack/build/1bde1859aa144d928f7123c78c57cb45 : SUCCESS in 26m 15s (non-voting)\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/a58ccce953f540439a2f2a4a4cef6a97 : SUCCESS in 5m 04s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/910ecdfc576c40f8b0aab83777cc03a3 : SUCCESS in 4m 25s\n- openstack-tox-py27 https://zuul.opendev.org/t/openstack/build/6f5c950f715d4593b80fe827e355fcae : SUCCESS in 5m 07s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/c24373d410cc44c89b4837021fff7dc3 : SUCCESS in 5m 14s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/20fc71a175c44ed08b344fdba01f48d9 : SUCCESS in 5m 41s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/8153023826a74cc2a8d679ffa1266809 : SUCCESS in 4m 00s\n- kolla-build-ubuntu-binary https://zuul.opendev.org/t/openstack/build/a171ca04d7e44d6d86c743be382ec63e : SUCCESS in 26m 38s (non-voting)\n- kolla-build-ubuntu-source https://zuul.opendev.org/t/openstack/build/a6b790db1ace4df48893a41a36ba48e7 : SUCCESS in 37m 27s\n- kolla-ansible-ubuntu-source https://zuul.opendev.org/t/openstack/build/0689ae3668ac453f820c8d4aad057674 : SUCCESS in 50m 52s\n- kolla-ansible-ubuntu-binary https://zuul.opendev.org/t/openstack/build/7c690b07133b4a0b8893867ab997a085 : SUCCESS in 51m 07s (non-voting)\n- kolla-ansible-ubuntu-source-upgrade https://zuul.opendev.org/t/openstack/build/0ed5783dfa184a7da09c5056b2a9a4a8 : SUCCESS in 1h 05m 57s","accounts_in_message":[],"_revision_number":1},{"id":"0ac3158ad01ec597d249e0fee09b5e5cc42b2a38","author":{"_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"},"date":"2019-09-19 07:11:34.000000000","message":"Patch Set 1: Code-Review+2 Workflow+1\n\nlovely","accounts_in_message":[],"_revision_number":1},{"id":"346cd0ef3395238efbb0dc67caad448d3640606a","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-09-19 07:11:51.000000000","message":"Patch Set 1: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":1},{"id":"62ecc1ba17dcfa46d96651359d934ecd2c76044a","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-09-19 08:46:10.000000000","message":"Change has been successfully merged by Zuul","accounts_in_message":[],"_revision_number":1},{"id":"56e0a247cd4ae3aab6c1efead55424750dfabe4a","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-09-19 08:46:10.000000000","message":"Patch Set 1: Verified+2\n\nBuild succeeded (gate pipeline).\n\n- kolla-build-centos-source https://zuul.opendev.org/t/openstack/build/90f3456bb335489f97a0212d529d553b : SUCCESS in 57m 02s\n- kolla-ansible-centos-source https://zuul.opendev.org/t/openstack/build/67055926236c47b0889c75264518767e : SUCCESS in 1h 21m 55s\n- kolla-ansible-centos-source-upgrade https://zuul.opendev.org/t/openstack/build/e73f6cc0778d47fb8f096bbbdf9d549e : SUCCESS in 1h 26m 59s\n- tripleo-build-containers-centos-7 https://zuul.opendev.org/t/openstack/build/3e762f774a614dc697d3e674903ca3c0 : SUCCESS in 1h 05m 02s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/866a80c729fc4add9ec90d9201402823 : SUCCESS in 5m 03s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/59b23d157fcd4a78a38af051975a9565 : SUCCESS in 3m 53s\n- openstack-tox-py27 https://zuul.opendev.org/t/openstack/build/7afb3df77d10495bb31b043fe5951306 : SUCCESS in 3m 48s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/16f501d2c90e4e4182b9979f5a6080a3 : SUCCESS in 3m 33s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/b32dc9f27af940149ee066cc132edbdd : SUCCESS in 3m 31s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/d1471e584c764f7597c91595716ec2a2 : SUCCESS in 4m 43s\n- kolla-build-ubuntu-source https://zuul.opendev.org/t/openstack/build/2cdca9c7c4e9429aac70cf389ab17afc : SUCCESS in 37m 38s\n- kolla-ansible-ubuntu-source https://zuul.opendev.org/t/openstack/build/e2db2172d380405eb5d66531999c2aac : SUCCESS in 1h 02m 01s\n- kolla-ansible-ubuntu-source-upgrade https://zuul.opendev.org/t/openstack/build/d1ea590483fc4a21aab7f851f7ff2145 : SUCCESS in 1h 04m 29s","accounts_in_message":[],"_revision_number":1},{"id":"1492e84d7fbb48b4ca269ccd3227939739574607","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-09-19 08:48:56.000000000","message":"Patch Set 1:\n\nBuild succeeded (promote pipeline).\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/131d390f4c2a41268d3ce139579ca33d : SUCCESS in 50s","accounts_in_message":[],"_revision_number":1}],"current_revision_number":1,"current_revision":"2ce77bc7f84a5a662a81c3cacb3a187e02fa7e98","revisions":{"2ce77bc7f84a5a662a81c3cacb3a187e02fa7e98":{"kind":"REWORK","_number":1,"created":"2019-09-17 15:11:48.000000000","uploader":{"_account_id":24072,"name":"Marcin Juszkiewicz","email":"mjuszkiewicz@redhat.com","username":"hrw"},"ref":"refs/changes/73/682673/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/kolla","ref":"refs/changes/73/682673/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/kolla refs/changes/73/682673/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/kolla refs/changes/73/682673/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/kolla refs/changes/73/682673/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/kolla refs/changes/73/682673/1"}}},"commit":{"parents":[{"commit":"d5f081a879b46427ad3d7720bd43767f0c3e7ae5","subject":"Merge \"Remove exemption for non-systemd Ubuntu in bifrost\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/d5f081a879b46427ad3d7720bd43767f0c3e7ae5"}]}],"author":{"name":"Marcin Juszkiewicz","email":"marcin.juszkiewicz@linaro.org","date":"2019-09-17 15:11:37.000000000","tz":120},"committer":{"name":"Marcin Juszkiewicz","email":"marcin.juszkiewicz@linaro.org","date":"2019-09-17 15:11:37.000000000","tz":120},"subject":"build.py: fix pep8 warnings","message":"build.py: fix pep8 warnings\n\nChange-Id: Ifaeed6917636e2bcb5ea3bda32ae0393ebb08a4f\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/2ce77bc7f84a5a662a81c3cacb3a187e02fa7e98"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/2ce77bc7f84a5a662a81c3cacb3a187e02fa7e98"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[],"submit_requirements":[]}
