)]}'
{"id":"openstack%2Fkolla~682927","triplet_id":"openstack%2Fkolla~stable%2Frocky~I7603ed700abd2d6ce3ac0f301dc433ed1ed11e64","project":"openstack/kolla","branch":"stable/rocky","hashtags":[],"change_id":"I7603ed700abd2d6ce3ac0f301dc433ed1ed11e64","subject":"base: fix OracleLinux build","status":"MERGED","created":"2019-09-18 15:08:54.000000000","updated":"2019-09-19 07:38:17.000000000","submitted":"2019-09-19 07:38:17.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"682927-1568878697877-d349e51d","meta_rev_id":"4ee3cc128415265a9e192952dac0a68585f0606a","_number":682927,"virtual_id_number":682927,"owner":{"_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":24072,"name":"Marcin Juszkiewicz","email":"mjuszkiewicz@redhat.com","username":"hrw"},{"value":0,"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},{"value":2,"date":"2019-09-19 07:38:17.000000000","post_submit":true,"permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":24072,"name":"Marcin Juszkiewicz","email":"mjuszkiewicz@redhat.com","username":"hrw"},"all":[{"value":2,"date":"2019-09-18 15:09:35.000000000","_account_id":24072,"name":"Marcin Juszkiewicz","email":"mjuszkiewicz@redhat.com","username":"hrw"},{"value":2,"date":"2019-09-18 16:18:07.000000000","_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":24072,"name":"Marcin Juszkiewicz","email":"mjuszkiewicz@redhat.com","username":"hrw"},"all":[{"value":1,"date":"2019-09-19 06:37:09.000000000","_account_id":24072,"name":"Marcin Juszkiewicz","email":"mjuszkiewicz@redhat.com","username":"hrw"},{"value":0,"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Backport-Candidate":{"all":[{"value":0,"_account_id":24072,"name":"Marcin Juszkiewicz","email":"mjuszkiewicz@redhat.com","username":"hrw"},{"value":0,"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Do Not Backport"," 0":"Backport Review Needed","+1":"Should Backport"},"description":"","default_value":0,"optional":true},"Review-Priority":{"all":[{"value":0,"_account_id":24072,"name":"Marcin Juszkiewicz","email":"mjuszkiewicz@redhat.com","username":"hrw"},{"value":0,"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Branch Freeze"," 0":"No Priority","+1":"Important Change","+2":"Gate Blocker Fix / Urgent Change"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":24072,"name":"Marcin Juszkiewicz","email":"mjuszkiewicz@redhat.com","username":"hrw"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2019-09-18 16:18:07.000000000","updated_by":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},"reviewer":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},"state":"REVIEWER"},{"updated":"2019-09-19 06:37:09.000000000","updated_by":{"_account_id":24072,"name":"Marcin Juszkiewicz","email":"mjuszkiewicz@redhat.com","username":"hrw"},"reviewer":{"_account_id":24072,"name":"Marcin Juszkiewicz","email":"mjuszkiewicz@redhat.com","username":"hrw"},"state":"REVIEWER"},{"updated":"2019-09-19 07:38:17.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"}],"messages":[{"id":"4797ba7d5aac9ef2c53b971eed734d971e4daa11","author":{"_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"},"date":"2019-09-18 15:08:54.000000000","message":"Patch Set 1: Cherry Picked from branch stable/stein.","accounts_in_message":[],"_revision_number":1},{"id":"4472db832940c35a638d22f7f34741495f1df361","author":{"_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"},"date":"2019-09-18 15:09:01.000000000","message":"Patch Set 1: Cherry Picked\n\nThis patchset was cherry picked to branch stable/queens as commit ea34f0e19f17fefd3897deb9332a7a4596de5f6b","accounts_in_message":[],"_revision_number":1},{"id":"da854251b21f3a828dfa35b4df7bfc6f1db7e210","author":{"_account_id":24072,"name":"Marcin Juszkiewicz","email":"mjuszkiewicz@redhat.com","username":"hrw"},"date":"2019-09-18 15:09:35.000000000","message":"Patch Set 1: Code-Review+2","accounts_in_message":[],"_revision_number":1},{"id":"1bf93b8ef9e5c499dbf61fb068b8ca1dfb074097","author":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},"date":"2019-09-18 16:18:07.000000000","message":"Patch Set 1: Code-Review+2","accounts_in_message":[],"_revision_number":1},{"id":"3c7fddd5649bd04533000c197110cf7b506a6994","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-09-18 17:12:59.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\n\n- kolla-build-centos-binary https://zuul.opendev.org/t/openstack/build/855a810d704a44a8b5322c1ee7603018 : SUCCESS in 43m 47s (non-voting)\n- kolla-build-centos-source https://zuul.opendev.org/t/openstack/build/1c3b833d14374210a8ce458477c45425 : SUCCESS in 42m 15s\n- kolla-ansible-centos-source https://zuul.opendev.org/t/openstack/build/de6ce53157d04d7083e63eef59541dad : SUCCESS in 1h 29m 04s\n- kolla-ansible-centos-binary https://zuul.opendev.org/t/openstack/build/129b6af84c654f8fbb373f952f2f2f1f : TIMED_OUT in 2h 01m 40s (non-voting)\n- kolla-build-oraclelinux-binary https://zuul.opendev.org/t/openstack/build/361eec6952914d8fbb48ab9642febee2 : SUCCESS in 56m 11s (non-voting)\n- kolla-build-oraclelinux-source https://zuul.opendev.org/t/openstack/build/5c91182924d046d6bcfb614e59aa7522 : SUCCESS in 1h 06m 23s\n- kolla-ansible-oraclelinux-source https://zuul.opendev.org/t/openstack/build/f07407eb222944e3a3b85f0ef6aac79d : SUCCESS in 1h 23m 08s (non-voting)\n- kolla-ansible-oraclelinux-binary https://zuul.opendev.org/t/openstack/build/ce41edcf47c4458b9371d1f3fdddaeec : SUCCESS in 1h 18m 19s (non-voting)\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/2258575236e24018a8a45e8a2258fb49 : SUCCESS in 3m 47s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/7d59c35f75884e4083ac1761d3edbdb6 : SUCCESS in 4m 52s\n- openstack-tox-py27 https://zuul.opendev.org/t/openstack/build/f38b0d33ed304415b1a4f23108d5c4d9 : SUCCESS in 2m 57s\n- openstack-tox-py35 https://zuul.opendev.org/t/openstack/build/564da7603231434a950ecae752b020df : SUCCESS in 4m 48s\n- build-openstack-sphinx-docs https://zuul.opendev.org/t/openstack/build/1a6245ad91664499a29f38e4aa204376 : SUCCESS in 5m 59s\n- kolla-build-ubuntu-binary https://zuul.opendev.org/t/openstack/build/37b16b1f4955450bad30c446e6f44b30 : SUCCESS in 1h 30m 08s (non-voting)\n- kolla-build-ubuntu-source https://zuul.opendev.org/t/openstack/build/9002d050f56d4e32b168add1e5dfd502 : SUCCESS in 33m 19s\n- kolla-ansible-ubuntu-source https://zuul.opendev.org/t/openstack/build/9e2d50d756824c87b70cd22485b7ddf5 : SUCCESS in 53m 00s\n- kolla-ansible-ubuntu-binary https://zuul.opendev.org/t/openstack/build/f9aef63c3eaa47d088e31e5145c74461 : SUCCESS in 1h 34m 43s (non-voting)","accounts_in_message":[],"_revision_number":1},{"id":"c323bc3a6ca2cf8195144d279c981ca22991aa97","author":{"_account_id":24072,"name":"Marcin Juszkiewicz","email":"mjuszkiewicz@redhat.com","username":"hrw"},"date":"2019-09-19 06:37:09.000000000","message":"Patch Set 1: Workflow+1","accounts_in_message":[],"_revision_number":1},{"id":"16df5359e70859c34b2d93b7efa6577a2e0bbac1","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-09-19 06:37:23.000000000","message":"Patch Set 1: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":1},{"id":"52ef7831124449219252b44dc6fefeeb2398fe57","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-09-19 07:38:17.000000000","message":"Change has been successfully merged by Zuul","accounts_in_message":[],"_revision_number":1},{"id":"4ee3cc128415265a9e192952dac0a68585f0606a","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-09-19 07:38:17.000000000","message":"Patch Set 1: Verified+2\n\nBuild succeeded (gate pipeline).\n\n- kolla-build-centos-source https://zuul.opendev.org/t/openstack/build/042b946abbf54f8e954c4eebbd34d6c7 : SUCCESS in 53m 53s\n- kolla-build-oraclelinux-source https://zuul.opendev.org/t/openstack/build/938e896c21a5434c95ddb88474da27a2 : SUCCESS in 59m 28s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/7663338d40aa45d292252865cb6c4386 : SUCCESS in 7m 59s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/72a2af0e91d64bfea8d4a14b35c0ceb0 : SUCCESS in 3m 30s\n- openstack-tox-py27 https://zuul.opendev.org/t/openstack/build/330e24cec4b446c1a5620629fb1909fa : SUCCESS in 4m 40s\n- openstack-tox-py35 https://zuul.opendev.org/t/openstack/build/e0d638c82a93477d987432b76d3362cd : SUCCESS in 2m 48s\n- build-openstack-sphinx-docs https://zuul.opendev.org/t/openstack/build/e726ae7f81b7470586f23b8ff2796843 : SUCCESS in 4m 04s\n- kolla-build-ubuntu-source https://zuul.opendev.org/t/openstack/build/f0b67fc21a884541a4d0783d4c6e5435 : SUCCESS in 55m 09s","accounts_in_message":[],"_revision_number":1}],"current_revision_number":1,"current_revision":"5693bfd698a9f00f5fefd0ab0c4c3c38628223bb","revisions":{"5693bfd698a9f00f5fefd0ab0c4c3c38628223bb":{"kind":"REWORK","_number":1,"created":"2019-09-18 15:08:54.000000000","uploader":{"_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"},"ref":"refs/changes/27/682927/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/kolla","ref":"refs/changes/27/682927/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/kolla refs/changes/27/682927/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/kolla refs/changes/27/682927/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/kolla refs/changes/27/682927/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/kolla refs/changes/27/682927/1"}}},"commit":{"parents":[{"commit":"8b6a7845f331a39e6eaef899b0d2a3ccc7336379","subject":"Add pciutils package to zun-compute","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/8b6a7845f331a39e6eaef899b0d2a3ccc7336379"}]}],"author":{"name":"Marcin Juszkiewicz","email":"marcin.juszkiewicz@linaro.org","date":"2019-09-18 09:45:49.000000000","tz":120},"committer":{"name":"Radosław Piliszek","email":"radoslaw.piliszek@gmail.com","date":"2019-09-18 15:08:54.000000000","tz":0},"subject":"base: fix OracleLinux build","message":"base: fix OracleLinux build\n\nChange-Id: I7603ed700abd2d6ce3ac0f301dc433ed1ed11e64\nCloses-Bug: #1844484\n(cherry picked from commit 187d6c52f1bbb1176d7129a2b24b0cf48ba29d79)\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/5693bfd698a9f00f5fefd0ab0c4c3c38628223bb"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/5693bfd698a9f00f5fefd0ab0c4c3c38628223bb"}]},"branch":"refs/heads/stable/rocky"}},"requirements":[],"submit_records":[],"submit_requirements":[]}
