)]}'
{"id":"openstack%2Fkolla~682928","triplet_id":"openstack%2Fkolla~stable%2Fqueens~I7603ed700abd2d6ce3ac0f301dc433ed1ed11e64","project":"openstack/kolla","branch":"stable/queens","hashtags":[],"change_id":"I7603ed700abd2d6ce3ac0f301dc433ed1ed11e64","subject":"base: fix OracleLinux build","status":"MERGED","created":"2019-09-18 15:09:01.000000000","updated":"2019-09-19 08:55:55.000000000","submitted":"2019-09-19 07:45:52.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"682928-1568879152427-ba73d174","meta_rev_id":"d06661d3345586af19b192c5a5390d16f9fabd2a","_number":682928,"virtual_id_number":682928,"owner":{"_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":24072,"name":"Marcin Juszkiewicz","email":"mjuszkiewicz@redhat.com","username":"hrw"},{"value":0,"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},{"value":2,"date":"2019-09-19 07:45:52.000000000","post_submit":true,"permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":24072,"name":"Marcin Juszkiewicz","email":"mjuszkiewicz@redhat.com","username":"hrw"},"all":[{"value":2,"date":"2019-09-18 15:09:27.000000000","_account_id":24072,"name":"Marcin Juszkiewicz","email":"mjuszkiewicz@redhat.com","username":"hrw"},{"value":2,"date":"2019-09-18 16:18:16.000000000","_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":24072,"name":"Marcin Juszkiewicz","email":"mjuszkiewicz@redhat.com","username":"hrw"},"all":[{"value":1,"date":"2019-09-19 06:37:20.000000000","_account_id":24072,"name":"Marcin Juszkiewicz","email":"mjuszkiewicz@redhat.com","username":"hrw"},{"value":0,"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Backport-Candidate":{"all":[{"value":0,"_account_id":24072,"name":"Marcin Juszkiewicz","email":"mjuszkiewicz@redhat.com","username":"hrw"},{"value":0,"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Do Not Backport"," 0":"Backport Review Needed","+1":"Should Backport"},"description":"","default_value":0,"optional":true},"Review-Priority":{"all":[{"value":0,"_account_id":24072,"name":"Marcin Juszkiewicz","email":"mjuszkiewicz@redhat.com","username":"hrw"},{"value":0,"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Branch Freeze"," 0":"No Priority","+1":"Important Change","+2":"Gate Blocker Fix / Urgent Change"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":24072,"name":"Marcin Juszkiewicz","email":"mjuszkiewicz@redhat.com","username":"hrw"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2019-09-18 16:18:16.000000000","updated_by":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},"reviewer":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},"state":"REVIEWER"},{"updated":"2019-09-19 06:37:20.000000000","updated_by":{"_account_id":24072,"name":"Marcin Juszkiewicz","email":"mjuszkiewicz@redhat.com","username":"hrw"},"reviewer":{"_account_id":24072,"name":"Marcin Juszkiewicz","email":"mjuszkiewicz@redhat.com","username":"hrw"},"state":"REVIEWER"},{"updated":"2019-09-19 07:45:52.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"}],"messages":[{"id":"ea04dbda94f7cc8f95653963ebd7f0f83f2898cc","author":{"_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"},"date":"2019-09-18 15:09:01.000000000","message":"Patch Set 1: Cherry Picked from branch stable/rocky.","accounts_in_message":[],"_revision_number":1},{"id":"9b75d432b72172aed3313630896c94da34d9b05d","author":{"_account_id":24072,"name":"Marcin Juszkiewicz","email":"mjuszkiewicz@redhat.com","username":"hrw"},"date":"2019-09-18 15:09:27.000000000","message":"Patch Set 1: Code-Review+2","accounts_in_message":[],"_revision_number":1},{"id":"be831ed886b0b7bfa0e80e8882b013c5c89a4d47","author":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},"date":"2019-09-18 16:18:16.000000000","message":"Patch Set 1: Code-Review+2","accounts_in_message":[],"_revision_number":1},{"id":"2697f215d08e137b2e0b4d7ac2a7e8af07233481","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-09-18 17:14:34.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\n\n- kolla-build-centos-binary https://zuul.opendev.org/t/openstack/build/23d49d946f7d4bc3bfb91a9db32c68e6 : SUCCESS in 35m 20s (non-voting)\n- kolla-build-centos-source https://zuul.opendev.org/t/openstack/build/5864bc4397b2442c96888481d08a60a1 : SUCCESS in 47m 43s\n- kolla-ansible-centos-source https://zuul.opendev.org/t/openstack/build/6673e1f23f66494fa5c50615a263fc0b : SUCCESS in 1h 22m 29s (non-voting)\n- kolla-ansible-centos-binary https://zuul.opendev.org/t/openstack/build/d33e4f69d6114223957e0bc07512dfef : SUCCESS in 1h 15m 26s (non-voting)\n- kolla-build-oraclelinux-binary https://zuul.opendev.org/t/openstack/build/b27632b98be8469480138bceb5bf293b : SUCCESS in 41m 49s (non-voting)\n- kolla-build-oraclelinux-source https://zuul.opendev.org/t/openstack/build/a47d10ac4e744b54902556b7a7643f76 : SUCCESS in 45m 19s\n- kolla-ansible-oraclelinux-source https://zuul.opendev.org/t/openstack/build/a05c8b03b67d49378d2b319a26860fe5 : SUCCESS in 1h 22m 34s (non-voting)\n- kolla-ansible-oraclelinux-binary https://zuul.opendev.org/t/openstack/build/72a5cbbf3c0043708ca7eecc206a2b09 : TIMED_OUT in 2h 01m 22s (non-voting)\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/8e5a69e19726438dbe8fec75a57a63ad : SUCCESS in 3m 24s\n- openstack-tox-py27 https://zuul.opendev.org/t/openstack/build/64f0ebf07c6841abb09fd5389ed1d35d : SUCCESS in 5m 00s\n- openstack-tox-py35 https://zuul.opendev.org/t/openstack/build/186c4cd5a13f478eaa3700b48c7b79bd : SUCCESS in 5m 17s\n- build-openstack-sphinx-docs https://zuul.opendev.org/t/openstack/build/975dde525b5741dba1f874eeb8863b38 : SUCCESS in 6m 45s\n- kolla-build-ubuntu-binary https://zuul.opendev.org/t/openstack/build/0eca67f981f449a8a7c22057e6b06bbf : SUCCESS in 27m 21s (non-voting)\n- kolla-build-ubuntu-source https://zuul.opendev.org/t/openstack/build/63343ece8c9442f2accbbfef70d9a929 : SUCCESS in 42m 12s\n- kolla-ansible-ubuntu-source https://zuul.opendev.org/t/openstack/build/d58d955847ff461a9e3cc55a2bf3f3f6 : SUCCESS in 53m 34s (non-voting)\n- kolla-ansible-ubuntu-binary https://zuul.opendev.org/t/openstack/build/783b43c48e65447e91bdc62b3d04dc37 : SUCCESS in 59m 18s (non-voting)","accounts_in_message":[],"_revision_number":1},{"id":"e94071940b4795457ed2965181b38d3e953c9e65","author":{"_account_id":24072,"name":"Marcin Juszkiewicz","email":"mjuszkiewicz@redhat.com","username":"hrw"},"date":"2019-09-19 06:37:20.000000000","message":"Patch Set 1: Workflow+1","accounts_in_message":[],"_revision_number":1},{"id":"fb048124b84364def30d5d2030b82c6014e95aeb","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-09-19 06:37:33.000000000","message":"Patch Set 1: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":1},{"id":"6b22bc5a632832900d06eb7e44958b21cacd127a","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-09-19 07:45:52.000000000","message":"Change has been successfully merged by Zuul","accounts_in_message":[],"_revision_number":1},{"id":"c14df13d1d65df6a8bed1ed8fcfdff9a75baf240","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-09-19 07:45:52.000000000","message":"Patch Set 1: Verified+2\n\nBuild succeeded (gate pipeline).\n\n- kolla-build-centos-source https://zuul.opendev.org/t/openstack/build/81dde676980c4cd19572aad20f961c45 : SUCCESS in 50m 15s\n- kolla-build-oraclelinux-source https://zuul.opendev.org/t/openstack/build/4bc3eea477f0476a921c9626c9b7e7f2 : SUCCESS in 50m 49s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/bd648a5f7e1346db9b784de0b4ddfe31 : SUCCESS in 3m 27s\n- openstack-tox-py27 https://zuul.opendev.org/t/openstack/build/4c0c4881acc4401294571ecbd77b32f4 : SUCCESS in 3m 41s\n- openstack-tox-py35 https://zuul.opendev.org/t/openstack/build/f02780172d02411d8043468e4d9bb818 : SUCCESS in 3m 48s\n- build-openstack-sphinx-docs https://zuul.opendev.org/t/openstack/build/60bddd4f5e5d418885cefe11cdac9125 : SUCCESS in 3m 57s\n- kolla-build-ubuntu-source https://zuul.opendev.org/t/openstack/build/4d5cbf1abc304d8987e052f682d01afa : SUCCESS in 1h 01m 07s","accounts_in_message":[],"_revision_number":1},{"id":"d06661d3345586af19b192c5a5390d16f9fabd2a","author":{"_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"},"date":"2019-09-19 08:55:55.000000000","message":"Patch Set 1: Cherry Picked\n\nThis patchset was cherry picked to branch stable/pike as commit 18507dbc1f2ea3ae31108ba00ae5a3267028c2fb","accounts_in_message":[],"_revision_number":1}],"current_revision_number":1,"current_revision":"ea34f0e19f17fefd3897deb9332a7a4596de5f6b","revisions":{"ea34f0e19f17fefd3897deb9332a7a4596de5f6b":{"kind":"REWORK","_number":1,"created":"2019-09-18 15:09:01.000000000","uploader":{"_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"},"ref":"refs/changes/28/682928/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/kolla","ref":"refs/changes/28/682928/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/kolla refs/changes/28/682928/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/kolla refs/changes/28/682928/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/kolla refs/changes/28/682928/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/kolla refs/changes/28/682928/1"}}},"commit":{"parents":[{"commit":"30142a2094959d2e013dc8e9f3e68f5bbb3076c2","subject":"Add config option use_dumb_init to add it to container images","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/30142a2094959d2e013dc8e9f3e68f5bbb3076c2"}]}],"author":{"name":"Marcin Juszkiewicz","email":"marcin.juszkiewicz@linaro.org","date":"2019-09-18 09:45:49.000000000","tz":120},"committer":{"name":"Radosław Piliszek","email":"radoslaw.piliszek@gmail.com","date":"2019-09-18 15:09:01.000000000","tz":0},"subject":"base: fix OracleLinux build","message":"base: fix OracleLinux build\n\nChange-Id: I7603ed700abd2d6ce3ac0f301dc433ed1ed11e64\nCloses-Bug: #1844484\n(cherry picked from commit 187d6c52f1bbb1176d7129a2b24b0cf48ba29d79)\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/ea34f0e19f17fefd3897deb9332a7a4596de5f6b"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/ea34f0e19f17fefd3897deb9332a7a4596de5f6b"}]},"branch":"refs/heads/stable/queens"}},"requirements":[],"submit_records":[],"submit_requirements":[]}
