)]}'
{"id":"openstack%2Fkolla~683117","triplet_id":"openstack%2Fkolla~master~I6ec9d58bf54db1ac69b879785f38fb492bc2dfc2","project":"openstack/kolla","branch":"master","topic":"683117","hashtags":[],"change_id":"I6ec9d58bf54db1ac69b879785f38fb492bc2dfc2","subject":"sensu-client: install older whois gem","status":"MERGED","created":"2019-09-19 11:51:21.000000000","updated":"2019-09-20 10:49:39.000000000","submitted":"2019-09-19 17:40:40.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":1,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"683117-1568914840719-8c653045","meta_rev_id":"968be684ea05fb129b4279be975971178bfe3133","_number":683117,"virtual_id_number":683117,"owner":{"_account_id":24072,"name":"Marcin Juszkiewicz","email":"mjuszkiewicz@redhat.com","username":"hrw"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},{"value":0,"_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"},{"value":2,"date":"2019-09-19 17:40:40.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"date":"2019-09-19 13:56:24.000000000","_account_id":23181,"name":"RDO Third Party CI","email":"dmsimard+rdothirdparty@redhat.com","username":"rdothirdparty","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"},"all":[{"value":0,"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},{"value":2,"date":"2019-09-19 16:03:57.000000000","_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":23181,"name":"RDO Third Party CI","email":"dmsimard+rdothirdparty@redhat.com","username":"rdothirdparty","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"},"all":[{"value":0,"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},{"value":1,"date":"2019-09-19 16:03:57.000000000","_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":23181,"name":"RDO Third Party CI","email":"dmsimard+rdothirdparty@redhat.com","username":"rdothirdparty","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Backport-Candidate":{"all":[{"value":0,"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},{"value":0,"_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":23181,"name":"RDO Third Party CI","email":"dmsimard+rdothirdparty@redhat.com","username":"rdothirdparty","tags":["SERVICE_USER"]}],"values":{"-1":"Do Not Backport"," 0":"Backport Review Needed","+1":"Should Backport"},"description":"","default_value":0,"optional":true},"Review-Priority":{"all":[{"value":0,"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},{"value":0,"_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":23181,"name":"RDO Third Party CI","email":"dmsimard+rdothirdparty@redhat.com","username":"rdothirdparty","tags":["SERVICE_USER"]}],"values":{"-1":"Branch Freeze"," 0":"No Priority","+1":"Important Change","+2":"Gate Blocker Fix / Urgent Change"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":23181,"name":"RDO Third Party CI","email":"dmsimard+rdothirdparty@redhat.com","username":"rdothirdparty","tags":["SERVICE_USER"]},{"_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2019-09-19 13:10:15.000000000","updated_by":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},"reviewer":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},"state":"REVIEWER"},{"updated":"2019-09-19 13:56:24.000000000","updated_by":{"_account_id":23181,"name":"RDO Third Party CI","email":"dmsimard+rdothirdparty@redhat.com","username":"rdothirdparty","tags":["SERVICE_USER"]},"reviewer":{"_account_id":23181,"name":"RDO Third Party CI","email":"dmsimard+rdothirdparty@redhat.com","username":"rdothirdparty","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2019-09-19 16:03:57.000000000","updated_by":{"_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"},"reviewer":{"_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"},"state":"REVIEWER"},{"updated":"2019-09-19 17:40:40.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"}],"messages":[{"id":"9ece97cff515400da3cf3b3ac5b2b2352b0eb15d","author":{"_account_id":24072,"name":"Marcin Juszkiewicz","email":"mjuszkiewicz@redhat.com","username":"hrw"},"date":"2019-09-19 11:51:21.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"2f061004e4b5fce39cb5b58c572844e499639d43","author":{"_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"},"date":"2019-09-19 11:52:56.000000000","message":"Patch Set 2: Published edit on patch set 1.","accounts_in_message":[],"_revision_number":2},{"id":"63a4b44369fe4128d1d88a8f99829bf6e758435e","author":{"_account_id":23181,"name":"RDO Third Party CI","email":"dmsimard+rdothirdparty@redhat.com","username":"rdothirdparty","tags":["SERVICE_USER"]},"date":"2019-09-19 12:03:40.000000000","message":"Patch Set 2:\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttp://docs.openstack.org/infra/manual/developers.html#automated-testing\n\n\n- tripleo-build-containers-rhel-8-master http://logs.rdoproject.org/17/683117/2/openstack-check/tripleo-build-containers-rhel-8-master/91cb0ce/ : FAILURE in 9m 56s","accounts_in_message":[],"_revision_number":2},{"id":"400b76ea7b0eabe790af1590022b2e1937cd86c4","author":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},"date":"2019-09-19 13:10:15.000000000","message":"Patch Set 2:\n\n(1 comment)","accounts_in_message":[],"_revision_number":2},{"id":"380fa52fda970a925801faeb330f3cd9da8e06b2","author":{"_account_id":24072,"name":"Marcin Juszkiewicz","email":"mjuszkiewicz@redhat.com","username":"hrw"},"date":"2019-09-19 13:45:39.000000000","message":"Uploaded patch set 3.","accounts_in_message":[],"_revision_number":3},{"id":"c503f00a056e3d2fc96dc26fabb86e77291deba0","author":{"_account_id":23181,"name":"RDO Third Party CI","email":"dmsimard+rdothirdparty@redhat.com","username":"rdothirdparty","tags":["SERVICE_USER"]},"date":"2019-09-19 13:56:24.000000000","message":"Patch Set 3:\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttp://docs.openstack.org/infra/manual/developers.html#automated-testing\n\n\n- tripleo-build-containers-rhel-8-master http://logs.rdoproject.org/17/683117/3/openstack-check/tripleo-build-containers-rhel-8-master/e18fd8f/ : FAILURE in 9m 53s","accounts_in_message":[],"_revision_number":3},{"id":"b50ab76ad05059bf10b983ca1ff1e6236c707619","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-09-19 15:35:21.000000000","message":"Patch Set 3: Verified+1\n\nBuild succeeded (check pipeline).\n\n- kolla-build-centos-binary https://zuul.opendev.org/t/openstack/build/46db295fe49b47779d4d78989eb363ad : FAILURE in 1h 06m 46s (non-voting)\n- kolla-build-centos-source https://zuul.opendev.org/t/openstack/build/8e09834d7ed94d77ab5598467ffd9d66 : SUCCESS in 48m 57s\n- kolla-ansible-centos-source https://zuul.opendev.org/t/openstack/build/a3a3a736f2b04e94b22517b896178662 : SUCCESS in 1h 29m 58s\n- kolla-ansible-centos-binary https://zuul.opendev.org/t/openstack/build/69cf3344d26f4e76bc665cf11c6fbd9e : SUCCESS in 1h 23m 48s (non-voting)\n- kolla-ansible-centos-source-upgrade https://zuul.opendev.org/t/openstack/build/55f38219b9434d44a4232ed92178bfb3 : SUCCESS in 1h 25m 54s\n- tripleo-build-containers-centos-7 https://zuul.opendev.org/t/openstack/build/03979cba04cd4c66896eecf0ca73c749 : SUCCESS in 57m 09s\n- kolla-build-debian-source https://zuul.opendev.org/t/openstack/build/9970fc356f4d44d4bbd5b95d6a602938 : FAILURE in 11m 11s (non-voting)\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/379307aa89fa40a8aa9dccffd791aa0d : SUCCESS in 5m 21s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/7d9ffd4e68f746e2a98772f290d30940 : SUCCESS in 5m 04s\n- openstack-tox-py27 https://zuul.opendev.org/t/openstack/build/1c6ec6d9b0524f65a08f19475df2ff3e : SUCCESS in 3m 13s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/4235c981f283412ea29d2bfc0e416876 : SUCCESS in 3m 17s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/a6301ece1aa04f6e8899d4022d64de82 : SUCCESS in 4m 16s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/702ddf37c612400192212535f65833af : SUCCESS in 5m 03s\n- kolla-build-ubuntu-binary https://zuul.opendev.org/t/openstack/build/1c4a870542474842b5a8dcb4449993a9 : SUCCESS in 45m 54s (non-voting)\n- kolla-build-ubuntu-source https://zuul.opendev.org/t/openstack/build/65bbd6c2613e4804bb55d1b0550e47c2 : SUCCESS in 31m 27s\n- kolla-ansible-ubuntu-source https://zuul.opendev.org/t/openstack/build/516eabd4c7dc42ed90c18909285910c1 : SUCCESS in 1h 04m 49s\n- kolla-ansible-ubuntu-binary https://zuul.opendev.org/t/openstack/build/7d34898e32cc43d0983f9d28814a9b29 : SUCCESS in 1h 05m 02s (non-voting)\n- kolla-ansible-ubuntu-source-upgrade https://zuul.opendev.org/t/openstack/build/91c822466f5d4c859f98bec4af8cd07a : SUCCESS in 1h 05m 38s","accounts_in_message":[],"_revision_number":3},{"id":"cb8f8fa6f028729ddc85c752cc3a8b67c54c6220","author":{"_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"},"date":"2019-09-19 16:03:57.000000000","message":"Patch Set 3: Code-Review+2 Workflow+1\n\nrandom issues with mirror, issuing unbreak","accounts_in_message":[],"_revision_number":3},{"id":"5fd1917afd11e326817dbb4d8de03d8801b670e9","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-09-19 16:04:10.000000000","message":"Patch Set 3: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":3},{"id":"0da00d94af60d807906cdc34a9bffe63a03b9d0c","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-09-19 17:40:40.000000000","message":"Patch Set 3: Verified+2\n\nBuild succeeded (gate pipeline).\n\n- kolla-build-centos-source https://zuul.opendev.org/t/openstack/build/7c11615c5cf24b948b2315b7128e779b : SUCCESS in 48m 25s\n- kolla-ansible-centos-source https://zuul.opendev.org/t/openstack/build/23e4e1a09f3c4e4e86574740861f1c4f : SUCCESS in 1h 22m 10s\n- kolla-ansible-centos-source-upgrade https://zuul.opendev.org/t/openstack/build/52d7a5114704422c84122751d568c684 : SUCCESS in 1h 34m 28s\n- tripleo-build-containers-centos-7 https://zuul.opendev.org/t/openstack/build/975f3ea68a324f69ab577ac7fa61892e : SUCCESS in 56m 24s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/0c663baf32694ad2a19c525f660555ef : SUCCESS in 3m 27s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/61849551ac994fa7a16f2b896df158d6 : SUCCESS in 4m 55s\n- openstack-tox-py27 https://zuul.opendev.org/t/openstack/build/3fe4045f049144bf9a7dd3812f545d96 : SUCCESS in 2m 41s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/f7188c0ca9454045b1e0ce25036535f9 : SUCCESS in 6m 59s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/d0cb89093ad8454d8ae5c7cebbbd666f : SUCCESS in 5m 03s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/8506279e8d5a477abb6c2322d057f13c : SUCCESS in 5m 09s\n- kolla-build-ubuntu-source https://zuul.opendev.org/t/openstack/build/b5db4f0099ac4810acd84c12c948a794 : SUCCESS in 33m 41s\n- kolla-ansible-ubuntu-source https://zuul.opendev.org/t/openstack/build/a678cccce30344819a2775c77d408ee5 : SUCCESS in 1h 15m 04s\n- kolla-ansible-ubuntu-source-upgrade https://zuul.opendev.org/t/openstack/build/83b0d28120e94872bc2a0ac876576160 : SUCCESS in 53m 09s","accounts_in_message":[],"_revision_number":3},{"id":"0b2c404f3f4b1fa4086c6f8b99dd1226221d35ff","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-09-19 17:40:40.000000000","message":"Change has been successfully merged by Zuul","accounts_in_message":[],"_revision_number":3},{"id":"fa9ba78a2f5b75f8113b0d903cc27ae9f17dd08c","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-09-19 17:42:43.000000000","message":"Patch Set 3:\n\nBuild succeeded (promote pipeline).\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/d0aba214cda44baea85b4d874794581a : SUCCESS in 1m 32s","accounts_in_message":[],"_revision_number":3},{"id":"ab7594ec1c2d369101a1d1a63b605447d256868f","author":{"_account_id":24072,"name":"Marcin Juszkiewicz","email":"mjuszkiewicz@redhat.com","username":"hrw"},"date":"2019-09-20 10:45:55.000000000","message":"Patch Set 3: Cherry Picked\n\nThis patchset was cherry picked to branch stable/stein as commit 44be1294c1b801f01c5f989355c4486afd6b7ef3","accounts_in_message":[],"_revision_number":3},{"id":"3382df1cb9f5e855afa4fd954b04ac2db8b2eb92","author":{"_account_id":24072,"name":"Marcin Juszkiewicz","email":"mjuszkiewicz@redhat.com","username":"hrw"},"date":"2019-09-20 10:49:39.000000000","message":"Patch Set 3: Cherry Picked\n\nThis patchset was cherry picked to branch stable/queens as commit dea7c9e08aeb5dbeac81178725ef2cbe0ef74971","accounts_in_message":[],"_revision_number":3}],"current_revision_number":3,"current_revision":"41408527ae88ac42de11ca2818ff405320795a8b","revisions":{"b680fe6619b424e3ea4520850d489032d85099f1":{"kind":"REWORK","_number":1,"created":"2019-09-19 11:51:21.000000000","uploader":{"_account_id":24072,"name":"Marcin Juszkiewicz","email":"mjuszkiewicz@redhat.com","username":"hrw"},"ref":"refs/changes/17/683117/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/kolla","ref":"refs/changes/17/683117/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/kolla refs/changes/17/683117/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/kolla refs/changes/17/683117/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/kolla refs/changes/17/683117/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/kolla refs/changes/17/683117/1"}}},"commit":{"parents":[{"commit":"49777e0072966154a0b0669fd4d5ea9cc3d12f44","subject":"Merge \"Update invalid link for README\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/49777e0072966154a0b0669fd4d5ea9cc3d12f44"}]}],"author":{"name":"Marcin Juszkiewicz","email":"marcin.juszkiewicz@linaro.org","date":"2019-09-19 11:50:41.000000000","tz":120},"committer":{"name":"Marcin Juszkiewicz","email":"marcin.juszkiewicz@linaro.org","date":"2019-09-19 11:50:44.000000000","tz":120},"subject":"sensu-client: install older whois gem","message":"sensu-client: install older whois gem\n\nWhois 5.0.0 requires Ruby 2.4 while CentOS has only 2.0 version.\n\nChange-Id: I6ec9d58bf54db1ac69b879785f38fb492bc2dfc2\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/b680fe6619b424e3ea4520850d489032d85099f1"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/b680fe6619b424e3ea4520850d489032d85099f1"}]},"branch":"refs/heads/master"},"aa255444a99e5b1e2fd964d5c8cd2c1b2d234a69":{"kind":"REWORK","_number":2,"created":"2019-09-19 11:52:56.000000000","uploader":{"_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"},"ref":"refs/changes/17/683117/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/kolla","ref":"refs/changes/17/683117/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/kolla refs/changes/17/683117/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/kolla refs/changes/17/683117/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/kolla refs/changes/17/683117/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/kolla refs/changes/17/683117/2"}}},"commit":{"parents":[{"commit":"49777e0072966154a0b0669fd4d5ea9cc3d12f44","subject":"Merge \"Update invalid link for README\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/49777e0072966154a0b0669fd4d5ea9cc3d12f44"}]}],"author":{"name":"Marcin Juszkiewicz","email":"marcin.juszkiewicz@linaro.org","date":"2019-09-19 11:50:41.000000000","tz":120},"committer":{"name":"Radosław Piliszek","email":"radoslaw.piliszek@gmail.com","date":"2019-09-19 11:52:50.000000000","tz":0},"subject":"sensu-client: install older whois gem","message":"sensu-client: install older whois gem\n\nWhois 5.0.0 requires Ruby 2.4 while CentOS has only 2.0 version.\n\nChange-Id: I6ec9d58bf54db1ac69b879785f38fb492bc2dfc2\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/aa255444a99e5b1e2fd964d5c8cd2c1b2d234a69"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/aa255444a99e5b1e2fd964d5c8cd2c1b2d234a69"}]},"branch":"refs/heads/master"},"41408527ae88ac42de11ca2818ff405320795a8b":{"kind":"REWORK","_number":3,"created":"2019-09-19 13:45:39.000000000","uploader":{"_account_id":24072,"name":"Marcin Juszkiewicz","email":"mjuszkiewicz@redhat.com","username":"hrw"},"ref":"refs/changes/17/683117/3","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/kolla","ref":"refs/changes/17/683117/3","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/kolla refs/changes/17/683117/3 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/kolla refs/changes/17/683117/3 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/kolla refs/changes/17/683117/3 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/kolla refs/changes/17/683117/3"}}},"commit":{"parents":[{"commit":"49777e0072966154a0b0669fd4d5ea9cc3d12f44","subject":"Merge \"Update invalid link for README\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/49777e0072966154a0b0669fd4d5ea9cc3d12f44"}]}],"author":{"name":"Marcin Juszkiewicz","email":"marcin.juszkiewicz@linaro.org","date":"2019-09-19 11:50:41.000000000","tz":120},"committer":{"name":"Marcin Juszkiewicz","email":"marcin.juszkiewicz@linaro.org","date":"2019-09-19 13:45:23.000000000","tz":120},"subject":"sensu-client: install older whois gem","message":"sensu-client: install older whois gem\n\nWhois 5.0.0 requires Ruby 2.4 while CentOS has only 2.0 version.\n\nChange-Id: I6ec9d58bf54db1ac69b879785f38fb492bc2dfc2\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/41408527ae88ac42de11ca2818ff405320795a8b"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/41408527ae88ac42de11ca2818ff405320795a8b"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[],"submit_requirements":[]}
