)]}'
{"id":"openstack%2Fkolla~683359","triplet_id":"openstack%2Fkolla~stable%2Fstein~I6ec9d58bf54db1ac69b879785f38fb492bc2dfc2","project":"openstack/kolla","branch":"stable/stein","topic":"683117-stable/stein","hashtags":[],"change_id":"I6ec9d58bf54db1ac69b879785f38fb492bc2dfc2","subject":"sensu-client: install older whois gem","status":"MERGED","created":"2019-09-20 10:45:55.000000000","updated":"2019-09-20 14:10:52.000000000","submitted":"2019-09-20 14:09:00.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"683359-1568988540677-657de88e","meta_rev_id":"6fdf5fe35bd07cf7528b31d2fe85b199c48954ef","_number":683359,"virtual_id_number":683359,"owner":{"_account_id":24072,"name":"Marcin Juszkiewicz","email":"mjuszkiewicz@redhat.com","username":"hrw"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},{"value":0,"_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"},{"value":2,"date":"2019-09-20 14:09:00.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},"all":[{"value":2,"date":"2019-09-20 11:35:11.000000000","_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},{"value":2,"date":"2019-09-20 11:30:48.000000000","_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},"all":[{"value":1,"date":"2019-09-20 12:28:41.000000000","_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},{"value":0,"_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Backport-Candidate":{"all":[{"value":0,"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},{"value":0,"_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Do Not Backport"," 0":"Backport Review Needed","+1":"Should Backport"},"description":"","default_value":0,"optional":true},"Review-Priority":{"approved":{"_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"},"all":[{"value":0,"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},{"value":2,"date":"2019-09-20 11:30:54.000000000","_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Branch Freeze"," 0":"No Priority","+1":"Important Change","+2":"Gate Blocker Fix / Urgent Change"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2019-09-20 11:30:54.000000000","updated_by":{"_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"},"reviewer":{"_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"},"state":"REVIEWER"},{"updated":"2019-09-20 12:28:41.000000000","updated_by":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},"reviewer":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},"state":"REVIEWER"},{"updated":"2019-09-20 14:09:00.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"}],"messages":[{"id":"27f688b742c9343b2bb4302ca45f47f72e49b389","author":{"_account_id":24072,"name":"Marcin Juszkiewicz","email":"mjuszkiewicz@redhat.com","username":"hrw"},"date":"2019-09-20 10:45:55.000000000","message":"Patch Set 1: Cherry Picked from branch master.","accounts_in_message":[],"_revision_number":1},{"id":"c46eb86c387001f4960aa2c44f5201941aab1a8f","author":{"_account_id":24072,"name":"Marcin Juszkiewicz","email":"mjuszkiewicz@redhat.com","username":"hrw"},"date":"2019-09-20 10:46:21.000000000","message":"Patch Set 1: Cherry Picked\n\nThis patchset was cherry picked to branch stable/rocky as commit cb47fccd13a9fe6b275dbe92cf3e3e7263671fff","accounts_in_message":[],"_revision_number":1},{"id":"be0ebf9fd71d1814ba7f3051f102ee99ad142cd6","author":{"_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"},"date":"2019-09-20 11:30:48.000000000","message":"Patch Set 1: Code-Review+2","accounts_in_message":[],"_revision_number":1},{"id":"755dd46d7136013b1e4d9cd75081a304e295df3e","author":{"_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"},"date":"2019-09-20 11:30:54.000000000","message":"Patch Set 1: Review-Priority+2","accounts_in_message":[],"_revision_number":1},{"id":"87d5b7c430346f60ef69a6f81a268f5b84edd89b","author":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},"date":"2019-09-20 11:35:11.000000000","message":"Patch Set 1: Code-Review+2","accounts_in_message":[],"_revision_number":1},{"id":"49427466c3fce898c23ba3e659240a783337ba93","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-09-20 12:19:51.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\n\n- kolla-build-centos-binary https://zuul.opendev.org/t/openstack/build/67ff6fe28eab4852acfff79a9e483145 : SUCCESS in 47m 24s (non-voting)\n- kolla-build-centos-source https://zuul.opendev.org/t/openstack/build/b87eaae92ae843688ca5e3c4c5938e49 : SUCCESS in 55m 31s\n- kolla-ansible-centos-source https://zuul.opendev.org/t/openstack/build/b7a952116ace40efb0e6033bbeea7de9 : SUCCESS in 1h 16m 49s\n- kolla-ansible-centos-binary https://zuul.opendev.org/t/openstack/build/d2f301d9cb6c4f238c64bb710ae45de1 : SUCCESS in 1h 15m 23s (non-voting)\n- kolla-ansible-centos-source-upgrade https://zuul.opendev.org/t/openstack/build/da5344bc5b394b2aab39ab562653ed49 : SUCCESS in 1h 27m 50s\n- tripleo-build-containers-centos-7 https://zuul.opendev.org/t/openstack/build/02350d22cf0d4806a900d76ae07378e7 : SUCCESS in 1h 04m 28s (non-voting)\n- kolla-build-oraclelinux-binary https://zuul.opendev.org/t/openstack/build/8066e16b7c36466f9d4d7c7ae4b3edad : SUCCESS in 1h 00m 10s (non-voting)\n- kolla-build-oraclelinux-source https://zuul.opendev.org/t/openstack/build/47c12c2ee17a4e3aa764db3af7e0a25e : SUCCESS in 51m 18s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/ef4a16e4a6924c7fbd6cdc0ad1f403cd : SUCCESS in 5m 50s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/d92e66807865453eaff974cbf05a902a : SUCCESS in 3m 06s\n- openstack-tox-py27 https://zuul.opendev.org/t/openstack/build/7065f281a2664e5f990a19cbbe32e22f : SUCCESS in 3m 44s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/62de2c9aa18e4369ab5bd4090e034efb : SUCCESS in 3m 15s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/0d382b7d9cbe4d3d86657f10c82dbae2 : SUCCESS in 4m 32s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/4b641618ab964a42ab9096d8253aa8e4 : SUCCESS in 3m 02s\n- kolla-build-ubuntu-binary https://zuul.opendev.org/t/openstack/build/1e71b6ecdaf04bdf9a0b6f5dac027a2f : SUCCESS in 35m 57s (non-voting)\n- kolla-build-ubuntu-source https://zuul.opendev.org/t/openstack/build/ec5d33e1cd1d481491024701dac2f9c0 : SUCCESS in 34m 23s\n- kolla-ansible-ubuntu-source https://zuul.opendev.org/t/openstack/build/c5def6778d8249b9b06ffcd062e8ad3a : SUCCESS in 55m 09s\n- kolla-ansible-ubuntu-binary https://zuul.opendev.org/t/openstack/build/d02bc1c40aa9422a90710432ad037f1c : SUCCESS in 52m 22s (non-voting)\n- kolla-ansible-ubuntu-source-upgrade https://zuul.opendev.org/t/openstack/build/edc40c3d07784b62b9317f5bcab75932 : SUCCESS in 55m 28s","accounts_in_message":[],"_revision_number":1},{"id":"b9eeea7052c388c3b86809bd5eb65804bbbc79b1","author":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},"date":"2019-09-20 12:28:41.000000000","message":"Patch Set 1: Workflow+1","accounts_in_message":[],"_revision_number":1},{"id":"930bf78d1d0125fa4ea6aa34c982159a33cbb73b","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-09-20 12:28:58.000000000","message":"Patch Set 1: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":1},{"id":"2b1a741b04b316063979e5d0704d8e73694f98b0","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-09-20 14:09:00.000000000","message":"Change has been successfully merged by Zuul","accounts_in_message":[],"_revision_number":1},{"id":"e4e5dbeb98bc3d907cf1c6e0aa1873a80919d37b","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-09-20 14:09:00.000000000","message":"Patch Set 1: Verified+2\n\nBuild succeeded (gate pipeline).\n\n- kolla-build-centos-source https://zuul.opendev.org/t/openstack/build/85015b0242b44adc98fdd98660d1038e : SUCCESS in 55m 13s\n- kolla-build-oraclelinux-source https://zuul.opendev.org/t/openstack/build/4b9da2d1bcfa4e52aacb3f16b7354b9d : SUCCESS in 58m 56s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/456c8f7987f948d791229dd0a88daada : SUCCESS in 4m 24s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/721ac8676cf347d187da3cc0bec5e042 : SUCCESS in 3m 03s\n- openstack-tox-py27 https://zuul.opendev.org/t/openstack/build/5934f39b0e144b0a83039bdc251fb5f1 : SUCCESS in 3m 08s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/c30ce9fc903649dcb1146510ca1b7277 : SUCCESS in 2m 50s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/ec2d98fc3c41456c96a11bb457fbb997 : SUCCESS in 4m 07s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/27bc96395a07467493e4b5eb11783f70 : SUCCESS in 4m 16s\n- kolla-build-ubuntu-source https://zuul.opendev.org/t/openstack/build/4f7df4f8adc542e4a6eb225e85cefac1 : SUCCESS in 34m 20s","accounts_in_message":[],"_revision_number":1},{"id":"07f0d358e3663274e97cda67f5cbf12ef1fc7046","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-09-20 14:10:52.000000000","message":"Patch Set 1:\n\nBuild succeeded (promote pipeline).\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/4d3fad3d6d83483ebf562f15d037b448 : SUCCESS in 1m 31s","accounts_in_message":[],"_revision_number":1}],"current_revision_number":1,"current_revision":"44be1294c1b801f01c5f989355c4486afd6b7ef3","revisions":{"44be1294c1b801f01c5f989355c4486afd6b7ef3":{"kind":"REWORK","_number":1,"created":"2019-09-20 10:45:55.000000000","uploader":{"_account_id":24072,"name":"Marcin Juszkiewicz","email":"mjuszkiewicz@redhat.com","username":"hrw"},"ref":"refs/changes/59/683359/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/kolla","ref":"refs/changes/59/683359/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/kolla refs/changes/59/683359/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/kolla refs/changes/59/683359/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/kolla refs/changes/59/683359/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/kolla refs/changes/59/683359/1"}}},"commit":{"parents":[{"commit":"187d6c52f1bbb1176d7129a2b24b0cf48ba29d79","subject":"base: fix OracleLinux build","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/187d6c52f1bbb1176d7129a2b24b0cf48ba29d79"}]}],"author":{"name":"Marcin Juszkiewicz","email":"marcin.juszkiewicz@linaro.org","date":"2019-09-19 11:50:41.000000000","tz":120},"committer":{"name":"Marcin Juszkiewicz","email":"marcin.juszkiewicz@linaro.org","date":"2019-09-20 10:45:55.000000000","tz":0},"subject":"sensu-client: install older whois gem","message":"sensu-client: install older whois gem\n\nWhois 5.0.0 requires Ruby 2.4 while CentOS has only 2.0 version.\n\nChange-Id: I6ec9d58bf54db1ac69b879785f38fb492bc2dfc2\n(cherry picked from commit 41408527ae88ac42de11ca2818ff405320795a8b)\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/44be1294c1b801f01c5f989355c4486afd6b7ef3"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/44be1294c1b801f01c5f989355c4486afd6b7ef3"}]},"branch":"refs/heads/stable/stein"}},"requirements":[],"submit_records":[],"submit_requirements":[]}
