)]}'
{"id":"openstack%2Fkolla~683360","triplet_id":"openstack%2Fkolla~stable%2Frocky~I6ec9d58bf54db1ac69b879785f38fb492bc2dfc2","project":"openstack/kolla","branch":"stable/rocky","topic":"683117-stable/stein-stable/rocky","hashtags":[],"change_id":"I6ec9d58bf54db1ac69b879785f38fb492bc2dfc2","subject":"sensu-client: install older whois gem","status":"MERGED","created":"2019-09-20 10:46:21.000000000","updated":"2019-09-20 16:05:53.000000000","submitted":"2019-09-20 16:05:53.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"683360-1568995553503-657de88e","meta_rev_id":"69946b552f59ff98053b08423e0a4603945b079e","_number":683360,"virtual_id_number":683360,"owner":{"_account_id":24072,"name":"Marcin Juszkiewicz","email":"mjuszkiewicz@redhat.com","username":"hrw"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},{"value":0,"_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"},{"value":2,"date":"2019-09-20 16:05:53.000000000","post_submit":true,"permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},"all":[{"value":2,"date":"2019-09-20 13:41:02.000000000","_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},{"value":2,"date":"2019-09-20 14:53:40.000000000","_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"},"all":[{"value":0,"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},{"value":1,"date":"2019-09-20 14:53:40.000000000","_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Backport-Candidate":{"all":[{"value":0,"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},{"value":0,"_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Do Not Backport"," 0":"Backport Review Needed","+1":"Should Backport"},"description":"","default_value":0,"optional":true},"Review-Priority":{"approved":{"_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"},"all":[{"value":0,"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},{"value":2,"date":"2019-09-20 12:41:24.000000000","_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Branch Freeze"," 0":"No Priority","+1":"Important Change","+2":"Gate Blocker Fix / Urgent Change"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2019-09-20 13:41:02.000000000","updated_by":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},"reviewer":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},"state":"REVIEWER"},{"updated":"2019-09-20 14:53:40.000000000","updated_by":{"_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"},"reviewer":{"_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"},"state":"REVIEWER"},{"updated":"2019-09-20 16:05:53.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"}],"messages":[{"id":"303316c26365ca5d6e371de464a60a55c0f09a94","author":{"_account_id":24072,"name":"Marcin Juszkiewicz","email":"mjuszkiewicz@redhat.com","username":"hrw"},"date":"2019-09-20 10:46:21.000000000","message":"Patch Set 1: Cherry Picked from branch stable/stein.","accounts_in_message":[],"_revision_number":1},{"id":"980d75fa33f87949b49757389d7285bdbf318cc8","author":{"_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"},"date":"2019-09-20 11:30:39.000000000","message":"Patch Set 1: Code-Review+2","accounts_in_message":[],"_revision_number":1},{"id":"c07d154bfe96c188fc8983e8226a3e45dc9287af","author":{"_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"},"date":"2019-09-20 11:31:00.000000000","message":"Patch Set 1: Review-Priority+2","accounts_in_message":[],"_revision_number":1},{"id":"9a37e37ae56a13c5a14d3e8688b7c0d01fed6fa7","author":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},"date":"2019-09-20 11:35:16.000000000","message":"Patch Set 1: Code-Review+2","accounts_in_message":[],"_revision_number":1},{"id":"025cf38fead2bfa609ebad6e0aa33f0fe28d98fd","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-09-20 12:37:01.000000000","message":"Patch Set 1: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttp://docs.openstack.org/infra/manual/developers.html#automated-testing\n\n\n- kolla-build-centos-binary https://zuul.opendev.org/t/openstack/build/b340c32b4f68457f9b819c43bb96afe2 : FAILURE in 40m 09s (non-voting)\n- kolla-build-centos-source https://zuul.opendev.org/t/openstack/build/ed6a6c1236a14e4a8dadfdba698f544b : FAILURE in 1h 11m 15s\n- kolla-ansible-centos-source https://zuul.opendev.org/t/openstack/build/3eeec283bbbe420ab106726b1ab2021a : SUCCESS in 1h 27m 09s\n- kolla-ansible-centos-binary https://zuul.opendev.org/t/openstack/build/7eaf461d0222416cb973cd1f6cc352e3 : SUCCESS in 1h 27m 51s (non-voting)\n- kolla-build-oraclelinux-binary https://zuul.opendev.org/t/openstack/build/93f374d5e7184f0aa5162d8cc0abe0ec : FAILURE in 52m 09s (non-voting)\n- kolla-build-oraclelinux-source https://zuul.opendev.org/t/openstack/build/ac147fd8a7ea4df797fd41a3d24afa3f : FAILURE in 50m 32s\n- kolla-ansible-oraclelinux-source https://zuul.opendev.org/t/openstack/build/c59a221d4cdf41499d1503612d0e7758 : SUCCESS in 1h 21m 28s (non-voting)\n- kolla-ansible-oraclelinux-binary https://zuul.opendev.org/t/openstack/build/1771cb4314c84a709738fb84e463533c : FAILURE in 40m 46s (non-voting)\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/bdc0acee6bea4adcac02d555515bbdf8 : SUCCESS in 5m 45s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/3ae09fb3ea3442da94c932ab03a9b419 : SUCCESS in 4m 21s\n- openstack-tox-py27 https://zuul.opendev.org/t/openstack/build/05189a1cd62745ebb1f4432bd808f218 : SUCCESS in 3m 47s\n- openstack-tox-py35 https://zuul.opendev.org/t/openstack/build/14a3b35b40b749faab36392f8b66f6e5 : SUCCESS in 4m 19s\n- build-openstack-sphinx-docs https://zuul.opendev.org/t/openstack/build/22b45754d56944458d10ea0d93f5edcb : SUCCESS in 6m 03s\n- kolla-build-ubuntu-binary https://zuul.opendev.org/t/openstack/build/9b62a40fd0c34e7dbdebb1421e9acec5 : SUCCESS in 58m 40s (non-voting)\n- kolla-build-ubuntu-source https://zuul.opendev.org/t/openstack/build/2079ffecfbe34392a66c3a577050d6e3 : SUCCESS in 37m 34s\n- kolla-ansible-ubuntu-source https://zuul.opendev.org/t/openstack/build/4f5a6f906f584187900d84c731169e84 : SUCCESS in 58m 51s\n- kolla-ansible-ubuntu-binary https://zuul.opendev.org/t/openstack/build/9d0a5388db4146fb95101391bbdd6196 : SUCCESS in 1h 30m 43s (non-voting)","accounts_in_message":[],"_revision_number":1},{"id":"258bb8fc2cab9583c79d37e866b3f766524a93d4","author":{"_account_id":24072,"name":"Marcin Juszkiewicz","email":"mjuszkiewicz@redhat.com","username":"hrw"},"date":"2019-09-20 12:41:24.000000000","message":"Patch Set 2: Published edit on patch set 1.","accounts_in_message":[],"_revision_number":2},{"id":"bf270d1bc79f0a8b64792cbf89fd5938c0c2cd00","author":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},"date":"2019-09-20 13:41:02.000000000","message":"Patch Set 2: Code-Review+2","accounts_in_message":[],"_revision_number":2},{"id":"8bf0d1221f92c8d724910f6fae5ae6a6b2ad7963","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-09-20 14:23:31.000000000","message":"Patch Set 2: Verified+1\n\nBuild succeeded (check pipeline).\n\n- kolla-build-centos-binary https://zuul.opendev.org/t/openstack/build/8ea5cf56f5c94ad99e57d942679f2fb4 : SUCCESS in 36m 25s (non-voting)\n- kolla-build-centos-source https://zuul.opendev.org/t/openstack/build/4227bb9994884fbf9c5d2feea9c58ecc : SUCCESS in 56m 52s\n- kolla-ansible-centos-source https://zuul.opendev.org/t/openstack/build/8b98faa7a08b4b06bf9d717837ab3fac : SUCCESS in 1h 15m 01s\n- kolla-ansible-centos-binary https://zuul.opendev.org/t/openstack/build/a96ad5574edd413cb39c6a296f9a6ae6 : SUCCESS in 1h 20m 41s (non-voting)\n- kolla-build-oraclelinux-binary https://zuul.opendev.org/t/openstack/build/6c3c0d877cdb45c8a6c986865018510b : SUCCESS in 39m 09s (non-voting)\n- kolla-build-oraclelinux-source https://zuul.opendev.org/t/openstack/build/7447e5748ba145c7be5528fd1b8f38bf : SUCCESS in 48m 33s\n- kolla-ansible-oraclelinux-source https://zuul.opendev.org/t/openstack/build/8b6cba62638d452db80ffbe22b58360a : SUCCESS in 1h 21m 07s (non-voting)\n- kolla-ansible-oraclelinux-binary https://zuul.opendev.org/t/openstack/build/dd07de7d13ed42f99505ba4869ef9c59 : SUCCESS in 1h 17m 29s (non-voting)\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/13e7fabd226444ff974dd498a0adcc0b : SUCCESS in 3m 23s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/3416dfc9a5db4e65a5678353ee08168f : SUCCESS in 3m 48s\n- openstack-tox-py27 https://zuul.opendev.org/t/openstack/build/a3b605e67db44e4685471433ec0c1ee9 : SUCCESS in 3m 21s\n- openstack-tox-py35 https://zuul.opendev.org/t/openstack/build/3beabea2024343228e1dcc5a82b1bfd8 : SUCCESS in 3m 01s\n- build-openstack-sphinx-docs https://zuul.opendev.org/t/openstack/build/ffd7d9648c0248bc8ff007e85e5fbe1f : SUCCESS in 5m 39s\n- kolla-build-ubuntu-binary https://zuul.opendev.org/t/openstack/build/1021f4aa34d34818b971dd24b2920d6d : SUCCESS in 1h 32m 47s (non-voting)\n- kolla-build-ubuntu-source https://zuul.opendev.org/t/openstack/build/219c513e9f72404a818b101bd7aa4ec7 : SUCCESS in 42m 15s\n- kolla-ansible-ubuntu-source https://zuul.opendev.org/t/openstack/build/d7329fb05aac4bb9af71c1856739fa49 : SUCCESS in 1h 12m 10s\n- kolla-ansible-ubuntu-binary https://zuul.opendev.org/t/openstack/build/04db4a3655a54cdf994f492633888399 : SUCCESS in 1h 33m 26s (non-voting)","accounts_in_message":[],"_revision_number":2},{"id":"d35caed53cc8870f410e91731c9691edfd7ff277","author":{"_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"},"date":"2019-09-20 14:53:40.000000000","message":"Patch Set 2: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":2},{"id":"46cb7be36b125efcfc586d318e917e67f56cf213","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-09-20 14:53:55.000000000","message":"Patch Set 2: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":2},{"id":"0a49e2e9c4515ecd222132b621c971619f10a712","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-09-20 16:05:53.000000000","message":"Change has been successfully merged by Zuul","accounts_in_message":[],"_revision_number":2},{"id":"a7a67541a1f9f16b6f05e9eaef5707b5e9eab39c","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-09-20 16:05:53.000000000","message":"Patch Set 2: Verified+2\n\nBuild succeeded (gate pipeline).\n\n- kolla-build-centos-source https://zuul.opendev.org/t/openstack/build/812ff55fabd349848af8b9cb0e672b5d : SUCCESS in 55m 10s\n- kolla-build-oraclelinux-source https://zuul.opendev.org/t/openstack/build/05937f5429f340a5aa614406714fc384 : SUCCESS in 53m 20s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/593581e153af404591cdebff25b071d0 : SUCCESS in 6m 02s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/092f3e59ed3a40058f17f271bbf1dd44 : SUCCESS in 3m 45s\n- openstack-tox-py27 https://zuul.opendev.org/t/openstack/build/a60e6537417549b2aecd7b83e598d691 : SUCCESS in 2m 59s\n- openstack-tox-py35 https://zuul.opendev.org/t/openstack/build/80facd77cd8247589b12de24f36b871b : SUCCESS in 4m 06s\n- build-openstack-sphinx-docs https://zuul.opendev.org/t/openstack/build/d05707a29abf4414965ba547c47a2cd6 : SUCCESS in 4m 01s\n- kolla-build-ubuntu-source https://zuul.opendev.org/t/openstack/build/b6a17e2ea84948eeb5700b5e262d5370 : SUCCESS in 37m 05s","accounts_in_message":[],"_revision_number":2}],"current_revision_number":2,"current_revision":"091d9a96ef0462017a1ea20cdf5c2b43271e226a","revisions":{"cb47fccd13a9fe6b275dbe92cf3e3e7263671fff":{"kind":"REWORK","_number":1,"created":"2019-09-20 10:46:21.000000000","uploader":{"_account_id":24072,"name":"Marcin Juszkiewicz","email":"mjuszkiewicz@redhat.com","username":"hrw"},"ref":"refs/changes/60/683360/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/kolla","ref":"refs/changes/60/683360/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/kolla refs/changes/60/683360/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/kolla refs/changes/60/683360/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/kolla refs/changes/60/683360/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/kolla refs/changes/60/683360/1"}}},"commit":{"parents":[{"commit":"5693bfd698a9f00f5fefd0ab0c4c3c38628223bb","subject":"base: fix OracleLinux build","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/5693bfd698a9f00f5fefd0ab0c4c3c38628223bb"}]}],"author":{"name":"Marcin Juszkiewicz","email":"marcin.juszkiewicz@linaro.org","date":"2019-09-19 11:50:41.000000000","tz":120},"committer":{"name":"Marcin Juszkiewicz","email":"marcin.juszkiewicz@linaro.org","date":"2019-09-20 10:46:21.000000000","tz":0},"subject":"sensu-client: install older whois gem","message":"sensu-client: install older whois gem\n\nWhois 5.0.0 requires Ruby 2.4 while CentOS has only 2.0 version.\n\nChange-Id: I6ec9d58bf54db1ac69b879785f38fb492bc2dfc2\n(cherry picked from commit 41408527ae88ac42de11ca2818ff405320795a8b)\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/cb47fccd13a9fe6b275dbe92cf3e3e7263671fff"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/cb47fccd13a9fe6b275dbe92cf3e3e7263671fff"}]},"branch":"refs/heads/stable/rocky"},"091d9a96ef0462017a1ea20cdf5c2b43271e226a":{"kind":"REWORK","_number":2,"created":"2019-09-20 12:41:24.000000000","uploader":{"_account_id":24072,"name":"Marcin Juszkiewicz","email":"mjuszkiewicz@redhat.com","username":"hrw"},"ref":"refs/changes/60/683360/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/kolla","ref":"refs/changes/60/683360/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/kolla refs/changes/60/683360/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/kolla refs/changes/60/683360/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/kolla refs/changes/60/683360/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/kolla refs/changes/60/683360/2"}}},"commit":{"parents":[{"commit":"5693bfd698a9f00f5fefd0ab0c4c3c38628223bb","subject":"base: fix OracleLinux build","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/5693bfd698a9f00f5fefd0ab0c4c3c38628223bb"}]}],"author":{"name":"Marcin Juszkiewicz","email":"marcin.juszkiewicz@linaro.org","date":"2019-09-19 11:50:41.000000000","tz":120},"committer":{"name":"Marcin Juszkiewicz","email":"marcin.juszkiewicz@linaro.org","date":"2019-09-20 12:41:20.000000000","tz":0},"subject":"sensu-client: install older whois gem","message":"sensu-client: install older whois gem\n\nWhois 5.0.0 requires Ruby 2.4 while CentOS has only 2.0 version.\n\nChange-Id: I6ec9d58bf54db1ac69b879785f38fb492bc2dfc2\n(cherry picked from commit 41408527ae88ac42de11ca2818ff405320795a8b)\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/091d9a96ef0462017a1ea20cdf5c2b43271e226a"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/091d9a96ef0462017a1ea20cdf5c2b43271e226a"}]},"branch":"refs/heads/stable/rocky"}},"requirements":[],"submit_records":[],"submit_requirements":[]}
