)]}'
{"id":"openstack%2Fkolla~683361","triplet_id":"openstack%2Fkolla~stable%2Fqueens~I6ec9d58bf54db1ac69b879785f38fb492bc2dfc2","project":"openstack/kolla","branch":"stable/queens","topic":"683117-stable/queens","hashtags":[],"change_id":"I6ec9d58bf54db1ac69b879785f38fb492bc2dfc2","subject":"sensu-client: install older whois gem","status":"MERGED","created":"2019-09-20 10:49:39.000000000","updated":"2019-09-23 14:14:57.000000000","submitted":"2019-09-23 14:14:57.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"683361-1569248097725-c425b5ec","meta_rev_id":"850bf9a09d3f226915e0e3d251fe35575cdf5989","_number":683361,"virtual_id_number":683361,"owner":{"_account_id":24072,"name":"Marcin Juszkiewicz","email":"mjuszkiewicz@redhat.com","username":"hrw"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":19316,"name":"Eduardo Gonzalez","email":"dabarren@gmail.com","username":"egonzalez90"},{"value":0,"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},{"value":0,"_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2019-09-23 14:14:57.000000000","post_submit":true,"permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":19316,"name":"Eduardo Gonzalez","email":"dabarren@gmail.com","username":"egonzalez90"},"all":[{"value":2,"date":"2019-09-23 11:36:32.000000000","_account_id":19316,"name":"Eduardo Gonzalez","email":"dabarren@gmail.com","username":"egonzalez90"},{"value":0,"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},{"value":2,"date":"2019-09-23 10:28:37.000000000","_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},"all":[{"value":0,"_account_id":19316,"name":"Eduardo Gonzalez","email":"dabarren@gmail.com","username":"egonzalez90"},{"value":1,"date":"2019-09-23 12:39:20.000000000","_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},{"value":0,"_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Backport-Candidate":{"all":[{"value":0,"_account_id":19316,"name":"Eduardo Gonzalez","email":"dabarren@gmail.com","username":"egonzalez90"},{"value":0,"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},{"value":0,"_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Do Not Backport"," 0":"Backport Review Needed","+1":"Should Backport"},"description":"","default_value":0,"optional":true},"Review-Priority":{"approved":{"_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"},"all":[{"value":0,"_account_id":19316,"name":"Eduardo Gonzalez","email":"dabarren@gmail.com","username":"egonzalez90"},{"value":0,"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},{"value":2,"date":"2019-09-23 10:27:16.000000000","_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Branch Freeze"," 0":"No Priority","+1":"Important Change","+2":"Gate Blocker Fix / Urgent Change"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},{"_account_id":19316,"name":"Eduardo Gonzalez","email":"dabarren@gmail.com","username":"egonzalez90"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2019-09-20 11:31:07.000000000","updated_by":{"_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"},"reviewer":{"_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"},"state":"REVIEWER"},{"updated":"2019-09-23 11:36:32.000000000","updated_by":{"_account_id":19316,"name":"Eduardo Gonzalez","email":"dabarren@gmail.com","username":"egonzalez90"},"reviewer":{"_account_id":19316,"name":"Eduardo Gonzalez","email":"dabarren@gmail.com","username":"egonzalez90"},"state":"REVIEWER"},{"updated":"2019-09-23 12:39:20.000000000","updated_by":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},"reviewer":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},"state":"REVIEWER"},{"updated":"2019-09-23 14:14:57.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"}],"messages":[{"id":"bf7e0297a995f00146496934c3be0e4d6a0618d0","author":{"_account_id":24072,"name":"Marcin Juszkiewicz","email":"mjuszkiewicz@redhat.com","username":"hrw"},"date":"2019-09-20 10:49:39.000000000","message":"Patch Set 1: Cherry Picked from branch master.","accounts_in_message":[],"_revision_number":1},{"id":"43169c3fee8522da817a188eeca30e365446a9a6","author":{"_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"},"date":"2019-09-20 11:30:30.000000000","message":"Patch Set 1: Code-Review+2","accounts_in_message":[],"_revision_number":1},{"id":"5d62469a7b3c1520053cb24d9364295d35ab4e4f","author":{"_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"},"date":"2019-09-20 11:31:07.000000000","message":"Patch Set 1: Review-Priority+2","accounts_in_message":[],"_revision_number":1},{"id":"37fbd5ae3b8208e7b37fbf0b8451770f67979baa","author":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},"date":"2019-09-20 11:35:03.000000000","message":"Patch Set 1: Code-Review+2","accounts_in_message":[],"_revision_number":1},{"id":"7f252918fcaea8ebc91a5eeacebf07faae7c3a50","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-09-20 12:27:19.000000000","message":"Patch Set 1: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttp://docs.openstack.org/infra/manual/developers.html#automated-testing\n\n\n- kolla-build-centos-binary https://zuul.opendev.org/t/openstack/build/bef631424e7445b081507356245ca397 : FAILURE in 36m 45s (non-voting)\n- kolla-build-centos-source https://zuul.opendev.org/t/openstack/build/58ae9dd49ccf4f3f8fa2d2de899638ba : FAILURE in 41m 05s\n- kolla-ansible-centos-source https://zuul.opendev.org/t/openstack/build/51570e7687f446ef889cf190feead2e5 : SUCCESS in 1h 24m 14s (non-voting)\n- kolla-ansible-centos-binary https://zuul.opendev.org/t/openstack/build/a8f23363f2164f5b93f8c3b289ea257e : SUCCESS in 1h 11m 02s (non-voting)\n- kolla-build-oraclelinux-binary https://zuul.opendev.org/t/openstack/build/6bc7671b73b045d8bd9693b8ebdbc2a9 : FAILURE in 46m 11s (non-voting)\n- kolla-build-oraclelinux-source https://zuul.opendev.org/t/openstack/build/a197b7b2e95f4221acd3c8732e175390 : FAILURE in 1h 00m 31s\n- kolla-ansible-oraclelinux-source https://zuul.opendev.org/t/openstack/build/7ea499e04e644df9815e3e1ed7c5ea49 : SUCCESS in 1h 23m 12s (non-voting)\n- kolla-ansible-oraclelinux-binary https://zuul.opendev.org/t/openstack/build/de079593ef3e495da5f482375ed180b2 : SUCCESS in 1h 24m 29s (non-voting)\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/f35d2a8b80d3443ca17634d47a2c85ee : SUCCESS in 5m 21s\n- openstack-tox-py27 https://zuul.opendev.org/t/openstack/build/bc4fe6ab5c0d4dde98ab4e848bfb8cdf : SUCCESS in 5m 02s\n- openstack-tox-py35 https://zuul.opendev.org/t/openstack/build/da1494da24b94b3cb7f301f37c9e68b5 : SUCCESS in 4m 44s\n- build-openstack-sphinx-docs https://zuul.opendev.org/t/openstack/build/c55efe720b3b4c8db0f4ac6f61004973 : SUCCESS in 5m 22s\n- kolla-build-ubuntu-binary https://zuul.opendev.org/t/openstack/build/80993435ba5a43819c324ca50dad1d88 : SUCCESS in 32m 05s (non-voting)\n- kolla-build-ubuntu-source https://zuul.opendev.org/t/openstack/build/49006e5fcac94cce8b21705720ec4380 : SUCCESS in 34m 46s\n- kolla-ansible-ubuntu-source https://zuul.opendev.org/t/openstack/build/84f8b60dd55d4b5b984d551fe1c4f223 : SUCCESS in 1h 00m 38s (non-voting)\n- kolla-ansible-ubuntu-binary https://zuul.opendev.org/t/openstack/build/654afefad8dd4fdd9b7a36363b287146 : SUCCESS in 1h 02m 35s (non-voting)","accounts_in_message":[],"_revision_number":1},{"id":"c6318859a5763f5b72c42763fcd915c59a8dadb8","author":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},"date":"2019-09-20 12:29:16.000000000","message":"Patch Set 1: -Code-Review","accounts_in_message":[],"_revision_number":1},{"id":"d312b2c701225daf490f146e5d3fe7c9dc39602e","author":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},"date":"2019-09-23 10:27:16.000000000","message":"Uploaded patch set 2.","accounts_in_message":[],"_revision_number":2},{"id":"56da5009b0d4359569b8afc293d16f61d086de08","author":{"_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"},"date":"2019-09-23 10:28:37.000000000","message":"Patch Set 2: Code-Review+2","accounts_in_message":[],"_revision_number":2},{"id":"8270eed357a662fd550d0dc600dc25173731a49b","author":{"_account_id":19316,"name":"Eduardo Gonzalez","email":"dabarren@gmail.com","username":"egonzalez90"},"date":"2019-09-23 11:36:32.000000000","message":"Patch Set 2: Code-Review+2","accounts_in_message":[],"_revision_number":2},{"id":"a70a115b3272b86a8510573f2bde08a20e232a8d","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-09-23 12:15:07.000000000","message":"Patch Set 2: Verified+1\n\nBuild succeeded (check pipeline).\n\n- kolla-build-centos-binary https://zuul.opendev.org/t/openstack/build/5b8ed84a02424ef890113658a8f43029 : SUCCESS in 39m 56s (non-voting)\n- kolla-build-centos-source https://zuul.opendev.org/t/openstack/build/99e9e0bdc3f248e0b0f48ab8a304d9b4 : SUCCESS in 48m 47s\n- kolla-ansible-centos-source https://zuul.opendev.org/t/openstack/build/8ceca543fc8741429e91a8d0f3863ddf : SUCCESS in 1h 28m 10s (non-voting)\n- kolla-ansible-centos-binary https://zuul.opendev.org/t/openstack/build/96d672bbca654aae9f549a2171a10947 : SUCCESS in 1h 15m 15s (non-voting)\n- kolla-build-oraclelinux-binary https://zuul.opendev.org/t/openstack/build/6bd7cb8cfb2b4c01a5eecea299c194e8 : SUCCESS in 49m 24s (non-voting)\n- kolla-build-oraclelinux-source https://zuul.opendev.org/t/openstack/build/0f2a94f1ffe1464aac4c99a1ac691877 : SUCCESS in 45m 17s\n- kolla-ansible-oraclelinux-source https://zuul.opendev.org/t/openstack/build/bb8fa7dbe082444aadef22a68a3eede8 : SUCCESS in 1h 23m 31s (non-voting)\n- kolla-ansible-oraclelinux-binary https://zuul.opendev.org/t/openstack/build/6942fd826ba64dddb98cec4138e86847 : SUCCESS in 1h 41m 50s (non-voting)\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/273c3c0ca1954f608129ad5dbe9d9c14 : SUCCESS in 5m 36s\n- openstack-tox-py27 https://zuul.opendev.org/t/openstack/build/e44b1a39e7dc4e2396851dff9b08c67e : SUCCESS in 3m 20s\n- openstack-tox-py35 https://zuul.opendev.org/t/openstack/build/bcd1cbf32ad048a4b20aed8f192da5d2 : SUCCESS in 3m 50s\n- build-openstack-sphinx-docs https://zuul.opendev.org/t/openstack/build/09212b09688e42208b85bc45ef082e71 : SUCCESS in 4m 15s\n- kolla-build-ubuntu-binary https://zuul.opendev.org/t/openstack/build/d59701e446954e01b09a0c51899ac2aa : SUCCESS in 31m 31s (non-voting)\n- kolla-build-ubuntu-source https://zuul.opendev.org/t/openstack/build/d2e6fb72fa73488b8bd4108748abd31d : SUCCESS in 36m 49s\n- kolla-ansible-ubuntu-source https://zuul.opendev.org/t/openstack/build/d76c05e8a8aa4c49bedf9be3148afbfb : SUCCESS in 43m 42s (non-voting)\n- kolla-ansible-ubuntu-binary https://zuul.opendev.org/t/openstack/build/7fb59e2019dd40a7a4615cb9f2ccdf46 : SUCCESS in 1h 05m 31s (non-voting)","accounts_in_message":[],"_revision_number":2},{"id":"ddde936bbd94a564b29d8bc1e9600e8a37791322","author":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},"date":"2019-09-23 12:39:20.000000000","message":"Patch Set 2: Workflow+1","accounts_in_message":[],"_revision_number":2},{"id":"e2e161075af852a6de5a4753d61465c5f651b481","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-09-23 12:39:34.000000000","message":"Patch Set 2: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":2},{"id":"d2cc21502799dd32e1df49611610bf011f43af8d","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-09-23 13:10:45.000000000","message":"Patch Set 2:\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":2},{"id":"2ecfc19783063c5245d29eea3821f5c560810e37","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-09-23 14:14:57.000000000","message":"Change has been successfully merged by Zuul","accounts_in_message":[],"_revision_number":2},{"id":"30ce0fa82aa9cfc0d0f27787147423583d7f2c12","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-09-23 14:14:57.000000000","message":"Patch Set 2: Verified+2\n\nBuild succeeded (gate pipeline).\n\n- kolla-build-centos-source https://zuul.opendev.org/t/openstack/build/29202779cd6f4eac8b3bcfe15e725d14 : SUCCESS in 41m 51s\n- kolla-build-oraclelinux-source https://zuul.opendev.org/t/openstack/build/c8905e46979544a486eec231003eb431 : SUCCESS in 46m 48s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/e2d8e99c594742ad95e39bd7b2308a9c : SUCCESS in 5m 47s\n- openstack-tox-py27 https://zuul.opendev.org/t/openstack/build/41788dfc5f1247e28dc1b22c053f65c7 : SUCCESS in 7m 07s\n- openstack-tox-py35 https://zuul.opendev.org/t/openstack/build/8f57174bb2b044ec9359ea60089096dc : SUCCESS in 6m 13s\n- build-openstack-sphinx-docs https://zuul.opendev.org/t/openstack/build/c9062182a3c547c7b1e8ab9926f836f0 : SUCCESS in 7m 36s\n- kolla-build-ubuntu-source https://zuul.opendev.org/t/openstack/build/79e7d0d4c29f40d498f240f6f9a8c62d : SUCCESS in 32m 35s","accounts_in_message":[],"_revision_number":2}],"current_revision_number":2,"current_revision":"29d12aebd679a3079fec6e46c6d0d1654796bee8","revisions":{"dea7c9e08aeb5dbeac81178725ef2cbe0ef74971":{"kind":"REWORK","_number":1,"created":"2019-09-20 10:49:39.000000000","uploader":{"_account_id":24072,"name":"Marcin Juszkiewicz","email":"mjuszkiewicz@redhat.com","username":"hrw"},"ref":"refs/changes/61/683361/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/kolla","ref":"refs/changes/61/683361/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/kolla refs/changes/61/683361/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/kolla refs/changes/61/683361/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/kolla refs/changes/61/683361/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/kolla refs/changes/61/683361/1"}}},"commit":{"parents":[{"commit":"ea34f0e19f17fefd3897deb9332a7a4596de5f6b","subject":"base: fix OracleLinux build","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/ea34f0e19f17fefd3897deb9332a7a4596de5f6b"}]}],"author":{"name":"Marcin Juszkiewicz","email":"marcin.juszkiewicz@linaro.org","date":"2019-09-19 11:50:41.000000000","tz":120},"committer":{"name":"Marcin Juszkiewicz","email":"marcin.juszkiewicz@linaro.org","date":"2019-09-20 10:49:39.000000000","tz":0},"subject":"sensu-client: install older whois gem","message":"sensu-client: install older whois gem\n\nWhois 5.0.0 requires Ruby 2.4 while CentOS has only 2.0 version.\n\nChange-Id: I6ec9d58bf54db1ac69b879785f38fb492bc2dfc2\n(cherry picked from commit 41408527ae88ac42de11ca2818ff405320795a8b)\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/dea7c9e08aeb5dbeac81178725ef2cbe0ef74971"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/dea7c9e08aeb5dbeac81178725ef2cbe0ef74971"}]},"branch":"refs/heads/stable/queens"},"29d12aebd679a3079fec6e46c6d0d1654796bee8":{"kind":"REWORK","_number":2,"created":"2019-09-23 10:27:16.000000000","uploader":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},"ref":"refs/changes/61/683361/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/kolla","ref":"refs/changes/61/683361/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/kolla refs/changes/61/683361/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/kolla refs/changes/61/683361/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/kolla refs/changes/61/683361/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/kolla refs/changes/61/683361/2"}}},"commit":{"parents":[{"commit":"ea34f0e19f17fefd3897deb9332a7a4596de5f6b","subject":"base: fix OracleLinux build","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/ea34f0e19f17fefd3897deb9332a7a4596de5f6b"}]}],"author":{"name":"Marcin Juszkiewicz","email":"marcin.juszkiewicz@linaro.org","date":"2019-09-19 11:50:41.000000000","tz":120},"committer":{"name":"Mark Goddard","email":"mark@stackhpc.com","date":"2019-09-23 10:27:16.000000000","tz":0},"subject":"sensu-client: install older whois gem","message":"sensu-client: install older whois gem\n\nWhois 5.0.0 requires Ruby 2.4 while CentOS has only 2.0 version.\n\nChange-Id: I6ec9d58bf54db1ac69b879785f38fb492bc2dfc2\n(cherry picked from commit 41408527ae88ac42de11ca2818ff405320795a8b)\n(cherry picked from commit 091d9a96ef0462017a1ea20cdf5c2b43271e226a)\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/29d12aebd679a3079fec6e46c6d0d1654796bee8"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/29d12aebd679a3079fec6e46c6d0d1654796bee8"}]},"branch":"refs/heads/stable/queens"}},"requirements":[],"submit_records":[],"submit_requirements":[]}
