)]}'
{"id":"openstack%2Fkolla~683378","triplet_id":"openstack%2Fkolla~stable%2Fstein~Iec78660049085483b207f4bdcf48f572d845c600","project":"openstack/kolla","branch":"stable/stein","hashtags":[],"change_id":"Iec78660049085483b207f4bdcf48f572d845c600","subject":"[horizon] Install octavia dashboard on deb","status":"MERGED","created":"2019-09-20 13:03:05.000000000","updated":"2019-09-24 08:47:49.000000000","submitted":"2019-09-24 08:45:07.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"683378-1569314708088-efb49f7a","meta_rev_id":"115f93080321d54371ae9c3f1b49dafc339cd0de","_number":683378,"virtual_id_number":683378,"owner":{"_account_id":30523,"name":"Dincer Celik","email":"hello@dincercelik.com","username":"osmanlicilegi"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},{"value":0,"_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"},{"value":0,"date":"2019-09-20 15:42:16.000000000","_account_id":30523,"name":"Dincer Celik","email":"hello@dincercelik.com","username":"osmanlicilegi"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2019-09-24 08:45:07.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"all":[{"value":2,"date":"2019-09-24 07:43:15.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},{"value":2,"date":"2019-09-20 14:56:37.000000000","_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"},{"value":0,"_account_id":30523,"name":"Dincer Celik","email":"hello@dincercelik.com","username":"osmanlicilegi"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"all":[{"value":1,"date":"2019-09-24 07:43:15.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},{"value":0,"_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"},{"value":0,"_account_id":30523,"name":"Dincer Celik","email":"hello@dincercelik.com","username":"osmanlicilegi"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Backport-Candidate":{"all":[{"value":0,"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},{"value":0,"_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"},{"value":0,"_account_id":30523,"name":"Dincer Celik","email":"hello@dincercelik.com","username":"osmanlicilegi"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Do Not Backport"," 0":"Backport Review Needed","+1":"Should Backport"},"description":"","default_value":0,"optional":true},"Review-Priority":{"all":[{"value":0,"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},{"value":0,"_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"},{"value":0,"_account_id":30523,"name":"Dincer Celik","email":"hello@dincercelik.com","username":"osmanlicilegi"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Branch Freeze"," 0":"No Priority","+1":"Important Change","+2":"Gate Blocker Fix / Urgent Change"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},{"_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"},{"_account_id":30523,"name":"Dincer Celik","email":"hello@dincercelik.com","username":"osmanlicilegi"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2019-09-20 14:56:37.000000000","updated_by":{"_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"},"reviewer":{"_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"},"state":"REVIEWER"},{"updated":"2019-09-24 07:43:15.000000000","updated_by":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"reviewer":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"state":"REVIEWER"},{"updated":"2019-09-24 08:45:07.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"}],"messages":[{"id":"bc56d4f6e5c191167f270fbc32bff0beb37c3e2d","author":{"_account_id":30523,"name":"Dincer Celik","email":"hello@dincercelik.com","username":"osmanlicilegi"},"date":"2019-09-20 13:03:05.000000000","message":"Patch Set 1: Cherry Picked from branch master.","accounts_in_message":[],"_revision_number":1},{"id":"5c4098e1e050792c60490952c82764812427542c","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-09-20 14:52:23.000000000","message":"Patch Set 1: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttp://docs.openstack.org/infra/manual/developers.html#automated-testing\n\n\n- kolla-build-centos-binary https://zuul.opendev.org/t/openstack/build/1027a39ae18241f5bef767bbb20b3c50 : FAILURE in 46m 49s (non-voting)\n- kolla-build-centos-source https://zuul.opendev.org/t/openstack/build/771e45547504478a832dc378603f10b9 : FAILURE in 57m 01s\n- kolla-ansible-centos-source https://zuul.opendev.org/t/openstack/build/15171240abbe4dd4aa4739967a6a7afa : SUCCESS in 1h 09m 40s\n- kolla-ansible-centos-binary https://zuul.opendev.org/t/openstack/build/3dabd31bb37a420e911f2a0bb42ad197 : FAILURE in 1h 03m 45s (non-voting)\n- kolla-ansible-centos-source-upgrade https://zuul.opendev.org/t/openstack/build/ed04f24d234f4763a1c97e3996b9d750 : SUCCESS in 1h 38m 26s\n- tripleo-build-containers-centos-7 https://zuul.opendev.org/t/openstack/build/5957b62364a24310a9f5c3be5091d75e : SUCCESS in 47m 26s (non-voting)\n- kolla-build-oraclelinux-binary https://zuul.opendev.org/t/openstack/build/95e6321f58be437bb9422811e352db62 : FAILURE in 53m 25s (non-voting)\n- kolla-build-oraclelinux-source https://zuul.opendev.org/t/openstack/build/cdfce1789558422da2d1801ee4acdecf : FAILURE in 1h 00m 59s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/90e3edd82e3f4775b38a0bc8ad18b57b : SUCCESS in 3m 46s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/52c2aa406647411098110eedf59011c7 : SUCCESS in 5m 28s\n- openstack-tox-py27 https://zuul.opendev.org/t/openstack/build/f7720e70aae647199568d5bdc452889c : SUCCESS in 3m 16s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/717d3a66945e414b904df0aa3800aced : SUCCESS in 3m 28s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/d98265bfd92a46b9b572ca0928f061ca : SUCCESS in 4m 58s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/e719d4e2bb09457d9298fd98ff5113f7 : SUCCESS in 4m 41s\n- kolla-build-ubuntu-binary https://zuul.opendev.org/t/openstack/build/83d0782e10fc47fc83e3c77ec887bc00 : SUCCESS in 33m 28s (non-voting)\n- kolla-build-ubuntu-source https://zuul.opendev.org/t/openstack/build/0007f224d3ea45aba74d6e593dcb8d2b : SUCCESS in 38m 55s\n- kolla-ansible-ubuntu-source https://zuul.opendev.org/t/openstack/build/7cb85f77733147288b94b76eefbcbff8 : SUCCESS in 1h 00m 31s\n- kolla-ansible-ubuntu-binary https://zuul.opendev.org/t/openstack/build/2fea03cf2e884268adb83ac35612a06a : SUCCESS in 49m 22s (non-voting)\n- kolla-ansible-ubuntu-source-upgrade https://zuul.opendev.org/t/openstack/build/8e662e5d3b0846e48d8715243af4db78 : SUCCESS in 1h 10m 31s","accounts_in_message":[],"_revision_number":1},{"id":"c65cb89356693a7487e63d89ee76c22790abb9eb","author":{"_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"},"date":"2019-09-20 14:55:51.000000000","message":"Patch Set 1:\n\nrecheck","accounts_in_message":[],"_revision_number":1},{"id":"d8cb5ce4271a4b36d3e2bc6b0bbdce6e6827a415","author":{"_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"},"date":"2019-09-20 14:56:37.000000000","message":"Patch Set 1: Code-Review+2\n\nthis is a fix \u003d backportable","accounts_in_message":[],"_revision_number":1},{"id":"fa2f19807a9b1aec3b470575012aa834960b7189","author":{"_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"},"date":"2019-09-20 14:56:58.000000000","message":"Patch Set 1:\n\n@Dincer: please create and refer to bug reports next time ;-)","accounts_in_message":[],"_revision_number":1},{"id":"c7fb2e64bdc877558b24522eea93aea2db47516a","author":{"_account_id":30523,"name":"Dincer Celik","email":"hello@dincercelik.com","username":"osmanlicilegi"},"date":"2019-09-20 15:42:16.000000000","message":"Patch Set 1:\n\nI will :) Thank you.","accounts_in_message":[],"_revision_number":1},{"id":"921912b4da86274f23f03c25a3a8555afae48174","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-09-20 17:54:56.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\n\n- kolla-build-centos-binary https://zuul.opendev.org/t/openstack/build/2c5fefcdc8814b8e9c5bde9ce80e700d : SUCCESS in 53m 53s (non-voting)\n- kolla-build-centos-source https://zuul.opendev.org/t/openstack/build/5419928a5fd84e658660c189935f4261 : SUCCESS in 53m 05s\n- kolla-ansible-centos-source https://zuul.opendev.org/t/openstack/build/2d191708bf8d456bb319e998352dd2e2 : SUCCESS in 1h 14m 22s\n- kolla-ansible-centos-binary https://zuul.opendev.org/t/openstack/build/9bf1cf58c21a42c7930e67d04b0f7b6a : SUCCESS in 1h 18m 36s (non-voting)\n- kolla-ansible-centos-source-upgrade https://zuul.opendev.org/t/openstack/build/4255fb106b9b4e39bca3ccbfa475ea28 : SUCCESS in 1h 48m 27s\n- tripleo-build-containers-centos-7 https://zuul.opendev.org/t/openstack/build/b4e4ac806f934b0baf1dee6188d9fb3d : SUCCESS in 1h 09m 51s (non-voting)\n- kolla-build-oraclelinux-binary https://zuul.opendev.org/t/openstack/build/c49f5a36dce04209a0fddc0401770b0e : SUCCESS in 47m 54s (non-voting)\n- kolla-build-oraclelinux-source https://zuul.opendev.org/t/openstack/build/224c280dacea4907bb8e4272ea09984d : SUCCESS in 45m 45s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/973dc1304d8e432787052ed36c9384ca : SUCCESS in 3m 27s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/c6ae8f87a6dd405ba776a3a609a60c28 : SUCCESS in 3m 30s\n- openstack-tox-py27 https://zuul.opendev.org/t/openstack/build/05599dd512244acc81bf6580a5ba54ea : SUCCESS in 3m 47s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/a581e8cc265f478e9a5957005b5002e7 : SUCCESS in 3m 15s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/4da73549a93f45308fad4a01fc72a5c1 : SUCCESS in 3m 25s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/2566c01e8a454b438fdddf9fdc6e87d7 : SUCCESS in 3m 16s\n- kolla-build-ubuntu-binary https://zuul.opendev.org/t/openstack/build/bc7212e763614172a9459f771bf0caaa : SUCCESS in 24m 45s (non-voting)\n- kolla-build-ubuntu-source https://zuul.opendev.org/t/openstack/build/d3fca5b959cf4ee6b079f7e6acc7bc50 : SUCCESS in 31m 08s\n- kolla-ansible-ubuntu-source https://zuul.opendev.org/t/openstack/build/0e9d5558b3234fcaac2082e3d391f832 : SUCCESS in 57m 34s\n- kolla-ansible-ubuntu-binary https://zuul.opendev.org/t/openstack/build/1604784c567a478ba6f7f14149e7a3d4 : SUCCESS in 51m 37s (non-voting)\n- kolla-ansible-ubuntu-source-upgrade https://zuul.opendev.org/t/openstack/build/5c63663c13914f0caf41f34356d9a67f : SUCCESS in 1h 16m 51s","accounts_in_message":[],"_revision_number":1},{"id":"48c228947c9d5a9cc14abcd4734e96f48dc98e58","author":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"date":"2019-09-24 07:43:15.000000000","message":"Patch Set 1: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":1},{"id":"a80fdfe060b0b5e2931cfff459f85030a0fecf92","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-09-24 07:43:33.000000000","message":"Patch Set 1: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":1},{"id":"3fd3b169ae677799fc47f0268f924fb9ff7ee289","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-09-24 08:45:07.000000000","message":"Patch Set 1: Verified+2\n\nBuild succeeded (gate pipeline).\n\n- kolla-build-centos-source https://zuul.opendev.org/t/openstack/build/5d887d0c14614807816ad3411cb3bf2b : SUCCESS in 49m 01s\n- kolla-build-oraclelinux-source https://zuul.opendev.org/t/openstack/build/02f0f7e9ec234b32882cd91ef03a4ae9 : SUCCESS in 51m 49s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/ece3739369424a4196f6110fe63b4744 : SUCCESS in 3m 38s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/6689908d118c4c889c53f4bd36a3ef9b : SUCCESS in 3m 23s\n- openstack-tox-py27 https://zuul.opendev.org/t/openstack/build/49e3056b733b4fa5993c4b7f8e8fec20 : SUCCESS in 3m 30s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/cb929b1b68ef40768e3cd509b010e0c9 : SUCCESS in 3m 18s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/a707cff162a24a3b905cba123e97b945 : SUCCESS in 3m 23s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/e368d06cc81e44b086fc990e6e07246f : SUCCESS in 3m 21s\n- kolla-build-ubuntu-source https://zuul.opendev.org/t/openstack/build/d9a493c7fd2a4c5b864487faf5f0abc8 : SUCCESS in 30m 35s","accounts_in_message":[],"_revision_number":1},{"id":"5ee7f7f705f378df68cb89643f8a0b849be6b1f3","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-09-24 08:45:08.000000000","message":"Change has been successfully merged by Zuul","accounts_in_message":[],"_revision_number":1},{"id":"115f93080321d54371ae9c3f1b49dafc339cd0de","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-09-24 08:47:49.000000000","message":"Patch Set 1:\n\nBuild succeeded (promote pipeline).\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/5bf2ce50465946ee87c4bfa125107fb8 : SUCCESS in 55s","accounts_in_message":[],"_revision_number":1}],"current_revision_number":1,"current_revision":"57fa08a92a0eaeed48648ba47e1a24400979ba21","revisions":{"57fa08a92a0eaeed48648ba47e1a24400979ba21":{"kind":"REWORK","_number":1,"created":"2019-09-20 13:03:05.000000000","uploader":{"_account_id":30523,"name":"Dincer Celik","email":"hello@dincercelik.com","username":"osmanlicilegi"},"ref":"refs/changes/78/683378/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/kolla","ref":"refs/changes/78/683378/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/kolla refs/changes/78/683378/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/kolla refs/changes/78/683378/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/kolla refs/changes/78/683378/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/kolla refs/changes/78/683378/1"}}},"commit":{"parents":[{"commit":"187d6c52f1bbb1176d7129a2b24b0cf48ba29d79","subject":"base: fix OracleLinux build","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/187d6c52f1bbb1176d7129a2b24b0cf48ba29d79"}]}],"author":{"name":"Dincer Celik","email":"hello@dincercelik.com","date":"2019-09-19 13:36:31.000000000","tz":180},"committer":{"name":"Dincer Celik","email":"hello@dincercelik.com","date":"2019-09-20 13:03:05.000000000","tz":0},"subject":"[horizon] Install octavia dashboard on deb","message":"[horizon] Install octavia dashboard on deb\n\nChange-Id: Iec78660049085483b207f4bdcf48f572d845c600\n(cherry picked from commit 77e8285e897feb4789d1b24c69bfa3be8a468f78)\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/57fa08a92a0eaeed48648ba47e1a24400979ba21"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/57fa08a92a0eaeed48648ba47e1a24400979ba21"}]},"branch":"refs/heads/stable/stein"}},"requirements":[],"submit_records":[],"submit_requirements":[]}
