)]}'
{"id":"openstack%2Fkolla~683399","triplet_id":"openstack%2Fkolla~stable%2Frocky~I93fc0e383da8fec6b3ca31f8094321c2a0c3af71","project":"openstack/kolla","branch":"stable/rocky","topic":"bug/1844697-stable/rocky","hashtags":[],"change_id":"I93fc0e383da8fec6b3ca31f8094321c2a0c3af71","subject":"Fix retries when pushing images","status":"MERGED","created":"2019-09-20 14:15:27.000000000","updated":"2019-09-23 22:12:09.000000000","submitted":"2019-09-23 22:12:09.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"683399-1569276729169-c425b5ec","meta_rev_id":"35969fdd73c58d14a018ca0f0ea5d9f6e4193f04","_number":683399,"virtual_id_number":683399,"owner":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":19316,"name":"Eduardo Gonzalez","email":"dabarren@gmail.com","username":"egonzalez90"},{"value":0,"date":"2019-09-23 19:01:05.000000000","_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},{"value":0,"_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2019-09-23 22:12:08.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":19316,"name":"Eduardo Gonzalez","email":"dabarren@gmail.com","username":"egonzalez90"},"all":[{"value":2,"date":"2019-09-23 11:39:04.000000000","_account_id":19316,"name":"Eduardo Gonzalez","email":"dabarren@gmail.com","username":"egonzalez90"},{"value":0,"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},{"value":2,"date":"2019-09-20 14:54:58.000000000","_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"},"all":[{"value":0,"_account_id":19316,"name":"Eduardo Gonzalez","email":"dabarren@gmail.com","username":"egonzalez90"},{"value":0,"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},{"value":1,"date":"2019-09-23 14:00:44.000000000","_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Backport-Candidate":{"all":[{"value":0,"_account_id":19316,"name":"Eduardo Gonzalez","email":"dabarren@gmail.com","username":"egonzalez90"},{"value":0,"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},{"value":0,"_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Do Not Backport"," 0":"Backport Review Needed","+1":"Should Backport"},"description":"","default_value":0,"optional":true},"Review-Priority":{"all":[{"value":0,"_account_id":19316,"name":"Eduardo Gonzalez","email":"dabarren@gmail.com","username":"egonzalez90"},{"value":0,"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},{"value":0,"_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Branch Freeze"," 0":"No Priority","+1":"Important Change","+2":"Gate Blocker Fix / Urgent Change"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},{"_account_id":19316,"name":"Eduardo Gonzalez","email":"dabarren@gmail.com","username":"egonzalez90"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2019-09-23 11:39:04.000000000","updated_by":{"_account_id":19316,"name":"Eduardo Gonzalez","email":"dabarren@gmail.com","username":"egonzalez90"},"reviewer":{"_account_id":19316,"name":"Eduardo Gonzalez","email":"dabarren@gmail.com","username":"egonzalez90"},"state":"REVIEWER"},{"updated":"2019-09-23 14:00:44.000000000","updated_by":{"_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"},"reviewer":{"_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"},"state":"REVIEWER"},{"updated":"2019-09-23 22:12:08.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"}],"messages":[{"id":"9a97da4fc542e7ca5825f97582676e6cdac323bc","author":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},"date":"2019-09-20 14:15:27.000000000","message":"Patch Set 1: Cherry Picked from branch master.","accounts_in_message":[],"_revision_number":1},{"id":"7b8d5f41b64bfbb1dd1050e284194cb84d0b50ba","author":{"_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"},"date":"2019-09-20 14:54:58.000000000","message":"Patch Set 1: Code-Review+2","accounts_in_message":[],"_revision_number":1},{"id":"7e1db8538d1733641ca1e9929eef0112590c55e4","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-09-20 16:34:58.000000000","message":"Patch Set 1: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttp://docs.openstack.org/infra/manual/developers.html#automated-testing\n\n\n- kolla-build-centos-binary https://zuul.opendev.org/t/openstack/build/eac8256321c748c08cb2a90c78f60170 : FAILURE in 50m 31s (non-voting)\n- kolla-build-centos-source https://zuul.opendev.org/t/openstack/build/db4bfbc1c5454cd8abd9a2858d289296 : FAILURE in 1h 13m 58s\n- kolla-ansible-centos-source https://zuul.opendev.org/t/openstack/build/c5670fc22a6b4abe8ab88b49d9af5ae2 : SUCCESS in 1h 37m 21s\n- kolla-ansible-centos-binary https://zuul.opendev.org/t/openstack/build/72ce01818ace4872b257d137922595e1 : SUCCESS in 1h 22m 04s (non-voting)\n- kolla-build-oraclelinux-binary https://zuul.opendev.org/t/openstack/build/45c868327d794a208e74def80e8a962b : FAILURE in 41m 42s (non-voting)\n- kolla-build-oraclelinux-source https://zuul.opendev.org/t/openstack/build/70651745ffef4f9fa2b5b7f692bf58b8 : FAILURE in 1h 00m 24s\n- kolla-ansible-oraclelinux-source https://zuul.opendev.org/t/openstack/build/cd41293ef9514781a78541ab8eedf45e : SUCCESS in 1h 38m 12s (non-voting)\n- kolla-ansible-oraclelinux-binary https://zuul.opendev.org/t/openstack/build/4ea033f75d6243d0a41b953620958b86 : SUCCESS in 1h 17m 56s (non-voting)\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/d0e98e2ae8e7463f8ea02ad0df55cf5f : SUCCESS in 5m 04s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/7dfa5597de1a4a088930581ac9778a33 : SUCCESS in 3m 17s\n- openstack-tox-py27 https://zuul.opendev.org/t/openstack/build/1f0f593b401f40c59ca0d79d817e17df : SUCCESS in 2m 48s\n- openstack-tox-py35 https://zuul.opendev.org/t/openstack/build/2d6f877a776049d0a8dda71eb16bc28e : SUCCESS in 4m 06s\n- build-openstack-sphinx-docs https://zuul.opendev.org/t/openstack/build/dcc79b73b414421a97fe7385137a073f : SUCCESS in 3m 30s\n- kolla-build-ubuntu-binary https://zuul.opendev.org/t/openstack/build/420ffaad03c148bdbf2e7e77d219ec19 : SUCCESS in 1h 32m 02s (non-voting)\n- kolla-build-ubuntu-source https://zuul.opendev.org/t/openstack/build/e625c4293d76429db75a8c48627b5ac7 : SUCCESS in 42m 27s\n- kolla-ansible-ubuntu-source https://zuul.opendev.org/t/openstack/build/554dbf2a3ff143bb9c59044af74fbed2 : SUCCESS in 58m 01s\n- kolla-ansible-ubuntu-binary https://zuul.opendev.org/t/openstack/build/cb7d9fab92134dbd857c31e099039484 : TIMED_OUT in 2h 01m 52s (non-voting)","accounts_in_message":[],"_revision_number":1},{"id":"f8345c7ae10af7f9527a68b2c8f0d53bd93f06ca","author":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},"date":"2019-09-23 10:24:21.000000000","message":"Patch Set 1:\n\nrecheck","accounts_in_message":[],"_revision_number":1},{"id":"58be6daf0c109e07e6aa95a312e3273a49b6725b","author":{"_account_id":19316,"name":"Eduardo Gonzalez","email":"dabarren@gmail.com","username":"egonzalez90"},"date":"2019-09-23 11:39:04.000000000","message":"Patch Set 1: Code-Review+2","accounts_in_message":[],"_revision_number":1},{"id":"e495557e22f559a6ad106fde9782545d8a5ea133","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-09-23 12:03:23.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\n\n- kolla-build-centos-binary https://zuul.opendev.org/t/openstack/build/a696ca5f2d79424d9a4a986f9bc42315 : SUCCESS in 49m 44s (non-voting)\n- kolla-build-centos-source https://zuul.opendev.org/t/openstack/build/141c3cf24bc84600a62f8019cea8c20d : SUCCESS in 51m 48s\n- kolla-ansible-centos-source https://zuul.opendev.org/t/openstack/build/b98b02624194453eb03651cfc69e66f7 : SUCCESS in 1h 16m 44s\n- kolla-ansible-centos-binary https://zuul.opendev.org/t/openstack/build/042a956df4aa4f87afd29f8270ac2838 : SUCCESS in 1h 15m 07s (non-voting)\n- kolla-build-oraclelinux-binary https://zuul.opendev.org/t/openstack/build/d505370387df48e1b020ffb614521665 : SUCCESS in 40m 12s (non-voting)\n- kolla-build-oraclelinux-source https://zuul.opendev.org/t/openstack/build/8d92fac3d61a438aa143346440efe68a : SUCCESS in 52m 08s\n- kolla-ansible-oraclelinux-source https://zuul.opendev.org/t/openstack/build/786a48d641ae489f9ef29355227d0c95 : SUCCESS in 1h 18m 40s (non-voting)\n- kolla-ansible-oraclelinux-binary https://zuul.opendev.org/t/openstack/build/ddfbbc4e20804109a079f37bdb351847 : SUCCESS in 1h 27m 22s (non-voting)\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/656721cdb7b84b38bb3a9aa5d0768238 : SUCCESS in 4m 17s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/5241d9a63b9e401b9b104a5843ae9f68 : SUCCESS in 4m 17s\n- openstack-tox-py27 https://zuul.opendev.org/t/openstack/build/a137311e2acb479689d37eeb4b3883d7 : SUCCESS in 5m 24s\n- openstack-tox-py35 https://zuul.opendev.org/t/openstack/build/2b6d5112200040e98818727665ec5a27 : SUCCESS in 5m 03s\n- build-openstack-sphinx-docs https://zuul.opendev.org/t/openstack/build/29454e09cb9a479990a30b3d085f9688 : SUCCESS in 5m 19s\n- kolla-build-ubuntu-binary https://zuul.opendev.org/t/openstack/build/2f6c148439584c5e995d355bec04f489 : SUCCESS in 1h 00m 02s (non-voting)\n- kolla-build-ubuntu-source https://zuul.opendev.org/t/openstack/build/94301d9cfefb4bd5a88d635fea029c7e : SUCCESS in 37m 07s\n- kolla-ansible-ubuntu-source https://zuul.opendev.org/t/openstack/build/f80e6c2552704e0e9b4e7f6b1415aa88 : SUCCESS in 1h 22m 05s\n- kolla-ansible-ubuntu-binary https://zuul.opendev.org/t/openstack/build/5ab954b7c11d42cc84fd9f52a9ef10ff : SUCCESS in 1h 27m 55s (non-voting)","accounts_in_message":[],"_revision_number":1},{"id":"24e73e0bf8e90d4536dd23bcd19a2d7a0948008e","author":{"_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"},"date":"2019-09-23 14:00:44.000000000","message":"Patch Set 1: Workflow+1","accounts_in_message":[],"_revision_number":1},{"id":"af1681544ceb736e314e37c0e5e46aa9f470091e","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-09-23 14:00:57.000000000","message":"Patch Set 1: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":1},{"id":"49d40ac61359ce31e3e442ba26dfacdf8d5a037c","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-09-23 15:00:09.000000000","message":"Patch Set 1: Verified-2\n\nBuild failed (gate pipeline).  For information on how to proceed, see\nhttp://docs.openstack.org/infra/manual/developers.html#automated-testing\n\n\n- kolla-build-centos-source https://zuul.opendev.org/t/openstack/build/2c257bcbab9047b8bedda3cd591ade31 : FAILURE in 24m 00s\n- kolla-build-oraclelinux-source https://zuul.opendev.org/t/openstack/build/87766b83915a4a6b8df69e2711a36960 : SUCCESS in 51m 53s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/75a0f6be027649158917d48fe304825e : SUCCESS in 5m 10s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/a301a333f62045bd883001b1b3b5c6c7 : SUCCESS in 4m 25s\n- openstack-tox-py27 https://zuul.opendev.org/t/openstack/build/d1cb2a82e77c4e8caf8d60968903144e : SUCCESS in 4m 21s\n- openstack-tox-py35 https://zuul.opendev.org/t/openstack/build/c11f6747e595448284e2d04e888c12c0 : SUCCESS in 3m 39s\n- build-openstack-sphinx-docs https://zuul.opendev.org/t/openstack/build/488d59c7f0fc4f36a8ee45c617c633d3 : SUCCESS in 4m 53s\n- kolla-build-ubuntu-source https://zuul.opendev.org/t/openstack/build/14a3bb348475440586c1b9cfa9bdcf77 : SUCCESS in 30m 35s","accounts_in_message":[],"_revision_number":1},{"id":"3b96e0f5935c74eb4493fa583431df61abfe32d6","author":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},"date":"2019-09-23 15:24:26.000000000","message":"Patch Set 1:\n\nrecheck","accounts_in_message":[],"_revision_number":1},{"id":"d68b4eb299aff077a127a6cebf36059e00067d9a","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-09-23 17:36:21.000000000","message":"Patch Set 1: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttp://docs.openstack.org/infra/manual/developers.html#automated-testing\n\n\n- kolla-build-centos-binary https://zuul.opendev.org/t/openstack/build/3d00c9a3a1414c9d94d7addc95c4b612 : SUCCESS in 43m 31s (non-voting)\n- kolla-build-centos-source https://zuul.opendev.org/t/openstack/build/7b2fa43e884e4c17b65b4ab770f16154 : SUCCESS in 51m 04s\n- kolla-ansible-centos-source https://zuul.opendev.org/t/openstack/build/dde21e7bb0a642789de68019fd37f6d2 : FAILURE in 58m 49s\n- kolla-ansible-centos-binary https://zuul.opendev.org/t/openstack/build/bfb236c5f60c46cb853ea3c5da79230f : SUCCESS in 1h 28m 14s (non-voting)\n- kolla-build-oraclelinux-binary https://zuul.opendev.org/t/openstack/build/f1a527d9110d44f7846b16afdde167b3 : SUCCESS in 40m 34s (non-voting)\n- kolla-build-oraclelinux-source https://zuul.opendev.org/t/openstack/build/60fd5627e480401999767970c5b8651c : SUCCESS in 55m 26s\n- kolla-ansible-oraclelinux-source https://zuul.opendev.org/t/openstack/build/4ac5173603d9473898a5547030041199 : SUCCESS in 1h 32m 18s (non-voting)\n- kolla-ansible-oraclelinux-binary https://zuul.opendev.org/t/openstack/build/a3559e16ff774169beb9329f3b1181b1 : SUCCESS in 1h 09m 45s (non-voting)\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/9e91502df4174c1191db5251896972e4 : SUCCESS in 3m 10s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/8f334a5970c4495c85b96c2da79c3d0f : SUCCESS in 6m 13s\n- openstack-tox-py27 https://zuul.opendev.org/t/openstack/build/f1df05adb5f04744aa964e1db24591fd : SUCCESS in 4m 13s\n- openstack-tox-py35 https://zuul.opendev.org/t/openstack/build/caa4953814d74399b026159b230af872 : SUCCESS in 3m 32s\n- build-openstack-sphinx-docs https://zuul.opendev.org/t/openstack/build/1b37224319544e2ebdc956ad39b0f037 : SUCCESS in 4m 43s\n- kolla-build-ubuntu-binary https://zuul.opendev.org/t/openstack/build/70b878226cd54c3fab2dbd7ef5ee3342 : SUCCESS in 1h 02m 29s (non-voting)\n- kolla-build-ubuntu-source https://zuul.opendev.org/t/openstack/build/43fda4ed10e24499b8b55a55098d4111 : SUCCESS in 36m 37s\n- kolla-ansible-ubuntu-source https://zuul.opendev.org/t/openstack/build/7e8ae0bb6cf142da87415d36ca7483c7 : SUCCESS in 1h 09m 58s\n- kolla-ansible-ubuntu-binary https://zuul.opendev.org/t/openstack/build/6d8576d2275249b0b1e6634ad73dbbd3 : TIMED_OUT in 2h 01m 28s (non-voting)","accounts_in_message":[],"_revision_number":1},{"id":"f5700ce7a948c35c94c88866c662048d8c88fb5a","author":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},"date":"2019-09-23 19:01:05.000000000","message":"Patch Set 1:\n\nrecheck","accounts_in_message":[],"_revision_number":1},{"id":"e6c2645cae8ca4daf59cd3f7833f912cff42b7ca","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-09-23 21:14:47.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\n\n- kolla-build-centos-binary https://zuul.opendev.org/t/openstack/build/ec58379c9a2845bb96094811bd9a760f : SUCCESS in 44m 18s (non-voting)\n- kolla-build-centos-source https://zuul.opendev.org/t/openstack/build/665d8c4c5d5e4e15abdbbe66b4174921 : SUCCESS in 1h 03m 31s\n- kolla-ansible-centos-source https://zuul.opendev.org/t/openstack/build/f72604a33bbf4a059d6fc3bc3e080f14 : SUCCESS in 1h 22m 45s\n- kolla-ansible-centos-binary https://zuul.opendev.org/t/openstack/build/edd8bb29402246f7bbfd4f0dc88c665d : SUCCESS in 1h 18m 33s (non-voting)\n- kolla-build-oraclelinux-binary https://zuul.opendev.org/t/openstack/build/fde50192658e4410ad1ce5b3b6fe4af6 : SUCCESS in 48m 38s (non-voting)\n- kolla-build-oraclelinux-source https://zuul.opendev.org/t/openstack/build/02800598da004689ab5f6f5b21e426d2 : SUCCESS in 49m 38s\n- kolla-ansible-oraclelinux-source https://zuul.opendev.org/t/openstack/build/136dd469e5124945935adc64885e277e : FAILURE in 1h 14m 24s (non-voting)\n- kolla-ansible-oraclelinux-binary https://zuul.opendev.org/t/openstack/build/9d4fd4ef5c57451a94313070c2897574 : SUCCESS in 1h 24m 11s (non-voting)\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/0f9d570277ca496885f870d7ac5b4d0e : SUCCESS in 3m 46s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/592ac6ac0e9d4390a57ba5caeecb9c0f : SUCCESS in 4m 08s\n- openstack-tox-py27 https://zuul.opendev.org/t/openstack/build/a699cd0d29f34613920165e02692f7e1 : SUCCESS in 3m 45s\n- openstack-tox-py35 https://zuul.opendev.org/t/openstack/build/c205c7f06936445eb59f762df49a8fa2 : SUCCESS in 3m 52s\n- build-openstack-sphinx-docs https://zuul.opendev.org/t/openstack/build/a90ffa3d25244460aa203a153cf9af49 : SUCCESS in 5m 43s\n- kolla-build-ubuntu-binary https://zuul.opendev.org/t/openstack/build/1a9cb6594bb64153a3f3ccb2461de49d : SUCCESS in 1h 24m 34s (non-voting)\n- kolla-build-ubuntu-source https://zuul.opendev.org/t/openstack/build/d7b1684d66b045f0849f345532882540 : SUCCESS in 32m 57s\n- kolla-ansible-ubuntu-source https://zuul.opendev.org/t/openstack/build/f610f25b0d504b17860a71d61cfb9bed : SUCCESS in 1h 08m 59s\n- kolla-ansible-ubuntu-binary https://zuul.opendev.org/t/openstack/build/4e97ff21496142fc800f181f8f30fe6f : TIMED_OUT in 2h 02m 27s (non-voting)","accounts_in_message":[],"_revision_number":1},{"id":"d7887d7ded37cbdca4d53b7ec62c292e2ce16a60","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-09-23 21:15:00.000000000","message":"Patch Set 1: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":1},{"id":"20c9d7cd65c102be0ef3958c87a19a5f96d02485","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-09-23 22:12:08.000000000","message":"Patch Set 1: Verified+2\n\nBuild succeeded (gate pipeline).\n\n- kolla-build-centos-source https://zuul.opendev.org/t/openstack/build/6c90406d2fac41759f110798b0009955 : SUCCESS in 53m 36s\n- kolla-build-oraclelinux-source https://zuul.opendev.org/t/openstack/build/5e128463df8947a19e248e2a20d135f9 : SUCCESS in 47m 03s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/43519c01ea6e45e18755184005c393f5 : SUCCESS in 5m 45s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/2190c15e758f43f695aa4e9d63636983 : SUCCESS in 4m 51s\n- openstack-tox-py27 https://zuul.opendev.org/t/openstack/build/0cbe4ffbd1a9441ab104c4056328c9c4 : SUCCESS in 4m 39s\n- openstack-tox-py35 https://zuul.opendev.org/t/openstack/build/925a7b11e8e64d978ce9c3395af5a3ed : SUCCESS in 5m 10s\n- build-openstack-sphinx-docs https://zuul.opendev.org/t/openstack/build/9dcd06ae678c404b8cc148fd2edc9a5d : SUCCESS in 5m 05s\n- kolla-build-ubuntu-source https://zuul.opendev.org/t/openstack/build/1d53f406904a4cbdb57d5d5d294d6c37 : SUCCESS in 36m 20s","accounts_in_message":[],"_revision_number":1},{"id":"52c889ebfd0cc4760774b3b696e854e4fe904748","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-09-23 22:12:09.000000000","message":"Change has been successfully merged by Zuul","accounts_in_message":[],"_revision_number":1}],"current_revision_number":1,"current_revision":"7d4b06f9b0fe9709385fd7af3fe39b5baa865a95","revisions":{"7d4b06f9b0fe9709385fd7af3fe39b5baa865a95":{"kind":"REWORK","_number":1,"created":"2019-09-20 14:15:27.000000000","uploader":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},"ref":"refs/changes/99/683399/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/kolla","ref":"refs/changes/99/683399/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/kolla refs/changes/99/683399/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/kolla refs/changes/99/683399/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/kolla refs/changes/99/683399/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/kolla refs/changes/99/683399/1"}}},"commit":{"parents":[{"commit":"5693bfd698a9f00f5fefd0ab0c4c3c38628223bb","subject":"base: fix OracleLinux build","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/5693bfd698a9f00f5fefd0ab0c4c3c38628223bb"}]}],"author":{"name":"Mark Goddard","email":"mark@stackhpc.com","date":"2019-09-19 16:20:50.000000000","tz":60},"committer":{"name":"Mark Goddard","email":"mark@stackhpc.com","date":"2019-09-20 14:15:27.000000000","tz":0},"subject":"Fix retries when pushing images","message":"Fix retries when pushing images\n\nCurrently the retry mechanism is broken for pushing because the image\nstate gets set to an error state, and is never reset to \u0027built\u0027. This\nprevents the PushTask from setting success to True.\n\nThis change sets the image state to \u0027built\u0027 if the push succeeds,\nensuring it overrides any previous failures.\n\nChange-Id: I93fc0e383da8fec6b3ca31f8094321c2a0c3af71\nCloses-Bug: #1844697\n(cherry picked from commit f8ded663891f79c289156ee61d4e02197d80ce7a)\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/7d4b06f9b0fe9709385fd7af3fe39b5baa865a95"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/7d4b06f9b0fe9709385fd7af3fe39b5baa865a95"}]},"branch":"refs/heads/stable/rocky"}},"requirements":[],"submit_records":[],"submit_requirements":[]}
