)]}'
{"id":"openstack%2Fkolla~683674","triplet_id":"openstack%2Fkolla~master~Ie140cf0d6308a5ee3c09f06ddfcf63be2e1cec64","project":"openstack/kolla","branch":"master","hashtags":[],"change_id":"Ie140cf0d6308a5ee3c09f06ddfcf63be2e1cec64","subject":"Add Debian Buster Stein backports APT repository","status":"MERGED","created":"2019-09-20 18:14:28.000000000","updated":"2019-09-24 11:56:35.000000000","submitted":"2019-09-24 11:55:28.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":3,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"683674-1569326128603-657de88e","meta_rev_id":"ca678de0bbff9f9ee98f0857fd4d13448a09275a","_number":683674,"virtual_id_number":683674,"owner":{"_account_id":28374,"name":"Viktor Michalek","email":"viktor.michalek@ultimum.io","username":"viktormichalek"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":24072,"name":"Marcin Juszkiewicz","email":"mjuszkiewicz@redhat.com","username":"hrw"},{"value":0,"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},{"value":0,"_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2019-09-24 11:55:28.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"date":"2019-09-24 07:54:13.000000000","_account_id":23181,"name":"RDO Third Party CI","email":"dmsimard+rdothirdparty@redhat.com","username":"rdothirdparty","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},"all":[{"value":0,"_account_id":24072,"name":"Marcin Juszkiewicz","email":"mjuszkiewicz@redhat.com","username":"hrw"},{"value":2,"date":"2019-09-24 08:29:45.000000000","_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},{"value":2,"date":"2019-09-24 07:16:52.000000000","_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":23181,"name":"RDO Third Party CI","email":"dmsimard+rdothirdparty@redhat.com","username":"rdothirdparty","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},"all":[{"value":0,"_account_id":24072,"name":"Marcin Juszkiewicz","email":"mjuszkiewicz@redhat.com","username":"hrw"},{"value":1,"date":"2019-09-24 08:29:45.000000000","_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},{"value":0,"_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":23181,"name":"RDO Third Party CI","email":"dmsimard+rdothirdparty@redhat.com","username":"rdothirdparty","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Backport-Candidate":{"all":[{"value":0,"_account_id":24072,"name":"Marcin Juszkiewicz","email":"mjuszkiewicz@redhat.com","username":"hrw"},{"value":0,"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},{"value":0,"_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":23181,"name":"RDO Third Party CI","email":"dmsimard+rdothirdparty@redhat.com","username":"rdothirdparty","tags":["SERVICE_USER"]}],"values":{"-1":"Do Not Backport"," 0":"Backport Review Needed","+1":"Should Backport"},"description":"","default_value":0,"optional":true},"Review-Priority":{"all":[{"value":0,"_account_id":24072,"name":"Marcin Juszkiewicz","email":"mjuszkiewicz@redhat.com","username":"hrw"},{"value":0,"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},{"value":0,"_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":23181,"name":"RDO Third Party CI","email":"dmsimard+rdothirdparty@redhat.com","username":"rdothirdparty","tags":["SERVICE_USER"]}],"values":{"-1":"Branch Freeze"," 0":"No Priority","+1":"Important Change","+2":"Gate Blocker Fix / Urgent Change"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":23181,"name":"RDO Third Party CI","email":"dmsimard+rdothirdparty@redhat.com","username":"rdothirdparty","tags":["SERVICE_USER"]},{"_account_id":24072,"name":"Marcin Juszkiewicz","email":"mjuszkiewicz@redhat.com","username":"hrw"},{"_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2019-09-20 20:37:34.000000000","updated_by":{"_account_id":24072,"name":"Marcin Juszkiewicz","email":"mjuszkiewicz@redhat.com","username":"hrw"},"reviewer":{"_account_id":24072,"name":"Marcin Juszkiewicz","email":"mjuszkiewicz@redhat.com","username":"hrw"},"state":"REVIEWER"},{"updated":"2019-09-24 07:16:52.000000000","updated_by":{"_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"},"reviewer":{"_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"},"state":"REVIEWER"},{"updated":"2019-09-24 07:54:13.000000000","updated_by":{"_account_id":23181,"name":"RDO Third Party CI","email":"dmsimard+rdothirdparty@redhat.com","username":"rdothirdparty","tags":["SERVICE_USER"]},"reviewer":{"_account_id":23181,"name":"RDO Third Party CI","email":"dmsimard+rdothirdparty@redhat.com","username":"rdothirdparty","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2019-09-24 08:29:45.000000000","updated_by":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},"reviewer":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},"state":"REVIEWER"},{"updated":"2019-09-24 11:55:28.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"}],"messages":[{"id":"ed7dfaef9ab571f1cddcfd48a548dc83de2f7098","author":{"_account_id":28374,"name":"Viktor Michalek","email":"viktor.michalek@ultimum.io","username":"viktormichalek"},"date":"2019-09-20 18:14:28.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"4791db37867a9610582c8da6ef55f22a40f8e49c","author":{"_account_id":23181,"name":"RDO Third Party CI","email":"dmsimard+rdothirdparty@redhat.com","username":"rdothirdparty","tags":["SERVICE_USER"]},"date":"2019-09-20 19:04:03.000000000","message":"Patch Set 1:\n\nBuild succeeded (check pipeline).\n\n- tripleo-build-containers-rhel-8-master http://logs.rdoproject.org/74/683674/1/openstack-check/tripleo-build-containers-rhel-8-master/c9b8d60/ : SUCCESS in 47m 47s","accounts_in_message":[],"_revision_number":1},{"id":"b2f737d7e0e7e0ddbfce36720bb2bcfbe178bb60","author":{"_account_id":24072,"name":"Marcin Juszkiewicz","email":"mjuszkiewicz@redhat.com","username":"hrw"},"date":"2019-09-20 20:37:34.000000000","message":"Patch Set 1:\n\nplease split out td-agent part\n\nthis patch looks like good stuff for stable/stein backport as it gives us better working Debian/binary stein images","accounts_in_message":[],"_revision_number":1},{"id":"635bba0c4dda3183da94f8b7992f899926d66695","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-09-20 20:55:43.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\n\n- kolla-build-centos-binary https://zuul.opendev.org/t/openstack/build/f6ae14d357b24c6b98b3933f8783032e : SUCCESS in 47m 34s (non-voting)\n- kolla-build-centos-source https://zuul.opendev.org/t/openstack/build/3865740f0b3b466b8c49d432ac34ac29 : SUCCESS in 1h 11m 37s\n- kolla-ansible-centos-source https://zuul.opendev.org/t/openstack/build/1229f5500e1e436abfcfbd9d9adae1d6 : SUCCESS in 1h 37m 06s\n- kolla-ansible-centos-binary https://zuul.opendev.org/t/openstack/build/93cea3c6acb7483c833261854e2a7272 : SUCCESS in 1h 40m 02s (non-voting)\n- kolla-ansible-centos-source-upgrade https://zuul.opendev.org/t/openstack/build/475a32ac4bc24971a2debbbe7486be09 : SUCCESS in 1h 31m 16s\n- tripleo-build-containers-centos-7 https://zuul.opendev.org/t/openstack/build/56d4722f774746f5b3ade73af55dcd71 : SUCCESS in 1h 14m 13s\n- kolla-ansible-bifrost-centos-source https://zuul.opendev.org/t/openstack/build/51eecf36ef3f47cab15f2079fb3ba671 : SUCCESS in 42m 02s (non-voting)\n- kolla-ansible-centos-source-ceph https://zuul.opendev.org/t/openstack/build/37183665002a4d1bb128b198c8d46f4d : SUCCESS in 2h 24m 54s (non-voting)\n- kolla-ansible-centos-source-upgrade-ceph https://zuul.opendev.org/t/openstack/build/3f7c214adf6b4d9c86025fb61d955d23 : TIMED_OUT in 2h 33m 04s (non-voting)\n- kolla-ansible-centos-source-cinder-lvm https://zuul.opendev.org/t/openstack/build/bca6f4946dd34fe2acd06fb3ce0d5212 : TIMED_OUT in 2h 04m 04s (non-voting)\n- kolla-ansible-centos-source-scenario-nfv https://zuul.opendev.org/t/openstack/build/6119d5a57dc84ae8a42f298a3b362d4f : TIMED_OUT in 2h 04m 16s (non-voting)\n- kolla-ansible-centos-source-ironic https://zuul.opendev.org/t/openstack/build/a788cd4a12724cef9f79ac41f7351fad : SUCCESS in 1h 50m 04s (non-voting)\n- kolla-build-debian-source https://zuul.opendev.org/t/openstack/build/63c9e51b1481493ab1529c085b64a13a : SUCCESS in 30m 22s (non-voting)\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/a9287ea96929458989af2ad776c70302 : SUCCESS in 4m 22s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/e0615a1098c747fca8a8cf2722243653 : SUCCESS in 4m 08s\n- openstack-tox-py27 https://zuul.opendev.org/t/openstack/build/70ec7901b958444ca5540c77db1d2a72 : SUCCESS in 3m 47s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/2e96255d5e004b74a01ada960abc4d4a : SUCCESS in 3m 01s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/60b2b6d6543b49b1ad6fe37f3476e4d4 : SUCCESS in 3m 43s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/080a344d91e64b988d7b4689e0e45b0f : SUCCESS in 7m 16s\n- kolla-build-ubuntu-binary https://zuul.opendev.org/t/openstack/build/a293fbdbe16a4fb9af2ca6ca3b39a007 : SUCCESS in 23m 59s (non-voting)\n- kolla-build-ubuntu-source https://zuul.opendev.org/t/openstack/build/2ca349fc4ef24390bc6e761194de9ccf : SUCCESS in 33m 47s\n- kolla-ansible-ubuntu-source https://zuul.opendev.org/t/openstack/build/d1861ce2ae2f4ca991e17d287b13bb2a : SUCCESS in 58m 35s\n- kolla-ansible-ubuntu-binary https://zuul.opendev.org/t/openstack/build/de760c5989ea4b61a2d403c6e2fe320b : SUCCESS in 1h 03m 21s (non-voting)\n- kolla-ansible-ubuntu-source-upgrade https://zuul.opendev.org/t/openstack/build/edff728b7b4d41b6a4c60a275b0a0be1 : SUCCESS in 1h 10m 24s\n- kolla-ansible-ubuntu-source-zun https://zuul.opendev.org/t/openstack/build/250cec0e446b4512a6abb6c7d8d29b05 : SUCCESS in 1h 30m 25s (non-voting)\n- kolla-ansible-ubuntu-source-ceph https://zuul.opendev.org/t/openstack/build/e095f33e1e3e47ffbb1e673e53731168 : SUCCESS in 1h 43m 14s (non-voting)\n- kolla-ansible-ubuntu-source-upgrade-ceph https://zuul.opendev.org/t/openstack/build/5614700f306e4422bb609967bba39d86 : SUCCESS in 1h 53m 18s (non-voting)\n- kolla-ansible-ubuntu-source-cinder-lvm https://zuul.opendev.org/t/openstack/build/03d581ee5bb24fe2bcf0966866fe9062 : SUCCESS in 1h 25m 23s (non-voting)\n- kolla-ansible-ubuntu-source-ironic https://zuul.opendev.org/t/openstack/build/9934e6a807d540fc9c8c9219dea07703 : SUCCESS in 1h 09m 13s (non-voting)","accounts_in_message":[],"_revision_number":1},{"id":"9c6de67e43d9be7a708b375f90b072c33b3bb78f","author":{"_account_id":28374,"name":"Viktor Michalek","email":"viktor.michalek@ultimum.io","username":"viktormichalek"},"date":"2019-09-22 16:28:44.000000000","message":"Patch Set 2: Commit message was updated.","accounts_in_message":[],"_revision_number":2},{"id":"051d986da32b3494ce2f58fae5f86ccc01cfd823","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-09-22 16:28:55.000000000","message":"Patch Set 2: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttp://docs.openstack.org/infra/manual/developers.html#automated-testing\n\n\n\nWarning:\n  Dependency cycle detected","accounts_in_message":[],"_revision_number":2},{"id":"539ddc0a9540ce5bdb2590e93394593f00ac3eb1","author":{"_account_id":23181,"name":"RDO Third Party CI","email":"dmsimard+rdothirdparty@redhat.com","username":"rdothirdparty","tags":["SERVICE_USER"]},"date":"2019-09-22 16:28:55.000000000","message":"Patch Set 2:\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttp://docs.openstack.org/infra/manual/developers.html#automated-testing\n\n\n\nWarning:\n  Dependency cycle detected","accounts_in_message":[],"_revision_number":2},{"id":"2687608fc39117d3dcb1e537cf92f35f2c9ef8e6","author":{"_account_id":23181,"name":"RDO Third Party CI","email":"dmsimard+rdothirdparty@redhat.com","username":"rdothirdparty","tags":["SERVICE_USER"]},"date":"2019-09-22 16:28:56.000000000","message":"Patch Set 2:\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttp://docs.openstack.org/infra/manual/developers.html#automated-testing\n\n\n\nWarning:\n  Dependency cycle detected","accounts_in_message":[],"_revision_number":2},{"id":"6836f8aaec39422749203d5918348160a81ede94","author":{"_account_id":28374,"name":"Viktor Michalek","email":"viktor.michalek@ultimum.io","username":"viktormichalek"},"date":"2019-09-22 16:31:00.000000000","message":"Patch Set 3: Commit message was updated.","accounts_in_message":[],"_revision_number":3},{"id":"f9b5045580d50c48f0e00901a3b5e861830bd62c","author":{"_account_id":23181,"name":"RDO Third Party CI","email":"dmsimard+rdothirdparty@redhat.com","username":"rdothirdparty","tags":["SERVICE_USER"]},"date":"2019-09-22 17:09:17.000000000","message":"Patch Set 3:\n\nBuild succeeded (check pipeline).\n\n- tripleo-build-containers-rhel-8-master http://logs.rdoproject.org/74/683674/3/openstack-check/tripleo-build-containers-rhel-8-master/bd4912c/ : SUCCESS in 36m 54s","accounts_in_message":[],"_revision_number":3},{"id":"d241fc7d00c7695640fdf3aa3a5312dbdd4a4ba6","author":{"_account_id":28374,"name":"Viktor Michalek","email":"viktor.michalek@ultimum.io","username":"viktormichalek"},"date":"2019-09-22 18:03:03.000000000","message":"Patch Set 4: Commit message was updated.","accounts_in_message":[],"_revision_number":4},{"id":"a85da3a87f22bbb1f1d7028cea2658c8a7d2822f","author":{"_account_id":23181,"name":"RDO Third Party CI","email":"dmsimard+rdothirdparty@redhat.com","username":"rdothirdparty","tags":["SERVICE_USER"]},"date":"2019-09-22 18:45:09.000000000","message":"Patch Set 4:\n\nBuild succeeded (check pipeline).\n\n- tripleo-build-containers-rhel-8-master http://logs.rdoproject.org/74/683674/4/openstack-check/tripleo-build-containers-rhel-8-master/f8a7c4c/ : SUCCESS in 40m 37s","accounts_in_message":[],"_revision_number":4},{"id":"a2a138c0a2469cfbb34d1814e58d51f89e409406","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-09-22 20:29:29.000000000","message":"Patch Set 4: Verified+1\n\nBuild succeeded (check pipeline).\n\n- kolla-build-centos-binary https://zuul.opendev.org/t/openstack/build/d55e0217f0ed4e15af47722c9a666370 : SUCCESS in 44m 20s (non-voting)\n- kolla-build-centos-source https://zuul.opendev.org/t/openstack/build/bc84668ed2e4408489fdf9f252e8d9fd : SUCCESS in 55m 07s\n- kolla-ansible-centos-source https://zuul.opendev.org/t/openstack/build/04ea1f23b4444b21a1b5221abdd39726 : SUCCESS in 1h 12m 59s\n- kolla-ansible-centos-binary https://zuul.opendev.org/t/openstack/build/27d560361f6a4008aad71a1357451137 : SUCCESS in 1h 32m 21s (non-voting)\n- kolla-ansible-centos-source-upgrade https://zuul.opendev.org/t/openstack/build/e6740adcd59c49419b4f28f6badb3832 : SUCCESS in 1h 33m 07s\n- tripleo-build-containers-centos-7 https://zuul.opendev.org/t/openstack/build/dbf5b5a7e8824c6b88e5616508889262 : SUCCESS in 1h 08m 06s\n- kolla-ansible-bifrost-centos-source https://zuul.opendev.org/t/openstack/build/a68c0e6c9ac648f1b565cdf85831c14d : SUCCESS in 40m 34s (non-voting)\n- kolla-ansible-centos-source-ceph https://zuul.opendev.org/t/openstack/build/d733049a58104287bd1741d38ce5baab : SUCCESS in 2h 09m 40s (non-voting)\n- kolla-ansible-centos-source-upgrade-ceph https://zuul.opendev.org/t/openstack/build/e4a091614775402d96a05d990c31f838 : FAILURE in 1h 56m 41s (non-voting)\n- kolla-ansible-centos-source-cinder-lvm https://zuul.opendev.org/t/openstack/build/aba520d2a9c54a9b8e419b2ef024d896 : TIMED_OUT in 2h 05m 55s (non-voting)\n- kolla-ansible-centos-source-scenario-nfv https://zuul.opendev.org/t/openstack/build/7d727409968b4db4ab46c694cbcb1950 : FAILURE in 2h 03m 00s (non-voting)\n- kolla-ansible-centos-source-ironic https://zuul.opendev.org/t/openstack/build/b1c228fab23d4269a5e2ecb653ce0980 : SUCCESS in 1h 25m 24s (non-voting)\n- kolla-build-debian-source https://zuul.opendev.org/t/openstack/build/384ab933ea0547a791aaf96deea12c37 : SUCCESS in 29m 26s (non-voting)\n- kolla-build-debian-binary https://zuul.opendev.org/t/openstack/build/03a8ceb57d1f465491c0e3359baa29b9 : SUCCESS in 21m 41s (non-voting)\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/0e0ea97ebbd1451996c8fbdbc613350d : SUCCESS in 3m 58s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/15c2efb92dc74a6d95f6719670759e67 : SUCCESS in 4m 37s\n- openstack-tox-py27 https://zuul.opendev.org/t/openstack/build/7e283706940f45abab9fb8d0645c209e : SUCCESS in 3m 16s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/bda3fb0878834b08ae1d0633a3f1e4c3 : SUCCESS in 3m 42s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/658de837aa7e4693aa6724e141a25198 : SUCCESS in 4m 05s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/92fdd0f8fe2a44399a1dbb681f632e06 : SUCCESS in 6m 35s\n- kolla-build-ubuntu-binary https://zuul.opendev.org/t/openstack/build/e3a1199420384c379378bf6aff0caa37 : SUCCESS in 36m 11s (non-voting)\n- kolla-build-ubuntu-source https://zuul.opendev.org/t/openstack/build/1e2f121c85894a36b54e14e37f658e20 : SUCCESS in 48m 11s\n- kolla-ansible-ubuntu-source https://zuul.opendev.org/t/openstack/build/9c9a5d6c4ceb491e835f65c931be23c5 : SUCCESS in 53m 12s\n- kolla-ansible-ubuntu-binary https://zuul.opendev.org/t/openstack/build/e2ee484f87d441db8ca3c40a67c1a0ed : SUCCESS in 1h 07m 30s (non-voting)\n- kolla-ansible-ubuntu-source-upgrade https://zuul.opendev.org/t/openstack/build/47a3658e3f9e4b569512ab80d74a59b4 : SUCCESS in 1h 28m 46s\n- kolla-ansible-ubuntu-source-zun https://zuul.opendev.org/t/openstack/build/c234baaeaf6743d1ad906ac87e09b999 : SUCCESS in 1h 16m 37s (non-voting)\n- kolla-ansible-ubuntu-source-ceph https://zuul.opendev.org/t/openstack/build/02518eb4042443e7b7619e45c00cf0a4 : SUCCESS in 1h 57m 34s (non-voting)\n- kolla-ansible-ubuntu-source-upgrade-ceph https://zuul.opendev.org/t/openstack/build/ee3b1ae2bf08473fb5d7ddf2b108e152 : SUCCESS in 2h 03m 24s (non-voting)\n- kolla-ansible-ubuntu-source-cinder-lvm https://zuul.opendev.org/t/openstack/build/4c3d99fff1a0478aa0c450df92ba0183 : SUCCESS in 1h 11m 16s (non-voting)\n- kolla-ansible-ubuntu-source-ironic https://zuul.opendev.org/t/openstack/build/0a1f0edf21d448e3bb6ccc72792d444d : SUCCESS in 1h 02m 10s (non-voting)","accounts_in_message":[],"_revision_number":4},{"id":"7766280429424608695e6851c7fc462ff75df4ed","author":{"_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"},"date":"2019-09-23 09:17:59.000000000","message":"Patch Set 4:\n\n(3 comments)","accounts_in_message":[],"_revision_number":4},{"id":"f48d89bd0c2dd0bc99f6bcf64c6803a9ba86f139","author":{"_account_id":28374,"name":"Viktor Michalek","email":"viktor.michalek@ultimum.io","username":"viktormichalek"},"date":"2019-09-23 17:38:48.000000000","message":"Uploaded patch set 5.","accounts_in_message":[],"_revision_number":5},{"id":"5f0eae8badf8ade3ac45231b91e176d99226104c","author":{"_account_id":23181,"name":"RDO Third Party CI","email":"dmsimard+rdothirdparty@redhat.com","username":"rdothirdparty","tags":["SERVICE_USER"]},"date":"2019-09-23 17:39:11.000000000","message":"Patch Set 5:\n\nMerge Failed.\n\nThis change or one of its cross-repo dependencies was unable to be automatically merged with the current state of its repository. Please rebase the change and upload a new patchset.","accounts_in_message":[],"_revision_number":5},{"id":"405e935b5f24f695e7ab549c4d03c14b228d178b","author":{"_account_id":28374,"name":"Viktor Michalek","email":"viktor.michalek@ultimum.io","username":"viktormichalek"},"date":"2019-09-23 17:41:04.000000000","message":"Uploaded patch set 6: Commit message was updated.","accounts_in_message":[],"_revision_number":6},{"id":"b6eeca395f6fc9b942f509825ac48bac6388a948","author":{"_account_id":23181,"name":"RDO Third Party CI","email":"dmsimard+rdothirdparty@redhat.com","username":"rdothirdparty","tags":["SERVICE_USER"]},"date":"2019-09-23 17:41:27.000000000","message":"Patch Set 6:\n\nMerge Failed.\n\nThis change or one of its cross-repo dependencies was unable to be automatically merged with the current state of its repository. Please rebase the change and upload a new patchset.","accounts_in_message":[],"_revision_number":6},{"id":"4d1098f23987689c12c1253976a5de26b0334b3e","author":{"_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"},"date":"2019-09-24 07:15:19.000000000","message":"Patch Set 7: Commit message was updated.","accounts_in_message":[],"_revision_number":7},{"id":"2db2ab0a28d91d8374922b50caa9cc61f60d873b","author":{"_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"},"date":"2019-09-24 07:15:33.000000000","message":"Patch Set 8: Patch Set 7 was rebased","accounts_in_message":[],"_revision_number":8},{"id":"64316e74ce314f5f516e83d6c920c54cdcbd5085","author":{"_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"},"date":"2019-09-24 07:16:09.000000000","message":"Patch Set 9: Published edit on patch set 8.","accounts_in_message":[],"_revision_number":9},{"id":"bb1f4f29390007bf4897251a2cc2899048c369db","author":{"_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"},"date":"2019-09-24 07:16:52.000000000","message":"Patch Set 9: Code-Review+2\n\nall right, should be untangled now","accounts_in_message":[],"_revision_number":9},{"id":"f192db0eee5159407e78406ce49fa0a51c2955de","author":{"_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"},"date":"2019-09-24 07:20:39.000000000","message":"Topic 683674 removed","accounts_in_message":[],"_revision_number":9},{"id":"a7d4a373613218570bc00d18ec12b80bfe46bf3f","author":{"_account_id":23181,"name":"RDO Third Party CI","email":"dmsimard+rdothirdparty@redhat.com","username":"rdothirdparty","tags":["SERVICE_USER"]},"date":"2019-09-24 07:54:13.000000000","message":"Patch Set 9:\n\nBuild succeeded (check pipeline).\n\n- tripleo-build-containers-rhel-8-master http://logs.rdoproject.org/74/683674/9/openstack-check/tripleo-build-containers-rhel-8-master/dc8271b/ : SUCCESS in 37m 25s","accounts_in_message":[],"_revision_number":9},{"id":"2f23cd1e63cbdb992ecfa50f06830b85569cca1c","author":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},"date":"2019-09-24 08:29:45.000000000","message":"Patch Set 9: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":9},{"id":"1353a5aebfb0db5875a176d2a33fdf191c2407ef","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-09-24 09:58:41.000000000","message":"Patch Set 9: Verified+1\n\nBuild succeeded (check pipeline).\n\n- kolla-build-centos-binary https://zuul.opendev.org/t/openstack/build/12ffea9130cf4be3b1863e74f2c4946d : SUCCESS in 43m 35s (non-voting)\n- kolla-build-centos-source https://zuul.opendev.org/t/openstack/build/78d2b86f58364eeb8a4a03ccc13eba47 : SUCCESS in 48m 24s\n- kolla-ansible-centos-source https://zuul.opendev.org/t/openstack/build/0f48d58d86bc423985610535bd39f6bc : SUCCESS in 1h 15m 31s\n- kolla-ansible-centos-binary https://zuul.opendev.org/t/openstack/build/cfcd31959ef148babf30c291433fdb73 : SUCCESS in 1h 12m 22s (non-voting)\n- kolla-ansible-centos-source-upgrade https://zuul.opendev.org/t/openstack/build/61ff43209c83404a8367de6cb8eb8a79 : SUCCESS in 1h 32m 38s\n- tripleo-build-containers-centos-7 https://zuul.opendev.org/t/openstack/build/fb12afa9c50745dbbb08dc6bbcc2912a : SUCCESS in 54m 10s\n- kolla-ansible-bifrost-centos-source https://zuul.opendev.org/t/openstack/build/3eeddbf9eb4c471ab365b30c037a1d88 : SUCCESS in 52m 14s (non-voting)\n- kolla-ansible-centos-source-ceph https://zuul.opendev.org/t/openstack/build/d118f3da1e81450aab59f2cad4911537 : SUCCESS in 2h 14m 03s (non-voting)\n- kolla-ansible-centos-source-upgrade-ceph https://zuul.opendev.org/t/openstack/build/99c629583758455fa2dc837083f0660f : TIMED_OUT in 2h 34m 20s (non-voting)\n- kolla-ansible-centos-source-cinder-lvm https://zuul.opendev.org/t/openstack/build/371944db10a846f18d7733a82e7e1769 : SUCCESS in 2h 00m 16s (non-voting)\n- kolla-ansible-centos-source-scenario-nfv https://zuul.opendev.org/t/openstack/build/fb5eeb15975643509a6ecdcb7f1f6670 : TIMED_OUT in 2h 03m 10s (non-voting)\n- kolla-ansible-centos-source-ironic https://zuul.opendev.org/t/openstack/build/4da4faddd9474458a8cf8e86739a3526 : SUCCESS in 1h 37m 38s (non-voting)\n- kolla-build-debian-source https://zuul.opendev.org/t/openstack/build/45bcc4cc69d34afcac804d5f57501eff : SUCCESS in 27m 21s (non-voting)\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/4542acd047514d64a13108948ccbba31 : SUCCESS in 4m 06s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/16dae409bc3647f1a37d3914e0a2016f : SUCCESS in 7m 28s\n- openstack-tox-py27 https://zuul.opendev.org/t/openstack/build/e47d57f7e7d946ce9945c1cb60e00003 : SUCCESS in 3m 29s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/a86d08f2afb94a00b975c0fed6ac8392 : SUCCESS in 5m 00s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/a64c0a0fab374a81bde2e3e5659d26e3 : SUCCESS in 3m 50s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/2ede3593fec54209b987390346c12311 : SUCCESS in 5m 50s\n- kolla-build-ubuntu-binary https://zuul.opendev.org/t/openstack/build/944e2540209b4b3a89c210d8e3bced16 : SUCCESS in 23m 13s (non-voting)\n- kolla-build-ubuntu-source https://zuul.opendev.org/t/openstack/build/5ea9704fb5364a80878b20116bb83650 : SUCCESS in 38m 51s\n- kolla-ansible-ubuntu-source https://zuul.opendev.org/t/openstack/build/35ad26260b38437f95596bec5e4e8ede : SUCCESS in 56m 40s\n- kolla-ansible-ubuntu-binary https://zuul.opendev.org/t/openstack/build/215f853bd5d543009cc5a1007b2ca6c4 : SUCCESS in 53m 03s (non-voting)\n- kolla-ansible-ubuntu-source-upgrade https://zuul.opendev.org/t/openstack/build/5152a5e34277407db8f57b1c5a23e1c4 : SUCCESS in 54m 46s\n- kolla-ansible-ubuntu-source-zun https://zuul.opendev.org/t/openstack/build/2370e4f5f4f74b5ebf114e8727f9ae74 : SUCCESS in 1h 18m 24s (non-voting)\n- kolla-ansible-ubuntu-source-ceph https://zuul.opendev.org/t/openstack/build/fbe360eab53a43a49be5ccdfb38e0533 : SUCCESS in 1h 36m 04s (non-voting)\n- kolla-ansible-ubuntu-source-upgrade-ceph https://zuul.opendev.org/t/openstack/build/4d4966f857fb4044afe9a6297c04ad46 : SUCCESS in 1h 55m 03s (non-voting)\n- kolla-ansible-ubuntu-source-cinder-lvm https://zuul.opendev.org/t/openstack/build/6d687cceb14e4c6fa39f183acb951342 : TIMED_OUT in 2h 02m 59s (non-voting)\n- kolla-ansible-ubuntu-source-ironic https://zuul.opendev.org/t/openstack/build/c5f0463065a34db2b14637ea98d15f97 : FAILURE in 6m 53s (non-voting)","accounts_in_message":[],"_revision_number":9},{"id":"744845bf786e060d573f60216220eef46646b5ef","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-09-24 09:58:54.000000000","message":"Patch Set 9: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":9},{"id":"c31fa723221e20acd6130820d55cae38c212f834","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-09-24 11:55:28.000000000","message":"Patch Set 9: Verified+2\n\nBuild succeeded (gate pipeline).\n\n- kolla-build-centos-source https://zuul.opendev.org/t/openstack/build/1008f7e781de45c0b797093eaa4f5ba5 : SUCCESS in 1h 04m 38s\n- kolla-ansible-centos-source https://zuul.opendev.org/t/openstack/build/bcbc7bdccee245dbbbdb61b2846c1ecb : SUCCESS in 1h 47m 46s\n- kolla-ansible-centos-source-upgrade https://zuul.opendev.org/t/openstack/build/26e586a11017481e8a12295a6bff67af : SUCCESS in 1h 46m 34s\n- tripleo-build-containers-centos-7 https://zuul.opendev.org/t/openstack/build/0b06d4b53a3a4d89ae8922e2e7e9b0ed : SUCCESS in 1h 32m 00s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/e22d743929f944e8bae2a23706d6372b : SUCCESS in 5m 26s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/387c23ed70d04baeb3546421b4a3bcb9 : SUCCESS in 4m 05s\n- openstack-tox-py27 https://zuul.opendev.org/t/openstack/build/fd5258138a5e46fa824bd27a59d18134 : SUCCESS in 4m 34s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/30b08c51135f4704b844e2ce8074b40c : SUCCESS in 3m 16s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/d0e6e1f5d9d34997bb683f64210db59c : SUCCESS in 3m 54s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/6b3ea60122f243a0acfaa35fdf65c63d : SUCCESS in 5m 29s\n- kolla-build-ubuntu-source https://zuul.opendev.org/t/openstack/build/5b106212800840f09994169f6e3960d2 : SUCCESS in 33m 26s\n- kolla-ansible-ubuntu-source https://zuul.opendev.org/t/openstack/build/1c3bafeb8e344085939b0afa8c6a879a : SUCCESS in 59m 49s\n- kolla-ansible-ubuntu-source-upgrade https://zuul.opendev.org/t/openstack/build/bc4a7a5ce2f8430bb5d09be15158fef3 : SUCCESS in 1h 03m 32s","accounts_in_message":[],"_revision_number":9},{"id":"b98c3f629a130d11f7adec58b088d6d7e85b5ed3","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-09-24 11:55:28.000000000","message":"Change has been successfully merged by Zuul","accounts_in_message":[],"_revision_number":9},{"id":"ca678de0bbff9f9ee98f0857fd4d13448a09275a","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-09-24 11:56:35.000000000","message":"Patch Set 9:\n\nBuild succeeded (promote pipeline).\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/6a0fec6afe7346e99179e7ce1948e142 : SUCCESS in 47s","accounts_in_message":[],"_revision_number":9}],"current_revision_number":9,"current_revision":"c08a4b5a6352f3d1ebf640687d68d00aabc11ff6","revisions":{"ada4184971b2786b055448ae5906a7a140ef0ffc":{"kind":"REWORK","_number":1,"created":"2019-09-20 18:14:28.000000000","uploader":{"_account_id":28374,"name":"Viktor Michalek","email":"viktor.michalek@ultimum.io","username":"viktormichalek"},"ref":"refs/changes/74/683674/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/kolla","ref":"refs/changes/74/683674/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/kolla refs/changes/74/683674/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/kolla refs/changes/74/683674/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/kolla refs/changes/74/683674/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/kolla refs/changes/74/683674/1"}}},"commit":{"parents":[{"commit":"29e4cb118ae14f300717c0a80c82ddf37cffa385","subject":"Merge \"Fix retries when pushing images\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/29e4cb118ae14f300717c0a80c82ddf37cffa385"}]}],"author":{"name":"Viktor Michalek","email":"viktor.michalek@ultimum.io","date":"2019-09-20 16:20:38.000000000","tz":120},"committer":{"name":"Viktor Michalek","email":"viktor.michalek@ultimum.io","date":"2019-09-20 16:20:38.000000000","tz":120},"subject":"Debian Buster Stein backports repository","message":"Debian Buster Stein backports repository\n\nChange-Id: Ie140cf0d6308a5ee3c09f06ddfcf63be2e1cec64\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/ada4184971b2786b055448ae5906a7a140ef0ffc"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/ada4184971b2786b055448ae5906a7a140ef0ffc"}]},"branch":"refs/heads/master"},"948d23578212b716399df2781bfe3b02e283878e":{"kind":"NO_CODE_CHANGE","_number":2,"created":"2019-09-22 16:28:44.000000000","uploader":{"_account_id":28374,"name":"Viktor Michalek","email":"viktor.michalek@ultimum.io","username":"viktormichalek"},"ref":"refs/changes/74/683674/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/kolla","ref":"refs/changes/74/683674/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/kolla refs/changes/74/683674/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/kolla refs/changes/74/683674/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/kolla refs/changes/74/683674/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/kolla refs/changes/74/683674/2"}}},"commit":{"parents":[{"commit":"29e4cb118ae14f300717c0a80c82ddf37cffa385","subject":"Merge \"Fix retries when pushing images\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/29e4cb118ae14f300717c0a80c82ddf37cffa385"}]}],"author":{"name":"Viktor Michalek","email":"viktor.michalek@ultimum.io","date":"2019-09-20 16:20:38.000000000","tz":120},"committer":{"name":"Viktor Michalek","email":"viktor.michalek@ultimum.io","date":"2019-09-22 16:28:35.000000000","tz":0},"subject":"Debian Buster Stein backports repository","message":"Debian Buster Stein backports repository\n\nDepends-On: https://review.opendev.org/#/c/683675\nChange-Id: Ie140cf0d6308a5ee3c09f06ddfcf63be2e1cec64\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/948d23578212b716399df2781bfe3b02e283878e"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/948d23578212b716399df2781bfe3b02e283878e"}]},"branch":"refs/heads/master"},"1cf43d028dc6ceddf204074234edf0bc81029fd3":{"kind":"NO_CODE_CHANGE","_number":3,"created":"2019-09-22 16:31:00.000000000","uploader":{"_account_id":28374,"name":"Viktor Michalek","email":"viktor.michalek@ultimum.io","username":"viktormichalek"},"ref":"refs/changes/74/683674/3","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/kolla","ref":"refs/changes/74/683674/3","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/kolla refs/changes/74/683674/3 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/kolla refs/changes/74/683674/3 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/kolla refs/changes/74/683674/3 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/kolla refs/changes/74/683674/3"}}},"commit":{"parents":[{"commit":"29e4cb118ae14f300717c0a80c82ddf37cffa385","subject":"Merge \"Fix retries when pushing images\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/29e4cb118ae14f300717c0a80c82ddf37cffa385"}]}],"author":{"name":"Viktor Michalek","email":"viktor.michalek@ultimum.io","date":"2019-09-20 16:20:38.000000000","tz":120},"committer":{"name":"Viktor Michalek","email":"viktor.michalek@ultimum.io","date":"2019-09-22 16:30:55.000000000","tz":0},"subject":"Debian Buster Stein backports repository","message":"Debian Buster Stein backports repository\n\nChange-Id: Ie140cf0d6308a5ee3c09f06ddfcf63be2e1cec64\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/1cf43d028dc6ceddf204074234edf0bc81029fd3"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/1cf43d028dc6ceddf204074234edf0bc81029fd3"}]},"branch":"refs/heads/master"},"93e5e91e22378bbd1df6f3e9b96efc9df53604c0":{"kind":"NO_CODE_CHANGE","_number":4,"created":"2019-09-22 18:03:03.000000000","uploader":{"_account_id":28374,"name":"Viktor Michalek","email":"viktor.michalek@ultimum.io","username":"viktormichalek"},"ref":"refs/changes/74/683674/4","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/kolla","ref":"refs/changes/74/683674/4","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/kolla refs/changes/74/683674/4 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/kolla refs/changes/74/683674/4 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/kolla refs/changes/74/683674/4 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/kolla refs/changes/74/683674/4"}}},"commit":{"parents":[{"commit":"29e4cb118ae14f300717c0a80c82ddf37cffa385","subject":"Merge \"Fix retries when pushing images\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/29e4cb118ae14f300717c0a80c82ddf37cffa385"}]}],"author":{"name":"Viktor Michalek","email":"viktor.michalek@ultimum.io","date":"2019-09-20 16:20:38.000000000","tz":120},"committer":{"name":"Viktor Michalek","email":"viktor.michalek@ultimum.io","date":"2019-09-22 18:02:55.000000000","tz":0},"subject":"Debian Buster Stein backports repository","message":"Debian Buster Stein backports repository\n\nDepends-On: https://review.opendev.org/#/c/683681\nChange-Id: Ie140cf0d6308a5ee3c09f06ddfcf63be2e1cec64\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/93e5e91e22378bbd1df6f3e9b96efc9df53604c0"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/93e5e91e22378bbd1df6f3e9b96efc9df53604c0"}]},"branch":"refs/heads/master"},"2865e4cf252c5ea2702a055e78a7278a83647a29":{"kind":"REWORK","_number":5,"created":"2019-09-23 17:38:48.000000000","uploader":{"_account_id":28374,"name":"Viktor Michalek","email":"viktor.michalek@ultimum.io","username":"viktormichalek"},"ref":"refs/changes/74/683674/5","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/kolla","ref":"refs/changes/74/683674/5","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/kolla refs/changes/74/683674/5 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/kolla refs/changes/74/683674/5 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/kolla refs/changes/74/683674/5 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/kolla refs/changes/74/683674/5"}}},"commit":{"parents":[{"commit":"29e4cb118ae14f300717c0a80c82ddf37cffa385","subject":"Merge \"Fix retries when pushing images\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/29e4cb118ae14f300717c0a80c82ddf37cffa385"}]}],"author":{"name":"Viktor Michalek","email":"viktor.michalek@ultimum.io","date":"2019-09-20 16:20:38.000000000","tz":120},"committer":{"name":"Viktor Michalek","email":"viktor.michalek@ultimum.io","date":"2019-09-23 17:32:58.000000000","tz":120},"subject":"Debian Buster Stein backports repository","message":"Debian Buster Stein backports repository\n\nDepends-On: https://review.opendev.org/#/c/683681\nChange-Id: Ie140cf0d6308a5ee3c09f06ddfcf63be2e1cec64\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/2865e4cf252c5ea2702a055e78a7278a83647a29"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/2865e4cf252c5ea2702a055e78a7278a83647a29"}]},"branch":"refs/heads/master"},"7ec87711186fca18b568fd84a18b8b0273d10144":{"kind":"NO_CODE_CHANGE","_number":6,"created":"2019-09-23 17:41:04.000000000","uploader":{"_account_id":28374,"name":"Viktor Michalek","email":"viktor.michalek@ultimum.io","username":"viktormichalek"},"ref":"refs/changes/74/683674/6","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/kolla","ref":"refs/changes/74/683674/6","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/kolla refs/changes/74/683674/6 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/kolla refs/changes/74/683674/6 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/kolla refs/changes/74/683674/6 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/kolla refs/changes/74/683674/6"}}},"commit":{"parents":[{"commit":"29e4cb118ae14f300717c0a80c82ddf37cffa385","subject":"Merge \"Fix retries when pushing images\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/29e4cb118ae14f300717c0a80c82ddf37cffa385"}]}],"author":{"name":"Viktor Michalek","email":"viktor.michalek@ultimum.io","date":"2019-09-20 16:20:38.000000000","tz":120},"committer":{"name":"Viktor Michalek","email":"viktor.michalek@ultimum.io","date":"2019-09-23 17:39:51.000000000","tz":120},"subject":"Debian Buster Stein backports repository","message":"Debian Buster Stein backports repository\n\nDepends-On: https://review.opendev.org/683681\nChange-Id: Ie140cf0d6308a5ee3c09f06ddfcf63be2e1cec64\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/7ec87711186fca18b568fd84a18b8b0273d10144"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/7ec87711186fca18b568fd84a18b8b0273d10144"}]},"branch":"refs/heads/master"},"57f4a4e32fc853d7bbe52172e0c5224b36c23417":{"kind":"NO_CODE_CHANGE","_number":7,"created":"2019-09-24 07:15:19.000000000","uploader":{"_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"},"ref":"refs/changes/74/683674/7","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/kolla","ref":"refs/changes/74/683674/7","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/kolla refs/changes/74/683674/7 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/kolla refs/changes/74/683674/7 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/kolla refs/changes/74/683674/7 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/kolla refs/changes/74/683674/7"}}},"commit":{"parents":[{"commit":"29e4cb118ae14f300717c0a80c82ddf37cffa385","subject":"Merge \"Fix retries when pushing images\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/29e4cb118ae14f300717c0a80c82ddf37cffa385"}]}],"author":{"name":"Viktor Michalek","email":"viktor.michalek@ultimum.io","date":"2019-09-20 16:20:38.000000000","tz":120},"committer":{"name":"Radosław Piliszek","email":"radoslaw.piliszek@gmail.com","date":"2019-09-24 07:15:13.000000000","tz":0},"subject":"Add Debian Buster Stein backports APT repository","message":"Add Debian Buster Stein backports APT repository\n\nChange-Id: Ie140cf0d6308a5ee3c09f06ddfcf63be2e1cec64\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/57f4a4e32fc853d7bbe52172e0c5224b36c23417"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/57f4a4e32fc853d7bbe52172e0c5224b36c23417"}]},"branch":"refs/heads/master"},"31f24734825fe6c3cb6e73b514fe8bd44f411e66":{"kind":"TRIVIAL_REBASE","_number":8,"created":"2019-09-24 07:15:33.000000000","uploader":{"_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"},"ref":"refs/changes/74/683674/8","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/kolla","ref":"refs/changes/74/683674/8","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/kolla refs/changes/74/683674/8 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/kolla refs/changes/74/683674/8 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/kolla refs/changes/74/683674/8 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/kolla refs/changes/74/683674/8"}}},"commit":{"parents":[{"commit":"e3b2bb78084866a4b2c8966bacfd8ce47161dc18","subject":"Merge \"CI: Use docker mirror proxy\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/e3b2bb78084866a4b2c8966bacfd8ce47161dc18"}]}],"author":{"name":"Viktor Michalek","email":"viktor.michalek@ultimum.io","date":"2019-09-20 16:20:38.000000000","tz":120},"committer":{"name":"Radosław Piliszek","email":"radoslaw.piliszek@gmail.com","date":"2019-09-24 07:15:33.000000000","tz":0},"subject":"Add Debian Buster Stein backports APT repository","message":"Add Debian Buster Stein backports APT repository\n\nChange-Id: Ie140cf0d6308a5ee3c09f06ddfcf63be2e1cec64\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/31f24734825fe6c3cb6e73b514fe8bd44f411e66"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/31f24734825fe6c3cb6e73b514fe8bd44f411e66"}]},"branch":"refs/heads/master"},"c08a4b5a6352f3d1ebf640687d68d00aabc11ff6":{"kind":"REWORK","_number":9,"created":"2019-09-24 07:16:09.000000000","uploader":{"_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"},"ref":"refs/changes/74/683674/9","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/kolla","ref":"refs/changes/74/683674/9","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/kolla refs/changes/74/683674/9 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/kolla refs/changes/74/683674/9 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/kolla refs/changes/74/683674/9 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/kolla refs/changes/74/683674/9"}}},"commit":{"parents":[{"commit":"e3b2bb78084866a4b2c8966bacfd8ce47161dc18","subject":"Merge \"CI: Use docker mirror proxy\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/e3b2bb78084866a4b2c8966bacfd8ce47161dc18"}]}],"author":{"name":"Viktor Michalek","email":"viktor.michalek@ultimum.io","date":"2019-09-20 16:20:38.000000000","tz":120},"committer":{"name":"Radosław Piliszek","email":"radoslaw.piliszek@gmail.com","date":"2019-09-24 07:16:04.000000000","tz":0},"subject":"Add Debian Buster Stein backports APT repository","message":"Add Debian Buster Stein backports APT repository\n\nChange-Id: Ie140cf0d6308a5ee3c09f06ddfcf63be2e1cec64\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/c08a4b5a6352f3d1ebf640687d68d00aabc11ff6"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/c08a4b5a6352f3d1ebf640687d68d00aabc11ff6"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[],"submit_requirements":[]}
