)]}'
{"id":"openstack%2Fkolla~683676","triplet_id":"openstack%2Fkolla~master~Ibd38101eb1ce0085d4ccb877dea7ecae48197563","project":"openstack/kolla","branch":"master","topic":"680447","hashtags":[],"change_id":"Ibd38101eb1ce0085d4ccb877dea7ecae48197563","subject":"Fluentd: fix missing /var/run/fluentd","status":"ABANDONED","created":"2019-09-20 18:14:28.000000000","updated":"2019-09-27 18:47:33.000000000","total_comment_count":1,"unresolved_comment_count":0,"has_review_started":true,"meta_rev_id":"014b9e71f680def38ea2a44917e10b8c15bfe3c4","_number":683676,"virtual_id_number":683676,"owner":{"_account_id":28374,"name":"Viktor Michalek","email":"viktor.michalek@ultimum.io","username":"viktormichalek"},"actions":{},"labels":{"Verified":{"disliked":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},{"_account_id":24072,"name":"Marcin Juszkiewicz","email":"mjuszkiewicz@redhat.com","username":"hrw"},{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},{"_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"},{"tag":"autogenerated:zuul:check","value":-1,"date":"2019-09-27 18:38:12.000000000","permitted_voting_range":{"min":-2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"date":"2019-09-27 17:28:48.000000000","_account_id":23181,"name":"RDO Third Party CI","email":"dmsimard+rdothirdparty@redhat.com","username":"rdothirdparty","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","value":-1,"default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":24072,"name":"Marcin Juszkiewicz","email":"mjuszkiewicz@redhat.com","username":"hrw"},"all":[{"value":0,"permitted_voting_range":{"min":-2,"max":2},"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},{"value":2,"date":"2019-09-27 16:47:27.000000000","permitted_voting_range":{"min":-1,"max":1},"_account_id":24072,"name":"Marcin Juszkiewicz","email":"mjuszkiewicz@redhat.com","username":"hrw"},{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},{"value":-2,"date":"2019-09-27 16:47:27.000000000","permitted_voting_range":{"min":-1,"max":1},"_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"},{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":23181,"name":"RDO Third Party CI","email":"dmsimard+rdothirdparty@redhat.com","username":"rdothirdparty","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"all":[{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},{"_account_id":24072,"name":"Marcin Juszkiewicz","email":"mjuszkiewicz@redhat.com","username":"hrw"},{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},{"_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":23181,"name":"RDO Third Party CI","email":"dmsimard+rdothirdparty@redhat.com","username":"rdothirdparty","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Backport-Candidate":{"all":[{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},{"_account_id":24072,"name":"Marcin Juszkiewicz","email":"mjuszkiewicz@redhat.com","username":"hrw"},{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},{"_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":23181,"name":"RDO Third Party CI","email":"dmsimard+rdothirdparty@redhat.com","username":"rdothirdparty","tags":["SERVICE_USER"]}],"values":{"-1":"Do Not Backport"," 0":"Backport Review Needed","+1":"Should Backport"},"description":"","default_value":0,"optional":true},"Review-Priority":{"all":[{"value":0,"date":"2019-09-27 16:47:27.000000000","permitted_voting_range":{"min":-1,"max":2},"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},{"_account_id":24072,"name":"Marcin Juszkiewicz","email":"mjuszkiewicz@redhat.com","username":"hrw"},{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},{"_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":23181,"name":"RDO Third Party CI","email":"dmsimard+rdothirdparty@redhat.com","username":"rdothirdparty","tags":["SERVICE_USER"]}],"values":{"-1":"Branch Freeze"," 0":"No Priority","+1":"Important Change","+2":"Gate Blocker Fix / Urgent Change"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},{"_account_id":23181,"name":"RDO Third Party CI","email":"dmsimard+rdothirdparty@redhat.com","username":"rdothirdparty","tags":["SERVICE_USER"]},{"_account_id":24072,"name":"Marcin Juszkiewicz","email":"mjuszkiewicz@redhat.com","username":"hrw"},{"_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2019-09-20 20:47:57.000000000","updated_by":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"reviewer":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"state":"REVIEWER"},{"updated":"2019-09-20 20:48:04.000000000","updated_by":{"_account_id":24072,"name":"Marcin Juszkiewicz","email":"mjuszkiewicz@redhat.com","username":"hrw"},"reviewer":{"_account_id":24072,"name":"Marcin Juszkiewicz","email":"mjuszkiewicz@redhat.com","username":"hrw"},"state":"REVIEWER"},{"updated":"2019-09-23 09:12:09.000000000","updated_by":{"_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"},"reviewer":{"_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"},"state":"REVIEWER"},{"updated":"2019-09-24 08:32:09.000000000","updated_by":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},"reviewer":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},"state":"REVIEWER"},{"updated":"2019-09-27 17:28:48.000000000","updated_by":{"_account_id":23181,"name":"RDO Third Party CI","email":"dmsimard+rdothirdparty@redhat.com","username":"rdothirdparty","tags":["SERVICE_USER"]},"reviewer":{"_account_id":23181,"name":"RDO Third Party CI","email":"dmsimard+rdothirdparty@redhat.com","username":"rdothirdparty","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2019-09-27 18:38:12.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"}],"messages":[{"id":"4e229ee7e06f9528889c1777cdd79d6cc51fb75d","author":{"_account_id":28374,"name":"Viktor Michalek","email":"viktor.michalek@ultimum.io","username":"viktormichalek"},"date":"2019-09-20 18:14:28.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"d1a86c6d37559d54737ab30222afd0ddb18ed923","author":{"_account_id":23181,"name":"RDO Third Party CI","email":"dmsimard+rdothirdparty@redhat.com","username":"rdothirdparty","tags":["SERVICE_USER"]},"date":"2019-09-20 18:53:12.000000000","message":"Patch Set 1:\n\nBuild succeeded (check pipeline).\n\n- tripleo-build-containers-rhel-8-master http://logs.rdoproject.org/76/683676/1/openstack-check/tripleo-build-containers-rhel-8-master/1ffb0c4/ : SUCCESS in 37m 03s","accounts_in_message":[],"_revision_number":1},{"id":"077d079fcef658b73a98a41366938a3993e2cae5","author":{"_account_id":24072,"name":"Marcin Juszkiewicz","email":"mjuszkiewicz@redhat.com","username":"hrw"},"date":"2019-09-20 20:48:04.000000000","message":"Patch Set 1: Code-Review+2","accounts_in_message":[],"_revision_number":1},{"id":"44526a9dfa86f5ccf558083b4c548866f2333823","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-09-20 21:45:21.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\n\n- kolla-build-centos-binary https://zuul.opendev.org/t/openstack/build/f081f1613b0e44f5a698d5cd8e4a7e61 : SUCCESS in 1h 01m 59s (non-voting)\n- kolla-build-centos-source https://zuul.opendev.org/t/openstack/build/2ba9d8ccd65246d285a1cf639d661adf : SUCCESS in 51m 04s\n- kolla-ansible-centos-source https://zuul.opendev.org/t/openstack/build/4f001a2153c74f27a6baa425fd157951 : SUCCESS in 1h 31m 15s\n- kolla-ansible-centos-binary https://zuul.opendev.org/t/openstack/build/f553f451760b4a9a9882b7a174eaf815 : SUCCESS in 1h 29m 29s (non-voting)\n- kolla-ansible-centos-source-upgrade https://zuul.opendev.org/t/openstack/build/c2f9042d5b5a45fca674f97cc5884eb2 : SUCCESS in 1h 35m 52s\n- tripleo-build-containers-centos-7 https://zuul.opendev.org/t/openstack/build/5e91c69b9cce4b648aef3bc7f03fb73f : SUCCESS in 56m 20s\n- kolla-build-debian-source https://zuul.opendev.org/t/openstack/build/d7488dfcd403452ead72b0e56b46ecb9 : SUCCESS in 26m 54s (non-voting)\n- kolla-build-debian-binary https://zuul.opendev.org/t/openstack/build/17ad277a6d214ebb8ad20a25178b3f1b : FAILURE in 17m 35s (non-voting)\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/78a5d1e0437640b8a808d7c09da8c34d : SUCCESS in 5m 02s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/44202fbbeeca4592b5b882354fed6aa2 : SUCCESS in 4m 41s\n- openstack-tox-py27 https://zuul.opendev.org/t/openstack/build/dbc6ac044ccc4295ab8c07afc18bdb4e : SUCCESS in 3m 21s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/b4bae1bc0739475a933012253143ae98 : SUCCESS in 3m 24s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/56cf22c2fbd0407b94b14f975d0a127a : SUCCESS in 3m 20s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/80514b965a754fd5ae201e71d2a8d6e8 : SUCCESS in 5m 56s\n- kolla-build-ubuntu-binary https://zuul.opendev.org/t/openstack/build/e2335e16c53d4b4991985736f38b5394 : SUCCESS in 38m 41s (non-voting)\n- kolla-build-ubuntu-source https://zuul.opendev.org/t/openstack/build/b45ff0aff7844f6892ed6d5e1551772f : SUCCESS in 48m 52s\n- kolla-ansible-ubuntu-source https://zuul.opendev.org/t/openstack/build/d9b9f3f98e9e4461b0ab40f0e145c702 : SUCCESS in 58m 30s\n- kolla-ansible-ubuntu-binary https://zuul.opendev.org/t/openstack/build/a6bc7f76f7e5415a8637b5800fc7bcf8 : SUCCESS in 59m 33s (non-voting)\n- kolla-ansible-ubuntu-source-upgrade https://zuul.opendev.org/t/openstack/build/0c5fd0a22f9f4ae4b9aab9e5cb95401e : SUCCESS in 1h 07m 03s","accounts_in_message":[],"_revision_number":1},{"id":"06267efdb21496dbf76da9ff1c9dae187bf76a30","author":{"_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"},"date":"2019-09-23 09:12:09.000000000","message":"Patch Set 1: Code-Review-2\n\nIt\u0027s not missing. td-agent simply uses td-agent","accounts_in_message":[],"_revision_number":1},{"id":"6ef35b3146f69ea1c6a2905291b24cb63c7eaefe","author":{"_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"},"date":"2019-09-23 09:13:41.000000000","message":"Patch Set 1:\n\n(1 comment)","accounts_in_message":[],"_revision_number":1},{"id":"1937e95023f15a4ecfa6b4c393e92316e9e9ab80","author":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},"date":"2019-09-24 08:32:09.000000000","message":"Patch Set 1:\n\nHere\u0027s a related bug: https://bugs.launchpad.net/kolla/+bug/1844574","accounts_in_message":[],"_revision_number":1},{"id":"4ea9f75892b2eee9f124ce21aa0d1f25360466c4","author":{"_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"},"date":"2019-09-24 08:43:30.000000000","message":"Patch Set 1:\n\nYeah, we fixed only Ubuntu before.","accounts_in_message":[],"_revision_number":1},{"id":"2dbcc5b1990265d9cd8bb194c7f970db0adcf672","author":{"_account_id":28374,"name":"Viktor Michalek","email":"viktor.michalek@ultimum.io","username":"viktormichalek"},"date":"2019-09-24 14:05:24.000000000","message":"Patch Set 2: Patch Set 1 was rebased","accounts_in_message":[],"_revision_number":2},{"id":"928242ba4d23b6065e6a4b58b4cc3d10ca354077","author":{"_account_id":23181,"name":"RDO Third Party CI","email":"dmsimard+rdothirdparty@redhat.com","username":"rdothirdparty","tags":["SERVICE_USER"]},"date":"2019-09-24 15:12:06.000000000","message":"Patch Set 2:\n\nBuild succeeded (check pipeline).\n\n- tripleo-build-containers-rhel-8-master http://logs.rdoproject.org/76/683676/2/openstack-check/tripleo-build-containers-rhel-8-master/624b56c/ : SUCCESS in 46m 48s","accounts_in_message":[],"_revision_number":2},{"id":"bc20c501fa4ac4ace567fc6be73d3174b6440862","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-09-24 16:15:00.000000000","message":"Patch Set 2: Verified+1\n\nBuild succeeded (check pipeline).\n\n- kolla-build-centos-binary https://zuul.opendev.org/t/openstack/build/a169d76be56e47f692056793c0cf0d9c : SUCCESS in 45m 38s (non-voting)\n- kolla-build-centos-source https://zuul.opendev.org/t/openstack/build/f1042cc7ab8d4b1ea5694fab7cf12247 : SUCCESS in 56m 08s\n- kolla-ansible-centos-source https://zuul.opendev.org/t/openstack/build/023fdf74224e4ee5a0be7ab1d4c7ed4e : SUCCESS in 1h 45m 03s\n- kolla-ansible-centos-binary https://zuul.opendev.org/t/openstack/build/9fd4a3fed29046e7a60de4c4f20e09a7 : SUCCESS in 1h 16m 11s (non-voting)\n- kolla-ansible-centos-source-upgrade https://zuul.opendev.org/t/openstack/build/a285c0ee83c44feb8794a3e35a40f6e5 : SUCCESS in 1h 29m 27s\n- tripleo-build-containers-centos-7 https://zuul.opendev.org/t/openstack/build/d5979fd242b7409fb6eab006fc247759 : SUCCESS in 46m 37s\n- kolla-build-debian-source https://zuul.opendev.org/t/openstack/build/db9ca3362a1f47a8b3b20d77d1592787 : SUCCESS in 22m 04s (non-voting)\n- kolla-build-debian-binary https://zuul.opendev.org/t/openstack/build/67d1ce2dddfa4349abe06ece58be2384 : FAILURE in 18m 51s (non-voting)\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/b46b6e8cbee04f2384044d57389dc093 : SUCCESS in 4m 04s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/f1d27cc2585c42a1971b72756c5f3505 : SUCCESS in 4m 07s\n- openstack-tox-py27 https://zuul.opendev.org/t/openstack/build/8dcf17b371cc43b49f2977feb15e2c6d : SUCCESS in 4m 38s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/1eb401836a714d269fe9c5b4080ac9fb : SUCCESS in 3m 02s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/45537dead46e4c8287c2de5a8530e8d7 : SUCCESS in 6m 50s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/d3eadd5c8f7b4c3c9f6057ae6cabcf8f : SUCCESS in 5m 35s\n- kolla-build-ubuntu-binary https://zuul.opendev.org/t/openstack/build/0bb2978cfe4642da86b51888b8d166bb : SUCCESS in 25m 03s (non-voting)\n- kolla-build-ubuntu-source https://zuul.opendev.org/t/openstack/build/8d81a3605ef74807a39e632271a77b59 : SUCCESS in 43m 44s\n- kolla-ansible-ubuntu-source https://zuul.opendev.org/t/openstack/build/f4f8106b2f9e457fb906e9bc156c66a0 : SUCCESS in 55m 34s\n- kolla-ansible-ubuntu-binary https://zuul.opendev.org/t/openstack/build/22bcf244b66c49c7ad3434f0093d6b4b : SUCCESS in 58m 28s (non-voting)\n- kolla-ansible-ubuntu-source-upgrade https://zuul.opendev.org/t/openstack/build/88852af7b7314701941cb681b25773b3 : SUCCESS in 1h 07m 10s","accounts_in_message":[],"_revision_number":2},{"id":"cf6c71146d76523d44f14f0ecc8e7a667437da1e","author":{"_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"},"date":"2019-09-25 07:34:26.000000000","message":"Patch Set 2:\n\nThe fluentd/td-agent mismatch is long fixed.","accounts_in_message":[],"_revision_number":2},{"id":"040b86bef9103422a27eef4b8d6c674e0e9d3649","author":{"_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"},"date":"2019-09-25 07:39:22.000000000","message":"Patch Set 2:\n\nDebian uses fluentd currently, not td-agent - this cannot have happened.","accounts_in_message":[],"_revision_number":2},{"id":"771c5a7dab718971fcd5eb8af2adce5bb79f2630","author":{"_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"},"date":"2019-09-25 07:57:50.000000000","message":"Patch Set 2:\n\nProposed the switch: https://review.opendev.org/684551","accounts_in_message":[],"_revision_number":2},{"id":"f2652d4adcb0e746037edd51dcac5dcffe64055a","author":{"_account_id":24072,"name":"Marcin Juszkiewicz","email":"mjuszkiewicz@redhat.com","username":"hrw"},"date":"2019-09-27 16:47:27.000000000","message":"Uploaded patch set 3: Patch Set 2 was rebased.","accounts_in_message":[],"_revision_number":3},{"id":"6701fe292e0a8c9a9d08732b07f40318a5797bdd","author":{"_account_id":23181,"name":"RDO Third Party CI","email":"dmsimard+rdothirdparty@redhat.com","username":"rdothirdparty","tags":["SERVICE_USER"]},"date":"2019-09-27 17:28:48.000000000","message":"Patch Set 3:\n\nBuild succeeded (check pipeline).\n\n- tripleo-build-containers-rhel-8-master http://logs.rdoproject.org/76/683676/3/openstack-check/tripleo-build-containers-rhel-8-master/ebf2747/ : SUCCESS in 40m 13s","accounts_in_message":[],"_revision_number":3},{"id":"0f243a91d12bd2da962cdc33716c250dc825a786","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-09-27 18:38:12.000000000","message":"Patch Set 3: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttp://docs.openstack.org/infra/manual/developers.html#automated-testing\n\n\n- kolla-build-centos-binary https://zuul.opendev.org/t/openstack/build/9fa93c3a51a44ab0a3291dc56fbc4d29 : FAILURE in 52m 56s (non-voting)\n- kolla-build-centos-source https://zuul.opendev.org/t/openstack/build/1e346f96da50414aa808ecc98f216187 : FAILURE in 56m 24s\n- kolla-ansible-centos-source https://zuul.opendev.org/t/openstack/build/7f63f531bf1343f4883e8c228383373c : SUCCESS in 1h 43m 40s\n- kolla-ansible-centos-binary https://zuul.opendev.org/t/openstack/build/45d22b5fe0d54403932b64fb89146c74 : SUCCESS in 1h 13m 26s (non-voting)\n- kolla-ansible-centos-source-upgrade https://zuul.opendev.org/t/openstack/build/78ca510d22c3493e9e1a9a7e25c783a2 : SUCCESS in 1h 27m 36s\n- tripleo-build-containers-centos-7 https://zuul.opendev.org/t/openstack/build/751f87ae1c0e422097fd586003a449d4 : SUCCESS in 1h 02m 31s\n- kolla-build-debian-source https://zuul.opendev.org/t/openstack/build/1dac121851734ccdba9465e9fadb9eb3 : SUCCESS in 32m 52s (non-voting)\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/f89b369aaa9d42089fdee82c71b3d501 : SUCCESS in 3m 41s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/f92c8eb60d5041a4b6bd8ed7100822bd : SUCCESS in 4m 46s\n- openstack-tox-py27 https://zuul.opendev.org/t/openstack/build/b110facdfa2e4a2d89e51e68631e9a09 : SUCCESS in 3m 20s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/fbd5b47eb6084d9ead2e17c017bbdacc : SUCCESS in 4m 45s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/81d0efbde0cc44cb8d45c93984c76157 : SUCCESS in 3m 30s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/40d47b6a38404a8b867b4bbcec84d394 : SUCCESS in 4m 51s\n- kolla-build-ubuntu-binary https://zuul.opendev.org/t/openstack/build/cb2f5bd169d747cd855cc98653c5b982 : SUCCESS in 25m 29s (non-voting)\n- kolla-build-ubuntu-source https://zuul.opendev.org/t/openstack/build/1bd7f1630dfb4ae98a834a45e36d8736 : SUCCESS in 30m 39s\n- kolla-ansible-ubuntu-source https://zuul.opendev.org/t/openstack/build/35287f1c2b3f431ab92afc4548726a9c : SUCCESS in 58m 30s\n- kolla-ansible-ubuntu-binary https://zuul.opendev.org/t/openstack/build/bee8b9167ac94c9790a64125351ea6cd : SUCCESS in 58m 01s (non-voting)\n- kolla-ansible-ubuntu-source-upgrade https://zuul.opendev.org/t/openstack/build/d0bd2012eae34da091d798bc4bf7d6d4 : SUCCESS in 1h 10m 05s","accounts_in_message":[],"_revision_number":3},{"id":"69f8ad6fd8af11839f853110b78ee72aa7c13c9f","author":{"_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"},"date":"2019-09-27 18:47:33.000000000","message":"Abandoned\n\nstop carrying it around","accounts_in_message":[],"_revision_number":3}],"current_revision_number":3,"current_revision":"7bb2f9bc391acd1c7ec02f8bfebb2bf634fda5a2","revisions":{"a7abc458a9322dd2aa49771ba67eb53f14a8463e":{"kind":"REWORK","_number":1,"created":"2019-09-20 18:14:28.000000000","uploader":{"_account_id":28374,"name":"Viktor Michalek","email":"viktor.michalek@ultimum.io","username":"viktormichalek"},"ref":"refs/changes/76/683676/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/kolla","ref":"refs/changes/76/683676/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/kolla refs/changes/76/683676/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/kolla refs/changes/76/683676/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/kolla refs/changes/76/683676/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/kolla refs/changes/76/683676/1"}}},"commit":{"parents":[{"commit":"db5b5246702a18d29bd32e9a96ad8e12f63b02cf","subject":"Debian/binary CI: added kolla-build-debian-binary","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/db5b5246702a18d29bd32e9a96ad8e12f63b02cf"}]}],"author":{"name":"Viktor Michalek","email":"viktor.michalek@ultimum.io","date":"2019-09-20 16:33:29.000000000","tz":120},"committer":{"name":"Viktor Michalek","email":"viktor.michalek@ultimum.io","date":"2019-09-20 16:33:29.000000000","tz":120},"subject":"Fluentd: fix missing /var/run/fluentd","message":"Fluentd: fix missing /var/run/fluentd\n\nChange-Id: Ibd38101eb1ce0085d4ccb877dea7ecae48197563\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/a7abc458a9322dd2aa49771ba67eb53f14a8463e"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/a7abc458a9322dd2aa49771ba67eb53f14a8463e"}]},"branch":"refs/heads/master"},"543b53ed2512cfb39d1ac234a37a067d3f95e063":{"kind":"TRIVIAL_REBASE","_number":2,"created":"2019-09-24 14:05:24.000000000","uploader":{"_account_id":28374,"name":"Viktor Michalek","email":"viktor.michalek@ultimum.io","username":"viktormichalek"},"ref":"refs/changes/76/683676/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/kolla","ref":"refs/changes/76/683676/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/kolla refs/changes/76/683676/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/kolla refs/changes/76/683676/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/kolla refs/changes/76/683676/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/kolla refs/changes/76/683676/2"}}},"commit":{"parents":[{"commit":"ba7e3f77d493fcdfb5442b56d962c0aa4b659d04","subject":"Debian/binary CI: added kolla-build-debian-binary","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/ba7e3f77d493fcdfb5442b56d962c0aa4b659d04"}]}],"author":{"name":"Viktor Michalek","email":"viktor.michalek@ultimum.io","date":"2019-09-20 16:33:29.000000000","tz":120},"committer":{"name":"Viktor Michalek","email":"viktor.michalek@ultimum.io","date":"2019-09-24 14:05:24.000000000","tz":0},"subject":"Fluentd: fix missing /var/run/fluentd","message":"Fluentd: fix missing /var/run/fluentd\n\nChange-Id: Ibd38101eb1ce0085d4ccb877dea7ecae48197563\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/543b53ed2512cfb39d1ac234a37a067d3f95e063"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/543b53ed2512cfb39d1ac234a37a067d3f95e063"}]},"branch":"refs/heads/master"},"7bb2f9bc391acd1c7ec02f8bfebb2bf634fda5a2":{"kind":"TRIVIAL_REBASE","_number":3,"created":"2019-09-27 16:47:27.000000000","uploader":{"_account_id":24072,"name":"Marcin Juszkiewicz","email":"mjuszkiewicz@redhat.com","username":"hrw"},"ref":"refs/changes/76/683676/3","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/kolla","ref":"refs/changes/76/683676/3","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/kolla refs/changes/76/683676/3 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/kolla refs/changes/76/683676/3 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/kolla refs/changes/76/683676/3 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/kolla refs/changes/76/683676/3"}}},"commit":{"parents":[{"commit":"f9b229841db5d27fe11b95d055d74af289893ef2","subject":"Merge \"ppc64le: Ensure that rabbitmq-server.repo exists before trying sed\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/f9b229841db5d27fe11b95d055d74af289893ef2"}]}],"author":{"name":"Viktor Michalek","email":"viktor.michalek@ultimum.io","date":"2019-09-20 16:33:29.000000000","tz":120},"committer":{"name":"Marcin Juszkiewicz","email":"marcin.juszkiewicz@linaro.org","date":"2019-09-27 16:47:19.000000000","tz":-420},"subject":"Fluentd: fix missing /var/run/fluentd","message":"Fluentd: fix missing /var/run/fluentd\n\nChange-Id: Ibd38101eb1ce0085d4ccb877dea7ecae48197563\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/7bb2f9bc391acd1c7ec02f8bfebb2bf634fda5a2"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/7bb2f9bc391acd1c7ec02f8bfebb2bf634fda5a2"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[],"submit_requirements":[]}
