)]}'
{"id":"openstack%2Fkolla~683677","triplet_id":"openstack%2Fkolla~master~I9604ebb4f2a5081fd1209b9841aa997a70a90a11","project":"openstack/kolla","branch":"master","topic":"683677","hashtags":[],"change_id":"I9604ebb4f2a5081fd1209b9841aa997a70a90a11","subject":"Horizon: fix for Debian","status":"MERGED","created":"2019-09-20 18:14:28.000000000","updated":"2019-10-02 09:25:07.000000000","submitted":"2019-10-01 08:23:09.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":29,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"683677-1569918189188-fa124ec6","meta_rev_id":"9c81639f0536eec2c9cb3be3d64e740a6abfe674","_number":683677,"virtual_id_number":683677,"owner":{"_account_id":28374,"name":"Viktor Michalek","email":"viktor.michalek@ultimum.io","username":"viktormichalek"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":28374,"name":"Viktor Michalek","email":"viktor.michalek@ultimum.io","username":"viktormichalek"},{"value":0,"_account_id":24072,"name":"Marcin Juszkiewicz","email":"mjuszkiewicz@redhat.com","username":"hrw"},{"value":0,"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},{"value":0,"_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2019-10-01 08:23:08.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"date":"2019-10-01 03:53:19.000000000","_account_id":23181,"name":"RDO Third Party CI","email":"dmsimard+rdothirdparty@redhat.com","username":"rdothirdparty","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":24072,"name":"Marcin Juszkiewicz","email":"mjuszkiewicz@redhat.com","username":"hrw"},"all":[{"value":0,"_account_id":28374,"name":"Viktor Michalek","email":"viktor.michalek@ultimum.io","username":"viktormichalek"},{"value":2,"date":"2019-10-01 03:15:50.000000000","_account_id":24072,"name":"Marcin Juszkiewicz","email":"mjuszkiewicz@redhat.com","username":"hrw"},{"value":0,"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},{"value":2,"date":"2019-10-01 06:17:38.000000000","_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":23181,"name":"RDO Third Party CI","email":"dmsimard+rdothirdparty@redhat.com","username":"rdothirdparty","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"},"all":[{"value":0,"_account_id":28374,"name":"Viktor Michalek","email":"viktor.michalek@ultimum.io","username":"viktormichalek"},{"value":0,"_account_id":24072,"name":"Marcin Juszkiewicz","email":"mjuszkiewicz@redhat.com","username":"hrw"},{"value":0,"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},{"value":1,"date":"2019-10-01 06:17:38.000000000","_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":23181,"name":"RDO Third Party CI","email":"dmsimard+rdothirdparty@redhat.com","username":"rdothirdparty","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Backport-Candidate":{"all":[{"value":0,"_account_id":28374,"name":"Viktor Michalek","email":"viktor.michalek@ultimum.io","username":"viktormichalek"},{"value":0,"_account_id":24072,"name":"Marcin Juszkiewicz","email":"mjuszkiewicz@redhat.com","username":"hrw"},{"value":0,"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},{"value":0,"_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":23181,"name":"RDO Third Party CI","email":"dmsimard+rdothirdparty@redhat.com","username":"rdothirdparty","tags":["SERVICE_USER"]}],"values":{"-1":"Do Not Backport"," 0":"Backport Review Needed","+1":"Should Backport"},"description":"","default_value":0,"optional":true},"Review-Priority":{"all":[{"value":0,"_account_id":28374,"name":"Viktor Michalek","email":"viktor.michalek@ultimum.io","username":"viktormichalek"},{"value":0,"_account_id":24072,"name":"Marcin Juszkiewicz","email":"mjuszkiewicz@redhat.com","username":"hrw"},{"value":0,"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},{"value":0,"_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":23181,"name":"RDO Third Party CI","email":"dmsimard+rdothirdparty@redhat.com","username":"rdothirdparty","tags":["SERVICE_USER"]}],"values":{"-1":"Branch Freeze"," 0":"No Priority","+1":"Important Change","+2":"Gate Blocker Fix / Urgent Change"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":23181,"name":"RDO Third Party CI","email":"dmsimard+rdothirdparty@redhat.com","username":"rdothirdparty","tags":["SERVICE_USER"]},{"_account_id":24072,"name":"Marcin Juszkiewicz","email":"mjuszkiewicz@redhat.com","username":"hrw"},{"_account_id":28374,"name":"Viktor Michalek","email":"viktor.michalek@ultimum.io","username":"viktormichalek"},{"_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2019-09-24 08:43:57.000000000","updated_by":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},"reviewer":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},"state":"REVIEWER"},{"updated":"2019-10-01 03:15:50.000000000","updated_by":{"_account_id":24072,"name":"Marcin Juszkiewicz","email":"mjuszkiewicz@redhat.com","username":"hrw"},"reviewer":{"_account_id":24072,"name":"Marcin Juszkiewicz","email":"mjuszkiewicz@redhat.com","username":"hrw"},"state":"REVIEWER"},{"updated":"2019-10-01 03:53:19.000000000","updated_by":{"_account_id":23181,"name":"RDO Third Party CI","email":"dmsimard+rdothirdparty@redhat.com","username":"rdothirdparty","tags":["SERVICE_USER"]},"reviewer":{"_account_id":23181,"name":"RDO Third Party CI","email":"dmsimard+rdothirdparty@redhat.com","username":"rdothirdparty","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2019-10-01 06:17:38.000000000","updated_by":{"_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"},"reviewer":{"_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"},"state":"REVIEWER"},{"updated":"2019-10-01 08:23:08.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"}],"messages":[{"id":"5a6691e1469317a9eff2a454c72e5e74b71010f5","author":{"_account_id":28374,"name":"Viktor Michalek","email":"viktor.michalek@ultimum.io","username":"viktormichalek"},"date":"2019-09-20 18:14:28.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"33ddc13e38155bf94c4f5ca29ae03765abd40206","author":{"_account_id":23181,"name":"RDO Third Party CI","email":"dmsimard+rdothirdparty@redhat.com","username":"rdothirdparty","tags":["SERVICE_USER"]},"date":"2019-09-20 18:58:05.000000000","message":"Patch Set 1:\n\nBuild succeeded (check pipeline).\n\n- tripleo-build-containers-rhel-8-master http://logs.rdoproject.org/77/683677/1/openstack-check/tripleo-build-containers-rhel-8-master/323ded9/ : SUCCESS in 41m 34s","accounts_in_message":[],"_revision_number":1},{"id":"09f446188dd9519d1937057f4572df3d3f69fcaa","author":{"_account_id":24072,"name":"Marcin Juszkiewicz","email":"mjuszkiewicz@redhat.com","username":"hrw"},"date":"2019-09-20 20:45:47.000000000","message":"Patch Set 1:\n\n(3 comments)","accounts_in_message":[],"_revision_number":1},{"id":"424b4e29c2c5a6a1b462cac1dc8e8cdaad6d2857","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-09-20 21:57:24.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\n\n- kolla-build-centos-binary https://zuul.opendev.org/t/openstack/build/22c3fa21e8834bbba6d4c2bd9433d379 : SUCCESS in 47m 17s (non-voting)\n- kolla-build-centos-source https://zuul.opendev.org/t/openstack/build/44a14a5b0f904ee380c554e8464abc60 : SUCCESS in 53m 34s\n- kolla-ansible-centos-source https://zuul.opendev.org/t/openstack/build/bd672a1893d74dc49e9800e04e6506e8 : SUCCESS in 1h 29m 32s\n- kolla-ansible-centos-binary https://zuul.opendev.org/t/openstack/build/78443d242a7f43b5b2fbfb89175bee22 : SUCCESS in 1h 23m 53s (non-voting)\n- kolla-ansible-centos-source-upgrade https://zuul.opendev.org/t/openstack/build/7ec5aa38cc044c389578310457db6819 : SUCCESS in 1h 32m 27s\n- tripleo-build-containers-centos-7 https://zuul.opendev.org/t/openstack/build/f3fa90680b284410a827b6adf2882954 : SUCCESS in 1h 05m 47s\n- kolla-build-debian-source https://zuul.opendev.org/t/openstack/build/92f81ad5f5c34ea99e69748acd223599 : SUCCESS in 30m 34s (non-voting)\n- kolla-build-debian-binary https://zuul.opendev.org/t/openstack/build/2cf084e369864c17aa326025911174f8 : FAILURE in 18m 25s (non-voting)\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/751f72d935d7449994212a7ed6171b52 : SUCCESS in 4m 45s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/9af086a0047b47b7a2742b677b63ef42 : SUCCESS in 5m 00s\n- openstack-tox-py27 https://zuul.opendev.org/t/openstack/build/ea00c5cf0efd4f58a6697ea1d23867dc : SUCCESS in 4m 10s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/47e7135bdb7b4946a87a68c10a274709 : SUCCESS in 4m 04s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/3037dad947d34ef98dfdeec9aedf0ff9 : SUCCESS in 3m 57s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/99d87adb9b8b45c594d0362944ae848b : SUCCESS in 4m 52s\n- kolla-build-ubuntu-binary https://zuul.opendev.org/t/openstack/build/04889e93e0d34954ac45d5f9aa9ef212 : SUCCESS in 34m 11s (non-voting)\n- kolla-build-ubuntu-source https://zuul.opendev.org/t/openstack/build/625f570744634f698f9b746a5b860e3a : SUCCESS in 38m 03s\n- kolla-ansible-ubuntu-source https://zuul.opendev.org/t/openstack/build/6d691f1493e14dea96b69988f5e3a818 : SUCCESS in 57m 17s\n- kolla-ansible-ubuntu-binary https://zuul.opendev.org/t/openstack/build/73cc1a0c87014bf4bf55a615de45aa2a : SUCCESS in 51m 01s (non-voting)\n- kolla-ansible-ubuntu-source-upgrade https://zuul.opendev.org/t/openstack/build/e5bd0f06cb014530a61ebf8807baa369 : SUCCESS in 1h 19m 50s","accounts_in_message":[],"_revision_number":1},{"id":"48762a3b334d6bca07ade5a6c5e96929016c9221","author":{"_account_id":28374,"name":"Viktor Michalek","email":"viktor.michalek@ultimum.io","username":"viktormichalek"},"date":"2019-09-22 16:37:09.000000000","message":"Patch Set 2: Commit message was updated.","accounts_in_message":[],"_revision_number":2},{"id":"61389bd2c889e815c6b831e4c173c3c78765b4f8","author":{"_account_id":23181,"name":"RDO Third Party CI","email":"dmsimard+rdothirdparty@redhat.com","username":"rdothirdparty","tags":["SERVICE_USER"]},"date":"2019-09-22 17:15:02.000000000","message":"Patch Set 2:\n\nBuild succeeded (check pipeline).\n\n- tripleo-build-containers-rhel-8-master http://logs.rdoproject.org/77/683677/2/openstack-check/tripleo-build-containers-rhel-8-master/585864e/ : SUCCESS in 36m 19s","accounts_in_message":[],"_revision_number":2},{"id":"e41c148ed8672a796e0ff7f64ec8a19e03257026","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-09-22 18:03:14.000000000","message":"Patch Set 2: Verified-1\n\nThis change depends on a change that failed to merge.","accounts_in_message":[],"_revision_number":2},{"id":"e28afdb300cd63290951630be7dc967b15566fb4","author":{"_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"},"date":"2019-09-23 09:27:52.000000000","message":"Patch Set 2:\n\n(4 comments)","accounts_in_message":[],"_revision_number":2},{"id":"e586deb1ae6c76e70e291b2c6c0adf58a10676df","author":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},"date":"2019-09-24 08:43:57.000000000","message":"Patch Set 2:\n\n(6 comments)","accounts_in_message":[],"_revision_number":2},{"id":"e5223553c89ac698823c98544ea73b9a2f341e3f","author":{"_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"},"date":"2019-09-24 08:48:09.000000000","message":"Patch Set 2:\n\n(2 comments)","accounts_in_message":[],"_revision_number":2},{"id":"70be87dc98a6dd26ade8dfcb1d17add4719808f5","author":{"_account_id":28374,"name":"Viktor Michalek","email":"viktor.michalek@ultimum.io","username":"viktormichalek"},"date":"2019-09-24 15:52:39.000000000","message":"Patch Set 3: Patch Set 2 was rebased","accounts_in_message":[],"_revision_number":3},{"id":"baf93a03353be4e9381736d5ccd15d3ef6a5fa1d","author":{"_account_id":23181,"name":"RDO Third Party CI","email":"dmsimard+rdothirdparty@redhat.com","username":"rdothirdparty","tags":["SERVICE_USER"]},"date":"2019-09-24 16:39:03.000000000","message":"Patch Set 3:\n\nBuild succeeded (check pipeline).\n\n- tripleo-build-containers-rhel-8-master http://logs.rdoproject.org/77/683677/3/openstack-check/tripleo-build-containers-rhel-8-master/738db25/ : SUCCESS in 41m 26s","accounts_in_message":[],"_revision_number":3},{"id":"af8d463dd19233ce00613396c42bdbabff4e94b0","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-09-24 17:54:32.000000000","message":"Patch Set 3: Verified+1\n\nBuild succeeded (check pipeline).\n\n- kolla-build-centos-binary https://zuul.opendev.org/t/openstack/build/fa2b26ff7b8947b5ad03d838edea7355 : SUCCESS in 49m 07s (non-voting)\n- kolla-build-centos-source https://zuul.opendev.org/t/openstack/build/124352a1915b4f248128723e00946ba8 : SUCCESS in 1h 04m 00s\n- kolla-ansible-centos-source https://zuul.opendev.org/t/openstack/build/67cb06cbe77541dabe97a19f957cca23 : SUCCESS in 1h 17m 55s\n- kolla-ansible-centos-binary https://zuul.opendev.org/t/openstack/build/db2c2f50d8ce4005ad388d98267ed8ec : SUCCESS in 1h 37m 06s (non-voting)\n- kolla-ansible-centos-source-upgrade https://zuul.opendev.org/t/openstack/build/580eaabcd3904938ac387f68fb5bcfa4 : SUCCESS in 1h 44m 16s\n- tripleo-build-containers-centos-7 https://zuul.opendev.org/t/openstack/build/285fa1e867d049aa9109581b51477a25 : SUCCESS in 1h 00m 59s\n- kolla-build-debian-source https://zuul.opendev.org/t/openstack/build/fd13dae8238b45bf87fda1457e8e4f12 : SUCCESS in 29m 31s (non-voting)\n- kolla-build-debian-binary https://zuul.opendev.org/t/openstack/build/0cb495bae7454ee68f66f5072b566654 : FAILURE in 19m 25s (non-voting)\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/231ff3dc8968420c8d6143023cd6bcdc : SUCCESS in 3m 31s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/4755a0e5bba04d6ebf4b5b2d7623b9ef : SUCCESS in 5m 14s\n- openstack-tox-py27 https://zuul.opendev.org/t/openstack/build/24330c989e5b43c3b1910fe60536dc9b : SUCCESS in 4m 41s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/d508ba7aff094daabcaa9f1051e8683e : SUCCESS in 4m 27s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/1a1b8999e38542738fb32022b467f9fe : SUCCESS in 3m 21s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/5da806fadc43437da440fdec43f5afce : SUCCESS in 4m 54s\n- kolla-build-ubuntu-binary https://zuul.opendev.org/t/openstack/build/4c44afaabcd84a198122cae77b48ae19 : SUCCESS in 37m 55s (non-voting)\n- kolla-build-ubuntu-source https://zuul.opendev.org/t/openstack/build/ba4741f48a394522a332c9410205845b : SUCCESS in 37m 20s\n- kolla-ansible-ubuntu-source https://zuul.opendev.org/t/openstack/build/334a6d7882bc4d65b5a6d64acf4a6104 : SUCCESS in 1h 03m 55s\n- kolla-ansible-ubuntu-binary https://zuul.opendev.org/t/openstack/build/6f2f1f84fd9e42968bd1a69acaac46da : SUCCESS in 1h 07m 20s (non-voting)\n- kolla-ansible-ubuntu-source-upgrade https://zuul.opendev.org/t/openstack/build/0a73197ce9a445e987aca1d104142f92 : SUCCESS in 1h 04m 00s","accounts_in_message":[],"_revision_number":3},{"id":"0163aa07983c76b3c28749371604c801367f5a31","author":{"_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"},"date":"2019-09-25 07:07:35.000000000","message":"Patch Set 3: Code-Review-1\n\nplease review Mark\u0027s and mine comments","accounts_in_message":[],"_revision_number":3},{"id":"5219dbddb036ec3ed216286f275baa6aa17294ce","author":{"_account_id":28374,"name":"Viktor Michalek","email":"viktor.michalek@ultimum.io","username":"viktormichalek"},"date":"2019-09-26 15:48:09.000000000","message":"Patch Set 4: Published edit on patch set 3.","accounts_in_message":[],"_revision_number":4},{"id":"3b9d966f225b39a68b8aad3bfa8eb9a38a8da48e","author":{"_account_id":28374,"name":"Viktor Michalek","email":"viktor.michalek@ultimum.io","username":"viktormichalek"},"date":"2019-09-26 16:03:21.000000000","message":"Patch Set 5: Published edit on patch set 4.","accounts_in_message":[],"_revision_number":5},{"id":"d3e83da2962cf2a93d7d356769d294723553be55","author":{"_account_id":28374,"name":"Viktor Michalek","email":"viktor.michalek@ultimum.io","username":"viktormichalek"},"date":"2019-09-26 16:24:30.000000000","message":"Patch Set 6: Published edit on patch set 5.","accounts_in_message":[],"_revision_number":6},{"id":"c3e3e45275e9dc6ed8a287a4a1d79b588bf506bb","author":{"_account_id":23181,"name":"RDO Third Party CI","email":"dmsimard+rdothirdparty@redhat.com","username":"rdothirdparty","tags":["SERVICE_USER"]},"date":"2019-09-26 17:03:51.000000000","message":"Patch Set 6:\n\nBuild succeeded (check pipeline).\n\n- tripleo-build-containers-rhel-8-master http://logs.rdoproject.org/77/683677/6/openstack-check/tripleo-build-containers-rhel-8-master/5eebe7b/ : SUCCESS in 38m 17s","accounts_in_message":[],"_revision_number":6},{"id":"d9b7f60f5d51c4bab04eee0e09226c6d8f5bd8e6","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-09-26 18:24:08.000000000","message":"Patch Set 6: Verified+1\n\nBuild succeeded (check pipeline).\n\n- kolla-build-centos-binary https://zuul.opendev.org/t/openstack/build/2c5fa9e3b7c1457cb3bd210645730ce3 : SUCCESS in 54m 37s (non-voting)\n- kolla-build-centos-source https://zuul.opendev.org/t/openstack/build/3a69ea88414d456ca50f256ec329a9f8 : SUCCESS in 46m 12s\n- kolla-ansible-centos-source https://zuul.opendev.org/t/openstack/build/b7f84b691f014542b47ecf8171d51016 : SUCCESS in 1h 41m 02s\n- kolla-ansible-centos-binary https://zuul.opendev.org/t/openstack/build/800147c4700e4e11b87fdbf72e2ceb97 : SUCCESS in 1h 31m 16s (non-voting)\n- kolla-ansible-centos-source-upgrade https://zuul.opendev.org/t/openstack/build/c79e2e0e4df74bd6be2de3ad8f7dc3cf : SUCCESS in 1h 51m 08s\n- tripleo-build-containers-centos-7 https://zuul.opendev.org/t/openstack/build/051a4f7851b04f9993df9e042172db30 : SUCCESS in 51m 59s\n- kolla-build-debian-source https://zuul.opendev.org/t/openstack/build/446a7cb9a491410a8139a0439e4a38af : SUCCESS in 30m 35s (non-voting)\n- kolla-build-debian-binary https://zuul.opendev.org/t/openstack/build/283ee962d60243e8b19c73131dd453cc : FAILURE in 24m 05s (non-voting)\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/bdb9d221addf4a5c86fe7bbbfaeb403b : SUCCESS in 3m 21s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/5498532736d0458196711763244b5652 : SUCCESS in 4m 35s\n- openstack-tox-py27 https://zuul.opendev.org/t/openstack/build/008efd691340446bbd7e347095aabba9 : SUCCESS in 4m 42s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/20f4c6d306004efca49d672f52a11b24 : SUCCESS in 4m 02s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/80cf842addbf4330966298d5c97b0394 : SUCCESS in 4m 12s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/a7fc4b566f83498ca85120e5ce571104 : SUCCESS in 5m 43s\n- kolla-build-ubuntu-binary https://zuul.opendev.org/t/openstack/build/cc918fe1169f40eaa95732525a526b0f : SUCCESS in 1h 00m 34s (non-voting)\n- kolla-build-ubuntu-source https://zuul.opendev.org/t/openstack/build/1b4013d315d5476e8f13f2116816ec08 : SUCCESS in 33m 47s\n- kolla-ansible-ubuntu-source https://zuul.opendev.org/t/openstack/build/8620bf682605413f98321bc6fec95012 : SUCCESS in 57m 56s\n- kolla-ansible-ubuntu-binary https://zuul.opendev.org/t/openstack/build/621c813db3884e2593a48e4dca66985b : SUCCESS in 54m 47s (non-voting)\n- kolla-ansible-ubuntu-source-upgrade https://zuul.opendev.org/t/openstack/build/f2f2330c3c2d4132b079996801972af3 : SUCCESS in 1h 19m 55s","accounts_in_message":[],"_revision_number":6},{"id":"bb58d1ba8c4757927c8c1fcb9559ae7175d7922c","author":{"_account_id":28374,"name":"Viktor Michalek","email":"viktor.michalek@ultimum.io","username":"viktormichalek"},"date":"2019-09-27 15:05:51.000000000","message":"Patch Set 6:\n\n(11 comments)\n\n\u003e please review Mark\u0027s and mine comments\n\nHave answered to comments.","accounts_in_message":[],"_revision_number":6},{"id":"a2a0d93d6bb3c925454c7d4d7463a181137fe649","author":{"_account_id":24072,"name":"Marcin Juszkiewicz","email":"mjuszkiewicz@redhat.com","username":"hrw"},"date":"2019-09-27 16:08:13.000000000","message":"Patch Set 6:\n\nrecheck","accounts_in_message":[],"_revision_number":6},{"id":"6270d9f6b31f9212fd7494894497c4fce3d6749a","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-09-27 16:11:09.000000000","message":"Patch Set 6: Verified-1\n\nThis change depends on a change that failed to merge.","accounts_in_message":[],"_revision_number":6},{"id":"66955cbb65919df7ed09944e4a628dc01ff476c2","author":{"_account_id":23181,"name":"RDO Third Party CI","email":"dmsimard+rdothirdparty@redhat.com","username":"rdothirdparty","tags":["SERVICE_USER"]},"date":"2019-09-27 16:11:11.000000000","message":"Patch Set 6:\n\nThis change depends on a change that failed to merge.","accounts_in_message":[],"_revision_number":6},{"id":"5e0a6204ffac60b91d444f8826dbf8cf1330bd47","author":{"_account_id":28374,"name":"Viktor Michalek","email":"viktor.michalek@ultimum.io","username":"viktormichalek"},"date":"2019-09-27 16:38:02.000000000","message":"Patch Set 7: Commit message was updated.","accounts_in_message":[],"_revision_number":7},{"id":"0faf0612cbc933dcc9fa17f4b470f8a4751c8017","author":{"_account_id":24072,"name":"Marcin Juszkiewicz","email":"mjuszkiewicz@redhat.com","username":"hrw"},"date":"2019-09-27 16:44:20.000000000","message":"Uploaded patch set 8: Patch Set 7 was rebased.","accounts_in_message":[],"_revision_number":8},{"id":"dd7f090ddc4382bf1770a7e3e4b5a1cf215f53d3","author":{"_account_id":23181,"name":"RDO Third Party CI","email":"dmsimard+rdothirdparty@redhat.com","username":"rdothirdparty","tags":["SERVICE_USER"]},"date":"2019-09-27 17:23:43.000000000","message":"Patch Set 8:\n\nBuild succeeded (check pipeline).\n\n- tripleo-build-containers-rhel-8-master http://logs.rdoproject.org/77/683677/8/openstack-check/tripleo-build-containers-rhel-8-master/df675ed/ : SUCCESS in 38m 37s","accounts_in_message":[],"_revision_number":8},{"id":"5e5cd96bbe8fba0b0649c95010939f150fab9dd3","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-09-27 18:32:25.000000000","message":"Patch Set 8: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttp://docs.openstack.org/infra/manual/developers.html#automated-testing\n\n\n- kolla-build-centos-binary https://zuul.opendev.org/t/openstack/build/4f91e62c184a4efbbfe43daddbd6ae6a : FAILURE in 45m 28s (non-voting)\n- kolla-build-centos-source https://zuul.opendev.org/t/openstack/build/854b5750b1934c19a444a23f56dd0a8b : FAILURE in 57m 57s\n- kolla-ansible-centos-source https://zuul.opendev.org/t/openstack/build/b7ad3f13404a4a95bd03cdf75bfb9a71 : SUCCESS in 1h 22m 04s\n- kolla-ansible-centos-binary https://zuul.opendev.org/t/openstack/build/e123deace90040d3aa39adba3c9d6972 : SUCCESS in 1h 14m 10s (non-voting)\n- kolla-ansible-centos-source-upgrade https://zuul.opendev.org/t/openstack/build/ccf99771913a4eeea34b9a4d9f59f4b0 : SUCCESS in 1h 43m 05s\n- tripleo-build-containers-centos-7 https://zuul.opendev.org/t/openstack/build/16475e9465e74e7fbcdfb65c908fc293 : SUCCESS in 55m 57s\n- kolla-build-debian-source https://zuul.opendev.org/t/openstack/build/ecc565db21c941b5b16740a03a1884af : SUCCESS in 35m 44s (non-voting)\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/f46d3ec39fd34d708e7ed223cc282d10 : SUCCESS in 3m 35s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/8616faeac45b459a9e69ea3f6c866911 : SUCCESS in 4m 41s\n- openstack-tox-py27 https://zuul.opendev.org/t/openstack/build/dfd20c46b6d2443fae12fc134687120a : SUCCESS in 4m 21s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/9c9af0b9d1db420eaae5049911127ab9 : SUCCESS in 3m 27s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/f1967bde92c9408c8f279da723c79686 : SUCCESS in 4m 14s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/dbfd7c65d2404c8b9c96780b8f3041f0 : SUCCESS in 5m 08s\n- kolla-build-ubuntu-binary https://zuul.opendev.org/t/openstack/build/477a92243a844e26bee597546c933d9c : SUCCESS in 30m 18s (non-voting)\n- kolla-build-ubuntu-source https://zuul.opendev.org/t/openstack/build/abc3f50cea0342afa54b191584b00985 : SUCCESS in 29m 05s\n- kolla-ansible-ubuntu-source https://zuul.opendev.org/t/openstack/build/75f1e9d270d046ce9f976ccf9fbf12e6 : SUCCESS in 1h 01m 03s\n- kolla-ansible-ubuntu-binary https://zuul.opendev.org/t/openstack/build/74ec2a67e74b47549747a46a6db474a3 : SUCCESS in 1h 00m 49s (non-voting)\n- kolla-ansible-ubuntu-source-upgrade https://zuul.opendev.org/t/openstack/build/0d7e633f4f744706886580f8f28b1cf8 : SUCCESS in 56m 28s","accounts_in_message":[],"_revision_number":8},{"id":"9dde754991d2fb507baabecf094a17ddf675db68","author":{"_account_id":28374,"name":"Viktor Michalek","email":"viktor.michalek@ultimum.io","username":"viktormichalek"},"date":"2019-09-27 20:52:50.000000000","message":"Patch Set 8:\n\nrecheck","accounts_in_message":[],"_revision_number":8},{"id":"e8f3a95d60b239023fc32c0111b41934fa05216f","author":{"_account_id":23181,"name":"RDO Third Party CI","email":"dmsimard+rdothirdparty@redhat.com","username":"rdothirdparty","tags":["SERVICE_USER"]},"date":"2019-09-27 21:32:02.000000000","message":"Patch Set 8:\n\nBuild succeeded (check pipeline).\n\n- tripleo-build-containers-rhel-8-master http://logs.rdoproject.org/77/683677/8/openstack-check/tripleo-build-containers-rhel-8-master/4a8c5a0/ : SUCCESS in 38m 29s","accounts_in_message":[],"_revision_number":8},{"id":"9ac75929cd5d17290c0304d77ed21ae0a4c7841a","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-09-27 22:36:20.000000000","message":"Patch Set 8: Verified+1\n\nBuild succeeded (check pipeline).\n\n- kolla-build-centos-binary https://zuul.opendev.org/t/openstack/build/e9e500f9f1ae4e1ab9eca0b654284b61 : SUCCESS in 43m 47s (non-voting)\n- kolla-build-centos-source https://zuul.opendev.org/t/openstack/build/fd4f7c76520a4a109ce2401505be243f : SUCCESS in 1h 12m 36s\n- kolla-ansible-centos-source https://zuul.opendev.org/t/openstack/build/a6a98166daee4359bf1f2f1660604cef : SUCCESS in 1h 37m 44s\n- kolla-ansible-centos-binary https://zuul.opendev.org/t/openstack/build/03edb1cc2dd94bebac655923971cbfed : SUCCESS in 1h 16m 49s (non-voting)\n- kolla-ansible-centos-source-upgrade https://zuul.opendev.org/t/openstack/build/28d62d09d36a472dadf7f13bf0b1a0c4 : SUCCESS in 1h 29m 28s\n- tripleo-build-containers-centos-7 https://zuul.opendev.org/t/openstack/build/bbfff1f523ed406bb7d680a982d7430d : SUCCESS in 1h 02m 34s\n- kolla-build-debian-source https://zuul.opendev.org/t/openstack/build/12fa26d39e31452383fc21485ec0c3be : SUCCESS in 32m 37s (non-voting)\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/f92e93f23fb74c8e9c3ac72c0bbe53ec : SUCCESS in 4m 18s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/1a8f3b2275c74bffbdfaf398a7221b68 : SUCCESS in 4m 31s\n- openstack-tox-py27 https://zuul.opendev.org/t/openstack/build/ebc7dc138b1c4b5f81ca2582ab742972 : SUCCESS in 4m 25s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/332aa6f9ac134dd68fee4d94f577980b : SUCCESS in 4m 21s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/cb188e229e5849748565f238ebe2f10a : SUCCESS in 4m 46s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/3176aa6ec02b47bbae3a874637f58f61 : SUCCESS in 5m 56s\n- kolla-build-ubuntu-binary https://zuul.opendev.org/t/openstack/build/38741c5111cb47fd98b1242549946cb6 : SUCCESS in 24m 42s (non-voting)\n- kolla-build-ubuntu-source https://zuul.opendev.org/t/openstack/build/db7cb5c9fad4481b826d0198fe3ed5bb : SUCCESS in 31m 35s\n- kolla-ansible-ubuntu-source https://zuul.opendev.org/t/openstack/build/31b241cae3e94a64b6a45d9bff03a045 : SUCCESS in 54m 43s\n- kolla-ansible-ubuntu-binary https://zuul.opendev.org/t/openstack/build/fd06e2b1a39348bdbb794b946d98f5a8 : SUCCESS in 55m 51s (non-voting)\n- kolla-ansible-ubuntu-source-upgrade https://zuul.opendev.org/t/openstack/build/b231ee388834451c9d03825035359db8 : SUCCESS in 1h 03m 30s","accounts_in_message":[],"_revision_number":8},{"id":"61b7c1604e4b4cd82426ac29bfe7e23cd2d03f11","author":{"_account_id":24072,"name":"Marcin Juszkiewicz","email":"mjuszkiewicz@redhat.com","username":"hrw"},"date":"2019-09-30 09:14:42.000000000","message":"Patch Set 8:\n\n(1 comment)","accounts_in_message":[],"_revision_number":8},{"id":"9df14c22852b1113549ffbd412fa0fe261ac5b7c","author":{"_account_id":28374,"name":"Viktor Michalek","email":"viktor.michalek@ultimum.io","username":"viktormichalek"},"date":"2019-09-30 14:49:46.000000000","message":"Uploaded patch set 9.","accounts_in_message":[],"_revision_number":9},{"id":"db1c85ff4bb9294b250d912e3d25eb4205496491","author":{"_account_id":24072,"name":"Marcin Juszkiewicz","email":"mjuszkiewicz@redhat.com","username":"hrw"},"date":"2019-09-30 14:53:30.000000000","message":"Patch Set 9: Code-Review+2","accounts_in_message":[],"_revision_number":9},{"id":"a46707f1a317f8df1f90f14e2416b44fecae6123","author":{"_account_id":23181,"name":"RDO Third Party CI","email":"dmsimard+rdothirdparty@redhat.com","username":"rdothirdparty","tags":["SERVICE_USER"]},"date":"2019-09-30 15:28:55.000000000","message":"Patch Set 9:\n\nBuild succeeded (check pipeline).\n\n- tripleo-build-containers-rhel-8-master http://logs.rdoproject.org/77/683677/9/openstack-check/tripleo-build-containers-rhel-8-master/ac5718d/ : SUCCESS in 38m 15s","accounts_in_message":[],"_revision_number":9},{"id":"f6890e4cddae6c32dcbc308dab2db66f66786fea","author":{"_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"},"date":"2019-09-30 16:12:03.000000000","message":"Patch Set 9:\n\n(1 comment)","accounts_in_message":[],"_revision_number":9},{"id":"32cafce2cbbf9300cca04ac2747398f186df58f7","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-09-30 16:25:38.000000000","message":"Patch Set 9: Verified+1\n\nBuild succeeded (check pipeline).\n\n- kolla-build-centos-binary https://zuul.opendev.org/t/openstack/build/1626e82ba4be4e68afcc534548658992 : SUCCESS in 51m 53s (non-voting)\n- kolla-build-centos-source https://zuul.opendev.org/t/openstack/build/b8ee4b223351487b88b370b4f8ef4bb9 : SUCCESS in 53m 29s\n- kolla-ansible-centos-source https://zuul.opendev.org/t/openstack/build/72a95eaf8cf3491a913acbf0a204c014 : SUCCESS in 1h 31m 03s\n- kolla-ansible-centos-binary https://zuul.opendev.org/t/openstack/build/46ff1368704e424ebfc911b6951e4f33 : SUCCESS in 1h 26m 50s (non-voting)\n- kolla-ansible-centos-source-upgrade https://zuul.opendev.org/t/openstack/build/73dddb261d7b4c04bae18315b149628c : SUCCESS in 1h 24m 41s\n- tripleo-build-containers-centos-7 https://zuul.opendev.org/t/openstack/build/d67748edd9c44851a1606176ebdcb2c5 : SUCCESS in 49m 49s\n- kolla-build-debian-source https://zuul.opendev.org/t/openstack/build/8000890622f5499f891e3f20ad635e1b : SUCCESS in 26m 34s (non-voting)\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/5ccb234c95da46e186eafa1f650ac35d : SUCCESS in 3m 47s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/64509b6fe11a44dfb80affadf5db1969 : SUCCESS in 4m 25s\n- openstack-tox-py27 https://zuul.opendev.org/t/openstack/build/4416d13ed81f45fc8c798f93647d6bb6 : SUCCESS in 3m 44s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/bcc7920dfbb2490b826968c988b60cd3 : SUCCESS in 3m 14s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/f8d82882563f4e50a60c39eab3b78edb : SUCCESS in 5m 05s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/91b00bcfdb17417dbfd18ca29b4302bd : SUCCESS in 5m 00s\n- kolla-build-ubuntu-binary https://zuul.opendev.org/t/openstack/build/207e040559a945ada789d4f7eab01c8e : SUCCESS in 29m 20s (non-voting)\n- kolla-build-ubuntu-source https://zuul.opendev.org/t/openstack/build/85b07291819d4cbdad5fde728255a7f2 : SUCCESS in 30m 51s\n- kolla-ansible-ubuntu-source https://zuul.opendev.org/t/openstack/build/85ffeb13234f4a19be4af2c7875dd1d5 : SUCCESS in 55m 52s\n- kolla-ansible-ubuntu-binary https://zuul.opendev.org/t/openstack/build/7cba044a882f4cbeb8e26196136bd8d5 : SUCCESS in 57m 08s (non-voting)\n- kolla-ansible-ubuntu-source-upgrade https://zuul.opendev.org/t/openstack/build/bbe2a078eedb490d8e60882b1e9b311a : SUCCESS in 55m 52s","accounts_in_message":[],"_revision_number":9},{"id":"6d6dd620e42bbe9108f96120c02a2a2868adc7b4","author":{"_account_id":24072,"name":"Marcin Juszkiewicz","email":"mjuszkiewicz@redhat.com","username":"hrw"},"date":"2019-10-01 03:15:34.000000000","message":"Patch Set 10: Published edit on patch set 9.","accounts_in_message":[],"_revision_number":10},{"id":"5ed391c29b49e8c9c667eb5c4847688b50cb039f","author":{"_account_id":24072,"name":"Marcin Juszkiewicz","email":"mjuszkiewicz@redhat.com","username":"hrw"},"date":"2019-10-01 03:15:50.000000000","message":"Patch Set 10: Code-Review+2","accounts_in_message":[],"_revision_number":10},{"id":"35536aa72662d8f6b71cc9c41b97e3bdea87e2f9","author":{"_account_id":23181,"name":"RDO Third Party CI","email":"dmsimard+rdothirdparty@redhat.com","username":"rdothirdparty","tags":["SERVICE_USER"]},"date":"2019-10-01 03:53:19.000000000","message":"Patch Set 10:\n\nBuild succeeded (check pipeline).\n\n- tripleo-build-containers-rhel-8-master http://logs.rdoproject.org/77/683677/10/openstack-check/tripleo-build-containers-rhel-8-master/ff695f9/ : SUCCESS in 37m 05s","accounts_in_message":[],"_revision_number":10},{"id":"d8e413bef1e1e23d42b912824897752c5bf05e5e","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-10-01 04:50:32.000000000","message":"Patch Set 10: Verified+1\n\nBuild succeeded (check pipeline).\n\n- kolla-build-centos-binary https://zuul.opendev.org/t/openstack/build/202c5e66a012454381fc2204a91f4de6 : SUCCESS in 45m 36s (non-voting)\n- kolla-build-centos-source https://zuul.opendev.org/t/openstack/build/8b7c51f418914b3fbf86b4c6d7ecee61 : SUCCESS in 50m 29s\n- kolla-ansible-centos-source https://zuul.opendev.org/t/openstack/build/8149503d020c4b599376eea345b01958 : SUCCESS in 1h 33m 26s\n- kolla-ansible-centos-binary https://zuul.opendev.org/t/openstack/build/630669c3d9014abc8a3c78261737fab0 : SUCCESS in 1h 30m 21s (non-voting)\n- kolla-ansible-centos-source-upgrade https://zuul.opendev.org/t/openstack/build/8fa6524330c341579a2edb2f17e3c823 : SUCCESS in 1h 25m 12s\n- tripleo-build-containers-centos-7 https://zuul.opendev.org/t/openstack/build/ec8e4534c63a4b81b965f8f5d5ac2ef8 : SUCCESS in 56m 34s\n- kolla-build-debian-source https://zuul.opendev.org/t/openstack/build/33129234dc9b4a5f83d6c0712c300bb4 : SUCCESS in 32m 19s (non-voting)\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/26e57dc1e2e343b29481f3a8330d9eb5 : SUCCESS in 3m 10s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/0566bd391b414fd193cc6bf183b751c2 : SUCCESS in 3m 34s\n- openstack-tox-py27 https://zuul.opendev.org/t/openstack/build/4060e043b3bb4a759009a603058aa9b0 : SUCCESS in 2m 46s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/a7418ef2d758453fae8c8c33fc6d9f15 : SUCCESS in 2m 52s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/94547422598c4c58a1813d67af97ded8 : SUCCESS in 3m 11s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/62178e273f404a63a0767b053fafe509 : SUCCESS in 4m 26s\n- kolla-build-ubuntu-binary https://zuul.opendev.org/t/openstack/build/5d33bbdc8bb145309b6f851a66fdf1f2 : SUCCESS in 28m 58s (non-voting)\n- kolla-build-ubuntu-source https://zuul.opendev.org/t/openstack/build/09a5ef0684cc41caae59d48c6eeb0354 : SUCCESS in 35m 19s\n- kolla-ansible-ubuntu-source https://zuul.opendev.org/t/openstack/build/2cc0916af0834710a93107d9a415e43b : SUCCESS in 51m 04s\n- kolla-ansible-ubuntu-binary https://zuul.opendev.org/t/openstack/build/93e4dc72431743008a9f7f1eb251da3c : SUCCESS in 56m 53s (non-voting)\n- kolla-ansible-ubuntu-source-upgrade https://zuul.opendev.org/t/openstack/build/68a69105ef95425e8833c3dbbee40f1e : SUCCESS in 58m 58s","accounts_in_message":[],"_revision_number":10},{"id":"ce43b4ffa9c02ce258fa4c6a6e12b8cd8f03c567","author":{"_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"},"date":"2019-10-01 06:17:38.000000000","message":"Patch Set 10: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":10},{"id":"a966386915dd39f0a4342a110a6bcd40d8abab37","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-10-01 06:20:31.000000000","message":"Patch Set 10: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":10},{"id":"c632a9c231cd0d1ec09982208b234186526e9e4a","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-10-01 08:23:08.000000000","message":"Patch Set 10: Verified+2\n\nBuild succeeded (gate pipeline).\n\n- kolla-build-centos-source https://zuul.opendev.org/t/openstack/build/fd9bc7a0921b45489c0140b55bb2fb8a : SUCCESS in 1h 10m 59s\n- kolla-ansible-centos-source https://zuul.opendev.org/t/openstack/build/5e0825d8eda34b10b43297e9ef3b8b02 : SUCCESS in 1h 36m 34s\n- kolla-ansible-centos-source-upgrade https://zuul.opendev.org/t/openstack/build/a7bfd783eaba4831b4e8bc5a2b10b323 : SUCCESS in 1h 58m 03s\n- tripleo-build-containers-centos-7 https://zuul.opendev.org/t/openstack/build/7a1047f3062543fcb9788acab37652b2 : SUCCESS in 1h 06m 49s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/8b52effce7114f2e949df27246cc03d7 : SUCCESS in 3m 36s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/ca0f33ff142a4b228c0874c1a1669842 : SUCCESS in 5m 51s\n- openstack-tox-py27 https://zuul.opendev.org/t/openstack/build/a3a2b18f97dc42aa909d931809656e2d : SUCCESS in 4m 57s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/39666228477b48bba6ec1104125c9e82 : SUCCESS in 5m 53s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/b4de50b4050b40e9a83ea4ca7aa2b828 : SUCCESS in 7m 01s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/3b284d2a2b4a47e59b5e2945ab098f9b : SUCCESS in 10m 06s\n- kolla-build-ubuntu-source https://zuul.opendev.org/t/openstack/build/8d53cd0f14af488da3d7b96b506d4cad : SUCCESS in 1h 05m 52s\n- kolla-ansible-ubuntu-source https://zuul.opendev.org/t/openstack/build/eeabf145d8ac40dbb6678091f01cff4e : SUCCESS in 1h 28m 57s\n- kolla-ansible-ubuntu-source-upgrade https://zuul.opendev.org/t/openstack/build/0bb9720943eb4232923e9ace4879a2e9 : SUCCESS in 1h 17m 50s","accounts_in_message":[],"_revision_number":10},{"id":"e3329773fed9727246e5f99cbb4469c48b3bfd67","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-10-01 08:23:09.000000000","message":"Change has been successfully merged by Zuul","accounts_in_message":[],"_revision_number":10},{"id":"3822c1e18707e485a9bcf3cd27287dff5838813f","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-10-01 08:24:25.000000000","message":"Patch Set 10:\n\nBuild succeeded (promote pipeline).\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/3e7e93ad980b474f96c620ad18c94e9a : SUCCESS in 54s","accounts_in_message":[],"_revision_number":10},{"id":"eaa48f5c559cf1fadbfb7876bdb3efe0e9dbcead","author":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},"date":"2019-10-02 09:25:07.000000000","message":"Patch Set 10:\n\n(1 comment)","accounts_in_message":[],"_revision_number":10}],"current_revision_number":10,"current_revision":"6f44a3282aeeef90caaba48daae068493e2b8606","revisions":{"2c45db8a0fa5e5e0e35d258b3a190e23376f075a":{"kind":"REWORK","_number":1,"created":"2019-09-20 18:14:28.000000000","uploader":{"_account_id":28374,"name":"Viktor Michalek","email":"viktor.michalek@ultimum.io","username":"viktormichalek"},"ref":"refs/changes/77/683677/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/kolla","ref":"refs/changes/77/683677/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/kolla refs/changes/77/683677/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/kolla refs/changes/77/683677/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/kolla refs/changes/77/683677/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/kolla refs/changes/77/683677/1"}}},"commit":{"parents":[{"commit":"a7abc458a9322dd2aa49771ba67eb53f14a8463e","subject":"Fluentd: fix missing /var/run/fluentd","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/a7abc458a9322dd2aa49771ba67eb53f14a8463e"}]}],"author":{"name":"Viktor Michalek","email":"viktor.michalek@ultimum.io","date":"2019-09-20 16:36:27.000000000","tz":120},"committer":{"name":"Viktor Michalek","email":"viktor.michalek@ultimum.io","date":"2019-09-20 16:36:27.000000000","tz":120},"subject":"Horizon: fix for Debian","message":"Horizon: fix for Debian\n\nChange-Id: I9604ebb4f2a5081fd1209b9841aa997a70a90a11\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/2c45db8a0fa5e5e0e35d258b3a190e23376f075a"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/2c45db8a0fa5e5e0e35d258b3a190e23376f075a"}]},"branch":"refs/heads/master"},"8ff7f1ef4ebba54d3e9b30f98e941ff180826c03":{"kind":"NO_CODE_CHANGE","_number":2,"created":"2019-09-22 16:37:09.000000000","uploader":{"_account_id":28374,"name":"Viktor Michalek","email":"viktor.michalek@ultimum.io","username":"viktormichalek"},"ref":"refs/changes/77/683677/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/kolla","ref":"refs/changes/77/683677/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/kolla refs/changes/77/683677/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/kolla refs/changes/77/683677/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/kolla refs/changes/77/683677/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/kolla refs/changes/77/683677/2"}}},"commit":{"parents":[{"commit":"a7abc458a9322dd2aa49771ba67eb53f14a8463e","subject":"Fluentd: fix missing /var/run/fluentd","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/a7abc458a9322dd2aa49771ba67eb53f14a8463e"}]}],"author":{"name":"Viktor Michalek","email":"viktor.michalek@ultimum.io","date":"2019-09-20 16:36:27.000000000","tz":120},"committer":{"name":"Viktor Michalek","email":"viktor.michalek@ultimum.io","date":"2019-09-22 16:37:03.000000000","tz":0},"subject":"Horizon: fix for Debian","message":"Horizon: fix for Debian\n\nDepends-On: https://review.opendev.org/#/c/683675\nChange-Id: I9604ebb4f2a5081fd1209b9841aa997a70a90a11\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/8ff7f1ef4ebba54d3e9b30f98e941ff180826c03"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/8ff7f1ef4ebba54d3e9b30f98e941ff180826c03"}]},"branch":"refs/heads/master"},"700d91506090d61d3d235292f5a2ca1ea1dd1fcb":{"kind":"TRIVIAL_REBASE","_number":3,"created":"2019-09-24 15:52:39.000000000","uploader":{"_account_id":28374,"name":"Viktor Michalek","email":"viktor.michalek@ultimum.io","username":"viktormichalek"},"ref":"refs/changes/77/683677/3","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/kolla","ref":"refs/changes/77/683677/3","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/kolla refs/changes/77/683677/3 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/kolla refs/changes/77/683677/3 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/kolla refs/changes/77/683677/3 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/kolla refs/changes/77/683677/3"}}},"commit":{"parents":[{"commit":"543b53ed2512cfb39d1ac234a37a067d3f95e063","subject":"Fluentd: fix missing /var/run/fluentd","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/543b53ed2512cfb39d1ac234a37a067d3f95e063"}]}],"author":{"name":"Viktor Michalek","email":"viktor.michalek@ultimum.io","date":"2019-09-20 16:36:27.000000000","tz":120},"committer":{"name":"Viktor Michalek","email":"viktor.michalek@ultimum.io","date":"2019-09-24 15:52:39.000000000","tz":0},"subject":"Horizon: fix for Debian","message":"Horizon: fix for Debian\n\nDepends-On: https://review.opendev.org/#/c/683675\nChange-Id: I9604ebb4f2a5081fd1209b9841aa997a70a90a11\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/700d91506090d61d3d235292f5a2ca1ea1dd1fcb"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/700d91506090d61d3d235292f5a2ca1ea1dd1fcb"}]},"branch":"refs/heads/master"},"f684dc3e9d8999251ee60e767395d2d55563b706":{"kind":"REWORK","_number":4,"created":"2019-09-26 15:48:09.000000000","uploader":{"_account_id":28374,"name":"Viktor Michalek","email":"viktor.michalek@ultimum.io","username":"viktormichalek"},"ref":"refs/changes/77/683677/4","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/kolla","ref":"refs/changes/77/683677/4","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/kolla refs/changes/77/683677/4 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/kolla refs/changes/77/683677/4 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/kolla refs/changes/77/683677/4 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/kolla refs/changes/77/683677/4"}}},"commit":{"parents":[{"commit":"543b53ed2512cfb39d1ac234a37a067d3f95e063","subject":"Fluentd: fix missing /var/run/fluentd","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/543b53ed2512cfb39d1ac234a37a067d3f95e063"}]}],"author":{"name":"Viktor Michalek","email":"viktor.michalek@ultimum.io","date":"2019-09-20 16:36:27.000000000","tz":120},"committer":{"name":"Viktor Michalek","email":"viktor.michalek@ultimum.io","date":"2019-09-26 15:48:01.000000000","tz":0},"subject":"Horizon: fix for Debian","message":"Horizon: fix for Debian\n\nDepends-On: https://review.opendev.org/683675\nChange-Id: I9604ebb4f2a5081fd1209b9841aa997a70a90a11\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/f684dc3e9d8999251ee60e767395d2d55563b706"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/f684dc3e9d8999251ee60e767395d2d55563b706"}]},"branch":"refs/heads/master"},"f49b41f299aa89ef00476269171c95b01b8ee121":{"kind":"REWORK","_number":5,"created":"2019-09-26 16:03:21.000000000","uploader":{"_account_id":28374,"name":"Viktor Michalek","email":"viktor.michalek@ultimum.io","username":"viktormichalek"},"ref":"refs/changes/77/683677/5","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/kolla","ref":"refs/changes/77/683677/5","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/kolla refs/changes/77/683677/5 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/kolla refs/changes/77/683677/5 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/kolla refs/changes/77/683677/5 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/kolla refs/changes/77/683677/5"}}},"commit":{"parents":[{"commit":"543b53ed2512cfb39d1ac234a37a067d3f95e063","subject":"Fluentd: fix missing /var/run/fluentd","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/543b53ed2512cfb39d1ac234a37a067d3f95e063"}]}],"author":{"name":"Viktor Michalek","email":"viktor.michalek@ultimum.io","date":"2019-09-20 16:36:27.000000000","tz":120},"committer":{"name":"Viktor Michalek","email":"viktor.michalek@ultimum.io","date":"2019-09-26 16:03:11.000000000","tz":0},"subject":"Horizon: fix for Debian","message":"Horizon: fix for Debian\n\nDepends-On: https://review.opendev.org/683675\nChange-Id: I9604ebb4f2a5081fd1209b9841aa997a70a90a11\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/f49b41f299aa89ef00476269171c95b01b8ee121"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/f49b41f299aa89ef00476269171c95b01b8ee121"}]},"branch":"refs/heads/master"},"cc3fb7c701c2392e511a0c7a6dfb2c7bdb8a9647":{"kind":"REWORK","_number":6,"created":"2019-09-26 16:24:30.000000000","uploader":{"_account_id":28374,"name":"Viktor Michalek","email":"viktor.michalek@ultimum.io","username":"viktormichalek"},"ref":"refs/changes/77/683677/6","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/kolla","ref":"refs/changes/77/683677/6","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/kolla refs/changes/77/683677/6 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/kolla refs/changes/77/683677/6 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/kolla refs/changes/77/683677/6 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/kolla refs/changes/77/683677/6"}}},"commit":{"parents":[{"commit":"543b53ed2512cfb39d1ac234a37a067d3f95e063","subject":"Fluentd: fix missing /var/run/fluentd","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/543b53ed2512cfb39d1ac234a37a067d3f95e063"}]}],"author":{"name":"Viktor Michalek","email":"viktor.michalek@ultimum.io","date":"2019-09-20 16:36:27.000000000","tz":120},"committer":{"name":"Viktor Michalek","email":"viktor.michalek@ultimum.io","date":"2019-09-26 16:24:25.000000000","tz":0},"subject":"Horizon: fix for Debian","message":"Horizon: fix for Debian\n\nDepends-On: https://review.opendev.org/683675\nChange-Id: I9604ebb4f2a5081fd1209b9841aa997a70a90a11\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/cc3fb7c701c2392e511a0c7a6dfb2c7bdb8a9647"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/cc3fb7c701c2392e511a0c7a6dfb2c7bdb8a9647"}]},"branch":"refs/heads/master"},"17463dfa11c030ee521347c5231f696b013ef9e8":{"kind":"NO_CODE_CHANGE","_number":7,"created":"2019-09-27 16:38:02.000000000","uploader":{"_account_id":28374,"name":"Viktor Michalek","email":"viktor.michalek@ultimum.io","username":"viktormichalek"},"ref":"refs/changes/77/683677/7","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/kolla","ref":"refs/changes/77/683677/7","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/kolla refs/changes/77/683677/7 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/kolla refs/changes/77/683677/7 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/kolla refs/changes/77/683677/7 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/kolla refs/changes/77/683677/7"}}},"commit":{"parents":[{"commit":"543b53ed2512cfb39d1ac234a37a067d3f95e063","subject":"Fluentd: fix missing /var/run/fluentd","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/543b53ed2512cfb39d1ac234a37a067d3f95e063"}]}],"author":{"name":"Viktor Michalek","email":"viktor.michalek@ultimum.io","date":"2019-09-20 16:36:27.000000000","tz":120},"committer":{"name":"Viktor Michalek","email":"viktor.michalek@ultimum.io","date":"2019-09-27 16:37:57.000000000","tz":0},"subject":"Horizon: fix for Debian","message":"Horizon: fix for Debian\n\nChange-Id: I9604ebb4f2a5081fd1209b9841aa997a70a90a11\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/17463dfa11c030ee521347c5231f696b013ef9e8"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/17463dfa11c030ee521347c5231f696b013ef9e8"}]},"branch":"refs/heads/master"},"30e3a57e09fc8f0873fea62e1af20b8bcc790fde":{"kind":"TRIVIAL_REBASE","_number":8,"created":"2019-09-27 16:44:20.000000000","uploader":{"_account_id":24072,"name":"Marcin Juszkiewicz","email":"mjuszkiewicz@redhat.com","username":"hrw"},"ref":"refs/changes/77/683677/8","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/kolla","ref":"refs/changes/77/683677/8","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/kolla refs/changes/77/683677/8 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/kolla refs/changes/77/683677/8 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/kolla refs/changes/77/683677/8 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/kolla refs/changes/77/683677/8"}}},"commit":{"parents":[{"commit":"f9b229841db5d27fe11b95d055d74af289893ef2","subject":"Merge \"ppc64le: Ensure that rabbitmq-server.repo exists before trying sed\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/f9b229841db5d27fe11b95d055d74af289893ef2"}]}],"author":{"name":"Viktor Michalek","email":"viktor.michalek@ultimum.io","date":"2019-09-20 16:36:27.000000000","tz":120},"committer":{"name":"Marcin Juszkiewicz","email":"marcin.juszkiewicz@linaro.org","date":"2019-09-27 16:44:09.000000000","tz":-420},"subject":"Horizon: fix for Debian","message":"Horizon: fix for Debian\n\nChange-Id: I9604ebb4f2a5081fd1209b9841aa997a70a90a11\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/30e3a57e09fc8f0873fea62e1af20b8bcc790fde"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/30e3a57e09fc8f0873fea62e1af20b8bcc790fde"}]},"branch":"refs/heads/master"},"d6b6019a960983abfe639a8cbd2aea63597205a5":{"kind":"REWORK","_number":9,"created":"2019-09-30 14:49:46.000000000","uploader":{"_account_id":28374,"name":"Viktor Michalek","email":"viktor.michalek@ultimum.io","username":"viktormichalek"},"ref":"refs/changes/77/683677/9","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/kolla","ref":"refs/changes/77/683677/9","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/kolla refs/changes/77/683677/9 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/kolla refs/changes/77/683677/9 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/kolla refs/changes/77/683677/9 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/kolla refs/changes/77/683677/9"}}},"commit":{"parents":[{"commit":"f9b229841db5d27fe11b95d055d74af289893ef2","subject":"Merge \"ppc64le: Ensure that rabbitmq-server.repo exists before trying sed\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/f9b229841db5d27fe11b95d055d74af289893ef2"}]}],"author":{"name":"Viktor Michalek","email":"viktor.michalek@ultimum.io","date":"2019-09-20 16:36:27.000000000","tz":120},"committer":{"name":"Viktor Michalek","email":"viktor.michalek@ultimum.io","date":"2019-09-30 14:46:09.000000000","tz":120},"subject":"Horizon: fix for Debian","message":"Horizon: fix for Debian\n\nStein/Buster packages are enabling plugins by default at its\ninstallation time, but we actually know what plugins should be enabled\njust only at the deploy time.\nTherefore we need to install all available plugins and disble them right\nat the image build time.\nPlugins will be enabled according to deployment (kolla-ansible)\nconfiguration at deploy time.\n\nChange-Id: I9604ebb4f2a5081fd1209b9841aa997a70a90a11\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/d6b6019a960983abfe639a8cbd2aea63597205a5"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/d6b6019a960983abfe639a8cbd2aea63597205a5"}]},"branch":"refs/heads/master"},"6f44a3282aeeef90caaba48daae068493e2b8606":{"kind":"REWORK","_number":10,"created":"2019-10-01 03:15:34.000000000","uploader":{"_account_id":24072,"name":"Marcin Juszkiewicz","email":"mjuszkiewicz@redhat.com","username":"hrw"},"ref":"refs/changes/77/683677/10","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/kolla","ref":"refs/changes/77/683677/10","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/kolla refs/changes/77/683677/10 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/kolla refs/changes/77/683677/10 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/kolla refs/changes/77/683677/10 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/kolla refs/changes/77/683677/10"}}},"commit":{"parents":[{"commit":"f9b229841db5d27fe11b95d055d74af289893ef2","subject":"Merge \"ppc64le: Ensure that rabbitmq-server.repo exists before trying sed\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/f9b229841db5d27fe11b95d055d74af289893ef2"}]}],"author":{"name":"Viktor Michalek","email":"viktor.michalek@ultimum.io","date":"2019-09-20 16:36:27.000000000","tz":120},"committer":{"name":"Marcin Juszkiewicz","email":"marcin.juszkiewicz@linaro.org","date":"2019-10-01 03:15:23.000000000","tz":0},"subject":"Horizon: fix for Debian","message":"Horizon: fix for Debian\n\nStein/Buster packages are enabling plugins by default at its\ninstallation time, but we actually know what plugins should be enabled\njust only at the deploy time.\nTherefore we need to install all available plugins and disble them right\nat the image build time.\nPlugins will be enabled according to deployment (kolla-ansible)\nconfiguration at deploy time.\n\nChange-Id: I9604ebb4f2a5081fd1209b9841aa997a70a90a11\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/6f44a3282aeeef90caaba48daae068493e2b8606"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/6f44a3282aeeef90caaba48daae068493e2b8606"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[],"submit_requirements":[]}
