)]}'
{"id":"openstack%2Fkolla~683679","triplet_id":"openstack%2Fkolla~master~I3b17bfdbc68a9a227bb22dc05a02ed6da607a114","project":"openstack/kolla","branch":"master","topic":"680447","hashtags":[],"change_id":"I3b17bfdbc68a9a227bb22dc05a02ed6da607a114","subject":"Neutron: add support to use legacy iptables","status":"MERGED","created":"2019-09-20 18:14:28.000000000","updated":"2019-10-16 12:27:42.000000000","submitted":"2019-10-16 12:22:13.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":3,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"683679-1571228533366-f62c7b6a","meta_rev_id":"6a1cb7cf5df1fd897a1d572130a87445e06e89c4","_number":683679,"virtual_id_number":683679,"owner":{"_account_id":28374,"name":"Viktor Michalek","email":"viktor.michalek@ultimum.io","username":"viktormichalek"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":26768,"name":"Jan Vondra","email":"jan.vondra@ultimum.io","username":"JanVondra"},{"value":0,"date":"2019-10-15 17:33:30.000000000","_account_id":28374,"name":"Viktor Michalek","email":"viktor.michalek@ultimum.io","username":"viktormichalek"},{"value":0,"_account_id":24072,"name":"Marcin Juszkiewicz","email":"mjuszkiewicz@redhat.com","username":"hrw"},{"value":0,"_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2019-10-16 12:22:13.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"date":"2019-10-16 09:32:05.000000000","_account_id":23181,"name":"RDO Third Party CI","email":"dmsimard+rdothirdparty@redhat.com","username":"rdothirdparty","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":24072,"name":"Marcin Juszkiewicz","email":"mjuszkiewicz@redhat.com","username":"hrw"},"all":[{"value":0,"_account_id":26768,"name":"Jan Vondra","email":"jan.vondra@ultimum.io","username":"JanVondra"},{"value":0,"_account_id":28374,"name":"Viktor Michalek","email":"viktor.michalek@ultimum.io","username":"viktormichalek"},{"value":2,"date":"2019-10-15 09:08:43.000000000","_account_id":24072,"name":"Marcin Juszkiewicz","email":"mjuszkiewicz@redhat.com","username":"hrw"},{"value":2,"date":"2019-10-14 16:53:15.000000000","_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":23181,"name":"RDO Third Party CI","email":"dmsimard+rdothirdparty@redhat.com","username":"rdothirdparty","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":24072,"name":"Marcin Juszkiewicz","email":"mjuszkiewicz@redhat.com","username":"hrw"},"all":[{"value":0,"_account_id":26768,"name":"Jan Vondra","email":"jan.vondra@ultimum.io","username":"JanVondra"},{"value":0,"_account_id":28374,"name":"Viktor Michalek","email":"viktor.michalek@ultimum.io","username":"viktormichalek"},{"value":1,"date":"2019-10-15 09:08:43.000000000","_account_id":24072,"name":"Marcin Juszkiewicz","email":"mjuszkiewicz@redhat.com","username":"hrw"},{"value":0,"_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":23181,"name":"RDO Third Party CI","email":"dmsimard+rdothirdparty@redhat.com","username":"rdothirdparty","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Backport-Candidate":{"all":[{"value":0,"_account_id":26768,"name":"Jan Vondra","email":"jan.vondra@ultimum.io","username":"JanVondra"},{"value":0,"_account_id":28374,"name":"Viktor Michalek","email":"viktor.michalek@ultimum.io","username":"viktormichalek"},{"value":0,"_account_id":24072,"name":"Marcin Juszkiewicz","email":"mjuszkiewicz@redhat.com","username":"hrw"},{"value":0,"_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":23181,"name":"RDO Third Party CI","email":"dmsimard+rdothirdparty@redhat.com","username":"rdothirdparty","tags":["SERVICE_USER"]}],"values":{"-1":"Do Not Backport"," 0":"Backport Review Needed","+1":"Should Backport"},"description":"","default_value":0,"optional":true},"Review-Priority":{"all":[{"value":0,"_account_id":26768,"name":"Jan Vondra","email":"jan.vondra@ultimum.io","username":"JanVondra"},{"value":0,"_account_id":28374,"name":"Viktor Michalek","email":"viktor.michalek@ultimum.io","username":"viktormichalek"},{"value":0,"_account_id":24072,"name":"Marcin Juszkiewicz","email":"mjuszkiewicz@redhat.com","username":"hrw"},{"value":0,"_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":23181,"name":"RDO Third Party CI","email":"dmsimard+rdothirdparty@redhat.com","username":"rdothirdparty","tags":["SERVICE_USER"]}],"values":{"-1":"Branch Freeze"," 0":"No Priority","+1":"Important Change","+2":"Gate Blocker Fix / Urgent Change"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":23181,"name":"RDO Third Party CI","email":"dmsimard+rdothirdparty@redhat.com","username":"rdothirdparty","tags":["SERVICE_USER"]},{"_account_id":24072,"name":"Marcin Juszkiewicz","email":"mjuszkiewicz@redhat.com","username":"hrw"},{"_account_id":26768,"name":"Jan Vondra","email":"jan.vondra@ultimum.io","username":"JanVondra"},{"_account_id":28374,"name":"Viktor Michalek","email":"viktor.michalek@ultimum.io","username":"viktormichalek"},{"_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2019-09-23 14:58:28.000000000","updated_by":{"_account_id":26768,"name":"Jan Vondra","email":"jan.vondra@ultimum.io","username":"JanVondra"},"reviewer":{"_account_id":26768,"name":"Jan Vondra","email":"jan.vondra@ultimum.io","username":"JanVondra"},"state":"REVIEWER"},{"updated":"2019-10-14 16:53:15.000000000","updated_by":{"_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"},"reviewer":{"_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"},"state":"REVIEWER"},{"updated":"2019-10-15 09:08:43.000000000","updated_by":{"_account_id":24072,"name":"Marcin Juszkiewicz","email":"mjuszkiewicz@redhat.com","username":"hrw"},"reviewer":{"_account_id":24072,"name":"Marcin Juszkiewicz","email":"mjuszkiewicz@redhat.com","username":"hrw"},"state":"REVIEWER"},{"updated":"2019-10-16 09:32:05.000000000","updated_by":{"_account_id":23181,"name":"RDO Third Party CI","email":"dmsimard+rdothirdparty@redhat.com","username":"rdothirdparty","tags":["SERVICE_USER"]},"reviewer":{"_account_id":23181,"name":"RDO Third Party CI","email":"dmsimard+rdothirdparty@redhat.com","username":"rdothirdparty","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2019-10-16 12:22:13.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"}],"messages":[{"id":"f42e289d32196c686eeb488cfb56d927d047f17e","author":{"_account_id":28374,"name":"Viktor Michalek","email":"viktor.michalek@ultimum.io","username":"viktormichalek"},"date":"2019-09-20 18:14:28.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"9845a9ef0b549a79015c04067f831b29387fedd2","author":{"_account_id":23181,"name":"RDO Third Party CI","email":"dmsimard+rdothirdparty@redhat.com","username":"rdothirdparty","tags":["SERVICE_USER"]},"date":"2019-09-20 18:54:17.000000000","message":"Patch Set 1:\n\nBuild succeeded (check pipeline).\n\n- tripleo-build-containers-rhel-8-master http://logs.rdoproject.org/79/683679/1/openstack-check/tripleo-build-containers-rhel-8-master/75d2711/ : SUCCESS in 37m 43s","accounts_in_message":[],"_revision_number":1},{"id":"e62ae6640600bf2bc5c469607dbcc176cf39a3da","author":{"_account_id":24072,"name":"Marcin Juszkiewicz","email":"mjuszkiewicz@redhat.com","username":"hrw"},"date":"2019-09-20 20:47:16.000000000","message":"Patch Set 1:\n\n(1 comment)","accounts_in_message":[],"_revision_number":1},{"id":"98dea1cf861f32efb79dc1e9d045a78e1c9030fc","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-09-20 22:26:43.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\n\n- kolla-build-centos-binary https://zuul.opendev.org/t/openstack/build/3483aae1fee240f594bc540168031830 : SUCCESS in 53m 54s (non-voting)\n- kolla-build-centos-source https://zuul.opendev.org/t/openstack/build/1cefd31aaebe4aafa1cfe44325637a94 : SUCCESS in 48m 32s\n- kolla-ansible-centos-source https://zuul.opendev.org/t/openstack/build/b5df50975c874d90b3c113e592f154d7 : SUCCESS in 1h 25m 05s\n- kolla-ansible-centos-binary https://zuul.opendev.org/t/openstack/build/3775a6994bca47fd94f60e350759a85f : SUCCESS in 1h 16m 37s (non-voting)\n- kolla-ansible-centos-source-upgrade https://zuul.opendev.org/t/openstack/build/2a9fd397536b47059792b665fa3559f4 : SUCCESS in 1h 35m 47s\n- tripleo-build-containers-centos-7 https://zuul.opendev.org/t/openstack/build/10e1fe011e3a45fcbf80998e22edd4d8 : SUCCESS in 58m 31s\n- kolla-build-debian-source https://zuul.opendev.org/t/openstack/build/f6dfd15f43b94dbfbf8429498b60597d : SUCCESS in 37m 33s (non-voting)\n- kolla-build-debian-binary https://zuul.opendev.org/t/openstack/build/404b4a2d93464c1b8fce4ef6ac3305c6 : FAILURE in 31m 32s (non-voting)\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/51fa2ae11b7746fd98acfe2cabfe8ea4 : SUCCESS in 4m 41s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/7b5aa1c8cc624019a3141147b75052c6 : SUCCESS in 4m 40s\n- openstack-tox-py27 https://zuul.opendev.org/t/openstack/build/8efd1e8274864a35a359d7c49fc6765d : SUCCESS in 4m 29s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/d1fc64f3b6054180acb911cf0849168e : SUCCESS in 3m 53s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/f680011c4b074a2e8a691ca06c32d3d1 : SUCCESS in 3m 37s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/2c0b6564cb2d44168bb99925b0da8fd0 : SUCCESS in 8m 54s\n- kolla-build-ubuntu-binary https://zuul.opendev.org/t/openstack/build/abf9951795334a118b7bd1a7a0843ee4 : SUCCESS in 35m 29s (non-voting)\n- kolla-build-ubuntu-source https://zuul.opendev.org/t/openstack/build/33ad0da98f494d7b89ab548ff8ef94a3 : SUCCESS in 44m 49s\n- kolla-ansible-ubuntu-source https://zuul.opendev.org/t/openstack/build/932004ff0f504cbf9fcc0f23cc004457 : SUCCESS in 51m 07s\n- kolla-ansible-ubuntu-binary https://zuul.opendev.org/t/openstack/build/1c69ab980f3b4745911ba0043e53e843 : SUCCESS in 58m 20s (non-voting)\n- kolla-ansible-ubuntu-source-upgrade https://zuul.opendev.org/t/openstack/build/0755861ac7d24955b966a59ccd5cde0c : SUCCESS in 59m 49s","accounts_in_message":[],"_revision_number":1},{"id":"94f2395addfe103a9ab45436790d454a5a5d0282","author":{"_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"},"date":"2019-09-23 09:20:07.000000000","message":"Patch Set 1:\n\n(1 comment)","accounts_in_message":[],"_revision_number":1},{"id":"caeca3be3566818fd2619b48d1d791da5586fc50","author":{"_account_id":26768,"name":"Jan Vondra","email":"jan.vondra@ultimum.io","username":"JanVondra"},"date":"2019-09-23 14:58:28.000000000","message":"Patch Set 1:\n\n\u003e (1 comment)\n \u003e (1 comment)\nHi,\nactually there is no issue with 4.19 in Buster, but in older kernels used by host OS.\nWe enountered the problem while upgrading deployment with host system Ubuntu 16.04 Xenial from our custom Queens debian kolla images (based on stretch) to Stein (those images we are trying to push to master in other topic - based on Buster). This causes problems when using a hybrid security group driver since Buster is using iptables-nft by default and Xenial kernel has no support for it resulting in port binding failure.\nSo basically this issue occurs \"only\" when upgrading kolla deployment without upgrading the host OS and I suppose that similar behaviour could happen even in other distros.","accounts_in_message":[],"_revision_number":1},{"id":"b1abd563574f024e377f436d55dfeecbd1f26f8c","author":{"_account_id":28374,"name":"Viktor Michalek","email":"viktor.michalek@ultimum.io","username":"viktormichalek"},"date":"2019-09-24 15:56:34.000000000","message":"Patch Set 2: Patch Set 1 was rebased","accounts_in_message":[],"_revision_number":2},{"id":"47f06262a6132d2d4208faeb1325d7c343084875","author":{"_account_id":23181,"name":"RDO Third Party CI","email":"dmsimard+rdothirdparty@redhat.com","username":"rdothirdparty","tags":["SERVICE_USER"]},"date":"2019-09-24 16:38:21.000000000","message":"Patch Set 2:\n\nBuild succeeded (check pipeline).\n\n- tripleo-build-containers-rhel-8-master http://logs.rdoproject.org/79/683679/2/openstack-check/tripleo-build-containers-rhel-8-master/6cf44f5/ : SUCCESS in 39m 53s","accounts_in_message":[],"_revision_number":2},{"id":"54357ef8d3a864f60556bb44df18dc332986ba46","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-09-24 18:04:48.000000000","message":"Patch Set 2: Verified+1\n\nBuild succeeded (check pipeline).\n\n- kolla-build-centos-binary https://zuul.opendev.org/t/openstack/build/1ceb2ccc75a64b77b8780d79fbeb9363 : SUCCESS in 42m 37s (non-voting)\n- kolla-build-centos-source https://zuul.opendev.org/t/openstack/build/f5ba2952e5c542ba803744315490ccbe : SUCCESS in 1h 16m 33s\n- kolla-ansible-centos-source https://zuul.opendev.org/t/openstack/build/e37b91ad7e964af7acec9f5e8656af58 : SUCCESS in 1h 31m 58s\n- kolla-ansible-centos-binary https://zuul.opendev.org/t/openstack/build/f1e8b3523995404fba34c6661987ebc8 : SUCCESS in 1h 15m 26s (non-voting)\n- kolla-ansible-centos-source-upgrade https://zuul.opendev.org/t/openstack/build/0e22a01933744d90b2a87e95bf98582f : SUCCESS in 1h 56m 59s\n- tripleo-build-containers-centos-7 https://zuul.opendev.org/t/openstack/build/a694ead96ef04b97b2ee5a3987ce00e2 : SUCCESS in 1h 03m 29s\n- kolla-build-debian-source https://zuul.opendev.org/t/openstack/build/6c5f5bcc5c21497ab292eff9370c4977 : SUCCESS in 32m 27s (non-voting)\n- kolla-build-debian-binary https://zuul.opendev.org/t/openstack/build/3c0c3664d90b46a8aff4b8599acc5446 : FAILURE in 22m 50s (non-voting)\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/b1845f87e2ca493c94b8bd4c0652a060 : SUCCESS in 5m 13s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/4b6723078a12432f853a4935d30bd788 : SUCCESS in 4m 13s\n- openstack-tox-py27 https://zuul.opendev.org/t/openstack/build/5936cd0319eb47d68b8b47b58476c728 : SUCCESS in 3m 30s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/3603ed6bf0294498ba23dbd5bfc4be4b : SUCCESS in 3m 21s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/5b6bcdd2f8124c1bbac667540ef63246 : SUCCESS in 8m 09s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/82f1e4ad8c044a23afdc75c520a0dc4c : SUCCESS in 4m 38s\n- kolla-build-ubuntu-binary https://zuul.opendev.org/t/openstack/build/3c6ec7de5fc344a8af7b3a3b84ff1418 : SUCCESS in 35m 15s (non-voting)\n- kolla-build-ubuntu-source https://zuul.opendev.org/t/openstack/build/c7e7ac4491c84f18955da762fcf608bc : SUCCESS in 42m 08s\n- kolla-ansible-ubuntu-source https://zuul.opendev.org/t/openstack/build/129ad2bcb15845b49f713892da5668da : SUCCESS in 48m 17s\n- kolla-ansible-ubuntu-binary https://zuul.opendev.org/t/openstack/build/65c7926ebf0c4ef18cd9648394dbe496 : SUCCESS in 1h 04m 11s (non-voting)\n- kolla-ansible-ubuntu-source-upgrade https://zuul.opendev.org/t/openstack/build/bfb24450f4a34bb887237ed93de119ca : SUCCESS in 1h 39m 21s","accounts_in_message":[],"_revision_number":2},{"id":"082753eed2d2f5de57f6ed8e16d421b6a5769397","author":{"_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"},"date":"2019-09-25 07:06:26.000000000","message":"Patch Set 2: Code-Review-1\n\n\u003e So basically this issue occurs \"only\" when upgrading kolla\n \u003e deployment without upgrading the host OS and I suppose that similar\n \u003e behaviour could happen even in other distros.\n\nThanks for the explanation, Jan. Then the problem with this patch is that this is not undoable by setting legacy back to \u0027no\u0027. Let\u0027s make it and also expand the commit message to included the explanation. In-code comment would be welcome as well.","accounts_in_message":[],"_revision_number":2},{"id":"2a0e71c96292234ec88f76c9c67e5075b82ebc3b","author":{"_account_id":24072,"name":"Marcin Juszkiewicz","email":"mjuszkiewicz@redhat.com","username":"hrw"},"date":"2019-09-27 17:17:31.000000000","message":"Patch Set 2:\n\nOnes commit message and in-code comment will be improved I will +2 it.","accounts_in_message":[],"_revision_number":2},{"id":"9824196015ecc5b76e84ddd7e4d605cf6f247dda","author":{"_account_id":28374,"name":"Viktor Michalek","email":"viktor.michalek@ultimum.io","username":"viktormichalek"},"date":"2019-09-27 17:36:54.000000000","message":"Patch Set 3: Published edit on patch set 2.","accounts_in_message":[],"_revision_number":3},{"id":"557b632aae75af37472e5452ee8b3a1df7c3b83e","author":{"_account_id":23181,"name":"RDO Third Party CI","email":"dmsimard+rdothirdparty@redhat.com","username":"rdothirdparty","tags":["SERVICE_USER"]},"date":"2019-09-27 17:37:16.000000000","message":"Patch Set 3:\n\nMerge Failed.\n\nThis change or one of its cross-repo dependencies was unable to be automatically merged with the current state of its repository. Please rebase the change and upload a new patchset.","accounts_in_message":[],"_revision_number":3},{"id":"23778e88307ad3588c453d90511aa142fd9372da","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-09-27 17:38:46.000000000","message":"Patch Set 3: Verified-1\n\nMerge Failed.\n\nThis change or one of its cross-repo dependencies was unable to be automatically merged with the current state of its repository. Please rebase the change and upload a new patchset.","accounts_in_message":[],"_revision_number":3},{"id":"aadf9bc5a180ac9bb7c7ffe6cdf644ac3163c681","author":{"_account_id":28374,"name":"Viktor Michalek","email":"viktor.michalek@ultimum.io","username":"viktormichalek"},"date":"2019-09-27 17:46:21.000000000","message":"Patch Set 4: Commit message was updated.","accounts_in_message":[],"_revision_number":4},{"id":"6edbc7b0ad4ec324265ac0d4ccf4913db73acdcc","author":{"_account_id":28374,"name":"Viktor Michalek","email":"viktor.michalek@ultimum.io","username":"viktormichalek"},"date":"2019-09-27 17:46:28.000000000","message":"Patch Set 5: Patch Set 4 was rebased","accounts_in_message":[],"_revision_number":5},{"id":"6cc51587af76418544c86459d2f9cff890fd5be5","author":{"_account_id":23181,"name":"RDO Third Party CI","email":"dmsimard+rdothirdparty@redhat.com","username":"rdothirdparty","tags":["SERVICE_USER"]},"date":"2019-09-27 18:26:10.000000000","message":"Patch Set 5:\n\nBuild succeeded (check pipeline).\n\n- tripleo-build-containers-rhel-8-master http://logs.rdoproject.org/79/683679/5/openstack-check/tripleo-build-containers-rhel-8-master/5d50c22/ : SUCCESS in 38m 45s","accounts_in_message":[],"_revision_number":5},{"id":"49d9e397b14710ffbbc9d254da2c33415d052644","author":{"_account_id":28374,"name":"Viktor Michalek","email":"viktor.michalek@ultimum.io","username":"viktormichalek"},"date":"2019-09-27 19:16:36.000000000","message":"Patch Set 6: Published edit on patch set 5.","accounts_in_message":[],"_revision_number":6},{"id":"ccf263aafd7546832ffc0ad4c14b346a9be50842","author":{"_account_id":23181,"name":"RDO Third Party CI","email":"dmsimard+rdothirdparty@redhat.com","username":"rdothirdparty","tags":["SERVICE_USER"]},"date":"2019-09-27 19:55:27.000000000","message":"Patch Set 6:\n\nBuild succeeded (check pipeline).\n\n- tripleo-build-containers-rhel-8-master http://logs.rdoproject.org/79/683679/6/openstack-check/tripleo-build-containers-rhel-8-master/7a6458d/ : SUCCESS in 37m 03s","accounts_in_message":[],"_revision_number":6},{"id":"f07407d43192be6094331b22374ba7dc381bbae6","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-09-27 21:20:11.000000000","message":"Patch Set 6: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttp://docs.openstack.org/infra/manual/developers.html#automated-testing\n\n\n- kolla-build-centos-binary https://zuul.opendev.org/t/openstack/build/b454a82dc8944444a8573c5be8d6892a : SUCCESS in 57m 40s (non-voting)\n- kolla-build-centos-source https://zuul.opendev.org/t/openstack/build/34ccaeb4cd0940bc8258e24d3808b5ee : SUCCESS in 55m 54s\n- kolla-ansible-centos-source https://zuul.opendev.org/t/openstack/build/d3f2e3250e15442da77168e6240a6ac8 : FAILURE in 1h 08m 42s\n- kolla-ansible-centos-binary https://zuul.opendev.org/t/openstack/build/917224702b8f4628849b4ddcc68152e9 : FAILURE in 57m 25s (non-voting)\n- kolla-ansible-centos-source-upgrade https://zuul.opendev.org/t/openstack/build/a91ab1ca790c4074902ba52f3f164ce6 : FAILURE in 1h 56m 06s\n- tripleo-build-containers-centos-7 https://zuul.opendev.org/t/openstack/build/29e86f24fbd343f9b8f63c0de5a69c23 : SUCCESS in 1h 08m 08s\n- kolla-build-debian-source https://zuul.opendev.org/t/openstack/build/6050b86a8af84c67b395748399a0f50a : SUCCESS in 33m 29s (non-voting)\n- kolla-build-debian-binary https://zuul.opendev.org/t/openstack/build/b68d82894a13495397c625e912ffe33b : SUCCESS in 27m 07s (non-voting)\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/279aaee5377b440ba254e124f0ce6b47 : SUCCESS in 2m 57s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/2f861d452bf043ebb033b931c5f9b023 : SUCCESS in 4m 09s\n- openstack-tox-py27 https://zuul.opendev.org/t/openstack/build/9092f7d8bef74a6f9d9987e4e10ad7d4 : SUCCESS in 4m 58s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/e02efe5fdb3e48ca89d15d9e2c156ff0 : SUCCESS in 3m 11s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/b767f39512e94dbfab6bab8b4c470f03 : SUCCESS in 6m 18s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/5be398ec30db4a72b0d81e8f091711ed : SUCCESS in 8m 40s\n- kolla-build-ubuntu-binary https://zuul.opendev.org/t/openstack/build/14467edf68d6406a8abb8f4e43b7705c : SUCCESS in 23m 29s (non-voting)\n- kolla-build-ubuntu-source https://zuul.opendev.org/t/openstack/build/703c7493cf0842f788c1ae8b91c37ec9 : SUCCESS in 56m 17s\n- kolla-ansible-ubuntu-source https://zuul.opendev.org/t/openstack/build/252804abd624410fbc9d99b7ab74a24e : FAILURE in 41m 59s\n- kolla-ansible-ubuntu-binary https://zuul.opendev.org/t/openstack/build/6260d7b4ca394ca68a3c9d354361b0be : FAILURE in 40m 31s (non-voting)\n- kolla-ansible-ubuntu-source-upgrade https://zuul.opendev.org/t/openstack/build/6de5223a986446e8b63d6275ba370e23 : FAILURE in 56m 53s","accounts_in_message":[],"_revision_number":6},{"id":"a58480ecb406cb419d1c79f04d697d9366238b1c","author":{"_account_id":28374,"name":"Viktor Michalek","email":"viktor.michalek@ultimum.io","username":"viktormichalek"},"date":"2019-09-28 17:45:00.000000000","message":"Patch Set 7: Patch Set 6 was rebased","accounts_in_message":[],"_revision_number":7},{"id":"e9ec4319befd7ab0ad4f81e2885903b2e1f3a1c7","author":{"_account_id":23181,"name":"RDO Third Party CI","email":"dmsimard+rdothirdparty@redhat.com","username":"rdothirdparty","tags":["SERVICE_USER"]},"date":"2019-09-28 18:21:57.000000000","message":"Patch Set 7:\n\nBuild succeeded (check pipeline).\n\n- tripleo-build-containers-rhel-8-master http://logs.rdoproject.org/79/683679/7/openstack-check/tripleo-build-containers-rhel-8-master/785c6d2/ : SUCCESS in 35m 45s","accounts_in_message":[],"_revision_number":7},{"id":"92a86133685018926a1d69ba884bd9653a6d7c6d","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-09-28 19:03:39.000000000","message":"Patch Set 7:\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttp://docs.openstack.org/infra/manual/developers.html#automated-testing\n\n\n- kolla-build-centos-binary https://zuul.opendev.org/t/openstack/build/8e9d1c1c2aa3457484198056dddee878 : FAILURE in 38m 56s (non-voting)\n- kolla-build-centos-source https://zuul.opendev.org/t/openstack/build/291ffb3bfcc44701a99347ee6798f3df : FAILURE in 1h 01m 47s\n- kolla-ansible-centos-source https://zuul.opendev.org/t/openstack/build/4fd54075be1240a8bd3644ebcaddbd64 : FAILURE in 49m 06s\n- kolla-ansible-centos-binary https://zuul.opendev.org/t/openstack/build/9bee5e64b9e64723af58c16a2653b193 : FAILURE in 47m 31s (non-voting)\n- kolla-ansible-centos-source-upgrade https://zuul.opendev.org/t/openstack/build/ae157dfe8cd841b689f1b7ad3b05dc47 : FAILURE in 1h 11m 59s\n- tripleo-build-containers-centos-7 https://zuul.opendev.org/t/openstack/build/60f5c58fe8304c72a4ec31e7f4ff042a : SUCCESS in 47m 18s\n- kolla-build-debian-source https://zuul.opendev.org/t/openstack/build/6796f87d67524dacb62682424d85f796 : SUCCESS in 25m 00s (non-voting)\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/aba967b9974b40ba86b5d158d8009a7a : SUCCESS in 4m 25s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/81d2ab88e7d64cb582e0a0fa127924c9 : SUCCESS in 3m 43s\n- openstack-tox-py27 https://zuul.opendev.org/t/openstack/build/49a6f6ae0d6f493b9acba5146863b9e5 : SUCCESS in 3m 15s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/6126479e78894ac1bfaeb9d15935c959 : SUCCESS in 4m 19s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/23f0888ee36e44f2a9d4a2d5bf8b89b4 : SUCCESS in 3m 35s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/0d6038968013467fb753e76a4625b804 : SUCCESS in 6m 40s\n- kolla-build-ubuntu-binary https://zuul.opendev.org/t/openstack/build/da0e714581ce4de6a7fa43abb12b36e7 : SUCCESS in 24m 43s (non-voting)\n- kolla-build-ubuntu-source https://zuul.opendev.org/t/openstack/build/986a9a3162614ae3963691ccd4d1ff87 : SUCCESS in 32m 20s\n- kolla-ansible-ubuntu-source https://zuul.opendev.org/t/openstack/build/4e6984647e54460f8b3606fd74a65c4f : FAILURE in 44m 35s\n- kolla-ansible-ubuntu-binary https://zuul.opendev.org/t/openstack/build/def7f059073545c29ac08dbb9536009b : FAILURE in 29m 43s (non-voting)\n- kolla-ansible-ubuntu-source-upgrade https://zuul.opendev.org/t/openstack/build/5d4d23522042462080f2d1cb06a33845 : FAILURE in 1h 17m 01s","accounts_in_message":[],"_revision_number":7},{"id":"34a0bc6ad3e7d278b82eae28aa5b1f3fdf30685c","author":{"_account_id":28374,"name":"Viktor Michalek","email":"viktor.michalek@ultimum.io","username":"viktormichalek"},"date":"2019-09-28 22:22:32.000000000","message":"Patch Set 7:\n\n(1 comment)\n\nrecheck","accounts_in_message":[],"_revision_number":7},{"id":"578167b7b0720e2cf374e83529d7a705f0ff8a2a","author":{"_account_id":28374,"name":"Viktor Michalek","email":"viktor.michalek@ultimum.io","username":"viktormichalek"},"date":"2019-09-29 13:58:45.000000000","message":"Patch Set 8: Patch Set 7 was rebased","accounts_in_message":[],"_revision_number":8},{"id":"ae5e329e642a2e3e16ee8095df5b322fc3953444","author":{"_account_id":23181,"name":"RDO Third Party CI","email":"dmsimard+rdothirdparty@redhat.com","username":"rdothirdparty","tags":["SERVICE_USER"]},"date":"2019-09-29 14:43:06.000000000","message":"Patch Set 8:\n\nBuild succeeded (check pipeline).\n\n- tripleo-build-containers-rhel-8-master http://logs.rdoproject.org/79/683679/8/openstack-check/tripleo-build-containers-rhel-8-master/2964312/ : SUCCESS in 43m 34s","accounts_in_message":[],"_revision_number":8},{"id":"f5347903e1b9ea14bbbe908d438a465ca40b513d","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-09-29 15:56:25.000000000","message":"Patch Set 8: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttp://docs.openstack.org/infra/manual/developers.html#automated-testing\n\n\n- kolla-build-centos-binary https://zuul.opendev.org/t/openstack/build/7e2669b83ef848208c3cb826aafb3c33 : SUCCESS in 38m 42s (non-voting)\n- kolla-build-centos-source https://zuul.opendev.org/t/openstack/build/f4b5a44a96d042778a75de41717a0eab : SUCCESS in 1h 16m 00s\n- kolla-ansible-centos-source https://zuul.opendev.org/t/openstack/build/bd4d1e1bf85040518c1aee9bc28544d5 : FAILURE in 1h 02m 36s\n- kolla-ansible-centos-binary https://zuul.opendev.org/t/openstack/build/abc55130b5624661977bc4c0a799b08e : FAILURE in 1h 20m 51s (non-voting)\n- kolla-ansible-centos-source-upgrade https://zuul.opendev.org/t/openstack/build/a1bee5bbbc454457b319fd10b5899306 : FAILURE in 1h 47m 29s\n- tripleo-build-containers-centos-7 https://zuul.opendev.org/t/openstack/build/63b326dcdfd34b6d962e818adde87e78 : SUCCESS in 1h 15m 50s\n- kolla-build-debian-source https://zuul.opendev.org/t/openstack/build/f293e9d8117e4e85a302bbb93a4fb8ea : SUCCESS in 35m 35s (non-voting)\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/1bf7e5356f9844c8814f76b7276baa40 : SUCCESS in 4m 54s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/7f8b69d64be74df59e0d04408e158aa8 : SUCCESS in 5m 36s\n- openstack-tox-py27 https://zuul.opendev.org/t/openstack/build/a56e6d3cc57a44fe9c5843e16c9fe5c1 : SUCCESS in 5m 16s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/b06c1435ef644129b1de87025606fe4c : SUCCESS in 5m 23s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/ca7898ff11d145dbba66d4bed8dd58ea : SUCCESS in 4m 49s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/9705f64b536b4a68afc4ef525957433f : SUCCESS in 8m 30s\n- kolla-build-ubuntu-binary https://zuul.opendev.org/t/openstack/build/ce8421b5343d49448fbc9fa503e89840 : SUCCESS in 1h 08m 16s (non-voting)\n- kolla-build-ubuntu-source https://zuul.opendev.org/t/openstack/build/d267c14c3fc8458fb5ffe8fa7de6bb84 : SUCCESS in 1h 05m 26s\n- kolla-ansible-ubuntu-source https://zuul.opendev.org/t/openstack/build/f9ff1491ebd74d1c8ac5ff9043ec0dd7 : FAILURE in 1h 14m 37s\n- kolla-ansible-ubuntu-binary https://zuul.opendev.org/t/openstack/build/3d27c2a0e2584a4db837f7bbdb330369 : FAILURE in 34m 19s (non-voting)\n- kolla-ansible-ubuntu-source-upgrade https://zuul.opendev.org/t/openstack/build/9c5cd3f987274dff9c5e5249321b24a6 : FAILURE in 1h 39m 18s","accounts_in_message":[],"_revision_number":8},{"id":"76947efa901a45b5dd0f8a804f6eba83310520fa","author":{"_account_id":28374,"name":"Viktor Michalek","email":"viktor.michalek@ultimum.io","username":"viktormichalek"},"date":"2019-09-30 06:40:52.000000000","message":"Patch Set 8:\n\nrecheck","accounts_in_message":[],"_revision_number":8},{"id":"4ff22d1df05f98012ff8e64099bda50026166d49","author":{"_account_id":23181,"name":"RDO Third Party CI","email":"dmsimard+rdothirdparty@redhat.com","username":"rdothirdparty","tags":["SERVICE_USER"]},"date":"2019-09-30 07:20:08.000000000","message":"Patch Set 8:\n\nBuild succeeded (check pipeline).\n\n- tripleo-build-containers-rhel-8-master http://logs.rdoproject.org/79/683679/8/openstack-check/tripleo-build-containers-rhel-8-master/a1fb84c/ : SUCCESS in 37m 54s","accounts_in_message":[],"_revision_number":8},{"id":"f1582bf46f58f713596d480f0bb3344a7aaf53e0","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-09-30 08:23:46.000000000","message":"Patch Set 8:\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttp://docs.openstack.org/infra/manual/developers.html#automated-testing\n\n\n- kolla-build-centos-binary https://zuul.opendev.org/t/openstack/build/832d13cb9a474d83bd86a971659f3f06 : SUCCESS in 44m 19s (non-voting)\n- kolla-build-centos-source https://zuul.opendev.org/t/openstack/build/10cc31abcec644e7ac452db896e182f1 : SUCCESS in 1h 25m 06s\n- kolla-ansible-centos-source https://zuul.opendev.org/t/openstack/build/e51996e20e824749976de6612a7cc988 : FAILURE in 1h 21m 21s\n- kolla-ansible-centos-binary https://zuul.opendev.org/t/openstack/build/301d07e661664f1b97e4e95b2d5acf3e : FAILURE in 54m 18s (non-voting)\n- kolla-ansible-centos-source-upgrade https://zuul.opendev.org/t/openstack/build/ebf056166280450bb3f9c64445633563 : FAILURE in 1h 21m 02s\n- tripleo-build-containers-centos-7 https://zuul.opendev.org/t/openstack/build/090d4917680c482f8d921fde340f57fe : SUCCESS in 57m 49s\n- kolla-build-debian-source https://zuul.opendev.org/t/openstack/build/6d835ea7b7ae477cac231dda910bad26 : SUCCESS in 30m 04s (non-voting)\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/6b0e9922d87f4397bf15f0839e86c7f1 : SUCCESS in 5m 13s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/5f83ce45a8a54fa184e440b98a2b03b5 : SUCCESS in 5m 50s\n- openstack-tox-py27 https://zuul.opendev.org/t/openstack/build/3d0a5c64e076471baff1fe2bb76d69ec : SUCCESS in 3m 25s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/45cbad6ee50846f5abbddda4f8ed1ef0 : SUCCESS in 3m 40s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/d4141e86424444579fe543d6edf4f384 : SUCCESS in 3m 52s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/1c92364ea50d48e98b7936c1ae729350 : SUCCESS in 8m 44s\n- kolla-build-ubuntu-binary https://zuul.opendev.org/t/openstack/build/29d18a6534124cbba6e27e38fb0f6bd8 : SUCCESS in 1h 12m 35s (non-voting)\n- kolla-build-ubuntu-source https://zuul.opendev.org/t/openstack/build/7302883c5aa944caab52b32929091bb6 : SUCCESS in 1h 12m 36s\n- kolla-ansible-ubuntu-source https://zuul.opendev.org/t/openstack/build/417864ea50b74cdfb390a9134db26435 : FAILURE in 48m 46s\n- kolla-ansible-ubuntu-binary https://zuul.opendev.org/t/openstack/build/ad40aeaa58e34ffcaaaf6789225c52dd : FAILURE in 45m 58s (non-voting)\n- kolla-ansible-ubuntu-source-upgrade https://zuul.opendev.org/t/openstack/build/0ed22d4b47974768bfd2082a5bca0062 : FAILURE in 1h 18m 22s","accounts_in_message":[],"_revision_number":8},{"id":"0ce554cb573c08d6406ee35f94ea866215596d16","author":{"_account_id":24072,"name":"Marcin Juszkiewicz","email":"mjuszkiewicz@redhat.com","username":"hrw"},"date":"2019-10-01 06:49:12.000000000","message":"Patch Set 9: Commit message was updated.","accounts_in_message":[],"_revision_number":9},{"id":"ddf7472727a9468d12f52d3916a2e16335a6957b","author":{"_account_id":24072,"name":"Marcin Juszkiewicz","email":"mjuszkiewicz@redhat.com","username":"hrw"},"date":"2019-10-01 06:49:19.000000000","message":"Patch Set 9: Code-Review+2","accounts_in_message":[],"_revision_number":9},{"id":"ebfcfa35643110958d587d484803f3c5b23bd56d","author":{"_account_id":23181,"name":"RDO Third Party CI","email":"dmsimard+rdothirdparty@redhat.com","username":"rdothirdparty","tags":["SERVICE_USER"]},"date":"2019-10-01 07:27:50.000000000","message":"Patch Set 9:\n\nBuild succeeded (check pipeline).\n\n- tripleo-build-containers-rhel-8-master http://logs.rdoproject.org/79/683679/9/openstack-check/tripleo-build-containers-rhel-8-master/5a17c30/ : SUCCESS in 37m 48s","accounts_in_message":[],"_revision_number":9},{"id":"898d9255add9bfa48538b8bace736c1388c52903","author":{"_account_id":24072,"name":"Marcin Juszkiewicz","email":"mjuszkiewicz@redhat.com","username":"hrw"},"date":"2019-10-01 07:55:09.000000000","message":"Removed Code-Review+2 by Marcin Juszkiewicz \u003cmarcin.juszkiewicz@linaro.org\u003e\n","accounts_in_message":[],"_revision_number":9},{"id":"77782b2fc5d4c864017c8248e2a6691c1df56e04","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-10-01 08:21:50.000000000","message":"Patch Set 9: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttp://docs.openstack.org/infra/manual/developers.html#automated-testing\n\n\n- kolla-build-centos-binary https://zuul.opendev.org/t/openstack/build/37ecea1f3c324e4f8231701a3080acce : SUCCESS in 44m 46s (non-voting)\n- kolla-build-centos-source https://zuul.opendev.org/t/openstack/build/99bd301ab358470aa4ba6974f0d92e6a : SUCCESS in 56m 10s\n- kolla-ansible-centos-source https://zuul.opendev.org/t/openstack/build/1dc1e29562354d8d9cecd29f50d7e7ef : FAILURE in 1h 06m 23s\n- kolla-ansible-centos-binary https://zuul.opendev.org/t/openstack/build/2d5554b270974a4f89f0fdccf536a7f2 : FAILURE in 1h 10m 07s (non-voting)\n- kolla-ansible-centos-source-upgrade https://zuul.opendev.org/t/openstack/build/d2c96511c92443d6bfa688dd848eeef3 : FAILURE in 1h 18m 34s\n- tripleo-build-containers-centos-7 https://zuul.opendev.org/t/openstack/build/eb273174fd1a4dac9538c8ca66c6166b : SUCCESS in 56m 00s\n- kolla-build-debian-source https://zuul.opendev.org/t/openstack/build/31edf29cc0984b6f8f2732c3c6282c5d : SUCCESS in 28m 06s (non-voting)\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/8e596c1a3c9448f080825378fa3c5422 : SUCCESS in 3m 12s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/1d85ded7cce64804b002f8c7ef73268a : SUCCESS in 4m 10s\n- openstack-tox-py27 https://zuul.opendev.org/t/openstack/build/f2026928b81e4dd5af91de4293805b45 : SUCCESS in 4m 07s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/bbe37f4c03dd4ee8865f2a212be76a54 : SUCCESS in 5m 02s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/ddb77a24cbda4a03b1aa6fc625e41a2a : SUCCESS in 3m 27s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/d05279b94efd4018a9f678f31b682548 : SUCCESS in 4m 54s\n- kolla-build-ubuntu-binary https://zuul.opendev.org/t/openstack/build/fa35cc647c3d4be7b9787ec5b9bd347f : SUCCESS in 32m 35s (non-voting)\n- kolla-build-ubuntu-source https://zuul.opendev.org/t/openstack/build/0f4337e347ca4938b8f2dcfb1becdc35 : SUCCESS in 37m 19s\n- kolla-ansible-ubuntu-source https://zuul.opendev.org/t/openstack/build/bc4f53330005413885d1914834301bde : FAILURE in 42m 40s\n- kolla-ansible-ubuntu-binary https://zuul.opendev.org/t/openstack/build/a8fb06f5c40c4d1296d5918c00e4c3bf : FAILURE in 44m 54s (non-voting)\n- kolla-ansible-ubuntu-source-upgrade https://zuul.opendev.org/t/openstack/build/0b9a1d639aee45659c5202237e093f31 : FAILURE in 1h 02m 39s","accounts_in_message":[],"_revision_number":9},{"id":"8abf81edf860291bd450c3b2ec60ba05d579ce36","author":{"_account_id":28374,"name":"Viktor Michalek","email":"viktor.michalek@ultimum.io","username":"viktormichalek"},"date":"2019-10-01 13:26:12.000000000","message":"Patch Set 10: Commit message was updated.","accounts_in_message":[],"_revision_number":10},{"id":"8ee91fb1a233e03dc7ce4c0215afc568a1745b98","author":{"_account_id":23181,"name":"RDO Third Party CI","email":"dmsimard+rdothirdparty@redhat.com","username":"rdothirdparty","tags":["SERVICE_USER"]},"date":"2019-10-01 14:05:46.000000000","message":"Patch Set 10:\n\nBuild succeeded (check pipeline).\n\n- tripleo-build-containers-rhel-8-master http://logs.rdoproject.org/79/683679/10/openstack-check/tripleo-build-containers-rhel-8-master/71b7e7b/ : SUCCESS in 37m 55s","accounts_in_message":[],"_revision_number":10},{"id":"ddbcc8fedf301988c7cd31aa853f44f43925ccd2","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-10-01 14:59:31.000000000","message":"Patch Set 10: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttp://docs.openstack.org/infra/manual/developers.html#automated-testing\n\n\n- kolla-build-centos-binary https://zuul.opendev.org/t/openstack/build/10d5eae344fd433980a1af0a1e3f750f : SUCCESS in 47m 49s (non-voting)\n- kolla-build-centos-source https://zuul.opendev.org/t/openstack/build/e74504d489814df8b2fb96257f28bb52 : SUCCESS in 48m 40s\n- kolla-ansible-centos-source https://zuul.opendev.org/t/openstack/build/a60ac58fbb3a4030b9c98bbf8f7be08e : FAILURE in 1h 00m 46s\n- kolla-ansible-centos-binary https://zuul.opendev.org/t/openstack/build/4ac34245b9104fef906e05cf6ea06435 : FAILURE in 59m 39s (non-voting)\n- kolla-ansible-centos-source-upgrade https://zuul.opendev.org/t/openstack/build/596594217d0d4d628bdbe25d227397bf : FAILURE in 1h 27m 48s\n- tripleo-build-containers-centos-7 https://zuul.opendev.org/t/openstack/build/1165e7ff5937450087ef18c0675c54d5 : SUCCESS in 1h 03m 29s\n- kolla-build-debian-source https://zuul.opendev.org/t/openstack/build/9070df1d8dc5457e81956ed19f1d7521 : SUCCESS in 24m 15s (non-voting)\n- kolla-build-debian-binary https://zuul.opendev.org/t/openstack/build/f8aedc4940dd43edae285b48f8cf9482 : SUCCESS in 16m 43s (non-voting)\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/e149c07d9b854d5aa31e5246f33cc4de : SUCCESS in 3m 20s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/6a588f3dc1904b5eadd270d3e47bc82e : SUCCESS in 3m 33s\n- openstack-tox-py27 https://zuul.opendev.org/t/openstack/build/44fac37429164cdca3b9108c227db924 : SUCCESS in 4m 02s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/c049dd30509f4221a704240e66f0a182 : SUCCESS in 3m 08s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/21bfcf620176436ab26c8a94592e4e2d : SUCCESS in 5m 02s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/94a0a135cd7b4deaac5abf3f82960e5c : SUCCESS in 5m 01s\n- kolla-build-ubuntu-binary https://zuul.opendev.org/t/openstack/build/f55a54f54af241a7a3e70bec375b0a65 : SUCCESS in 32m 09s (non-voting)\n- kolla-build-ubuntu-source https://zuul.opendev.org/t/openstack/build/ec40f7a00763405595c8b5cc060fc869 : SUCCESS in 36m 52s\n- kolla-ansible-ubuntu-source https://zuul.opendev.org/t/openstack/build/11cb48876b3348e4b362461a94eeb555 : FAILURE in 38m 25s\n- kolla-ansible-ubuntu-binary https://zuul.opendev.org/t/openstack/build/0157561d99a6455ba02d0f98cea6c992 : FAILURE in 31m 59s (non-voting)\n- kolla-ansible-ubuntu-source-upgrade https://zuul.opendev.org/t/openstack/build/79e924080c6343c5b0e11fda9b2afdbe : FAILURE in 53m 27s","accounts_in_message":[],"_revision_number":10},{"id":"81bff89c6f9ca7750ecb51d6f288bfb8d395898e","author":{"_account_id":28374,"name":"Viktor Michalek","email":"viktor.michalek@ultimum.io","username":"viktormichalek"},"date":"2019-10-01 15:10:19.000000000","message":"Patch Set 11: Patch Set 10 was rebased","accounts_in_message":[],"_revision_number":11},{"id":"708eb842620e306d668e753d38528a6f5b04c1f3","author":{"_account_id":23181,"name":"RDO Third Party CI","email":"dmsimard+rdothirdparty@redhat.com","username":"rdothirdparty","tags":["SERVICE_USER"]},"date":"2019-10-01 15:49:53.000000000","message":"Patch Set 11:\n\nBuild succeeded (check pipeline).\n\n- tripleo-build-containers-rhel-8-master http://logs.rdoproject.org/79/683679/11/openstack-check/tripleo-build-containers-rhel-8-master/d6f6e25/ : SUCCESS in 36m 46s","accounts_in_message":[],"_revision_number":11},{"id":"12e153036e98efd507056474b5229e95acfa10eb","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-10-01 16:33:50.000000000","message":"Patch Set 11: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttp://docs.openstack.org/infra/manual/developers.html#automated-testing\n\n\n- kolla-build-centos-binary https://zuul.opendev.org/t/openstack/build/ed3c41c73a6941f3983c93c1881eed27 : SUCCESS in 40m 58s (non-voting)\n- kolla-build-centos-source https://zuul.opendev.org/t/openstack/build/00cfbd6fdf964f5b926dd5c54c3d2d0f : SUCCESS in 54m 56s\n- kolla-ansible-centos-source https://zuul.opendev.org/t/openstack/build/7b783364b56d4571a1fcc4a5eb837217 : FAILURE in 54m 23s\n- kolla-ansible-centos-binary https://zuul.opendev.org/t/openstack/build/01deecf0761c4f47aa6cf981d2379fed : FAILURE in 46m 37s (non-voting)\n- kolla-ansible-centos-source-upgrade https://zuul.opendev.org/t/openstack/build/41b19596b8884a358c60dc374ef3e105 : FAILURE in 1h 19m 07s\n- tripleo-build-containers-centos-7 https://zuul.opendev.org/t/openstack/build/490d7fbb3f0b453dbd9757bf1d92c9cd : SUCCESS in 55m 03s\n- kolla-build-debian-source https://zuul.opendev.org/t/openstack/build/490cdb581290445db5e665e1faac8d8c : SUCCESS in 27m 49s (non-voting)\n- kolla-build-debian-binary https://zuul.opendev.org/t/openstack/build/a454834c18fb49bba61c5f5af0b6e87f : SUCCESS in 21m 37s (non-voting)\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/c80da407c98c4250b72608994525819b : SUCCESS in 5m 30s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/2d05ba34cf0a431fb2042b99a97e3022 : SUCCESS in 4m 29s\n- openstack-tox-py27 https://zuul.opendev.org/t/openstack/build/95aeaf975ef74c74ac254b27736f4120 : SUCCESS in 4m 31s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/e4593d09e28e48f1aafdc0040c42a99c : SUCCESS in 4m 01s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/4e3107f23d2d4ac682405f82bc4f3fcd : SUCCESS in 3m 18s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/99a0cb86fc7f461d9b81fee300f9071b : SUCCESS in 5m 20s\n- kolla-build-ubuntu-binary https://zuul.opendev.org/t/openstack/build/99ca165956964965a2c5f61edec105bf : SUCCESS in 24m 21s (non-voting)\n- kolla-build-ubuntu-source https://zuul.opendev.org/t/openstack/build/0b6dcfb1764e4f50b9ab780cdf1ac712 : SUCCESS in 34m 49s\n- kolla-ansible-ubuntu-source https://zuul.opendev.org/t/openstack/build/68ddf2a366d442cd9b1f136b50c11a53 : FAILURE in 40m 15s\n- kolla-ansible-ubuntu-binary https://zuul.opendev.org/t/openstack/build/ec9a6b6f30f84fa38c89f2ee6d3bce2c : FAILURE in 37m 34s (non-voting)\n- kolla-ansible-ubuntu-source-upgrade https://zuul.opendev.org/t/openstack/build/6a4087875c5d451781a60446dd9842e6 : FAILURE in 1h 01m 07s","accounts_in_message":[],"_revision_number":11},{"id":"8eadd7506d1d0ce8cadc5aefabc37862c88672a4","author":{"_account_id":28374,"name":"Viktor Michalek","email":"viktor.michalek@ultimum.io","username":"viktormichalek"},"date":"2019-10-02 20:16:59.000000000","message":"Patch Set 12: Commit message was updated.","accounts_in_message":[],"_revision_number":12},{"id":"8b4370f595857f38040087739a21828eab8637fe","author":{"_account_id":23181,"name":"RDO Third Party CI","email":"dmsimard+rdothirdparty@redhat.com","username":"rdothirdparty","tags":["SERVICE_USER"]},"date":"2019-10-02 20:57:39.000000000","message":"Patch Set 12:\n\nBuild succeeded (check pipeline).\n\n- tripleo-build-containers-rhel-8-master http://logs.rdoproject.org/79/683679/12/openstack-check/tripleo-build-containers-rhel-8-master/c3d861f/ : SUCCESS in 39m 46s","accounts_in_message":[],"_revision_number":12},{"id":"14fb6a24ed4362505802ce2e949121f214fe9346","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-10-02 22:05:17.000000000","message":"Patch Set 12: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttp://docs.openstack.org/infra/manual/developers.html#automated-testing\n\n\n- kolla-build-centos-binary https://zuul.opendev.org/t/openstack/build/039f54ae89f44ac1b3436ac6f19babd5 : SUCCESS in 1h 31m 12s (non-voting)\n- kolla-build-centos-source https://zuul.opendev.org/t/openstack/build/1015c4054bd74483be8dbd79fa167072 : SUCCESS in 1h 31m 22s\n- kolla-ansible-centos-source https://zuul.opendev.org/t/openstack/build/493ada51ef284ce3838308fc67da4c3b : FAILURE in 53m 59s\n- kolla-ansible-centos-binary https://zuul.opendev.org/t/openstack/build/1aed2a9d241641caac7acb2e88ec04da : FAILURE in 55m 10s (non-voting)\n- kolla-ansible-centos-source-upgrade https://zuul.opendev.org/t/openstack/build/d1b1e2239ae348f29017a02761e006df : FAILURE in 1h 37m 52s\n- tripleo-build-containers-centos-7 https://zuul.opendev.org/t/openstack/build/7a9e38d9c3ab490b9be684900112b3af : SUCCESS in 1h 05m 01s\n- kolla-build-debian-source https://zuul.opendev.org/t/openstack/build/e1c56361942a4b91beaee0218f61e1cc : SUCCESS in 28m 27s (non-voting)\n- kolla-build-debian-binary https://zuul.opendev.org/t/openstack/build/5767c8c71a284414bf1aa125da8c021f : SUCCESS in 19m 34s (non-voting)\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/83fbb5c45a6e4015a4952ddd4ff80543 : SUCCESS in 4m 43s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/0a2be22236aa46f8a7a2046681db2c6e : SUCCESS in 4m 09s\n- openstack-tox-py27 https://zuul.opendev.org/t/openstack/build/6bc7f5dbc3df4b3a8781e6de08274176 : SUCCESS in 4m 16s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/05c8389c9b644fadb68791cb288bde2e : SUCCESS in 3m 12s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/de38fe22e74b4e408099e0f7fad3ee7e : SUCCESS in 3m 59s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/8daa3b73740642a894750bb5613a18e4 : SUCCESS in 5m 04s\n- kolla-build-ubuntu-binary https://zuul.opendev.org/t/openstack/build/31408dcbdedc40fe8453044b562f1a24 : FAILURE in 48m 55s (non-voting)\n- kolla-build-ubuntu-source https://zuul.opendev.org/t/openstack/build/984bab95f9d34367afb6785d124a7cfa : SUCCESS in 32m 01s\n- kolla-ansible-ubuntu-source https://zuul.opendev.org/t/openstack/build/1ddc2a7cca9f4a2d83388b051d5a0767 : FAILURE in 34m 55s\n- kolla-ansible-ubuntu-binary https://zuul.opendev.org/t/openstack/build/8df56abe34a146c8b8b851a72dd98664 : FAILURE in 33m 31s (non-voting)\n- kolla-ansible-ubuntu-source-upgrade https://zuul.opendev.org/t/openstack/build/f7722f23a4e74ce48a782e25ee3c4189 : FAILURE in 52m 02s","accounts_in_message":[],"_revision_number":12},{"id":"d3ddd06e0e0805d6ca26f036318d6cef2afd9c35","author":{"_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"},"date":"2019-10-04 07:20:53.000000000","message":"Patch Set 12: Code-Review-1\n\ndo not check again, sudoers need to be fixed first","accounts_in_message":[],"_revision_number":12},{"id":"9884bf98d7a285ad36ba8e49361919e08b00af4e","author":{"_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"},"date":"2019-10-11 18:17:27.000000000","message":"Uploaded patch set 13.","accounts_in_message":[],"_revision_number":13},{"id":"359bdb0096943486d6037614a0d1b07fce80910c","author":{"_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"},"date":"2019-10-11 18:19:15.000000000","message":"Patch Set 13:\n\nsudoers did not have an empty line at the end...","accounts_in_message":[],"_revision_number":13},{"id":"960ae6aeca45f8e159a799704e649b3ad6e36e2e","author":{"_account_id":28374,"name":"Viktor Michalek","email":"viktor.michalek@ultimum.io","username":"viktormichalek"},"date":"2019-10-11 18:45:26.000000000","message":"Patch Set 14: Published edit on patch set 13.","accounts_in_message":[],"_revision_number":14},{"id":"3f13aab91d4807add14ee75e6d3b836cac0f8c68","author":{"_account_id":28374,"name":"Viktor Michalek","email":"viktor.michalek@ultimum.io","username":"viktormichalek"},"date":"2019-10-11 18:45:32.000000000","message":"Patch Set 15: Patch Set 14 was rebased","accounts_in_message":[],"_revision_number":15},{"id":"f2ba0936367d08483961d92f184dac0b8897976f","author":{"_account_id":23181,"name":"RDO Third Party CI","email":"dmsimard+rdothirdparty@redhat.com","username":"rdothirdparty","tags":["SERVICE_USER"]},"date":"2019-10-11 19:26:19.000000000","message":"Patch Set 15:\n\nBuild succeeded (check pipeline).\n\n- tripleo-build-containers-rhel-8-master http://logs.rdoproject.org/79/683679/15/openstack-check/tripleo-build-containers-rhel-8-master/9708404/ : SUCCESS in 37m 21s","accounts_in_message":[],"_revision_number":15},{"id":"63b0ecba1cf5b4d96bf9f189185c85d912db7ca0","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-10-11 20:08:59.000000000","message":"Patch Set 15: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttp://docs.openstack.org/infra/manual/developers.html#automated-testing\n\n\n- kolla-build-centos-binary https://zuul.opendev.org/t/openstack/build/31dfb217c24846eb92b2a24d97b7c5b6 : SUCCESS in 42m 23s (non-voting)\n- kolla-build-centos-source https://zuul.opendev.org/t/openstack/build/0ce6a01a5d404cfb8f368152513a6086 : SUCCESS in 49m 16s\n- kolla-ansible-centos-source https://zuul.opendev.org/t/openstack/build/2e87fc410bc248d6bb9505da7aa6bcea : FAILURE in 57m 47s\n- kolla-ansible-centos-binary https://zuul.opendev.org/t/openstack/build/6e1d1acebf7c41a9b92b47f1ebec2366 : FAILURE in 58m 52s (non-voting)\n- kolla-ansible-centos-source-upgrade https://zuul.opendev.org/t/openstack/build/5678391d8ad5499999f5c39868b4ffb5 : FAILURE in 1h 18m 52s\n- tripleo-build-containers-centos-7 https://zuul.opendev.org/t/openstack/build/b728f47c70464ed0a00690ee0ce21515 : SUCCESS in 53m 34s\n- kolla-build-debian-source https://zuul.opendev.org/t/openstack/build/eb3a1a2beecf42719fa84515aa2851d1 : SUCCESS in 22m 20s (non-voting)\n- kolla-build-debian-binary https://zuul.opendev.org/t/openstack/build/9f499ddaf5a24aa7b3a712178ea62f01 : SUCCESS in 22m 20s (non-voting)\n- kolla-ansible-debian-source https://zuul.opendev.org/t/openstack/build/6232bd371035466abe1e29740dcc8ccd : FAILURE in 40m 06s (non-voting)\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/1f360609f6ef4eab9fba81427a892c8e : SUCCESS in 3m 39s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/c924a723eec14cf68065144e73131007 : SUCCESS in 3m 53s\n- openstack-tox-py27 https://zuul.opendev.org/t/openstack/build/d8724d6bd46f47828af687cb6bd5dbf6 : SUCCESS in 3m 34s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/9b22acf31b8a4cf59c0d71473b14516b : SUCCESS in 4m 59s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/fe9c4f2af1774afe94292cefaf63c748 : SUCCESS in 6m 53s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/761e1d62bc7847e895cf58f168397cbf : SUCCESS in 9m 12s\n- kolla-build-ubuntu-binary https://zuul.opendev.org/t/openstack/build/018e03700cca47719be2df6c80662bda : FAILURE in 25m 57s (non-voting)\n- kolla-build-ubuntu-source https://zuul.opendev.org/t/openstack/build/98afdc9ce10042fa83107bb633aa80a7 : SUCCESS in 26m 21s\n- kolla-ansible-ubuntu-source https://zuul.opendev.org/t/openstack/build/221421168ac741e98b9d1552263cad32 : SUCCESS in 51m 15s\n- kolla-ansible-ubuntu-binary https://zuul.opendev.org/t/openstack/build/1c7c1e95a49249e0bf4915d50ae1748d : SUCCESS in 52m 46s (non-voting)\n- kolla-ansible-ubuntu-source-upgrade https://zuul.opendev.org/t/openstack/build/c81715be7b9e48aead0fa9951085b6ab : SUCCESS in 57m 37s","accounts_in_message":[],"_revision_number":15},{"id":"1b86173d58a9ffa91b50ba9ef590940ec73542c1","author":{"_account_id":28374,"name":"Viktor Michalek","email":"viktor.michalek@ultimum.io","username":"viktormichalek"},"date":"2019-10-11 21:52:15.000000000","message":"Patch Set 15:\n\nrecheck","accounts_in_message":[],"_revision_number":15},{"id":"d5663aa3a0d6ee139700b7306ac168506e8921cc","author":{"_account_id":23181,"name":"RDO Third Party CI","email":"dmsimard+rdothirdparty@redhat.com","username":"rdothirdparty","tags":["SERVICE_USER"]},"date":"2019-10-11 22:31:13.000000000","message":"Patch Set 15:\n\nBuild succeeded (check pipeline).\n\n- tripleo-build-containers-rhel-8-master http://logs.rdoproject.org/79/683679/15/openstack-check/tripleo-build-containers-rhel-8-master/7246463/ : SUCCESS in 37m 27s","accounts_in_message":[],"_revision_number":15},{"id":"d861ef4f098913aab8bb77fd9d369a4a3136c64f","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-10-11 23:27:34.000000000","message":"Patch Set 15:\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttp://docs.openstack.org/infra/manual/developers.html#automated-testing\n\n\n- kolla-build-centos-binary https://zuul.opendev.org/t/openstack/build/a8d141a09eee4394894467c1331d79ae : SUCCESS in 41m 05s (non-voting)\n- kolla-build-centos-source https://zuul.opendev.org/t/openstack/build/cb8bb9ea5ced4b86937498da97be486c : SUCCESS in 52m 01s\n- kolla-ansible-centos-source https://zuul.opendev.org/t/openstack/build/a1ab013115734f5a973261c1165dc2cb : FAILURE in 55m 30s\n- kolla-ansible-centos-binary https://zuul.opendev.org/t/openstack/build/1e6adf41df1d4ce4a63ba51bf9b6c0fc : FAILURE in 52m 55s (non-voting)\n- kolla-ansible-centos-source-upgrade https://zuul.opendev.org/t/openstack/build/540cc3de5c50423d9d6893a18c3ae535 : FAILURE in 1h 19m 33s\n- tripleo-build-containers-centos-7 https://zuul.opendev.org/t/openstack/build/cb04e3a8859b482299e0d20937b413dc : SUCCESS in 48m 42s\n- kolla-build-debian-source https://zuul.opendev.org/t/openstack/build/de1e513752cd432fb5b6b93669c77bcf : SUCCESS in 24m 27s (non-voting)\n- kolla-build-debian-binary https://zuul.opendev.org/t/openstack/build/5b6e05a3748d4875b030c386b6b5f645 : SUCCESS in 21m 30s (non-voting)\n- kolla-ansible-debian-source https://zuul.opendev.org/t/openstack/build/6d2e1f7b69d743c3b21218714ba9bf0f : FAILURE in 34m 45s (non-voting)\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/9e3168e8e2af45ccab788be89d81288e : SUCCESS in 3m 11s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/1d9bf0930d644472b6b4d6683fffd5ef : SUCCESS in 3m 58s\n- openstack-tox-py27 https://zuul.opendev.org/t/openstack/build/749afe1c284944a982dda5609a742d8f : SUCCESS in 3m 56s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/bf7225934dd842a497f17efad3262705 : SUCCESS in 3m 21s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/4f7e1594ab394dc1bc8fe09f6357140b : SUCCESS in 3m 57s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/e269130cafd74581b6d20c9d5c6cc093 : SUCCESS in 15m 11s\n- kolla-build-ubuntu-binary https://zuul.opendev.org/t/openstack/build/59492a588f5c4c4abe2604bc7b4eaeec : FAILURE in 26m 32s (non-voting)\n- kolla-build-ubuntu-source https://zuul.opendev.org/t/openstack/build/1a1b9c94722345fdaa7bda4f5fe29d25 : SUCCESS in 58m 44s\n- kolla-ansible-ubuntu-source https://zuul.opendev.org/t/openstack/build/3844c336136a4ee490f40ee2a2ab8c57 : SUCCESS in 58m 46s\n- kolla-ansible-ubuntu-binary https://zuul.opendev.org/t/openstack/build/bc103f46d4be49a1abc136ebbc3aa6df : SUCCESS in 1h 26m 19s (non-voting)\n- kolla-ansible-ubuntu-source-upgrade https://zuul.opendev.org/t/openstack/build/5b28d7e4f3294ad297a1a6d77e1bcc6a : SUCCESS in 55m 16s","accounts_in_message":[],"_revision_number":15},{"id":"650b93dbbb4ed8f9b1effd547d493dcce8560086","author":{"_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"},"date":"2019-10-12 07:27:00.000000000","message":"Patch Set 15:\n\ncentos broken by update-alternatives, debian by grep (wonder why not ubuntu though)","accounts_in_message":[],"_revision_number":15},{"id":"43b6537176089ccac213561e9861ff1728141493","author":{"_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"},"date":"2019-10-12 15:37:45.000000000","message":"Patch Set 16: Published edit on patch set 15.","accounts_in_message":[],"_revision_number":16},{"id":"837c1c7dd8c683964683e40dfe6ffbcd548e9d85","author":{"_account_id":23181,"name":"RDO Third Party CI","email":"dmsimard+rdothirdparty@redhat.com","username":"rdothirdparty","tags":["SERVICE_USER"]},"date":"2019-10-12 16:19:36.000000000","message":"Patch Set 16:\n\nBuild succeeded (check pipeline).\n\n- tripleo-build-containers-rhel-8-master http://logs.rdoproject.org/79/683679/16/openstack-check/tripleo-build-containers-rhel-8-master/d696eb4/ : SUCCESS in 37m 23s","accounts_in_message":[],"_revision_number":16},{"id":"1137062677715c3f247c52ea251b848f3d000cae","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-10-12 17:13:18.000000000","message":"Patch Set 16: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttp://docs.openstack.org/infra/manual/developers.html#automated-testing\n\n\n- kolla-build-centos-binary https://zuul.opendev.org/t/openstack/build/f200789ff76746dbbeb17d91c09f254f : SUCCESS in 51m 38s (non-voting)\n- kolla-build-centos-source https://zuul.opendev.org/t/openstack/build/3ff904b984644442869ca718c5c57a5e : SUCCESS in 48m 51s\n- kolla-ansible-centos-source https://zuul.opendev.org/t/openstack/build/2d39ae9f420747e4a8bdd84e7287343e : SUCCESS in 1h 16m 05s\n- kolla-ansible-centos-binary https://zuul.opendev.org/t/openstack/build/59ea4de2e49149b3a1e6ee5cc471e6aa : SUCCESS in 1h 14m 53s (non-voting)\n- kolla-ansible-centos-source-upgrade https://zuul.opendev.org/t/openstack/build/bd36fb5141a94613a79cc71a08919d87 : SUCCESS in 1h 32m 15s\n- tripleo-build-containers-centos-7 https://zuul.opendev.org/t/openstack/build/06b150b8423143c390f498931230548c : SUCCESS in 50m 09s\n- kolla-build-debian-source https://zuul.opendev.org/t/openstack/build/07c96babf5444d2da978b9b100bec73d : SUCCESS in 30m 20s (non-voting)\n- kolla-build-debian-binary https://zuul.opendev.org/t/openstack/build/6cbe102af2a544a984cdd07749d08698 : SUCCESS in 16m 02s (non-voting)\n- kolla-ansible-debian-source https://zuul.opendev.org/t/openstack/build/9a84c1dd728246a0b75cad30c360aa7b : FAILURE in 37m 52s (non-voting)\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/ec68954872e74f779e2fd9e35b389862 : SUCCESS in 3m 24s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/0b1c9b943c724976abf2a9eac023a41a : SUCCESS in 4m 00s\n- openstack-tox-py27 https://zuul.opendev.org/t/openstack/build/0f4b0db730964214aeeab1f850322cb6 : SUCCESS in 2m 57s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/cbaa3e3eaa08481fa2e378878f296050 : SUCCESS in 4m 18s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/c4a1ac52e65b48fc8b82de24ecbb146c : SUCCESS in 3m 23s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/b00c5a70297d402996a041f14ec8049b : SUCCESS in 5m 17s\n- kolla-build-ubuntu-binary https://zuul.opendev.org/t/openstack/build/dc289697cd9b491198f2e69923667e10 : FAILURE in 37m 15s (non-voting)\n- kolla-build-ubuntu-source https://zuul.opendev.org/t/openstack/build/d3a1205a4e4544a88e874f0f7dbde4fe : SUCCESS in 36m 41s\n- kolla-ansible-ubuntu-source https://zuul.opendev.org/t/openstack/build/94aac0b5ac8d467bb21c406ad69e8e49 : FAILURE in 31m 30s\n- kolla-ansible-ubuntu-binary https://zuul.opendev.org/t/openstack/build/c3dfffca88a640e5ab4b122c34ad66ef : FAILURE in 46m 18s (non-voting)\n- kolla-ansible-ubuntu-source-upgrade https://zuul.opendev.org/t/openstack/build/655c440889d340eaa27c8864141fb1b9 : FAILURE in 58m 33s","accounts_in_message":[],"_revision_number":16},{"id":"976dfee8bad24d42f3e1011fc49d7d5d69ea44b4","author":{"_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"},"date":"2019-10-12 18:53:42.000000000","message":"Patch Set 17: Published edit on patch set 16.","accounts_in_message":[],"_revision_number":17},{"id":"b8cc823b34030afaf19b61cfc0922fff65329e31","author":{"_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"},"date":"2019-10-12 19:02:18.000000000","message":"Patch Set 18: Published edit on patch set 17.","accounts_in_message":[],"_revision_number":18},{"id":"a4f471fdcead2a509e4716fe2c213e3ddbf39942","author":{"_account_id":23181,"name":"RDO Third Party CI","email":"dmsimard+rdothirdparty@redhat.com","username":"rdothirdparty","tags":["SERVICE_USER"]},"date":"2019-10-12 19:41:20.000000000","message":"Patch Set 18:\n\nBuild succeeded (check pipeline).\n\n- tripleo-build-containers-rhel-8-master http://logs.rdoproject.org/79/683679/18/openstack-check/tripleo-build-containers-rhel-8-master/ceef2fc/ : SUCCESS in 37m 38s","accounts_in_message":[],"_revision_number":18},{"id":"70ec617ddeabbe8bbeb0597ce7b159f10c137ae9","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-10-12 20:35:41.000000000","message":"Patch Set 18: Verified+1\n\nBuild succeeded (check pipeline).\n\n- kolla-build-centos-binary https://zuul.opendev.org/t/openstack/build/4ffc68a0343c4d46a0117007daf61e4b : SUCCESS in 38m 13s (non-voting)\n- kolla-build-centos-source https://zuul.opendev.org/t/openstack/build/d42baa8d4e2f4d4db08a920c61d504fd : SUCCESS in 51m 28s\n- kolla-ansible-centos-source https://zuul.opendev.org/t/openstack/build/110d0899fb8c433cb324c1adcc1fa83b : SUCCESS in 1h 25m 38s\n- kolla-ansible-centos-binary https://zuul.opendev.org/t/openstack/build/124ea28601674fdd82324d23d8429947 : SUCCESS in 1h 13m 48s (non-voting)\n- kolla-ansible-centos-source-upgrade https://zuul.opendev.org/t/openstack/build/14807184349e475fbdcedb597fa75dcb : SUCCESS in 1h 29m 02s\n- tripleo-build-containers-centos-7 https://zuul.opendev.org/t/openstack/build/8d5c7bc9d61d4eefbdd7ef0e15028343 : SUCCESS in 50m 00s\n- kolla-build-debian-source https://zuul.opendev.org/t/openstack/build/1e5722f062e740c285658e9f65597ccb : SUCCESS in 29m 21s (non-voting)\n- kolla-build-debian-binary https://zuul.opendev.org/t/openstack/build/eae54a1faec5470d847d9726cdbcc19b : SUCCESS in 18m 44s (non-voting)\n- kolla-ansible-debian-source https://zuul.opendev.org/t/openstack/build/18d2f571b5d0401a88769672b289eaf6 : SUCCESS in 47m 05s (non-voting)\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/8c337d470bb348b5add8f3856aaeb3aa : SUCCESS in 3m 30s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/e2df94bdafd34150bb577cb410d49802 : SUCCESS in 3m 39s\n- openstack-tox-py27 https://zuul.opendev.org/t/openstack/build/aa275bc318ac4641aa3bd1a006a0af9a : SUCCESS in 3m 01s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/1c2925d56fd54cb1bae28627d6c70a95 : SUCCESS in 2m 52s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/36d7c3b452664476b5dcfabdc9357a37 : SUCCESS in 3m 18s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/7cb2a5e590a04c2ead884469277dcdc5 : SUCCESS in 4m 30s\n- kolla-build-ubuntu-binary https://zuul.opendev.org/t/openstack/build/25d55d18e16c4256893bc4d6c327daae : FAILURE in 24m 00s (non-voting)\n- kolla-build-ubuntu-source https://zuul.opendev.org/t/openstack/build/a24042facc4b46e3b598ec73fb60dd52 : SUCCESS in 27m 14s\n- kolla-ansible-ubuntu-source https://zuul.opendev.org/t/openstack/build/26c500279ed44f459fb3cd901f957bce : SUCCESS in 49m 40s\n- kolla-ansible-ubuntu-binary https://zuul.opendev.org/t/openstack/build/473b36d033ea4133b173fffbccb169d1 : SUCCESS in 50m 07s (non-voting)\n- kolla-ansible-ubuntu-source-upgrade https://zuul.opendev.org/t/openstack/build/714d3fff8e0643c3aa7122f36c8fb39f : SUCCESS in 54m 30s","accounts_in_message":[],"_revision_number":18},{"id":"3cc0a2b6c560a70af3070f38791a46878345844b","author":{"_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"},"date":"2019-10-14 16:53:15.000000000","message":"Patch Set 18: Code-Review+2","accounts_in_message":[],"_revision_number":18},{"id":"e8fe8b0add2f533105a24dd853f2d68f4f145f92","author":{"_account_id":24072,"name":"Marcin Juszkiewicz","email":"mjuszkiewicz@redhat.com","username":"hrw"},"date":"2019-10-15 09:08:43.000000000","message":"Patch Set 18: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":18},{"id":"c8cb3f4df0a6c7644a04e01f613bcbb42328d235","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-10-15 09:08:56.000000000","message":"Patch Set 18: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":18},{"id":"2a67045a03a5211c0a2e940cb0fdf056a301066d","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-10-15 10:43:09.000000000","message":"Patch Set 18: Verified-2\n\nBuild failed (gate pipeline).  For information on how to proceed, see\nhttp://docs.openstack.org/infra/manual/developers.html#automated-testing\n\n\n- kolla-build-centos-source https://zuul.opendev.org/t/openstack/build/7db830c37c5f44cb9dc52f2c1b6e2bd2 : SUCCESS in 51m 36s\n- kolla-ansible-centos-source https://zuul.opendev.org/t/openstack/build/345f85914ecc40c69f0d413b97ec9978 : SUCCESS in 1h 22m 51s\n- kolla-ansible-centos-source-upgrade https://zuul.opendev.org/t/openstack/build/c3c3994803054280a3f9de614ea61934 : SUCCESS in 1h 18m 29s\n- tripleo-build-containers-centos-7 https://zuul.opendev.org/t/openstack/build/29145f1aa25646ed98d80dcff3ed1900 : SUCCESS in 54m 27s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/71c45b17653840eabf694d00523f1060 : SUCCESS in 4m 04s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/5796e9ce609e49f39cbabf3a37d54d9c : FAILURE in 4m 30s\n- openstack-tox-py27 https://zuul.opendev.org/t/openstack/build/0c911bb443a246d9a80da06e8129864b : SUCCESS in 3m 05s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/479f86d7f9974dbeb5857b16f0198abf : SUCCESS in 3m 33s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/2b4b0161c2e9463e893eb34919acadce : SUCCESS in 6m 33s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/1bf63e573edc4d1f8bf383380623f0a5 : SUCCESS in 8m 49s\n- kolla-build-ubuntu-source https://zuul.opendev.org/t/openstack/build/23604a2ededb46b292b36bd27cba22fb : SUCCESS in 28m 45s\n- kolla-ansible-ubuntu-source https://zuul.opendev.org/t/openstack/build/6814dc7b8d02408ab049ee306fb6ac75 : SUCCESS in 59m 17s\n- kolla-ansible-ubuntu-source-upgrade https://zuul.opendev.org/t/openstack/build/1fdf3f19df794b99be6ee5d582df1174 : SUCCESS in 1h 07m 09s","accounts_in_message":[],"_revision_number":18},{"id":"e2e854f40978ded1d359697c4f612d27a72846e6","author":{"_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"},"date":"2019-10-15 16:56:12.000000000","message":"Patch Set 18:\n\nrecheck","accounts_in_message":[],"_revision_number":18},{"id":"f2a1b371e3feecedfb857c9ad6c8f100cea5000b","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-10-15 17:13:51.000000000","message":"Patch Set 18: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttp://docs.openstack.org/infra/manual/developers.html#automated-testing\n\n\n- kolla-build-centos-binary https://zuul.opendev.org/t/openstack/build/72dc6651f7f64cde812006849a5f41ee : FAILURE in 5m 05s (non-voting)\n- kolla-build-centos-source https://zuul.opendev.org/t/openstack/build/67146d1a562d4829a260f0ae94ae6209 : FAILURE in 4m 55s\n- kolla-ansible-centos-source https://zuul.opendev.org/t/openstack/build/77627a5cf92d41d1a2e2421678036bec : FAILURE in 7m 27s\n- kolla-ansible-centos-binary https://zuul.opendev.org/t/openstack/build/50a1b390f1684be5b02d51b3dec8da72 : FAILURE in 9m 34s (non-voting)\n- kolla-ansible-centos-source-upgrade https://zuul.opendev.org/t/openstack/build/6c65d95a13094ca49e89be869d698c9c : FAILURE in 10m 00s\n- tripleo-build-containers-centos-7 https://zuul.opendev.org/t/openstack/build/e1dbb1d3d85b4593829a698452e355a4 : FAILURE in 14m 30s\n- kolla-build-debian-source https://zuul.opendev.org/t/openstack/build/5f363e6406504f8a9813d54c713edb6c : RETRY_LIMIT in 1m 50s (non-voting)\n- kolla-build-debian-binary https://zuul.opendev.org/t/openstack/build/0cb7f4b4334f4f1297855ae20b1c37f8 : RETRY_LIMIT in 1m 50s (non-voting)\n- kolla-ansible-debian-source https://zuul.opendev.org/t/openstack/build/3bea81059d62426baaf3f16329577c62 : RETRY_LIMIT in 1m 29s (non-voting)\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/3894e144d10b439ebb8a3c7233c41d8e : SUCCESS in 4m 31s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/8458deb165d747479e41c70eac0b4ccf : FAILURE in 6m 49s\n- openstack-tox-py27 https://zuul.opendev.org/t/openstack/build/f7faea5c610c48458b506dc20921aa77 : SUCCESS in 5m 16s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/4292c23613854e038f16006ab956b8df : SUCCESS in 4m 17s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/3533c8656d584edb91b0b6bfe3b62342 : SUCCESS in 4m 25s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/fc7631510da545e780db4de16e9c1a74 : SUCCESS in 7m 42s\n- kolla-build-ubuntu-binary https://zuul.opendev.org/t/openstack/build/f3ee2a90b4d04c4bae61677521512f5f : FAILURE in 4m 16s (non-voting)\n- kolla-build-ubuntu-source https://zuul.opendev.org/t/openstack/build/625cf6331fdc4629b94cd5edea0414a0 : FAILURE in 6m 05s\n- kolla-ansible-ubuntu-source https://zuul.opendev.org/t/openstack/build/014568f7047d4fa08d765b08ca600aed : FAILURE in 8m 31s\n- kolla-ansible-ubuntu-binary https://zuul.opendev.org/t/openstack/build/059cc09db88f41abbf684d2d703fabf8 : FAILURE in 6m 07s (non-voting)\n- kolla-ansible-ubuntu-source-upgrade https://zuul.opendev.org/t/openstack/build/923c4591c3d24bef98b22391e75c97ed : FAILURE in 8m 45s","accounts_in_message":[],"_revision_number":18},{"id":"f07eef66758ec6a712ba1d7ccd311ee43d4a645c","author":{"_account_id":28374,"name":"Viktor Michalek","email":"viktor.michalek@ultimum.io","username":"viktormichalek"},"date":"2019-10-15 17:33:30.000000000","message":"Patch Set 18:\n\nrecheck","accounts_in_message":[],"_revision_number":18},{"id":"426034d99aeeba4b31f49270458d203727ae90b6","author":{"_account_id":23181,"name":"RDO Third Party CI","email":"dmsimard+rdothirdparty@redhat.com","username":"rdothirdparty","tags":["SERVICE_USER"]},"date":"2019-10-15 17:47:29.000000000","message":"Patch Set 18:\n\nBuild succeeded (check pipeline).\n\n- tripleo-build-containers-rhel-8-master http://logs.rdoproject.org/79/683679/18/openstack-check/tripleo-build-containers-rhel-8-master/f9aadae/ : SUCCESS in 39m 49s","accounts_in_message":[],"_revision_number":18},{"id":"e777d3f642106f510088f0d5b8989c7d37d32540","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-10-15 18:03:07.000000000","message":"Patch Set 18:\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttp://docs.openstack.org/infra/manual/developers.html#automated-testing\n\n\n- kolla-build-centos-binary https://zuul.opendev.org/t/openstack/build/c21cd774cae348069d80958f7f3f08f4 : FAILURE in 4m 53s (non-voting)\n- kolla-build-centos-source https://zuul.opendev.org/t/openstack/build/980c6f8dc2434e24aebfb62c8c68bd4a : FAILURE in 5m 04s\n- kolla-ansible-centos-source https://zuul.opendev.org/t/openstack/build/558f0c52cd8e4556a18e4f0141c3ee4a : FAILURE in 9m 14s\n- kolla-ansible-centos-binary https://zuul.opendev.org/t/openstack/build/b9e90f29dec74bf7ae19210ca24c6a59 : FAILURE in 7m 57s (non-voting)\n- kolla-ansible-centos-source-upgrade https://zuul.opendev.org/t/openstack/build/52404ebaf77c41c4b4a03418f9825b2c : FAILURE in 9m 49s\n- tripleo-build-containers-centos-7 https://zuul.opendev.org/t/openstack/build/37c9942bab5f4aa0a80ba9cffb511e4d : FAILURE in 26m 30s\n- kolla-build-debian-source https://zuul.opendev.org/t/openstack/build/2a60223e73614a81920920df71f7ec27 : RETRY_LIMIT in 2m 30s (non-voting)\n- kolla-build-debian-binary https://zuul.opendev.org/t/openstack/build/95e2c98dc6c64d8c844fa0f826371b4a : RETRY_LIMIT in 1m 52s (non-voting)\n- kolla-ansible-debian-source https://zuul.opendev.org/t/openstack/build/b98dcbe3b8024c65bfed57e439a85661 : RETRY_LIMIT in 2m 36s (non-voting)\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/f8565992c06349f788def2671b8e74a2 : SUCCESS in 3m 55s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/16843836ea6d41428fa3e555a3777592 : FAILURE in 4m 34s\n- openstack-tox-py27 https://zuul.opendev.org/t/openstack/build/41ba7899fcc54244b11660aec65a37d6 : SUCCESS in 3m 13s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/836e099583194419a96ec37d6e661f19 : SUCCESS in 3m 33s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/ed94c828a7cc40f6b567353c2db260ab : SUCCESS in 3m 52s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/6ab0150e147142dcac8d28fa84bafa37 : SUCCESS in 5m 24s\n- kolla-build-ubuntu-binary https://zuul.opendev.org/t/openstack/build/4b0ae90687274dc5b2b51b94e9f9aaf2 : FAILURE in 3m 16s (non-voting)\n- kolla-build-ubuntu-source https://zuul.opendev.org/t/openstack/build/d44db017f95d4d71aaa49f140549ad35 : FAILURE in 3m 25s\n- kolla-ansible-ubuntu-source https://zuul.opendev.org/t/openstack/build/adb522f2cc47466a8d6311e60e42a630 : FAILURE in 7m 02s\n- kolla-ansible-ubuntu-binary https://zuul.opendev.org/t/openstack/build/9d2a9d3501b64367a9ff64892d58ae75 : FAILURE in 9m 50s (non-voting)\n- kolla-ansible-ubuntu-source-upgrade https://zuul.opendev.org/t/openstack/build/bd9629cbafe64bfdba397b1f10b6e5eb : FAILURE in 6m 40s","accounts_in_message":[],"_revision_number":18},{"id":"fda3a8742f0ba6b769e3984f41bffd46039216a4","author":{"_account_id":24072,"name":"Marcin Juszkiewicz","email":"mjuszkiewicz@redhat.com","username":"hrw"},"date":"2019-10-16 05:03:01.000000000","message":"Patch Set 18:\n\nrecheck","accounts_in_message":[],"_revision_number":18},{"id":"045df98a095bf059fa447bdbd6ee490410664722","author":{"_account_id":23181,"name":"RDO Third Party CI","email":"dmsimard+rdothirdparty@redhat.com","username":"rdothirdparty","tags":["SERVICE_USER"]},"date":"2019-10-16 05:44:21.000000000","message":"Patch Set 18:\n\nBuild succeeded (check pipeline).\n\n- tripleo-build-containers-rhel-8-master http://logs.rdoproject.org/79/683679/18/openstack-check/tripleo-build-containers-rhel-8-master/24e96f5/ : SUCCESS in 37m 51s","accounts_in_message":[],"_revision_number":18},{"id":"91be6c653c2f25909a0a667151962bec4ddca553","author":{"_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"},"date":"2019-10-16 06:28:50.000000000","message":"Patch Set 18:\n\nrecheck\n\nregistry again","accounts_in_message":[],"_revision_number":18},{"id":"b84ec91627aed9f3b4786ef17183965cdb753620","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-10-16 06:42:03.000000000","message":"Patch Set 18: Verified+1\n\nBuild succeeded (check pipeline).\n\n- kolla-build-centos-binary https://zuul.opendev.org/t/openstack/build/9200868680d5483ca3288b3f13d54162 : SUCCESS in 55m 19s (non-voting)\n- kolla-build-centos-source https://zuul.opendev.org/t/openstack/build/2848577bab074799a1d4a9673bd3351c : SUCCESS in 57m 43s\n- kolla-ansible-centos-source https://zuul.opendev.org/t/openstack/build/3c7d0750b9e0403a990a7d343c253fec : SUCCESS in 1h 28m 11s\n- kolla-ansible-centos-binary https://zuul.opendev.org/t/openstack/build/4e67fe1d0de847278ead69557438350a : SUCCESS in 1h 21m 54s (non-voting)\n- kolla-ansible-centos-source-upgrade https://zuul.opendev.org/t/openstack/build/4a30a565778240c387b4a23b644114ca : SUCCESS in 1h 35m 05s\n- tripleo-build-containers-centos-7 https://zuul.opendev.org/t/openstack/build/4fe8f2a63e8e453e8a099e97a67e6363 : SUCCESS in 58m 08s\n- kolla-build-debian-source https://zuul.opendev.org/t/openstack/build/ae0d6c765ab244e39bae2c1c8aad62c1 : SUCCESS in 30m 04s (non-voting)\n- kolla-build-debian-binary https://zuul.opendev.org/t/openstack/build/602ba0340efe4ac4b948cd900fcb4de6 : SUCCESS in 16m 48s (non-voting)\n- kolla-ansible-debian-source https://zuul.opendev.org/t/openstack/build/883df5eb6cad4a218f1f8ad885209f44 : SUCCESS in 1h 06m 30s (non-voting)\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/15038f8027ad4178ae4a4a73d664aa3b : SUCCESS in 3m 03s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/6df8b8ea976a4ae38479179acf007b51 : SUCCESS in 3m 44s\n- openstack-tox-py27 https://zuul.opendev.org/t/openstack/build/09c42418561549938bf2ac757c104a34 : SUCCESS in 2m 46s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/8c3d971c84af4d679ab67efc5a54884e : SUCCESS in 4m 05s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/755a17a2dc634d488cb1fbd4987083a7 : SUCCESS in 3m 12s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/3195bccfad724ebb9703b2c6f14378c6 : SUCCESS in 5m 06s\n- kolla-build-ubuntu-binary https://zuul.opendev.org/t/openstack/build/4f4b440dc37d49f39665e18a53b81de2 : FAILURE in 23m 07s (non-voting)\n- kolla-build-ubuntu-source https://zuul.opendev.org/t/openstack/build/a90ea5a69edd40c6a677da924f93be53 : SUCCESS in 44m 29s\n- kolla-ansible-ubuntu-source https://zuul.opendev.org/t/openstack/build/8d480ab3ad3d4eac94c843bf8c6818ed : SUCCESS in 52m 09s\n- kolla-ansible-ubuntu-binary https://zuul.opendev.org/t/openstack/build/7de58e6eaa2b4addab54116e6e070ffc : SUCCESS in 50m 30s (non-voting)\n- kolla-ansible-ubuntu-source-upgrade https://zuul.opendev.org/t/openstack/build/6416f2d84e6e49bba9c8ed17d2e85112 : SUCCESS in 1h 15m 47s","accounts_in_message":[],"_revision_number":18},{"id":"944c2b69d39504d1229c87fb1d1d77e3fabb01e5","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-10-16 06:42:15.000000000","message":"Patch Set 18: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":18},{"id":"482a6ebf05e425ef7000816d6c81af4a5ca6dd3c","author":{"_account_id":23181,"name":"RDO Third Party CI","email":"dmsimard+rdothirdparty@redhat.com","username":"rdothirdparty","tags":["SERVICE_USER"]},"date":"2019-10-16 07:06:11.000000000","message":"Patch Set 18:\n\nBuild succeeded (check pipeline).\n\n- tripleo-build-containers-rhel-8-master http://logs.rdoproject.org/79/683679/18/openstack-check/tripleo-build-containers-rhel-8-master/c28eede/ : SUCCESS in 35m 53s","accounts_in_message":[],"_revision_number":18},{"id":"a4b2711ea3de00ce9c3315fdbfa34f12fcd99314","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-10-16 08:49:01.000000000","message":"Patch Set 18: Verified-2\n\nBuild failed (gate pipeline).  For information on how to proceed, see\nhttp://docs.openstack.org/infra/manual/developers.html#automated-testing\n\n\n- kolla-build-centos-source https://zuul.opendev.org/t/openstack/build/b5e6d5625de04eca9b9534d58cf68dcd : SUCCESS in 1h 00m 20s\n- kolla-ansible-centos-source https://zuul.opendev.org/t/openstack/build/cb6e0e04dca34d4b953272ef9ef93e08 : SUCCESS in 1h 28m 16s\n- kolla-ansible-centos-source-upgrade https://zuul.opendev.org/t/openstack/build/31c7ce25cd0f433a9513c30257a588e0 : TIMED_OUT in 2h 01m 50s\n- tripleo-build-containers-centos-7 https://zuul.opendev.org/t/openstack/build/6ee49ceea4fe40be823dc488bcfc5662 : SUCCESS in 1h 02m 16s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/e0a44afc52d54eae8be1b76365733679 : SUCCESS in 5m 53s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/b2aedc264b1447f59a19f2dfb22eb755 : SUCCESS in 6m 37s\n- openstack-tox-py27 https://zuul.opendev.org/t/openstack/build/9d4f58e1e8c1448f97967b5c1704edbd : SUCCESS in 5m 53s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/6ab01ab5dd484fc183a9a2322d45f9df : SUCCESS in 6m 30s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/890c83496fc447b79fb89b26d3d8e097 : SUCCESS in 8m 55s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/80dac4275b194b4f92b2871ec933ef51 : SUCCESS in 13m 47s\n- kolla-build-ubuntu-source https://zuul.opendev.org/t/openstack/build/d003a6f0bebb4f389fb286a0525eb693 : SUCCESS in 1h 11m 17s\n- kolla-ansible-ubuntu-source https://zuul.opendev.org/t/openstack/build/2f504a21844b482e8d7ee55b7a39d3ac : SUCCESS in 1h 23m 56s\n- kolla-ansible-ubuntu-source-upgrade https://zuul.opendev.org/t/openstack/build/62ed77218ae1438c964dbbc75ee2fff6 : SUCCESS in 1h 58m 03s","accounts_in_message":[],"_revision_number":18},{"id":"ebdb11a8cc7f279c32f5dfb763bc1732683e0349","author":{"_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"},"date":"2019-10-16 08:52:27.000000000","message":"Patch Set 18:\n\nrecheck\n\nhmm, upgrades with builds often take too much time","accounts_in_message":[],"_revision_number":18},{"id":"5f4efa69ff61429fced946d3c946a5dc6857d161","author":{"_account_id":23181,"name":"RDO Third Party CI","email":"dmsimard+rdothirdparty@redhat.com","username":"rdothirdparty","tags":["SERVICE_USER"]},"date":"2019-10-16 09:32:05.000000000","message":"Patch Set 18:\n\nBuild succeeded (check pipeline).\n\n- tripleo-build-containers-rhel-8-master http://logs.rdoproject.org/79/683679/18/openstack-check/tripleo-build-containers-rhel-8-master/d0be74e/ : SUCCESS in 38m 00s","accounts_in_message":[],"_revision_number":18},{"id":"99b28a270616c36db5017f17cd04769499ea8b28","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-10-16 10:40:10.000000000","message":"Patch Set 18: Verified+1\n\nBuild succeeded (check pipeline).\n\n- kolla-build-centos-binary https://zuul.opendev.org/t/openstack/build/29781d91be904a469df1922956f68322 : SUCCESS in 41m 31s (non-voting)\n- kolla-build-centos-source https://zuul.opendev.org/t/openstack/build/f36478979bd540248b08bcca0543f60c : SUCCESS in 56m 37s\n- kolla-ansible-centos-source https://zuul.opendev.org/t/openstack/build/cd00b8dc14664427a2f0af29a8c23301 : SUCCESS in 1h 22m 13s\n- kolla-ansible-centos-binary https://zuul.opendev.org/t/openstack/build/dc56917d1a2b4b449a0370df5f0e2ab7 : SUCCESS in 1h 34m 00s (non-voting)\n- kolla-ansible-centos-source-upgrade https://zuul.opendev.org/t/openstack/build/0b183a3dee6e41f79cf03ea9f4d82314 : SUCCESS in 1h 37m 43s\n- tripleo-build-containers-centos-7 https://zuul.opendev.org/t/openstack/build/9179faa7aa1149ea980835a14ef1c0d5 : SUCCESS in 58m 09s\n- kolla-build-debian-source https://zuul.opendev.org/t/openstack/build/922925e527924db798a344368722f032 : SUCCESS in 24m 35s (non-voting)\n- kolla-build-debian-binary https://zuul.opendev.org/t/openstack/build/c8f2caf82d7d46e4a6ee4d5406aa0228 : SUCCESS in 23m 16s (non-voting)\n- kolla-ansible-debian-source https://zuul.opendev.org/t/openstack/build/ad001118b0cc4caba951222e008321d6 : SUCCESS in 49m 41s (non-voting)\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/1123142aa062487fbbfd26be54e8001d : SUCCESS in 6m 45s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/899b97e471a84eefa9141a762f25c8a9 : SUCCESS in 4m 08s\n- openstack-tox-py27 https://zuul.opendev.org/t/openstack/build/3863dde956094a44922de2e68496c80c : SUCCESS in 2m 40s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/de2012a1d73c4b07ab8d7f73e42a314a : SUCCESS in 4m 18s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/efcf4c72b3754d299be63f5068c9919b : SUCCESS in 3m 58s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/aee542f5c351462d80787ab3d45de841 : SUCCESS in 4m 42s\n- kolla-build-ubuntu-binary https://zuul.opendev.org/t/openstack/build/ff10bc7040634611bd532c45410b7ad5 : FAILURE in 27m 49s (non-voting)\n- kolla-build-ubuntu-source https://zuul.opendev.org/t/openstack/build/295b1fd61a3b4f7394f5ad93999c8f07 : SUCCESS in 37m 23s\n- kolla-ansible-ubuntu-source https://zuul.opendev.org/t/openstack/build/ef657c70c9974ccb9b16d0b42209d296 : SUCCESS in 55m 00s\n- kolla-ansible-ubuntu-binary https://zuul.opendev.org/t/openstack/build/057e2a74a72a44e5b93efedce8fc8cca : SUCCESS in 52m 28s (non-voting)\n- kolla-ansible-ubuntu-source-upgrade https://zuul.opendev.org/t/openstack/build/47a2d14f263c441f95d4f25c0f248ea0 : SUCCESS in 1h 01m 34s","accounts_in_message":[],"_revision_number":18},{"id":"6a850a7e93a6eaf40eb5fbcfae8a122f36815550","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-10-16 10:40:24.000000000","message":"Patch Set 18: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":18},{"id":"05f5115fb116908e93523fbc2de37d0217134810","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-10-16 12:22:13.000000000","message":"Change has been successfully merged by Zuul","accounts_in_message":[],"_revision_number":18},{"id":"fac1bfe878d0e1a71ecddfd7951288ece502c5cb","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-10-16 12:22:13.000000000","message":"Patch Set 18: Verified+2\n\nBuild succeeded (gate pipeline).\n\n- kolla-build-centos-source https://zuul.opendev.org/t/openstack/build/35215a3aa4ee46f2b1f3f060e544a72d : SUCCESS in 1h 03m 42s\n- kolla-ansible-centos-source https://zuul.opendev.org/t/openstack/build/b0fe1beb044a464ebeb6d10bedcadaf0 : SUCCESS in 1h 36m 38s\n- kolla-ansible-centos-source-upgrade https://zuul.opendev.org/t/openstack/build/ed13ef6022f54c3f925ccd61b0ab6c5c : SUCCESS in 1h 22m 25s\n- tripleo-build-containers-centos-7 https://zuul.opendev.org/t/openstack/build/f2b7ecd2eefa44de8a094da39537b708 : SUCCESS in 43m 33s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/cd5d6e72a72e4596b1de97476f608af2 : SUCCESS in 4m 15s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/14f0a895f91a4eadb5bdd072ec5ed160 : SUCCESS in 3m 57s\n- openstack-tox-py27 https://zuul.opendev.org/t/openstack/build/089f4aec9adb4f0cab39387d4621fc72 : SUCCESS in 4m 22s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/f7a6bf5bc11a429c862c77bea943c3b3 : SUCCESS in 3m 16s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/c570429f793d4f15be4d54e15726e32e : SUCCESS in 3m 42s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/232b7cd3006c4a8c9d8df58bc402db6f : SUCCESS in 5m 04s\n- kolla-build-ubuntu-source https://zuul.opendev.org/t/openstack/build/ba1e0eb987674d3fb4f19d7884205829 : SUCCESS in 31m 55s\n- kolla-ansible-ubuntu-source https://zuul.opendev.org/t/openstack/build/d2426eb376bf43cdb6c868e74e18ce27 : SUCCESS in 51m 32s\n- kolla-ansible-ubuntu-source-upgrade https://zuul.opendev.org/t/openstack/build/ac2d7b18a9fb4fa8ba949632158d697a : SUCCESS in 57m 29s","accounts_in_message":[],"_revision_number":18},{"id":"bf3ae152448af655bb30084d032940e8dd788610","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-10-16 12:23:49.000000000","message":"Patch Set 18:\n\nBuild succeeded (promote pipeline).\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/9e6b7a817f2243ebb508519164079a13 : SUCCESS in 1m 15s","accounts_in_message":[],"_revision_number":18},{"id":"0d0e5332d15d40e559a43bcd079f5f0987b80c74","author":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},"date":"2019-10-16 12:27:42.000000000","message":"Patch Set 18:\n\nWhat happened to feature freeze?","accounts_in_message":[],"_revision_number":18}],"current_revision_number":18,"current_revision":"e8289ff6df6b305a2f154a4aa6210c5d0bbb006f","revisions":{"6a2dd78ffae1a2410a84baa3d2d34c98d9212f6e":{"kind":"REWORK","_number":1,"created":"2019-09-20 18:14:28.000000000","uploader":{"_account_id":28374,"name":"Viktor Michalek","email":"viktor.michalek@ultimum.io","username":"viktormichalek"},"ref":"refs/changes/79/683679/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/kolla","ref":"refs/changes/79/683679/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/kolla refs/changes/79/683679/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/kolla refs/changes/79/683679/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/kolla refs/changes/79/683679/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/kolla refs/changes/79/683679/1"}}},"commit":{"parents":[{"commit":"ebeccf17aae046a89e083a360d0dcfcd671f1ff7","subject":"Neutron: fwaas should be installed also in Debian","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/ebeccf17aae046a89e083a360d0dcfcd671f1ff7"}]}],"author":{"name":"Jan Vondra","email":"jan.vondra@ultimum.io","date":"2019-09-20 16:45:52.000000000","tz":120},"committer":{"name":"Viktor Michalek","email":"viktor.michalek@ultimum.io","date":"2019-09-20 16:45:52.000000000","tz":120},"subject":"Neutron: use iptables-legacy","message":"Neutron: use iptables-legacy\n\nUse iptables-legacy if KOLLA_LEGACY_IPTABLES\u003dyes is set.\nForces usage of legacy iptables binary (some kernels have issues\nwith nftables).\n\nChange-Id: I3b17bfdbc68a9a227bb22dc05a02ed6da607a114\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/6a2dd78ffae1a2410a84baa3d2d34c98d9212f6e"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/6a2dd78ffae1a2410a84baa3d2d34c98d9212f6e"}]},"branch":"refs/heads/master"},"95170cec480268822a4c629ab11ca13cb4564e37":{"kind":"TRIVIAL_REBASE","_number":2,"created":"2019-09-24 15:56:34.000000000","uploader":{"_account_id":28374,"name":"Viktor Michalek","email":"viktor.michalek@ultimum.io","username":"viktormichalek"},"ref":"refs/changes/79/683679/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/kolla","ref":"refs/changes/79/683679/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/kolla refs/changes/79/683679/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/kolla refs/changes/79/683679/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/kolla refs/changes/79/683679/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/kolla refs/changes/79/683679/2"}}},"commit":{"parents":[{"commit":"07102a0ad29ebb2526860ab058332e8379d6c9ec","subject":"Neutron: fwaas should be installed also in Debian","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/07102a0ad29ebb2526860ab058332e8379d6c9ec"}]}],"author":{"name":"Jan Vondra","email":"jan.vondra@ultimum.io","date":"2019-09-20 16:45:52.000000000","tz":120},"committer":{"name":"Viktor Michalek","email":"viktor.michalek@ultimum.io","date":"2019-09-24 15:56:34.000000000","tz":0},"subject":"Neutron: use iptables-legacy","message":"Neutron: use iptables-legacy\n\nUse iptables-legacy if KOLLA_LEGACY_IPTABLES\u003dyes is set.\nForces usage of legacy iptables binary (some kernels have issues\nwith nftables).\n\nChange-Id: I3b17bfdbc68a9a227bb22dc05a02ed6da607a114\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/95170cec480268822a4c629ab11ca13cb4564e37"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/95170cec480268822a4c629ab11ca13cb4564e37"}]},"branch":"refs/heads/master"},"be225eae06c19cfce3d692e2c16cf1c2c2df2221":{"kind":"REWORK","_number":3,"created":"2019-09-27 17:36:54.000000000","uploader":{"_account_id":28374,"name":"Viktor Michalek","email":"viktor.michalek@ultimum.io","username":"viktormichalek"},"ref":"refs/changes/79/683679/3","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/kolla","ref":"refs/changes/79/683679/3","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/kolla refs/changes/79/683679/3 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/kolla refs/changes/79/683679/3 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/kolla refs/changes/79/683679/3 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/kolla refs/changes/79/683679/3"}}},"commit":{"parents":[{"commit":"07102a0ad29ebb2526860ab058332e8379d6c9ec","subject":"Neutron: fwaas should be installed also in Debian","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/07102a0ad29ebb2526860ab058332e8379d6c9ec"}]}],"author":{"name":"Jan Vondra","email":"jan.vondra@ultimum.io","date":"2019-09-20 16:45:52.000000000","tz":120},"committer":{"name":"Viktor Michalek","email":"viktor.michalek@ultimum.io","date":"2019-09-27 17:36:46.000000000","tz":0},"subject":"Neutron: use iptables-legacy","message":"Neutron: use iptables-legacy\n\nUse iptables-legacy if KOLLA_LEGACY_IPTABLES\u003dyes is set.\nForces usage of legacy iptables binary (some kernels have issues\nwith nftables).\n\nChange-Id: I3b17bfdbc68a9a227bb22dc05a02ed6da607a114\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/be225eae06c19cfce3d692e2c16cf1c2c2df2221"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/be225eae06c19cfce3d692e2c16cf1c2c2df2221"}]},"branch":"refs/heads/master"},"f8d43b84dd7729fd765d0bfd722c894faf22e989":{"kind":"NO_CODE_CHANGE","_number":4,"created":"2019-09-27 17:46:21.000000000","uploader":{"_account_id":28374,"name":"Viktor Michalek","email":"viktor.michalek@ultimum.io","username":"viktormichalek"},"ref":"refs/changes/79/683679/4","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/kolla","ref":"refs/changes/79/683679/4","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/kolla refs/changes/79/683679/4 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/kolla refs/changes/79/683679/4 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/kolla refs/changes/79/683679/4 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/kolla refs/changes/79/683679/4"}}},"commit":{"parents":[{"commit":"07102a0ad29ebb2526860ab058332e8379d6c9ec","subject":"Neutron: fwaas should be installed also in Debian","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/07102a0ad29ebb2526860ab058332e8379d6c9ec"}]}],"author":{"name":"Jan Vondra","email":"jan.vondra@ultimum.io","date":"2019-09-20 16:45:52.000000000","tz":120},"committer":{"name":"Viktor Michalek","email":"viktor.michalek@ultimum.io","date":"2019-09-27 17:46:14.000000000","tz":0},"subject":"Neutron: use iptables-legacy","message":"Neutron: use iptables-legacy\n\nForces usage of legacy iptables binary (some kernels have issues\nwith nftables). \nWe enountered the problem while upgrading deployment with host system \nUbuntu 16.04 Xenial from Queens debian kolla images (based on stretch) \nto Stein. \nThis causes problems when using a hybrid security group driver since\nBuster is using iptables-nft by default and Xenial kernel has no support\nfor it resulting in port binding failure.\nSo basically this issue occurs \"only\" when upgrading kolla deployment \nwithout upgrading the host OS.\n\nChange-Id: I3b17bfdbc68a9a227bb22dc05a02ed6da607a114\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/f8d43b84dd7729fd765d0bfd722c894faf22e989"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/f8d43b84dd7729fd765d0bfd722c894faf22e989"}]},"branch":"refs/heads/master"},"b1dcbb6985f6c7af49d22935e37ff7a9882f1926":{"kind":"TRIVIAL_REBASE","_number":5,"created":"2019-09-27 17:46:28.000000000","uploader":{"_account_id":28374,"name":"Viktor Michalek","email":"viktor.michalek@ultimum.io","username":"viktormichalek"},"ref":"refs/changes/79/683679/5","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/kolla","ref":"refs/changes/79/683679/5","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/kolla refs/changes/79/683679/5 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/kolla refs/changes/79/683679/5 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/kolla refs/changes/79/683679/5 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/kolla refs/changes/79/683679/5"}}},"commit":{"parents":[{"commit":"9f4197c69202f546fd7b67840dcd7540ba6db407","subject":"Neutron: fwaas should be installed also in Debian","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/9f4197c69202f546fd7b67840dcd7540ba6db407"}]}],"author":{"name":"Jan Vondra","email":"jan.vondra@ultimum.io","date":"2019-09-20 16:45:52.000000000","tz":120},"committer":{"name":"Viktor Michalek","email":"viktor.michalek@ultimum.io","date":"2019-09-27 17:46:28.000000000","tz":0},"subject":"Neutron: use iptables-legacy","message":"Neutron: use iptables-legacy\n\nForces usage of legacy iptables binary (some kernels have issues\nwith nftables). \nWe enountered the problem while upgrading deployment with host system \nUbuntu 16.04 Xenial from Queens debian kolla images (based on stretch) \nto Stein. \nThis causes problems when using a hybrid security group driver since\nBuster is using iptables-nft by default and Xenial kernel has no support\nfor it resulting in port binding failure.\nSo basically this issue occurs \"only\" when upgrading kolla deployment \nwithout upgrading the host OS.\n\nChange-Id: I3b17bfdbc68a9a227bb22dc05a02ed6da607a114\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/b1dcbb6985f6c7af49d22935e37ff7a9882f1926"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/b1dcbb6985f6c7af49d22935e37ff7a9882f1926"}]},"branch":"refs/heads/master"},"ab86dad6dabcc20b2d4d188e2b8579d7fc5a6b79":{"kind":"REWORK","_number":6,"created":"2019-09-27 19:16:36.000000000","uploader":{"_account_id":28374,"name":"Viktor Michalek","email":"viktor.michalek@ultimum.io","username":"viktormichalek"},"ref":"refs/changes/79/683679/6","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/kolla","ref":"refs/changes/79/683679/6","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/kolla refs/changes/79/683679/6 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/kolla refs/changes/79/683679/6 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/kolla refs/changes/79/683679/6 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/kolla refs/changes/79/683679/6"}}},"commit":{"parents":[{"commit":"9f4197c69202f546fd7b67840dcd7540ba6db407","subject":"Neutron: fwaas should be installed also in Debian","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/9f4197c69202f546fd7b67840dcd7540ba6db407"}]}],"author":{"name":"Jan Vondra","email":"jan.vondra@ultimum.io","date":"2019-09-20 16:45:52.000000000","tz":120},"committer":{"name":"Viktor Michalek","email":"viktor.michalek@ultimum.io","date":"2019-09-27 19:16:20.000000000","tz":0},"subject":"Neutron: use iptables-legacy","message":"Neutron: use iptables-legacy\n\nSetting legacy iptables for host system kernels not supporting \niptables-nft.\n\nChange-Id: I3b17bfdbc68a9a227bb22dc05a02ed6da607a114\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/ab86dad6dabcc20b2d4d188e2b8579d7fc5a6b79"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/ab86dad6dabcc20b2d4d188e2b8579d7fc5a6b79"}]},"branch":"refs/heads/master"},"340306f483dbac1e3df1865472f114d1b784516c":{"kind":"NO_CHANGE","_number":7,"created":"2019-09-28 17:45:00.000000000","uploader":{"_account_id":28374,"name":"Viktor Michalek","email":"viktor.michalek@ultimum.io","username":"viktormichalek"},"ref":"refs/changes/79/683679/7","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/kolla","ref":"refs/changes/79/683679/7","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/kolla refs/changes/79/683679/7 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/kolla refs/changes/79/683679/7 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/kolla refs/changes/79/683679/7 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/kolla refs/changes/79/683679/7"}}},"commit":{"parents":[{"commit":"223a7c0f050405b680e24bb185cdef677734ee16","subject":"Neutron: fwaas should be installed also in Debian","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/223a7c0f050405b680e24bb185cdef677734ee16"}]}],"author":{"name":"Jan Vondra","email":"jan.vondra@ultimum.io","date":"2019-09-20 16:45:52.000000000","tz":120},"committer":{"name":"Viktor Michalek","email":"viktor.michalek@ultimum.io","date":"2019-09-28 17:45:00.000000000","tz":0},"subject":"Neutron: use iptables-legacy","message":"Neutron: use iptables-legacy\n\nSetting legacy iptables for host system kernels not supporting \niptables-nft.\n\nChange-Id: I3b17bfdbc68a9a227bb22dc05a02ed6da607a114\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/340306f483dbac1e3df1865472f114d1b784516c"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/340306f483dbac1e3df1865472f114d1b784516c"}]},"branch":"refs/heads/master"},"1deee09ae4e1c9ff595b92597bc7f474998cb318":{"kind":"TRIVIAL_REBASE","_number":8,"created":"2019-09-29 13:58:45.000000000","uploader":{"_account_id":28374,"name":"Viktor Michalek","email":"viktor.michalek@ultimum.io","username":"viktormichalek"},"ref":"refs/changes/79/683679/8","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/kolla","ref":"refs/changes/79/683679/8","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/kolla refs/changes/79/683679/8 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/kolla refs/changes/79/683679/8 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/kolla refs/changes/79/683679/8 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/kolla refs/changes/79/683679/8"}}},"commit":{"parents":[{"commit":"baf9a10da0a7b7cf7047b4ba4905f69ace48803e","subject":"Neutron: fwaas should be installed also in Debian","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/baf9a10da0a7b7cf7047b4ba4905f69ace48803e"}]}],"author":{"name":"Jan Vondra","email":"jan.vondra@ultimum.io","date":"2019-09-20 16:45:52.000000000","tz":120},"committer":{"name":"Viktor Michalek","email":"viktor.michalek@ultimum.io","date":"2019-09-29 13:58:45.000000000","tz":0},"subject":"Neutron: use iptables-legacy","message":"Neutron: use iptables-legacy\n\nSetting legacy iptables for host system kernels not supporting \niptables-nft.\n\nChange-Id: I3b17bfdbc68a9a227bb22dc05a02ed6da607a114\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/1deee09ae4e1c9ff595b92597bc7f474998cb318"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/1deee09ae4e1c9ff595b92597bc7f474998cb318"}]},"branch":"refs/heads/master"},"9e02cdaa591b7dbe352865b44e0c7e8ad789a4f9":{"kind":"NO_CODE_CHANGE","_number":9,"created":"2019-10-01 06:49:12.000000000","uploader":{"_account_id":24072,"name":"Marcin Juszkiewicz","email":"mjuszkiewicz@redhat.com","username":"hrw"},"ref":"refs/changes/79/683679/9","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/kolla","ref":"refs/changes/79/683679/9","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/kolla refs/changes/79/683679/9 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/kolla refs/changes/79/683679/9 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/kolla refs/changes/79/683679/9 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/kolla refs/changes/79/683679/9"}}},"commit":{"parents":[{"commit":"baf9a10da0a7b7cf7047b4ba4905f69ace48803e","subject":"Neutron: fwaas should be installed also in Debian","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/baf9a10da0a7b7cf7047b4ba4905f69ace48803e"}]}],"author":{"name":"Jan Vondra","email":"jan.vondra@ultimum.io","date":"2019-09-20 16:45:52.000000000","tz":120},"committer":{"name":"Marcin Juszkiewicz","email":"marcin.juszkiewicz@linaro.org","date":"2019-10-01 06:48:53.000000000","tz":0},"subject":"Neutron: add support to use legacy iptables","message":"Neutron: add support to use legacy iptables\n\nSetting legacy iptables for host system kernels not supporting \niptables-nft.\n\nChange-Id: I3b17bfdbc68a9a227bb22dc05a02ed6da607a114\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/9e02cdaa591b7dbe352865b44e0c7e8ad789a4f9"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/9e02cdaa591b7dbe352865b44e0c7e8ad789a4f9"}]},"branch":"refs/heads/master"},"5e8ce6518aa999c8f5ec6fd6df8b8737d8b95f95":{"kind":"NO_CODE_CHANGE","_number":10,"created":"2019-10-01 13:26:12.000000000","uploader":{"_account_id":28374,"name":"Viktor Michalek","email":"viktor.michalek@ultimum.io","username":"viktormichalek"},"ref":"refs/changes/79/683679/10","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/kolla","ref":"refs/changes/79/683679/10","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/kolla refs/changes/79/683679/10 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/kolla refs/changes/79/683679/10 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/kolla refs/changes/79/683679/10 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/kolla refs/changes/79/683679/10"}}},"commit":{"parents":[{"commit":"baf9a10da0a7b7cf7047b4ba4905f69ace48803e","subject":"Neutron: fwaas should be installed also in Debian","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/baf9a10da0a7b7cf7047b4ba4905f69ace48803e"}]}],"author":{"name":"Jan Vondra","email":"jan.vondra@ultimum.io","date":"2019-09-20 16:45:52.000000000","tz":120},"committer":{"name":"Viktor Michalek","email":"viktor.michalek@ultimum.io","date":"2019-10-01 13:26:07.000000000","tz":0},"subject":"Neutron: add support to use legacy iptables","message":"Neutron: add support to use legacy iptables\n\nSetting legacy iptables for host system kernels not supporting \niptables-nft.\n\nDepends-On: https://review.opendev.org/685967\nChange-Id: I3b17bfdbc68a9a227bb22dc05a02ed6da607a114\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/5e8ce6518aa999c8f5ec6fd6df8b8737d8b95f95"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/5e8ce6518aa999c8f5ec6fd6df8b8737d8b95f95"}]},"branch":"refs/heads/master"},"11033fa82b68b94a8d3c8d910d0b1906f0826dd8":{"kind":"TRIVIAL_REBASE","_number":11,"created":"2019-10-01 15:10:19.000000000","uploader":{"_account_id":28374,"name":"Viktor Michalek","email":"viktor.michalek@ultimum.io","username":"viktormichalek"},"ref":"refs/changes/79/683679/11","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/kolla","ref":"refs/changes/79/683679/11","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/kolla refs/changes/79/683679/11 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/kolla refs/changes/79/683679/11 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/kolla refs/changes/79/683679/11 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/kolla refs/changes/79/683679/11"}}},"commit":{"parents":[{"commit":"a9181bf89529756e4d58b22324988b4e57770ec1","subject":"Merge \"Update mention of Kayobe in documentation\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/a9181bf89529756e4d58b22324988b4e57770ec1"}]}],"author":{"name":"Jan Vondra","email":"jan.vondra@ultimum.io","date":"2019-09-20 16:45:52.000000000","tz":120},"committer":{"name":"Viktor Michalek","email":"viktor.michalek@ultimum.io","date":"2019-10-01 15:10:19.000000000","tz":0},"subject":"Neutron: add support to use legacy iptables","message":"Neutron: add support to use legacy iptables\n\nSetting legacy iptables for host system kernels not supporting \niptables-nft.\n\nDepends-On: https://review.opendev.org/685967\nChange-Id: I3b17bfdbc68a9a227bb22dc05a02ed6da607a114\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/11033fa82b68b94a8d3c8d910d0b1906f0826dd8"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/11033fa82b68b94a8d3c8d910d0b1906f0826dd8"}]},"branch":"refs/heads/master"},"fa511d87a6f9169cbbef624de0b96052c034bda8":{"kind":"NO_CODE_CHANGE","_number":12,"created":"2019-10-02 20:16:59.000000000","uploader":{"_account_id":28374,"name":"Viktor Michalek","email":"viktor.michalek@ultimum.io","username":"viktormichalek"},"ref":"refs/changes/79/683679/12","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/kolla","ref":"refs/changes/79/683679/12","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/kolla refs/changes/79/683679/12 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/kolla refs/changes/79/683679/12 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/kolla refs/changes/79/683679/12 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/kolla refs/changes/79/683679/12"}}},"commit":{"parents":[{"commit":"a9181bf89529756e4d58b22324988b4e57770ec1","subject":"Merge \"Update mention of Kayobe in documentation\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/a9181bf89529756e4d58b22324988b4e57770ec1"}]}],"author":{"name":"Jan Vondra","email":"jan.vondra@ultimum.io","date":"2019-09-20 16:45:52.000000000","tz":120},"committer":{"name":"Viktor Michalek","email":"viktor.michalek@ultimum.io","date":"2019-10-02 20:16:52.000000000","tz":0},"subject":"Neutron: add support to use legacy iptables","message":"Neutron: add support to use legacy iptables\n\nSetting legacy iptables for host system kernels not supporting \niptables-nft.\n\nChange-Id: I3b17bfdbc68a9a227bb22dc05a02ed6da607a114\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/fa511d87a6f9169cbbef624de0b96052c034bda8"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/fa511d87a6f9169cbbef624de0b96052c034bda8"}]},"branch":"refs/heads/master"},"6fb4d076391c8ca5447c6ede6f99ff830b9a54ff":{"kind":"REWORK","_number":13,"created":"2019-10-11 18:17:27.000000000","uploader":{"_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"},"ref":"refs/changes/79/683679/13","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/kolla","ref":"refs/changes/79/683679/13","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/kolla refs/changes/79/683679/13 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/kolla refs/changes/79/683679/13 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/kolla refs/changes/79/683679/13 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/kolla refs/changes/79/683679/13"}}},"commit":{"parents":[{"commit":"a9181bf89529756e4d58b22324988b4e57770ec1","subject":"Merge \"Update mention of Kayobe in documentation\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/a9181bf89529756e4d58b22324988b4e57770ec1"}]}],"author":{"name":"Jan Vondra","email":"jan.vondra@ultimum.io","date":"2019-09-20 16:45:52.000000000","tz":120},"committer":{"name":"Radosław Piliszek","email":"radoslaw.piliszek@gmail.com","date":"2019-10-11 18:17:03.000000000","tz":120},"subject":"Neutron: add support to use legacy iptables","message":"Neutron: add support to use legacy iptables\n\nSetting legacy iptables for host system kernels not supporting\niptables-nft.\n\nChange-Id: I3b17bfdbc68a9a227bb22dc05a02ed6da607a114\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/6fb4d076391c8ca5447c6ede6f99ff830b9a54ff"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/6fb4d076391c8ca5447c6ede6f99ff830b9a54ff"}]},"branch":"refs/heads/master"},"452f3ed3baef54babe4c77773702a21b81210d80":{"kind":"REWORK","_number":14,"created":"2019-10-11 18:45:26.000000000","uploader":{"_account_id":28374,"name":"Viktor Michalek","email":"viktor.michalek@ultimum.io","username":"viktormichalek"},"ref":"refs/changes/79/683679/14","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/kolla","ref":"refs/changes/79/683679/14","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/kolla refs/changes/79/683679/14 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/kolla refs/changes/79/683679/14 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/kolla refs/changes/79/683679/14 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/kolla refs/changes/79/683679/14"}}},"commit":{"parents":[{"commit":"a9181bf89529756e4d58b22324988b4e57770ec1","subject":"Merge \"Update mention of Kayobe in documentation\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/a9181bf89529756e4d58b22324988b4e57770ec1"}]}],"author":{"name":"Jan Vondra","email":"jan.vondra@ultimum.io","date":"2019-09-20 16:45:52.000000000","tz":120},"committer":{"name":"Viktor Michalek","email":"viktor.michalek@ultimum.io","date":"2019-10-11 18:45:19.000000000","tz":0},"subject":"Neutron: add support to use legacy iptables","message":"Neutron: add support to use legacy iptables\n\nSetting legacy iptables for host system kernels not supporting\niptables-nft.\n\nChange-Id: I3b17bfdbc68a9a227bb22dc05a02ed6da607a114\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/452f3ed3baef54babe4c77773702a21b81210d80"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/452f3ed3baef54babe4c77773702a21b81210d80"}]},"branch":"refs/heads/master"},"58f094432af277fcb8a428d78577865da8afdb62":{"kind":"TRIVIAL_REBASE","_number":15,"created":"2019-10-11 18:45:32.000000000","uploader":{"_account_id":28374,"name":"Viktor Michalek","email":"viktor.michalek@ultimum.io","username":"viktormichalek"},"ref":"refs/changes/79/683679/15","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/kolla","ref":"refs/changes/79/683679/15","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/kolla refs/changes/79/683679/15 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/kolla refs/changes/79/683679/15 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/kolla refs/changes/79/683679/15 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/kolla refs/changes/79/683679/15"}}},"commit":{"parents":[{"commit":"cf525a990059b684dae971a1905ea280d4866ed5","subject":"Mark images as deprecated in support matrix","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/cf525a990059b684dae971a1905ea280d4866ed5"}]}],"author":{"name":"Jan Vondra","email":"jan.vondra@ultimum.io","date":"2019-09-20 16:45:52.000000000","tz":120},"committer":{"name":"Viktor Michalek","email":"viktor.michalek@ultimum.io","date":"2019-10-11 18:45:32.000000000","tz":0},"subject":"Neutron: add support to use legacy iptables","message":"Neutron: add support to use legacy iptables\n\nSetting legacy iptables for host system kernels not supporting\niptables-nft.\n\nChange-Id: I3b17bfdbc68a9a227bb22dc05a02ed6da607a114\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/58f094432af277fcb8a428d78577865da8afdb62"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/58f094432af277fcb8a428d78577865da8afdb62"}]},"branch":"refs/heads/master"},"df56d5ad9ee44a73c201caf59fa7cb63abd625b5":{"kind":"REWORK","_number":16,"created":"2019-10-12 15:37:45.000000000","uploader":{"_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"},"ref":"refs/changes/79/683679/16","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/kolla","ref":"refs/changes/79/683679/16","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/kolla refs/changes/79/683679/16 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/kolla refs/changes/79/683679/16 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/kolla refs/changes/79/683679/16 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/kolla refs/changes/79/683679/16"}}},"commit":{"parents":[{"commit":"cf525a990059b684dae971a1905ea280d4866ed5","subject":"Mark images as deprecated in support matrix","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/cf525a990059b684dae971a1905ea280d4866ed5"}]}],"author":{"name":"Jan Vondra","email":"jan.vondra@ultimum.io","date":"2019-09-20 16:45:52.000000000","tz":120},"committer":{"name":"Radosław Piliszek","email":"radoslaw.piliszek@gmail.com","date":"2019-10-12 15:37:27.000000000","tz":0},"subject":"Neutron: add support to use legacy iptables","message":"Neutron: add support to use legacy iptables\n\nSetting legacy iptables for host system kernels not supporting\niptables-nft.\n\nChange-Id: I3b17bfdbc68a9a227bb22dc05a02ed6da607a114\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/df56d5ad9ee44a73c201caf59fa7cb63abd625b5"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/df56d5ad9ee44a73c201caf59fa7cb63abd625b5"}]},"branch":"refs/heads/master"},"a4ce044707579f5e781eaa703299119302971d84":{"kind":"REWORK","_number":17,"created":"2019-10-12 18:53:42.000000000","uploader":{"_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"},"ref":"refs/changes/79/683679/17","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/kolla","ref":"refs/changes/79/683679/17","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/kolla refs/changes/79/683679/17 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/kolla refs/changes/79/683679/17 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/kolla refs/changes/79/683679/17 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/kolla refs/changes/79/683679/17"}}},"commit":{"parents":[{"commit":"cf525a990059b684dae971a1905ea280d4866ed5","subject":"Mark images as deprecated in support matrix","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/cf525a990059b684dae971a1905ea280d4866ed5"}]}],"author":{"name":"Jan Vondra","email":"jan.vondra@ultimum.io","date":"2019-09-20 16:45:52.000000000","tz":120},"committer":{"name":"Radosław Piliszek","email":"radoslaw.piliszek@gmail.com","date":"2019-10-12 18:53:39.000000000","tz":0},"subject":"Neutron: add support to use legacy iptables","message":"Neutron: add support to use legacy iptables\n\nSetting legacy iptables for host system kernels not supporting\niptables-nft.\n\nChange-Id: I3b17bfdbc68a9a227bb22dc05a02ed6da607a114\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/a4ce044707579f5e781eaa703299119302971d84"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/a4ce044707579f5e781eaa703299119302971d84"}]},"branch":"refs/heads/master"},"e8289ff6df6b305a2f154a4aa6210c5d0bbb006f":{"kind":"REWORK","_number":18,"created":"2019-10-12 19:02:18.000000000","uploader":{"_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"},"ref":"refs/changes/79/683679/18","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/kolla","ref":"refs/changes/79/683679/18","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/kolla refs/changes/79/683679/18 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/kolla refs/changes/79/683679/18 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/kolla refs/changes/79/683679/18 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/kolla refs/changes/79/683679/18"}}},"commit":{"parents":[{"commit":"cf525a990059b684dae971a1905ea280d4866ed5","subject":"Mark images as deprecated in support matrix","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/cf525a990059b684dae971a1905ea280d4866ed5"}]}],"author":{"name":"Jan Vondra","email":"jan.vondra@ultimum.io","date":"2019-09-20 16:45:52.000000000","tz":120},"committer":{"name":"Radosław Piliszek","email":"radoslaw.piliszek@gmail.com","date":"2019-10-12 19:02:16.000000000","tz":0},"subject":"Neutron: add support to use legacy iptables","message":"Neutron: add support to use legacy iptables\n\nSetting legacy iptables for host system kernels not supporting\niptables-nft.\n\nChange-Id: I3b17bfdbc68a9a227bb22dc05a02ed6da607a114\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/e8289ff6df6b305a2f154a4aa6210c5d0bbb006f"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/e8289ff6df6b305a2f154a4aa6210c5d0bbb006f"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[],"submit_requirements":[]}
