)]}'
{"id":"openstack%2Fkolla~683681","triplet_id":"openstack%2Fkolla~master~I52933c535023e8681bbde887364a188450c15663","project":"openstack/kolla","branch":"master","topic":"680447","hashtags":[],"change_id":"I52933c535023e8681bbde887364a188450c15663","subject":"added broken images to UNBUILDABLE_IMAGES","status":"MERGED","created":"2019-09-20 18:14:28.000000000","updated":"2019-09-26 16:04:24.000000000","submitted":"2019-09-26 16:03:01.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"683681-1569513781730-91403195","meta_rev_id":"1ec450f7853e902ec562f6fdd99d0a69bced86bd","_number":683681,"virtual_id_number":683681,"owner":{"_account_id":28374,"name":"Viktor Michalek","email":"viktor.michalek@ultimum.io","username":"viktormichalek"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":28614,"name":"jacky06","email":"zhang.min@99cloud.net","username":"jacky06"},{"value":0,"_account_id":24072,"name":"Marcin Juszkiewicz","email":"mjuszkiewicz@redhat.com","username":"hrw"},{"value":0,"_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2019-09-26 16:03:01.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"date":"2019-09-26 13:22:23.000000000","_account_id":23181,"name":"RDO Third Party CI","email":"dmsimard+rdothirdparty@redhat.com","username":"rdothirdparty","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":24072,"name":"Marcin Juszkiewicz","email":"mjuszkiewicz@redhat.com","username":"hrw"},"recommended":{"_account_id":28614,"name":"jacky06","email":"zhang.min@99cloud.net","username":"jacky06"},"all":[{"value":1,"date":"2019-09-26 12:25:34.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":28614,"name":"jacky06","email":"zhang.min@99cloud.net","username":"jacky06"},{"value":2,"date":"2019-09-26 12:25:34.000000000","_account_id":24072,"name":"Marcin Juszkiewicz","email":"mjuszkiewicz@redhat.com","username":"hrw"},{"value":2,"date":"2019-09-26 12:25:34.000000000","_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":23181,"name":"RDO Third Party CI","email":"dmsimard+rdothirdparty@redhat.com","username":"rdothirdparty","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","value":1,"default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"},"all":[{"value":0,"_account_id":28614,"name":"jacky06","email":"zhang.min@99cloud.net","username":"jacky06"},{"value":0,"_account_id":24072,"name":"Marcin Juszkiewicz","email":"mjuszkiewicz@redhat.com","username":"hrw"},{"value":1,"date":"2019-09-26 12:25:46.000000000","_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":23181,"name":"RDO Third Party CI","email":"dmsimard+rdothirdparty@redhat.com","username":"rdothirdparty","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Backport-Candidate":{"all":[{"value":0,"_account_id":28614,"name":"jacky06","email":"zhang.min@99cloud.net","username":"jacky06"},{"value":0,"_account_id":24072,"name":"Marcin Juszkiewicz","email":"mjuszkiewicz@redhat.com","username":"hrw"},{"value":0,"_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":23181,"name":"RDO Third Party CI","email":"dmsimard+rdothirdparty@redhat.com","username":"rdothirdparty","tags":["SERVICE_USER"]}],"values":{"-1":"Do Not Backport"," 0":"Backport Review Needed","+1":"Should Backport"},"description":"","default_value":0,"optional":true},"Review-Priority":{"all":[{"value":0,"_account_id":28614,"name":"jacky06","email":"zhang.min@99cloud.net","username":"jacky06"},{"value":0,"_account_id":24072,"name":"Marcin Juszkiewicz","email":"mjuszkiewicz@redhat.com","username":"hrw"},{"value":0,"_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":23181,"name":"RDO Third Party CI","email":"dmsimard+rdothirdparty@redhat.com","username":"rdothirdparty","tags":["SERVICE_USER"]}],"values":{"-1":"Branch Freeze"," 0":"No Priority","+1":"Important Change","+2":"Gate Blocker Fix / Urgent Change"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":23181,"name":"RDO Third Party CI","email":"dmsimard+rdothirdparty@redhat.com","username":"rdothirdparty","tags":["SERVICE_USER"]},{"_account_id":24072,"name":"Marcin Juszkiewicz","email":"mjuszkiewicz@redhat.com","username":"hrw"},{"_account_id":28614,"name":"jacky06","email":"zhang.min@99cloud.net","username":"jacky06"},{"_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2019-09-20 20:43:22.000000000","updated_by":{"_account_id":24072,"name":"Marcin Juszkiewicz","email":"mjuszkiewicz@redhat.com","username":"hrw"},"reviewer":{"_account_id":24072,"name":"Marcin Juszkiewicz","email":"mjuszkiewicz@redhat.com","username":"hrw"},"state":"REVIEWER"},{"updated":"2019-09-25 15:13:18.000000000","updated_by":{"_account_id":28614,"name":"jacky06","email":"zhang.min@99cloud.net","username":"jacky06"},"reviewer":{"_account_id":28614,"name":"jacky06","email":"zhang.min@99cloud.net","username":"jacky06"},"state":"REVIEWER"},{"updated":"2019-09-26 12:25:46.000000000","updated_by":{"_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"},"reviewer":{"_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"},"state":"REVIEWER"},{"updated":"2019-09-26 13:22:23.000000000","updated_by":{"_account_id":23181,"name":"RDO Third Party CI","email":"dmsimard+rdothirdparty@redhat.com","username":"rdothirdparty","tags":["SERVICE_USER"]},"reviewer":{"_account_id":23181,"name":"RDO Third Party CI","email":"dmsimard+rdothirdparty@redhat.com","username":"rdothirdparty","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2019-09-26 16:03:01.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"}],"messages":[{"id":"83ac4a73fa4d7a3f775084dd12ed398bcbc85d08","author":{"_account_id":28374,"name":"Viktor Michalek","email":"viktor.michalek@ultimum.io","username":"viktormichalek"},"date":"2019-09-20 18:14:28.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"564cf2605f52a9d802a12c3d78cf9ba5e147b94c","author":{"_account_id":23181,"name":"RDO Third Party CI","email":"dmsimard+rdothirdparty@redhat.com","username":"rdothirdparty","tags":["SERVICE_USER"]},"date":"2019-09-20 18:59:09.000000000","message":"Patch Set 1:\n\nBuild succeeded (check pipeline).\n\n- tripleo-build-containers-rhel-8-master http://logs.rdoproject.org/81/683681/1/openstack-check/tripleo-build-containers-rhel-8-master/c6db157/ : SUCCESS in 42m 36s","accounts_in_message":[],"_revision_number":1},{"id":"bbaa4fba2ca45364241b6436385ae56998fc44ec","author":{"_account_id":24072,"name":"Marcin Juszkiewicz","email":"mjuszkiewicz@redhat.com","username":"hrw"},"date":"2019-09-20 20:43:22.000000000","message":"Patch Set 1: Code-Review+2","accounts_in_message":[],"_revision_number":1},{"id":"834feb2925f9a9c8d87f0d097950703b71fe549f","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-09-20 22:26:24.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\n\n- kolla-build-centos-binary https://zuul.opendev.org/t/openstack/build/f7bb2b9dd0f6493b92f16b0a9f0c0968 : SUCCESS in 53m 18s (non-voting)\n- kolla-build-centos-source https://zuul.opendev.org/t/openstack/build/5d5e56fb04e44b1aa89c3a150267db19 : SUCCESS in 1h 18m 42s\n- kolla-ansible-centos-source https://zuul.opendev.org/t/openstack/build/56d407f333be4fef88009dc1150b3d2e : SUCCESS in 1h 37m 12s\n- kolla-ansible-centos-binary https://zuul.opendev.org/t/openstack/build/d6d5629f63ce470380c243ffdb3fe353 : SUCCESS in 1h 26m 10s (non-voting)\n- kolla-ansible-centos-source-upgrade https://zuul.opendev.org/t/openstack/build/572896a165b14e4381a123d0b00a94bf : SUCCESS in 1h 34m 18s\n- tripleo-build-containers-centos-7 https://zuul.opendev.org/t/openstack/build/bde9f56bcaa54148ac4e791ded2d176a : SUCCESS in 1h 04m 04s\n- kolla-build-debian-source https://zuul.opendev.org/t/openstack/build/6b011e8383f64b3abcec2987d59ebf72 : SUCCESS in 24m 31s (non-voting)\n- kolla-build-debian-binary https://zuul.opendev.org/t/openstack/build/861413a6d1f94b08864e629117dc600c : SUCCESS in 19m 59s (non-voting)\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/6d387f8878644905bba3790e153dc2b5 : SUCCESS in 4m 18s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/1fe7e21049a74f91b61e803ef2716548 : SUCCESS in 5m 26s\n- openstack-tox-py27 https://zuul.opendev.org/t/openstack/build/39e991104264492ca95d037d49109966 : SUCCESS in 3m 49s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/c1f4bf2cb68242dc8542ec77f09679a6 : SUCCESS in 4m 04s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/79b829a986ed47c7a2e7c9a601e9bb80 : SUCCESS in 4m 33s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/95383699721842fcb3159afa9fe9d47b : SUCCESS in 7m 40s\n- kolla-build-ubuntu-binary https://zuul.opendev.org/t/openstack/build/b94193dc0daf4da6a0e6f175ac065edf : SUCCESS in 37m 24s (non-voting)\n- kolla-build-ubuntu-source https://zuul.opendev.org/t/openstack/build/25ef9326cfe84d24a583256dc495ad1e : SUCCESS in 45m 27s\n- kolla-ansible-ubuntu-source https://zuul.opendev.org/t/openstack/build/e2e9ce913fca49f186bb6214262379ce : SUCCESS in 1h 08m 16s\n- kolla-ansible-ubuntu-binary https://zuul.opendev.org/t/openstack/build/12240ec0790a4b0ab1109222a768f644 : SUCCESS in 1h 06m 44s (non-voting)\n- kolla-ansible-ubuntu-source-upgrade https://zuul.opendev.org/t/openstack/build/51eb5ad9750d4dada4091392d09cc54c : SUCCESS in 1h 07m 41s","accounts_in_message":[],"_revision_number":1},{"id":"c3099e97baf8a806ab4f74fd09b2131b5e93e99f","author":{"_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"},"date":"2019-09-23 09:20:47.000000000","message":"Patch Set 1: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":1},{"id":"31caf2023f2275a5f4635298b8bbf64233ee727e","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-09-23 17:38:59.000000000","message":"Patch Set 1: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttp://docs.openstack.org/infra/manual/developers.html#automated-testing\n\n\n\nWarning:\n  Dependency cycle detected","accounts_in_message":[],"_revision_number":1},{"id":"8eaf13ce173d6c4bdaa56382201319d0d3126c4e","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-09-23 17:41:15.000000000","message":"Patch Set 1:\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttp://docs.openstack.org/infra/manual/developers.html#automated-testing\n\n\n\nWarning:\n  Dependency cycle detected","accounts_in_message":[],"_revision_number":1},{"id":"4f1757f9592b65a9e02ac456fb671b3f4d00e960","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-09-23 19:04:08.000000000","message":"Patch Set 1:\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttp://docs.openstack.org/infra/manual/developers.html#automated-testing\n\n\n\nWarning:\n  Dependency cycle detected","accounts_in_message":[],"_revision_number":1},{"id":"54fa92b31c811e1a875f5483af20bcfee20d4ae8","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-09-23 19:18:30.000000000","message":"Patch Set 1:\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttp://docs.openstack.org/infra/manual/developers.html#automated-testing\n\n\n\nWarning:\n  Dependency cycle detected","accounts_in_message":[],"_revision_number":1},{"id":"821768d081d3f792daa2c85c9b00d82d0ad5d45d","author":{"_account_id":28374,"name":"Viktor Michalek","email":"viktor.michalek@ultimum.io","username":"viktormichalek"},"date":"2019-09-24 15:57:17.000000000","message":"Patch Set 2: Patch Set 1 was rebased","accounts_in_message":[],"_revision_number":2},{"id":"dbd0d98d09fe9abec348b892ce78999a2a4835e1","author":{"_account_id":23181,"name":"RDO Third Party CI","email":"dmsimard+rdothirdparty@redhat.com","username":"rdothirdparty","tags":["SERVICE_USER"]},"date":"2019-09-24 16:40:10.000000000","message":"Patch Set 2:\n\nBuild succeeded (check pipeline).\n\n- tripleo-build-containers-rhel-8-master http://logs.rdoproject.org/81/683681/2/openstack-check/tripleo-build-containers-rhel-8-master/41f21c9/ : SUCCESS in 39m 13s","accounts_in_message":[],"_revision_number":2},{"id":"ced2137342591ca883a980f05aac04fbc439c0c6","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-09-24 17:56:34.000000000","message":"Patch Set 2: Verified+1\n\nBuild succeeded (check pipeline).\n\n- kolla-build-centos-binary https://zuul.opendev.org/t/openstack/build/f39636e1f7954980aafc705888784df2 : SUCCESS in 1h 07m 13s (non-voting)\n- kolla-build-centos-source https://zuul.opendev.org/t/openstack/build/4e5a33385a064cd4aba924534b55d774 : SUCCESS in 1h 09m 29s\n- kolla-ansible-centos-source https://zuul.opendev.org/t/openstack/build/c2fc7a383d3c4a3e87781d40af85d7c1 : SUCCESS in 1h 32m 48s\n- kolla-ansible-centos-binary https://zuul.opendev.org/t/openstack/build/dd96c16f63234590b4b1c5521c4b08c0 : SUCCESS in 1h 24m 59s (non-voting)\n- kolla-ansible-centos-source-upgrade https://zuul.opendev.org/t/openstack/build/5ffa1cfde76848cebf9b879d1b2dbb6b : SUCCESS in 1h 43m 42s\n- tripleo-build-containers-centos-7 https://zuul.opendev.org/t/openstack/build/4d9eda07c4a8479ab579218900ae9f18 : SUCCESS in 1h 10m 53s\n- kolla-build-debian-source https://zuul.opendev.org/t/openstack/build/55a63cdb10814474b07c64b8b408de86 : SUCCESS in 32m 58s (non-voting)\n- kolla-build-debian-binary https://zuul.opendev.org/t/openstack/build/f7d509f186f9469fbacaddd9e7c0464f : FAILURE in 17m 26s (non-voting)\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/12c708357a694d1dbe41fa603f5a4308 : SUCCESS in 6m 01s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/40d3006ca5e5491aa3b676b904cc5116 : SUCCESS in 6m 38s\n- openstack-tox-py27 https://zuul.opendev.org/t/openstack/build/af1770d830154986bd8a74bcab2fc3f2 : SUCCESS in 2m 59s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/27ede457c862443babe8eb503710ebb3 : SUCCESS in 3m 51s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/e969668a463148d3beb60a3a0344f5e5 : SUCCESS in 3m 59s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/8b13c134ca274c9c8ad99385d1e732db : SUCCESS in 6m 58s\n- kolla-build-ubuntu-binary https://zuul.opendev.org/t/openstack/build/dc79aaa66f9b49a2b7da53be40cce5c9 : SUCCESS in 38m 37s (non-voting)\n- kolla-build-ubuntu-source https://zuul.opendev.org/t/openstack/build/5c160b7ec22146f3bba60dd5819ddb1b : SUCCESS in 35m 25s\n- kolla-ansible-ubuntu-source https://zuul.opendev.org/t/openstack/build/dc769b2f62e54dd6aa7fc12d88fecd44 : SUCCESS in 1h 22m 14s\n- kolla-ansible-ubuntu-binary https://zuul.opendev.org/t/openstack/build/15613d032579403cb2b45e364af01a77 : SUCCESS in 1h 15m 11s (non-voting)\n- kolla-ansible-ubuntu-source-upgrade https://zuul.opendev.org/t/openstack/build/f95115ca0629426fb38c4129912e116a : SUCCESS in 1h 38m 51s","accounts_in_message":[],"_revision_number":2},{"id":"3be80dbe563ff0efcd27d706523d7bf06e86e8bb","author":{"_account_id":28614,"name":"jacky06","email":"zhang.min@99cloud.net","username":"jacky06"},"date":"2019-09-25 15:13:18.000000000","message":"Patch Set 2: Code-Review+1","accounts_in_message":[],"_revision_number":2},{"id":"507c901e67eef978091cd88231db84ce61a9759c","author":{"_account_id":24072,"name":"Marcin Juszkiewicz","email":"mjuszkiewicz@redhat.com","username":"hrw"},"date":"2019-09-25 22:57:17.000000000","message":"Patch Set 2: Workflow+1","accounts_in_message":[],"_revision_number":2},{"id":"e01289e6401c905c21e7e17a3fbb914257f70c2d","author":{"_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"},"date":"2019-09-26 12:25:25.000000000","message":"Patch Set 3: Patch Set 2 was rebased","accounts_in_message":[],"_revision_number":3},{"id":"f7c63ddbd6504909949f49d187a9a0169e5ee551","author":{"_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"},"date":"2019-09-26 12:25:34.000000000","message":"Patch Set 4: Patch Set 3 was rebased","accounts_in_message":[],"_revision_number":4},{"id":"89604a102475a6c1f86ce1a579e359a76791b05f","author":{"_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"},"date":"2019-09-26 12:25:46.000000000","message":"Patch Set 4: Workflow+1\n\nuntangled","accounts_in_message":[],"_revision_number":4},{"id":"033cb97d9d5e0c1be8d205cc19a5d2b652db2d3c","author":{"_account_id":23181,"name":"RDO Third Party CI","email":"dmsimard+rdothirdparty@redhat.com","username":"rdothirdparty","tags":["SERVICE_USER"]},"date":"2019-09-26 13:22:23.000000000","message":"Patch Set 4:\n\nBuild succeeded (check pipeline).\n\n- tripleo-build-containers-rhel-8-master http://logs.rdoproject.org/81/683681/4/openstack-check/tripleo-build-containers-rhel-8-master/5cf1761/ : SUCCESS in 37m 40s","accounts_in_message":[],"_revision_number":4},{"id":"6b4773a9ff5e984d2d3fe368883fed1d509f893a","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-09-26 14:16:57.000000000","message":"Patch Set 4: Verified+1\n\nBuild succeeded (check pipeline).\n\n- kolla-build-centos-binary https://zuul.opendev.org/t/openstack/build/e72b9b8096bb454aac9729c7a725a1e3 : SUCCESS in 48m 49s (non-voting)\n- kolla-build-centos-source https://zuul.opendev.org/t/openstack/build/e7470c2011fc41d6800955675846372f : SUCCESS in 1h 06m 18s\n- kolla-ansible-centos-source https://zuul.opendev.org/t/openstack/build/7c3fd494fc954a3881d7d3149d245c30 : SUCCESS in 1h 17m 25s\n- kolla-ansible-centos-binary https://zuul.opendev.org/t/openstack/build/4f6f27e9fd374d8f839a9bc296c8f373 : SUCCESS in 1h 26m 22s (non-voting)\n- kolla-ansible-centos-source-upgrade https://zuul.opendev.org/t/openstack/build/faf78b8ffadc4c7d9468f8dc3785ad9e : SUCCESS in 1h 48m 41s\n- tripleo-build-containers-centos-7 https://zuul.opendev.org/t/openstack/build/07032d77e9c24df0a3e633dea5cd5650 : SUCCESS in 1h 05m 28s\n- kolla-build-debian-source https://zuul.opendev.org/t/openstack/build/b1d0555586b843fab096b899ddbbcb86 : SUCCESS in 37m 06s (non-voting)\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/ec9ef62688aa405492f0abb3ee841471 : SUCCESS in 4m 54s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/82732acf94694ef9b266dcbb7e3fafe1 : SUCCESS in 3m 53s\n- openstack-tox-py27 https://zuul.opendev.org/t/openstack/build/e0a2af686f31457aa5d94dec01b6611a : SUCCESS in 3m 19s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/4ab7913cc0374239bf12d3e8a78f25a5 : SUCCESS in 2m 49s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/997002ca63824aad9c1d9f5c4474c8e2 : SUCCESS in 3m 40s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/5c9a253e09ad437e853ed67c1b065a5a : SUCCESS in 6m 01s\n- kolla-build-ubuntu-binary https://zuul.opendev.org/t/openstack/build/b1d05a1fcd2048dda7e3a1e3fb3a8d30 : SUCCESS in 49m 38s (non-voting)\n- kolla-build-ubuntu-source https://zuul.opendev.org/t/openstack/build/eb6f6bf2e4904ae18582e97db6ff5341 : SUCCESS in 35m 20s\n- kolla-ansible-ubuntu-source https://zuul.opendev.org/t/openstack/build/6c6e5110917445fe948f4e12481a1ce4 : SUCCESS in 1h 03m 29s\n- kolla-ansible-ubuntu-binary https://zuul.opendev.org/t/openstack/build/b94ac5d0c36d4159acda837cc922c2eb : SUCCESS in 54m 00s (non-voting)\n- kolla-ansible-ubuntu-source-upgrade https://zuul.opendev.org/t/openstack/build/660b08036f804fc89dd2bebfeeba4f3e : SUCCESS in 1h 04m 24s","accounts_in_message":[],"_revision_number":4},{"id":"117ff6bbbc20b77c1e4f3339fe6c24dc89817f98","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-09-26 14:17:10.000000000","message":"Patch Set 4: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":4},{"id":"22a05abb1957c8586591d6f7aac7d6d5098494e9","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-09-26 16:03:01.000000000","message":"Patch Set 4: Verified+2\n\nBuild succeeded (gate pipeline).\n\n- kolla-build-centos-source https://zuul.opendev.org/t/openstack/build/d62ac11561d446d8abda8bac9d62496c : SUCCESS in 1h 20m 34s\n- kolla-ansible-centos-source https://zuul.opendev.org/t/openstack/build/d535cb08896e40e0a6f2d7a217d87ec3 : SUCCESS in 1h 33m 30s\n- kolla-ansible-centos-source-upgrade https://zuul.opendev.org/t/openstack/build/0e6d8fdc248b42ddbd94012398d02d7c : SUCCESS in 1h 40m 15s\n- tripleo-build-containers-centos-7 https://zuul.opendev.org/t/openstack/build/da4a31ba37034984860e2415b4d33072 : SUCCESS in 54m 00s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/e119f3df9d1b4a78a29c4a7b3f7db51b : SUCCESS in 4m 11s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/10661d9ec4d94db78d5d061673e08d01 : SUCCESS in 5m 20s\n- openstack-tox-py27 https://zuul.opendev.org/t/openstack/build/25fe277786064b35a7b99619e3de90e5 : SUCCESS in 3m 49s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/5d559d9e4ccb4d5d91176dc92f9c3a24 : SUCCESS in 3m 11s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/32efe7dca02d4f69927ccfac923376b8 : SUCCESS in 4m 32s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/0c45ba1e772d4188881a66d815b4befb : SUCCESS in 6m 33s\n- kolla-build-ubuntu-source https://zuul.opendev.org/t/openstack/build/8f6dd80a50d743538b3fa212df4f94c7 : SUCCESS in 35m 42s\n- kolla-ansible-ubuntu-source https://zuul.opendev.org/t/openstack/build/20f0c9285fb74d4d8b80185ad17f26b1 : SUCCESS in 54m 43s\n- kolla-ansible-ubuntu-source-upgrade https://zuul.opendev.org/t/openstack/build/37830d28421a410ea4f3128294433dff : SUCCESS in 1h 02m 24s","accounts_in_message":[],"_revision_number":4},{"id":"e7fc42ec014aa9fcbd9b9d8b1ffb420ff8b97409","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-09-26 16:03:01.000000000","message":"Change has been successfully merged by Zuul","accounts_in_message":[],"_revision_number":4},{"id":"fbe62ef9d10ed3cebd601009c06b2496ed07200e","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-09-26 16:04:24.000000000","message":"Patch Set 4:\n\nBuild succeeded (promote pipeline).\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/be1700a11d064e3a89b2a2a0ad7b95d0 : SUCCESS in 1m 00s","accounts_in_message":[],"_revision_number":4}],"current_revision_number":4,"current_revision":"867c5d4bc4a9cadc87ed00bb4f15d359b4d038cc","revisions":{"ba4de452c8c28aa5be31062675baaa229d0f5968":{"kind":"REWORK","_number":1,"created":"2019-09-20 18:14:28.000000000","uploader":{"_account_id":28374,"name":"Viktor Michalek","email":"viktor.michalek@ultimum.io","username":"viktormichalek"},"ref":"refs/changes/81/683681/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/kolla","ref":"refs/changes/81/683681/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/kolla refs/changes/81/683681/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/kolla refs/changes/81/683681/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/kolla refs/changes/81/683681/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/kolla refs/changes/81/683681/1"}}},"commit":{"parents":[{"commit":"4817324324718a63b0522450eab4e1bb0e0ee5a1","subject":"Placement: Debian pkgs do not include migration script","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/4817324324718a63b0522450eab4e1bb0e0ee5a1"}]}],"author":{"name":"Marcin Juszkiewicz","email":"marcin.juszkiewicz@linaro.org","date":"2019-09-20 17:08:09.000000000","tz":120},"committer":{"name":"Viktor Michalek","email":"viktor.michalek@ultimum.io","date":"2019-09-20 17:08:09.000000000","tz":120},"subject":"added broken images to UNBUILDABLE_IMAGES","message":"added broken images to UNBUILDABLE_IMAGES\n\nChange-Id: I52933c535023e8681bbde887364a188450c15663\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/ba4de452c8c28aa5be31062675baaa229d0f5968"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/ba4de452c8c28aa5be31062675baaa229d0f5968"}]},"branch":"refs/heads/master"},"7f7b8527dc54c3150e735c571ecc3a07632b9c07":{"kind":"TRIVIAL_REBASE","_number":2,"created":"2019-09-24 15:57:17.000000000","uploader":{"_account_id":28374,"name":"Viktor Michalek","email":"viktor.michalek@ultimum.io","username":"viktormichalek"},"ref":"refs/changes/81/683681/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/kolla","ref":"refs/changes/81/683681/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/kolla refs/changes/81/683681/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/kolla refs/changes/81/683681/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/kolla refs/changes/81/683681/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/kolla refs/changes/81/683681/2"}}},"commit":{"parents":[{"commit":"924cab6549ca29c99b18ad88a51229c871d0e2ef","subject":"Placement: Debian pkgs has migration script in different path","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/924cab6549ca29c99b18ad88a51229c871d0e2ef"}]}],"author":{"name":"Marcin Juszkiewicz","email":"marcin.juszkiewicz@linaro.org","date":"2019-09-20 17:08:09.000000000","tz":120},"committer":{"name":"Viktor Michalek","email":"viktor.michalek@ultimum.io","date":"2019-09-24 15:57:17.000000000","tz":0},"subject":"added broken images to UNBUILDABLE_IMAGES","message":"added broken images to UNBUILDABLE_IMAGES\n\nChange-Id: I52933c535023e8681bbde887364a188450c15663\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/7f7b8527dc54c3150e735c571ecc3a07632b9c07"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/7f7b8527dc54c3150e735c571ecc3a07632b9c07"}]},"branch":"refs/heads/master"},"ea3387d8a5de9a622a7ef160f4f01301074e18eb":{"kind":"TRIVIAL_REBASE","_number":3,"created":"2019-09-26 12:25:25.000000000","uploader":{"_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"},"ref":"refs/changes/81/683681/3","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/kolla","ref":"refs/changes/81/683681/3","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/kolla refs/changes/81/683681/3 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/kolla refs/changes/81/683681/3 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/kolla refs/changes/81/683681/3 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/kolla refs/changes/81/683681/3"}}},"commit":{"parents":[{"commit":"76c44d34883023c0b31eff8b0342da0380b84578","subject":"Placement: Debian pkg has migration script in different path","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/76c44d34883023c0b31eff8b0342da0380b84578"}]}],"author":{"name":"Marcin Juszkiewicz","email":"marcin.juszkiewicz@linaro.org","date":"2019-09-20 17:08:09.000000000","tz":120},"committer":{"name":"Radosław Piliszek","email":"radoslaw.piliszek@gmail.com","date":"2019-09-26 12:25:25.000000000","tz":0},"subject":"added broken images to UNBUILDABLE_IMAGES","message":"added broken images to UNBUILDABLE_IMAGES\n\nChange-Id: I52933c535023e8681bbde887364a188450c15663\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/ea3387d8a5de9a622a7ef160f4f01301074e18eb"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/ea3387d8a5de9a622a7ef160f4f01301074e18eb"}]},"branch":"refs/heads/master"},"867c5d4bc4a9cadc87ed00bb4f15d359b4d038cc":{"kind":"TRIVIAL_REBASE","_number":4,"created":"2019-09-26 12:25:34.000000000","uploader":{"_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"},"ref":"refs/changes/81/683681/4","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/kolla","ref":"refs/changes/81/683681/4","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/kolla refs/changes/81/683681/4 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/kolla refs/changes/81/683681/4 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/kolla refs/changes/81/683681/4 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/kolla refs/changes/81/683681/4"}}},"commit":{"parents":[{"commit":"2131ef9875efd67a9dd45dc3329df8de16cacff1","subject":"Fluentd: td-agent repository for Debian","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/2131ef9875efd67a9dd45dc3329df8de16cacff1"}]}],"author":{"name":"Marcin Juszkiewicz","email":"marcin.juszkiewicz@linaro.org","date":"2019-09-20 17:08:09.000000000","tz":120},"committer":{"name":"Radosław Piliszek","email":"radoslaw.piliszek@gmail.com","date":"2019-09-26 12:25:34.000000000","tz":0},"subject":"added broken images to UNBUILDABLE_IMAGES","message":"added broken images to UNBUILDABLE_IMAGES\n\nChange-Id: I52933c535023e8681bbde887364a188450c15663\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/867c5d4bc4a9cadc87ed00bb4f15d359b4d038cc"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/867c5d4bc4a9cadc87ed00bb4f15d359b4d038cc"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[],"submit_requirements":[]}
