)]}'
{"id":"openstack%2Fkolla~683982","triplet_id":"openstack%2Fkolla~stable%2Fstein~I6e893c25a9dfbbbac6bd69f2fa26e6c72a4154d7","project":"openstack/kolla","branch":"stable/stein","hashtags":[],"change_id":"I6e893c25a9dfbbbac6bd69f2fa26e6c72a4154d7","subject":"[docker/openstack-exporter] Bump to 0.2.1 version","status":"MERGED","created":"2019-09-23 14:41:59.000000000","updated":"2019-09-24 08:46:44.000000000","submitted":"2019-09-24 08:45:03.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"683982-1569314703839-f2a16230","meta_rev_id":"e42a96f9472cedbdaabc67adf183af036747d070","_number":683982,"virtual_id_number":683982,"owner":{"_account_id":30272,"name":"Keith Plant","email":"kplantjr@gmail.com","username":"kplant"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":19316,"name":"Eduardo Gonzalez","email":"dabarren@gmail.com","username":"egonzalez90"},{"value":0,"_account_id":2276,"name":"Jorge Niedbalski","email":"jorge.niedbalski@canonical.com","username":"niedbalski"},{"value":0,"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},{"value":0,"_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2019-09-24 08:45:03.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":19316,"name":"Eduardo Gonzalez","email":"dabarren@gmail.com","username":"egonzalez90"},"all":[{"value":2,"date":"2019-09-24 07:18:09.000000000","_account_id":19316,"name":"Eduardo Gonzalez","email":"dabarren@gmail.com","username":"egonzalez90"},{"value":0,"_account_id":2276,"name":"Jorge Niedbalski","email":"jorge.niedbalski@canonical.com","username":"niedbalski"},{"value":2,"date":"2019-09-23 17:00:56.000000000","_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},{"value":2,"date":"2019-09-24 07:17:14.000000000","_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"},"all":[{"value":0,"_account_id":19316,"name":"Eduardo Gonzalez","email":"dabarren@gmail.com","username":"egonzalez90"},{"value":0,"_account_id":2276,"name":"Jorge Niedbalski","email":"jorge.niedbalski@canonical.com","username":"niedbalski"},{"value":0,"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},{"value":1,"date":"2019-09-24 07:17:14.000000000","_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Backport-Candidate":{"all":[{"value":0,"_account_id":19316,"name":"Eduardo Gonzalez","email":"dabarren@gmail.com","username":"egonzalez90"},{"value":0,"_account_id":2276,"name":"Jorge Niedbalski","email":"jorge.niedbalski@canonical.com","username":"niedbalski"},{"value":0,"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},{"value":0,"_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Do Not Backport"," 0":"Backport Review Needed","+1":"Should Backport"},"description":"","default_value":0,"optional":true},"Review-Priority":{"all":[{"value":0,"_account_id":19316,"name":"Eduardo Gonzalez","email":"dabarren@gmail.com","username":"egonzalez90"},{"value":0,"_account_id":2276,"name":"Jorge Niedbalski","email":"jorge.niedbalski@canonical.com","username":"niedbalski"},{"value":0,"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},{"value":0,"_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Branch Freeze"," 0":"No Priority","+1":"Important Change","+2":"Gate Blocker Fix / Urgent Change"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":2276,"name":"Jorge Niedbalski","email":"jorge.niedbalski@canonical.com","username":"niedbalski"},{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},{"_account_id":19316,"name":"Eduardo Gonzalez","email":"dabarren@gmail.com","username":"egonzalez90"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2019-09-23 14:41:59.000000000","updated_by":{"_account_id":2276,"name":"Jorge Niedbalski","email":"jorge.niedbalski@canonical.com","username":"niedbalski"},"reviewer":{"_account_id":2276,"name":"Jorge Niedbalski","email":"jorge.niedbalski@canonical.com","username":"niedbalski"},"state":"REVIEWER"},{"updated":"2019-09-23 17:00:56.000000000","updated_by":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},"reviewer":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},"state":"REVIEWER"},{"updated":"2019-09-24 07:17:14.000000000","updated_by":{"_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"},"reviewer":{"_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"},"state":"REVIEWER"},{"updated":"2019-09-24 07:18:09.000000000","updated_by":{"_account_id":19316,"name":"Eduardo Gonzalez","email":"dabarren@gmail.com","username":"egonzalez90"},"reviewer":{"_account_id":19316,"name":"Eduardo Gonzalez","email":"dabarren@gmail.com","username":"egonzalez90"},"state":"REVIEWER"},{"updated":"2019-09-24 08:45:03.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"}],"messages":[{"id":"847ba5fc5f029f49ca85c1f78c7e1fb4a45903c8","author":{"_account_id":30272,"name":"Keith Plant","email":"kplantjr@gmail.com","username":"kplant"},"date":"2019-09-23 14:41:59.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"c17b00454c174f055d1d2760d66849cff2cae299","author":{"_account_id":30272,"name":"Keith Plant","email":"kplantjr@gmail.com","username":"kplant"},"date":"2019-09-23 14:43:20.000000000","message":"Uploaded patch set 2.","accounts_in_message":[],"_revision_number":2},{"id":"d9b8086e2758500fcf3b6859f64cec8a285d79d4","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-09-23 16:57:34.000000000","message":"Patch Set 2: Verified+1\n\nBuild succeeded (check pipeline).\n\n- kolla-build-centos-binary https://zuul.opendev.org/t/openstack/build/1bf394c21f79424aaf93816dde5d6f42 : SUCCESS in 53m 49s (non-voting)\n- kolla-build-centos-source https://zuul.opendev.org/t/openstack/build/408152015258407d82f327f25e390bf1 : SUCCESS in 1h 00m 39s\n- kolla-ansible-centos-source https://zuul.opendev.org/t/openstack/build/86c157f77c5543b28d5496cbff9f25e9 : SUCCESS in 1h 27m 19s\n- kolla-ansible-centos-binary https://zuul.opendev.org/t/openstack/build/b32e2d4a847e4bf58567ccf7ecde8c5c : SUCCESS in 1h 06m 31s (non-voting)\n- kolla-ansible-centos-source-upgrade https://zuul.opendev.org/t/openstack/build/6a7ad2bd7e45455faa7ad222f39e2f20 : SUCCESS in 1h 36m 24s\n- tripleo-build-containers-centos-7 https://zuul.opendev.org/t/openstack/build/8f5aac098fe04e9993355d2b5c8d06bc : SUCCESS in 1h 05m 16s (non-voting)\n- kolla-build-oraclelinux-binary https://zuul.opendev.org/t/openstack/build/b562786e4aa8415eb87c238202e1036b : SUCCESS in 52m 13s (non-voting)\n- kolla-build-oraclelinux-source https://zuul.opendev.org/t/openstack/build/5962251f9ccf45e997d473c9698d5035 : SUCCESS in 47m 38s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/a6bda009696f45c0b6957cb85e3ffc50 : SUCCESS in 4m 42s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/7bf7e98608874d108c9ee1dad410d05e : SUCCESS in 3m 54s\n- openstack-tox-py27 https://zuul.opendev.org/t/openstack/build/2aef9b35f13c456eaf1079d6f8e764f5 : SUCCESS in 4m 07s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/7551ea8077824f8084f1b7d3818f3345 : SUCCESS in 2m 37s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/125213ca3c0d4633a3b7918dcfc80bd2 : SUCCESS in 3m 36s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/e6a51c52b13447bd893d0472a51c5353 : SUCCESS in 3m 57s\n- kolla-build-ubuntu-binary https://zuul.opendev.org/t/openstack/build/3728839719f24d20a72f218f8b0aa2b3 : SUCCESS in 40m 45s (non-voting)\n- kolla-build-ubuntu-source https://zuul.opendev.org/t/openstack/build/1f798aac3d514396b89eb5b981385622 : SUCCESS in 31m 45s\n- kolla-ansible-ubuntu-source https://zuul.opendev.org/t/openstack/build/3e87e2180e9542bcb7571057502a7835 : SUCCESS in 53m 11s\n- kolla-ansible-ubuntu-binary https://zuul.opendev.org/t/openstack/build/6d3daace6d984a2391dde7e190f8d403 : SUCCESS in 48m 25s (non-voting)\n- kolla-ansible-ubuntu-source-upgrade https://zuul.opendev.org/t/openstack/build/8e93247deb5d4c4280beda9e450b9dd0 : SUCCESS in 53m 58s","accounts_in_message":[],"_revision_number":2},{"id":"bc1464615a0082e9a9fa9f23cdba2d567938d1eb","author":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},"date":"2019-09-23 17:00:56.000000000","message":"Patch Set 2: Code-Review+2","accounts_in_message":[],"_revision_number":2},{"id":"738d8d53cdffc9951029bd22a0c9b32bbb528d65","author":{"_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"},"date":"2019-09-24 07:17:14.000000000","message":"Patch Set 2: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":2},{"id":"ff024404a660123e71f70e86333026091a35620d","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-09-24 07:17:27.000000000","message":"Patch Set 2: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":2},{"id":"df288c40307b4cd507a66807c853901bec4b56f1","author":{"_account_id":19316,"name":"Eduardo Gonzalez","email":"dabarren@gmail.com","username":"egonzalez90"},"date":"2019-09-24 07:18:09.000000000","message":"Patch Set 2: Code-Review+2","accounts_in_message":[],"_revision_number":2},{"id":"fd516b1d93db739e61435c9e420d95787108bf69","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-09-24 08:45:03.000000000","message":"Change has been successfully merged by Zuul","accounts_in_message":[],"_revision_number":2},{"id":"b56e1b6a8415e9f1af2d8abd05f3eed5298e6d9f","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-09-24 08:45:03.000000000","message":"Patch Set 2: Verified+2\n\nBuild succeeded (gate pipeline).\n\n- kolla-build-centos-source https://zuul.opendev.org/t/openstack/build/17cc7b87b073403d96c4424b82926de9 : SUCCESS in 42m 43s\n- kolla-build-oraclelinux-source https://zuul.opendev.org/t/openstack/build/60f71a8843ad4d03988552b49499d5ce : SUCCESS in 46m 45s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/22a138f78b3e46b6bb343f55872dcdd3 : SUCCESS in 3m 42s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/c743486856e1414187b9323ceea7918d : SUCCESS in 3m 32s\n- openstack-tox-py27 https://zuul.opendev.org/t/openstack/build/87c38585c83449df87922bb30ddcf0c5 : SUCCESS in 3m 25s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/2ea9d1ae1f5944bd8d63bc334229dff3 : SUCCESS in 3m 25s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/6cc9c04e626d4a4a9f9a7895ff800aeb : SUCCESS in 3m 48s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/e99b814db86d4ed5b1a7532f15d29d23 : SUCCESS in 6m 44s\n- kolla-build-ubuntu-source https://zuul.opendev.org/t/openstack/build/d68caafae06542dcb7d5195c85ec8dd1 : SUCCESS in 28m 35s","accounts_in_message":[],"_revision_number":2},{"id":"e42a96f9472cedbdaabc67adf183af036747d070","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-09-24 08:46:44.000000000","message":"Patch Set 2:\n\nBuild succeeded (promote pipeline).\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/76ce790b11894e81acd9b4441a5e4971 : SUCCESS in 1m 16s","accounts_in_message":[],"_revision_number":2}],"current_revision_number":2,"current_revision":"89bf48dac0b64d9386761cbc29942905b0d8033c","revisions":{"f331dfa89832a35eb611035654e4471bb5966352":{"kind":"REWORK","_number":1,"created":"2019-09-23 14:41:59.000000000","uploader":{"_account_id":30272,"name":"Keith Plant","email":"kplantjr@gmail.com","username":"kplant"},"ref":"refs/changes/82/683982/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/kolla","ref":"refs/changes/82/683982/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/kolla refs/changes/82/683982/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/kolla refs/changes/82/683982/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/kolla refs/changes/82/683982/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/kolla refs/changes/82/683982/1"}}},"commit":{"parents":[{"commit":"44be1294c1b801f01c5f989355c4486afd6b7ef3","subject":"sensu-client: install older whois gem","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/44be1294c1b801f01c5f989355c4486afd6b7ef3"}]}],"author":{"name":"Jorge Niedbalski","email":"jnr@metaklass.org","date":"2019-07-29 15:15:38.000000000","tz":-240},"committer":{"name":"Keith Plant","email":"kplantjr@gmail.com","date":"2019-09-23 14:41:58.000000000","tz":-240},"subject":"[docker/openstack-exporter] Bump to 0.2.1 version","message":"[docker/openstack-exporter] Bump to 0.2.1 version\n\nBump the default version to 0.2.1 of the\nopenstack-exporter.\n\nChange-Id: I6e893c25a9dfbbbac6bd69f2fa26e6c72a4154d7\nSigned-off-by: Jorge Niedbalski \u003cjnr@metaklass.org\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/f331dfa89832a35eb611035654e4471bb5966352"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/f331dfa89832a35eb611035654e4471bb5966352"}]},"branch":"refs/heads/stable/stein"},"89bf48dac0b64d9386761cbc29942905b0d8033c":{"kind":"REWORK","_number":2,"created":"2019-09-23 14:43:20.000000000","uploader":{"_account_id":30272,"name":"Keith Plant","email":"kplantjr@gmail.com","username":"kplant"},"ref":"refs/changes/82/683982/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/kolla","ref":"refs/changes/82/683982/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/kolla refs/changes/82/683982/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/kolla refs/changes/82/683982/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/kolla refs/changes/82/683982/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/kolla refs/changes/82/683982/2"}}},"commit":{"parents":[{"commit":"44be1294c1b801f01c5f989355c4486afd6b7ef3","subject":"sensu-client: install older whois gem","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/44be1294c1b801f01c5f989355c4486afd6b7ef3"}]}],"author":{"name":"Jorge Niedbalski","email":"jnr@metaklass.org","date":"2019-07-29 15:15:38.000000000","tz":-240},"committer":{"name":"Keith Plant","email":"kplantjr@gmail.com","date":"2019-09-23 14:43:13.000000000","tz":-240},"subject":"[docker/openstack-exporter] Bump to 0.2.1 version","message":"[docker/openstack-exporter] Bump to 0.2.1 version\n\nBump the default version to 0.2.1 of the\nopenstack-exporter.\n\nChange-Id: I6e893c25a9dfbbbac6bd69f2fa26e6c72a4154d7\nSigned-off-by: Jorge Niedbalski \u003cjnr@metaklass.org\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/89bf48dac0b64d9386761cbc29942905b0d8033c"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/89bf48dac0b64d9386761cbc29942905b0d8033c"}]},"branch":"refs/heads/stable/stein"}},"requirements":[],"submit_records":[],"submit_requirements":[]}
