)]}'
{"id":"openstack%2Fkolla~684551","triplet_id":"openstack%2Fkolla~master~I4d2449d7d100c68084f937ef9981596fc1625375","project":"openstack/kolla","branch":"master","topic":"debian-td-agent","hashtags":[],"change_id":"I4d2449d7d100c68084f937ef9981596fc1625375","subject":"Debian: switch fluentd to td-agent on x86_64","status":"MERGED","created":"2019-09-25 07:56:31.000000000","updated":"2019-09-27 19:02:48.000000000","submitted":"2019-09-27 19:01:26.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":4,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"684551-1569610887231-91403195","meta_rev_id":"d4639f1b7c7f221debda549a723f08520fef14c2","_number":684551,"virtual_id_number":684551,"owner":{"_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":24072,"name":"Marcin Juszkiewicz","email":"mjuszkiewicz@redhat.com","username":"hrw"},{"value":0,"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},{"value":0,"date":"2019-09-26 07:49:50.000000000","_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2019-09-27 19:01:26.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"date":"2019-09-26 08:30:13.000000000","_account_id":23181,"name":"RDO Third Party CI","email":"dmsimard+rdothirdparty@redhat.com","username":"rdothirdparty","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":24072,"name":"Marcin Juszkiewicz","email":"mjuszkiewicz@redhat.com","username":"hrw"},"all":[{"value":2,"date":"2019-09-27 17:03:46.000000000","_account_id":24072,"name":"Marcin Juszkiewicz","email":"mjuszkiewicz@redhat.com","username":"hrw"},{"value":2,"date":"2019-09-26 09:56:21.000000000","_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},{"value":0,"_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":23181,"name":"RDO Third Party CI","email":"dmsimard+rdothirdparty@redhat.com","username":"rdothirdparty","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":24072,"name":"Marcin Juszkiewicz","email":"mjuszkiewicz@redhat.com","username":"hrw"},"all":[{"value":1,"date":"2019-09-27 17:03:46.000000000","_account_id":24072,"name":"Marcin Juszkiewicz","email":"mjuszkiewicz@redhat.com","username":"hrw"},{"value":0,"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},{"value":0,"_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":23181,"name":"RDO Third Party CI","email":"dmsimard+rdothirdparty@redhat.com","username":"rdothirdparty","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Backport-Candidate":{"all":[{"value":0,"_account_id":24072,"name":"Marcin Juszkiewicz","email":"mjuszkiewicz@redhat.com","username":"hrw"},{"value":0,"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},{"value":0,"_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":23181,"name":"RDO Third Party CI","email":"dmsimard+rdothirdparty@redhat.com","username":"rdothirdparty","tags":["SERVICE_USER"]}],"values":{"-1":"Do Not Backport"," 0":"Backport Review Needed","+1":"Should Backport"},"description":"","default_value":0,"optional":true},"Review-Priority":{"all":[{"value":0,"_account_id":24072,"name":"Marcin Juszkiewicz","email":"mjuszkiewicz@redhat.com","username":"hrw"},{"value":0,"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},{"value":0,"_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":23181,"name":"RDO Third Party CI","email":"dmsimard+rdothirdparty@redhat.com","username":"rdothirdparty","tags":["SERVICE_USER"]}],"values":{"-1":"Branch Freeze"," 0":"No Priority","+1":"Important Change","+2":"Gate Blocker Fix / Urgent Change"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":23181,"name":"RDO Third Party CI","email":"dmsimard+rdothirdparty@redhat.com","username":"rdothirdparty","tags":["SERVICE_USER"]},{"_account_id":24072,"name":"Marcin Juszkiewicz","email":"mjuszkiewicz@redhat.com","username":"hrw"},{"_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2019-09-26 08:30:13.000000000","updated_by":{"_account_id":23181,"name":"RDO Third Party CI","email":"dmsimard+rdothirdparty@redhat.com","username":"rdothirdparty","tags":["SERVICE_USER"]},"reviewer":{"_account_id":23181,"name":"RDO Third Party CI","email":"dmsimard+rdothirdparty@redhat.com","username":"rdothirdparty","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2019-09-26 09:56:21.000000000","updated_by":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},"reviewer":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},"state":"REVIEWER"},{"updated":"2019-09-27 17:03:46.000000000","updated_by":{"_account_id":24072,"name":"Marcin Juszkiewicz","email":"mjuszkiewicz@redhat.com","username":"hrw"},"reviewer":{"_account_id":24072,"name":"Marcin Juszkiewicz","email":"mjuszkiewicz@redhat.com","username":"hrw"},"state":"REVIEWER"},{"updated":"2019-09-27 19:01:26.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"}],"messages":[{"id":"fc6162dea3d8bcfd73aaeda295e8059492e6989e","author":{"_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"},"date":"2019-09-25 07:56:31.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"f2518b80b62d6c0948e66e177be7a8c3c29e55cf","author":{"_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"},"date":"2019-09-25 07:56:40.000000000","message":"Patch Set 2: Patch Set 1 was rebased","accounts_in_message":[],"_revision_number":2},{"id":"94005ab61cd453cb0c83c70aa51d4a4140fbf205","author":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},"date":"2019-09-25 08:18:01.000000000","message":"Patch Set 2:\n\n(2 comments)","accounts_in_message":[],"_revision_number":2},{"id":"bd31665ddf67559b722ef74b204edb0975d7a3a7","author":{"_account_id":23181,"name":"RDO Third Party CI","email":"dmsimard+rdothirdparty@redhat.com","username":"rdothirdparty","tags":["SERVICE_USER"]},"date":"2019-09-25 08:38:26.000000000","message":"Patch Set 2:\n\nBuild succeeded (check pipeline).\n\n- tripleo-build-containers-rhel-8-master http://logs.rdoproject.org/51/684551/2/openstack-check/tripleo-build-containers-rhel-8-master/3fb781b/ : SUCCESS in 40m 39s","accounts_in_message":[],"_revision_number":2},{"id":"9677ed7e4c1222c9b895fd404843a9208618c3d9","author":{"_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"},"date":"2019-09-25 08:42:47.000000000","message":"Patch Set 2:\n\n(2 comments)","accounts_in_message":[],"_revision_number":2},{"id":"56971e9a04f882a33f16e4c5e5fafbc03ad60b2a","author":{"_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"},"date":"2019-09-25 08:43:34.000000000","message":"Patch Set 3: Commit message was updated.","accounts_in_message":[],"_revision_number":3},{"id":"be55fb52df94a6d48281cf478b8c91edeff4d157","author":{"_account_id":23181,"name":"RDO Third Party CI","email":"dmsimard+rdothirdparty@redhat.com","username":"rdothirdparty","tags":["SERVICE_USER"]},"date":"2019-09-25 09:30:28.000000000","message":"Patch Set 3:\n\nBuild succeeded (check pipeline).\n\n- tripleo-build-containers-rhel-8-master http://logs.rdoproject.org/51/684551/3/openstack-check/tripleo-build-containers-rhel-8-master/63c8c96/ : SUCCESS in 46m 03s","accounts_in_message":[],"_revision_number":3},{"id":"8029db70c67c9d8e51ac79163f4544c1416a8599","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-09-25 11:23:34.000000000","message":"Patch Set 3: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttp://docs.openstack.org/infra/manual/developers.html#automated-testing\n\n\n- kolla-build-centos-binary https://zuul.opendev.org/t/openstack/build/29a57783ae4a456aa1be7f8cec65e09f : SUCCESS in 49m 26s (non-voting)\n- kolla-build-centos-source https://zuul.opendev.org/t/openstack/build/2e4e918a04924c5d8e843e653ec345dd : SUCCESS in 54m 25s\n- kolla-ansible-centos-source https://zuul.opendev.org/t/openstack/build/9bf2d1902c754ee68dedde6a09a9e86d : SUCCESS in 1h 27m 08s\n- kolla-ansible-centos-binary https://zuul.opendev.org/t/openstack/build/6af13218346044ec8b2152ad582397f7 : SUCCESS in 1h 18m 43s (non-voting)\n- kolla-ansible-centos-source-upgrade https://zuul.opendev.org/t/openstack/build/5c02f12c62e047f1b5086a0ea6424a83 : SUCCESS in 1h 30m 36s\n- tripleo-build-containers-centos-7 https://zuul.opendev.org/t/openstack/build/f404c15712ee487592a2c284fe960eff : SUCCESS in 1h 00m 12s\n- kolla-build-debian-source https://zuul.opendev.org/t/openstack/build/ebbbcfa75ff348aa94fb0b4be1e56d80 : SUCCESS in 26m 58s (non-voting)\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/799089545539441a950da082a9f01c9a : SUCCESS in 3m 39s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/07a79f23c42a4a06a3e5c955af13326f : SUCCESS in 4m 28s\n- openstack-tox-py27 https://zuul.opendev.org/t/openstack/build/5a966e37bd1d4d16a966e9a43af4a5e1 : SUCCESS in 4m 30s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/a6951f2f258a4915afc3580a3152e49e : SUCCESS in 3m 29s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/e2d7ca56dbe64d02a80b02c247cb4eb2 : SUCCESS in 4m 45s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/33d9e1e06dc64b6cae3420667f85aa18 : SUCCESS in 5m 47s\n- kolla-build-ubuntu-binary https://zuul.opendev.org/t/openstack/build/8cbaa262c4dc4546b22ce6b62be50995 : SUCCESS in 30m 33s (non-voting)\n- kolla-build-ubuntu-source https://zuul.opendev.org/t/openstack/build/45b8e298e58542f788c31f3720f5aeaf : FAILURE in 28m 35s\n- kolla-ansible-ubuntu-source https://zuul.opendev.org/t/openstack/build/fb0cec64dc8542e6a937aff0dd34a98e : SUCCESS in 58m 48s\n- kolla-ansible-ubuntu-binary https://zuul.opendev.org/t/openstack/build/7dc493a260f74c089bc360e3ed754eef : SUCCESS in 56m 25s (non-voting)\n- kolla-ansible-ubuntu-source-upgrade https://zuul.opendev.org/t/openstack/build/9373c61f329642da8c963ad4e9758d04 : SUCCESS in 1h 04m 20s","accounts_in_message":[],"_revision_number":3},{"id":"e62288707b373759e6ab621c0a30c8cc901ddb25","author":{"_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"},"date":"2019-09-25 11:24:55.000000000","message":"Patch Set 3:\n\nrecheck","accounts_in_message":[],"_revision_number":3},{"id":"df7e99d5fcf1610012e5682d5bdf131a02a36c9d","author":{"_account_id":23181,"name":"RDO Third Party CI","email":"dmsimard+rdothirdparty@redhat.com","username":"rdothirdparty","tags":["SERVICE_USER"]},"date":"2019-09-25 12:02:21.000000000","message":"Patch Set 3:\n\nBuild succeeded (check pipeline).\n\n- tripleo-build-containers-rhel-8-master http://logs.rdoproject.org/51/684551/3/openstack-check/tripleo-build-containers-rhel-8-master/2f3578b/ : SUCCESS in 36m 42s","accounts_in_message":[],"_revision_number":3},{"id":"4e11867f373032b10246d4de80d7a7d5c8a376a0","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-09-25 12:59:43.000000000","message":"Patch Set 3:\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttp://docs.openstack.org/infra/manual/developers.html#automated-testing\n\n\n- kolla-build-centos-binary https://zuul.opendev.org/t/openstack/build/fc50d3be9e18413989363944c0569eee : SUCCESS in 53m 29s (non-voting)\n- kolla-build-centos-source https://zuul.opendev.org/t/openstack/build/a6255f06ba924332aa763657914b3adb : SUCCESS in 1h 16m 02s\n- kolla-ansible-centos-source https://zuul.opendev.org/t/openstack/build/cff3f0c7b12d42dd833a9433d6ae2578 : SUCCESS in 1h 24m 16s\n- kolla-ansible-centos-binary https://zuul.opendev.org/t/openstack/build/145d926f9ce64518b45c3f7556c49caa : SUCCESS in 1h 26m 34s (non-voting)\n- kolla-ansible-centos-source-upgrade https://zuul.opendev.org/t/openstack/build/181d44b8991940da9af9a8bbf862da47 : SUCCESS in 1h 28m 30s\n- tripleo-build-containers-centos-7 https://zuul.opendev.org/t/openstack/build/9c87a560e5974762aa4c9b20dee192d6 : SUCCESS in 1h 02m 08s\n- kolla-build-debian-source https://zuul.opendev.org/t/openstack/build/26757bb54aca48bf8b8f0170e8ccc451 : SUCCESS in 29m 08s (non-voting)\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/3c2b41e4862b4b92b71ee62c45cda41a : SUCCESS in 3m 24s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/ee679b081bc84bdc921895701ea845b0 : SUCCESS in 3m 51s\n- openstack-tox-py27 https://zuul.opendev.org/t/openstack/build/fe0016fdda594beba6c1386004dd80d9 : SUCCESS in 3m 21s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/aa575a15df654774b31ddda1885146f1 : SUCCESS in 3m 20s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/fc8ebf4a12d049ef88f40ee870132786 : SUCCESS in 4m 49s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/db65cda34edb4825aa2aa55d434fd95c : SUCCESS in 4m 20s\n- kolla-build-ubuntu-binary https://zuul.opendev.org/t/openstack/build/bff7e7f36b2041e2b74182b19e1b39c7 : SUCCESS in 33m 20s (non-voting)\n- kolla-build-ubuntu-source https://zuul.opendev.org/t/openstack/build/838c0808b53f4c4dad6152531cbe243c : FAILURE in 50m 14s\n- kolla-ansible-ubuntu-source https://zuul.opendev.org/t/openstack/build/b1ad95f5840d4cd4b55c70777879ef80 : SUCCESS in 59m 57s\n- kolla-ansible-ubuntu-binary https://zuul.opendev.org/t/openstack/build/f77cd666e1e94cbcbb1d3401b12a6600 : SUCCESS in 1h 06m 53s (non-voting)\n- kolla-ansible-ubuntu-source-upgrade https://zuul.opendev.org/t/openstack/build/891114a7369340f1b30dd8c41cdb878b : SUCCESS in 1h 03m 59s","accounts_in_message":[],"_revision_number":3},{"id":"793a201db0c39b80f1231b13fd3c4b1d06e6247c","author":{"_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"},"date":"2019-09-26 07:49:50.000000000","message":"Patch Set 3:\n\nrecheck","accounts_in_message":[],"_revision_number":3},{"id":"cd461e0b15e5b2bf5c37217d38099202d78a6ab3","author":{"_account_id":23181,"name":"RDO Third Party CI","email":"dmsimard+rdothirdparty@redhat.com","username":"rdothirdparty","tags":["SERVICE_USER"]},"date":"2019-09-26 08:30:13.000000000","message":"Patch Set 3:\n\nBuild succeeded (check pipeline).\n\n- tripleo-build-containers-rhel-8-master http://logs.rdoproject.org/51/684551/3/openstack-check/tripleo-build-containers-rhel-8-master/6e7f913/ : SUCCESS in 39m 24s","accounts_in_message":[],"_revision_number":3},{"id":"9159602510f711e8364b568e1b3af960daecbb23","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-09-26 09:44:44.000000000","message":"Patch Set 3: Verified+1\n\nBuild succeeded (check pipeline).\n\n- kolla-build-centos-binary https://zuul.opendev.org/t/openstack/build/2caa51ac20554412b37ddb6e8fdd4967 : SUCCESS in 46m 01s (non-voting)\n- kolla-build-centos-source https://zuul.opendev.org/t/openstack/build/fe74029f2c664c119602f31823869f39 : SUCCESS in 52m 28s\n- kolla-ansible-centos-source https://zuul.opendev.org/t/openstack/build/3bdf80304bfe45eca53db09d42dd844f : SUCCESS in 1h 32m 12s\n- kolla-ansible-centos-binary https://zuul.opendev.org/t/openstack/build/49e7eb7cf2b741c3a3681bcaf6e3acc8 : SUCCESS in 1h 26m 53s (non-voting)\n- kolla-ansible-centos-source-upgrade https://zuul.opendev.org/t/openstack/build/ce2cbc4f0e7b4a8fa78617d537b80c81 : SUCCESS in 1h 51m 10s\n- tripleo-build-containers-centos-7 https://zuul.opendev.org/t/openstack/build/f57ae9ab554144a7b8675a7daa300915 : SUCCESS in 1h 00m 27s\n- kolla-build-debian-source https://zuul.opendev.org/t/openstack/build/cd6d52bd0bf4460b96f212e942289319 : SUCCESS in 29m 18s (non-voting)\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/f0c342f948ec4bda945ae661433af725 : SUCCESS in 4m 32s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/7eac45f8edfc46d89b14732e04eeead3 : SUCCESS in 4m 17s\n- openstack-tox-py27 https://zuul.opendev.org/t/openstack/build/a251571dc6f64624aef19e207eac2d0b : SUCCESS in 3m 52s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/f2f3b08f50344c2aa0775234b9d51e61 : SUCCESS in 4m 05s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/b49fd7da875e4095b3b69df233ef2cd4 : SUCCESS in 4m 47s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/97362170cc1741a7b0bf16931c7c9e14 : SUCCESS in 5m 55s\n- kolla-build-ubuntu-binary https://zuul.opendev.org/t/openstack/build/8966471ffbe24ac39401c8302c4a0289 : SUCCESS in 23m 11s (non-voting)\n- kolla-build-ubuntu-source https://zuul.opendev.org/t/openstack/build/80b32d5ff8764f74b6615bf120ad5dcf : SUCCESS in 30m 41s\n- kolla-ansible-ubuntu-source https://zuul.opendev.org/t/openstack/build/1db2d14bce4143d699088518b8c1a9fd : SUCCESS in 50m 59s\n- kolla-ansible-ubuntu-binary https://zuul.opendev.org/t/openstack/build/fb3254abf4554179ba6eb9e3d33ab049 : SUCCESS in 55m 57s (non-voting)\n- kolla-ansible-ubuntu-source-upgrade https://zuul.opendev.org/t/openstack/build/54309e30c6c54d0cb4426034b9e66b86 : SUCCESS in 58m 51s","accounts_in_message":[],"_revision_number":3},{"id":"65665939798d6433bdc4a7dc9409123c9eb5ede9","author":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},"date":"2019-09-26 09:56:21.000000000","message":"Patch Set 3: Code-Review+2","accounts_in_message":[],"_revision_number":3},{"id":"b5b0dee5ffd622c10a86deb1d544b36c1e75f714","author":{"_account_id":24072,"name":"Marcin Juszkiewicz","email":"mjuszkiewicz@redhat.com","username":"hrw"},"date":"2019-09-27 17:03:46.000000000","message":"Patch Set 3: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":3},{"id":"aa42e682c332a7826d78e5f6339bd7ff4f9487aa","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-09-27 17:06:24.000000000","message":"Patch Set 3: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":3},{"id":"9fc34d7cff5c46be214a0c555a3035f93a7a9aa0","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-09-27 19:01:26.000000000","message":"Patch Set 3: Verified+2\n\nBuild succeeded (gate pipeline).\n\n- kolla-build-centos-source https://zuul.opendev.org/t/openstack/build/37a16717db744fdca389134bf482237a : SUCCESS in 1h 04m 22s\n- kolla-ansible-centos-source https://zuul.opendev.org/t/openstack/build/01334addcfc64999bc60b21c71e84c32 : SUCCESS in 1h 30m 31s\n- kolla-ansible-centos-source-upgrade https://zuul.opendev.org/t/openstack/build/135a1d401af24b088220b3e01624cefd : SUCCESS in 1h 25m 49s\n- tripleo-build-containers-centos-7 https://zuul.opendev.org/t/openstack/build/1f0e6c9399c14d8a800bc156e75f82e0 : SUCCESS in 56m 11s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/61bd1ef9a3c14b2d97edee179912ed07 : SUCCESS in 4m 16s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/607a138b5a1249c5bb7ce48afa2d2ca9 : SUCCESS in 5m 06s\n- openstack-tox-py27 https://zuul.opendev.org/t/openstack/build/ad22de259d194603b78db4eae3fa2d39 : SUCCESS in 3m 20s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/692a9e5b26ea4fc9ad62b008509a4ca4 : SUCCESS in 3m 38s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/9e074c0015a141df8c5bb8414c2836e0 : SUCCESS in 3m 51s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/22705c563b9e4e69929a12ee072e1bc8 : SUCCESS in 8m 45s\n- kolla-build-ubuntu-source https://zuul.opendev.org/t/openstack/build/12f7d349bd48493db53fdcaec807d486 : SUCCESS in 35m 26s\n- kolla-ansible-ubuntu-source https://zuul.opendev.org/t/openstack/build/96400513046243bb83b57f458c03944c : SUCCESS in 1h 07m 10s\n- kolla-ansible-ubuntu-source-upgrade https://zuul.opendev.org/t/openstack/build/adb09626ba5b48488399a91c6c084182 : SUCCESS in 1h 18m 37s","accounts_in_message":[],"_revision_number":3},{"id":"c6d53c1d43ea53cf4fcacfeffc8c3ed80b79c271","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-09-27 19:01:27.000000000","message":"Change has been successfully merged by Zuul","accounts_in_message":[],"_revision_number":3},{"id":"050d8271ba52a2313d1b7cf883cb4ff3866207d6","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-09-27 19:02:48.000000000","message":"Patch Set 3:\n\nBuild succeeded (promote pipeline).\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/5a12c1072f4b440eb122bd08bb4c549e : SUCCESS in 53s","accounts_in_message":[],"_revision_number":3}],"current_revision_number":3,"current_revision":"76fbf2c08b17f73dd304c1316fbabbbbe9d9468a","revisions":{"779941c6a30a651f18aba4652a54870bbcd643ef":{"kind":"REWORK","_number":1,"created":"2019-09-25 07:56:31.000000000","uploader":{"_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"},"ref":"refs/changes/51/684551/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/kolla","ref":"refs/changes/51/684551/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/kolla refs/changes/51/684551/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/kolla refs/changes/51/684551/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/kolla refs/changes/51/684551/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/kolla refs/changes/51/684551/1"}}},"commit":{"parents":[{"commit":"c08a4b5a6352f3d1ebf640687d68d00aabc11ff6","subject":"Add Debian Buster Stein backports APT repository","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/c08a4b5a6352f3d1ebf640687d68d00aabc11ff6"}]}],"author":{"name":"Radosław Piliszek","email":"radoslaw.piliszek@gmail.com","date":"2019-09-25 07:52:38.000000000","tz":120},"committer":{"name":"Radosław Piliszek","email":"radoslaw.piliszek@gmail.com","date":"2019-09-25 07:52:38.000000000","tz":120},"subject":"Debian: switch fluentd to td-agent","message":"Debian: switch fluentd to td-agent\n\nThis patch uses the newly added td-agent repo on Debian.\n\nIt also makes code more robust regarding different distro handling.\n\nChange-Id: I4d2449d7d100c68084f937ef9981596fc1625375\nSigned-off-by: Radosław Piliszek \u003cradoslaw.piliszek@gmail.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/779941c6a30a651f18aba4652a54870bbcd643ef"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/779941c6a30a651f18aba4652a54870bbcd643ef"}]},"branch":"refs/heads/master"},"eb34d329e12f5a174dc9cecbeba18e0f51cc022a":{"kind":"TRIVIAL_REBASE","_number":2,"created":"2019-09-25 07:56:40.000000000","uploader":{"_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"},"ref":"refs/changes/51/684551/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/kolla","ref":"refs/changes/51/684551/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/kolla refs/changes/51/684551/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/kolla refs/changes/51/684551/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/kolla refs/changes/51/684551/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/kolla refs/changes/51/684551/2"}}},"commit":{"parents":[{"commit":"2131ef9875efd67a9dd45dc3329df8de16cacff1","subject":"Fluentd: td-agent repository for Debian","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/2131ef9875efd67a9dd45dc3329df8de16cacff1"}]}],"author":{"name":"Radosław Piliszek","email":"radoslaw.piliszek@gmail.com","date":"2019-09-25 07:52:38.000000000","tz":120},"committer":{"name":"Radosław Piliszek","email":"radoslaw.piliszek@gmail.com","date":"2019-09-25 07:56:40.000000000","tz":0},"subject":"Debian: switch fluentd to td-agent","message":"Debian: switch fluentd to td-agent\n\nThis patch uses the newly added td-agent repo on Debian.\n\nIt also makes code more robust regarding different distro handling.\n\nChange-Id: I4d2449d7d100c68084f937ef9981596fc1625375\nSigned-off-by: Radosław Piliszek \u003cradoslaw.piliszek@gmail.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/eb34d329e12f5a174dc9cecbeba18e0f51cc022a"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/eb34d329e12f5a174dc9cecbeba18e0f51cc022a"}]},"branch":"refs/heads/master"},"76fbf2c08b17f73dd304c1316fbabbbbe9d9468a":{"kind":"NO_CODE_CHANGE","_number":3,"created":"2019-09-25 08:43:34.000000000","uploader":{"_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"},"ref":"refs/changes/51/684551/3","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/kolla","ref":"refs/changes/51/684551/3","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/kolla refs/changes/51/684551/3 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/kolla refs/changes/51/684551/3 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/kolla refs/changes/51/684551/3 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/kolla refs/changes/51/684551/3"}}},"commit":{"parents":[{"commit":"2131ef9875efd67a9dd45dc3329df8de16cacff1","subject":"Fluentd: td-agent repository for Debian","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/2131ef9875efd67a9dd45dc3329df8de16cacff1"}]}],"author":{"name":"Radosław Piliszek","email":"radoslaw.piliszek@gmail.com","date":"2019-09-25 07:52:38.000000000","tz":120},"committer":{"name":"Radosław Piliszek","email":"radoslaw.piliszek@gmail.com","date":"2019-09-25 08:43:30.000000000","tz":0},"subject":"Debian: switch fluentd to td-agent on x86_64","message":"Debian: switch fluentd to td-agent on x86_64\n\nThis patch uses the newly added td-agent repo on Debian.\n\nIt also makes code more robust regarding different distro handling.\n\nWe now have td-agent everywhere on x86_64.\n\nChange-Id: I4d2449d7d100c68084f937ef9981596fc1625375\nSigned-off-by: Radosław Piliszek \u003cradoslaw.piliszek@gmail.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/76fbf2c08b17f73dd304c1316fbabbbbe9d9468a"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/76fbf2c08b17f73dd304c1316fbabbbbe9d9468a"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[],"submit_requirements":[]}
