)]}'
{"id":"openstack%2Fkolla~684742","triplet_id":"openstack%2Fkolla~master~I2c10c7b4f7dbdd87f2dda7c40fa44cded07c57a5","project":"openstack/kolla","branch":"master","topic":"bp/support-nova-cells","hashtags":[],"change_id":"I2c10c7b4f7dbdd87f2dda7c40fa44cded07c57a5","subject":"Install RabbitMQ in kolla-toolbox image","status":"MERGED","created":"2019-09-25 12:52:25.000000000","updated":"2019-09-27 19:04:00.000000000","submitted":"2019-09-27 19:02:49.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"684742-1569610970141-e03b07bf","meta_rev_id":"aa635c2723a55e55f0b0de5db8784a1f0d055ed3","_number":684742,"virtual_id_number":684742,"owner":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":24072,"name":"Marcin Juszkiewicz","email":"mjuszkiewicz@redhat.com","username":"hrw"},{"value":0,"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},{"value":0,"_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2019-09-27 19:02:49.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"date":"2019-09-27 08:33:26.000000000","_account_id":23181,"name":"RDO Third Party CI","email":"dmsimard+rdothirdparty@redhat.com","username":"rdothirdparty","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":24072,"name":"Marcin Juszkiewicz","email":"mjuszkiewicz@redhat.com","username":"hrw"},"all":[{"value":2,"date":"2019-09-27 17:15:32.000000000","_account_id":24072,"name":"Marcin Juszkiewicz","email":"mjuszkiewicz@redhat.com","username":"hrw"},{"value":0,"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},{"value":2,"date":"2019-09-26 09:39:47.000000000","_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":23181,"name":"RDO Third Party CI","email":"dmsimard+rdothirdparty@redhat.com","username":"rdothirdparty","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":24072,"name":"Marcin Juszkiewicz","email":"mjuszkiewicz@redhat.com","username":"hrw"},"all":[{"value":1,"date":"2019-09-27 17:15:32.000000000","_account_id":24072,"name":"Marcin Juszkiewicz","email":"mjuszkiewicz@redhat.com","username":"hrw"},{"value":0,"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},{"value":0,"_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":23181,"name":"RDO Third Party CI","email":"dmsimard+rdothirdparty@redhat.com","username":"rdothirdparty","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Backport-Candidate":{"all":[{"value":0,"_account_id":24072,"name":"Marcin Juszkiewicz","email":"mjuszkiewicz@redhat.com","username":"hrw"},{"value":0,"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},{"value":0,"_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":23181,"name":"RDO Third Party CI","email":"dmsimard+rdothirdparty@redhat.com","username":"rdothirdparty","tags":["SERVICE_USER"]}],"values":{"-1":"Do Not Backport"," 0":"Backport Review Needed","+1":"Should Backport"},"description":"","default_value":0,"optional":true},"Review-Priority":{"recommended":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},"all":[{"value":0,"_account_id":24072,"name":"Marcin Juszkiewicz","email":"mjuszkiewicz@redhat.com","username":"hrw"},{"value":1,"date":"2019-09-26 11:18:50.000000000","_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},{"value":0,"_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":23181,"name":"RDO Third Party CI","email":"dmsimard+rdothirdparty@redhat.com","username":"rdothirdparty","tags":["SERVICE_USER"]}],"values":{"-1":"Branch Freeze"," 0":"No Priority","+1":"Important Change","+2":"Gate Blocker Fix / Urgent Change"},"description":"","value":1,"default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":23181,"name":"RDO Third Party CI","email":"dmsimard+rdothirdparty@redhat.com","username":"rdothirdparty","tags":["SERVICE_USER"]},{"_account_id":24072,"name":"Marcin Juszkiewicz","email":"mjuszkiewicz@redhat.com","username":"hrw"},{"_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2019-09-26 09:39:47.000000000","updated_by":{"_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"},"reviewer":{"_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"},"state":"REVIEWER"},{"updated":"2019-09-27 08:33:26.000000000","updated_by":{"_account_id":23181,"name":"RDO Third Party CI","email":"dmsimard+rdothirdparty@redhat.com","username":"rdothirdparty","tags":["SERVICE_USER"]},"reviewer":{"_account_id":23181,"name":"RDO Third Party CI","email":"dmsimard+rdothirdparty@redhat.com","username":"rdothirdparty","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2019-09-27 17:15:32.000000000","updated_by":{"_account_id":24072,"name":"Marcin Juszkiewicz","email":"mjuszkiewicz@redhat.com","username":"hrw"},"reviewer":{"_account_id":24072,"name":"Marcin Juszkiewicz","email":"mjuszkiewicz@redhat.com","username":"hrw"},"state":"REVIEWER"},{"updated":"2019-09-27 19:02:49.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"}],"messages":[{"id":"87e9e05fabd1beb2a33bb1a5b2243e0c4f88d3fb","author":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},"date":"2019-09-25 12:52:25.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"ab8021d16e7151c818b778f449aeb1d2655ff3d0","author":{"_account_id":23181,"name":"RDO Third Party CI","email":"dmsimard+rdothirdparty@redhat.com","username":"rdothirdparty","tags":["SERVICE_USER"]},"date":"2019-09-25 13:35:21.000000000","message":"Patch Set 1:\n\nBuild succeeded (check pipeline).\n\n- tripleo-build-containers-rhel-8-master http://logs.rdoproject.org/42/684742/1/openstack-check/tripleo-build-containers-rhel-8-master/c473f92/ : SUCCESS in 42m 11s","accounts_in_message":[],"_revision_number":1},{"id":"863483698e3ecbcb0978b55ea4ab19046bf53527","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-09-25 14:28:34.000000000","message":"Patch Set 1: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttp://docs.openstack.org/infra/manual/developers.html#automated-testing\n\n\n- kolla-build-centos-binary https://zuul.opendev.org/t/openstack/build/aff9242d505543afbc70812dd4709689 : SUCCESS in 49m 47s (non-voting)\n- kolla-build-centos-source https://zuul.opendev.org/t/openstack/build/49b19e80508848f88c72709f34665d9f : SUCCESS in 1h 02m 25s\n- kolla-ansible-centos-source https://zuul.opendev.org/t/openstack/build/8c07f4acfd27472cacf4c8534022e62b : SUCCESS in 1h 28m 45s\n- kolla-ansible-centos-binary https://zuul.opendev.org/t/openstack/build/925581dd7a414bcdac789046d1b3864f : SUCCESS in 1h 17m 03s (non-voting)\n- kolla-ansible-centos-source-upgrade https://zuul.opendev.org/t/openstack/build/f7529c2ca6bb4317b1d74d1f4af8c772 : SUCCESS in 1h 18m 43s\n- tripleo-build-containers-centos-7 https://zuul.opendev.org/t/openstack/build/6f2d10d578014261a387e8fd08cac453 : SUCCESS in 45m 25s\n- kolla-build-debian-source https://zuul.opendev.org/t/openstack/build/822c62fe14c445a1b103bff591140447 : SUCCESS in 27m 33s (non-voting)\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/68c0361b0da64213907f04dd9215e225 : SUCCESS in 4m 22s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/44dc688c2f1b48a1930929e1e54d0778 : SUCCESS in 3m 59s\n- openstack-tox-py27 https://zuul.opendev.org/t/openstack/build/a1779fe5d3eb466d89aabf68ffeb8713 : SUCCESS in 3m 54s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/4329e66c0ba246a6962a65aa18eadcd6 : SUCCESS in 4m 24s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/347e127ce11e41cab9e33dde844fe6b3 : SUCCESS in 4m 03s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/8c14364227b044428545c191d9200135 : SUCCESS in 5m 13s\n- kolla-build-ubuntu-binary https://zuul.opendev.org/t/openstack/build/fe706b778cc0401a9c88bf8011ad8959 : SUCCESS in 30m 26s (non-voting)\n- kolla-build-ubuntu-source https://zuul.opendev.org/t/openstack/build/f1455b993df04a86bdaddb84d12120fe : FAILURE in 32m 47s\n- kolla-ansible-ubuntu-source https://zuul.opendev.org/t/openstack/build/83e463c0e09b40988d64c5b96f3d8e7d : SUCCESS in 59m 46s\n- kolla-ansible-ubuntu-binary https://zuul.opendev.org/t/openstack/build/6483ecbfbeb446d9b4e93231da9ce49e : SUCCESS in 52m 20s (non-voting)\n- kolla-ansible-ubuntu-source-upgrade https://zuul.opendev.org/t/openstack/build/02fb42ec5b0249738e05737cb74dec55 : SUCCESS in 1h 02m 07s","accounts_in_message":[],"_revision_number":1},{"id":"a3f8ee62021acc7011909b5352d20485e42678ab","author":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},"date":"2019-09-26 09:33:38.000000000","message":"Patch Set 1:\n\nrecheck","accounts_in_message":[],"_revision_number":1},{"id":"c7a0f05a766eec907ca50607a34a383322770b82","author":{"_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"},"date":"2019-09-26 09:39:47.000000000","message":"Patch Set 1: Code-Review+2\n\ncopied semi-verbatim from rabbitmq image","accounts_in_message":[],"_revision_number":1},{"id":"bfd303fc56f04b90ed11d24bd000f7d0e33d0c47","author":{"_account_id":23181,"name":"RDO Third Party CI","email":"dmsimard+rdothirdparty@redhat.com","username":"rdothirdparty","tags":["SERVICE_USER"]},"date":"2019-09-26 11:01:58.000000000","message":"Patch Set 1:\n\nBuild succeeded (check pipeline).\n\n- tripleo-build-containers-rhel-8-master http://logs.rdoproject.org/42/684742/1/openstack-check/tripleo-build-containers-rhel-8-master/d0d2f06/ : SUCCESS in 39m 39s","accounts_in_message":[],"_revision_number":1},{"id":"a461bf22ca2cd76c5181122d79218aefbe5c5b03","author":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},"date":"2019-09-26 11:18:50.000000000","message":"Patch Set 1: Review-Priority+1","accounts_in_message":[],"_revision_number":1},{"id":"eeff43bd20843679923bcde97fcb7fc37f91ba45","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-09-26 11:56:26.000000000","message":"Patch Set 1:\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttp://docs.openstack.org/infra/manual/developers.html#automated-testing\n\n\n- kolla-build-centos-binary https://zuul.opendev.org/t/openstack/build/4dafecde3a2c4aeabee82198c27697a7 : SUCCESS in 40m 34s (non-voting)\n- kolla-build-centos-source https://zuul.opendev.org/t/openstack/build/3662b82d5fb04128a757028757aea80a : SUCCESS in 1h 16m 19s\n- kolla-ansible-centos-source https://zuul.opendev.org/t/openstack/build/20a858cebe744f2f9f187f700180c579 : SUCCESS in 1h 34m 30s\n- kolla-ansible-centos-binary https://zuul.opendev.org/t/openstack/build/add7a001503246049cbee4df05f83ccb : SUCCESS in 1h 21m 24s (non-voting)\n- kolla-ansible-centos-source-upgrade https://zuul.opendev.org/t/openstack/build/79f06b9a2979477a901e18c9eef5ef63 : TIMED_OUT in 2h 02m 45s\n- tripleo-build-containers-centos-7 https://zuul.opendev.org/t/openstack/build/88bdbb3a4f1e4662b3e45a3271a6f3d2 : SUCCESS in 1h 00m 20s\n- kolla-build-debian-source https://zuul.opendev.org/t/openstack/build/892b164342a8418890b98d1182738fdd : SUCCESS in 25m 37s (non-voting)\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/30905ea505724f668708b35c4b123290 : SUCCESS in 4m 35s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/86694acc642b40a8a6313f2b0fa3d16b : SUCCESS in 4m 56s\n- openstack-tox-py27 https://zuul.opendev.org/t/openstack/build/fe00d543024f4295b62a753c0b0e08d8 : SUCCESS in 3m 56s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/3969f8ebeb194ca68a697cb3b913f0ad : SUCCESS in 5m 05s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/fd02c770280845728f47bbd56a54ea65 : SUCCESS in 5m 18s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/a1b59251e3b0464dbad2e7f5bbcb7b8b : SUCCESS in 4m 52s\n- kolla-build-ubuntu-binary https://zuul.opendev.org/t/openstack/build/60c9f6d7db7f4d52a5e46e9d9d70fd0d : SUCCESS in 1h 34m 37s (non-voting)\n- kolla-build-ubuntu-source https://zuul.opendev.org/t/openstack/build/5eab580666c744d19172044a11781a63 : SUCCESS in 29m 02s\n- kolla-ansible-ubuntu-source https://zuul.opendev.org/t/openstack/build/337460571bca423b87c937900e456f10 : SUCCESS in 1h 04m 49s\n- kolla-ansible-ubuntu-binary https://zuul.opendev.org/t/openstack/build/162ec5ef3f1f49c8bc943e529915f058 : SUCCESS in 54m 55s (non-voting)\n- kolla-ansible-ubuntu-source-upgrade https://zuul.opendev.org/t/openstack/build/f0f60b71b6554babb072fe84e868cf43 : SUCCESS in 1h 14m 08s","accounts_in_message":[],"_revision_number":1},{"id":"a0d82598f2c0594013d78394087ee733b2c292fd","author":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},"date":"2019-09-27 07:47:43.000000000","message":"Patch Set 1:\n\nrecheck","accounts_in_message":[],"_revision_number":1},{"id":"c9d68ba92bf260e17d552f6a256a358af5e93094","author":{"_account_id":23181,"name":"RDO Third Party CI","email":"dmsimard+rdothirdparty@redhat.com","username":"rdothirdparty","tags":["SERVICE_USER"]},"date":"2019-09-27 08:33:26.000000000","message":"Patch Set 1:\n\nBuild succeeded (check pipeline).\n\n- tripleo-build-containers-rhel-8-master http://logs.rdoproject.org/42/684742/1/openstack-check/tripleo-build-containers-rhel-8-master/9297eaa/ : SUCCESS in 44m 53s","accounts_in_message":[],"_revision_number":1},{"id":"cead77aefdab43a977fd95e8a89aec8e17ed8c58","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-09-27 09:28:45.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\n\n- kolla-build-centos-binary https://zuul.opendev.org/t/openstack/build/6f87febca6de4271a800d4b7785f4bf7 : SUCCESS in 42m 32s (non-voting)\n- kolla-build-centos-source https://zuul.opendev.org/t/openstack/build/d2fd6d3aebbe4a8ea499349963988775 : SUCCESS in 59m 52s\n- kolla-ansible-centos-source https://zuul.opendev.org/t/openstack/build/b9a65544d03348409b65b462502cb3c5 : SUCCESS in 1h 28m 10s\n- kolla-ansible-centos-binary https://zuul.opendev.org/t/openstack/build/3d9daf770d174a9eb635c60f09fa19c2 : SUCCESS in 1h 21m 38s (non-voting)\n- kolla-ansible-centos-source-upgrade https://zuul.opendev.org/t/openstack/build/b4692d3304ae4b9c9088ee117e0fae58 : SUCCESS in 1h 29m 59s\n- tripleo-build-containers-centos-7 https://zuul.opendev.org/t/openstack/build/27bec131e2f748159f29317a559cceb2 : SUCCESS in 50m 56s\n- kolla-build-debian-source https://zuul.opendev.org/t/openstack/build/1271e879c3c54e8dbe7878df736c8b39 : SUCCESS in 29m 09s (non-voting)\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/b84e0a7cd9fd45e8a9cb4a0251a07135 : SUCCESS in 3m 16s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/7a5b0b1302fa4d01b9eb5b98808229bb : SUCCESS in 4m 46s\n- openstack-tox-py27 https://zuul.opendev.org/t/openstack/build/129c39c886dc4aa7aec19d5a1f24c08c : SUCCESS in 4m 29s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/9eaf29ae3c8a45d6825aed9ab5d62dda : SUCCESS in 2m 46s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/4d4355f591c3463d893927d6f47d38cc : SUCCESS in 3m 24s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/6beed8cf1f84403d9603ab15335be14d : SUCCESS in 5m 35s\n- kolla-build-ubuntu-binary https://zuul.opendev.org/t/openstack/build/7abcebad90234fe3a6ffabdadb2e2c54 : SUCCESS in 28m 56s (non-voting)\n- kolla-build-ubuntu-source https://zuul.opendev.org/t/openstack/build/21ff8be91ef24409bef4eb5e8ea50503 : SUCCESS in 35m 27s\n- kolla-ansible-ubuntu-source https://zuul.opendev.org/t/openstack/build/95eb190a260448b2bf3e9682ec5c3fe5 : SUCCESS in 1h 03m 09s\n- kolla-ansible-ubuntu-binary https://zuul.opendev.org/t/openstack/build/3ddb6e2c80f5485b9664906504f09768 : SUCCESS in 1h 00m 48s (non-voting)\n- kolla-ansible-ubuntu-source-upgrade https://zuul.opendev.org/t/openstack/build/3a70fef8d38b403eb07ddeedbb47dfcb : SUCCESS in 55m 41s","accounts_in_message":[],"_revision_number":1},{"id":"11838afd6560a0ee17d7f17dab501bc00aef3ee7","author":{"_account_id":24072,"name":"Marcin Juszkiewicz","email":"mjuszkiewicz@redhat.com","username":"hrw"},"date":"2019-09-27 17:15:32.000000000","message":"Patch Set 1: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":1},{"id":"502a6a7034f15fdb560599cba698474acb367a2e","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-09-27 17:15:45.000000000","message":"Patch Set 1: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":1},{"id":"b9b3fab4edcb44e22a3a093a944742ad5ce50001","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-09-27 19:02:49.000000000","message":"Patch Set 1: Verified+2\n\nBuild succeeded (gate pipeline).\n\n- kolla-build-centos-source https://zuul.opendev.org/t/openstack/build/43b521811c144c419c6f9a66e0c92bdd : SUCCESS in 1h 16m 30s\n- kolla-ansible-centos-source https://zuul.opendev.org/t/openstack/build/9807f63ec64049c9822dee357272a1e3 : SUCCESS in 1h 44m 30s\n- kolla-ansible-centos-source-upgrade https://zuul.opendev.org/t/openstack/build/1addda5c6c874cbabcf3a7426194d558 : SUCCESS in 1h 35m 02s\n- tripleo-build-containers-centos-7 https://zuul.opendev.org/t/openstack/build/ecaebcddb4b443288c310ea5190be549 : SUCCESS in 58m 56s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/3248d4f6be4846d4956127a123601cc0 : SUCCESS in 5m 03s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/fd9ac96145904ea98319206230893623 : SUCCESS in 5m 38s\n- openstack-tox-py27 https://zuul.opendev.org/t/openstack/build/50179aac2a4e4ade9125c3f35fe1a858 : SUCCESS in 3m 42s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/5dbf2e5664114322b75b88e977f1de9e : SUCCESS in 3m 52s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/866b7656f0154856a566f2c4bf24249c : SUCCESS in 3m 27s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/7493a39f1e5c4d389cabd71815e80f80 : SUCCESS in 4m 56s\n- kolla-build-ubuntu-source https://zuul.opendev.org/t/openstack/build/a131b69e83f14a5faf70346e2915e3b0 : SUCCESS in 30m 51s\n- kolla-ansible-ubuntu-source https://zuul.opendev.org/t/openstack/build/cd342e212d344744b6a303fd41098c82 : SUCCESS in 1h 03m 46s\n- kolla-ansible-ubuntu-source-upgrade https://zuul.opendev.org/t/openstack/build/c89e3d5693f94670ac56091d1522812f : SUCCESS in 1h 22m 07s","accounts_in_message":[],"_revision_number":1},{"id":"dcbab2ad2c297ccff2c972b25b9c99fa76b42612","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-09-27 19:02:50.000000000","message":"Change has been successfully merged by Zuul","accounts_in_message":[],"_revision_number":1},{"id":"e13eecc2233f2743aa50b5a5eacfbaa34b00aca8","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-09-27 19:04:00.000000000","message":"Patch Set 1:\n\nBuild succeeded (promote pipeline).\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/d0a02f9bed534b17ba031b5eb0a230c0 : SUCCESS in 47s","accounts_in_message":[],"_revision_number":1}],"current_revision_number":1,"current_revision":"1e2a6dba0348a9692aeb668c8b10b7de3ebf66f9","revisions":{"1e2a6dba0348a9692aeb668c8b10b7de3ebf66f9":{"kind":"REWORK","_number":1,"created":"2019-09-25 12:52:25.000000000","uploader":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},"ref":"refs/changes/42/684742/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/kolla","ref":"refs/changes/42/684742/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/kolla refs/changes/42/684742/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/kolla refs/changes/42/684742/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/kolla refs/changes/42/684742/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/kolla refs/changes/42/684742/1"}}},"commit":{"parents":[{"commit":"e3b2bb78084866a4b2c8966bacfd8ce47161dc18","subject":"Merge \"CI: Use docker mirror proxy\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/e3b2bb78084866a4b2c8966bacfd8ce47161dc18"}]}],"author":{"name":"Mark Goddard","email":"mark@stackhpc.com","date":"2019-09-25 12:48:47.000000000","tz":60},"committer":{"name":"Mark Goddard","email":"mark@stackhpc.com","date":"2019-09-25 12:48:47.000000000","tz":60},"subject":"Install RabbitMQ in kolla-toolbox image","message":"Install RabbitMQ in kolla-toolbox image\n\nThis allows us to execute RabbitMQ Ansible modules in the kolla-toolbox\ncontainer using the kolla ansible kolla_toolbox module.\n\nPartially Implements: blueprint support-nova-cells\n\nChange-Id: I2c10c7b4f7dbdd87f2dda7c40fa44cded07c57a5\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/1e2a6dba0348a9692aeb668c8b10b7de3ebf66f9"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/1e2a6dba0348a9692aeb668c8b10b7de3ebf66f9"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[],"submit_requirements":[]}
