)]}'
{"id":"openstack%2Fkolla~684808","triplet_id":"openstack%2Fkolla~master~I8b55a6a260cab14c2292cac02a1a07729223c281","project":"openstack/kolla","branch":"master","topic":"kolla-unicode","hashtags":[],"change_id":"I8b55a6a260cab14c2292cac02a1a07729223c281","subject":"Use C.UTF-8 locale","status":"MERGED","created":"2019-09-25 16:40:45.000000000","updated":"2019-10-03 21:10:55.000000000","submitted":"2019-10-03 21:09:37.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":6,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"684808-1570136977699-ef2ef3aa","meta_rev_id":"e0dd205c2dab2d0c982e4dca15b426e7c1eb9867","_number":684808,"virtual_id_number":684808,"owner":{"_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"date":"2019-10-02 08:40:00.000000000","_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},{"value":0,"_account_id":24072,"name":"Marcin Juszkiewicz","email":"mjuszkiewicz@redhat.com","username":"hrw"},{"value":0,"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},{"value":0,"_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2019-10-03 21:09:37.000000000","post_submit":true,"permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"date":"2019-10-03 17:34:01.000000000","_account_id":23181,"name":"RDO Third Party CI","email":"dmsimard+rdothirdparty@redhat.com","username":"rdothirdparty","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":24072,"name":"Marcin Juszkiewicz","email":"mjuszkiewicz@redhat.com","username":"hrw"},"all":[{"value":0,"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},{"value":2,"date":"2019-10-02 08:34:25.000000000","_account_id":24072,"name":"Marcin Juszkiewicz","email":"mjuszkiewicz@redhat.com","username":"hrw"},{"value":2,"date":"2019-10-03 08:43:21.000000000","_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},{"value":0,"_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":23181,"name":"RDO Third Party CI","email":"dmsimard+rdothirdparty@redhat.com","username":"rdothirdparty","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},"all":[{"value":0,"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},{"value":0,"_account_id":24072,"name":"Marcin Juszkiewicz","email":"mjuszkiewicz@redhat.com","username":"hrw"},{"value":1,"date":"2019-10-03 08:43:21.000000000","_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},{"value":0,"_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":23181,"name":"RDO Third Party CI","email":"dmsimard+rdothirdparty@redhat.com","username":"rdothirdparty","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Backport-Candidate":{"all":[{"value":0,"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},{"value":0,"_account_id":24072,"name":"Marcin Juszkiewicz","email":"mjuszkiewicz@redhat.com","username":"hrw"},{"value":0,"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},{"value":0,"_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":23181,"name":"RDO Third Party CI","email":"dmsimard+rdothirdparty@redhat.com","username":"rdothirdparty","tags":["SERVICE_USER"]}],"values":{"-1":"Do Not Backport"," 0":"Backport Review Needed","+1":"Should Backport"},"description":"","default_value":0,"optional":true},"Review-Priority":{"recommended":{"_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"},"all":[{"value":0,"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},{"value":0,"_account_id":24072,"name":"Marcin Juszkiewicz","email":"mjuszkiewicz@redhat.com","username":"hrw"},{"value":0,"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},{"value":1,"date":"2019-10-02 07:26:18.000000000","_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":23181,"name":"RDO Third Party CI","email":"dmsimard+rdothirdparty@redhat.com","username":"rdothirdparty","tags":["SERVICE_USER"]}],"values":{"-1":"Branch Freeze"," 0":"No Priority","+1":"Important Change","+2":"Gate Blocker Fix / Urgent Change"},"description":"","value":1,"default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},{"_account_id":23181,"name":"RDO Third Party CI","email":"dmsimard+rdothirdparty@redhat.com","username":"rdothirdparty","tags":["SERVICE_USER"]},{"_account_id":24072,"name":"Marcin Juszkiewicz","email":"mjuszkiewicz@redhat.com","username":"hrw"},{"_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2019-10-02 08:34:25.000000000","updated_by":{"_account_id":24072,"name":"Marcin Juszkiewicz","email":"mjuszkiewicz@redhat.com","username":"hrw"},"reviewer":{"_account_id":24072,"name":"Marcin Juszkiewicz","email":"mjuszkiewicz@redhat.com","username":"hrw"},"state":"REVIEWER"},{"updated":"2019-10-02 08:40:00.000000000","updated_by":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"reviewer":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"state":"REVIEWER"},{"updated":"2019-10-03 08:43:21.000000000","updated_by":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},"reviewer":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},"state":"REVIEWER"},{"updated":"2019-10-03 17:34:01.000000000","updated_by":{"_account_id":23181,"name":"RDO Third Party CI","email":"dmsimard+rdothirdparty@redhat.com","username":"rdothirdparty","tags":["SERVICE_USER"]},"reviewer":{"_account_id":23181,"name":"RDO Third Party CI","email":"dmsimard+rdothirdparty@redhat.com","username":"rdothirdparty","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2019-10-03 21:09:37.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"}],"messages":[{"id":"808133f5683f069c844eae497f17a44b322254f2","author":{"_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"},"date":"2019-09-25 16:40:45.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"227c44a16b19a2b7f09dad14ad397102fe20231e","author":{"_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"},"date":"2019-09-25 16:48:53.000000000","message":"Patch Set 1:\n\nTODO: add releasenote","accounts_in_message":[],"_revision_number":1},{"id":"a637073c91802494d0e57bfd19ce801295f0ec28","author":{"_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"},"date":"2019-09-25 17:06:50.000000000","message":"Patch Set 1:\n\nAll distros we support default to UTF-8 locale* in normal installs so it should be a more expected experience.\n\n* Usually en_US.UTF-8 rather than C.UTF-8 but at least raw glibc seems to support it.","accounts_in_message":[],"_revision_number":1},{"id":"4e64759ea9bd4733d0352e9fd14dbf8befde13c0","author":{"_account_id":23181,"name":"RDO Third Party CI","email":"dmsimard+rdothirdparty@redhat.com","username":"rdothirdparty","tags":["SERVICE_USER"]},"date":"2019-09-25 17:19:23.000000000","message":"Patch Set 1:\n\nBuild succeeded (check pipeline).\n\n- tripleo-build-containers-rhel-8-master http://logs.rdoproject.org/08/684808/1/openstack-check/tripleo-build-containers-rhel-8-master/f8ad649/ : SUCCESS in 37m 48s","accounts_in_message":[],"_revision_number":1},{"id":"a9ba5a866976974dd51c186536db920f662a45e0","author":{"_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"},"date":"2019-09-25 17:29:07.000000000","message":"Patch Set 1:\n\ntabulate 0.8.4\n0.8.5 is friendly with non-unicode again :-)","accounts_in_message":[],"_revision_number":1},{"id":"4a2d33cc70326dcf740d8feb5334110a3033ddae","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-09-25 19:14:21.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\n\n- kolla-build-centos-binary https://zuul.opendev.org/t/openstack/build/ad592f9edb294516814d6f978954da36 : SUCCESS in 44m 54s (non-voting)\n- kolla-build-centos-source https://zuul.opendev.org/t/openstack/build/a2310981105f404bb401464a18ab0122 : SUCCESS in 57m 33s\n- kolla-ansible-centos-source https://zuul.opendev.org/t/openstack/build/5777be5f16ed4289a302d0dd9983b342 : SUCCESS in 1h 21m 14s\n- kolla-ansible-centos-binary https://zuul.opendev.org/t/openstack/build/184ccfc1bd1544c38f1bb9f75602aca2 : SUCCESS in 1h 25m 36s (non-voting)\n- kolla-ansible-centos-source-upgrade https://zuul.opendev.org/t/openstack/build/50677d26b72343269071ba923c3e97c5 : SUCCESS in 1h 25m 50s\n- tripleo-build-containers-centos-7 https://zuul.opendev.org/t/openstack/build/0417145722554017acda0a0a9c3a649a : SUCCESS in 48m 54s\n- kolla-ansible-bifrost-centos-source https://zuul.opendev.org/t/openstack/build/60289fb763fd4d0182624595eb126c07 : SUCCESS in 56m 01s (non-voting)\n- kolla-ansible-centos-source-ceph https://zuul.opendev.org/t/openstack/build/3f2beb9dc5394a59a78ef40d164281f8 : FAILURE in 1h 37m 00s (non-voting)\n- kolla-ansible-centos-source-upgrade-ceph https://zuul.opendev.org/t/openstack/build/6bd321ee3cbc47f4b93868c0ecaec2be : SUCCESS in 2h 27m 25s (non-voting)\n- kolla-ansible-centos-source-cinder-lvm https://zuul.opendev.org/t/openstack/build/20e5c4fba2ca4072be1a938b82e8b533 : TIMED_OUT in 2h 04m 15s (non-voting)\n- kolla-ansible-centos-source-scenario-nfv https://zuul.opendev.org/t/openstack/build/a09f80e4f10e4c8eaaa2112e74905886 : TIMED_OUT in 2h 07m 20s (non-voting)\n- kolla-ansible-centos-source-ironic https://zuul.opendev.org/t/openstack/build/5dd50b17f1454c26b7f4458c83d2cb3d : SUCCESS in 1h 15m 02s (non-voting)\n- kolla-build-debian-source https://zuul.opendev.org/t/openstack/build/cc0e89f03d604407915e59b375f6f5d7 : SUCCESS in 27m 34s (non-voting)\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/da36034d7b014f06b88c59f61a5d4557 : SUCCESS in 4m 13s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/414acf9e520d4c038d68d5219a518eac : SUCCESS in 5m 19s\n- openstack-tox-py27 https://zuul.opendev.org/t/openstack/build/42ccf1eccbd1429a9eb284c44c759c89 : SUCCESS in 3m 55s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/4164efa1204349c9be8ec331743fedc8 : SUCCESS in 2m 50s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/138d2b937ac54cc980eff1a0613e11bc : SUCCESS in 4m 16s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/fcb68102b7584747b50759cae3ac6a69 : SUCCESS in 5m 19s\n- kolla-build-ubuntu-binary https://zuul.opendev.org/t/openstack/build/be16ff09a22447d5a51491b2bebbb18e : SUCCESS in 29m 18s (non-voting)\n- kolla-build-ubuntu-source https://zuul.opendev.org/t/openstack/build/2b8fcdf4f9464be1887b3f00344421fd : SUCCESS in 30m 30s\n- kolla-ansible-ubuntu-source https://zuul.opendev.org/t/openstack/build/c2305a5515684e5db60d3202dd2a6318 : SUCCESS in 55m 26s\n- kolla-ansible-ubuntu-binary https://zuul.opendev.org/t/openstack/build/585cd5eb8a2f42fe9665e7cbb305c24e : SUCCESS in 1h 01m 19s (non-voting)\n- kolla-ansible-ubuntu-source-upgrade https://zuul.opendev.org/t/openstack/build/5f38d5febbde452dbf06b30a8a9491bb : SUCCESS in 55m 07s\n- kolla-ansible-ubuntu-source-zun https://zuul.opendev.org/t/openstack/build/0c83fe93068244a4be27a6ba47d6bb42 : SUCCESS in 1h 09m 18s (non-voting)\n- kolla-ansible-ubuntu-source-ceph https://zuul.opendev.org/t/openstack/build/719540772c454e6cbc8ebe3185bea120 : SUCCESS in 1h 48m 41s (non-voting)\n- kolla-ansible-ubuntu-source-upgrade-ceph https://zuul.opendev.org/t/openstack/build/6473e951cff8452595aa999a02f15095 : SUCCESS in 1h 46m 09s (non-voting)\n- kolla-ansible-ubuntu-source-cinder-lvm https://zuul.opendev.org/t/openstack/build/81869e8a2e204eb2a6add855c2d98a19 : SUCCESS in 1h 24m 12s (non-voting)\n- kolla-ansible-ubuntu-source-ironic https://zuul.opendev.org/t/openstack/build/ddf4129576bd4634a585d9b1aa72be67 : SUCCESS in 1h 15m 41s (non-voting)","accounts_in_message":[],"_revision_number":1},{"id":"7db64e9ea384dbb1ed7f7111664a3d192f4ad566","author":{"_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"},"date":"2019-09-26 07:29:18.000000000","message":"Uploaded patch set 2.","accounts_in_message":[],"_revision_number":2},{"id":"c903d5555c938941f68458eb8bbc8819f23077ff","author":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},"date":"2019-09-26 07:55:14.000000000","message":"Patch Set 2:\n\nIs this still required, given tabulate was updated?","accounts_in_message":[],"_revision_number":2},{"id":"b7e026ba30dae542f53e03f1b6336dec99cd9fe0","author":{"_account_id":23181,"name":"RDO Third Party CI","email":"dmsimard+rdothirdparty@redhat.com","username":"rdothirdparty","tags":["SERVICE_USER"]},"date":"2019-09-26 08:07:41.000000000","message":"Patch Set 2:\n\nBuild succeeded (check pipeline).\n\n- tripleo-build-containers-rhel-8-master http://logs.rdoproject.org/08/684808/2/openstack-check/tripleo-build-containers-rhel-8-master/ff0435b/ : SUCCESS in 37m 31s","accounts_in_message":[],"_revision_number":2},{"id":"dbc5e9122766e5acb361f868288b275cbca14faf","author":{"_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"},"date":"2019-09-26 08:08:58.000000000","message":"Patch Set 2:\n\n\u003e Is this still required, given tabulate was updated?\n\nRequired no but don\u0027t we want saner defaults anyway?","accounts_in_message":[],"_revision_number":2},{"id":"712907812fb61933b288886fb4ae571be82ea31f","author":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},"date":"2019-09-26 09:26:32.000000000","message":"Patch Set 2:\n\n(3 comments)\n\nIt\u0027s not an area I\u0027m too familiar with, but seems like one that could easily introduce subtle issues.","accounts_in_message":[],"_revision_number":2},{"id":"c13f906e15f34e2fb8bca23d5d394c0004ba35cb","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-09-26 10:07:21.000000000","message":"Patch Set 2: Verified+1\n\nBuild succeeded (check pipeline).\n\n- kolla-build-centos-binary https://zuul.opendev.org/t/openstack/build/9d8680b5a33c48469ad221a25153e3ea : SUCCESS in 48m 32s (non-voting)\n- kolla-build-centos-source https://zuul.opendev.org/t/openstack/build/51ad66e54c934676aa8e0f92087b9fa8 : SUCCESS in 51m 03s\n- kolla-ansible-centos-source https://zuul.opendev.org/t/openstack/build/c5acfff2004b424d828b10188394c6ce : SUCCESS in 1h 57m 12s\n- kolla-ansible-centos-binary https://zuul.opendev.org/t/openstack/build/2ca6084f5e57470f80e1aeb7e981b76f : SUCCESS in 1h 22m 23s (non-voting)\n- kolla-ansible-centos-source-upgrade https://zuul.opendev.org/t/openstack/build/61b3e07bb3114b7e9cb1aa7083c3678e : SUCCESS in 1h 27m 21s\n- tripleo-build-containers-centos-7 https://zuul.opendev.org/t/openstack/build/4c0b5938bcd14c0cb43274a4a5a2dd15 : SUCCESS in 1h 08m 00s\n- kolla-ansible-bifrost-centos-source https://zuul.opendev.org/t/openstack/build/7e1e38a5356a445cbba5d48891771853 : SUCCESS in 52m 26s (non-voting)\n- kolla-ansible-centos-source-ceph https://zuul.opendev.org/t/openstack/build/4a13f98b55644337a0fef7d65a0563bb : FAILURE in 2h 08m 20s (non-voting)\n- kolla-ansible-centos-source-upgrade-ceph https://zuul.opendev.org/t/openstack/build/722716814c29442db8b222334b8355c6 : SUCCESS in 2h 32m 13s (non-voting)\n- kolla-ansible-centos-source-cinder-lvm https://zuul.opendev.org/t/openstack/build/4a3aedb4050a4734a0b0808c05aa8adc : SUCCESS in 1h 50m 16s (non-voting)\n- kolla-ansible-centos-source-scenario-nfv https://zuul.opendev.org/t/openstack/build/76b13aea6c014656be5f8ce2d9784ec2 : TIMED_OUT in 2h 02m 59s (non-voting)\n- kolla-ansible-centos-source-ironic https://zuul.opendev.org/t/openstack/build/f845a02e7fa744f9b585c35126d22855 : SUCCESS in 1h 35m 17s (non-voting)\n- kolla-build-debian-source https://zuul.opendev.org/t/openstack/build/3e67d703a1e54790bd6e13fef1be8a9b : SUCCESS in 25m 34s (non-voting)\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/f3e25f7c40ef4fa1b76642747af98ee9 : SUCCESS in 3m 27s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/4ee5f7cc917a4cf2a87270e39ceb3127 : SUCCESS in 5m 34s\n- openstack-tox-py27 https://zuul.opendev.org/t/openstack/build/7ed52025b3e94a2ea9f517b5eaa699b6 : SUCCESS in 3m 04s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/db2dff79ec514672b460f5aeecab1ac0 : SUCCESS in 3m 49s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/6ff78e0873544d49a52369e7d02e5bfd : SUCCESS in 3m 30s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/3f8a4314e78f4f5a873d0c274643069b : SUCCESS in 5m 12s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/2bbb1348f4ac432f883c2e0ea1e05418 : SUCCESS in 5m 04s\n- kolla-build-ubuntu-binary https://zuul.opendev.org/t/openstack/build/e0d77fb4e2914ee3a20f1b52ac07147a : SUCCESS in 30m 13s (non-voting)\n- kolla-build-ubuntu-source https://zuul.opendev.org/t/openstack/build/20ab72bcd08b45fabd1b98bdaffdcf70 : SUCCESS in 27m 02s\n- kolla-ansible-ubuntu-source https://zuul.opendev.org/t/openstack/build/cc8576199c5b476887f1111b347ca58c : SUCCESS in 58m 41s\n- kolla-ansible-ubuntu-binary https://zuul.opendev.org/t/openstack/build/d7f755bc7e5c468cb8b003fa34a54963 : SUCCESS in 1h 01m 47s (non-voting)\n- kolla-ansible-ubuntu-source-upgrade https://zuul.opendev.org/t/openstack/build/9be7c54e06124858a2d6c18fb2d60962 : SUCCESS in 1h 01m 35s\n- kolla-ansible-ubuntu-source-zun https://zuul.opendev.org/t/openstack/build/610db8dde8064c02bb13617dfafd0dcd : FAILURE in 1h 17m 45s (non-voting)\n- kolla-ansible-ubuntu-source-ceph https://zuul.opendev.org/t/openstack/build/3970892a601744f4b070a5c57f810250 : SUCCESS in 1h 41m 00s (non-voting)\n- kolla-ansible-ubuntu-source-upgrade-ceph https://zuul.opendev.org/t/openstack/build/82ab8189ebbf46e4b84fc4631bd68185 : SUCCESS in 1h 47m 31s (non-voting)\n- kolla-ansible-ubuntu-source-cinder-lvm https://zuul.opendev.org/t/openstack/build/39509cf589e24087bf81575155e08ae4 : SUCCESS in 1h 30m 49s (non-voting)\n- kolla-ansible-ubuntu-source-ironic https://zuul.opendev.org/t/openstack/build/5618e08e510945a3a0cde2f971e0564f : SUCCESS in 1h 06m 45s (non-voting)","accounts_in_message":[],"_revision_number":2},{"id":"6b1d266689f21f9eeacd0986743ec010886787c6","author":{"_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"},"date":"2019-09-26 10:30:11.000000000","message":"Uploaded patch set 3.","accounts_in_message":[],"_revision_number":3},{"id":"881ae5303830a6934a652996b407bf758488b7be","author":{"_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"},"date":"2019-09-26 10:36:31.000000000","message":"Patch Set 3:\n\n(1 comment)","accounts_in_message":[],"_revision_number":3},{"id":"7721b2252ca6520471be89957489f1ae761e4bb6","author":{"_account_id":23181,"name":"RDO Third Party CI","email":"dmsimard+rdothirdparty@redhat.com","username":"rdothirdparty","tags":["SERVICE_USER"]},"date":"2019-09-26 11:54:00.000000000","message":"Patch Set 3:\n\nBuild succeeded (check pipeline).\n\n- tripleo-build-containers-rhel-8-master http://logs.rdoproject.org/08/684808/3/openstack-check/tripleo-build-containers-rhel-8-master/3ae62ea/ : SUCCESS in 46m 27s","accounts_in_message":[],"_revision_number":3},{"id":"967f718e7aa6e7c3413892924d0693ca2fbd45fb","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-09-26 13:14:36.000000000","message":"Patch Set 3: Verified+1\n\nBuild succeeded (check pipeline).\n\n- kolla-build-centos-binary https://zuul.opendev.org/t/openstack/build/63575fcdc84c4dfc8e9cc5ca74ff5484 : SUCCESS in 44m 26s (non-voting)\n- kolla-build-centos-source https://zuul.opendev.org/t/openstack/build/6e551abbd1ff48f0a95e771f37ba2b15 : SUCCESS in 59m 02s\n- kolla-ansible-centos-source https://zuul.opendev.org/t/openstack/build/8865ba5368d94189956e850238992ce3 : SUCCESS in 1h 11m 42s\n- kolla-ansible-centos-binary https://zuul.opendev.org/t/openstack/build/194ff12ca1aa4dfa9da38c7106dcb061 : SUCCESS in 1h 18m 21s (non-voting)\n- kolla-ansible-centos-source-upgrade https://zuul.opendev.org/t/openstack/build/36ef0eb3712949968222fbba12b6e7ad : SUCCESS in 1h 49m 05s\n- tripleo-build-containers-centos-7 https://zuul.opendev.org/t/openstack/build/cba88508f9c743348562c05c1a7f673b : SUCCESS in 58m 26s\n- kolla-ansible-bifrost-centos-source https://zuul.opendev.org/t/openstack/build/6c9eee30daa84c0e85cf4c298e707593 : SUCCESS in 48m 21s (non-voting)\n- kolla-ansible-centos-source-ceph https://zuul.opendev.org/t/openstack/build/fa86bab0fc334bf19e4df0f88e6e5907 : SUCCESS in 2h 16m 03s (non-voting)\n- kolla-ansible-centos-source-upgrade-ceph https://zuul.opendev.org/t/openstack/build/3e2f644117d64d11b60b164285da6b55 : TIMED_OUT in 2h 34m 29s (non-voting)\n- kolla-ansible-centos-source-cinder-lvm https://zuul.opendev.org/t/openstack/build/22bb989101434a16b0954e2d5f6a9560 : SUCCESS in 2h 01m 02s (non-voting)\n- kolla-ansible-centos-source-scenario-nfv https://zuul.opendev.org/t/openstack/build/a5dd822566f548caae274253515eed6a : TIMED_OUT in 2h 02m 33s (non-voting)\n- kolla-ansible-centos-source-ironic https://zuul.opendev.org/t/openstack/build/3366d5750e9444c594082b61de6e633a : SUCCESS in 1h 39m 56s (non-voting)\n- kolla-build-debian-source https://zuul.opendev.org/t/openstack/build/e847fdceba884f80815b21fc45091ab2 : SUCCESS in 27m 05s (non-voting)\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/df17cf3bf8f249b2b779409d7e3f92cd : SUCCESS in 4m 59s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/84dad4790f464932884c704d375fcdce : SUCCESS in 6m 56s\n- openstack-tox-py27 https://zuul.opendev.org/t/openstack/build/61761f902c4146b3b2c8b163c3797666 : SUCCESS in 2m 52s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/afedc427f849499f9941b69966a5cb92 : SUCCESS in 3m 22s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/8c67def022a74e8a88ac524b7cf232d8 : SUCCESS in 4m 00s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/a9330ea3b03b4edc8aabba3f1c30b0ef : SUCCESS in 12m 26s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/2d27bce315cf41f09da04d48681f65ed : SUCCESS in 5m 53s\n- kolla-build-ubuntu-binary https://zuul.opendev.org/t/openstack/build/8ccca66a643548f2946827a452eb2185 : SUCCESS in 29m 34s (non-voting)\n- kolla-build-ubuntu-source https://zuul.opendev.org/t/openstack/build/15dcf713767f4cf6b7404e6fe093aae4 : SUCCESS in 30m 41s\n- kolla-ansible-ubuntu-source https://zuul.opendev.org/t/openstack/build/de9df0332bf941a68bdcc2224984a88a : SUCCESS in 51m 02s\n- kolla-ansible-ubuntu-binary https://zuul.opendev.org/t/openstack/build/145a561f04c246d4863ce7db792fab16 : SUCCESS in 48m 30s (non-voting)\n- kolla-ansible-ubuntu-source-upgrade https://zuul.opendev.org/t/openstack/build/11ee6ee17aef4a598808d9321ee3556c : SUCCESS in 1h 16m 46s\n- kolla-ansible-ubuntu-source-zun https://zuul.opendev.org/t/openstack/build/0affefa2e46247a69341d129840a6558 : SUCCESS in 1h 08m 23s (non-voting)\n- kolla-ansible-ubuntu-source-ceph https://zuul.opendev.org/t/openstack/build/d30feb1fa6694dc5a98a03e9cecde737 : SUCCESS in 1h 47m 54s (non-voting)\n- kolla-ansible-ubuntu-source-upgrade-ceph https://zuul.opendev.org/t/openstack/build/766ae3f8fdaf4bb2b2467cac50f6d474 : SUCCESS in 1h 54m 55s (non-voting)\n- kolla-ansible-ubuntu-source-cinder-lvm https://zuul.opendev.org/t/openstack/build/64fa80eca45d4c7686f3531b3198ebf8 : SUCCESS in 1h 35m 32s (non-voting)\n- kolla-ansible-ubuntu-source-ironic https://zuul.opendev.org/t/openstack/build/198abcd879fb4f729ed4b33ecefda320 : SUCCESS in 1h 01m 46s (non-voting)","accounts_in_message":[],"_revision_number":3},{"id":"5e3c970d580dec85b2c679702d2476c8075f6bb9","author":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"date":"2019-09-30 07:28:39.000000000","message":"Patch Set 3:\n\n(1 comment)","accounts_in_message":[],"_revision_number":3},{"id":"c90bfb1ca9617a257b65bd48b14363b94e4ca873","author":{"_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"},"date":"2019-09-30 10:56:41.000000000","message":"Patch Set 3:\n\n(1 comment)","accounts_in_message":[],"_revision_number":3},{"id":"586755d15d632b50d7d1f59c9f63e4ea0f605486","author":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"date":"2019-09-30 12:32:31.000000000","message":"Patch Set 3:\n\n\u003e (1 comment)\n\nWell, it would be easier to add the locale to the top of the Dockerfile, and do kolla-build --no-locale if you don\u0027t want it set for a reason. I don\u0027t have a problem with a block override, but I think from a user experience kolla-build option would be better (if anybody ever will want to change it) :-)","accounts_in_message":[],"_revision_number":3},{"id":"ee7940612c88e75c3798fb40662151b43f2ded4a","author":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},"date":"2019-09-30 12:42:45.000000000","message":"Patch Set 3:\n\n\u003e \u003e (1 comment)\n \u003e \n \u003e Well, it would be easier to add the locale to the top of the\n \u003e Dockerfile, and do kolla-build --no-locale if you don\u0027t want it set\n \u003e for a reason. I don\u0027t have a problem with a block override, but I\n \u003e think from a user experience kolla-build option would be better (if\n \u003e anybody ever will want to change it) :-)\n\nOTOH, if we make it easier to configure it looks like we support changing it arbitrarily. Are we sure we don\u0027t depend on US english text output in some places?","accounts_in_message":[],"_revision_number":3},{"id":"d93d46daf06cdaa4612b1fcbaddecc499ee62170","author":{"_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"},"date":"2019-10-01 06:20:36.000000000","message":"Patch Set 3:\n\n\u003e \u003e \u003e (1 comment)\n \u003e \u003e\n \u003e \u003e Well, it would be easier to add the locale to the top of the\n \u003e \u003e Dockerfile, and do kolla-build --no-locale if you don\u0027t want it\n \u003e set\n \u003e \u003e for a reason. I don\u0027t have a problem with a block override, but I\n \u003e \u003e think from a user experience kolla-build option would be better\n \u003e (if\n \u003e \u003e anybody ever will want to change it) :-)\n \u003e \n \u003e OTOH, if we make it easier to configure it looks like we support\n \u003e changing it arbitrarily. Are we sure we don\u0027t depend on US english\n \u003e text output in some places?\n\nAnd C (aka POSIX) specifically :-)\nThis is also the reason why I moved to C.UTF-8 (which is just C/POSIX telling glibc Unicode is available) rather than en_US.UTF-8 which may change messages (unlikely though compared to trying de_DE.UTF-8 of course).","accounts_in_message":[],"_revision_number":3},{"id":"a1c56b2492fbd007c95d824042b54a97566b0d2d","author":{"_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"},"date":"2019-10-02 07:26:18.000000000","message":"Patch Set 3: Review-Priority+1\n\nI strongly feel we should be getting this in Train.","accounts_in_message":[],"_revision_number":3},{"id":"71a654e2b525ca267db071de13abc794810bfda3","author":{"_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"},"date":"2019-10-02 08:10:38.000000000","message":"Patch Set 3:\n\nYup, we get hit by that not knowing it:\n2019-10-02 07:21:00.045 17 ERROR glance.db.sqlalchemy.metadata [-] Failed to parse json file /etc/glance/metadefs/compute-hypervisor.json while populating metadata due to: \u0027ascii\u0027 codec can\u0027t decode byte 0xc2 in position 898: ordinal not in range(128): UnicodeDecodeError: \u0027ascii\u0027 codec can\u0027t decode byte 0xc2 in position 898: ordinal not in range(128)\n2019-10-02 07:21:00.656 17 ERROR glance.db.sqlalchemy.metadata [-] Failed to parse json file /etc/glance/metadefs/software-runtimes.json while populating metadata due to: \u0027ascii\u0027 codec can\u0027t decode byte 0xe2 in position 1367: ordinal not in range(128): UnicodeDecodeError: \u0027ascii\u0027 codec can\u0027t decode byte 0xe2 in position 1367: ordinal not in range(128)\n2019-10-02 07:21:00.707 17 ERROR glance.db.sqlalchemy.metadata [-] Failed to parse json file /etc/glance/metadefs/compute-trust.json while populating metadata due to: \u0027ascii\u0027 codec can\u0027t decode byte 0xc2 in position 90: ordinal not in range(128): UnicodeDecodeError: \u0027ascii\u0027 codec can\u0027t decode byte 0xc2 in position 90: ordinal not in range(128)\n2019-10-02 07:39:48.431 16 ERROR glance.db.sqlalchemy.metadata [-] Failed to parse json file /etc/glance/metadefs/compute-hypervisor.json while populating metadata due to: \u0027ascii\u0027 codec can\u0027t decode byte 0xc2 in position 898: ordinal not in range(128): UnicodeDecodeError: \u0027ascii\u0027 codec can\u0027t decode byte 0xc2 in position 898: ordinal not in range(128)\n2019-10-02 07:39:48.472 16 ERROR glance.db.sqlalchemy.metadata [-] Failed to parse json file /etc/glance/metadefs/software-runtimes.json while populating metadata due to: \u0027ascii\u0027 codec can\u0027t decode byte 0xe2 in position 1367: ordinal not in range(128): UnicodeDecodeError: \u0027ascii\u0027 codec can\u0027t decode byte 0xe2 in position 1367: ordinal not in range(128)\n2019-10-02 07:39:48.477 16 ERROR glance.db.sqlalchemy.metadata [-] Failed to parse json file /etc/glance/metadefs/compute-trust.json while populating metadata due to: \u0027ascii\u0027 codec can\u0027t decode byte 0xc2 in position 90: ordinal not in range(128): UnicodeDecodeError: \u0027ascii\u0027 codec can\u0027t decode byte 0xc2 in position 90: ordinal not in range(128)","accounts_in_message":[],"_revision_number":3},{"id":"bc5ad0bd6054b22b1d00c7621b8eabc94f9f3836","author":{"_account_id":24072,"name":"Marcin Juszkiewicz","email":"mjuszkiewicz@redhat.com","username":"hrw"},"date":"2019-10-02 08:34:25.000000000","message":"Patch Set 3: Code-Review+2","accounts_in_message":[],"_revision_number":3},{"id":"f2c839bdf2f758da2c63882b5a22d0ab1cb74567","author":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"date":"2019-10-02 08:40:00.000000000","message":"Patch Set 3:\n\nSeems docker python image had the same issue: https://github.com/docker-library/python/pull/14/files","accounts_in_message":[],"_revision_number":3},{"id":"b171b5a997cb7a5cd217cf4e7e156db551b4ed41","author":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},"date":"2019-10-03 08:43:21.000000000","message":"Patch Set 3: Code-Review+2 Workflow+1\n\nOk, let\u0027s do it.","accounts_in_message":[],"_revision_number":3},{"id":"5535c63a6f3c93646f1df0cdee966491435551c5","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-10-03 08:43:35.000000000","message":"Patch Set 3: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":3},{"id":"a4711719cc6e4ce1698d93033c2cbd43f63fca23","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-10-03 10:33:19.000000000","message":"Patch Set 3: Verified-2\n\nBuild failed (gate pipeline).  For information on how to proceed, see\nhttp://docs.openstack.org/infra/manual/developers.html#automated-testing\n\n\n- kolla-build-centos-source https://zuul.opendev.org/t/openstack/build/1841aa439db840df9ec8f3a99f728399 : SUCCESS in 56m 21s\n- kolla-ansible-centos-source https://zuul.opendev.org/t/openstack/build/fc08ed7fc5f3465f874d48d778b57ac0 : SUCCESS in 1h 32m 42s\n- kolla-ansible-centos-source-upgrade https://zuul.opendev.org/t/openstack/build/2ab63e4cffc745f792d17e5375ce3530 : SUCCESS in 1h 42m 46s\n- tripleo-build-containers-centos-7 https://zuul.opendev.org/t/openstack/build/03ad34d9fc8247e9abe04a782ee468ce : SUCCESS in 56m 24s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/00e583c32e3146b3a41affa1f039cfc4 : SUCCESS in 3m 22s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/fa19039d31ea48ddb6fe2323619365fa : SUCCESS in 3m 55s\n- openstack-tox-py27 https://zuul.opendev.org/t/openstack/build/3701060ca10c4f5bb7b9729864855a59 : SUCCESS in 3m 12s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/cf959863ecda41fe82c7ec9cdcb05259 : POST_FAILURE in 18m 24s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/4b1446e5db8245719934aa547fdf2bac : SUCCESS in 3m 11s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/b7f607d96b7c4975b9f2a9bb38cf53e5 : SUCCESS in 5m 09s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/5d81edc6c6b7457193c86fe387f46687 : SUCCESS in 4m 32s\n- kolla-build-ubuntu-source https://zuul.opendev.org/t/openstack/build/617bcbfa8d3f43d28fb651d6dace71b7 : SUCCESS in 33m 17s\n- kolla-ansible-ubuntu-source https://zuul.opendev.org/t/openstack/build/96b89baab0da4e34bbeae847b6d8d5a6 : SUCCESS in 59m 06s\n- kolla-ansible-ubuntu-source-upgrade https://zuul.opendev.org/t/openstack/build/47860d30b8eb4098a664f12edb08fadb : SUCCESS in 56m 25s","accounts_in_message":[],"_revision_number":3},{"id":"a52b077ca3ea5eec922223e1015fb4174c761c2c","author":{"_account_id":24072,"name":"Marcin Juszkiewicz","email":"mjuszkiewicz@redhat.com","username":"hrw"},"date":"2019-10-03 16:48:09.000000000","message":"Patch Set 3:\n\nrecheck","accounts_in_message":[],"_revision_number":3},{"id":"fc3c1f2a2820c3a6ddb45c13a252c74c8ad26325","author":{"_account_id":23181,"name":"RDO Third Party CI","email":"dmsimard+rdothirdparty@redhat.com","username":"rdothirdparty","tags":["SERVICE_USER"]},"date":"2019-10-03 17:34:01.000000000","message":"Patch Set 3:\n\nBuild succeeded (check pipeline).\n\n- tripleo-build-containers-rhel-8-master http://logs.rdoproject.org/08/684808/3/openstack-check/tripleo-build-containers-rhel-8-master/0f8391f/ : SUCCESS in 44m 10s","accounts_in_message":[],"_revision_number":3},{"id":"a329151da3ca128b6a5b9482d23eed0dbd317c37","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-10-03 19:34:55.000000000","message":"Patch Set 3: Verified+1\n\nBuild succeeded (check pipeline).\n\n- kolla-build-centos-binary https://zuul.opendev.org/t/openstack/build/3b07d91356c54296b3e649e81f937af5 : SUCCESS in 53m 34s (non-voting)\n- kolla-build-centos-source https://zuul.opendev.org/t/openstack/build/206c516d675e40fbabcbf6af9cf615cd : SUCCESS in 1h 07m 50s\n- kolla-ansible-centos-source https://zuul.opendev.org/t/openstack/build/4f38c20863f846b2a05ce5ef6df9f64f : SUCCESS in 1h 28m 18s\n- kolla-ansible-centos-binary https://zuul.opendev.org/t/openstack/build/a9a9f560d3ae44148f3dbb9c5cedd0f8 : SUCCESS in 1h 23m 28s (non-voting)\n- kolla-ansible-centos-source-upgrade https://zuul.opendev.org/t/openstack/build/ee81deb8bf69426bbf0246e2127f13d0 : SUCCESS in 1h 25m 55s\n- tripleo-build-containers-centos-7 https://zuul.opendev.org/t/openstack/build/e0599e8345ce4ad09c2da512e10fccf0 : SUCCESS in 1h 05m 21s\n- kolla-ansible-bifrost-centos-source https://zuul.opendev.org/t/openstack/build/73e57bcaa7bb44489528724a5aefeb36 : SUCCESS in 59m 16s (non-voting)\n- kolla-ansible-centos-source-ceph https://zuul.opendev.org/t/openstack/build/f3a71abcaef840e59e474a9d22fd6815 : SUCCESS in 2h 07m 05s (non-voting)\n- kolla-ansible-centos-source-upgrade-ceph https://zuul.opendev.org/t/openstack/build/bfcc291936764d95a16737288762b1f8 : SUCCESS in 2h 32m 56s (non-voting)\n- kolla-ansible-centos-source-cinder-lvm https://zuul.opendev.org/t/openstack/build/ea86df4b855f40529f7c99328bbda355 : SUCCESS in 2h 00m 16s (non-voting)\n- kolla-ansible-centos-source-scenario-nfv https://zuul.opendev.org/t/openstack/build/5edbd607b0f8497f93cea8acdbc0d59c : TIMED_OUT in 2h 03m 24s (non-voting)\n- kolla-ansible-centos-source-ironic https://zuul.opendev.org/t/openstack/build/bf3aa3f4c22649ed9ae47945906b6721 : SUCCESS in 1h 40m 30s (non-voting)\n- kolla-build-debian-source https://zuul.opendev.org/t/openstack/build/85cdb851bb3a45cfa7c00c8ffe34826a : SUCCESS in 33m 12s (non-voting)\n- kolla-build-debian-binary https://zuul.opendev.org/t/openstack/build/633bba24d6b04613831fcb1b863aade1 : SUCCESS in 20m 05s (non-voting)\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/1f7c91d9dfc04168a9e39549f83c76ce : SUCCESS in 2m 52s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/2a3b756e10cc41418b397fb50d303c60 : SUCCESS in 4m 27s\n- openstack-tox-py27 https://zuul.opendev.org/t/openstack/build/4ec1a52a9c694b5b8ec13cdd54755a2e : SUCCESS in 3m 26s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/09f0aa65d43942ad95a8f002dee8a1f4 : SUCCESS in 3m 14s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/c969346e51854ee9889f2c43d5460141 : SUCCESS in 4m 16s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/0997cb0874bf48e3a130109620c6bfaf : SUCCESS in 6m 12s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/8c5e2669525447e2b1d0ed87a6102502 : SUCCESS in 5m 13s\n- kolla-build-ubuntu-binary https://zuul.opendev.org/t/openstack/build/9781c782854b4035a8290022dfaa9bb8 : FAILURE in 23m 45s (non-voting)\n- kolla-build-ubuntu-source https://zuul.opendev.org/t/openstack/build/d91b9d873b9a45f0af6abeb820c29573 : SUCCESS in 43m 38s\n- kolla-ansible-ubuntu-source https://zuul.opendev.org/t/openstack/build/d82badf8565c439eacaacbe6f07b997f : SUCCESS in 1h 05m 15s\n- kolla-ansible-ubuntu-binary https://zuul.opendev.org/t/openstack/build/f882f344a6a644f084f27d6d5b2e3fec : SUCCESS in 55m 17s (non-voting)\n- kolla-ansible-ubuntu-source-upgrade https://zuul.opendev.org/t/openstack/build/bb4552af396b4eaa841c2dccb6317026 : SUCCESS in 59m 35s\n- kolla-ansible-ubuntu-source-zun https://zuul.opendev.org/t/openstack/build/901f36623de44df999b6caa93b2bf736 : SUCCESS in 1h 20m 41s (non-voting)\n- kolla-ansible-ubuntu-source-ceph https://zuul.opendev.org/t/openstack/build/3b8e094578474bc59a2afab2e0913700 : SUCCESS in 1h 29m 37s (non-voting)\n- kolla-ansible-ubuntu-source-upgrade-ceph https://zuul.opendev.org/t/openstack/build/e94b3d03e9604242b13b53a242b7570c : SUCCESS in 1h 50m 07s (non-voting)\n- kolla-ansible-ubuntu-source-cinder-lvm https://zuul.opendev.org/t/openstack/build/d414b12ca74246a2ac4255133dd6a9f8 : SUCCESS in 1h 15m 59s (non-voting)\n- kolla-ansible-ubuntu-source-ironic https://zuul.opendev.org/t/openstack/build/9233d93a700f4016b30b06cd30e334e0 : SUCCESS in 1h 07m 52s (non-voting)","accounts_in_message":[],"_revision_number":3},{"id":"b56326b09c79caab4018894634fe6a1fa146fc4a","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-10-03 19:35:14.000000000","message":"Patch Set 3: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":3},{"id":"5178a6b59277c216aaca62ca51a2a96c8c910855","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-10-03 21:09:37.000000000","message":"Change has been successfully merged by Zuul","accounts_in_message":[],"_revision_number":3},{"id":"e20959ea5304abf7632654645c8d1204d66e8a6a","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-10-03 21:09:37.000000000","message":"Patch Set 3: Verified+2\n\nBuild succeeded (gate pipeline).\n\n- kolla-build-centos-source https://zuul.opendev.org/t/openstack/build/1e21b878027844588709bcaed3c64d19 : SUCCESS in 1h 00m 19s\n- kolla-ansible-centos-source https://zuul.opendev.org/t/openstack/build/0ea4b541030f428297d55b7497b5b731 : SUCCESS in 1h 25m 05s\n- kolla-ansible-centos-source-upgrade https://zuul.opendev.org/t/openstack/build/fb81c2698a1c44b8b0f2fa9ff9bc716e : SUCCESS in 1h 28m 47s\n- tripleo-build-containers-centos-7 https://zuul.opendev.org/t/openstack/build/741955397c814d44895b6ebbc60c85f7 : SUCCESS in 56m 42s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/e38d1f2bdaff42119980d59bb4d3151b : SUCCESS in 3m 56s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/163a1faaa52843aeb13a337bccb6c83c : SUCCESS in 3m 48s\n- openstack-tox-py27 https://zuul.opendev.org/t/openstack/build/2080aeaf65504ec59ec58656e73fe02c : SUCCESS in 3m 22s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/299726a1c0c448caa68b46477669d4f2 : SUCCESS in 2m 58s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/613b1d9e5afc4f68822f60abf8c67ee3 : SUCCESS in 3m 32s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/f03a020a268f446c969d20e0ff417096 : SUCCESS in 4m 47s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/50a63b6017a44f30972a385b082e0fde : SUCCESS in 4m 37s\n- kolla-build-ubuntu-source https://zuul.opendev.org/t/openstack/build/7f698a3bb82741479231880cff532360 : SUCCESS in 30m 30s\n- kolla-ansible-ubuntu-source https://zuul.opendev.org/t/openstack/build/ddb29a48daff45a881ac115fa12edca8 : SUCCESS in 56m 57s\n- kolla-ansible-ubuntu-source-upgrade https://zuul.opendev.org/t/openstack/build/3949b14d3a24438da161d4f435d90919 : SUCCESS in 58m 07s","accounts_in_message":[],"_revision_number":3},{"id":"0b1edd920832a90f5300500e776dd0cb5d171f18","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-10-03 21:10:55.000000000","message":"Patch Set 3:\n\nBuild succeeded (promote pipeline).\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/896eff1213bd4524992af4df64c070c9 : SUCCESS in 50s\n- promote-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/5b3c31ae32a74fd2944ec3cd3407b021 : SUCCESS in 57s","accounts_in_message":[],"_revision_number":3}],"current_revision_number":3,"current_revision":"a266023ab572f6dc728271a2f3936c385674c08a","revisions":{"1aaf462b5b59d8b37954777093fe96ea590e27a1":{"kind":"REWORK","_number":1,"created":"2019-09-25 16:40:45.000000000","uploader":{"_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"},"ref":"refs/changes/08/684808/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/kolla","ref":"refs/changes/08/684808/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/kolla refs/changes/08/684808/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/kolla refs/changes/08/684808/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/kolla refs/changes/08/684808/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/kolla refs/changes/08/684808/1"}}},"commit":{"parents":[{"commit":"c08a4b5a6352f3d1ebf640687d68d00aabc11ff6","subject":"Add Debian Buster Stein backports APT repository","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/c08a4b5a6352f3d1ebf640687d68d00aabc11ff6"}]}],"author":{"name":"Radosław Piliszek","email":"radoslaw.piliszek@gmail.com","date":"2019-09-25 16:28:38.000000000","tz":120},"committer":{"name":"Radosław Piliszek","email":"radoslaw.piliszek@gmail.com","date":"2019-09-25 16:38:45.000000000","tz":120},"subject":"Use C.UTF-8 locale","message":"Use C.UTF-8 locale\n\nThis ought to fix problems with Unicode,\nincluding the recent vmtp issue with tabulate on Ubuntu.\n\nChange-Id: I8b55a6a260cab14c2292cac02a1a07729223c281\nCloses-bug: #1845306\nSigned-off-by: Radosław Piliszek \u003cradoslaw.piliszek@gmail.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/1aaf462b5b59d8b37954777093fe96ea590e27a1"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/1aaf462b5b59d8b37954777093fe96ea590e27a1"}]},"branch":"refs/heads/master"},"852d6afb86eb371826b4af4f6537089d27cc96fa":{"kind":"REWORK","_number":2,"created":"2019-09-26 07:29:18.000000000","uploader":{"_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"},"ref":"refs/changes/08/684808/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/kolla","ref":"refs/changes/08/684808/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/kolla refs/changes/08/684808/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/kolla refs/changes/08/684808/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/kolla refs/changes/08/684808/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/kolla refs/changes/08/684808/2"}}},"commit":{"parents":[{"commit":"c08a4b5a6352f3d1ebf640687d68d00aabc11ff6","subject":"Add Debian Buster Stein backports APT repository","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/c08a4b5a6352f3d1ebf640687d68d00aabc11ff6"}]}],"author":{"name":"Radosław Piliszek","email":"radoslaw.piliszek@gmail.com","date":"2019-09-25 16:28:38.000000000","tz":120},"committer":{"name":"Radosław Piliszek","email":"radoslaw.piliszek@gmail.com","date":"2019-09-26 07:26:18.000000000","tz":120},"subject":"Use C.UTF-8 locale","message":"Use C.UTF-8 locale\n\nThis ought to fix problems with Unicode,\nincluding the recent vmtp issue with tabulate 0.8.4 failing\ndue to non-Unicode locale (Ubuntu Bionic 18.04 Python 3.6).\n\nChange-Id: I8b55a6a260cab14c2292cac02a1a07729223c281\nRelated-bug: #1845306\nSigned-off-by: Radosław Piliszek \u003cradoslaw.piliszek@gmail.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/852d6afb86eb371826b4af4f6537089d27cc96fa"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/852d6afb86eb371826b4af4f6537089d27cc96fa"}]},"branch":"refs/heads/master"},"a266023ab572f6dc728271a2f3936c385674c08a":{"kind":"REWORK","_number":3,"created":"2019-09-26 10:30:11.000000000","uploader":{"_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"},"ref":"refs/changes/08/684808/3","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/kolla","ref":"refs/changes/08/684808/3","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/kolla refs/changes/08/684808/3 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/kolla refs/changes/08/684808/3 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/kolla refs/changes/08/684808/3 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/kolla refs/changes/08/684808/3"}}},"commit":{"parents":[{"commit":"c08a4b5a6352f3d1ebf640687d68d00aabc11ff6","subject":"Add Debian Buster Stein backports APT repository","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/c08a4b5a6352f3d1ebf640687d68d00aabc11ff6"}]}],"author":{"name":"Radosław Piliszek","email":"radoslaw.piliszek@gmail.com","date":"2019-09-25 16:28:38.000000000","tz":120},"committer":{"name":"Radosław Piliszek","email":"radoslaw.piliszek@gmail.com","date":"2019-09-26 10:29:19.000000000","tz":120},"subject":"Use C.UTF-8 locale","message":"Use C.UTF-8 locale\n\nThis ought to fix problems with Unicode,\nincluding the recent vmtp issue with tabulate 0.8.4 failing\ndue to non-Unicode locale (Ubuntu Bionic 18.04 Python 3.6).\n\nChange-Id: I8b55a6a260cab14c2292cac02a1a07729223c281\nRelated-bug: #1845306\nSigned-off-by: Radosław Piliszek \u003cradoslaw.piliszek@gmail.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/a266023ab572f6dc728271a2f3936c385674c08a"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/a266023ab572f6dc728271a2f3936c385674c08a"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[],"submit_requirements":[]}
