)]}'
{"id":"openstack%2Fkolla~685410","triplet_id":"openstack%2Fkolla~stable%2Fstein~Id9a76f459aff016fbc920370c0a3ec58e0397fcb","project":"openstack/kolla","branch":"stable/stein","topic":"bug/1845647-stable/stein","hashtags":[],"change_id":"Id9a76f459aff016fbc920370c0a3ec58e0397fcb","subject":"Fix monasca-grafana - pin rake to \u003c13","status":"MERGED","created":"2019-09-27 17:03:19.000000000","updated":"2019-09-27 20:43:45.000000000","submitted":"2019-09-27 20:42:40.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"685410-1569616960333-d890bd4f","meta_rev_id":"56f190b302222429a0e2717e04af0575273d9259","_number":685410,"virtual_id_number":685410,"owner":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},{"value":0,"_account_id":24072,"name":"Marcin Juszkiewicz","email":"mjuszkiewicz@redhat.com","username":"hrw"},{"value":0,"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2019-09-27 20:42:40.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"all":[{"value":2,"date":"2019-09-27 17:09:43.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},{"value":2,"date":"2019-09-27 17:05:06.000000000","_account_id":24072,"name":"Marcin Juszkiewicz","email":"mjuszkiewicz@redhat.com","username":"hrw"},{"value":0,"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"all":[{"value":1,"date":"2019-09-27 17:09:43.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},{"value":0,"_account_id":24072,"name":"Marcin Juszkiewicz","email":"mjuszkiewicz@redhat.com","username":"hrw"},{"value":0,"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Backport-Candidate":{"all":[{"value":0,"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},{"value":0,"_account_id":24072,"name":"Marcin Juszkiewicz","email":"mjuszkiewicz@redhat.com","username":"hrw"},{"value":0,"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Do Not Backport"," 0":"Backport Review Needed","+1":"Should Backport"},"description":"","default_value":0,"optional":true},"Review-Priority":{"approved":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},"all":[{"value":0,"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},{"value":0,"_account_id":24072,"name":"Marcin Juszkiewicz","email":"mjuszkiewicz@redhat.com","username":"hrw"},{"value":2,"date":"2019-09-27 17:03:55.000000000","_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Branch Freeze"," 0":"No Priority","+1":"Important Change","+2":"Gate Blocker Fix / Urgent Change"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},{"_account_id":24072,"name":"Marcin Juszkiewicz","email":"mjuszkiewicz@redhat.com","username":"hrw"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2019-09-27 17:05:06.000000000","updated_by":{"_account_id":24072,"name":"Marcin Juszkiewicz","email":"mjuszkiewicz@redhat.com","username":"hrw"},"reviewer":{"_account_id":24072,"name":"Marcin Juszkiewicz","email":"mjuszkiewicz@redhat.com","username":"hrw"},"state":"REVIEWER"},{"updated":"2019-09-27 17:09:43.000000000","updated_by":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"reviewer":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"state":"REVIEWER"},{"updated":"2019-09-27 20:42:40.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"}],"messages":[{"id":"2484fd46a08dd5f35ca9d30db6dd0a3043b8a1af","author":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},"date":"2019-09-27 17:03:19.000000000","message":"Patch Set 1: Cherry Picked from branch master.","accounts_in_message":[],"_revision_number":1},{"id":"cef62d8af7424d50f73ab9b109724c57ba72a23b","author":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},"date":"2019-09-27 17:03:55.000000000","message":"Patch Set 1: Review-Priority+2","accounts_in_message":[],"_revision_number":1},{"id":"bd427d048e75e4be7dce05dc0af51935131b2efa","author":{"_account_id":24072,"name":"Marcin Juszkiewicz","email":"mjuszkiewicz@redhat.com","username":"hrw"},"date":"2019-09-27 17:05:06.000000000","message":"Patch Set 1: Code-Review+2","accounts_in_message":[],"_revision_number":1},{"id":"0afc19436b76a002270b6f55635b656a9876dc0c","author":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"date":"2019-09-27 17:09:43.000000000","message":"Patch Set 1: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":1},{"id":"9dedd2565e753d6dfdd1852d16b4f48eadfeb2de","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-09-27 19:27:00.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\n\n- kolla-build-centos-binary https://zuul.opendev.org/t/openstack/build/3f953cab47ab426fb52c5a83ded019cd : SUCCESS in 39m 09s (non-voting)\n- kolla-build-centos-source https://zuul.opendev.org/t/openstack/build/3389ac641be6422286df352524acbf4a : SUCCESS in 56m 27s\n- kolla-ansible-centos-source https://zuul.opendev.org/t/openstack/build/6c2eb347aa37420888e64a44c3bcaad6 : SUCCESS in 1h 21m 28s\n- kolla-ansible-centos-binary https://zuul.opendev.org/t/openstack/build/46e47351fb4047b0a1e3eb41756f65c8 : SUCCESS in 1h 14m 45s (non-voting)\n- kolla-ansible-centos-source-upgrade https://zuul.opendev.org/t/openstack/build/b6b978cf543648939ff48714f26d7655 : SUCCESS in 1h 46m 17s\n- tripleo-build-containers-centos-7 https://zuul.opendev.org/t/openstack/build/309074af1e5148be8a8b198d0367466f : SUCCESS in 1h 05m 49s (non-voting)\n- kolla-build-oraclelinux-binary https://zuul.opendev.org/t/openstack/build/0e442332a46f4a408ee636379fcf242c : SUCCESS in 52m 34s (non-voting)\n- kolla-build-oraclelinux-source https://zuul.opendev.org/t/openstack/build/b88b59d49bec40c78a8d98d151878f71 : SUCCESS in 41m 36s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/7f52381cc54049ff94ec8fe34a0e468d : SUCCESS in 4m 05s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/88723a06fa0641359ec19b565c65e43d : SUCCESS in 2m 57s\n- openstack-tox-py27 https://zuul.opendev.org/t/openstack/build/d94f56434416452c9de8370c61e22b6c : SUCCESS in 3m 38s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/bed436616f7549b88526226c9bbcbc9e : SUCCESS in 3m 28s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/8c3c87aa222845929df73c35887b1b7a : SUCCESS in 4m 01s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/ea0a66b951bb4f1d80a9aef1cfbe6d90 : SUCCESS in 3m 52s\n- kolla-build-ubuntu-binary https://zuul.opendev.org/t/openstack/build/a8478dcb847944db9e8fdd2779fb561d : SUCCESS in 30m 53s (non-voting)\n- kolla-build-ubuntu-source https://zuul.opendev.org/t/openstack/build/5a36f00014fa42848a1a6f1c04beb5ca : SUCCESS in 32m 53s\n- kolla-ansible-ubuntu-source https://zuul.opendev.org/t/openstack/build/25062d99378e468fb63a135d8aa7fa50 : SUCCESS in 49m 51s\n- kolla-ansible-ubuntu-binary https://zuul.opendev.org/t/openstack/build/c6dd809a579e4454939411aa92439e46 : SUCCESS in 52m 40s (non-voting)\n- kolla-ansible-ubuntu-source-upgrade https://zuul.opendev.org/t/openstack/build/3544cb29bd7b45ad884e5952e1155973 : SUCCESS in 1h 08m 21s","accounts_in_message":[],"_revision_number":1},{"id":"2ba27808adaeb05d46fd39684716e06de20a550a","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-09-27 19:27:14.000000000","message":"Patch Set 1: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":1},{"id":"eff166565ffa44c6eb73264ecc928a2ec772ad3e","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-09-27 20:42:40.000000000","message":"Change has been successfully merged by Zuul","accounts_in_message":[],"_revision_number":1},{"id":"b588e96618d0c9203b27f1293a602321f6309efb","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-09-27 20:42:40.000000000","message":"Patch Set 1: Verified+2\n\nBuild succeeded (gate pipeline).\n\n- kolla-build-centos-source https://zuul.opendev.org/t/openstack/build/bdb204523caf43d8bf2d7686831fa767 : SUCCESS in 1h 06m 58s\n- kolla-build-oraclelinux-source https://zuul.opendev.org/t/openstack/build/3a251aab98e74fc2aa53115c100d54a7 : SUCCESS in 56m 52s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/d7783bad568d446b9ecf5ff42c6bec84 : SUCCESS in 5m 02s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/cf43b3628b8b42809d4d9200fad0db17 : SUCCESS in 4m 59s\n- openstack-tox-py27 https://zuul.opendev.org/t/openstack/build/c5d1a5db46f4409f91b30942573cd26a : SUCCESS in 4m 57s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/8f7bbfe06b4e43c2980a8f99cc2b772f : SUCCESS in 4m 06s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/516b3460d0294fdd9b4524ac6c24c982 : SUCCESS in 3m 53s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/0ec8349af6ea4aeea47aa2d46314f4a1 : SUCCESS in 3m 58s\n- kolla-build-ubuntu-source https://zuul.opendev.org/t/openstack/build/a3fc4fc73e9b4b35a253a9ce3a476ea1 : SUCCESS in 28m 04s","accounts_in_message":[],"_revision_number":1},{"id":"766db6013c5f78438fdf94f5b5ab88063816e249","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-09-27 20:43:45.000000000","message":"Patch Set 1:\n\nBuild succeeded (promote pipeline).\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/59f3b3f1a7fd43d18f2f4e5d73f33745 : SUCCESS in 45s","accounts_in_message":[],"_revision_number":1}],"current_revision_number":1,"current_revision":"b847daea68a358652414018297afb44a540b1c79","revisions":{"b847daea68a358652414018297afb44a540b1c79":{"kind":"REWORK","_number":1,"created":"2019-09-27 17:03:19.000000000","uploader":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},"ref":"refs/changes/10/685410/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/kolla","ref":"refs/changes/10/685410/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/kolla refs/changes/10/685410/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/kolla refs/changes/10/685410/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/kolla refs/changes/10/685410/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/kolla refs/changes/10/685410/1"}}},"commit":{"parents":[{"commit":"2761f3b62aaf0988c2c28c72a83f43db1197c854","subject":"Merge \"[horizon] Install octavia dashboard on deb\" into stable/stein","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/2761f3b62aaf0988c2c28c72a83f43db1197c854"}]}],"author":{"name":"Mark Goddard","email":"mark@stackhpc.com","date":"2019-09-27 13:12:55.000000000","tz":60},"committer":{"name":"Mark Goddard","email":"mark@stackhpc.com","date":"2019-09-27 17:03:19.000000000","tz":0},"subject":"Fix monasca-grafana - pin rake to \u003c13","message":"Fix monasca-grafana - pin rake to \u003c13\n\nRake 13.0.0 (released today) depends on Ruby 2.2, but CentOS 7 only\nprovides Ruby 2.0.\n\nThis change pins rake in the monasca-grafana image to versions earlier\nthan 13.0.0.\n\nChange-Id: Id9a76f459aff016fbc920370c0a3ec58e0397fcb\nCloses-Bug: #1845647\n(cherry picked from commit a0712ceef3d6c20cdd79417dcff10525c527fb3d)\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/b847daea68a358652414018297afb44a540b1c79"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/b847daea68a358652414018297afb44a540b1c79"}]},"branch":"refs/heads/stable/stein"}},"requirements":[],"submit_records":[],"submit_requirements":[]}
