)]}'
{"id":"openstack%2Fkolla~685411","triplet_id":"openstack%2Fkolla~stable%2Frocky~Id9a76f459aff016fbc920370c0a3ec58e0397fcb","project":"openstack/kolla","branch":"stable/rocky","topic":"bug/1845647","hashtags":[],"change_id":"Id9a76f459aff016fbc920370c0a3ec58e0397fcb","subject":"Fix monasca-grafana, fluentd and sensu images","status":"MERGED","created":"2019-09-27 17:03:27.000000000","updated":"2019-09-30 15:17:47.000000000","submitted":"2019-09-30 15:17:47.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"685411-1569856667343-953ecec0","meta_rev_id":"c5b88b1f191934064dcf1f2e8b4c6b63f046147f","_number":685411,"virtual_id_number":685411,"owner":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},{"value":0,"_account_id":24072,"name":"Marcin Juszkiewicz","email":"mjuszkiewicz@redhat.com","username":"hrw"},{"value":0,"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},{"value":0,"_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2019-09-30 15:17:47.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":24072,"name":"Marcin Juszkiewicz","email":"mjuszkiewicz@redhat.com","username":"hrw"},"all":[{"value":0,"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},{"value":2,"date":"2019-09-30 14:23:03.000000000","_account_id":24072,"name":"Marcin Juszkiewicz","email":"mjuszkiewicz@redhat.com","username":"hrw"},{"value":0,"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},{"value":0,"_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":24072,"name":"Marcin Juszkiewicz","email":"mjuszkiewicz@redhat.com","username":"hrw"},"all":[{"value":0,"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},{"value":1,"date":"2019-09-30 14:23:03.000000000","_account_id":24072,"name":"Marcin Juszkiewicz","email":"mjuszkiewicz@redhat.com","username":"hrw"},{"value":0,"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},{"value":0,"_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Backport-Candidate":{"all":[{"value":0,"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},{"value":0,"_account_id":24072,"name":"Marcin Juszkiewicz","email":"mjuszkiewicz@redhat.com","username":"hrw"},{"value":0,"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},{"value":0,"_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Do Not Backport"," 0":"Backport Review Needed","+1":"Should Backport"},"description":"","default_value":0,"optional":true},"Review-Priority":{"approved":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},"all":[{"value":0,"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},{"value":0,"_account_id":24072,"name":"Marcin Juszkiewicz","email":"mjuszkiewicz@redhat.com","username":"hrw"},{"value":2,"date":"2019-09-30 12:41:26.000000000","_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},{"value":0,"_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Branch Freeze"," 0":"No Priority","+1":"Important Change","+2":"Gate Blocker Fix / Urgent Change"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},{"_account_id":24072,"name":"Marcin Juszkiewicz","email":"mjuszkiewicz@redhat.com","username":"hrw"},{"_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2019-09-27 19:20:13.000000000","updated_by":{"_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"},"reviewer":{"_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"},"state":"REVIEWER"},{"updated":"2019-09-30 10:26:56.000000000","updated_by":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"reviewer":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"state":"REVIEWER"},{"updated":"2019-09-30 14:23:03.000000000","updated_by":{"_account_id":24072,"name":"Marcin Juszkiewicz","email":"mjuszkiewicz@redhat.com","username":"hrw"},"reviewer":{"_account_id":24072,"name":"Marcin Juszkiewicz","email":"mjuszkiewicz@redhat.com","username":"hrw"},"state":"REVIEWER"},{"updated":"2019-09-30 15:17:47.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"}],"messages":[{"id":"25a906a8bf2de8423ed2c2066c531c174227e2e2","author":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},"date":"2019-09-27 17:03:27.000000000","message":"Patch Set 1: Cherry Picked from branch master.","accounts_in_message":[],"_revision_number":1},{"id":"baf00deada51066de6bf05072465f98d6ae63c1f","author":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},"date":"2019-09-27 17:03:37.000000000","message":"Patch Set 1: Review-Priority+2","accounts_in_message":[],"_revision_number":1},{"id":"3bcfdead3d8496e914201120727c514febe792e9","author":{"_account_id":24072,"name":"Marcin Juszkiewicz","email":"mjuszkiewicz@redhat.com","username":"hrw"},"date":"2019-09-27 17:05:20.000000000","message":"Patch Set 1: Code-Review+2","accounts_in_message":[],"_revision_number":1},{"id":"8869e2afe6a0148249db328baed987baf35f6b30","author":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"date":"2019-09-27 17:10:05.000000000","message":"Patch Set 1: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":1},{"id":"0002414566d751bca9b75279404aae2e33f8d93c","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-09-27 19:17:57.000000000","message":"Patch Set 1: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttp://docs.openstack.org/infra/manual/developers.html#automated-testing\n\n\n- kolla-build-centos-binary https://zuul.opendev.org/t/openstack/build/c1969f0b94e14e59a8d10b771ed8317b : SUCCESS in 40m 42s (non-voting)\n- kolla-build-centos-source https://zuul.opendev.org/t/openstack/build/b5c9de242052448c90bc0ccbacd2d17e : SUCCESS in 59m 07s\n- kolla-ansible-centos-source https://zuul.opendev.org/t/openstack/build/12a5668da5d04f9f8ff77e3ba3ceded1 : SUCCESS in 1h 09m 01s\n- kolla-ansible-centos-binary https://zuul.opendev.org/t/openstack/build/cc551c29b1a24d6285cbb0984e84fbba : SUCCESS in 1h 06m 01s (non-voting)\n- kolla-build-oraclelinux-binary https://zuul.opendev.org/t/openstack/build/2e2dd3a1cbcf4322939e5b18a13e0227 : SUCCESS in 48m 39s (non-voting)\n- kolla-build-oraclelinux-source https://zuul.opendev.org/t/openstack/build/7cc73ff979564566b7bd9fab92c01a3f : SUCCESS in 43m 07s\n- kolla-ansible-oraclelinux-source https://zuul.opendev.org/t/openstack/build/e2656d7c1caf4416a7a539f36a6ef8f3 : SUCCESS in 1h 31m 45s (non-voting)\n- kolla-ansible-oraclelinux-binary https://zuul.opendev.org/t/openstack/build/0602e95d7f684c33b6b69a5629491e92 : SUCCESS in 1h 11m 48s (non-voting)\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/e61f1fbc533846e7852a457c11398e01 : SUCCESS in 3m 20s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/b2906a6d559d48c9914254e2c2f1d9ff : SUCCESS in 4m 46s\n- openstack-tox-py27 https://zuul.opendev.org/t/openstack/build/8042a9d5bb2343eabcbec354f7bbb9a1 : SUCCESS in 3m 27s\n- openstack-tox-py35 https://zuul.opendev.org/t/openstack/build/8f528fb8b4c149f1944ba235ff24875b : SUCCESS in 3m 21s\n- build-openstack-sphinx-docs https://zuul.opendev.org/t/openstack/build/a817362512fb4982892877f2b74ad052 : SUCCESS in 3m 23s\n- kolla-build-ubuntu-binary https://zuul.opendev.org/t/openstack/build/717f8b96c7a545f8bbd483044d4171c1 : FAILURE in 1h 06m 01s (non-voting)\n- kolla-build-ubuntu-source https://zuul.opendev.org/t/openstack/build/e7ae4f93b4c649c383d1fbae13ff1136 : FAILURE in 40m 13s\n- kolla-ansible-ubuntu-source https://zuul.opendev.org/t/openstack/build/f09e033d231040cc9e814c9f4d5bbb13 : FAILURE in 27m 01s\n- kolla-ansible-ubuntu-binary https://zuul.opendev.org/t/openstack/build/96b8068070e741ecafa2387ae9a0d7bd : FAILURE in 58m 16s (non-voting)","accounts_in_message":[],"_revision_number":1},{"id":"b3581cd1497ed50fb072b1f4c2029ddfef58bdb8","author":{"_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"},"date":"2019-09-27 19:20:13.000000000","message":"Patch Set 1:\n\nError installing fluent-plugin-elasticsearch:\nfaraday requires Ruby version \u003e\u003d 2.3.","accounts_in_message":[],"_revision_number":1},{"id":"8998cd64b15a1308e399f936dbe3a559934bb00a","author":{"_account_id":24072,"name":"Marcin Juszkiewicz","email":"mjuszkiewicz@redhat.com","username":"hrw"},"date":"2019-09-28 02:30:09.000000000","message":"Patch Set 1:\n\nwe need to either drop elasticsearch plugin or pin faraday gem to \u003c0.16","accounts_in_message":[],"_revision_number":1},{"id":"9a8e933693eb250a41414d94a18cb8e7a77f679a","author":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},"date":"2019-09-30 09:03:16.000000000","message":"Patch Set 1:\n\n\u003e we need to either drop elasticsearch plugin or pin faraday gem to\n \u003e \u003c0.16\n\nLet\u0027s not drop plugins on stable branches.","accounts_in_message":[],"_revision_number":1},{"id":"ae149363b0a03e4634197774afae0f698c7680bd","author":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},"date":"2019-09-30 09:03:39.000000000","message":"Patch Set 1:\n\n\u003e \u003e we need to either drop elasticsearch plugin or pin faraday gem to\n \u003e \u003e \u003c0.16\n \u003e \n \u003e Let\u0027s not drop plugins on stable branches.\n\nEspecially ones that are useful :)","accounts_in_message":[],"_revision_number":1},{"id":"f2e4bdd41fbee55347a9359a33741eafd8c42b50","author":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},"date":"2019-09-30 10:22:45.000000000","message":"Uploaded patch set 2.","accounts_in_message":[],"_revision_number":2},{"id":"156d3237732e7ee26537a3579a956b8a911f3ac1","author":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"date":"2019-09-30 10:26:56.000000000","message":"Patch Set 2: Code-Review+2","accounts_in_message":[],"_revision_number":2},{"id":"ebf0e638a943256e68a83d683f46194218c80253","author":{"_account_id":24072,"name":"Marcin Juszkiewicz","email":"mjuszkiewicz@redhat.com","username":"hrw"},"date":"2019-09-30 10:38:56.000000000","message":"Patch Set 2: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":2},{"id":"3582a03c30a10a041417b4fd45efd3a3f28f7f93","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-09-30 12:18:33.000000000","message":"Patch Set 2: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttp://docs.openstack.org/infra/manual/developers.html#automated-testing\n\n\n- kolla-build-centos-binary https://zuul.opendev.org/t/openstack/build/bdf8b725d0744fb9b0234bd271ad4a6b : FAILURE in 46m 43s (non-voting)\n- kolla-build-centos-source https://zuul.opendev.org/t/openstack/build/86d600f738624c918c97d9f7285490a7 : FAILURE in 53m 47s\n- kolla-ansible-centos-source https://zuul.opendev.org/t/openstack/build/7c1f5588fea74cca83c024303c0c6f1f : SUCCESS in 1h 30m 32s\n- kolla-ansible-centos-binary https://zuul.opendev.org/t/openstack/build/6e8cc829676c40ebabe66608f54ece43 : SUCCESS in 1h 51m 21s (non-voting)\n- kolla-build-oraclelinux-binary https://zuul.opendev.org/t/openstack/build/7339820885524515bdd74aa7c98cdcd0 : FAILURE in 54m 50s (non-voting)\n- kolla-build-oraclelinux-source https://zuul.opendev.org/t/openstack/build/24b4a566da014623877bd6b324b17452 : FAILURE in 56m 12s\n- kolla-ansible-oraclelinux-source https://zuul.opendev.org/t/openstack/build/789491f3e31346d0a2217a29936f6c24 : SUCCESS in 1h 29m 44s (non-voting)\n- kolla-ansible-oraclelinux-binary https://zuul.opendev.org/t/openstack/build/71a154f2d7ed4c17a0ff3dffe7f42dd0 : SUCCESS in 1h 17m 37s (non-voting)\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/45d127a2e3b94b23a49c08c270d0da0a : SUCCESS in 4m 38s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/2c62ce79377c410383d718a29eb92c46 : SUCCESS in 4m 52s\n- openstack-tox-py27 https://zuul.opendev.org/t/openstack/build/c85a53d6b34644748fda3008e4bf024a : SUCCESS in 4m 32s\n- openstack-tox-py35 https://zuul.opendev.org/t/openstack/build/9b2a4fd858334b6598092b8a4c9175b1 : SUCCESS in 3m 01s\n- build-openstack-sphinx-docs https://zuul.opendev.org/t/openstack/build/870daf18c3ad49adaa15ff636e486ad6 : SUCCESS in 5m 09s\n- kolla-build-ubuntu-binary https://zuul.opendev.org/t/openstack/build/bc694f9152ed4e3491899d251b3f581b : SUCCESS in 57m 45s (non-voting)\n- kolla-build-ubuntu-source https://zuul.opendev.org/t/openstack/build/bdf6c62dfa2446aea024c238e76fca8a : SUCCESS in 33m 04s\n- kolla-ansible-ubuntu-source https://zuul.opendev.org/t/openstack/build/32eeb18097d3432e94a80478ad3122af : SUCCESS in 55m 58s\n- kolla-ansible-ubuntu-binary https://zuul.opendev.org/t/openstack/build/398894f415b3411993906baa4220a238 : SUCCESS in 1h 30m 03s (non-voting)","accounts_in_message":[],"_revision_number":2},{"id":"3b63645172dde85e167ad11e9cc0020ab959f5c1","author":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},"date":"2019-09-30 12:41:26.000000000","message":"Uploaded patch set 3.","accounts_in_message":[],"_revision_number":3},{"id":"fd8ed84a540815a65b240b10f3596565868861dc","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-09-30 14:17:04.000000000","message":"Patch Set 3: Verified+1\n\nBuild succeeded (check pipeline).\n\n- kolla-build-centos-binary https://zuul.opendev.org/t/openstack/build/4b5d407c6693484795fa37959a5fe1d9 : SUCCESS in 39m 50s (non-voting)\n- kolla-build-centos-source https://zuul.opendev.org/t/openstack/build/fd5a612aa29242c3bc6bc98c2a669a50 : SUCCESS in 48m 29s\n- kolla-ansible-centos-source https://zuul.opendev.org/t/openstack/build/891553959dcf4578b8bf062963618b8d : SUCCESS in 1h 18m 22s\n- kolla-ansible-centos-binary https://zuul.opendev.org/t/openstack/build/2ea5b538a8ce4872a7fe8c8a6aecd7f7 : SUCCESS in 1h 16m 25s (non-voting)\n- kolla-build-oraclelinux-binary https://zuul.opendev.org/t/openstack/build/fd994652e0a94a86be7c45bc64c6a167 : SUCCESS in 50m 57s (non-voting)\n- kolla-build-oraclelinux-source https://zuul.opendev.org/t/openstack/build/dec57093edce4e8896399df7b970a619 : SUCCESS in 45m 28s\n- kolla-ansible-oraclelinux-source https://zuul.opendev.org/t/openstack/build/6f675eb6ae4e4803b66dbc71e3cf2036 : SUCCESS in 1h 15m 53s (non-voting)\n- kolla-ansible-oraclelinux-binary https://zuul.opendev.org/t/openstack/build/4bb6c6c549434507b368f9c0a5da6d60 : SUCCESS in 1h 24m 37s (non-voting)\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/c92879a5b2084c74bf483e2109a45ea6 : SUCCESS in 3m 18s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/7333a53274e5472583fdd4993a8e3d4b : SUCCESS in 4m 00s\n- openstack-tox-py27 https://zuul.opendev.org/t/openstack/build/a88039d2627c4b018a8d5aa5ae8d32cf : SUCCESS in 3m 54s\n- openstack-tox-py35 https://zuul.opendev.org/t/openstack/build/9791f52c91da402cb587f7f749c58f99 : SUCCESS in 3m 56s\n- build-openstack-sphinx-docs https://zuul.opendev.org/t/openstack/build/fb11fc67ab5140288313ffe881169606 : SUCCESS in 6m 04s\n- kolla-build-ubuntu-binary https://zuul.opendev.org/t/openstack/build/73fd263ac9f04575b433d8eaa85aeaf9 : SUCCESS in 1h 01m 39s (non-voting)\n- kolla-build-ubuntu-source https://zuul.opendev.org/t/openstack/build/e719443511354957b39f8fa8d41d2d7a : SUCCESS in 40m 34s\n- kolla-ansible-ubuntu-source https://zuul.opendev.org/t/openstack/build/b0a4932689e148c2beff61962f98ff3c : SUCCESS in 49m 04s\n- kolla-ansible-ubuntu-binary https://zuul.opendev.org/t/openstack/build/196257f3d67a456687559ca7890af939 : SUCCESS in 1h 33m 06s (non-voting)","accounts_in_message":[],"_revision_number":3},{"id":"fdaf1d0491efaa6a2473a183618a554f7b8921a0","author":{"_account_id":24072,"name":"Marcin Juszkiewicz","email":"mjuszkiewicz@redhat.com","username":"hrw"},"date":"2019-09-30 14:23:03.000000000","message":"Patch Set 3: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":3},{"id":"dbbebeebcf256fc412c29e72c4d14819499d2538","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-09-30 14:23:21.000000000","message":"Patch Set 3: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":3},{"id":"cce93438428e2f4d34e9f68e459b525d10637044","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-09-30 15:17:47.000000000","message":"Patch Set 3: Verified+2\n\nBuild succeeded (gate pipeline).\n\n- kolla-build-centos-source https://zuul.opendev.org/t/openstack/build/f80b55ff31fd4ca7b53637714525d354 : SUCCESS in 52m 18s\n- kolla-build-oraclelinux-source https://zuul.opendev.org/t/openstack/build/5348773db74d4aa9a4ff263efcb3deba : SUCCESS in 46m 53s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/ecb8ccfbd4e74e7899b4019b4c6ab014 : SUCCESS in 3m 45s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/0d24a6bbaa25401691f8edea1489c9b9 : SUCCESS in 4m 01s\n- openstack-tox-py27 https://zuul.opendev.org/t/openstack/build/e9b79c59234d42268cb38346d4738344 : SUCCESS in 3m 03s\n- openstack-tox-py35 https://zuul.opendev.org/t/openstack/build/221728027f5b4047818c00c188ba5575 : SUCCESS in 4m 40s\n- build-openstack-sphinx-docs https://zuul.opendev.org/t/openstack/build/bccce70a69c34c98bbd6d9d653dca135 : SUCCESS in 4m 43s\n- kolla-build-ubuntu-source https://zuul.opendev.org/t/openstack/build/7b57e66fb57b4c1389956ebbecf57ac9 : SUCCESS in 33m 18s","accounts_in_message":[],"_revision_number":3},{"id":"b97b6d02f6dc0e93546eb3c81ec7d73d3f46482f","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-09-30 15:17:47.000000000","message":"Change has been successfully merged by Zuul","accounts_in_message":[],"_revision_number":3}],"current_revision_number":3,"current_revision":"74fc8ba940c27fd706716161a3cd66e26e34510e","revisions":{"3566be5f331f6adc6417ee8a451a1a2c08506832":{"kind":"REWORK","_number":1,"created":"2019-09-27 17:03:27.000000000","uploader":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},"ref":"refs/changes/11/685411/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/kolla","ref":"refs/changes/11/685411/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/kolla refs/changes/11/685411/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/kolla refs/changes/11/685411/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/kolla refs/changes/11/685411/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/kolla refs/changes/11/685411/1"}}},"commit":{"parents":[{"commit":"e1feb998eeb99202832f68bb25170ce18a6d587a","subject":"Merge \"Fix retries when pushing images\" into stable/rocky","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/e1feb998eeb99202832f68bb25170ce18a6d587a"}]}],"author":{"name":"Mark Goddard","email":"mark@stackhpc.com","date":"2019-09-27 13:12:55.000000000","tz":60},"committer":{"name":"Mark Goddard","email":"mark@stackhpc.com","date":"2019-09-27 17:03:27.000000000","tz":0},"subject":"Fix monasca-grafana - pin rake to \u003c13","message":"Fix monasca-grafana - pin rake to \u003c13\n\nRake 13.0.0 (released today) depends on Ruby 2.2, but CentOS 7 only\nprovides Ruby 2.0.\n\nThis change pins rake in the monasca-grafana image to versions earlier\nthan 13.0.0.\n\nChange-Id: Id9a76f459aff016fbc920370c0a3ec58e0397fcb\nCloses-Bug: #1845647\n(cherry picked from commit a0712ceef3d6c20cdd79417dcff10525c527fb3d)\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/3566be5f331f6adc6417ee8a451a1a2c08506832"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/3566be5f331f6adc6417ee8a451a1a2c08506832"}]},"branch":"refs/heads/stable/rocky"},"db6242e5e800daf179ba40670eb177d72c2b2102":{"kind":"REWORK","_number":2,"created":"2019-09-30 10:22:45.000000000","uploader":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},"ref":"refs/changes/11/685411/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/kolla","ref":"refs/changes/11/685411/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/kolla refs/changes/11/685411/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/kolla refs/changes/11/685411/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/kolla refs/changes/11/685411/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/kolla refs/changes/11/685411/2"}}},"commit":{"parents":[{"commit":"e1feb998eeb99202832f68bb25170ce18a6d587a","subject":"Merge \"Fix retries when pushing images\" into stable/rocky","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/e1feb998eeb99202832f68bb25170ce18a6d587a"}]}],"author":{"name":"Mark Goddard","email":"mark@stackhpc.com","date":"2019-09-27 13:12:55.000000000","tz":60},"committer":{"name":"Mark Goddard","email":"mark@stackhpc.com","date":"2019-09-30 10:22:14.000000000","tz":60},"subject":"Fix monasca-grafana on CentOS and fluentd on Debian/Ubuntu","message":"Fix monasca-grafana on CentOS and fluentd on Debian/Ubuntu\n\n1. monasca-grafana - pin rake to \u003c13\n\nRake 13.0.0 (released today) depends on Ruby 2.2, but CentOS 7 only\nprovides Ruby 2.0.\n\nThis change pins rake in the monasca-grafana image to versions earlier\nthan 13.0.0.\n\n2. fluentd on Ubuntu - pin faraday to \u003c0.16\n\nRocky and earlier only.\n\nFaraday 0.16.x depends on Ruby 2.3+, but td-agent on Ubuntu uses its own\nold version of Ruby.\n\nThis change pins faraday in the Ubuntu and Debian fluentd images to\nversions earlier than 0.16.\n\nChange-Id: Id9a76f459aff016fbc920370c0a3ec58e0397fcb\nCloses-Bug: #1845647\nCloses-Bug: #1845705\n(cherry picked from commit a0712ceef3d6c20cdd79417dcff10525c527fb3d)\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/db6242e5e800daf179ba40670eb177d72c2b2102"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/db6242e5e800daf179ba40670eb177d72c2b2102"}]},"branch":"refs/heads/stable/rocky"},"74fc8ba940c27fd706716161a3cd66e26e34510e":{"kind":"REWORK","_number":3,"created":"2019-09-30 12:41:26.000000000","uploader":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},"ref":"refs/changes/11/685411/3","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/kolla","ref":"refs/changes/11/685411/3","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/kolla refs/changes/11/685411/3 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/kolla refs/changes/11/685411/3 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/kolla refs/changes/11/685411/3 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/kolla refs/changes/11/685411/3"}}},"commit":{"parents":[{"commit":"e1feb998eeb99202832f68bb25170ce18a6d587a","subject":"Merge \"Fix retries when pushing images\" into stable/rocky","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/e1feb998eeb99202832f68bb25170ce18a6d587a"}]}],"author":{"name":"Mark Goddard","email":"mark@stackhpc.com","date":"2019-09-27 13:12:55.000000000","tz":60},"committer":{"name":"Mark Goddard","email":"mark@stackhpc.com","date":"2019-09-30 12:38:10.000000000","tz":60},"subject":"Fix monasca-grafana, fluentd and sensu images","message":"Fix monasca-grafana, fluentd and sensu images\n\n1. monasca-grafana - pin rake to \u003c13\n\nRake 13.0.0 (released today) depends on Ruby 2.2, but CentOS 7 only\nprovides Ruby 2.0.\n\nThis change pins rake in the monasca-grafana image to versions earlier\nthan 13.0.0.\n\n2. fluentd on Ubuntu - pin faraday to \u003c0.16\n\nRocky and earlier only.\n\nFaraday 0.16.x depends on Ruby 2.3+, but td-agent on Ubuntu uses its own\nold version of Ruby.\n\nThis change pins faraday in the Ubuntu and Debian fluentd images to\nversions earlier than 0.16.\n\n3. sensu-client: Pin minitest Ruby gem on CentOS\n\nThis was originally in change I1afd2d50c0dc2de6d305b56bd5f9182745fee843\non master and stable/stein.\n\n(cherry picked from commit 305312f1070e80ba849f8c4feb00ad02196b8306)\n\nChange-Id: Id9a76f459aff016fbc920370c0a3ec58e0397fcb\nCloses-Bug: #1845647\nCloses-Bug: #1845705\nCloses-bug: #1845843\n(cherry picked from commit a0712ceef3d6c20cdd79417dcff10525c527fb3d)\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/74fc8ba940c27fd706716161a3cd66e26e34510e"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/74fc8ba940c27fd706716161a3cd66e26e34510e"}]},"branch":"refs/heads/stable/rocky"}},"requirements":[],"submit_records":[],"submit_requirements":[]}
