)]}'
{"id":"openstack%2Fkolla~685901","triplet_id":"openstack%2Fkolla~master~I88e1c8c1ce78fc138fa2541b15a7620fa7300043","project":"openstack/kolla","branch":"master","hashtags":[],"change_id":"I88e1c8c1ce78fc138fa2541b15a7620fa7300043","subject":"debian: switch to Train binary packages","status":"MERGED","created":"2019-10-01 08:40:21.000000000","updated":"2019-10-02 10:44:24.000000000","submitted":"2019-10-02 10:43:15.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"685901-1570012995324-b42f53c0","meta_rev_id":"8356a47ce3ef876fb1f1922877b74f05e6cb9d2a","_number":685901,"virtual_id_number":685901,"owner":{"_account_id":24072,"name":"Marcin Juszkiewicz","email":"mjuszkiewicz@redhat.com","username":"hrw"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"date":"2019-10-01 09:32:21.000000000","_account_id":24072,"name":"Marcin Juszkiewicz","email":"mjuszkiewicz@redhat.com","username":"hrw"},{"value":0,"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},{"value":0,"_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2019-10-02 10:43:15.000000000","post_submit":true,"permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"date":"2019-10-01 09:25:28.000000000","_account_id":23181,"name":"RDO Third Party CI","email":"dmsimard+rdothirdparty@redhat.com","username":"rdothirdparty","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},"all":[{"value":0,"_account_id":24072,"name":"Marcin Juszkiewicz","email":"mjuszkiewicz@redhat.com","username":"hrw"},{"value":2,"date":"2019-10-02 09:13:29.000000000","_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},{"value":2,"date":"2019-10-01 08:52:37.000000000","_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":23181,"name":"RDO Third Party CI","email":"dmsimard+rdothirdparty@redhat.com","username":"rdothirdparty","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},"all":[{"value":0,"_account_id":24072,"name":"Marcin Juszkiewicz","email":"mjuszkiewicz@redhat.com","username":"hrw"},{"value":1,"date":"2019-10-02 09:13:29.000000000","_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},{"value":0,"_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":23181,"name":"RDO Third Party CI","email":"dmsimard+rdothirdparty@redhat.com","username":"rdothirdparty","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Backport-Candidate":{"all":[{"value":0,"_account_id":24072,"name":"Marcin Juszkiewicz","email":"mjuszkiewicz@redhat.com","username":"hrw"},{"value":0,"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},{"value":0,"_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":23181,"name":"RDO Third Party CI","email":"dmsimard+rdothirdparty@redhat.com","username":"rdothirdparty","tags":["SERVICE_USER"]}],"values":{"-1":"Do Not Backport"," 0":"Backport Review Needed","+1":"Should Backport"},"description":"","default_value":0,"optional":true},"Review-Priority":{"all":[{"value":0,"_account_id":24072,"name":"Marcin Juszkiewicz","email":"mjuszkiewicz@redhat.com","username":"hrw"},{"value":0,"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},{"value":0,"_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":23181,"name":"RDO Third Party CI","email":"dmsimard+rdothirdparty@redhat.com","username":"rdothirdparty","tags":["SERVICE_USER"]}],"values":{"-1":"Branch Freeze"," 0":"No Priority","+1":"Important Change","+2":"Gate Blocker Fix / Urgent Change"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":23181,"name":"RDO Third Party CI","email":"dmsimard+rdothirdparty@redhat.com","username":"rdothirdparty","tags":["SERVICE_USER"]},{"_account_id":24072,"name":"Marcin Juszkiewicz","email":"mjuszkiewicz@redhat.com","username":"hrw"},{"_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2019-10-01 08:52:37.000000000","updated_by":{"_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"},"reviewer":{"_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"},"state":"REVIEWER"},{"updated":"2019-10-01 09:25:28.000000000","updated_by":{"_account_id":23181,"name":"RDO Third Party CI","email":"dmsimard+rdothirdparty@redhat.com","username":"rdothirdparty","tags":["SERVICE_USER"]},"reviewer":{"_account_id":23181,"name":"RDO Third Party CI","email":"dmsimard+rdothirdparty@redhat.com","username":"rdothirdparty","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2019-10-02 09:13:29.000000000","updated_by":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},"reviewer":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},"state":"REVIEWER"},{"updated":"2019-10-02 10:43:15.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"}],"messages":[{"id":"0db2ff0ec34c7a82517ce21f6e4d6a3bbc781561","author":{"_account_id":24072,"name":"Marcin Juszkiewicz","email":"mjuszkiewicz@redhat.com","username":"hrw"},"date":"2019-10-01 08:40:21.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"692227c01ff9cc4a1d76d41bfd972e2b80fd4be1","author":{"_account_id":24072,"name":"Marcin Juszkiewicz","email":"mjuszkiewicz@redhat.com","username":"hrw"},"date":"2019-10-01 08:42:57.000000000","message":"Patch Set 2: Patch Set 1 was rebased","accounts_in_message":[],"_revision_number":2},{"id":"12735e52ad8c2e0806fb69730bcc688e7e66b957","author":{"_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"},"date":"2019-10-01 08:52:37.000000000","message":"Patch Set 2: Code-Review+2\n\nbackport stein to stein too","accounts_in_message":[],"_revision_number":2},{"id":"f4de6b4c97f26235b2d82d4a721f9e89be0deaec","author":{"_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"},"date":"2019-10-01 08:53:10.000000000","message":"Patch Set 2:\n\nunless we don\u0027t want that\n\nmaybe we don\u0027t want that","accounts_in_message":[],"_revision_number":2},{"id":"ad24aafd2b59afe433df3263721baa6a2f4722a4","author":{"_account_id":24072,"name":"Marcin Juszkiewicz","email":"mjuszkiewicz@redhat.com","username":"hrw"},"date":"2019-10-01 09:03:21.000000000","message":"Patch Set 2:\n\n\u003e unless we don\u0027t want that\n \u003e \n \u003e maybe we don\u0027t want that\n\nI have to check which of recent Debian/binary changes we need in Stein.","accounts_in_message":[],"_revision_number":2},{"id":"aba5d59164f278e8871717ef070006e51237c240","author":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},"date":"2019-10-01 09:11:10.000000000","message":"Patch Set 2:\n\n\u003e backport stein to stein too\n\nBackport train packages to Stein?","accounts_in_message":[],"_revision_number":2},{"id":"e135d58b86386fd9fe71dd9db20a81fbc719eb54","author":{"_account_id":23181,"name":"RDO Third Party CI","email":"dmsimard+rdothirdparty@redhat.com","username":"rdothirdparty","tags":["SERVICE_USER"]},"date":"2019-10-01 09:25:28.000000000","message":"Patch Set 2:\n\nBuild succeeded (check pipeline).\n\n- tripleo-build-containers-rhel-8-master http://logs.rdoproject.org/01/685901/2/openstack-check/tripleo-build-containers-rhel-8-master/8ae5d3a/ : SUCCESS in 41m 27s","accounts_in_message":[],"_revision_number":2},{"id":"068c0806ea9ad098030024f017e89dfec4d68214","author":{"_account_id":24072,"name":"Marcin Juszkiewicz","email":"mjuszkiewicz@redhat.com","username":"hrw"},"date":"2019-10-01 09:32:21.000000000","message":"Patch Set 2:\n\ndebian-binary fails on swift-proxy-server which is fixed in separate patch","accounts_in_message":[],"_revision_number":2},{"id":"a6788c6ec7c3fd4c021a60bc217c93ac8fd70231","author":{"_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"},"date":"2019-10-01 09:40:16.000000000","message":"Patch Set 2:\n\n\u003e \u003e backport stein to stein too\n \u003e \n \u003e Backport train packages to Stein?\n\nstein does not yet have stein","accounts_in_message":[],"_revision_number":2},{"id":"e7fb3cf820315bb111e218dafea48113846936a5","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-10-01 11:24:45.000000000","message":"Patch Set 2: Verified+1\n\nBuild succeeded (check pipeline).\n\n- kolla-build-centos-binary https://zuul.opendev.org/t/openstack/build/4871a67e267942a685a96d76780dd3a4 : SUCCESS in 47m 33s (non-voting)\n- kolla-build-centos-source https://zuul.opendev.org/t/openstack/build/733d092d475d4399b782d32d0748b152 : SUCCESS in 59m 24s\n- kolla-ansible-centos-source https://zuul.opendev.org/t/openstack/build/fefbcaa5e94145f181b065daa2ce2b53 : SUCCESS in 1h 26m 56s\n- kolla-ansible-centos-binary https://zuul.opendev.org/t/openstack/build/59515c92b319439b8cd3a220dd5fc275 : SUCCESS in 1h 14m 41s (non-voting)\n- kolla-ansible-centos-source-upgrade https://zuul.opendev.org/t/openstack/build/edc1fb5b44464b80b2c84d63c782c083 : SUCCESS in 1h 38m 27s\n- tripleo-build-containers-centos-7 https://zuul.opendev.org/t/openstack/build/a40561c006d14c9e8b2564eb0eeaa423 : SUCCESS in 55m 18s\n- kolla-ansible-bifrost-centos-source https://zuul.opendev.org/t/openstack/build/12575e60a14e45aabe79797c5bfe069f : SUCCESS in 56m 48s (non-voting)\n- kolla-ansible-centos-source-ceph https://zuul.opendev.org/t/openstack/build/a83a854919414db49dbc8cb73ac92f62 : SUCCESS in 2h 12m 50s (non-voting)\n- kolla-ansible-centos-source-upgrade-ceph https://zuul.opendev.org/t/openstack/build/7884a01225fa47059eb007af29f072d7 : SUCCESS in 2h 26m 28s (non-voting)\n- kolla-ansible-centos-source-cinder-lvm https://zuul.opendev.org/t/openstack/build/3952baabaef6445483944d2676791283 : TIMED_OUT in 2h 03m 40s (non-voting)\n- kolla-ansible-centos-source-scenario-nfv https://zuul.opendev.org/t/openstack/build/4931cf725d9d41a8ae17c933e8a7446c : TIMED_OUT in 2h 03m 44s (non-voting)\n- kolla-ansible-centos-source-ironic https://zuul.opendev.org/t/openstack/build/47bc7729ba134b75bdcf10212113bc56 : SUCCESS in 1h 37m 17s (non-voting)\n- kolla-build-debian-source https://zuul.opendev.org/t/openstack/build/0b3958150f274cab86e40db0282bda3a : SUCCESS in 28m 46s (non-voting)\n- kolla-build-debian-binary https://zuul.opendev.org/t/openstack/build/c2a3c78381b24d399c4f83ed0c405b5c : FAILURE in 21m 14s (non-voting)\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/3263861643c54fa298b2d3c070fa12e7 : SUCCESS in 3m 02s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/c34e54f3a8e64e8483ffe3e6109cc6c0 : SUCCESS in 3m 57s\n- openstack-tox-py27 https://zuul.opendev.org/t/openstack/build/f0b32d5275d5492394c6bbe1db029358 : SUCCESS in 4m 48s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/1feae341d83a4c52b4aee7376d5bb2d2 : SUCCESS in 4m 22s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/3e36b581d0524ce6bfc07c4ca0c04e40 : SUCCESS in 3m 42s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/f9ac56b4f0204daa9d5cdd52dcfd77cb : SUCCESS in 4m 42s\n- kolla-build-ubuntu-binary https://zuul.opendev.org/t/openstack/build/7a5d99b625494a71b43fa33bdf0e7053 : SUCCESS in 30m 09s (non-voting)\n- kolla-build-ubuntu-source https://zuul.opendev.org/t/openstack/build/50200d4d28fe458983c0fcf0d58eba65 : SUCCESS in 33m 19s\n- kolla-ansible-ubuntu-source https://zuul.opendev.org/t/openstack/build/bd28f022239b4ac781cf1721c0eb3a17 : SUCCESS in 53m 14s\n- kolla-ansible-ubuntu-binary https://zuul.opendev.org/t/openstack/build/2b0d5af75aad4dcdb3334992bee4cb74 : SUCCESS in 55m 39s (non-voting)\n- kolla-ansible-ubuntu-source-upgrade https://zuul.opendev.org/t/openstack/build/5a693900e8944da092954085c4546156 : SUCCESS in 55m 07s\n- kolla-ansible-ubuntu-source-zun https://zuul.opendev.org/t/openstack/build/755e217db68a4c768486cfd6c212d55a : SUCCESS in 1h 15m 07s (non-voting)\n- kolla-ansible-ubuntu-source-ceph https://zuul.opendev.org/t/openstack/build/2c90b67a97a8446183ce13fb96198eaa : FAILURE in 1h 21m 53s (non-voting)\n- kolla-ansible-ubuntu-source-upgrade-ceph https://zuul.opendev.org/t/openstack/build/6c6e206803f247888ae7530a8781d658 : RETRY_LIMIT in 9m 27s (non-voting)\n- kolla-ansible-ubuntu-source-cinder-lvm https://zuul.opendev.org/t/openstack/build/7db5755e2cdd4c8d89c97d78fd523d1c : SUCCESS in 1h 17m 57s (non-voting)\n- kolla-ansible-ubuntu-source-ironic https://zuul.opendev.org/t/openstack/build/46989c2baa7a4afca1db2b88f7282fe6 : SUCCESS in 1h 03m 04s (non-voting)","accounts_in_message":[],"_revision_number":2},{"id":"63b0b84855dbefceadd63d9d2e67d9607611125b","author":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},"date":"2019-10-02 09:13:29.000000000","message":"Patch Set 2: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":2},{"id":"87aa61a9d7764e3553c1a9c48eb8d65be1a3701a","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-10-02 09:13:44.000000000","message":"Patch Set 2: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":2},{"id":"25459ceb664be2eef442141cef8220e4c416ad80","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-10-02 10:43:15.000000000","message":"Patch Set 2: Verified+2\n\nBuild succeeded (gate pipeline).\n\n- kolla-build-centos-source https://zuul.opendev.org/t/openstack/build/5a2f5cd722ac45248da4141e63c24039 : SUCCESS in 59m 12s\n- kolla-ansible-centos-source https://zuul.opendev.org/t/openstack/build/3312b5da73724138b990cc0d9a74c242 : SUCCESS in 1h 29m 11s\n- kolla-ansible-centos-source-upgrade https://zuul.opendev.org/t/openstack/build/b40c336408d94b97b9f7acd90e652977 : SUCCESS in 1h 13m 43s\n- tripleo-build-containers-centos-7 https://zuul.opendev.org/t/openstack/build/a82eda2192a64629be1ab3492d04f088 : SUCCESS in 59m 52s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/20dc472f142440c0890659f7f7dd0133 : SUCCESS in 5m 15s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/230a9f0ab99f4592abfe4e677468b265 : SUCCESS in 4m 22s\n- openstack-tox-py27 https://zuul.opendev.org/t/openstack/build/0480b450b8f5488891754a330e196b98 : SUCCESS in 4m 45s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/057c3e5b2a5d49d7a849413eb83eda85 : SUCCESS in 4m 03s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/8ca8216fe91b4af6839c50a995997b70 : SUCCESS in 3m 36s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/897afeaedcbe4c3aa83c5df43009f6bb : SUCCESS in 5m 53s\n- kolla-build-ubuntu-source https://zuul.opendev.org/t/openstack/build/f43233b97c324c189e46158e361d02c5 : SUCCESS in 38m 46s\n- kolla-ansible-ubuntu-source https://zuul.opendev.org/t/openstack/build/52096d6be5e548edba54a2decf950dec : SUCCESS in 1h 02m 07s\n- kolla-ansible-ubuntu-source-upgrade https://zuul.opendev.org/t/openstack/build/4ce6d1c24eae44899a0fbb0ab22785c5 : SUCCESS in 1h 04m 35s","accounts_in_message":[],"_revision_number":2},{"id":"a58f43314e3cd4b67d9bd2f4f8ec288143ca5193","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-10-02 10:43:15.000000000","message":"Change has been successfully merged by Zuul","accounts_in_message":[],"_revision_number":2},{"id":"8356a47ce3ef876fb1f1922877b74f05e6cb9d2a","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-10-02 10:44:24.000000000","message":"Patch Set 2:\n\nBuild succeeded (promote pipeline).\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/4adc36b6f8bf4fa88acf80defabce7f4 : SUCCESS in 49s","accounts_in_message":[],"_revision_number":2}],"current_revision_number":2,"current_revision":"dcd93070dfb30e96bc5c19bdb9b4d999c03d28da","revisions":{"1bb64bd62ee65c1a540495df5a2b309380d663dd":{"kind":"REWORK","_number":1,"created":"2019-10-01 08:40:21.000000000","uploader":{"_account_id":24072,"name":"Marcin Juszkiewicz","email":"mjuszkiewicz@redhat.com","username":"hrw"},"ref":"refs/changes/01/685901/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/kolla","ref":"refs/changes/01/685901/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/kolla refs/changes/01/685901/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/kolla refs/changes/01/685901/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/kolla refs/changes/01/685901/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/kolla refs/changes/01/685901/1"}}},"commit":{"parents":[{"commit":"9b199efd9b25ff52058e5216a1c7ee47a2eae5cb","subject":"Merge \"Horizon: fix for Debian\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/9b199efd9b25ff52058e5216a1c7ee47a2eae5cb"}]}],"author":{"name":"Marcin Juszkiewicz","email":"marcin.juszkiewicz@linaro.org","date":"2019-10-01 08:39:44.000000000","tz":-420},"committer":{"name":"Marcin Juszkiewicz","email":"marcin.juszkiewicz@linaro.org","date":"2019-10-01 08:39:44.000000000","tz":-420},"subject":"debian: switch to Train binary packages","message":"debian: switch to Train binary packages\n\nRepository may not contain all packages yet.\n\nChange-Id: I88e1c8c1ce78fc138fa2541b15a7620fa7300043\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/1bb64bd62ee65c1a540495df5a2b309380d663dd"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/1bb64bd62ee65c1a540495df5a2b309380d663dd"}]},"branch":"refs/heads/master"},"dcd93070dfb30e96bc5c19bdb9b4d999c03d28da":{"kind":"TRIVIAL_REBASE","_number":2,"created":"2019-10-01 08:42:57.000000000","uploader":{"_account_id":24072,"name":"Marcin Juszkiewicz","email":"mjuszkiewicz@redhat.com","username":"hrw"},"ref":"refs/changes/01/685901/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/kolla","ref":"refs/changes/01/685901/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/kolla refs/changes/01/685901/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/kolla refs/changes/01/685901/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/kolla refs/changes/01/685901/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/kolla refs/changes/01/685901/2"}}},"commit":{"parents":[{"commit":"6afefd522b70ad57e25efbc61da94053ebd691fb","subject":"Merge \"CI: add Debian/binary job\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/6afefd522b70ad57e25efbc61da94053ebd691fb"}]}],"author":{"name":"Marcin Juszkiewicz","email":"marcin.juszkiewicz@linaro.org","date":"2019-10-01 08:39:44.000000000","tz":-420},"committer":{"name":"Marcin Juszkiewicz","email":"marcin.juszkiewicz@linaro.org","date":"2019-10-01 08:42:57.000000000","tz":0},"subject":"debian: switch to Train binary packages","message":"debian: switch to Train binary packages\n\nRepository may not contain all packages yet.\n\nChange-Id: I88e1c8c1ce78fc138fa2541b15a7620fa7300043\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/dcd93070dfb30e96bc5c19bdb9b4d999c03d28da"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/dcd93070dfb30e96bc5c19bdb9b4d999c03d28da"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[],"submit_requirements":[]}
