)]}'
{"id":"openstack%2Fkolla~685908","triplet_id":"openstack%2Fkolla~stable%2Fpike~Id9a76f459aff016fbc920370c0a3ec58e0397fcb","project":"openstack/kolla","branch":"stable/pike","topic":"bug/1845705","hashtags":[],"change_id":"Id9a76f459aff016fbc920370c0a3ec58e0397fcb","subject":"Fix fluentd image","status":"MERGED","created":"2019-10-01 09:05:38.000000000","updated":"2019-10-02 16:15:48.000000000","submitted":"2019-10-02 16:15:48.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":1,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"685908-1570032948413-b674a50a","meta_rev_id":"ae4f842c80525dcb25a4f47bcbc80b34080cae86","_number":685908,"virtual_id_number":685908,"owner":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":24072,"name":"Marcin Juszkiewicz","email":"mjuszkiewicz@redhat.com","username":"hrw"},{"value":0,"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},{"value":0,"_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2019-10-02 16:15:48.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":24072,"name":"Marcin Juszkiewicz","email":"mjuszkiewicz@redhat.com","username":"hrw"},"all":[{"value":2,"date":"2019-10-02 08:59:17.000000000","_account_id":24072,"name":"Marcin Juszkiewicz","email":"mjuszkiewicz@redhat.com","username":"hrw"},{"value":0,"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},{"value":0,"_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":24072,"name":"Marcin Juszkiewicz","email":"mjuszkiewicz@redhat.com","username":"hrw"},"all":[{"value":1,"date":"2019-10-02 15:11:31.000000000","_account_id":24072,"name":"Marcin Juszkiewicz","email":"mjuszkiewicz@redhat.com","username":"hrw"},{"value":0,"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},{"value":0,"_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Backport-Candidate":{"all":[{"value":0,"_account_id":24072,"name":"Marcin Juszkiewicz","email":"mjuszkiewicz@redhat.com","username":"hrw"},{"value":0,"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},{"value":0,"_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Do Not Backport"," 0":"Backport Review Needed","+1":"Should Backport"},"description":"","default_value":0,"optional":true},"Review-Priority":{"approved":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},"all":[{"value":0,"_account_id":24072,"name":"Marcin Juszkiewicz","email":"mjuszkiewicz@redhat.com","username":"hrw"},{"value":2,"date":"2019-10-02 08:57:19.000000000","_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},{"value":0,"_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Branch Freeze"," 0":"No Priority","+1":"Important Change","+2":"Gate Blocker Fix / Urgent Change"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":24072,"name":"Marcin Juszkiewicz","email":"mjuszkiewicz@redhat.com","username":"hrw"},{"_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2019-10-01 11:33:19.000000000","updated_by":{"_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"},"reviewer":{"_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"},"state":"REVIEWER"},{"updated":"2019-10-02 15:11:31.000000000","updated_by":{"_account_id":24072,"name":"Marcin Juszkiewicz","email":"mjuszkiewicz@redhat.com","username":"hrw"},"reviewer":{"_account_id":24072,"name":"Marcin Juszkiewicz","email":"mjuszkiewicz@redhat.com","username":"hrw"},"state":"REVIEWER"},{"updated":"2019-10-02 16:15:48.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"}],"messages":[{"id":"008f8f8aa3c1700a9e8ac1d88a9b73cad9e11a69","author":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},"date":"2019-10-01 09:05:38.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"f09515f0268f3182e8fd52b08dc80cee886296b1","author":{"_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"},"date":"2019-10-01 09:09:14.000000000","message":"Patch Set 1: Code-Review+2","accounts_in_message":[],"_revision_number":1},{"id":"226c7f74a9d61719a0fde417c79b4cbadc5e20c8","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-10-01 10:11:56.000000000","message":"Patch Set 1: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttp://docs.openstack.org/infra/manual/developers.html#automated-testing\n\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/dc2878dc565844bcbb5cb29c0d8abee7 : SUCCESS in 3m 49s\n- openstack-tox-py27 https://zuul.opendev.org/t/openstack/build/ce092aeec14047beb19a306c03bf6221 : SUCCESS in 3m 33s\n- openstack-tox-py35 https://zuul.opendev.org/t/openstack/build/a4168295787347f197bf1b4fcee6a2f2 : SUCCESS in 5m 04s\n- build-openstack-sphinx-docs https://zuul.opendev.org/t/openstack/build/6e393ece1f5247358e435e7681df4cdf : SUCCESS in 3m 48s\n- kolla-build-centos-binary https://zuul.opendev.org/t/openstack/build/aa10a1bf855c470ebbd8318c6ace6ad4 : FAILURE in 34m 09s (non-voting)\n- kolla-build-centos-source https://zuul.opendev.org/t/openstack/build/a5be2385e4374d3a9e613a8261c5fc69 : FAILURE in 50m 50s\n- kolla-build-ubuntu-binary https://zuul.opendev.org/t/openstack/build/c465a916796448919c1ac607fb967bac : SUCCESS in 28m 11s (non-voting)\n- kolla-build-ubuntu-source https://zuul.opendev.org/t/openstack/build/7dd4d619a50a48d19e933faed9db8cca : SUCCESS in 32m 45s\n- kolla-build-oraclelinux-binary https://zuul.opendev.org/t/openstack/build/a09296d4f4e3400aac5ec2371dfd3afb : FAILURE in 44m 23s (non-voting)\n- kolla-build-oraclelinux-source https://zuul.opendev.org/t/openstack/build/85652168cba642259c36aaa13bc81854 : FAILURE in 56m 08s\n- kolla-ansible-centos-source https://zuul.opendev.org/t/openstack/build/b32d025fa0ac44ffa848b2ffcca68320 : FAILURE in 40m 39s (non-voting)\n- kolla-ansible-centos-binary https://zuul.opendev.org/t/openstack/build/5c7f7a5d4f1241659a1d630c4f88b0f4 : FAILURE in 36m 19s (non-voting)\n- kolla-ansible-ubuntu-source https://zuul.opendev.org/t/openstack/build/f12ce765849e4ef0bd849fc07c1ec698 : SUCCESS in 47m 42s (non-voting)\n- kolla-ansible-ubuntu-binary https://zuul.opendev.org/t/openstack/build/b1310914fa2349d9bcdc702633ec9075 : SUCCESS in 48m 26s (non-voting)\n- kolla-ansible-oraclelinux-source https://zuul.opendev.org/t/openstack/build/d319e0832a504bb182a9a6843babf75a : FAILURE in 18m 05s (non-voting)\n- kolla-ansible-oraclelinux-binary https://zuul.opendev.org/t/openstack/build/b0081981d4164ef0a1a78a0e147bd9fa : FAILURE in 34m 22s (non-voting)","accounts_in_message":[],"_revision_number":1},{"id":"7e13eaed85f9fcc76614ae0ac91d335e89f098d5","author":{"_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"},"date":"2019-10-01 11:33:19.000000000","message":"Patch Set 1: -Code-Review\n\ncentos needs fixing as well","accounts_in_message":[],"_revision_number":1},{"id":"b824e31e94d90c5f2c596674dd6cf5a5ebb96e1b","author":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},"date":"2019-10-01 13:23:55.000000000","message":"Uploaded patch set 2.","accounts_in_message":[],"_revision_number":2},{"id":"b500b72f5fbdc754715c6ba4a58ffe236a15d33b","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-10-01 14:24:22.000000000","message":"Patch Set 2: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttp://docs.openstack.org/infra/manual/developers.html#automated-testing\n\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/7ce7a73be8054f91906f230ccdebefb2 : SUCCESS in 3m 44s\n- openstack-tox-py27 https://zuul.opendev.org/t/openstack/build/09ed5432946848c5a172bbdb741dd77c : SUCCESS in 3m 09s\n- openstack-tox-py35 https://zuul.opendev.org/t/openstack/build/86770be3c2574b7183768924e7c525ef : SUCCESS in 4m 55s\n- build-openstack-sphinx-docs https://zuul.opendev.org/t/openstack/build/3bb33b7d944e4a2dbc9b833f2e80a0ef : SUCCESS in 5m 04s\n- kolla-build-centos-binary https://zuul.opendev.org/t/openstack/build/73b55d5d302242e4a8efd761ad09a83f : FAILURE in 38m 53s (non-voting)\n- kolla-build-centos-source https://zuul.opendev.org/t/openstack/build/db8adcf1b78a4266a93d6dc4e0c28dfc : FAILURE in 47m 47s\n- kolla-build-ubuntu-binary https://zuul.opendev.org/t/openstack/build/4f9d57e890b24cd59f7b0f0edfe7ee49 : SUCCESS in 31m 16s (non-voting)\n- kolla-build-ubuntu-source https://zuul.opendev.org/t/openstack/build/fcb751e9d81a43d585f6c86fee5684e4 : SUCCESS in 30m 06s\n- kolla-build-oraclelinux-binary https://zuul.opendev.org/t/openstack/build/fd9598397f0240a6917a1bcc61ab22c9 : FAILURE in 45m 38s (non-voting)\n- kolla-build-oraclelinux-source https://zuul.opendev.org/t/openstack/build/0301e36b3831419b9770b31a21563c7d : FAILURE in 47m 02s\n- kolla-ansible-centos-source https://zuul.opendev.org/t/openstack/build/5637ecd206984b538019644adf7923a4 : FAILURE in 42m 49s (non-voting)\n- kolla-ansible-centos-binary https://zuul.opendev.org/t/openstack/build/9198fb2d29b843cd83af949fee191b0a : FAILURE in 40m 37s (non-voting)\n- kolla-ansible-ubuntu-source https://zuul.opendev.org/t/openstack/build/2879cfca23864a458d3438a574e4ee76 : SUCCESS in 46m 02s (non-voting)\n- kolla-ansible-ubuntu-binary https://zuul.opendev.org/t/openstack/build/2dfffcdb886d4aad93aebd5d4cd348c8 : SUCCESS in 50m 03s (non-voting)\n- kolla-ansible-oraclelinux-source https://zuul.opendev.org/t/openstack/build/873ccafa083e42c3997b0459878ff37b : FAILURE in 51m 48s (non-voting)\n- kolla-ansible-oraclelinux-binary https://zuul.opendev.org/t/openstack/build/d9fd9c7017e5477c873045758599129b : FAILURE in 33m 06s (non-voting)","accounts_in_message":[],"_revision_number":2},{"id":"036ba56004ff740c6d97db90e506c97d3738de16","author":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},"date":"2019-10-01 14:31:21.000000000","message":"Uploaded patch set 3.","accounts_in_message":[],"_revision_number":3},{"id":"77e2569ce74ef8ddd3e6ce8b8af1352bc60600de","author":{"_account_id":24072,"name":"Marcin Juszkiewicz","email":"mjuszkiewicz@redhat.com","username":"hrw"},"date":"2019-10-01 14:34:33.000000000","message":"Patch Set 3: Code-Review+2\n\n(1 comment)","accounts_in_message":[],"_revision_number":3},{"id":"39b6423e69fc8569b5b5015b48fd16773728a0e3","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-10-01 16:07:27.000000000","message":"Patch Set 3: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttp://docs.openstack.org/infra/manual/developers.html#automated-testing\n\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/4cbc8c3b1398464da06b7338d0506c2a : SUCCESS in 4m 41s\n- openstack-tox-py27 https://zuul.opendev.org/t/openstack/build/6772d0211b9746e1ae3798a40963d093 : SUCCESS in 3m 34s\n- openstack-tox-py35 https://zuul.opendev.org/t/openstack/build/7e5079e077af4562b15c49d38bc652bf : SUCCESS in 3m 46s\n- build-openstack-sphinx-docs https://zuul.opendev.org/t/openstack/build/58445f91279a4ccb8fd6d5e3d2456fcd : SUCCESS in 4m 13s\n- kolla-build-centos-binary https://zuul.opendev.org/t/openstack/build/3314d5f1b5584881bcb0614797b9a481 : FAILURE in 38m 47s (non-voting)\n- kolla-build-centos-source https://zuul.opendev.org/t/openstack/build/69ea62916d044a649100ee155ec7b568 : FAILURE in 51m 58s\n- kolla-build-ubuntu-binary https://zuul.opendev.org/t/openstack/build/d028286b303342dea8adcc84c58bcb36 : SUCCESS in 30m 24s (non-voting)\n- kolla-build-ubuntu-source https://zuul.opendev.org/t/openstack/build/4e2dba8f2c1b46e8946390744959c866 : SUCCESS in 29m 12s\n- kolla-build-oraclelinux-binary https://zuul.opendev.org/t/openstack/build/b484c2a9b3ff40fca5cccfed9f08f813 : FAILURE in 45m 05s (non-voting)\n- kolla-build-oraclelinux-source https://zuul.opendev.org/t/openstack/build/65791ed7c44a4406b56068a7b84fb581 : FAILURE in 57m 16s\n- kolla-ansible-centos-source https://zuul.opendev.org/t/openstack/build/7e7b3628ccb14a109253d6e755aba6db : FAILURE in 1h 16m 03s (non-voting)\n- kolla-ansible-centos-binary https://zuul.opendev.org/t/openstack/build/5694ba9436e4481f927380bb59dc0df7 : FAILURE in 36m 25s (non-voting)\n- kolla-ansible-ubuntu-source https://zuul.opendev.org/t/openstack/build/a3aaa1c3174a42ff8f5cc7813060704e : SUCCESS in 46m 17s (non-voting)\n- kolla-ansible-ubuntu-binary https://zuul.opendev.org/t/openstack/build/711d96be38634ed6a3515484d1f37bbb : SUCCESS in 44m 44s (non-voting)\n- kolla-ansible-oraclelinux-source https://zuul.opendev.org/t/openstack/build/2968dca833994a89a64c6be8b853c334 : FAILURE in 45m 14s (non-voting)\n- kolla-ansible-oraclelinux-binary https://zuul.opendev.org/t/openstack/build/e1b21ef995c14cdda77b27119648ed69 : FAILURE in 44m 44s (non-voting)","accounts_in_message":[],"_revision_number":3},{"id":"a8bda40ce854334957e6df40e540117e18c75046","author":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},"date":"2019-10-01 16:25:22.000000000","message":"Uploaded patch set 4.","accounts_in_message":[],"_revision_number":4},{"id":"de0c372bf30ca2d9848b4af10b64357e65b5a961","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-10-01 18:02:52.000000000","message":"Patch Set 4: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/ddf1f0139e1142fc89230c653cbf7ad8 : SUCCESS in 4m 53s\n- openstack-tox-py27 https://zuul.opendev.org/t/openstack/build/1aef4f40115848bb99c631a8edb988fe : SUCCESS in 4m 40s\n- openstack-tox-py35 https://zuul.opendev.org/t/openstack/build/1123ab78c2734ece9de066c9e15a79f1 : SUCCESS in 4m 58s\n- build-openstack-sphinx-docs https://zuul.opendev.org/t/openstack/build/e4c3b48cd8fc422e9822e6a5e0f6efa1 : SUCCESS in 6m 11s\n- kolla-build-centos-binary https://zuul.opendev.org/t/openstack/build/eb2ebd52a883457887198c648ccf2e7c : SUCCESS in 41m 42s (non-voting)\n- kolla-build-centos-source https://zuul.opendev.org/t/openstack/build/3a4c455ab0d54c20839f270242bcadef : SUCCESS in 45m 21s\n- kolla-build-ubuntu-binary https://zuul.opendev.org/t/openstack/build/7943bd8483484e83b810567cdbf17b7a : SUCCESS in 31m 11s (non-voting)\n- kolla-build-ubuntu-source https://zuul.opendev.org/t/openstack/build/5603f90f74eb42e9b652323708f3ffe1 : SUCCESS in 37m 48s\n- kolla-build-oraclelinux-binary https://zuul.opendev.org/t/openstack/build/f35cf0f6cef1460aa77d571ad857c097 : SUCCESS in 53m 07s (non-voting)\n- kolla-build-oraclelinux-source https://zuul.opendev.org/t/openstack/build/f8e3b28d9ae941078582d9ae7f9f52fd : SUCCESS in 58m 05s\n- kolla-ansible-centos-source https://zuul.opendev.org/t/openstack/build/a7ec4434a7e548c495f4a7a0c3d4b960 : SUCCESS in 1h 07m 26s (non-voting)\n- kolla-ansible-centos-binary https://zuul.opendev.org/t/openstack/build/046c9b1f1adb4e6a829d119ace52dc0b : SUCCESS in 1h 03m 52s (non-voting)\n- kolla-ansible-ubuntu-source https://zuul.opendev.org/t/openstack/build/cbd9778334224997aeda016e550f4eb9 : SUCCESS in 54m 32s (non-voting)\n- kolla-ansible-ubuntu-binary https://zuul.opendev.org/t/openstack/build/25230a2c6b444f6992499ba52d355f9e : SUCCESS in 48m 57s (non-voting)\n- kolla-ansible-oraclelinux-source https://zuul.opendev.org/t/openstack/build/ac8febf7306b4414abba935eb8b4236c : FAILURE in 21m 21s (non-voting)\n- kolla-ansible-oraclelinux-binary https://zuul.opendev.org/t/openstack/build/01637a2de3284f38b949471e18b848af : SUCCESS in 1h 03m 14s (non-voting)","accounts_in_message":[],"_revision_number":4},{"id":"e4ffc23b8a497f61c58513e5b6674a91c19513f1","author":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},"date":"2019-10-02 08:57:19.000000000","message":"Patch Set 4: Review-Priority+2","accounts_in_message":[],"_revision_number":4},{"id":"51a5614b04b4abf7a65e2ec98970cbc0acc576a8","author":{"_account_id":24072,"name":"Marcin Juszkiewicz","email":"mjuszkiewicz@redhat.com","username":"hrw"},"date":"2019-10-02 08:59:17.000000000","message":"Patch Set 4: Code-Review+2","accounts_in_message":[],"_revision_number":4},{"id":"4e112434492342f116417cb1032793cf20f7be90","author":{"_account_id":24072,"name":"Marcin Juszkiewicz","email":"mjuszkiewicz@redhat.com","username":"hrw"},"date":"2019-10-02 15:11:31.000000000","message":"Patch Set 4: Workflow+1","accounts_in_message":[],"_revision_number":4},{"id":"d09ed07e86d168d99c43b80bfa8d64abff0e7350","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-10-02 15:11:44.000000000","message":"Patch Set 4: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":4},{"id":"0a09743a52be6245e06fd1f11b97e76b89144a2a","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-10-02 16:15:48.000000000","message":"Patch Set 4: Verified+2\n\nBuild succeeded (gate pipeline).\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/fd79b180ea484a1c99ebb5f35c0b0a13 : SUCCESS in 4m 50s\n- openstack-tox-py27 https://zuul.opendev.org/t/openstack/build/0347f6fdcd204d479274cbce142c7df7 : SUCCESS in 3m 26s\n- openstack-tox-py35 https://zuul.opendev.org/t/openstack/build/a734ba92942e46e68969c23069decda3 : SUCCESS in 5m 10s\n- build-openstack-sphinx-docs https://zuul.opendev.org/t/openstack/build/08b37797c6114643813a3fa3c1cdcce6 : SUCCESS in 4m 06s\n- kolla-build-centos-source https://zuul.opendev.org/t/openstack/build/27159cc997d94f349283c9d3f88692ee : SUCCESS in 46m 28s\n- kolla-build-ubuntu-source https://zuul.opendev.org/t/openstack/build/e08d6ba4483c4f799e98d2b2e9927ce1 : SUCCESS in 39m 25s\n- kolla-build-oraclelinux-source https://zuul.opendev.org/t/openstack/build/ed39b933788349328fb6f337ea82c378 : SUCCESS in 56m 53s","accounts_in_message":[],"_revision_number":4},{"id":"9c5e371bb0795a30bea3bf0cfe691a1a8bfe50cd","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-10-02 16:15:48.000000000","message":"Change has been successfully merged by Zuul","accounts_in_message":[],"_revision_number":4}],"current_revision_number":4,"current_revision":"093fc1dda0788223a91c08e65ce8623789cebe8f","revisions":{"e359e1d4872800f9cfb7084ab1f38a577a7ae36c":{"kind":"REWORK","_number":1,"created":"2019-10-01 09:05:38.000000000","uploader":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},"ref":"refs/changes/08/685908/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/kolla","ref":"refs/changes/08/685908/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/kolla refs/changes/08/685908/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/kolla refs/changes/08/685908/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/kolla refs/changes/08/685908/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/kolla refs/changes/08/685908/1"}}},"commit":{"parents":[{"commit":"18507dbc1f2ea3ae31108ba00ae5a3267028c2fb","subject":"base: fix OracleLinux build","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/18507dbc1f2ea3ae31108ba00ae5a3267028c2fb"}]}],"author":{"name":"Mark Goddard","email":"mark@stackhpc.com","date":"2019-09-27 13:12:55.000000000","tz":60},"committer":{"name":"Mark Goddard","email":"mark@stackhpc.com","date":"2019-10-01 09:04:20.000000000","tz":60},"subject":"Fix fluentd image on Ubuntu","message":"Fix fluentd image on Ubuntu\n\nfluentd on Ubuntu - pin faraday to \u003c0.16\n\nRocky and earlier only.\n\nFaraday 0.16.x depends on Ruby 2.3+, but td-agent on Ubuntu uses its own\nold version of Ruby.\n\nThis change pins faraday in the Ubuntu and Debian fluentd images to\nversions earlier than 0.16.\n\nChange-Id: Id9a76f459aff016fbc920370c0a3ec58e0397fcb\nCloses-Bug: #1845705\n(cherry picked from commit a0712ceef3d6c20cdd79417dcff10525c527fb3d)\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/e359e1d4872800f9cfb7084ab1f38a577a7ae36c"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/e359e1d4872800f9cfb7084ab1f38a577a7ae36c"}]},"branch":"refs/heads/stable/pike"},"c70198fadeb4893d1c90ea6f9548ef96e5ba40fa":{"kind":"REWORK","_number":2,"created":"2019-10-01 13:23:55.000000000","uploader":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},"ref":"refs/changes/08/685908/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/kolla","ref":"refs/changes/08/685908/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/kolla refs/changes/08/685908/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/kolla refs/changes/08/685908/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/kolla refs/changes/08/685908/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/kolla refs/changes/08/685908/2"}}},"commit":{"parents":[{"commit":"18507dbc1f2ea3ae31108ba00ae5a3267028c2fb","subject":"base: fix OracleLinux build","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/18507dbc1f2ea3ae31108ba00ae5a3267028c2fb"}]}],"author":{"name":"Mark Goddard","email":"mark@stackhpc.com","date":"2019-09-27 13:12:55.000000000","tz":60},"committer":{"name":"Mark Goddard","email":"mark@stackhpc.com","date":"2019-10-01 13:23:18.000000000","tz":60},"subject":"Fix fluentd image","message":"Fix fluentd image\n\nfluentd - pin faraday to \u003c0.16\n\nRocky and earlier only.\n\nFaraday 0.16.x depends on Ruby 2.3+, but td-agent on Ubuntu uses its own\nold version of Ruby.\n\nThis change pins faraday in the Ubuntu and Debian fluentd images to\nversions earlier than 0.16.\n\nChange-Id: Id9a76f459aff016fbc920370c0a3ec58e0397fcb\nCloses-Bug: #1845705\n(cherry picked from commit a0712ceef3d6c20cdd79417dcff10525c527fb3d)\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/c70198fadeb4893d1c90ea6f9548ef96e5ba40fa"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/c70198fadeb4893d1c90ea6f9548ef96e5ba40fa"}]},"branch":"refs/heads/stable/pike"},"a7d9c85caf60330e91ee401e319a905156e10e2a":{"kind":"REWORK","_number":3,"created":"2019-10-01 14:31:21.000000000","uploader":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},"ref":"refs/changes/08/685908/3","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/kolla","ref":"refs/changes/08/685908/3","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/kolla refs/changes/08/685908/3 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/kolla refs/changes/08/685908/3 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/kolla refs/changes/08/685908/3 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/kolla refs/changes/08/685908/3"}}},"commit":{"parents":[{"commit":"18507dbc1f2ea3ae31108ba00ae5a3267028c2fb","subject":"base: fix OracleLinux build","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/18507dbc1f2ea3ae31108ba00ae5a3267028c2fb"}]}],"author":{"name":"Mark Goddard","email":"mark@stackhpc.com","date":"2019-09-27 13:12:55.000000000","tz":60},"committer":{"name":"Mark Goddard","email":"mark@stackhpc.com","date":"2019-10-01 14:30:58.000000000","tz":60},"subject":"Fix fluentd image","message":"Fix fluentd image\n\nfluentd - pin faraday to \u003c0.16\n\nRocky and earlier only.\n\nFaraday 0.16.x depends on Ruby 2.3+, but td-agent on Ubuntu uses its own\nold version of Ruby.\n\nThis change pins faraday in the Ubuntu and Debian fluentd images to\nversions earlier than 0.16.\n\nChange-Id: Id9a76f459aff016fbc920370c0a3ec58e0397fcb\nCloses-Bug: #1845705\n(cherry picked from commit a0712ceef3d6c20cdd79417dcff10525c527fb3d)\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/a7d9c85caf60330e91ee401e319a905156e10e2a"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/a7d9c85caf60330e91ee401e319a905156e10e2a"}]},"branch":"refs/heads/stable/pike"},"093fc1dda0788223a91c08e65ce8623789cebe8f":{"kind":"REWORK","_number":4,"created":"2019-10-01 16:25:22.000000000","uploader":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},"ref":"refs/changes/08/685908/4","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/kolla","ref":"refs/changes/08/685908/4","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/kolla refs/changes/08/685908/4 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/kolla refs/changes/08/685908/4 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/kolla refs/changes/08/685908/4 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/kolla refs/changes/08/685908/4"}}},"commit":{"parents":[{"commit":"18507dbc1f2ea3ae31108ba00ae5a3267028c2fb","subject":"base: fix OracleLinux build","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/18507dbc1f2ea3ae31108ba00ae5a3267028c2fb"}]}],"author":{"name":"Mark Goddard","email":"mark@stackhpc.com","date":"2019-09-27 13:12:55.000000000","tz":60},"committer":{"name":"Mark Goddard","email":"mark@stackhpc.com","date":"2019-10-01 16:25:04.000000000","tz":60},"subject":"Fix fluentd image","message":"Fix fluentd image\n\nfluentd - pin faraday to \u003c0.16\n\nRocky and earlier only.\n\nFaraday 0.16.x depends on Ruby 2.3+, but td-agent on Ubuntu uses its own\nold version of Ruby.\n\nThis change pins faraday in the Ubuntu and Debian fluentd images to\nversions earlier than 0.16.\n\nChange-Id: Id9a76f459aff016fbc920370c0a3ec58e0397fcb\nCloses-Bug: #1845705\n(cherry picked from commit a0712ceef3d6c20cdd79417dcff10525c527fb3d)\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/093fc1dda0788223a91c08e65ce8623789cebe8f"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/093fc1dda0788223a91c08e65ce8623789cebe8f"}]},"branch":"refs/heads/stable/pike"}},"requirements":[],"submit_records":[],"submit_requirements":[]}
