)]}'
{"id":"openstack%2Fkolla~685923","triplet_id":"openstack%2Fkolla~master~I331d8ed68ab32d7c56e891cf2e71c4a8157a26de","project":"openstack/kolla","branch":"master","topic":"685901","hashtags":[],"change_id":"I331d8ed68ab32d7c56e891cf2e71c4a8157a26de","subject":"debian: enable congress and zaqar images","status":"MERGED","created":"2019-10-01 10:02:39.000000000","updated":"2019-10-02 14:43:44.000000000","submitted":"2019-10-02 14:42:11.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"685923-1570027332083-e03b07bf","meta_rev_id":"b276c251cff59f3c95d8de3588d7eceb3d006d93","_number":685923,"virtual_id_number":685923,"owner":{"_account_id":24072,"name":"Marcin Juszkiewicz","email":"mjuszkiewicz@redhat.com","username":"hrw"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},{"value":0,"_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2019-10-02 14:42:11.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"date":"2019-10-01 10:41:52.000000000","_account_id":23181,"name":"RDO Third Party CI","email":"dmsimard+rdothirdparty@redhat.com","username":"rdothirdparty","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},"all":[{"value":2,"date":"2019-10-02 09:41:24.000000000","_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},{"value":2,"date":"2019-10-02 12:25:45.000000000","_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":23181,"name":"RDO Third Party CI","email":"dmsimard+rdothirdparty@redhat.com","username":"rdothirdparty","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"},"all":[{"value":0,"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},{"value":1,"date":"2019-10-02 12:25:45.000000000","_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":23181,"name":"RDO Third Party CI","email":"dmsimard+rdothirdparty@redhat.com","username":"rdothirdparty","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Backport-Candidate":{"all":[{"value":0,"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},{"value":0,"_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":23181,"name":"RDO Third Party CI","email":"dmsimard+rdothirdparty@redhat.com","username":"rdothirdparty","tags":["SERVICE_USER"]}],"values":{"-1":"Do Not Backport"," 0":"Backport Review Needed","+1":"Should Backport"},"description":"","default_value":0,"optional":true},"Review-Priority":{"all":[{"value":0,"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},{"value":0,"_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":23181,"name":"RDO Third Party CI","email":"dmsimard+rdothirdparty@redhat.com","username":"rdothirdparty","tags":["SERVICE_USER"]}],"values":{"-1":"Branch Freeze"," 0":"No Priority","+1":"Important Change","+2":"Gate Blocker Fix / Urgent Change"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":23181,"name":"RDO Third Party CI","email":"dmsimard+rdothirdparty@redhat.com","username":"rdothirdparty","tags":["SERVICE_USER"]},{"_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2019-10-01 10:41:52.000000000","updated_by":{"_account_id":23181,"name":"RDO Third Party CI","email":"dmsimard+rdothirdparty@redhat.com","username":"rdothirdparty","tags":["SERVICE_USER"]},"reviewer":{"_account_id":23181,"name":"RDO Third Party CI","email":"dmsimard+rdothirdparty@redhat.com","username":"rdothirdparty","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2019-10-02 09:41:24.000000000","updated_by":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},"reviewer":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},"state":"REVIEWER"},{"updated":"2019-10-02 12:25:45.000000000","updated_by":{"_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"},"reviewer":{"_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"},"state":"REVIEWER"},{"updated":"2019-10-02 14:42:11.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"}],"messages":[{"id":"11fa1032f7420f09006e14946be456ae69a9f1e5","author":{"_account_id":24072,"name":"Marcin Juszkiewicz","email":"mjuszkiewicz@redhat.com","username":"hrw"},"date":"2019-10-01 10:02:39.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"f156a595379c9af71c0858ac4758efca170944b2","author":{"_account_id":23181,"name":"RDO Third Party CI","email":"dmsimard+rdothirdparty@redhat.com","username":"rdothirdparty","tags":["SERVICE_USER"]},"date":"2019-10-01 10:41:52.000000000","message":"Patch Set 1:\n\nBuild succeeded (check pipeline).\n\n- tripleo-build-containers-rhel-8-master http://logs.rdoproject.org/23/685923/1/openstack-check/tripleo-build-containers-rhel-8-master/8ff6457/ : SUCCESS in 38m 29s","accounts_in_message":[],"_revision_number":1},{"id":"63f471ac5b94f5667c31c7d4502eb829e4d97360","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-10-01 11:43:42.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\n\n- kolla-build-centos-binary https://zuul.opendev.org/t/openstack/build/03fbe5c8012147ebb2342464863ee249 : SUCCESS in 47m 45s (non-voting)\n- kolla-build-centos-source https://zuul.opendev.org/t/openstack/build/15f6f3c6cdd0463c809814276d73d41f : SUCCESS in 57m 01s\n- kolla-ansible-centos-source https://zuul.opendev.org/t/openstack/build/61ddfaa6b0474d7c93641a78e71ddc19 : SUCCESS in 1h 18m 54s\n- kolla-ansible-centos-binary https://zuul.opendev.org/t/openstack/build/9184784bb008478c930a982189c02fcf : SUCCESS in 1h 14m 58s (non-voting)\n- kolla-ansible-centos-source-upgrade https://zuul.opendev.org/t/openstack/build/6881dd611610412f8c8fece83bb6f77b : SUCCESS in 1h 35m 02s\n- tripleo-build-containers-centos-7 https://zuul.opendev.org/t/openstack/build/e67ccce1e8b044f09458c4714cf49924 : SUCCESS in 59m 06s\n- kolla-build-debian-source https://zuul.opendev.org/t/openstack/build/72d3848aeb764ef8a0d1a337b906f055 : SUCCESS in 31m 19s (non-voting)\n- kolla-build-debian-binary https://zuul.opendev.org/t/openstack/build/d281898f67f1462f991438c3aaac4574 : SUCCESS in 30m 05s (non-voting)\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/b411166fbc3e4ec98260ebfa816dd41c : SUCCESS in 3m 20s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/7ab13e385503482ca80adb703e33c8a2 : SUCCESS in 5m 41s\n- openstack-tox-py27 https://zuul.opendev.org/t/openstack/build/73c760f9f7bd407583a2cbd06edfe571 : SUCCESS in 3m 23s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/5c986a4706f7481981036abb2c90b299 : SUCCESS in 5m 17s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/1651ba9777fc493eb04891bf30354a02 : SUCCESS in 5m 00s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/ab90b03ef6784ca8b76ac901e99f2798 : SUCCESS in 5m 56s\n- kolla-build-ubuntu-binary https://zuul.opendev.org/t/openstack/build/049fa3b47b3c4a338318981a21b2d387 : FAILURE in 23m 34s (non-voting)\n- kolla-build-ubuntu-source https://zuul.opendev.org/t/openstack/build/f65e71cc5f4c4270934c734e6f0c8261 : SUCCESS in 27m 21s\n- kolla-ansible-ubuntu-source https://zuul.opendev.org/t/openstack/build/e1718d09010d477f9ce7a386f247220e : SUCCESS in 50m 42s\n- kolla-ansible-ubuntu-binary https://zuul.opendev.org/t/openstack/build/c6d0be03e917478bbe3a36a73af10549 : SUCCESS in 57m 45s (non-voting)\n- kolla-ansible-ubuntu-source-upgrade https://zuul.opendev.org/t/openstack/build/455f0de89fd94303a28a41dde42f8497 : SUCCESS in 1h 03m 20s","accounts_in_message":[],"_revision_number":1},{"id":"dc0752c1fc9df63b23cb9e369ad8899b34925d81","author":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},"date":"2019-10-02 09:41:24.000000000","message":"Patch Set 1: Code-Review+2","accounts_in_message":[],"_revision_number":1},{"id":"28db786fa6735b6bd37f88b254968fb0421209bb","author":{"_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"},"date":"2019-10-02 12:25:45.000000000","message":"Patch Set 1: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":1},{"id":"9d14c2a9130e8c4622c8e9d04a9893d5c70a91d5","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-10-02 12:25:58.000000000","message":"Patch Set 1: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":1},{"id":"f8d32d2ea9d741c426d80a1886f47af0beefbf50","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-10-02 14:42:11.000000000","message":"Patch Set 1: Verified+2\n\nBuild succeeded (gate pipeline).\n\n- kolla-build-centos-source https://zuul.opendev.org/t/openstack/build/31ffcad0b543426e9983837699a8d6b4 : SUCCESS in 1h 10m 22s\n- kolla-ansible-centos-source https://zuul.opendev.org/t/openstack/build/8be814fdefcb40038485fdcef1bd8ca2 : SUCCESS in 1h 29m 00s\n- kolla-ansible-centos-source-upgrade https://zuul.opendev.org/t/openstack/build/13ac3f87a07f4eeabd6c400406719d5b : SUCCESS in 1h 43m 10s\n- tripleo-build-containers-centos-7 https://zuul.opendev.org/t/openstack/build/8ed52d2319d4497db48d12852bce8cb5 : SUCCESS in 1h 02m 11s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/faa93cf445c449c3850ac7cf55f238ee : SUCCESS in 3m 58s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/519d4abc931148f988e7dba32848f170 : SUCCESS in 5m 04s\n- openstack-tox-py27 https://zuul.opendev.org/t/openstack/build/b35e1a12e20c4029a13c75e2641f8670 : SUCCESS in 2m 52s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/50b1485eca094319987e9c7f53ffe84a : SUCCESS in 3m 02s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/1406f260b99045518c5e6e1db2a2a10c : SUCCESS in 4m 13s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/8b2ec2d9c2ba4a10972326c13bea1eb2 : SUCCESS in 5m 45s\n- kolla-build-ubuntu-source https://zuul.opendev.org/t/openstack/build/8b7ba5b050f94151ba28cc8293ea6669 : SUCCESS in 26m 54s\n- kolla-ansible-ubuntu-source https://zuul.opendev.org/t/openstack/build/47414a5d6ba84b2189182b25bfa2dfb2 : SUCCESS in 59m 16s\n- kolla-ansible-ubuntu-source-upgrade https://zuul.opendev.org/t/openstack/build/02c018b0ed634cf79cf8609b735f44e7 : SUCCESS in 58m 10s","accounts_in_message":[],"_revision_number":1},{"id":"178fe7abdc35c9698a9615c4239c755606c56152","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-10-02 14:42:12.000000000","message":"Change has been successfully merged by Zuul","accounts_in_message":[],"_revision_number":1},{"id":"8f48b344056d8b712fa5c6e36070ab1fb91310bf","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-10-02 14:43:44.000000000","message":"Patch Set 1:\n\nBuild succeeded (promote pipeline).\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/88ac801b386741d88f55bb7e5ceb822a : SUCCESS in 1m 10s","accounts_in_message":[],"_revision_number":1}],"current_revision_number":1,"current_revision":"2ba60953cd95d78d0e207a26ce9a826278325348","revisions":{"2ba60953cd95d78d0e207a26ce9a826278325348":{"kind":"REWORK","_number":1,"created":"2019-10-01 10:02:39.000000000","uploader":{"_account_id":24072,"name":"Marcin Juszkiewicz","email":"mjuszkiewicz@redhat.com","username":"hrw"},"ref":"refs/changes/23/685923/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/kolla","ref":"refs/changes/23/685923/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/kolla refs/changes/23/685923/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/kolla refs/changes/23/685923/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/kolla refs/changes/23/685923/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/kolla refs/changes/23/685923/1"}}},"commit":{"parents":[{"commit":"061f9ce040e72dc496510b6f302eda8e5c639eef","subject":"swift-proxy-server: use Python 3 for Debian/Ubuntu","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/061f9ce040e72dc496510b6f302eda8e5c639eef"}]}],"author":{"name":"Marcin Juszkiewicz","email":"marcin.juszkiewicz@linaro.org","date":"2019-10-01 10:01:49.000000000","tz":-420},"committer":{"name":"Marcin Juszkiewicz","email":"marcin.juszkiewicz@linaro.org","date":"2019-10-01 10:01:49.000000000","tz":-420},"subject":"debian: enable congress and zaqar images","message":"debian: enable congress and zaqar images\n\nChange-Id: I331d8ed68ab32d7c56e891cf2e71c4a8157a26de\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/2ba60953cd95d78d0e207a26ce9a826278325348"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/2ba60953cd95d78d0e207a26ce9a826278325348"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[],"submit_requirements":[]}
