)]}'
{"id":"openstack%2Fkolla~685934","triplet_id":"openstack%2Fkolla~stable%2Fstein~I33d45a5bdd4f45153a78c695e7a9a97f2444dd74","project":"openstack/kolla","branch":"stable/stein","topic":"683680-stable/stein","hashtags":[],"change_id":"I33d45a5bdd4f45153a78c695e7a9a97f2444dd74","subject":"Placement: Debian pkg has migration script in different path","status":"MERGED","created":"2019-10-01 11:06:02.000000000","updated":"2019-10-02 11:22:23.000000000","submitted":"2019-10-02 11:20:17.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"685934-1570015217230-fa124ec6","meta_rev_id":"152ec93d2633a39756056a9bc4b07c2fb63316e0","_number":685934,"virtual_id_number":685934,"owner":{"_account_id":24072,"name":"Marcin Juszkiewicz","email":"mjuszkiewicz@redhat.com","username":"hrw"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":28374,"name":"Viktor Michalek","email":"viktor.michalek@ultimum.io","username":"viktormichalek"},{"value":0,"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},{"value":0,"_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2019-10-02 11:20:17.000000000","post_submit":true,"permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},"all":[{"value":0,"_account_id":28374,"name":"Viktor Michalek","email":"viktor.michalek@ultimum.io","username":"viktormichalek"},{"value":2,"date":"2019-10-02 09:44:18.000000000","_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},{"value":2,"date":"2019-10-01 11:27:22.000000000","_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},"all":[{"value":0,"_account_id":28374,"name":"Viktor Michalek","email":"viktor.michalek@ultimum.io","username":"viktormichalek"},{"value":1,"date":"2019-10-02 09:44:18.000000000","_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},{"value":0,"_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Backport-Candidate":{"all":[{"value":0,"_account_id":28374,"name":"Viktor Michalek","email":"viktor.michalek@ultimum.io","username":"viktormichalek"},{"value":0,"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},{"value":0,"_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Do Not Backport"," 0":"Backport Review Needed","+1":"Should Backport"},"description":"","default_value":0,"optional":true},"Review-Priority":{"all":[{"value":0,"date":"2019-10-01 13:29:30.000000000","_account_id":28374,"name":"Viktor Michalek","email":"viktor.michalek@ultimum.io","username":"viktormichalek"},{"value":0,"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},{"value":0,"_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Branch Freeze"," 0":"No Priority","+1":"Important Change","+2":"Gate Blocker Fix / Urgent Change"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":28374,"name":"Viktor Michalek","email":"viktor.michalek@ultimum.io","username":"viktormichalek"},{"_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2019-10-01 11:27:22.000000000","updated_by":{"_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"},"reviewer":{"_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"},"state":"REVIEWER"},{"updated":"2019-10-01 13:29:30.000000000","updated_by":{"_account_id":28374,"name":"Viktor Michalek","email":"viktor.michalek@ultimum.io","username":"viktormichalek"},"reviewer":{"_account_id":28374,"name":"Viktor Michalek","email":"viktor.michalek@ultimum.io","username":"viktormichalek"},"state":"REVIEWER"},{"updated":"2019-10-02 09:44:18.000000000","updated_by":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},"reviewer":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},"state":"REVIEWER"},{"updated":"2019-10-02 11:20:17.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"}],"messages":[{"id":"a66741cd864a2e483ba16e56eb6c8a9d13b86767","author":{"_account_id":24072,"name":"Marcin Juszkiewicz","email":"mjuszkiewicz@redhat.com","username":"hrw"},"date":"2019-10-01 11:06:02.000000000","message":"Patch Set 1: Cherry Picked from branch master.","accounts_in_message":[],"_revision_number":1},{"id":"2653e5f104ac109273ffbec99356d5ae29016f7e","author":{"_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"},"date":"2019-10-01 11:27:22.000000000","message":"Patch Set 1: Code-Review+2","accounts_in_message":[],"_revision_number":1},{"id":"bb49f75cdff8626ebce0fcf1bcae6fc8ce851f75","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-10-01 12:54:18.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\n\n- kolla-build-centos-binary https://zuul.opendev.org/t/openstack/build/2ab180ebbc3149e2a93f1dbe43755e8b : SUCCESS in 51m 52s (non-voting)\n- kolla-build-centos-source https://zuul.opendev.org/t/openstack/build/c68b9848234f4938868e429a6da58e93 : SUCCESS in 55m 11s\n- kolla-ansible-centos-source https://zuul.opendev.org/t/openstack/build/69f2d6cb9d9042bb9cf11c67ceec1035 : SUCCESS in 1h 19m 44s\n- kolla-ansible-centos-binary https://zuul.opendev.org/t/openstack/build/985557451b0a49c39afa9b7fde1300e7 : SUCCESS in 1h 16m 46s (non-voting)\n- kolla-ansible-centos-source-upgrade https://zuul.opendev.org/t/openstack/build/76b7dd8d76f24bfb9ab1f561596f3e5e : SUCCESS in 1h 44m 55s\n- tripleo-build-containers-centos-7 https://zuul.opendev.org/t/openstack/build/e4d745408e4b4fbab1ec63eeafabe4f8 : SUCCESS in 49m 17s (non-voting)\n- kolla-build-oraclelinux-binary https://zuul.opendev.org/t/openstack/build/4478ee10d627451699880f363330ac8f : SUCCESS in 53m 22s (non-voting)\n- kolla-build-oraclelinux-source https://zuul.opendev.org/t/openstack/build/6f310533fd0a4c298d7a63f75131cad3 : SUCCESS in 49m 45s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/f19f46d14cc1412c818b0542dfbf7b8e : SUCCESS in 3m 24s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/6f849de419cb4a718c6ce6621fbe49af : SUCCESS in 3m 34s\n- openstack-tox-py27 https://zuul.opendev.org/t/openstack/build/e591da971d9b499c9445826c5fc16f98 : SUCCESS in 3m 34s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/9016a2630b0844d8804c0dd5688e5586 : SUCCESS in 5m 10s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/ff2dbf8d47064138b5482363a17d86fb : SUCCESS in 5m 55s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/6db0e83ec20c4ecd940a31664f4f167c : SUCCESS in 3m 07s\n- kolla-build-ubuntu-binary https://zuul.opendev.org/t/openstack/build/7a75ce60158b4589b53504c72f55c40a : SUCCESS in 27m 32s (non-voting)\n- kolla-build-ubuntu-source https://zuul.opendev.org/t/openstack/build/cf24723e8cb447ba904d0be451c6ed7b : SUCCESS in 40m 03s\n- kolla-ansible-ubuntu-source https://zuul.opendev.org/t/openstack/build/28a4c12cd7e3424db7cc60d66dedd583 : SUCCESS in 1h 01m 56s\n- kolla-ansible-ubuntu-binary https://zuul.opendev.org/t/openstack/build/ef87fbf29b60429b9cb7a13cb4f135b1 : SUCCESS in 1h 00m 19s (non-voting)\n- kolla-ansible-ubuntu-source-upgrade https://zuul.opendev.org/t/openstack/build/aaac4939a4b143a69188bef4537384fb : SUCCESS in 59m 57s","accounts_in_message":[],"_revision_number":1},{"id":"4d05baf18366a726e0143322116cd6008425e5b7","author":{"_account_id":28374,"name":"Viktor Michalek","email":"viktor.michalek@ultimum.io","username":"viktormichalek"},"date":"2019-10-01 13:29:30.000000000","message":"Patch Set 1:\n\nCode-Review+2","accounts_in_message":[],"_revision_number":1},{"id":"711c7481dd48e9b24b08d097ea7950291d01cf4b","author":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},"date":"2019-10-02 09:44:18.000000000","message":"Patch Set 1: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":1},{"id":"6bbf39f2b0c8ed953fa92e986c9bb3e9f6971d65","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-10-02 09:44:37.000000000","message":"Patch Set 1: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":1},{"id":"b27380328fa5957fcf78632f8de68756d5aef80b","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-10-02 11:20:17.000000000","message":"Change has been successfully merged by Zuul","accounts_in_message":[],"_revision_number":1},{"id":"a79a0410b01c3b26ceb1c787a815b4548fbdfeec","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-10-02 11:20:17.000000000","message":"Patch Set 1: Verified+2\n\nBuild succeeded (gate pipeline).\n\n- kolla-build-centos-source https://zuul.opendev.org/t/openstack/build/83c11bda5df2485d9b933371fac88f86 : SUCCESS in 41m 46s\n- kolla-build-oraclelinux-source https://zuul.opendev.org/t/openstack/build/184579411a174329af0b5bfbf0c0e5c8 : SUCCESS in 47m 18s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/4bb3bb7af09c4f6082720f01f1d1eb23 : SUCCESS in 4m 13s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/766cd6bd62364d8484e6467e4ad1081b : SUCCESS in 3m 26s\n- openstack-tox-py27 https://zuul.opendev.org/t/openstack/build/6cd97579343145c08c773c2264569f92 : SUCCESS in 3m 12s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/7565be46d0c84596ac8052778d3a0cda : SUCCESS in 3m 15s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/b357e916f7f04d34a039fab2725755f0 : SUCCESS in 3m 04s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/29aa53048cf54dc9a3e863ae2f8caac7 : SUCCESS in 4m 58s\n- kolla-build-ubuntu-source https://zuul.opendev.org/t/openstack/build/f2484eec87d44475814462fd3c155175 : SUCCESS in 33m 22s","accounts_in_message":[],"_revision_number":1},{"id":"7cb11c2c09e701144abb143e69170ebe7bcffac8","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-10-02 11:22:23.000000000","message":"Patch Set 1:\n\nBuild succeeded (promote pipeline).\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/bdabfefa564f4a4f968e8080b1087cf3 : SUCCESS in 58s","accounts_in_message":[],"_revision_number":1}],"current_revision_number":1,"current_revision":"0ef702332ea7502f5db23b58e3b10ab17b25e54a","revisions":{"0ef702332ea7502f5db23b58e3b10ab17b25e54a":{"kind":"REWORK","_number":1,"created":"2019-10-01 11:06:02.000000000","uploader":{"_account_id":24072,"name":"Marcin Juszkiewicz","email":"mjuszkiewicz@redhat.com","username":"hrw"},"ref":"refs/changes/34/685934/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/kolla","ref":"refs/changes/34/685934/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/kolla refs/changes/34/685934/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/kolla refs/changes/34/685934/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/kolla refs/changes/34/685934/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/kolla refs/changes/34/685934/1"}}},"commit":{"parents":[{"commit":"bcac8b82ad228f58ad1e422330ad60701046e5d5","subject":"sensu-client: Pin minitest Ruby gem on CentOS","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/bcac8b82ad228f58ad1e422330ad60701046e5d5"}]}],"author":{"name":"Viktor Michalek","email":"viktor.michalek@ultimum.io","date":"2019-09-20 16:48:42.000000000","tz":120},"committer":{"name":"Marcin Juszkiewicz","email":"marcin.juszkiewicz@linaro.org","date":"2019-10-01 11:06:02.000000000","tz":0},"subject":"Placement: Debian pkg has migration script in different path","message":"Placement: Debian pkg has migration script in different path\n\nChange-Id: I33d45a5bdd4f45153a78c695e7a9a97f2444dd74\n(cherry picked from commit 2d7a14012d2c50bc9fd61fad1e74ca99ad9a98ab)\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/0ef702332ea7502f5db23b58e3b10ab17b25e54a"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/0ef702332ea7502f5db23b58e3b10ab17b25e54a"}]},"branch":"refs/heads/stable/stein"}},"requirements":[],"submit_records":[],"submit_requirements":[]}
