)]}'
{"id":"openstack%2Fkolla~685935","triplet_id":"openstack%2Fkolla~stable%2Fstein~Ie140cf0d6308a5ee3c09f06ddfcf63be2e1cec64","project":"openstack/kolla","branch":"stable/stein","hashtags":[],"change_id":"Ie140cf0d6308a5ee3c09f06ddfcf63be2e1cec64","subject":"Add Debian Buster Stein backports APT repository","status":"MERGED","created":"2019-10-01 11:11:22.000000000","updated":"2019-10-02 11:21:18.000000000","submitted":"2019-10-02 11:20:14.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"685935-1570015214588-953ecec0","meta_rev_id":"ca67e346aa16e82db3cdc9015384e2328b7ed5fb","_number":685935,"virtual_id_number":685935,"owner":{"_account_id":24072,"name":"Marcin Juszkiewicz","email":"mjuszkiewicz@redhat.com","username":"hrw"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":28374,"name":"Viktor Michalek","email":"viktor.michalek@ultimum.io","username":"viktormichalek"},{"value":0,"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},{"value":0,"_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2019-10-02 11:20:14.000000000","post_submit":true,"permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},"recommended":{"_account_id":28374,"name":"Viktor Michalek","email":"viktor.michalek@ultimum.io","username":"viktormichalek"},"all":[{"value":1,"date":"2019-10-01 13:38:52.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":28374,"name":"Viktor Michalek","email":"viktor.michalek@ultimum.io","username":"viktormichalek"},{"value":2,"date":"2019-10-02 09:43:52.000000000","_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},{"value":2,"date":"2019-10-01 11:27:53.000000000","_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","value":1,"default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},"all":[{"value":0,"_account_id":28374,"name":"Viktor Michalek","email":"viktor.michalek@ultimum.io","username":"viktormichalek"},{"value":1,"date":"2019-10-02 09:43:52.000000000","_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},{"value":0,"_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Backport-Candidate":{"all":[{"value":0,"_account_id":28374,"name":"Viktor Michalek","email":"viktor.michalek@ultimum.io","username":"viktormichalek"},{"value":0,"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},{"value":0,"_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Do Not Backport"," 0":"Backport Review Needed","+1":"Should Backport"},"description":"","default_value":0,"optional":true},"Review-Priority":{"all":[{"value":0,"_account_id":28374,"name":"Viktor Michalek","email":"viktor.michalek@ultimum.io","username":"viktormichalek"},{"value":0,"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},{"value":0,"_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Branch Freeze"," 0":"No Priority","+1":"Important Change","+2":"Gate Blocker Fix / Urgent Change"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":28374,"name":"Viktor Michalek","email":"viktor.michalek@ultimum.io","username":"viktormichalek"},{"_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2019-10-01 11:27:53.000000000","updated_by":{"_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"},"reviewer":{"_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"},"state":"REVIEWER"},{"updated":"2019-10-01 13:38:52.000000000","updated_by":{"_account_id":28374,"name":"Viktor Michalek","email":"viktor.michalek@ultimum.io","username":"viktormichalek"},"reviewer":{"_account_id":28374,"name":"Viktor Michalek","email":"viktor.michalek@ultimum.io","username":"viktormichalek"},"state":"REVIEWER"},{"updated":"2019-10-02 09:43:52.000000000","updated_by":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},"reviewer":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},"state":"REVIEWER"},{"updated":"2019-10-02 11:20:14.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"}],"messages":[{"id":"8aa2c1856a3651df5018238f36b20eee4906fff6","author":{"_account_id":24072,"name":"Marcin Juszkiewicz","email":"mjuszkiewicz@redhat.com","username":"hrw"},"date":"2019-10-01 11:11:22.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"b241ba75464b6f052b1b29098927b05eaf9e7ac4","author":{"_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"},"date":"2019-10-01 11:27:53.000000000","message":"Patch Set 1: Code-Review+2","accounts_in_message":[],"_revision_number":1},{"id":"c1e2c243ad55a1256ec8795994614a156f85cac7","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-10-01 12:45:01.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\n\n- kolla-build-centos-binary https://zuul.opendev.org/t/openstack/build/9eeb8244c86c4bb5afa6c607341b0714 : SUCCESS in 43m 10s (non-voting)\n- kolla-build-centos-source https://zuul.opendev.org/t/openstack/build/44632180a4294706afeb70eca1dbbbdc : SUCCESS in 53m 18s\n- kolla-ansible-centos-source https://zuul.opendev.org/t/openstack/build/66d55647b6964667bbd7abf807f2e0c7 : SUCCESS in 1h 25m 18s\n- kolla-ansible-centos-binary https://zuul.opendev.org/t/openstack/build/70c5550d85af4e2da48d76607d0997e6 : SUCCESS in 1h 04m 35s (non-voting)\n- kolla-ansible-centos-source-upgrade https://zuul.opendev.org/t/openstack/build/db34cf8d486848f7aad570f4b07e41f7 : SUCCESS in 1h 31m 01s\n- tripleo-build-containers-centos-7 https://zuul.opendev.org/t/openstack/build/45b267afd6b24c2383a7122214ca23dd : SUCCESS in 1h 02m 55s (non-voting)\n- kolla-build-oraclelinux-binary https://zuul.opendev.org/t/openstack/build/3eb6ef166cf842388a73cc320e1304a5 : SUCCESS in 42m 14s (non-voting)\n- kolla-build-oraclelinux-source https://zuul.opendev.org/t/openstack/build/8b4530984917461fb2f7f4c9d004fb33 : SUCCESS in 52m 05s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/25187b19f3414e20954670f39424ef96 : SUCCESS in 3m 40s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/e2a02318cae446b2ad277c7b79239c14 : SUCCESS in 3m 10s\n- openstack-tox-py27 https://zuul.opendev.org/t/openstack/build/1a263c7d03be46d8b0c195ad01d20a3c : SUCCESS in 3m 05s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/ae053549be004afeacb8ec9932601373 : SUCCESS in 2m 49s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/7ac3f4b31cc3417fab48ca3897dd0323 : SUCCESS in 4m 26s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/158c2d400b854df8892a56c726e2184e : SUCCESS in 3m 28s\n- kolla-build-ubuntu-binary https://zuul.opendev.org/t/openstack/build/a7174535e9434462901533b9049f0e3f : SUCCESS in 29m 42s (non-voting)\n- kolla-build-ubuntu-source https://zuul.opendev.org/t/openstack/build/2c1ec04725f14aa0968fd85f15f7a4ab : SUCCESS in 31m 18s\n- kolla-ansible-ubuntu-source https://zuul.opendev.org/t/openstack/build/4b92f89522e04817af4d423d84780112 : SUCCESS in 1h 01m 04s\n- kolla-ansible-ubuntu-binary https://zuul.opendev.org/t/openstack/build/d7bd897f73e44d10bf26da071226931a : SUCCESS in 49m 50s (non-voting)\n- kolla-ansible-ubuntu-source-upgrade https://zuul.opendev.org/t/openstack/build/96792c55c9e9438cbee3e1120dd2888f : SUCCESS in 1h 05m 46s","accounts_in_message":[],"_revision_number":1},{"id":"893384db689605ca831b571e331f18d0b8e02960","author":{"_account_id":28374,"name":"Viktor Michalek","email":"viktor.michalek@ultimum.io","username":"viktormichalek"},"date":"2019-10-01 13:38:52.000000000","message":"Patch Set 1: Code-Review+1","accounts_in_message":[],"_revision_number":1},{"id":"cb9958c8898b979540ece090fe757736b5d52927","author":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},"date":"2019-10-02 09:43:52.000000000","message":"Patch Set 1: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":1},{"id":"7802173f1fa36966eb85295209bfc5fb7a031a11","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-10-02 09:44:10.000000000","message":"Patch Set 1: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":1},{"id":"bf9da09acbc51d3a9f0cf45b9766850a45224fb9","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-10-02 11:20:14.000000000","message":"Change has been successfully merged by Zuul","accounts_in_message":[],"_revision_number":1},{"id":"da7888f0e06da462eca0af5caa34530527acece3","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-10-02 11:20:14.000000000","message":"Patch Set 1: Verified+2\n\nBuild succeeded (gate pipeline).\n\n- kolla-build-centos-source https://zuul.opendev.org/t/openstack/build/040545ca36464f82a63fea7abf2736ca : SUCCESS in 43m 34s\n- kolla-build-oraclelinux-source https://zuul.opendev.org/t/openstack/build/276b24cbad0c4fc6a729b544bc2fdf66 : SUCCESS in 42m 35s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/2731ab439b144970929c1c60ea40b133 : SUCCESS in 5m 26s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/48e7bba9cc354699a5f69682411900ed : SUCCESS in 3m 19s\n- openstack-tox-py27 https://zuul.opendev.org/t/openstack/build/396aae98be614fe5b64212eac1a547b3 : SUCCESS in 3m 35s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/0bf8cd45375242069aa2dc6c3ab512c1 : SUCCESS in 2m 43s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/3a17848b93ca4566ae45dd1b19e2ca70 : SUCCESS in 3m 23s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/0fc5bf04ef97477fb9954b3e64a2ef9b : SUCCESS in 3m 00s\n- kolla-build-ubuntu-source https://zuul.opendev.org/t/openstack/build/c0a0c129739a446ba4007c31bd38df23 : SUCCESS in 43m 41s","accounts_in_message":[],"_revision_number":1},{"id":"ca67e346aa16e82db3cdc9015384e2328b7ed5fb","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-10-02 11:21:18.000000000","message":"Patch Set 1:\n\nBuild succeeded (promote pipeline).\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/d4c484c259724774acc2b548079e138d : SUCCESS in 47s","accounts_in_message":[],"_revision_number":1}],"current_revision_number":1,"current_revision":"15f74ab82ee7261ef32d92d37c7b9c1ac6ef4e45","revisions":{"15f74ab82ee7261ef32d92d37c7b9c1ac6ef4e45":{"kind":"REWORK","_number":1,"created":"2019-10-01 11:11:22.000000000","uploader":{"_account_id":24072,"name":"Marcin Juszkiewicz","email":"mjuszkiewicz@redhat.com","username":"hrw"},"ref":"refs/changes/35/685935/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/kolla","ref":"refs/changes/35/685935/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/kolla refs/changes/35/685935/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/kolla refs/changes/35/685935/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/kolla refs/changes/35/685935/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/kolla refs/changes/35/685935/1"}}},"commit":{"parents":[{"commit":"bcac8b82ad228f58ad1e422330ad60701046e5d5","subject":"sensu-client: Pin minitest Ruby gem on CentOS","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/bcac8b82ad228f58ad1e422330ad60701046e5d5"}]}],"author":{"name":"Viktor Michalek","email":"viktor.michalek@ultimum.io","date":"2019-09-20 16:20:38.000000000","tz":120},"committer":{"name":"Marcin Juszkiewicz","email":"marcin.juszkiewicz@linaro.org","date":"2019-10-01 10:13:32.000000000","tz":-420},"subject":"Add Debian Buster Stein backports APT repository","message":"Add Debian Buster Stein backports APT repository\n\nChange-Id: Ie140cf0d6308a5ee3c09f06ddfcf63be2e1cec64\n(manual picked from commit c08a4b5a6352f3d1ebf640687d68d00aabc11ff6)\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/15f74ab82ee7261ef32d92d37c7b9c1ac6ef4e45"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/15f74ab82ee7261ef32d92d37c7b9c1ac6ef4e45"}]},"branch":"refs/heads/stable/stein"}},"requirements":[],"submit_records":[],"submit_requirements":[]}
