)]}'
{"id":"openstack%2Fkolla~685936","triplet_id":"openstack%2Fkolla~stable%2Fstein~I0f05183889971dcc014663a849d3fc1d87a19e8e","project":"openstack/kolla","branch":"stable/stein","hashtags":[],"change_id":"I0f05183889971dcc014663a849d3fc1d87a19e8e","subject":"Fluentd: td-agent repository for Debian","status":"MERGED","created":"2019-10-01 11:11:22.000000000","updated":"2019-10-02 11:20:15.000000000","submitted":"2019-10-02 11:20:15.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"685936-1570015215819-e365b183","meta_rev_id":"014600a25b451dfd6c25e3fb4f4f01fcbb9ceb17","_number":685936,"virtual_id_number":685936,"owner":{"_account_id":24072,"name":"Marcin Juszkiewicz","email":"mjuszkiewicz@redhat.com","username":"hrw"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":28374,"name":"Viktor Michalek","email":"viktor.michalek@ultimum.io","username":"viktormichalek"},{"value":0,"date":"2019-10-01 11:12:07.000000000","_account_id":24072,"name":"Marcin Juszkiewicz","email":"mjuszkiewicz@redhat.com","username":"hrw"},{"value":0,"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},{"value":0,"_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2019-10-02 11:20:15.000000000","post_submit":true,"permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},"recommended":{"_account_id":28374,"name":"Viktor Michalek","email":"viktor.michalek@ultimum.io","username":"viktormichalek"},"all":[{"value":1,"date":"2019-10-01 13:39:56.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":28374,"name":"Viktor Michalek","email":"viktor.michalek@ultimum.io","username":"viktormichalek"},{"value":0,"_account_id":24072,"name":"Marcin Juszkiewicz","email":"mjuszkiewicz@redhat.com","username":"hrw"},{"value":2,"date":"2019-10-02 09:43:28.000000000","_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},{"value":2,"date":"2019-10-01 11:29:23.000000000","_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","value":1,"default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},"all":[{"value":0,"_account_id":28374,"name":"Viktor Michalek","email":"viktor.michalek@ultimum.io","username":"viktormichalek"},{"value":0,"_account_id":24072,"name":"Marcin Juszkiewicz","email":"mjuszkiewicz@redhat.com","username":"hrw"},{"value":1,"date":"2019-10-02 09:43:28.000000000","_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},{"value":0,"_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Backport-Candidate":{"all":[{"value":0,"_account_id":28374,"name":"Viktor Michalek","email":"viktor.michalek@ultimum.io","username":"viktormichalek"},{"value":0,"_account_id":24072,"name":"Marcin Juszkiewicz","email":"mjuszkiewicz@redhat.com","username":"hrw"},{"value":0,"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},{"value":0,"_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Do Not Backport"," 0":"Backport Review Needed","+1":"Should Backport"},"description":"","default_value":0,"optional":true},"Review-Priority":{"all":[{"value":0,"_account_id":28374,"name":"Viktor Michalek","email":"viktor.michalek@ultimum.io","username":"viktormichalek"},{"value":0,"_account_id":24072,"name":"Marcin Juszkiewicz","email":"mjuszkiewicz@redhat.com","username":"hrw"},{"value":0,"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},{"value":0,"_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Branch Freeze"," 0":"No Priority","+1":"Important Change","+2":"Gate Blocker Fix / Urgent Change"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":24072,"name":"Marcin Juszkiewicz","email":"mjuszkiewicz@redhat.com","username":"hrw"},{"_account_id":28374,"name":"Viktor Michalek","email":"viktor.michalek@ultimum.io","username":"viktormichalek"},{"_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2019-10-01 11:29:23.000000000","updated_by":{"_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"},"reviewer":{"_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"},"state":"REVIEWER"},{"updated":"2019-10-01 13:39:56.000000000","updated_by":{"_account_id":28374,"name":"Viktor Michalek","email":"viktor.michalek@ultimum.io","username":"viktormichalek"},"reviewer":{"_account_id":28374,"name":"Viktor Michalek","email":"viktor.michalek@ultimum.io","username":"viktormichalek"},"state":"REVIEWER"},{"updated":"2019-10-02 09:43:28.000000000","updated_by":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},"reviewer":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},"state":"REVIEWER"},{"updated":"2019-10-02 11:20:15.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"}],"messages":[{"id":"e9495da873e26b1c563f64a43d9968e224f0e002","author":{"_account_id":24072,"name":"Marcin Juszkiewicz","email":"mjuszkiewicz@redhat.com","username":"hrw"},"date":"2019-10-01 11:11:22.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"691e5477b42bc8a843f6c18987173c261e585b03","author":{"_account_id":24072,"name":"Marcin Juszkiewicz","email":"mjuszkiewicz@redhat.com","username":"hrw"},"date":"2019-10-01 11:12:07.000000000","message":"Patch Set 1:\n\nother fluentd patches are not needed - by default we use \u0027td-agent\u0027 on x86 here","accounts_in_message":[],"_revision_number":1},{"id":"d4f770abcde045fa9cee0a6ebc9154c4eefdd15b","author":{"_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"},"date":"2019-10-01 11:29:23.000000000","message":"Patch Set 1: Code-Review+2\n\nmakes sense","accounts_in_message":[],"_revision_number":1},{"id":"1eac4a39a19304b8031c14d9e9eb35ccd479835b","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-10-01 13:05:31.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\n\n- kolla-build-centos-binary https://zuul.opendev.org/t/openstack/build/0aff1e786b4441dd99ae48680ef6d9ae : SUCCESS in 44m 19s (non-voting)\n- kolla-build-centos-source https://zuul.opendev.org/t/openstack/build/41021c80a6a6484795c98322e98a4a6c : SUCCESS in 49m 35s\n- kolla-ansible-centos-source https://zuul.opendev.org/t/openstack/build/c067b0789b4b41a993aedce30f758f7f : SUCCESS in 1h 16m 43s\n- kolla-ansible-centos-binary https://zuul.opendev.org/t/openstack/build/713d8a46193244c9b3abe52ef6bd3d03 : SUCCESS in 1h 14m 00s (non-voting)\n- kolla-ansible-centos-source-upgrade https://zuul.opendev.org/t/openstack/build/f3626acc4adf48b8b06635cd0d1b624e : SUCCESS in 1h 46m 19s\n- tripleo-build-containers-centos-7 https://zuul.opendev.org/t/openstack/build/4aa8ecc12a9c4c2bbcdcd8d9dc054c1d : SUCCESS in 1h 00m 39s (non-voting)\n- kolla-build-oraclelinux-binary https://zuul.opendev.org/t/openstack/build/de5c498b74ef4160afb88fc60155d608 : SUCCESS in 51m 21s (non-voting)\n- kolla-build-oraclelinux-source https://zuul.opendev.org/t/openstack/build/a7b6bff121c644918cac5466866570f5 : SUCCESS in 59m 44s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/bfb7f0819aca411f9432f941da88c068 : SUCCESS in 3m 49s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/e9539b20a06340e6a109da704bcb8837 : SUCCESS in 3m 48s\n- openstack-tox-py27 https://zuul.opendev.org/t/openstack/build/94b5f036065a4823bd60353a7ee94159 : SUCCESS in 3m 01s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/4b9e7ee3aecb4326a656e2f22d45ca62 : SUCCESS in 4m 28s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/5757ac995a864a2298d8dc424e1998e9 : SUCCESS in 3m 24s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/35b33e21333f4605ba8d6233a1b696be : SUCCESS in 3m 27s\n- kolla-build-ubuntu-binary https://zuul.opendev.org/t/openstack/build/0386ee0b2cbf443ea4420a5b7ae3bc8c : SUCCESS in 23m 53s (non-voting)\n- kolla-build-ubuntu-source https://zuul.opendev.org/t/openstack/build/d77f82517204477998d96bd32bd7b1ee : SUCCESS in 31m 36s\n- kolla-ansible-ubuntu-source https://zuul.opendev.org/t/openstack/build/29f934e4ea3c4b28a5832a9ea13c306c : SUCCESS in 58m 08s\n- kolla-ansible-ubuntu-binary https://zuul.opendev.org/t/openstack/build/8367713c2a58473ca78ef5bfb96a396d : SUCCESS in 51m 37s (non-voting)\n- kolla-ansible-ubuntu-source-upgrade https://zuul.opendev.org/t/openstack/build/46321c8c22f9406293e5b2304d9c59ad : SUCCESS in 1h 05m 51s","accounts_in_message":[],"_revision_number":1},{"id":"3fef14a295ed8cd38250575cbfc7f3de5f90c630","author":{"_account_id":28374,"name":"Viktor Michalek","email":"viktor.michalek@ultimum.io","username":"viktormichalek"},"date":"2019-10-01 13:39:56.000000000","message":"Patch Set 1: Code-Review+1","accounts_in_message":[],"_revision_number":1},{"id":"82cdf76ff8c5eb25c593b530471c79debc8ec4c4","author":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},"date":"2019-10-02 09:43:28.000000000","message":"Patch Set 1: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":1},{"id":"09ab8c2956418bb255cb56173a199c227ca56930","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-10-02 09:44:14.000000000","message":"Patch Set 1: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":1},{"id":"e511c3155db986fc59e91823309ce1eac1a5d3ec","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-10-02 11:20:15.000000000","message":"Patch Set 1: Verified+2\n\nBuild succeeded (gate pipeline).\n\n- kolla-build-centos-source https://zuul.opendev.org/t/openstack/build/97f76e32aeca4de69d233627990f6a60 : SUCCESS in 51m 57s\n- kolla-build-oraclelinux-source https://zuul.opendev.org/t/openstack/build/724bab7fa8e34fcfafbc07fed65a8b20 : SUCCESS in 52m 28s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/728cd14ffda44b6184654af71898e0f9 : SUCCESS in 3m 31s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/b636d8e2c9ba457f812e18e9c8a55e61 : SUCCESS in 4m 13s\n- openstack-tox-py27 https://zuul.opendev.org/t/openstack/build/b5100f5c20ad40f9aa44a6e66bb540e5 : SUCCESS in 3m 46s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/153b9f1ddd5844b5a926f94657066691 : SUCCESS in 4m 25s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/066a76aad0394bebb94198c4087802ce : SUCCESS in 3m 40s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/4ce951cd4338485cbcc2ba0c442fed32 : SUCCESS in 4m 14s\n- kolla-build-ubuntu-source https://zuul.opendev.org/t/openstack/build/ed7cd5089c9b47078c9488a9684342cd : SUCCESS in 30m 32s","accounts_in_message":[],"_revision_number":1},{"id":"2f9501a2e76aa2ba8cfc6e2db7fa12a92db2c5be","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-10-02 11:20:15.000000000","message":"Change has been successfully merged by Zuul","accounts_in_message":[],"_revision_number":1}],"current_revision_number":1,"current_revision":"02bb2b71429ad4b6f82f329e8a7980a3f5695b7c","revisions":{"02bb2b71429ad4b6f82f329e8a7980a3f5695b7c":{"kind":"REWORK","_number":1,"created":"2019-10-01 11:11:22.000000000","uploader":{"_account_id":24072,"name":"Marcin Juszkiewicz","email":"mjuszkiewicz@redhat.com","username":"hrw"},"ref":"refs/changes/36/685936/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/kolla","ref":"refs/changes/36/685936/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/kolla refs/changes/36/685936/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/kolla refs/changes/36/685936/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/kolla refs/changes/36/685936/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/kolla refs/changes/36/685936/1"}}},"commit":{"parents":[{"commit":"15f74ab82ee7261ef32d92d37c7b9c1ac6ef4e45","subject":"Add Debian Buster Stein backports APT repository","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/15f74ab82ee7261ef32d92d37c7b9c1ac6ef4e45"}]}],"author":{"name":"Viktor Michalek","email":"viktor.michalek@ultimum.io","date":"2019-09-23 19:02:59.000000000","tz":120},"committer":{"name":"Marcin Juszkiewicz","email":"marcin.juszkiewicz@linaro.org","date":"2019-10-01 11:10:58.000000000","tz":-420},"subject":"Fluentd: td-agent repository for Debian","message":"Fluentd: td-agent repository for Debian\n\nChange-Id: I0f05183889971dcc014663a849d3fc1d87a19e8e\n(manual picked from commit 2131ef9875efd67a9dd45dc3329df8de16cacff1)\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/02bb2b71429ad4b6f82f329e8a7980a3f5695b7c"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/02bb2b71429ad4b6f82f329e8a7980a3f5695b7c"}]},"branch":"refs/heads/stable/stein"}},"requirements":[],"submit_records":[],"submit_requirements":[]}
