)]}'
{"id":"openstack%2Fkolla~685947","triplet_id":"openstack%2Fkolla~master~I0c5e5cc895238094e30fac49dfc6e7bd880ca55e","project":"openstack/kolla","branch":"master","topic":"enable-mistral","hashtags":[],"change_id":"I0c5e5cc895238094e30fac49dfc6e7bd880ca55e","subject":"mistral-event-engine: enable for Debian/Ubuntu","status":"MERGED","created":"2019-10-01 11:49:56.000000000","updated":"2019-10-02 11:21:42.000000000","submitted":"2019-10-02 11:20:13.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"685947-1570015213072-aecd3d01","meta_rev_id":"c2b567bec845debf215d37e42bd6c1de4bdd71fc","_number":685947,"virtual_id_number":685947,"owner":{"_account_id":24072,"name":"Marcin Juszkiewicz","email":"mjuszkiewicz@redhat.com","username":"hrw"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2019-10-02 11:20:12.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"date":"2019-10-01 12:29:50.000000000","_account_id":23181,"name":"RDO Third Party CI","email":"dmsimard+rdothirdparty@redhat.com","username":"rdothirdparty","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},"all":[{"value":2,"date":"2019-10-02 09:43:02.000000000","_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":23181,"name":"RDO Third Party CI","email":"dmsimard+rdothirdparty@redhat.com","username":"rdothirdparty","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},"all":[{"value":1,"date":"2019-10-02 09:43:02.000000000","_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":23181,"name":"RDO Third Party CI","email":"dmsimard+rdothirdparty@redhat.com","username":"rdothirdparty","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Backport-Candidate":{"all":[{"value":0,"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":23181,"name":"RDO Third Party CI","email":"dmsimard+rdothirdparty@redhat.com","username":"rdothirdparty","tags":["SERVICE_USER"]}],"values":{"-1":"Do Not Backport"," 0":"Backport Review Needed","+1":"Should Backport"},"description":"","default_value":0,"optional":true},"Review-Priority":{"all":[{"value":0,"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":23181,"name":"RDO Third Party CI","email":"dmsimard+rdothirdparty@redhat.com","username":"rdothirdparty","tags":["SERVICE_USER"]}],"values":{"-1":"Branch Freeze"," 0":"No Priority","+1":"Important Change","+2":"Gate Blocker Fix / Urgent Change"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":23181,"name":"RDO Third Party CI","email":"dmsimard+rdothirdparty@redhat.com","username":"rdothirdparty","tags":["SERVICE_USER"]}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2019-10-01 12:29:50.000000000","updated_by":{"_account_id":23181,"name":"RDO Third Party CI","email":"dmsimard+rdothirdparty@redhat.com","username":"rdothirdparty","tags":["SERVICE_USER"]},"reviewer":{"_account_id":23181,"name":"RDO Third Party CI","email":"dmsimard+rdothirdparty@redhat.com","username":"rdothirdparty","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2019-10-02 09:43:02.000000000","updated_by":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},"reviewer":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},"state":"REVIEWER"},{"updated":"2019-10-02 11:20:12.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"}],"messages":[{"id":"9d7bb5413a8d5e1f07c8e710cabad7f42e714c8d","author":{"_account_id":24072,"name":"Marcin Juszkiewicz","email":"mjuszkiewicz@redhat.com","username":"hrw"},"date":"2019-10-01 11:49:56.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"edb6e5b04c4cb0725a2b17e6d80b4e71cd648033","author":{"_account_id":23181,"name":"RDO Third Party CI","email":"dmsimard+rdothirdparty@redhat.com","username":"rdothirdparty","tags":["SERVICE_USER"]},"date":"2019-10-01 12:29:50.000000000","message":"Patch Set 1:\n\nBuild succeeded (check pipeline).\n\n- tripleo-build-containers-rhel-8-master http://logs.rdoproject.org/47/685947/1/openstack-check/tripleo-build-containers-rhel-8-master/fac5991/ : SUCCESS in 39m 02s","accounts_in_message":[],"_revision_number":1},{"id":"44923e97f3b4b810c55d49620cfc3a11f9b72b91","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-10-01 13:57:46.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\n\n- kolla-build-centos-binary https://zuul.opendev.org/t/openstack/build/40c94b0e76114d72a5717a8946c54088 : SUCCESS in 43m 29s (non-voting)\n- kolla-build-centos-source https://zuul.opendev.org/t/openstack/build/e2462067c52f411cb8d04a71afd93df2 : SUCCESS in 58m 48s\n- kolla-ansible-centos-source https://zuul.opendev.org/t/openstack/build/2d06dcdb981d4c61a47d12476fb6c920 : SUCCESS in 1h 22m 37s\n- kolla-ansible-centos-binary https://zuul.opendev.org/t/openstack/build/bd88c07cfe6d4b1193569a6bc0ac638d : SUCCESS in 1h 15m 12s (non-voting)\n- kolla-ansible-centos-source-upgrade https://zuul.opendev.org/t/openstack/build/5ffb6bfd3c664f12bcdd8590a98257fe : SUCCESS in 1h 41m 14s\n- tripleo-build-containers-centos-7 https://zuul.opendev.org/t/openstack/build/3de2b5dc29174cdab2eb18516b1b4c97 : SUCCESS in 1h 00m 29s\n- kolla-ansible-centos-source-scenario-nfv https://zuul.opendev.org/t/openstack/build/2777bfd6089c4836ab320831143a28dd : TIMED_OUT in 2h 03m 44s (non-voting)\n- kolla-build-debian-source https://zuul.opendev.org/t/openstack/build/de696345df5548c1a1ca0569fc81a573 : SUCCESS in 29m 31s (non-voting)\n- kolla-build-debian-binary https://zuul.opendev.org/t/openstack/build/ab6b603f6fbd45879de94ab643751e7d : FAILURE in 27m 58s (non-voting)\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/2953b6a65069477ba40b8fdd2787c9bb : SUCCESS in 4m 55s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/7badc68ff91949b2b36a9eceefeca7c1 : SUCCESS in 4m 55s\n- openstack-tox-py27 https://zuul.opendev.org/t/openstack/build/094853e122564f04a7ce268d894b11a1 : SUCCESS in 3m 31s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/f580442e44ec46c898edf52259f09438 : SUCCESS in 3m 11s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/8c9a822f21054061b5e16ef15aa2b8d7 : SUCCESS in 3m 27s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/218cd5f1c87d496cbca56500a53bd845 : SUCCESS in 6m 00s\n- kolla-build-ubuntu-binary https://zuul.opendev.org/t/openstack/build/2f7005b9ac4f4410be0349eda0381068 : SUCCESS in 29m 05s (non-voting)\n- kolla-build-ubuntu-source https://zuul.opendev.org/t/openstack/build/cc68b712f69348d9bfabc17f89e8ea1b : SUCCESS in 30m 45s\n- kolla-ansible-ubuntu-source https://zuul.opendev.org/t/openstack/build/8c9f491bbf154f63a94f3d36d923f691 : SUCCESS in 57m 45s\n- kolla-ansible-ubuntu-binary https://zuul.opendev.org/t/openstack/build/4d3149082e6f4438a6d56d36ba83476a : SUCCESS in 57m 04s (non-voting)\n- kolla-ansible-ubuntu-source-upgrade https://zuul.opendev.org/t/openstack/build/6ace7100518441e6b61843a85159fc6f : SUCCESS in 1h 12m 32s","accounts_in_message":[],"_revision_number":1},{"id":"f2583038bd3cb643f04e9c6cb37e72ba763926bc","author":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},"date":"2019-10-02 09:43:02.000000000","message":"Patch Set 1: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":1},{"id":"f1cb3f5b45b4ec66f9c800a05e9e5b67c417ca5a","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-10-02 09:43:16.000000000","message":"Patch Set 1: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":1},{"id":"6b67830a04203f099f95a9671ddc4cc79852b9e5","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-10-02 11:20:12.000000000","message":"Patch Set 1: Verified+2\n\nBuild succeeded (gate pipeline).\n\n- kolla-build-centos-source https://zuul.opendev.org/t/openstack/build/878e6ec5be8d469a9882afdfeb5fa8a3 : SUCCESS in 58m 26s\n- kolla-ansible-centos-source https://zuul.opendev.org/t/openstack/build/1ef51469d4194751804ab693557832aa : SUCCESS in 1h 32m 24s\n- kolla-ansible-centos-source-upgrade https://zuul.opendev.org/t/openstack/build/7e987340710f4e268457746728cdc00f : SUCCESS in 1h 30m 55s\n- tripleo-build-containers-centos-7 https://zuul.opendev.org/t/openstack/build/1fe85963f1364200ab5877285b6222ad : SUCCESS in 1h 01m 20s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/24126a5c695c4678b9cfcfc995998c3c : SUCCESS in 3m 43s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/ff443c2b14d34b619a119ac72f795b2e : SUCCESS in 3m 58s\n- openstack-tox-py27 https://zuul.opendev.org/t/openstack/build/b0663f646945448d80728394b645c431 : SUCCESS in 3m 18s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/e48ef07bdef74eeb963b0620ae79a6d2 : SUCCESS in 3m 17s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/0a070cd3dced4adfb8f9a9086ba93e5a : SUCCESS in 3m 33s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/b25d6b34bae54e2c8c5f30f7e1be2ce6 : SUCCESS in 5m 16s\n- kolla-build-ubuntu-source https://zuul.opendev.org/t/openstack/build/3fcaa8e04fb6406ab1c613a6f2c144ac : SUCCESS in 32m 36s\n- kolla-ansible-ubuntu-source https://zuul.opendev.org/t/openstack/build/387a4699073f4df8a6692eaee1cc9b36 : SUCCESS in 53m 30s\n- kolla-ansible-ubuntu-source-upgrade https://zuul.opendev.org/t/openstack/build/b7575e6b4ec34b1e91118379cdcb46f4 : SUCCESS in 1h 01m 52s","accounts_in_message":[],"_revision_number":1},{"id":"b3d00ee2cbbd52624d364b0f8693ddd7e33b78dc","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-10-02 11:20:13.000000000","message":"Change has been successfully merged by Zuul","accounts_in_message":[],"_revision_number":1},{"id":"87f1f6ae6337dd36948322378338c24470538863","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-10-02 11:21:42.000000000","message":"Patch Set 1:\n\nBuild succeeded (promote pipeline).\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/517a5ee831bd4500a00b4d93ec96b8b8 : SUCCESS in 1m 09s","accounts_in_message":[],"_revision_number":1}],"current_revision_number":1,"current_revision":"d7d6ebdd3462cb2ffe4b0199508a174ce0cc2171","revisions":{"d7d6ebdd3462cb2ffe4b0199508a174ce0cc2171":{"kind":"REWORK","_number":1,"created":"2019-10-01 11:49:56.000000000","uploader":{"_account_id":24072,"name":"Marcin Juszkiewicz","email":"mjuszkiewicz@redhat.com","username":"hrw"},"ref":"refs/changes/47/685947/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/kolla","ref":"refs/changes/47/685947/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/kolla refs/changes/47/685947/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/kolla refs/changes/47/685947/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/kolla refs/changes/47/685947/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/kolla refs/changes/47/685947/1"}}},"commit":{"parents":[{"commit":"a9181bf89529756e4d58b22324988b4e57770ec1","subject":"Merge \"Update mention of Kayobe in documentation\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/a9181bf89529756e4d58b22324988b4e57770ec1"}]}],"author":{"name":"Marcin Juszkiewicz","email":"marcin.juszkiewicz@linaro.org","date":"2019-10-01 11:48:26.000000000","tz":-420},"committer":{"name":"Marcin Juszkiewicz","email":"marcin.juszkiewicz@linaro.org","date":"2019-10-01 11:49:46.000000000","tz":-420},"subject":"mistral-event-engine: enable for Debian/Ubuntu","message":"mistral-event-engine: enable for Debian/Ubuntu\n\nDepends-On: https://review.opendev.org/685901\nChange-Id: I0c5e5cc895238094e30fac49dfc6e7bd880ca55e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/d7d6ebdd3462cb2ffe4b0199508a174ce0cc2171"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/d7d6ebdd3462cb2ffe4b0199508a174ce0cc2171"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[],"submit_requirements":[]}
