)]}'
{"id":"openstack%2Fkolla~685952","triplet_id":"openstack%2Fkolla~master~Iea8fd1ac30400c67f76aec6d37ee7614f9c51770","project":"openstack/kolla","branch":"master","topic":"pyngus","hashtags":[],"change_id":"Iea8fd1ac30400c67f76aec6d37ee7614f9c51770","subject":"Revert \"Adding python*-pyngus package to openstack base\"","status":"MERGED","created":"2019-10-01 12:23:01.000000000","updated":"2019-10-04 14:03:36.000000000","submitted":"2019-10-04 14:01:53.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"685952-1570197713783-244dc94e","meta_rev_id":"823277cc4c0d123c1f8200aba5ec01211cb65a37","_number":685952,"virtual_id_number":685952,"owner":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":8449,"name":"Marios Andreou","email":"marios.andreou@gmail.com","username":"marios"},{"value":0,"_account_id":24162,"name":"Sorin Sbârnea","display_name":"zbr","email":"ssbarnea@redhat.com","username":"ssbarnea","status":"do not feed the troll 🥕"},{"value":0,"_account_id":24072,"name":"Marcin Juszkiewicz","email":"mjuszkiewicz@redhat.com","username":"hrw"},{"value":0,"_account_id":14985,"name":"Alex Schultz","email":"aschultz@next-development.com","username":"mwhahaha"},{"value":0,"_account_id":12393,"name":"chandan kumar","display_name":"Chandan Kumar","email":"chkumar@redhat.com","username":"chkumar246"},{"value":0,"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2019-10-04 14:01:53.000000000","post_submit":true,"permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"date":"2019-10-04 09:59:15.000000000","_account_id":23181,"name":"RDO Third Party CI","email":"dmsimard+rdothirdparty@redhat.com","username":"rdothirdparty","tags":["SERVICE_USER"]},{"value":0,"_account_id":8367,"name":"Arx Cruz","email":"arxcruz@redhat.com","username":"arxcruz"},{"value":0,"_account_id":8175,"name":"Rafael Folco","email":"rfolco@redhat.com","username":"rfolco"},{"value":0,"_account_id":3153,"name":"Emilien Macchi","email":"emilien@redhat.com","username":"emilienm"},{"value":0,"_account_id":9976,"name":"Ronelle Landy","email":"rlandy@redhat.com","username":"rlandy"},{"value":0,"_account_id":9592,"name":"wes hayutin","email":"whayutin@redhat.com","username":"weshayutin"},{"value":0,"_account_id":10969,"name":"Shnaidman Sagi (Sergey)","display_name":"Shnaidman Sagi","email":"sshnaidm@redhat.com","username":"sergsh"},{"value":0,"_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":24072,"name":"Marcin Juszkiewicz","email":"mjuszkiewicz@redhat.com","username":"hrw"},"all":[{"value":0,"_account_id":8449,"name":"Marios Andreou","email":"marios.andreou@gmail.com","username":"marios"},{"value":0,"_account_id":24162,"name":"Sorin Sbârnea","display_name":"zbr","email":"ssbarnea@redhat.com","username":"ssbarnea","status":"do not feed the troll 🥕"},{"value":2,"date":"2019-10-04 11:06:38.000000000","_account_id":24072,"name":"Marcin Juszkiewicz","email":"mjuszkiewicz@redhat.com","username":"hrw"},{"value":0,"_account_id":14985,"name":"Alex Schultz","email":"aschultz@next-development.com","username":"mwhahaha"},{"value":0,"_account_id":12393,"name":"chandan kumar","display_name":"Chandan Kumar","email":"chkumar@redhat.com","username":"chkumar246"},{"value":0,"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":23181,"name":"RDO Third Party CI","email":"dmsimard+rdothirdparty@redhat.com","username":"rdothirdparty","tags":["SERVICE_USER"]},{"value":1,"date":"2019-10-04 09:32:47.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":8367,"name":"Arx Cruz","email":"arxcruz@redhat.com","username":"arxcruz"},{"value":0,"_account_id":8175,"name":"Rafael Folco","email":"rfolco@redhat.com","username":"rfolco"},{"value":0,"_account_id":3153,"name":"Emilien Macchi","email":"emilien@redhat.com","username":"emilienm"},{"value":0,"_account_id":9976,"name":"Ronelle Landy","email":"rlandy@redhat.com","username":"rlandy"},{"value":0,"_account_id":9592,"name":"wes hayutin","email":"whayutin@redhat.com","username":"weshayutin"},{"value":0,"_account_id":10969,"name":"Shnaidman Sagi (Sergey)","display_name":"Shnaidman Sagi","email":"sshnaidm@redhat.com","username":"sergsh"},{"value":2,"date":"2019-10-04 09:19:58.000000000","_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":24072,"name":"Marcin Juszkiewicz","email":"mjuszkiewicz@redhat.com","username":"hrw"},"all":[{"value":0,"_account_id":8449,"name":"Marios Andreou","email":"marios.andreou@gmail.com","username":"marios"},{"value":0,"_account_id":24162,"name":"Sorin Sbârnea","display_name":"zbr","email":"ssbarnea@redhat.com","username":"ssbarnea","status":"do not feed the troll 🥕"},{"value":1,"date":"2019-10-04 11:06:38.000000000","_account_id":24072,"name":"Marcin Juszkiewicz","email":"mjuszkiewicz@redhat.com","username":"hrw"},{"value":0,"_account_id":14985,"name":"Alex Schultz","email":"aschultz@next-development.com","username":"mwhahaha"},{"value":0,"_account_id":12393,"name":"chandan kumar","display_name":"Chandan Kumar","email":"chkumar@redhat.com","username":"chkumar246"},{"value":0,"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":23181,"name":"RDO Third Party CI","email":"dmsimard+rdothirdparty@redhat.com","username":"rdothirdparty","tags":["SERVICE_USER"]},{"value":0,"_account_id":8367,"name":"Arx Cruz","email":"arxcruz@redhat.com","username":"arxcruz"},{"value":0,"_account_id":8175,"name":"Rafael Folco","email":"rfolco@redhat.com","username":"rfolco"},{"value":0,"_account_id":3153,"name":"Emilien Macchi","email":"emilien@redhat.com","username":"emilienm"},{"value":0,"_account_id":9976,"name":"Ronelle Landy","email":"rlandy@redhat.com","username":"rlandy"},{"value":0,"_account_id":9592,"name":"wes hayutin","email":"whayutin@redhat.com","username":"weshayutin"},{"value":0,"_account_id":10969,"name":"Shnaidman Sagi (Sergey)","display_name":"Shnaidman Sagi","email":"sshnaidm@redhat.com","username":"sergsh"},{"value":0,"_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Backport-Candidate":{"all":[{"value":0,"_account_id":8449,"name":"Marios Andreou","email":"marios.andreou@gmail.com","username":"marios"},{"value":0,"_account_id":24162,"name":"Sorin Sbârnea","display_name":"zbr","email":"ssbarnea@redhat.com","username":"ssbarnea","status":"do not feed the troll 🥕"},{"value":0,"_account_id":24072,"name":"Marcin Juszkiewicz","email":"mjuszkiewicz@redhat.com","username":"hrw"},{"value":0,"_account_id":14985,"name":"Alex Schultz","email":"aschultz@next-development.com","username":"mwhahaha"},{"value":0,"_account_id":12393,"name":"chandan kumar","display_name":"Chandan Kumar","email":"chkumar@redhat.com","username":"chkumar246"},{"value":0,"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":23181,"name":"RDO Third Party CI","email":"dmsimard+rdothirdparty@redhat.com","username":"rdothirdparty","tags":["SERVICE_USER"]},{"value":0,"_account_id":8367,"name":"Arx Cruz","email":"arxcruz@redhat.com","username":"arxcruz"},{"value":0,"_account_id":8175,"name":"Rafael Folco","email":"rfolco@redhat.com","username":"rfolco"},{"value":0,"_account_id":3153,"name":"Emilien Macchi","email":"emilien@redhat.com","username":"emilienm"},{"value":0,"_account_id":9976,"name":"Ronelle Landy","email":"rlandy@redhat.com","username":"rlandy"},{"value":0,"_account_id":9592,"name":"wes hayutin","email":"whayutin@redhat.com","username":"weshayutin"},{"value":0,"_account_id":10969,"name":"Shnaidman Sagi (Sergey)","display_name":"Shnaidman Sagi","email":"sshnaidm@redhat.com","username":"sergsh"},{"value":0,"_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"}],"values":{"-1":"Do Not Backport"," 0":"Backport Review Needed","+1":"Should Backport"},"description":"","default_value":0,"optional":true},"Review-Priority":{"all":[{"value":0,"_account_id":8449,"name":"Marios Andreou","email":"marios.andreou@gmail.com","username":"marios"},{"value":0,"_account_id":24162,"name":"Sorin Sbârnea","display_name":"zbr","email":"ssbarnea@redhat.com","username":"ssbarnea","status":"do not feed the troll 🥕"},{"value":0,"_account_id":24072,"name":"Marcin Juszkiewicz","email":"mjuszkiewicz@redhat.com","username":"hrw"},{"value":0,"_account_id":14985,"name":"Alex Schultz","email":"aschultz@next-development.com","username":"mwhahaha"},{"value":0,"_account_id":12393,"name":"chandan kumar","display_name":"Chandan Kumar","email":"chkumar@redhat.com","username":"chkumar246"},{"value":0,"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":23181,"name":"RDO Third Party CI","email":"dmsimard+rdothirdparty@redhat.com","username":"rdothirdparty","tags":["SERVICE_USER"]},{"value":0,"_account_id":8367,"name":"Arx Cruz","email":"arxcruz@redhat.com","username":"arxcruz"},{"value":0,"_account_id":8175,"name":"Rafael Folco","email":"rfolco@redhat.com","username":"rfolco"},{"value":0,"_account_id":3153,"name":"Emilien Macchi","email":"emilien@redhat.com","username":"emilienm"},{"value":0,"_account_id":9976,"name":"Ronelle Landy","email":"rlandy@redhat.com","username":"rlandy"},{"value":0,"_account_id":9592,"name":"wes hayutin","email":"whayutin@redhat.com","username":"weshayutin"},{"value":0,"_account_id":10969,"name":"Shnaidman Sagi (Sergey)","display_name":"Shnaidman Sagi","email":"sshnaidm@redhat.com","username":"sergsh"},{"value":0,"_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"}],"values":{"-1":"Branch Freeze"," 0":"No Priority","+1":"Important Change","+2":"Gate Blocker Fix / Urgent Change"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":3153,"name":"Emilien Macchi","email":"emilien@redhat.com","username":"emilienm"},{"_account_id":8175,"name":"Rafael Folco","email":"rfolco@redhat.com","username":"rfolco"},{"_account_id":8367,"name":"Arx Cruz","email":"arxcruz@redhat.com","username":"arxcruz"},{"_account_id":8449,"name":"Marios Andreou","email":"marios.andreou@gmail.com","username":"marios"},{"_account_id":9592,"name":"wes hayutin","email":"whayutin@redhat.com","username":"weshayutin"},{"_account_id":9976,"name":"Ronelle Landy","email":"rlandy@redhat.com","username":"rlandy"},{"_account_id":10969,"name":"Shnaidman Sagi (Sergey)","display_name":"Shnaidman Sagi","email":"sshnaidm@redhat.com","username":"sergsh"},{"_account_id":12393,"name":"chandan kumar","display_name":"Chandan Kumar","email":"chkumar@redhat.com","username":"chkumar246"},{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},{"_account_id":14985,"name":"Alex Schultz","email":"aschultz@next-development.com","username":"mwhahaha"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":23181,"name":"RDO Third Party CI","email":"dmsimard+rdothirdparty@redhat.com","username":"rdothirdparty","tags":["SERVICE_USER"]},{"_account_id":24072,"name":"Marcin Juszkiewicz","email":"mjuszkiewicz@redhat.com","username":"hrw"},{"_account_id":24162,"name":"Sorin Sbârnea","display_name":"zbr","email":"ssbarnea@redhat.com","username":"ssbarnea","status":"do not feed the troll 🥕"},{"_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2019-10-01 12:23:01.000000000","updated_by":{"_account_id":3153,"name":"Emilien Macchi","email":"emilien@redhat.com","username":"emilienm"},"reviewer":{"_account_id":3153,"name":"Emilien Macchi","email":"emilien@redhat.com","username":"emilienm"},"state":"REVIEWER"},{"updated":"2019-10-01 12:23:01.000000000","updated_by":{"_account_id":8175,"name":"Rafael Folco","email":"rfolco@redhat.com","username":"rfolco"},"reviewer":{"_account_id":8175,"name":"Rafael Folco","email":"rfolco@redhat.com","username":"rfolco"},"state":"REVIEWER"},{"updated":"2019-10-01 12:23:01.000000000","updated_by":{"_account_id":8449,"name":"Marios Andreou","email":"marios.andreou@gmail.com","username":"marios"},"reviewer":{"_account_id":8449,"name":"Marios Andreou","email":"marios.andreou@gmail.com","username":"marios"},"state":"REVIEWER"},{"updated":"2019-10-01 12:23:01.000000000","updated_by":{"_account_id":9592,"name":"wes hayutin","email":"whayutin@redhat.com","username":"weshayutin"},"reviewer":{"_account_id":9592,"name":"wes hayutin","email":"whayutin@redhat.com","username":"weshayutin"},"state":"REVIEWER"},{"updated":"2019-10-01 12:23:01.000000000","updated_by":{"_account_id":9976,"name":"Ronelle Landy","email":"rlandy@redhat.com","username":"rlandy"},"reviewer":{"_account_id":9976,"name":"Ronelle Landy","email":"rlandy@redhat.com","username":"rlandy"},"state":"REVIEWER"},{"updated":"2019-10-01 12:23:01.000000000","updated_by":{"_account_id":10969,"name":"Shnaidman Sagi (Sergey)","display_name":"Shnaidman Sagi","email":"sshnaidm@redhat.com","username":"sergsh"},"reviewer":{"_account_id":10969,"name":"Shnaidman Sagi (Sergey)","display_name":"Shnaidman Sagi","email":"sshnaidm@redhat.com","username":"sergsh"},"state":"REVIEWER"},{"updated":"2019-10-01 12:23:01.000000000","updated_by":{"_account_id":12393,"name":"chandan kumar","display_name":"Chandan Kumar","email":"chkumar@redhat.com","username":"chkumar246"},"reviewer":{"_account_id":12393,"name":"chandan kumar","display_name":"Chandan Kumar","email":"chkumar@redhat.com","username":"chkumar246"},"state":"REVIEWER"},{"updated":"2019-10-01 12:23:01.000000000","updated_by":{"_account_id":14985,"name":"Alex Schultz","email":"aschultz@next-development.com","username":"mwhahaha"},"reviewer":{"_account_id":14985,"name":"Alex Schultz","email":"aschultz@next-development.com","username":"mwhahaha"},"state":"REVIEWER"},{"updated":"2019-10-01 12:23:01.000000000","updated_by":{"_account_id":24162,"name":"Sorin Sbârnea","display_name":"zbr","email":"ssbarnea@redhat.com","username":"ssbarnea","status":"do not feed the troll 🥕"},"reviewer":{"_account_id":24162,"name":"Sorin Sbârnea","display_name":"zbr","email":"ssbarnea@redhat.com","username":"ssbarnea","status":"do not feed the troll 🥕"},"state":"REVIEWER"},{"updated":"2019-10-04 09:19:58.000000000","updated_by":{"_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"},"reviewer":{"_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"},"state":"REVIEWER"},{"updated":"2019-10-04 09:32:47.000000000","updated_by":{"_account_id":8367,"name":"Arx Cruz","email":"arxcruz@redhat.com","username":"arxcruz"},"reviewer":{"_account_id":8367,"name":"Arx Cruz","email":"arxcruz@redhat.com","username":"arxcruz"},"state":"REVIEWER"},{"updated":"2019-10-04 09:59:15.000000000","updated_by":{"_account_id":23181,"name":"RDO Third Party CI","email":"dmsimard+rdothirdparty@redhat.com","username":"rdothirdparty","tags":["SERVICE_USER"]},"reviewer":{"_account_id":23181,"name":"RDO Third Party CI","email":"dmsimard+rdothirdparty@redhat.com","username":"rdothirdparty","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2019-10-04 11:06:38.000000000","updated_by":{"_account_id":24072,"name":"Marcin Juszkiewicz","email":"mjuszkiewicz@redhat.com","username":"hrw"},"reviewer":{"_account_id":24072,"name":"Marcin Juszkiewicz","email":"mjuszkiewicz@redhat.com","username":"hrw"},"state":"REVIEWER"},{"updated":"2019-10-04 14:01:53.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"}],"messages":[{"id":"d53be3a4515716ac4b5ef172f132fa60c668230b","author":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},"date":"2019-10-01 12:23:01.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"77afc6e7103796d1cdc6e70dba626d9cbf54880a","author":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},"date":"2019-10-01 12:23:10.000000000","message":"Patch Set 1: Code-Review+2","accounts_in_message":[],"_revision_number":1},{"id":"daf132d8e412845175745e198377413a0993bde7","author":{"_account_id":23181,"name":"RDO Third Party CI","email":"dmsimard+rdothirdparty@redhat.com","username":"rdothirdparty","tags":["SERVICE_USER"]},"date":"2019-10-01 13:05:15.000000000","message":"Patch Set 1:\n\nBuild succeeded (check pipeline).\n\n- tripleo-build-containers-rhel-8-master http://logs.rdoproject.org/52/685952/1/openstack-check/tripleo-build-containers-rhel-8-master/9593233/ : SUCCESS in 40m 58s","accounts_in_message":[],"_revision_number":1},{"id":"423cbb8eea805ca032200f3899b3665bd7508b74","author":{"_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"},"date":"2019-10-01 13:06:33.000000000","message":"Patch Set 1:\n\nwhen ubuntu/debian started using rdo?","accounts_in_message":[],"_revision_number":1},{"id":"fd849564d270235ee0efdfe834eeb552061e1aac","author":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},"date":"2019-10-01 13:28:31.000000000","message":"Patch Set 1: -Code-Review\n\n\u003e when ubuntu/debian started using rdo?\n\nGood question. Would need to check if they have this dependency pulled in automatically.","accounts_in_message":[],"_revision_number":1},{"id":"30de78a4299cc2f89510e2ad8a669a5910f10932","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-10-01 15:03:46.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\n\n- kolla-build-centos-binary https://zuul.opendev.org/t/openstack/build/1d40ea5820684f01897e509b89ef67c4 : SUCCESS in 50m 28s (non-voting)\n- kolla-build-centos-source https://zuul.opendev.org/t/openstack/build/6904ce6971fe411c9a1e123d94dd555e : SUCCESS in 50m 05s\n- kolla-ansible-centos-source https://zuul.opendev.org/t/openstack/build/e807a80cb30f4f3d858ff874102e415a : SUCCESS in 1h 21m 58s\n- kolla-ansible-centos-binary https://zuul.opendev.org/t/openstack/build/d4e0c4b88c9b40c083cca616541f60a2 : SUCCESS in 1h 20m 52s (non-voting)\n- kolla-ansible-centos-source-upgrade https://zuul.opendev.org/t/openstack/build/196cc5cf2bc24f3da1881d779cd90eb2 : SUCCESS in 1h 42m 26s\n- tripleo-build-containers-centos-7 https://zuul.opendev.org/t/openstack/build/298286ecd0864f0ebbd37c1484ba5f72 : SUCCESS in 54m 23s\n- kolla-ansible-bifrost-centos-source https://zuul.opendev.org/t/openstack/build/5f943ac70a8849b692c0dd40cd0cbb6a : SUCCESS in 56m 32s (non-voting)\n- kolla-ansible-centos-source-ceph https://zuul.opendev.org/t/openstack/build/01041f66273a4f3a9bf045f0c6644c77 : FAILURE in 1h 44m 02s (non-voting)\n- kolla-ansible-centos-source-upgrade-ceph https://zuul.opendev.org/t/openstack/build/c8954ea4d34a46e59f073a6c55247cf8 : TIMED_OUT in 2h 33m 49s (non-voting)\n- kolla-ansible-centos-source-cinder-lvm https://zuul.opendev.org/t/openstack/build/b017bb6416294e6184335753038afd0b : TIMED_OUT in 2h 03m 54s (non-voting)\n- kolla-ansible-centos-source-scenario-nfv https://zuul.opendev.org/t/openstack/build/8e32470bea9340c1b155d7012709b2e8 : TIMED_OUT in 2h 03m 43s (non-voting)\n- kolla-ansible-centos-source-ironic https://zuul.opendev.org/t/openstack/build/0ab066bf5dee449e8e76f07f702e7312 : SUCCESS in 1h 41m 46s (non-voting)\n- kolla-build-debian-source https://zuul.opendev.org/t/openstack/build/611de181e97d4a1986ce677de1dbd078 : SUCCESS in 24m 17s (non-voting)\n- kolla-build-debian-binary https://zuul.opendev.org/t/openstack/build/7d3d2f91823e4859a0d98b674d39523d : SUCCESS in 16m 51s (non-voting)\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/48fbb507159c439eb2b7ffdbf3e9457f : SUCCESS in 4m 46s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/1949f1cc079945a8b1d969a7088d9bc3 : SUCCESS in 5m 25s\n- openstack-tox-py27 https://zuul.opendev.org/t/openstack/build/ab7d09abf3804353a43c5db213ef1633 : SUCCESS in 3m 10s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/d5d3de4ac51b4a8f98604ca71c8f2c22 : SUCCESS in 2m 48s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/080523d0b358433f8ea01ef959f45f61 : SUCCESS in 5m 17s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/af5abfc761f847cb8fff876a30c61623 : SUCCESS in 6m 51s\n- kolla-build-ubuntu-binary https://zuul.opendev.org/t/openstack/build/0ac00ae659604abeb13aeac0771bc2d8 : SUCCESS in 31m 32s (non-voting)\n- kolla-build-ubuntu-source https://zuul.opendev.org/t/openstack/build/d60b996d4b8745b5ad7cd8119711e660 : SUCCESS in 47m 21s\n- kolla-ansible-ubuntu-source https://zuul.opendev.org/t/openstack/build/d3f7d6b35e6748c1bf30f7f33246e753 : SUCCESS in 55m 00s\n- kolla-ansible-ubuntu-binary https://zuul.opendev.org/t/openstack/build/4f34b53fa0ed43af8c3c96d7c1f3ae7b : SUCCESS in 1h 04m 37s (non-voting)\n- kolla-ansible-ubuntu-source-upgrade https://zuul.opendev.org/t/openstack/build/cc0f4884747d412f81e2785253a970e3 : SUCCESS in 1h 00m 01s\n- kolla-ansible-ubuntu-source-zun https://zuul.opendev.org/t/openstack/build/98ed279eb64a4715a83a0dc385da2835 : SUCCESS in 1h 08m 46s (non-voting)\n- kolla-ansible-ubuntu-source-ceph https://zuul.opendev.org/t/openstack/build/0f6141a2d99e4b48aa19401dfd167246 : SUCCESS in 1h 41m 43s (non-voting)\n- kolla-ansible-ubuntu-source-upgrade-ceph https://zuul.opendev.org/t/openstack/build/bc958004103643c1a93722103394d048 : SUCCESS in 1h 43m 59s (non-voting)\n- kolla-ansible-ubuntu-source-cinder-lvm https://zuul.opendev.org/t/openstack/build/1347b8fbb5964e6d9712545588f0a4f4 : SUCCESS in 1h 18m 03s (non-voting)\n- kolla-ansible-ubuntu-source-ironic https://zuul.opendev.org/t/openstack/build/b8652b7a76384bb480b840fc8929a292 : SUCCESS in 1h 02m 53s (non-voting)","accounts_in_message":[],"_revision_number":1},{"id":"0b96dcb457ef2dbdd907d82be4721ca490a387db","author":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},"date":"2019-10-04 09:17:56.000000000","message":"Uploaded patch set 2.","accounts_in_message":[],"_revision_number":2},{"id":"65cb0d31d38105b742578483016bf8db7b8ba9a5","author":{"_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"},"date":"2019-10-04 09:19:58.000000000","message":"Patch Set 2: Code-Review+2","accounts_in_message":[],"_revision_number":2},{"id":"b7536aee6ae65ac024aae43b86a2bb509d58192d","author":{"_account_id":8367,"name":"Arx Cruz","email":"arxcruz@redhat.com","username":"arxcruz"},"date":"2019-10-04 09:32:47.000000000","message":"Patch Set 2: Code-Review+1","accounts_in_message":[],"_revision_number":2},{"id":"b7cf664ab343842302a537642522abf51111f377","author":{"_account_id":23181,"name":"RDO Third Party CI","email":"dmsimard+rdothirdparty@redhat.com","username":"rdothirdparty","tags":["SERVICE_USER"]},"date":"2019-10-04 09:59:15.000000000","message":"Patch Set 2:\n\nBuild succeeded (check pipeline).\n\n- tripleo-build-containers-rhel-8-master http://logs.rdoproject.org/52/685952/2/openstack-check/tripleo-build-containers-rhel-8-master/865070d/ : SUCCESS in 39m 19s","accounts_in_message":[],"_revision_number":2},{"id":"c98144448d44e4fdc41a6064b910a3299ce8b1e4","author":{"_account_id":24072,"name":"Marcin Juszkiewicz","email":"mjuszkiewicz@redhat.com","username":"hrw"},"date":"2019-10-04 11:06:38.000000000","message":"Patch Set 2: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":2},{"id":"3f3fa3052b8858eb7253ed9ee4a423e9b776291c","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-10-04 11:37:19.000000000","message":"Patch Set 2: Verified+1\n\nBuild succeeded (check pipeline).\n\n- kolla-build-centos-binary https://zuul.opendev.org/t/openstack/build/4f6f35b5c314414ea26e87aaa5c9318c : SUCCESS in 52m 26s (non-voting)\n- kolla-build-centos-source https://zuul.opendev.org/t/openstack/build/52bafe8e144f46b0a6c6dc1c2e15d81e : SUCCESS in 51m 53s\n- kolla-ansible-centos-source https://zuul.opendev.org/t/openstack/build/55e3f7a723454cd7a8924bface56d334 : SUCCESS in 1h 23m 00s\n- kolla-ansible-centos-binary https://zuul.opendev.org/t/openstack/build/81ea72ee4d9f44c8afaafe31facd04d9 : SUCCESS in 1h 17m 19s (non-voting)\n- kolla-ansible-centos-source-upgrade https://zuul.opendev.org/t/openstack/build/6d5b608ced224ba4bfb9b2935b1e2294 : SUCCESS in 1h 22m 46s\n- tripleo-build-containers-centos-7 https://zuul.opendev.org/t/openstack/build/d4801efe97864c6192e4269beb5944f8 : SUCCESS in 59m 27s\n- kolla-ansible-bifrost-centos-source https://zuul.opendev.org/t/openstack/build/cbff57858e0c4705a160929f284b79c2 : SUCCESS in 49m 48s (non-voting)\n- kolla-ansible-centos-source-ceph https://zuul.opendev.org/t/openstack/build/5051cb87de3d457485a04b2ffbd92dbd : SUCCESS in 2h 13m 02s (non-voting)\n- kolla-ansible-centos-source-upgrade-ceph https://zuul.opendev.org/t/openstack/build/14a2e203b7364db0b14220b2b666d78f : SUCCESS in 1h 57m 50s (non-voting)\n- kolla-ansible-centos-source-cinder-lvm https://zuul.opendev.org/t/openstack/build/8c8bd996281143a79b7f3fc3325540a9 : SUCCESS in 1h 52m 52s (non-voting)\n- kolla-ansible-centos-source-scenario-nfv https://zuul.opendev.org/t/openstack/build/09090c629dfd45c0a2419ee1bf869c2c : FAILURE in 1h 50m 23s (non-voting)\n- kolla-ansible-centos-source-ironic https://zuul.opendev.org/t/openstack/build/7b47b4258a4f4786849e274bf0b71e8c : FAILURE in 1h 04m 59s (non-voting)\n- kolla-build-debian-source https://zuul.opendev.org/t/openstack/build/dfb1bbdb7ad44541a7164099a23dc575 : SUCCESS in 28m 34s (non-voting)\n- kolla-build-debian-binary https://zuul.opendev.org/t/openstack/build/182ef446fa8b44b39292df5e6087fb95 : FAILURE in 34m 26s (non-voting)\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/d17ab2efdc9e4c6fa59003040fa6b1a3 : SUCCESS in 3m 29s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/d15eaa150cc441cdb234e7ae11ddb2e8 : SUCCESS in 5m 57s\n- openstack-tox-py27 https://zuul.opendev.org/t/openstack/build/47adf85f7dd14afe909fa8bddee25f6c : SUCCESS in 3m 15s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/f2cd84dab76a47c9b3dcc1abd9952857 : SUCCESS in 3m 04s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/fb2ef2e3bcbc436d89d23511509b9d58 : SUCCESS in 5m 08s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/a88c2e6423f24eb4b4b3fe2da51dc740 : SUCCESS in 6m 06s\n- kolla-build-ubuntu-binary https://zuul.opendev.org/t/openstack/build/c159404f5ff4446d83a286bb4bc41c8b : SUCCESS in 24m 03s (non-voting)\n- kolla-build-ubuntu-source https://zuul.opendev.org/t/openstack/build/b82c1a918a9f45d58815b9e4d1c6f683 : SUCCESS in 36m 00s\n- kolla-ansible-ubuntu-source https://zuul.opendev.org/t/openstack/build/06e54efe49f745ee83a254a1b0755ba2 : SUCCESS in 1h 01m 34s\n- kolla-ansible-ubuntu-binary https://zuul.opendev.org/t/openstack/build/cb98dc5de85742ba95fdec86db037069 : SUCCESS in 54m 46s (non-voting)\n- kolla-ansible-ubuntu-source-upgrade https://zuul.opendev.org/t/openstack/build/500584e89d9542598fd6d4d00e92c0d1 : SUCCESS in 1h 02m 16s\n- kolla-ansible-ubuntu-source-zun https://zuul.opendev.org/t/openstack/build/356990353a7041bbb43801d87438e7cd : SUCCESS in 1h 24m 12s (non-voting)\n- kolla-ansible-ubuntu-source-ceph https://zuul.opendev.org/t/openstack/build/3fc75ea713db47748d5937dbc8bc63da : SUCCESS in 1h 26m 39s (non-voting)\n- kolla-ansible-ubuntu-source-upgrade-ceph https://zuul.opendev.org/t/openstack/build/3cdd02c056604a2da8fd5b5ba9997ef8 : SUCCESS in 1h 39m 56s (non-voting)\n- kolla-ansible-ubuntu-source-cinder-lvm https://zuul.opendev.org/t/openstack/build/0235d97d4f064fd2b535308a7893c3bc : SUCCESS in 1h 23m 33s (non-voting)\n- kolla-ansible-ubuntu-source-ironic https://zuul.opendev.org/t/openstack/build/93a64522f6764531af4d9ba41a6b2c1b : SUCCESS in 1h 02m 59s (non-voting)","accounts_in_message":[],"_revision_number":2},{"id":"b7980b5979c9ad665c126370855aab1c7e877806","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-10-04 11:37:35.000000000","message":"Patch Set 2: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":2},{"id":"0ab7abac9925b0da4425db4fc34e09908bdcb8bf","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-10-04 14:01:53.000000000","message":"Patch Set 2: Verified+2\n\nBuild succeeded (gate pipeline).\n\n- kolla-build-centos-source https://zuul.opendev.org/t/openstack/build/6ca1a078b37640dc9cd342ac1274e0bf : SUCCESS in 51m 26s\n- kolla-ansible-centos-source https://zuul.opendev.org/t/openstack/build/7b500eebad0446fc966ee6611d7665ae : SUCCESS in 1h 27m 29s\n- kolla-ansible-centos-source-upgrade https://zuul.opendev.org/t/openstack/build/b4a46ecb6b5a484db9dcc864c42f97f6 : SUCCESS in 1h 43m 07s\n- tripleo-build-containers-centos-7 https://zuul.opendev.org/t/openstack/build/91f829ad66024a438144132361a37841 : SUCCESS in 1h 10m 32s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/de1c2be949e2483f96a9e4e5167a7957 : SUCCESS in 3m 01s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/9ff29a6688d04a89a478680890ba21bd : SUCCESS in 4m 03s\n- openstack-tox-py27 https://zuul.opendev.org/t/openstack/build/412d4a736e0846b3bd4c6d6cf74dc6b7 : SUCCESS in 2m 55s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/23ce5148eef141e49dedbf178ae6c5fe : SUCCESS in 3m 09s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/feaa755976a74ad0a0876727f3c878a6 : SUCCESS in 3m 09s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/5412ff4ffd0541418558e8860db7819b : SUCCESS in 4m 35s\n- kolla-build-ubuntu-source https://zuul.opendev.org/t/openstack/build/817ab1068caa479797345aaa1257a664 : SUCCESS in 31m 14s\n- kolla-ansible-ubuntu-source https://zuul.opendev.org/t/openstack/build/e17838e143934ac5a18f65934363695d : SUCCESS in 55m 13s\n- kolla-ansible-ubuntu-source-upgrade https://zuul.opendev.org/t/openstack/build/a1b0d74acdae431e9f6aee5eee02340f : SUCCESS in 1h 01m 51s","accounts_in_message":[],"_revision_number":2},{"id":"4f4e1fe925f8ca7f6ee6d86ebe95947ce5e869cd","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-10-04 14:01:53.000000000","message":"Change has been successfully merged by Zuul","accounts_in_message":[],"_revision_number":2},{"id":"24ed234b3096c7560e75f09cc018bb9d3f3b4fba","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-10-04 14:03:36.000000000","message":"Patch Set 2:\n\nBuild succeeded (promote pipeline).\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/43c01da2b9fb438c9d61b0dbe851af12 : SUCCESS in 1m 15s","accounts_in_message":[],"_revision_number":2}],"current_revision_number":2,"current_revision":"0d0da5d99071231b83d41441bfb609406c7271a5","revisions":{"356597a466914c132d9bae838714202f3f3c4881":{"kind":"REWORK","_number":1,"created":"2019-10-01 12:23:01.000000000","uploader":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},"ref":"refs/changes/52/685952/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/kolla","ref":"refs/changes/52/685952/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/kolla refs/changes/52/685952/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/kolla refs/changes/52/685952/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/kolla refs/changes/52/685952/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/kolla refs/changes/52/685952/1"}}},"commit":{"parents":[{"commit":"73f9b5e47c7cab658ef75ac3dcdc2f006ebc7557","subject":"Adding python*-pyngus package to openstack base","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/73f9b5e47c7cab658ef75ac3dcdc2f006ebc7557"}]}],"author":{"name":"Mark Goddard","email":"mark@stackhpc.com","date":"2019-10-01 12:23:01.000000000","tz":0},"committer":{"name":"Mark Goddard","email":"mark@stackhpc.com","date":"2019-10-01 12:23:01.000000000","tz":0},"subject":"Revert \"Adding python*-pyngus package to openstack base\"","message":"Revert \"Adding python*-pyngus package to openstack base\"\n\nThis reverts commit 73f9b5e47c7cab658ef75ac3dcdc2f006ebc7557.\n\nIt was fixed on the packaging side: https://review.rdoproject.org/r/#/c/22745/\n\nChange-Id: Iea8fd1ac30400c67f76aec6d37ee7614f9c51770\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/356597a466914c132d9bae838714202f3f3c4881"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/356597a466914c132d9bae838714202f3f3c4881"}]},"branch":"refs/heads/master"},"0d0da5d99071231b83d41441bfb609406c7271a5":{"kind":"REWORK","_number":2,"created":"2019-10-04 09:17:56.000000000","uploader":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},"ref":"refs/changes/52/685952/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/kolla","ref":"refs/changes/52/685952/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/kolla refs/changes/52/685952/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/kolla refs/changes/52/685952/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/kolla refs/changes/52/685952/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/kolla refs/changes/52/685952/2"}}},"commit":{"parents":[{"commit":"73f9b5e47c7cab658ef75ac3dcdc2f006ebc7557","subject":"Adding python*-pyngus package to openstack base","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/73f9b5e47c7cab658ef75ac3dcdc2f006ebc7557"}]}],"author":{"name":"Mark Goddard","email":"mark@stackhpc.com","date":"2019-10-01 12:23:01.000000000","tz":0},"committer":{"name":"Mark Goddard","email":"mark@stackhpc.com","date":"2019-10-04 09:16:43.000000000","tz":60},"subject":"Revert \"Adding python*-pyngus package to openstack base\"","message":"Revert \"Adding python*-pyngus package to openstack base\"\n\nThis partially reverts commit 73f9b5e47c7cab658ef75ac3dcdc2f006ebc7557.\n\nIt was fixed in RDO on the packaging side:\nhttps://review.rdoproject.org/r/#/c/22745/\n\nUbuntu packages do not have a dependency on python3-pyngus, so that\npackage has been left in place.\n\nChange-Id: Iea8fd1ac30400c67f76aec6d37ee7614f9c51770\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/0d0da5d99071231b83d41441bfb609406c7271a5"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/0d0da5d99071231b83d41441bfb609406c7271a5"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[],"submit_requirements":[]}
