)]}'
{"id":"openstack%2Fkolla~685959","triplet_id":"openstack%2Fkolla~stable%2Fstein~I66eacbf2f51b93b94f2aea91d8db5e7f12000c40","project":"openstack/kolla","branch":"stable/stein","hashtags":[],"change_id":"I66eacbf2f51b93b94f2aea91d8db5e7f12000c40","subject":"Bump up version for magnum stein to 8.1.0","status":"MERGED","created":"2019-10-01 13:06:12.000000000","updated":"2019-10-02 16:32:22.000000000","submitted":"2019-10-02 16:30:20.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"685959-1570033820351-116e89a0","meta_rev_id":"a6e5f5a0f29bd08549cb534a69da4bd96543abf3","_number":685959,"virtual_id_number":685959,"owner":{"_account_id":28022,"name":"Bharat Kunwar","email":"brtknr@bath.edu","username":"brtknr"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"date":"2019-10-01 14:47:56.000000000","_account_id":28022,"name":"Bharat Kunwar","email":"brtknr@bath.edu","username":"brtknr"},{"value":0,"_account_id":24072,"name":"Marcin Juszkiewicz","email":"mjuszkiewicz@redhat.com","username":"hrw"},{"value":0,"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2019-10-02 16:30:20.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":24072,"name":"Marcin Juszkiewicz","email":"mjuszkiewicz@redhat.com","username":"hrw"},"all":[{"value":0,"_account_id":28022,"name":"Bharat Kunwar","email":"brtknr@bath.edu","username":"brtknr"},{"value":2,"date":"2019-10-02 15:28:49.000000000","_account_id":24072,"name":"Marcin Juszkiewicz","email":"mjuszkiewicz@redhat.com","username":"hrw"},{"value":2,"date":"2019-10-01 14:59:33.000000000","_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":24072,"name":"Marcin Juszkiewicz","email":"mjuszkiewicz@redhat.com","username":"hrw"},"all":[{"value":0,"_account_id":28022,"name":"Bharat Kunwar","email":"brtknr@bath.edu","username":"brtknr"},{"value":1,"date":"2019-10-02 15:28:49.000000000","_account_id":24072,"name":"Marcin Juszkiewicz","email":"mjuszkiewicz@redhat.com","username":"hrw"},{"value":0,"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Backport-Candidate":{"all":[{"value":0,"_account_id":28022,"name":"Bharat Kunwar","email":"brtknr@bath.edu","username":"brtknr"},{"value":0,"_account_id":24072,"name":"Marcin Juszkiewicz","email":"mjuszkiewicz@redhat.com","username":"hrw"},{"value":0,"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Do Not Backport"," 0":"Backport Review Needed","+1":"Should Backport"},"description":"","default_value":0,"optional":true},"Review-Priority":{"all":[{"value":0,"_account_id":28022,"name":"Bharat Kunwar","email":"brtknr@bath.edu","username":"brtknr"},{"value":0,"_account_id":24072,"name":"Marcin Juszkiewicz","email":"mjuszkiewicz@redhat.com","username":"hrw"},{"value":0,"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Branch Freeze"," 0":"No Priority","+1":"Important Change","+2":"Gate Blocker Fix / Urgent Change"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":24072,"name":"Marcin Juszkiewicz","email":"mjuszkiewicz@redhat.com","username":"hrw"},{"_account_id":28022,"name":"Bharat Kunwar","email":"brtknr@bath.edu","username":"brtknr"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2019-10-01 14:59:33.000000000","updated_by":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},"reviewer":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},"state":"REVIEWER"},{"updated":"2019-10-02 15:28:49.000000000","updated_by":{"_account_id":24072,"name":"Marcin Juszkiewicz","email":"mjuszkiewicz@redhat.com","username":"hrw"},"reviewer":{"_account_id":24072,"name":"Marcin Juszkiewicz","email":"mjuszkiewicz@redhat.com","username":"hrw"},"state":"REVIEWER"},{"updated":"2019-10-02 16:30:20.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"}],"messages":[{"id":"8cf59d6debe70eb82fda49574983df4fd2ada5f6","author":{"_account_id":28022,"name":"Bharat Kunwar","email":"brtknr@bath.edu","username":"brtknr"},"date":"2019-10-01 13:06:12.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"da4fac529a1e7581dc3c0847ca9928cee40b6b43","author":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},"date":"2019-10-01 13:27:14.000000000","message":"Patch Set 1:\n\nWas this the only changed version?","accounts_in_message":[],"_revision_number":1},{"id":"9401c5f67caf1a9f61c7c3565a751375b27b95bd","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-10-01 14:47:08.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\n\n- kolla-build-centos-binary https://zuul.opendev.org/t/openstack/build/a219b82efc24480dae167819562b6b1b : SUCCESS in 49m 45s (non-voting)\n- kolla-build-centos-source https://zuul.opendev.org/t/openstack/build/9a6f1ff282bd40e68d98b48ce8f49fbe : SUCCESS in 48m 06s\n- kolla-ansible-centos-source https://zuul.opendev.org/t/openstack/build/9b05cc56290d4f4d8c83ade0c91a1d1e : SUCCESS in 1h 14m 57s\n- kolla-ansible-centos-binary https://zuul.opendev.org/t/openstack/build/90c1c23d7ee24c38b566d7a3a63be35d : SUCCESS in 1h 21m 05s (non-voting)\n- kolla-ansible-centos-source-upgrade https://zuul.opendev.org/t/openstack/build/867824b1d52b492e9d5ada1e7633f562 : SUCCESS in 1h 34m 23s\n- tripleo-build-containers-centos-7 https://zuul.opendev.org/t/openstack/build/9fdb9d7da1b8413980b845dc14cb9a24 : SUCCESS in 49m 16s (non-voting)\n- kolla-build-oraclelinux-binary https://zuul.opendev.org/t/openstack/build/32b7bc4cea864a95946f295ad2f45a8f : SUCCESS in 1h 01m 38s (non-voting)\n- kolla-build-oraclelinux-source https://zuul.opendev.org/t/openstack/build/cb5e05e686f74b829d54c8d01610f88f : SUCCESS in 48m 46s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/353b051dad104c408bccb8d7af8569df : SUCCESS in 3m 45s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/252e6209b0454c8393577d872e123adb : SUCCESS in 4m 05s\n- openstack-tox-py27 https://zuul.opendev.org/t/openstack/build/378e984856aa446f9b80cc79ff5d79a1 : SUCCESS in 2m 42s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/0e1343b0af964536919fafcb1499b8f1 : SUCCESS in 4m 36s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/c2780f3305ec4965852e9a46ec493e88 : SUCCESS in 3m 21s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/137ad5b556074d5ba5b0cf5a5fad2731 : SUCCESS in 5m 11s\n- kolla-build-ubuntu-binary https://zuul.opendev.org/t/openstack/build/cff32398cef6463a8253f30685d03077 : RETRY_LIMIT in 1m 00s (non-voting)\n- kolla-build-ubuntu-source https://zuul.opendev.org/t/openstack/build/8f510b02e7bb40c9b33ade5eb927e72e : SUCCESS in 33m 13s\n- kolla-ansible-ubuntu-source https://zuul.opendev.org/t/openstack/build/92cb669a0c7b4bae91cde272e6eabba9 : SUCCESS in 53m 34s\n- kolla-ansible-ubuntu-binary https://zuul.opendev.org/t/openstack/build/cd425da9e5f6478bb8411abb113ead62 : SUCCESS in 53m 28s (non-voting)\n- kolla-ansible-ubuntu-source-upgrade https://zuul.opendev.org/t/openstack/build/b240647b566f4cf58d01a28fc37011fa : SUCCESS in 1h 07m 45s","accounts_in_message":[],"_revision_number":1},{"id":"08e22ddc9606a7f8d54fb51fec31573075d2cd81","author":{"_account_id":28022,"name":"Bharat Kunwar","email":"brtknr@bath.edu","username":"brtknr"},"date":"2019-10-01 14:47:56.000000000","message":"Patch Set 1:\n\nyes sir","accounts_in_message":[],"_revision_number":1},{"id":"7192bf02409eb5a3a0cf59c9a2d0811cefb44378","author":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},"date":"2019-10-01 14:59:33.000000000","message":"Patch Set 1: Code-Review+2","accounts_in_message":[],"_revision_number":1},{"id":"d3d1376935e90cf768c6089065fce602f67c39d7","author":{"_account_id":24072,"name":"Marcin Juszkiewicz","email":"mjuszkiewicz@redhat.com","username":"hrw"},"date":"2019-10-02 15:28:49.000000000","message":"Patch Set 1: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":1},{"id":"1a25aa134236becac543fa93bf8abfeac3d98b93","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-10-02 15:29:03.000000000","message":"Patch Set 1: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":1},{"id":"3392664a8a4759bbf70e010d2a753a6c0ed9385d","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-10-02 16:30:20.000000000","message":"Change has been successfully merged by Zuul","accounts_in_message":[],"_revision_number":1},{"id":"81e13ca60e847fed25b9f25f294b6e05c354918a","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-10-02 16:30:20.000000000","message":"Patch Set 1: Verified+2\n\nBuild succeeded (gate pipeline).\n\n- kolla-build-centos-source https://zuul.opendev.org/t/openstack/build/3b98110539064488a529e356a864c0e6 : SUCCESS in 54m 52s\n- kolla-build-oraclelinux-source https://zuul.opendev.org/t/openstack/build/6920c34f9f464a17ad7dc97a72fee924 : SUCCESS in 55m 43s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/2d6db98ebbf2446fa354250332c19418 : SUCCESS in 3m 47s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/7cbcc3ed7ffa4550add2d8c7a78c6074 : SUCCESS in 3m 24s\n- openstack-tox-py27 https://zuul.opendev.org/t/openstack/build/f13997d7b92047f8917c3bf11488518d : SUCCESS in 3m 37s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/4b1de1b0edcc4fff89f4b5ac3aa2b14e : SUCCESS in 4m 02s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/9904b15aa8474b149e4dce8b059d3061 : SUCCESS in 4m 29s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/892545cc89ff4e33b18a347789894085 : SUCCESS in 3m 10s\n- kolla-build-ubuntu-source https://zuul.opendev.org/t/openstack/build/70643917ef9a4d38a66f89de1deb8426 : SUCCESS in 38m 53s","accounts_in_message":[],"_revision_number":1},{"id":"a6e5f5a0f29bd08549cb534a69da4bd96543abf3","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-10-02 16:32:22.000000000","message":"Patch Set 1:\n\nBuild succeeded (promote pipeline).\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/7953816df0c94b50aa8a7dacc44dc5aa : SUCCESS in 1m 31s","accounts_in_message":[],"_revision_number":1}],"current_revision_number":1,"current_revision":"dcb13a9e93ee71b35048b18993ab444f0101c7f3","revisions":{"dcb13a9e93ee71b35048b18993ab444f0101c7f3":{"kind":"REWORK","_number":1,"created":"2019-10-01 13:06:12.000000000","uploader":{"_account_id":28022,"name":"Bharat Kunwar","email":"brtknr@bath.edu","username":"brtknr"},"ref":"refs/changes/59/685959/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/kolla","ref":"refs/changes/59/685959/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/kolla refs/changes/59/685959/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/kolla refs/changes/59/685959/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/kolla refs/changes/59/685959/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/kolla refs/changes/59/685959/1"}}},"commit":{"parents":[{"commit":"bcac8b82ad228f58ad1e422330ad60701046e5d5","subject":"sensu-client: Pin minitest Ruby gem on CentOS","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/bcac8b82ad228f58ad1e422330ad60701046e5d5"}]}],"author":{"name":"Bharat Kunwar","email":"brtknr@bath.edu","date":"2019-10-01 13:05:38.000000000","tz":0},"committer":{"name":"Bharat Kunwar","email":"brtknr@bath.edu","date":"2019-10-01 13:05:38.000000000","tz":0},"subject":"Bump up version for magnum stein to 8.1.0","message":"Bump up version for magnum stein to 8.1.0\n\nChange-Id: I66eacbf2f51b93b94f2aea91d8db5e7f12000c40\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/dcb13a9e93ee71b35048b18993ab444f0101c7f3"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/dcb13a9e93ee71b35048b18993ab444f0101c7f3"}]},"branch":"refs/heads/stable/stein"}},"requirements":[],"submit_records":[],"submit_requirements":[]}
