)]}'
{"id":"openstack%2Fkolla~685984","triplet_id":"openstack%2Fkolla~master~Iad7ad20542a159469d95e990c20564633046b251","project":"openstack/kolla","branch":"master","topic":"bug/1846207","hashtags":[],"change_id":"Iad7ad20542a159469d95e990c20564633046b251","subject":"Install rsync in swift-base containers on Debian/Ubuntu","status":"MERGED","created":"2019-10-01 14:38:15.000000000","updated":"2019-10-02 08:40:52.000000000","submitted":"2019-10-01 18:25:44.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":1,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"685984-1569954344931-b674a50a","meta_rev_id":"eabebd74d55d3aaf65c1bd68a3f64e944e51cc3f","_number":685984,"virtual_id_number":685984,"owner":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},{"value":0,"_account_id":24072,"name":"Marcin Juszkiewicz","email":"mjuszkiewicz@redhat.com","username":"hrw"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2019-10-01 18:25:44.000000000","post_submit":true,"permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"date":"2019-10-01 15:41:36.000000000","_account_id":23181,"name":"RDO Third Party CI","email":"dmsimard+rdothirdparty@redhat.com","username":"rdothirdparty","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"all":[{"value":2,"date":"2019-10-01 14:59:36.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},{"value":2,"date":"2019-10-01 14:57:32.000000000","_account_id":24072,"name":"Marcin Juszkiewicz","email":"mjuszkiewicz@redhat.com","username":"hrw"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":23181,"name":"RDO Third Party CI","email":"dmsimard+rdothirdparty@redhat.com","username":"rdothirdparty","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"all":[{"value":1,"date":"2019-10-01 14:59:36.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},{"value":0,"_account_id":24072,"name":"Marcin Juszkiewicz","email":"mjuszkiewicz@redhat.com","username":"hrw"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":23181,"name":"RDO Third Party CI","email":"dmsimard+rdothirdparty@redhat.com","username":"rdothirdparty","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Backport-Candidate":{"all":[{"value":0,"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},{"value":0,"_account_id":24072,"name":"Marcin Juszkiewicz","email":"mjuszkiewicz@redhat.com","username":"hrw"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":23181,"name":"RDO Third Party CI","email":"dmsimard+rdothirdparty@redhat.com","username":"rdothirdparty","tags":["SERVICE_USER"]}],"values":{"-1":"Do Not Backport"," 0":"Backport Review Needed","+1":"Should Backport"},"description":"","default_value":0,"optional":true},"Review-Priority":{"all":[{"value":0,"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},{"value":0,"_account_id":24072,"name":"Marcin Juszkiewicz","email":"mjuszkiewicz@redhat.com","username":"hrw"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":23181,"name":"RDO Third Party CI","email":"dmsimard+rdothirdparty@redhat.com","username":"rdothirdparty","tags":["SERVICE_USER"]}],"values":{"-1":"Branch Freeze"," 0":"No Priority","+1":"Important Change","+2":"Gate Blocker Fix / Urgent Change"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},{"_account_id":23181,"name":"RDO Third Party CI","email":"dmsimard+rdothirdparty@redhat.com","username":"rdothirdparty","tags":["SERVICE_USER"]},{"_account_id":24072,"name":"Marcin Juszkiewicz","email":"mjuszkiewicz@redhat.com","username":"hrw"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2019-10-01 14:57:32.000000000","updated_by":{"_account_id":24072,"name":"Marcin Juszkiewicz","email":"mjuszkiewicz@redhat.com","username":"hrw"},"reviewer":{"_account_id":24072,"name":"Marcin Juszkiewicz","email":"mjuszkiewicz@redhat.com","username":"hrw"},"state":"REVIEWER"},{"updated":"2019-10-01 14:59:36.000000000","updated_by":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"reviewer":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"state":"REVIEWER"},{"updated":"2019-10-01 15:41:36.000000000","updated_by":{"_account_id":23181,"name":"RDO Third Party CI","email":"dmsimard+rdothirdparty@redhat.com","username":"rdothirdparty","tags":["SERVICE_USER"]},"reviewer":{"_account_id":23181,"name":"RDO Third Party CI","email":"dmsimard+rdothirdparty@redhat.com","username":"rdothirdparty","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2019-10-01 18:25:44.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"}],"messages":[{"id":"a9976560e8e6ee48b87b8cda50a34c18d96ef7ea","author":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},"date":"2019-10-01 14:38:15.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"b38e53b0b146b55f5ddfe8a89734681e72c0b4fd","author":{"_account_id":24072,"name":"Marcin Juszkiewicz","email":"mjuszkiewicz@redhat.com","username":"hrw"},"date":"2019-10-01 14:45:51.000000000","message":"Patch Set 1: Code-Review-1\n\n\"Install rsync in swift containers on Debian/Ubuntu\"\n\nAnd install rsync in swift-base?","accounts_in_message":[],"_revision_number":1},{"id":"211f36329f366b67608c09c5feebe6fab03281f4","author":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},"date":"2019-10-01 14:51:16.000000000","message":"Uploaded patch set 2.","accounts_in_message":[],"_revision_number":2},{"id":"45165e6c6fb40ada495d4fe3c0283ecd08ff59b9","author":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"date":"2019-10-01 14:54:20.000000000","message":"Patch Set 2: Code-Review+2\n\n(1 comment)","accounts_in_message":[],"_revision_number":2},{"id":"39503d619a4608df3a1c0f509be3a9850adfb0c8","author":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},"date":"2019-10-01 14:55:43.000000000","message":"Uploaded patch set 3.","accounts_in_message":[],"_revision_number":3},{"id":"eef17760a677428051b182d85c417f1c4216f2bb","author":{"_account_id":24072,"name":"Marcin Juszkiewicz","email":"mjuszkiewicz@redhat.com","username":"hrw"},"date":"2019-10-01 14:57:32.000000000","message":"Patch Set 3: Code-Review+2","accounts_in_message":[],"_revision_number":3},{"id":"0a506bd504ce7aa9ab514262f7d55e98e365631c","author":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"date":"2019-10-01 14:59:36.000000000","message":"Patch Set 3: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":3},{"id":"261f1d8ebbe0f0eeb4f2b562c0c69b29a51627b3","author":{"_account_id":23181,"name":"RDO Third Party CI","email":"dmsimard+rdothirdparty@redhat.com","username":"rdothirdparty","tags":["SERVICE_USER"]},"date":"2019-10-01 15:41:36.000000000","message":"Patch Set 3:\n\nBuild succeeded (check pipeline).\n\n- tripleo-build-containers-rhel-8-master http://logs.rdoproject.org/84/685984/3/openstack-check/tripleo-build-containers-rhel-8-master/0179248/ : SUCCESS in 36m 27s","accounts_in_message":[],"_revision_number":3},{"id":"6727d9ca08f30b76f423ebe7994219b4ed8e4c98","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-10-01 16:46:22.000000000","message":"Patch Set 3: Verified+1\n\nBuild succeeded (check pipeline).\n\n- kolla-build-centos-binary https://zuul.opendev.org/t/openstack/build/fc8e9c48d7d0465b9fef840da93f8bef : SUCCESS in 1h 05m 12s (non-voting)\n- kolla-build-centos-source https://zuul.opendev.org/t/openstack/build/3f12fe7ef61740bca0aae11813762dd9 : SUCCESS in 52m 18s\n- kolla-ansible-centos-source https://zuul.opendev.org/t/openstack/build/0f16e829b0194d06b93f53d015acf78a : SUCCESS in 1h 41m 17s\n- kolla-ansible-centos-binary https://zuul.opendev.org/t/openstack/build/2fbdb90d82774aadb4db2016b879dd31 : SUCCESS in 1h 39m 18s (non-voting)\n- kolla-ansible-centos-source-upgrade https://zuul.opendev.org/t/openstack/build/3ca18f8348304c3c805b8829e513d19f : SUCCESS in 1h 45m 43s\n- tripleo-build-containers-centos-7 https://zuul.opendev.org/t/openstack/build/49e5aafc0b7f4647aea7a26c232cbed2 : SUCCESS in 1h 07m 48s\n- kolla-build-debian-source https://zuul.opendev.org/t/openstack/build/b793d1fdd4ea4fb29964607d883cae11 : SUCCESS in 35m 25s (non-voting)\n- kolla-build-debian-binary https://zuul.opendev.org/t/openstack/build/1fc18f43b0b444d99159ffe6a2f5e2ad : SUCCESS in 45m 49s (non-voting)\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/71305fa2364a404cabeab13df5ace6a7 : SUCCESS in 4m 00s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/abd9e6673bd242b881bdd13eb20f9b0e : SUCCESS in 5m 14s\n- openstack-tox-py27 https://zuul.opendev.org/t/openstack/build/08d32c5d3d164b71a93d856b28be1515 : SUCCESS in 4m 08s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/6c575c0507d54b07a357212270f7c31c : SUCCESS in 4m 17s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/f7a0028cf6f84c66b17f3e8feddb73dc : SUCCESS in 3m 54s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/10080efe852d43fd9d210e6c23f8e6b4 : SUCCESS in 6m 00s\n- kolla-build-ubuntu-binary https://zuul.opendev.org/t/openstack/build/77480e52b135403e9652034176739f7e : SUCCESS in 36m 28s (non-voting)\n- kolla-build-ubuntu-source https://zuul.opendev.org/t/openstack/build/3ba2f49d197248d08b1246baf6ef9f32 : SUCCESS in 34m 05s\n- kolla-ansible-ubuntu-source https://zuul.opendev.org/t/openstack/build/4dae0f986ba34ea39f79c36195424f3b : SUCCESS in 1h 01m 39s\n- kolla-ansible-ubuntu-binary https://zuul.opendev.org/t/openstack/build/6902f52e757d41a08fd82ff5cb52a656 : SUCCESS in 1h 05m 48s (non-voting)\n- kolla-ansible-ubuntu-source-upgrade https://zuul.opendev.org/t/openstack/build/a6687838386e48ff83dd935704c5f732 : SUCCESS in 1h 02m 07s","accounts_in_message":[],"_revision_number":3},{"id":"ea0e030a9b390bfdf338bf2dcf26f109cd5b3b23","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-10-01 16:46:36.000000000","message":"Patch Set 3: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":3},{"id":"a8ee92f206bb6bf54562659556284eb61fd2e01c","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-10-01 18:25:44.000000000","message":"Patch Set 3: Verified+2\n\nBuild succeeded (gate pipeline).\n\n- kolla-build-centos-source https://zuul.opendev.org/t/openstack/build/030bec7c764342dfa145cb6f8e6cf25f : SUCCESS in 1h 01m 02s\n- kolla-ansible-centos-source https://zuul.opendev.org/t/openstack/build/cc18a21ef81c47d5ad601c9300ef1cdb : SUCCESS in 1h 23m 00s\n- kolla-ansible-centos-source-upgrade https://zuul.opendev.org/t/openstack/build/1cdc9ed836984f98a4c372a06a21e6d9 : SUCCESS in 1h 34m 08s\n- tripleo-build-containers-centos-7 https://zuul.opendev.org/t/openstack/build/b6a8b8db950e4a87a2592dbf92cd0e3e : SUCCESS in 59m 19s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/ad9b11e90a354e8b82e38f68b48448a1 : SUCCESS in 3m 40s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/0a09d1d8d4f74d6480cf640b70cca1b5 : SUCCESS in 3m 58s\n- openstack-tox-py27 https://zuul.opendev.org/t/openstack/build/d0333bfed2554fb1922f12bad4e5e66e : SUCCESS in 3m 52s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/9d2a4c46797844c08cf291971e9329e5 : SUCCESS in 3m 50s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/b181067d8b5143a994b2e8d3ec59b296 : SUCCESS in 3m 04s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/becfdfa0df1a49aba12bd35e0df1cdf5 : SUCCESS in 5m 49s\n- kolla-build-ubuntu-source https://zuul.opendev.org/t/openstack/build/3b2be17c72c84eb4abe7fc443bc7fb06 : SUCCESS in 36m 58s\n- kolla-ansible-ubuntu-source https://zuul.opendev.org/t/openstack/build/31df2ce352994e72a9fc1774bbc804f6 : SUCCESS in 51m 16s\n- kolla-ansible-ubuntu-source-upgrade https://zuul.opendev.org/t/openstack/build/d79ecb85dc394f59bf262dff75f43f88 : SUCCESS in 1h 01m 07s","accounts_in_message":[],"_revision_number":3},{"id":"156e90007a2857448d54f584248748df3e058f38","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-10-01 18:25:44.000000000","message":"Change has been successfully merged by Zuul","accounts_in_message":[],"_revision_number":3},{"id":"4efe9708e0f53e818184bddedf4256e45cc1c14a","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-10-01 18:27:04.000000000","message":"Patch Set 3:\n\nBuild succeeded (promote pipeline).\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/da86038eaa3b4014ab3be22ef716de71 : SUCCESS in 55s","accounts_in_message":[],"_revision_number":3},{"id":"3307b5a5b033473bf7f9ed78cf82bdaf198c9805","author":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},"date":"2019-10-02 08:40:52.000000000","message":"Patch Set 3: Cherry Picked\n\nThis patchset was cherry picked to branch stable/stein as commit f8a8d87af079fbe509d6654c42f0ae2ca8f9515c","accounts_in_message":[],"_revision_number":3}],"current_revision_number":3,"current_revision":"d61c54be9d281a024a448f783dbde037d5fd1cda","revisions":{"dd9fd729c9849aa119052ad27bcf82ce9f281f15":{"kind":"REWORK","_number":1,"created":"2019-10-01 14:38:15.000000000","uploader":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},"ref":"refs/changes/84/685984/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/kolla","ref":"refs/changes/84/685984/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/kolla refs/changes/84/685984/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/kolla refs/changes/84/685984/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/kolla refs/changes/84/685984/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/kolla refs/changes/84/685984/1"}}},"commit":{"parents":[{"commit":"6afefd522b70ad57e25efbc61da94053ebd691fb","subject":"Merge \"CI: add Debian/binary job\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/6afefd522b70ad57e25efbc61da94053ebd691fb"}]}],"author":{"name":"Mark Goddard","email":"mark@stackhpc.com","date":"2019-10-01 14:36:47.000000000","tz":60},"committer":{"name":"Mark Goddard","email":"mark@stackhpc.com","date":"2019-10-01 14:36:47.000000000","tz":60},"subject":"Install rsync in swift-{account,container,object} containers on Ubuntu","message":"Install rsync in swift-{account,container,object} containers on Ubuntu\n\nThis is used by the replicator services.\n\nChange-Id: Iad7ad20542a159469d95e990c20564633046b251\nCloses-Bug: #1846207\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/dd9fd729c9849aa119052ad27bcf82ce9f281f15"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/dd9fd729c9849aa119052ad27bcf82ce9f281f15"}]},"branch":"refs/heads/master"},"770de8688646938e323839b043f7e99fde27808a":{"kind":"REWORK","_number":2,"created":"2019-10-01 14:51:16.000000000","uploader":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},"ref":"refs/changes/84/685984/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/kolla","ref":"refs/changes/84/685984/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/kolla refs/changes/84/685984/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/kolla refs/changes/84/685984/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/kolla refs/changes/84/685984/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/kolla refs/changes/84/685984/2"}}},"commit":{"parents":[{"commit":"6afefd522b70ad57e25efbc61da94053ebd691fb","subject":"Merge \"CI: add Debian/binary job\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/6afefd522b70ad57e25efbc61da94053ebd691fb"}]}],"author":{"name":"Mark Goddard","email":"mark@stackhpc.com","date":"2019-10-01 14:36:47.000000000","tz":60},"committer":{"name":"Mark Goddard","email":"mark@stackhpc.com","date":"2019-10-01 14:50:45.000000000","tz":60},"subject":"Install rsync in swift-base containers on Ubuntu","message":"Install rsync in swift-base containers on Ubuntu\n\nThis is used by the replicator services.\n\nChange-Id: Iad7ad20542a159469d95e990c20564633046b251\nCloses-Bug: #1846207\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/770de8688646938e323839b043f7e99fde27808a"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/770de8688646938e323839b043f7e99fde27808a"}]},"branch":"refs/heads/master"},"d61c54be9d281a024a448f783dbde037d5fd1cda":{"kind":"REWORK","_number":3,"created":"2019-10-01 14:55:43.000000000","uploader":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},"ref":"refs/changes/84/685984/3","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/kolla","ref":"refs/changes/84/685984/3","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/kolla refs/changes/84/685984/3 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/kolla refs/changes/84/685984/3 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/kolla refs/changes/84/685984/3 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/kolla refs/changes/84/685984/3"}}},"commit":{"parents":[{"commit":"6afefd522b70ad57e25efbc61da94053ebd691fb","subject":"Merge \"CI: add Debian/binary job\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/6afefd522b70ad57e25efbc61da94053ebd691fb"}]}],"author":{"name":"Mark Goddard","email":"mark@stackhpc.com","date":"2019-10-01 14:36:47.000000000","tz":60},"committer":{"name":"Mark Goddard","email":"mark@stackhpc.com","date":"2019-10-01 14:55:17.000000000","tz":60},"subject":"Install rsync in swift-base containers on Debian/Ubuntu","message":"Install rsync in swift-base containers on Debian/Ubuntu\n\nThis is used by the replicator services.\n\nChange-Id: Iad7ad20542a159469d95e990c20564633046b251\nCloses-Bug: #1846207\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/d61c54be9d281a024a448f783dbde037d5fd1cda"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/d61c54be9d281a024a448f783dbde037d5fd1cda"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[],"submit_requirements":[]}
