)]}'
{"id":"openstack%2Fkolla~686150","triplet_id":"openstack%2Fkolla~stable%2Fstein~I0f83c3498028135751b0b887665d009e5e19410f","project":"openstack/kolla","branch":"stable/stein","topic":"671507-stable/stein","hashtags":[],"change_id":"I0f83c3498028135751b0b887665d009e5e19410f","subject":"There is no Java 8 in Debian \u0027buster\u0027 (part 3)","status":"MERGED","created":"2019-10-02 13:11:54.000000000","updated":"2019-10-02 17:52:19.000000000","submitted":"2019-10-02 17:51:06.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"686150-1570038666838-fa124ec6","meta_rev_id":"c19439474c002fb360e7af9145edbae35f35a13b","_number":686150,"virtual_id_number":686150,"owner":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":24072,"name":"Marcin Juszkiewicz","email":"mjuszkiewicz@redhat.com","username":"hrw"},{"value":0,"date":"2019-10-02 15:02:26.000000000","_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},{"value":0,"_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2019-10-02 17:51:06.000000000","post_submit":true,"permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":24072,"name":"Marcin Juszkiewicz","email":"mjuszkiewicz@redhat.com","username":"hrw"},"all":[{"value":2,"date":"2019-10-02 13:13:07.000000000","_account_id":24072,"name":"Marcin Juszkiewicz","email":"mjuszkiewicz@redhat.com","username":"hrw"},{"value":0,"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},{"value":2,"date":"2019-10-02 14:29:28.000000000","_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"},"all":[{"value":0,"_account_id":24072,"name":"Marcin Juszkiewicz","email":"mjuszkiewicz@redhat.com","username":"hrw"},{"value":0,"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},{"value":1,"date":"2019-10-02 14:29:28.000000000","_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Backport-Candidate":{"all":[{"value":0,"_account_id":24072,"name":"Marcin Juszkiewicz","email":"mjuszkiewicz@redhat.com","username":"hrw"},{"value":0,"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},{"value":0,"_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Do Not Backport"," 0":"Backport Review Needed","+1":"Should Backport"},"description":"","default_value":0,"optional":true},"Review-Priority":{"all":[{"value":0,"_account_id":24072,"name":"Marcin Juszkiewicz","email":"mjuszkiewicz@redhat.com","username":"hrw"},{"value":0,"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},{"value":0,"_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Branch Freeze"," 0":"No Priority","+1":"Important Change","+2":"Gate Blocker Fix / Urgent Change"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":24072,"name":"Marcin Juszkiewicz","email":"mjuszkiewicz@redhat.com","username":"hrw"},{"_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2019-10-02 13:13:07.000000000","updated_by":{"_account_id":24072,"name":"Marcin Juszkiewicz","email":"mjuszkiewicz@redhat.com","username":"hrw"},"reviewer":{"_account_id":24072,"name":"Marcin Juszkiewicz","email":"mjuszkiewicz@redhat.com","username":"hrw"},"state":"REVIEWER"},{"updated":"2019-10-02 14:29:28.000000000","updated_by":{"_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"},"reviewer":{"_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"},"state":"REVIEWER"},{"updated":"2019-10-02 17:51:06.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"}],"messages":[{"id":"02eb2e3c5af0a2985ac6c56fc6d0923ac0e62547","author":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},"date":"2019-10-02 13:11:54.000000000","message":"Patch Set 1: Cherry Picked from branch master.","accounts_in_message":[],"_revision_number":1},{"id":"700aa62507fa51719201c9e14aec7ab784e920aa","author":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},"date":"2019-10-02 13:12:10.000000000","message":"Patch Set 1:\n\nThis is required to fix https://bugs.launchpad.net/kolla-ansible/+bug/1846336.","accounts_in_message":[],"_revision_number":1},{"id":"e752f6c9d93ae072bd5de69ab24f0f618eef9088","author":{"_account_id":24072,"name":"Marcin Juszkiewicz","email":"mjuszkiewicz@redhat.com","username":"hrw"},"date":"2019-10-02 13:13:07.000000000","message":"Patch Set 1: Code-Review+2","accounts_in_message":[],"_revision_number":1},{"id":"e374f363197e028ee043a52c3e06bd95e11a00f0","author":{"_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"},"date":"2019-10-02 14:29:28.000000000","message":"Patch Set 1: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":1},{"id":"12c69fdb32e30f15819aaec435c66da632d5cfa5","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-10-02 14:55:27.000000000","message":"Patch Set 1: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttp://docs.openstack.org/infra/manual/developers.html#automated-testing\n\n\n- kolla-build-centos-binary https://zuul.opendev.org/t/openstack/build/2b5c854c7c16490e9b792b6d75c2fd53 : SUCCESS in 42m 19s (non-voting)\n- kolla-build-centos-source https://zuul.opendev.org/t/openstack/build/0947f4cdc38246e2a2ec0271fc0164e8 : SUCCESS in 51m 25s\n- kolla-ansible-centos-source https://zuul.opendev.org/t/openstack/build/992c958150bc40fb92528d40c04a264d : SUCCESS in 1h 42m 04s\n- kolla-ansible-centos-binary https://zuul.opendev.org/t/openstack/build/eaa20b80946740e6a2672404f7817e45 : SUCCESS in 1h 34m 33s (non-voting)\n- kolla-ansible-centos-source-upgrade https://zuul.opendev.org/t/openstack/build/69bfde81e531406eb3ce25c2ccc3819e : FAILURE in 1h 33m 34s\n- tripleo-build-containers-centos-7 https://zuul.opendev.org/t/openstack/build/92636bef8da34ea0b914d8618fdc14c3 : SUCCESS in 1h 04m 18s (non-voting)\n- kolla-build-oraclelinux-binary https://zuul.opendev.org/t/openstack/build/44ec74b62e8d474aafc17be48005d6c5 : SUCCESS in 1h 12m 14s (non-voting)\n- kolla-build-oraclelinux-source https://zuul.opendev.org/t/openstack/build/0e6653a96db045c496aa0b80840340d0 : SUCCESS in 1h 15m 57s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/ad39c177466e4a47a0f34ede23791aa3 : SUCCESS in 4m 07s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/1df55e090e55414c8f57f0b94d46410b : SUCCESS in 3m 01s\n- openstack-tox-py27 https://zuul.opendev.org/t/openstack/build/b201702de5ea45579b5cf665e6a087b2 : SUCCESS in 3m 53s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/632fdf5fc1344479b5f4133ea0ee985a : SUCCESS in 4m 28s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/1db75b6cdcb34e59af85c3c262174468 : SUCCESS in 3m 59s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/331adac4857640d1884129c5a196e5e2 : SUCCESS in 3m 16s\n- kolla-build-ubuntu-binary https://zuul.opendev.org/t/openstack/build/7508bbfae02643a8abd8abfd0a4d2a6d : SUCCESS in 27m 41s (non-voting)\n- kolla-build-ubuntu-source https://zuul.opendev.org/t/openstack/build/732a834d34fa40f9872e83d3ca218ca5 : SUCCESS in 37m 33s\n- kolla-ansible-ubuntu-source https://zuul.opendev.org/t/openstack/build/78fd8f2d33534bae82d956e02e21cebf : SUCCESS in 58m 46s\n- kolla-ansible-ubuntu-binary https://zuul.opendev.org/t/openstack/build/f29545ce330f47a8ab910c5d23474d57 : SUCCESS in 45m 07s (non-voting)\n- kolla-ansible-ubuntu-source-upgrade https://zuul.opendev.org/t/openstack/build/5a74dc7433df4a84b4dc8937cf3cd254 : SUCCESS in 1h 10m 56s","accounts_in_message":[],"_revision_number":1},{"id":"49b555d2f7d96fa096337b33505e6a9ff8af8d8f","author":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},"date":"2019-10-02 15:02:26.000000000","message":"Patch Set 1:\n\nrecheck","accounts_in_message":[],"_revision_number":1},{"id":"f82ca86bdaf204159c1ee2ad0b96f771ed8e5203","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-10-02 16:51:30.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\n\n- kolla-build-centos-binary https://zuul.opendev.org/t/openstack/build/c9a556d07a304f8d97b81ba2436cd4d3 : SUCCESS in 49m 15s (non-voting)\n- kolla-build-centos-source https://zuul.opendev.org/t/openstack/build/d26718a1c09a4660836e480930d9abd1 : SUCCESS in 46m 49s\n- kolla-ansible-centos-source https://zuul.opendev.org/t/openstack/build/860cab68c1894415983ff7b416bd6b08 : SUCCESS in 1h 19m 31s\n- kolla-ansible-centos-binary https://zuul.opendev.org/t/openstack/build/711a13171481442b9ead112e3bac7ba2 : SUCCESS in 1h 20m 36s (non-voting)\n- kolla-ansible-centos-source-upgrade https://zuul.opendev.org/t/openstack/build/d80b5f89350440429e38e11fb375962e : SUCCESS in 1h 42m 39s\n- tripleo-build-containers-centos-7 https://zuul.opendev.org/t/openstack/build/004508ffdb554a03aeaf5810609b8376 : SUCCESS in 58m 50s (non-voting)\n- kolla-build-oraclelinux-binary https://zuul.opendev.org/t/openstack/build/1f3f932f3cac4fbea7a280a7d46ceaeb : SUCCESS in 49m 43s (non-voting)\n- kolla-build-oraclelinux-source https://zuul.opendev.org/t/openstack/build/a165aa52bcae4deab1b9ad4c90011926 : SUCCESS in 46m 52s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/72914adb1ac644f195aa31eadb910888 : SUCCESS in 3m 34s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/9247307de319406aa2a1cdb0f0d51879 : SUCCESS in 3m 26s\n- openstack-tox-py27 https://zuul.opendev.org/t/openstack/build/6b8fa3ee365847ab93636debb9737b2e : SUCCESS in 3m 28s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/215e48f75b874a468ac6dd3106e1b5b2 : SUCCESS in 3m 17s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/5a3fd49b199940709b20ec3617c49509 : SUCCESS in 4m 47s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/0a87c5c838314fc3a4caeb61caccfed7 : SUCCESS in 4m 22s\n- kolla-build-ubuntu-binary https://zuul.opendev.org/t/openstack/build/8c8db1bc700849309cde6bdb20a2a4c4 : SUCCESS in 25m 43s (non-voting)\n- kolla-build-ubuntu-source https://zuul.opendev.org/t/openstack/build/9cb5e8dc0266471cadc3976db23b6a96 : SUCCESS in 35m 09s\n- kolla-ansible-ubuntu-source https://zuul.opendev.org/t/openstack/build/a4801943173944daa9b6be5866aa545d : SUCCESS in 1h 02m 06s\n- kolla-ansible-ubuntu-binary https://zuul.opendev.org/t/openstack/build/05c6094f576546359480ea56c7cc3327 : SUCCESS in 1h 01m 27s (non-voting)\n- kolla-ansible-ubuntu-source-upgrade https://zuul.opendev.org/t/openstack/build/a9b195dfee1746bdb7cc93a6a7ebbc27 : SUCCESS in 1h 18m 35s","accounts_in_message":[],"_revision_number":1},{"id":"49ccc0ee49c5a2a940d99fc01e1b62b7ad6d12d4","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-10-02 16:51:45.000000000","message":"Patch Set 1: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":1},{"id":"b7ec168f33ba715666de33f502d638badeb11c9a","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-10-02 17:51:06.000000000","message":"Patch Set 1: Verified+2\n\nBuild succeeded (gate pipeline).\n\n- kolla-build-centos-source https://zuul.opendev.org/t/openstack/build/4977ca05937246bd8d5ffb65a08b2ed1 : SUCCESS in 52m 48s\n- kolla-build-oraclelinux-source https://zuul.opendev.org/t/openstack/build/24c4ec35ac434342a52e6e8b1b030181 : SUCCESS in 51m 15s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/749f42ea101a411a80eb0eb81df54927 : SUCCESS in 4m 14s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/b3ec2a5cc62f4b89be65cc4acbc1d321 : SUCCESS in 3m 24s\n- openstack-tox-py27 https://zuul.opendev.org/t/openstack/build/d25aa4f42b844acd94a2d6c0d4f2815f : SUCCESS in 3m 01s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/208c807777af4cbb914a656c1f3fe714 : SUCCESS in 2m 53s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/fa484b558ee94da0b00b55fca716a26d : SUCCESS in 3m 39s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/0b55c0da24084ef485da134658a41f90 : SUCCESS in 4m 18s\n- kolla-build-ubuntu-source https://zuul.opendev.org/t/openstack/build/02bdab7903d5485d9a1c7457fdab6543 : SUCCESS in 37m 35s","accounts_in_message":[],"_revision_number":1},{"id":"47a619f0c1d3ba70acf03f767566b413a54b3bdb","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-10-02 17:51:06.000000000","message":"Change has been successfully merged by Zuul","accounts_in_message":[],"_revision_number":1},{"id":"7b1204e89611102cbab07e2b2926d504c97c69c6","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-10-02 17:52:19.000000000","message":"Patch Set 1:\n\nBuild succeeded (promote pipeline).\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/f9572f467e3b40f788ea44781912d36a : SUCCESS in 50s","accounts_in_message":[],"_revision_number":1}],"current_revision_number":1,"current_revision":"7f8dd3063b9939a4d6346bd4a114805a2484c9c2","revisions":{"7f8dd3063b9939a4d6346bd4a114805a2484c9c2":{"kind":"REWORK","_number":1,"created":"2019-10-02 13:11:54.000000000","uploader":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},"ref":"refs/changes/50/686150/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/kolla","ref":"refs/changes/50/686150/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/kolla refs/changes/50/686150/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/kolla refs/changes/50/686150/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/kolla refs/changes/50/686150/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/kolla refs/changes/50/686150/1"}}},"commit":{"parents":[{"commit":"07750e32a03e8f9af347f2de36a3517192d50324","subject":"Merge \"Placement: Debian pkg has migration script in different path\" into stable/stein","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/07750e32a03e8f9af347f2de36a3517192d50324"}]}],"author":{"name":"Marcin Juszkiewicz","email":"marcin.juszkiewicz@linaro.org","date":"2019-07-18 15:04:36.000000000","tz":120},"committer":{"name":"Mark Goddard","email":"mark@stackhpc.com","date":"2019-10-02 13:11:54.000000000","tz":0},"subject":"There is no Java 8 in Debian \u0027buster\u0027 (part 3)","message":"There is no Java 8 in Debian \u0027buster\u0027 (part 3)\n\nGood to have proper Java version. Better is when JAVA_HOME is also\nadjusted to point to proper directory. And to have it set proper way.\n\nChange-Id: I0f83c3498028135751b0b887665d009e5e19410f\n(cherry picked from commit 0383aaca7b6282e69761e4c44a7d8e509305e107)\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/7f8dd3063b9939a4d6346bd4a114805a2484c9c2"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/7f8dd3063b9939a4d6346bd4a114805a2484c9c2"}]},"branch":"refs/heads/stable/stein"}},"requirements":[],"submit_records":[],"submit_requirements":[]}
