)]}'
{"id":"openstack%2Fkolla~686316","triplet_id":"openstack%2Fkolla~master~I7f818141506adee67d568061a93e3c13c65f24db","project":"openstack/kolla","branch":"master","hashtags":[],"change_id":"I7f818141506adee67d568061a93e3c13c65f24db","subject":"Horizon: change location of policy files in Debian","status":"MERGED","created":"2019-10-02 23:38:08.000000000","updated":"2019-10-03 10:34:38.000000000","submitted":"2019-10-03 10:33:18.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"686316-1570098798726-adebb769","meta_rev_id":"da52875073f19811b454000348308fae4c9bbb62","_number":686316,"virtual_id_number":686316,"owner":{"_account_id":28374,"name":"Viktor Michalek","email":"viktor.michalek@ultimum.io","username":"viktormichalek"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},{"value":0,"_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2019-10-03 10:33:18.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"date":"2019-10-03 00:17:57.000000000","_account_id":23181,"name":"RDO Third Party CI","email":"dmsimard+rdothirdparty@redhat.com","username":"rdothirdparty","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},"all":[{"value":2,"date":"2019-10-03 08:10:45.000000000","_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},{"value":2,"date":"2019-10-03 07:03:12.000000000","_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":23181,"name":"RDO Third Party CI","email":"dmsimard+rdothirdparty@redhat.com","username":"rdothirdparty","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},"all":[{"value":1,"date":"2019-10-03 08:10:45.000000000","_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},{"value":0,"_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":23181,"name":"RDO Third Party CI","email":"dmsimard+rdothirdparty@redhat.com","username":"rdothirdparty","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Backport-Candidate":{"all":[{"value":0,"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},{"value":0,"_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":23181,"name":"RDO Third Party CI","email":"dmsimard+rdothirdparty@redhat.com","username":"rdothirdparty","tags":["SERVICE_USER"]}],"values":{"-1":"Do Not Backport"," 0":"Backport Review Needed","+1":"Should Backport"},"description":"","default_value":0,"optional":true},"Review-Priority":{"all":[{"value":0,"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},{"value":0,"_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":23181,"name":"RDO Third Party CI","email":"dmsimard+rdothirdparty@redhat.com","username":"rdothirdparty","tags":["SERVICE_USER"]}],"values":{"-1":"Branch Freeze"," 0":"No Priority","+1":"Important Change","+2":"Gate Blocker Fix / Urgent Change"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":23181,"name":"RDO Third Party CI","email":"dmsimard+rdothirdparty@redhat.com","username":"rdothirdparty","tags":["SERVICE_USER"]},{"_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2019-10-03 00:17:57.000000000","updated_by":{"_account_id":23181,"name":"RDO Third Party CI","email":"dmsimard+rdothirdparty@redhat.com","username":"rdothirdparty","tags":["SERVICE_USER"]},"reviewer":{"_account_id":23181,"name":"RDO Third Party CI","email":"dmsimard+rdothirdparty@redhat.com","username":"rdothirdparty","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2019-10-03 07:03:12.000000000","updated_by":{"_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"},"reviewer":{"_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"},"state":"REVIEWER"},{"updated":"2019-10-03 08:10:45.000000000","updated_by":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},"reviewer":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},"state":"REVIEWER"},{"updated":"2019-10-03 10:33:18.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"}],"messages":[{"id":"e7f27bfcee2caaa4d735224328ac82e5c8c2eeb1","author":{"_account_id":28374,"name":"Viktor Michalek","email":"viktor.michalek@ultimum.io","username":"viktormichalek"},"date":"2019-10-02 23:38:08.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"61dd1ed8089b70377017505f7d970fa6ac335cc3","author":{"_account_id":23181,"name":"RDO Third Party CI","email":"dmsimard+rdothirdparty@redhat.com","username":"rdothirdparty","tags":["SERVICE_USER"]},"date":"2019-10-03 00:17:57.000000000","message":"Patch Set 1:\n\nBuild succeeded (check pipeline).\n\n- tripleo-build-containers-rhel-8-master http://logs.rdoproject.org/16/686316/1/openstack-check/tripleo-build-containers-rhel-8-master/3afd199/ : SUCCESS in 38m 23s","accounts_in_message":[],"_revision_number":1},{"id":"1f8a6ab14ba2fcf7bf8be7885f8bbca5a1c83e69","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-10-03 01:17:09.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\n\n- kolla-build-centos-binary https://zuul.opendev.org/t/openstack/build/66744db1b340494b9388453705ca2a30 : SUCCESS in 44m 42s (non-voting)\n- kolla-build-centos-source https://zuul.opendev.org/t/openstack/build/ec3fbe3ec486439fb57d8ec07700e77d : SUCCESS in 58m 36s\n- kolla-ansible-centos-source https://zuul.opendev.org/t/openstack/build/9358501858a94832af3722fad84b76cf : SUCCESS in 1h 23m 19s\n- kolla-ansible-centos-binary https://zuul.opendev.org/t/openstack/build/79463b69a1ac43cabbe376555fe51b7f : SUCCESS in 1h 35m 13s (non-voting)\n- kolla-ansible-centos-source-upgrade https://zuul.opendev.org/t/openstack/build/4a3bf3430141433f97a8e1af5a6649dc : SUCCESS in 1h 24m 53s\n- tripleo-build-containers-centos-7 https://zuul.opendev.org/t/openstack/build/d3bbc698c6324c7685b6a421ec46c97b : SUCCESS in 58m 19s\n- kolla-build-debian-source https://zuul.opendev.org/t/openstack/build/31bd72b67eca4d8298a4bad10fe4bb0e : SUCCESS in 26m 10s (non-voting)\n- kolla-build-debian-binary https://zuul.opendev.org/t/openstack/build/9cd8bdb7895a47009eb02189beb706ee : SUCCESS in 18m 33s (non-voting)\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/bd9e3724463c450280b8a26dd91b078c : SUCCESS in 4m 30s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/d4e15ed80afc44c986e06411ea6d38ae : SUCCESS in 5m 31s\n- openstack-tox-py27 https://zuul.opendev.org/t/openstack/build/7aed25031f434916825f6d18aa3b9c43 : SUCCESS in 4m 18s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/1971fa65def143ccb37fb722497720ef : SUCCESS in 3m 43s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/67d185354a934371bfc1fbbc87a6e65b : SUCCESS in 4m 47s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/97c8ae5b29a94272855f98959d372d8f : SUCCESS in 6m 41s\n- kolla-build-ubuntu-binary https://zuul.opendev.org/t/openstack/build/b57ccca0301047b78514830cb604a774 : FAILURE in 27m 01s (non-voting)\n- kolla-build-ubuntu-source https://zuul.opendev.org/t/openstack/build/92bebde2d9b54352ba20d8c24d33bdcf : SUCCESS in 34m 21s\n- kolla-ansible-ubuntu-source https://zuul.opendev.org/t/openstack/build/de738d906fcf4d1fa5f0e47fb5d41a5f : SUCCESS in 55m 18s\n- kolla-ansible-ubuntu-binary https://zuul.opendev.org/t/openstack/build/eca2f72827914425849c5dcf58610c84 : SUCCESS in 1h 02m 00s (non-voting)\n- kolla-ansible-ubuntu-source-upgrade https://zuul.opendev.org/t/openstack/build/309d464488364179b57d18182585ed59 : SUCCESS in 1h 19m 03s","accounts_in_message":[],"_revision_number":1},{"id":"d9ddab045bbe8bece121c3f60a69f463ece30062","author":{"_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"},"date":"2019-10-03 07:03:12.000000000","message":"Patch Set 1: Code-Review+2","accounts_in_message":[],"_revision_number":1},{"id":"53baeaffbe315b5d4bcb4ad7a4b9510fee340004","author":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},"date":"2019-10-03 08:10:45.000000000","message":"Patch Set 1: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":1},{"id":"fa5b9aa2c15587111dcbb72471c6bd916f685d05","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-10-03 08:10:57.000000000","message":"Patch Set 1: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":1},{"id":"7e45140f1cc4602658aa10a39c11cf6f1a094e12","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-10-03 10:33:18.000000000","message":"Change has been successfully merged by Zuul","accounts_in_message":[],"_revision_number":1},{"id":"e6e9e261b45e90ae48a3dd20cc71d412972db9fb","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-10-03 10:33:18.000000000","message":"Patch Set 1: Verified+2\n\nBuild succeeded (gate pipeline).\n\n- kolla-build-centos-source https://zuul.opendev.org/t/openstack/build/50433fd76ddb42f7830be61804bb394e : SUCCESS in 57m 25s\n- kolla-ansible-centos-source https://zuul.opendev.org/t/openstack/build/c09872d3a5894583b496eca1849e8b27 : SUCCESS in 1h 36m 24s\n- kolla-ansible-centos-source-upgrade https://zuul.opendev.org/t/openstack/build/e5aae5397a0f45919cdad7bd6b7068a7 : SUCCESS in 1h 30m 20s\n- tripleo-build-containers-centos-7 https://zuul.opendev.org/t/openstack/build/469d5c9e9f874e849941ee88c3029672 : SUCCESS in 1h 29m 57s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/1b6fffb386a947fdb8a75950f5a5b93a : SUCCESS in 3m 25s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/ae95a7254baf44f6bd2eeed923d742a6 : SUCCESS in 3m 58s\n- openstack-tox-py27 https://zuul.opendev.org/t/openstack/build/2eaf410d9e9340ce8ce0daadb2eb520a : SUCCESS in 3m 00s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/57493d04b4424fd7ac402dc144733154 : SUCCESS in 3m 19s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/4385c882430d491ebf4c49a049cf2561 : SUCCESS in 3m 38s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/92c370976690462097f16f7ee44bf609 : SUCCESS in 4m 29s\n- kolla-build-ubuntu-source https://zuul.opendev.org/t/openstack/build/d4be439644bd47e686428c9be3909ea3 : SUCCESS in 31m 15s\n- kolla-ansible-ubuntu-source https://zuul.opendev.org/t/openstack/build/24a76f93e6eb478d83f236b40d864a4c : SUCCESS in 58m 41s\n- kolla-ansible-ubuntu-source-upgrade https://zuul.opendev.org/t/openstack/build/7dd7efce0a1b43fda011ca5de91f3e03 : SUCCESS in 58m 14s","accounts_in_message":[],"_revision_number":1},{"id":"da52875073f19811b454000348308fae4c9bbb62","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-10-03 10:34:38.000000000","message":"Patch Set 1:\n\nBuild succeeded (promote pipeline).\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/d4a302e82bb54087ab7a522f2d2d8958 : SUCCESS in 58s","accounts_in_message":[],"_revision_number":1}],"current_revision_number":1,"current_revision":"531255d8d956990885fad2c052bc47cf518dfc28","revisions":{"531255d8d956990885fad2c052bc47cf518dfc28":{"kind":"REWORK","_number":1,"created":"2019-10-02 23:38:08.000000000","uploader":{"_account_id":28374,"name":"Viktor Michalek","email":"viktor.michalek@ultimum.io","username":"viktormichalek"},"ref":"refs/changes/16/686316/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/kolla","ref":"refs/changes/16/686316/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/kolla refs/changes/16/686316/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/kolla refs/changes/16/686316/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/kolla refs/changes/16/686316/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/kolla refs/changes/16/686316/1"}}},"commit":{"parents":[{"commit":"989bec307a40c91d4ec01a7407783d7d304df26c","subject":"Merge \"Fluentd: There is available td-agent pkg for Buster\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/989bec307a40c91d4ec01a7407783d7d304df26c"}]}],"author":{"name":"Viktor Michalek","email":"viktor.michalek@ultimum.io","date":"2019-10-02 23:34:03.000000000","tz":120},"committer":{"name":"Viktor Michalek","email":"viktor.michalek@ultimum.io","date":"2019-10-02 23:38:04.000000000","tz":120},"subject":"Horizon: change location of policy files in Debian","message":"Horizon: change location of policy files in Debian\n\nChange-Id: I7f818141506adee67d568061a93e3c13c65f24db\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/531255d8d956990885fad2c052bc47cf518dfc28"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/531255d8d956990885fad2c052bc47cf518dfc28"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[],"submit_requirements":[]}
