)]}'
{"id":"openstack%2Fkolla~686317","triplet_id":"openstack%2Fkolla~stable%2Fstein~Ie8c533cff44e185116d803f4858652f747cc5878","project":"openstack/kolla","branch":"stable/stein","topic":"bug/1845650","hashtags":[],"change_id":"Ie8c533cff44e185116d803f4858652f747cc5878","subject":"Bump zun-ui package to 3.0.1","status":"MERGED","created":"2019-10-03 01:00:57.000000000","updated":"2019-10-03 06:55:16.000000000","submitted":"2019-10-03 06:54:08.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"686317-1570085648168-d890bd4f","meta_rev_id":"1deeb3d8b4a3c6b53813d12758d7f297f2169808","_number":686317,"virtual_id_number":686317,"owner":{"_account_id":11536,"name":"hongbin","email":"hongbin034@gmail.com","username":"hongbin"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2019-10-03 06:54:07.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"},"all":[{"value":2,"date":"2019-10-03 05:57:12.000000000","_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"},"all":[{"value":1,"date":"2019-10-03 05:57:12.000000000","_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Backport-Candidate":{"all":[{"value":0,"_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Do Not Backport"," 0":"Backport Review Needed","+1":"Should Backport"},"description":"","default_value":0,"optional":true},"Review-Priority":{"all":[{"value":0,"_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Branch Freeze"," 0":"No Priority","+1":"Important Change","+2":"Gate Blocker Fix / Urgent Change"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2019-10-03 05:57:12.000000000","updated_by":{"_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"},"reviewer":{"_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"},"state":"REVIEWER"},{"updated":"2019-10-03 06:54:07.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"}],"messages":[{"id":"ae1b6e3dffbbea27337bbe4e3b244f73c1381a5c","author":{"_account_id":11536,"name":"hongbin","email":"hongbin034@gmail.com","username":"hongbin"},"date":"2019-10-03 01:00:57.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"4cf9242243eb187fd578a7874b432827fa473356","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-10-03 02:33:05.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\n\n- kolla-build-centos-binary https://zuul.opendev.org/t/openstack/build/1d07e951a9764caa83322e16b918cd2c : SUCCESS in 50m 07s (non-voting)\n- kolla-build-centos-source https://zuul.opendev.org/t/openstack/build/23ed0f117cab4f16ba35f02ac04b6ad5 : SUCCESS in 50m 54s\n- kolla-ansible-centos-source https://zuul.opendev.org/t/openstack/build/76e17b4080f84cfe9fe59027abe94ff5 : SUCCESS in 1h 10m 16s\n- kolla-ansible-centos-binary https://zuul.opendev.org/t/openstack/build/a57aa07325aa4a42b4653961bf557f09 : SUCCESS in 1h 07m 22s (non-voting)\n- kolla-ansible-centos-source-upgrade https://zuul.opendev.org/t/openstack/build/c52b4b3485de4c6c817c7ca6d6b9fd52 : SUCCESS in 1h 26m 30s\n- tripleo-build-containers-centos-7 https://zuul.opendev.org/t/openstack/build/5969ca9f669e421ca575f6c567b48ae9 : SUCCESS in 1h 00m 18s (non-voting)\n- kolla-build-oraclelinux-binary https://zuul.opendev.org/t/openstack/build/1c53a0d3b3c74eb9a586cfb1b974fda0 : SUCCESS in 51m 51s (non-voting)\n- kolla-build-oraclelinux-source https://zuul.opendev.org/t/openstack/build/6f774c84bb4f4303aef67b6d277b0a6a : SUCCESS in 53m 42s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/3fdbfcacdd8045ae8fbe356b9595d85b : SUCCESS in 3m 23s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/cabd8cdf6785407fb8bfd306904088ef : SUCCESS in 3m 05s\n- openstack-tox-py27 https://zuul.opendev.org/t/openstack/build/c8b87102113e4114a60cc58c9c05a30c : SUCCESS in 2m 59s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/fe56f5fdfb0549b1a6721c3699bd18ad : SUCCESS in 4m 00s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/df2ba51a6bf84757957713fdc4074ad6 : SUCCESS in 4m 45s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/9084a1092c3947f6a8891e9d454066bd : SUCCESS in 3m 59s\n- kolla-build-ubuntu-binary https://zuul.opendev.org/t/openstack/build/ff140dbf0ae343ac9e52908b449acc34 : SUCCESS in 38m 08s (non-voting)\n- kolla-build-ubuntu-source https://zuul.opendev.org/t/openstack/build/3bab5b260813414c902d786b7f2125ea : SUCCESS in 39m 41s\n- kolla-ansible-ubuntu-source https://zuul.opendev.org/t/openstack/build/e19d249df2094ee1a961b54e3113828c : SUCCESS in 1h 06m 50s\n- kolla-ansible-ubuntu-binary https://zuul.opendev.org/t/openstack/build/abfa62f61a014c129e3a722ca22ad466 : SUCCESS in 1h 04m 42s (non-voting)\n- kolla-ansible-ubuntu-source-upgrade https://zuul.opendev.org/t/openstack/build/55b7466923db40da94991f3070fc10fe : SUCCESS in 1h 13m 14s","accounts_in_message":[],"_revision_number":1},{"id":"1f89cafb40661fd5a4c97136da242a57c2649a5f","author":{"_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"},"date":"2019-10-03 05:57:12.000000000","message":"Patch Set 1: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":1},{"id":"837f5cf55bb3617be270e5718fdf0d3b8935993e","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-10-03 05:57:25.000000000","message":"Patch Set 1: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":1},{"id":"101029f5179e6fc339f3378fc3e6ca01f9055615","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-10-03 06:54:07.000000000","message":"Patch Set 1: Verified+2\n\nBuild succeeded (gate pipeline).\n\n- kolla-build-centos-source https://zuul.opendev.org/t/openstack/build/8df1b16f4edf49feab3e3c65d5c10c60 : SUCCESS in 50m 50s\n- kolla-build-oraclelinux-source https://zuul.opendev.org/t/openstack/build/e16e956a21894ce3a949873ea4ba6435 : SUCCESS in 55m 39s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/a77d7b737f6e467b8eb7a15a34735ea0 : SUCCESS in 2m 44s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/01c8dec6210d4ef8bad346838d0c4930 : SUCCESS in 4m 33s\n- openstack-tox-py27 https://zuul.opendev.org/t/openstack/build/b24b5340812a4f57b16a5d0169641218 : SUCCESS in 2m 39s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/202e329449f041cc93a6901cf99258b1 : SUCCESS in 3m 15s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/9a3362fc5a8d41b68a8e84c27df2a888 : SUCCESS in 3m 18s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/88125a33b29b4fa7b7081e711b4e3304 : SUCCESS in 2m 56s\n- kolla-build-ubuntu-source https://zuul.opendev.org/t/openstack/build/fa071c3dfa9e49a7a1266f974813758d : SUCCESS in 29m 10s","accounts_in_message":[],"_revision_number":1},{"id":"a50f0ecc8b96788b09a981762cd40bc9636a57b5","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-10-03 06:54:08.000000000","message":"Change has been successfully merged by Zuul","accounts_in_message":[],"_revision_number":1},{"id":"d5bae1309aa8d4bcef46613a3031f9a95f1b1fd5","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-10-03 06:55:16.000000000","message":"Patch Set 1:\n\nBuild succeeded (promote pipeline).\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/6a166a20b78f4c98a6ad345e7c1776cc : SUCCESS in 47s","accounts_in_message":[],"_revision_number":1}],"current_revision_number":1,"current_revision":"74ff528d93456fda2d265115641d6799a1a9c2a3","revisions":{"74ff528d93456fda2d265115641d6799a1a9c2a3":{"kind":"REWORK","_number":1,"created":"2019-10-03 01:00:57.000000000","uploader":{"_account_id":11536,"name":"hongbin","email":"hongbin034@gmail.com","username":"hongbin"},"ref":"refs/changes/17/686317/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/kolla","ref":"refs/changes/17/686317/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/kolla refs/changes/17/686317/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/kolla refs/changes/17/686317/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/kolla refs/changes/17/686317/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/kolla refs/changes/17/686317/1"}}},"commit":{"parents":[{"commit":"a3e5eceb34e5ca0a455e7aba916af413644a6896","subject":"Merge \"There is no Java 8 in Debian \u0027buster\u0027 (part 3)\" into stable/stein","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/a3e5eceb34e5ca0a455e7aba916af413644a6896"}]}],"author":{"name":"Hongbin Lu","email":"hongbin034@gmail.com","date":"2019-10-03 00:59:56.000000000","tz":0},"committer":{"name":"Hongbin Lu","email":"hongbin034@gmail.com","date":"2019-10-03 01:00:53.000000000","tz":0},"subject":"Bump zun-ui package to 3.0.1","message":"Bump zun-ui package to 3.0.1\n\nChange-Id: Ie8c533cff44e185116d803f4858652f747cc5878\nCloses-Bug: #1845650\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/74ff528d93456fda2d265115641d6799a1a9c2a3"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/74ff528d93456fda2d265115641d6799a1a9c2a3"}]},"branch":"refs/heads/stable/stein"}},"requirements":[],"submit_records":[],"submit_requirements":[]}
