)]}'
{"id":"openstack%2Fkolla~686561","triplet_id":"openstack%2Fkolla~master~Id9312830fb9961927ac8d317c432f52a4ad7af24","project":"openstack/kolla","branch":"master","topic":"686134","hashtags":[],"change_id":"Id9312830fb9961927ac8d317c432f52a4ad7af24","subject":"CI: increase both limits to 3h","status":"MERGED","created":"2019-10-04 09:41:19.000000000","updated":"2019-10-16 15:08:02.000000000","submitted":"2019-10-04 13:41:53.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"686561-1570196513801-79b23883","meta_rev_id":"4de49c7788fc4b2c9cfcef2ab90d4058a2e56490","_number":686561,"virtual_id_number":686561,"owner":{"_account_id":24072,"name":"Marcin Juszkiewicz","email":"mjuszkiewicz@redhat.com","username":"hrw"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},{"value":0,"_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2019-10-04 13:41:53.000000000","post_submit":true,"permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},"all":[{"value":2,"date":"2019-10-04 10:26:42.000000000","_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},{"value":2,"date":"2019-10-04 09:49:49.000000000","_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},"all":[{"value":1,"date":"2019-10-04 10:26:42.000000000","_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},{"value":0,"_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Backport-Candidate":{"all":[{"value":0,"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},{"value":0,"_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Do Not Backport"," 0":"Backport Review Needed","+1":"Should Backport"},"description":"","default_value":0,"optional":true},"Review-Priority":{"all":[{"value":0,"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},{"value":0,"_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Branch Freeze"," 0":"No Priority","+1":"Important Change","+2":"Gate Blocker Fix / Urgent Change"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2019-10-04 09:49:49.000000000","updated_by":{"_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"},"reviewer":{"_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"},"state":"REVIEWER"},{"updated":"2019-10-04 10:26:42.000000000","updated_by":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},"reviewer":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},"state":"REVIEWER"},{"updated":"2019-10-04 13:41:53.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"}],"messages":[{"id":"1fe68f203b9a400769daa6fab5affa464c688445","author":{"_account_id":24072,"name":"Marcin Juszkiewicz","email":"mjuszkiewicz@redhat.com","username":"hrw"},"date":"2019-10-04 09:41:19.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"49dcac4af76aed3f98942081c81634854446a8ac","author":{"_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"},"date":"2019-10-04 09:43:33.000000000","message":"Patch Set 1: Code-Review+2","accounts_in_message":[],"_revision_number":1},{"id":"0baf8f6c62dfa7f82f96b3a9e40c264c02afd597","author":{"_account_id":24072,"name":"Marcin Juszkiewicz","email":"mjuszkiewicz@redhat.com","username":"hrw"},"date":"2019-10-04 09:45:50.000000000","message":"Uploaded patch set 2: Commit message was updated.","accounts_in_message":[],"_revision_number":2},{"id":"a7a040eb2b247a85b0eddb9335df03e53da2dc34","author":{"_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"},"date":"2019-10-04 09:49:49.000000000","message":"Patch Set 2: Code-Review+2","accounts_in_message":[],"_revision_number":2},{"id":"6db0a7e851ba02db612004a027b126dea6d8b314","author":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},"date":"2019-10-04 10:26:42.000000000","message":"Patch Set 2: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":2},{"id":"0da34eb35ddfa46f6ac8c893f280c5bbc4f71ae0","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-10-04 11:16:33.000000000","message":"Patch Set 2: Verified+1\n\nBuild succeeded (check pipeline).\n\n- kolla-build-centos-binary https://zuul.opendev.org/t/openstack/build/847c84e42c874098b49c97dd1502712f : SUCCESS in 46m 38s (non-voting)\n- kolla-build-centos-source https://zuul.opendev.org/t/openstack/build/8c2f7a1fdafa43bba2a9c8a77fb47ff5 : SUCCESS in 44m 58s\n- kolla-ansible-centos-source https://zuul.opendev.org/t/openstack/build/482b98debb794a8d9cdc10609be70a1f : SUCCESS in 1h 17m 52s\n- kolla-ansible-centos-binary https://zuul.opendev.org/t/openstack/build/5a9a7f2a5a4040118a397af795a08cb4 : RETRY_LIMIT in 24m 20s (non-voting)\n- kolla-ansible-centos-source-upgrade https://zuul.opendev.org/t/openstack/build/c7fdca3b79d44a318b042b959817196e : SUCCESS in 1h 25m 16s\n- kolla-build-debian-source https://zuul.opendev.org/t/openstack/build/6d10ab2c022d4255bd2a14aaa197c917 : SUCCESS in 28m 01s (non-voting)\n- kolla-build-debian-binary https://zuul.opendev.org/t/openstack/build/5d6ed809c1b3467ba24fbe536d23ced5 : SUCCESS in 29m 42s (non-voting)\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/682af92ab05d43aaba4ce3e98e362556 : SUCCESS in 3m 53s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/722704979f6b46239396780a8514bdd5 : SUCCESS in 4m 31s\n- openstack-tox-py27 https://zuul.opendev.org/t/openstack/build/db65603fd139496c82820e48d54e5e61 : SUCCESS in 3m 56s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/fff659b2369b4aac9a7d4e1b946fa126 : SUCCESS in 2m 48s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/21bbb48da7bf4d3cb7f074593bda2c4f : SUCCESS in 3m 55s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/c8042689b6ff4ae6af2e1ceddb3d3e59 : SUCCESS in 4m 59s\n- kolla-build-ubuntu-binary https://zuul.opendev.org/t/openstack/build/daf0e408b93543c0ae068acb493da9e6 : SUCCESS in 27m 19s (non-voting)\n- kolla-build-ubuntu-source https://zuul.opendev.org/t/openstack/build/97e3c2c9ea76454083242d264a68ab56 : SUCCESS in 31m 06s\n- kolla-ansible-ubuntu-source https://zuul.opendev.org/t/openstack/build/067127010dce4d2a9e4b0047297c9936 : SUCCESS in 1h 01m 17s\n- kolla-ansible-ubuntu-binary https://zuul.opendev.org/t/openstack/build/dcf6fb59f0f948b48b1336e1352bc77d : SUCCESS in 56m 44s (non-voting)\n- kolla-ansible-ubuntu-source-upgrade https://zuul.opendev.org/t/openstack/build/9798e5610bc9433f928abc1cf9cf5ea6 : SUCCESS in 1h 02m 00s","accounts_in_message":[],"_revision_number":2},{"id":"85ffbf149136e756de90e2e080fe5ffb51585728","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-10-04 11:16:51.000000000","message":"Patch Set 2: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":2},{"id":"00b64205bfdd2e2a9cf97c8ca6ddcefc5fcb5f4c","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-10-04 13:41:53.000000000","message":"Change has been successfully merged by Zuul","accounts_in_message":[],"_revision_number":2},{"id":"6fb709b780f1e907d1c5574a08c42a8414041ce2","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-10-04 13:41:53.000000000","message":"Patch Set 2: Verified+2\n\nBuild succeeded (gate pipeline).\n\n- kolla-build-centos-source https://zuul.opendev.org/t/openstack/build/f8258de232be4f3b9975a0698b8d2c59 : SUCCESS in 52m 29s\n- kolla-ansible-centos-source https://zuul.opendev.org/t/openstack/build/7b1b93e881ed49bbbc0ce7eee5c47c51 : SUCCESS in 1h 21m 35s\n- kolla-ansible-centos-source-upgrade https://zuul.opendev.org/t/openstack/build/ff46cc0c9a83454eb48116aabe571b16 : SUCCESS in 1h 24m 42s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/44f927458c794347a85863c97b8defd9 : SUCCESS in 3m 41s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/14392aa3a899499b94d96220203d5325 : SUCCESS in 3m 40s\n- openstack-tox-py27 https://zuul.opendev.org/t/openstack/build/6e211a545d1643869652fd3a98c4f27c : SUCCESS in 3m 23s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/16e82b4801a74e25a81802fb6afd4a13 : SUCCESS in 4m 12s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/bc5a61f3d20143d689a5472bdcb84a9a : SUCCESS in 3m 40s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/5c778e641e6343109fa0821737450556 : SUCCESS in 6m 23s\n- kolla-build-ubuntu-source https://zuul.opendev.org/t/openstack/build/1d9fc5d43fea493d9c43c59dac2fcc44 : SUCCESS in 40m 48s\n- kolla-ansible-ubuntu-source https://zuul.opendev.org/t/openstack/build/517a950874d0404d82234cf4e21c91d6 : SUCCESS in 56m 23s\n- kolla-ansible-ubuntu-source-upgrade https://zuul.opendev.org/t/openstack/build/963839740aca4216b40ae1309e375dd7 : SUCCESS in 1h 01m 57s","accounts_in_message":[],"_revision_number":2},{"id":"6079246989727fb8e19ecfb3446b2202eedce957","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-10-04 13:45:49.000000000","message":"Patch Set 2:\n\nBuild succeeded (promote pipeline).\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/7345fc26aafe477281d1be0ec750fb6b : SUCCESS in 56s","accounts_in_message":[],"_revision_number":2},{"id":"05ee3a4e94050fb71c8fafc3874c5328f8c11b06","author":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},"date":"2019-10-16 15:08:02.000000000","message":"Patch Set 2: Cherry Picked\n\nThis patchset was cherry picked to branch stable/stein as commit c8cddefbe7f9e89cda08530aa4ce772e7b28abcc","accounts_in_message":[],"_revision_number":2}],"current_revision_number":2,"current_revision":"7e1d3eba928ec54899f90368b0235b6737fda3b0","revisions":{"7170de7b6bdb45ae5bac0be1b37e3d24872c6468":{"kind":"REWORK","_number":1,"created":"2019-10-04 09:41:19.000000000","uploader":{"_account_id":24072,"name":"Marcin Juszkiewicz","email":"mjuszkiewicz@redhat.com","username":"hrw"},"ref":"refs/changes/61/686561/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/kolla","ref":"refs/changes/61/686561/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/kolla refs/changes/61/686561/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/kolla refs/changes/61/686561/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/kolla refs/changes/61/686561/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/kolla refs/changes/61/686561/1"}}},"commit":{"parents":[{"commit":"531255d8d956990885fad2c052bc47cf518dfc28","subject":"Horizon: change location of policy files in Debian","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/531255d8d956990885fad2c052bc47cf518dfc28"}]}],"author":{"name":"Marcin Juszkiewicz","email":"marcin@juszkiewicz.com.pl","date":"2019-10-04 09:41:03.000000000","tz":-420},"committer":{"name":"Marcin Juszkiewicz","email":"marcin@juszkiewicz.com.pl","date":"2019-10-04 09:41:03.000000000","tz":-420},"subject":"CI: increase both limits to 3h","message":"CI: increase both limits to 3h\n\nChange-Id: Id9312830fb9961927ac8d317c432f52a4ad7af24\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/7170de7b6bdb45ae5bac0be1b37e3d24872c6468"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/7170de7b6bdb45ae5bac0be1b37e3d24872c6468"}]},"branch":"refs/heads/master"},"7e1d3eba928ec54899f90368b0235b6737fda3b0":{"kind":"NO_CODE_CHANGE","_number":2,"created":"2019-10-04 09:45:50.000000000","uploader":{"_account_id":24072,"name":"Marcin Juszkiewicz","email":"mjuszkiewicz@redhat.com","username":"hrw"},"ref":"refs/changes/61/686561/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/kolla","ref":"refs/changes/61/686561/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/kolla refs/changes/61/686561/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/kolla refs/changes/61/686561/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/kolla refs/changes/61/686561/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/kolla refs/changes/61/686561/2"}}},"commit":{"parents":[{"commit":"531255d8d956990885fad2c052bc47cf518dfc28","subject":"Horizon: change location of policy files in Debian","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/531255d8d956990885fad2c052bc47cf518dfc28"}]}],"author":{"name":"Marcin Juszkiewicz","email":"marcin@juszkiewicz.com.pl","date":"2019-10-04 09:41:03.000000000","tz":-420},"committer":{"name":"Marcin Juszkiewicz","email":"marcin@juszkiewicz.com.pl","date":"2019-10-04 09:44:36.000000000","tz":-420},"subject":"CI: increase both limits to 3h","message":"CI: increase both limits to 3h\n\nOur publishing job is barely managing to fit in 2h time so let add\nanother one.\n\nChange-Id: Id9312830fb9961927ac8d317c432f52a4ad7af24\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/7e1d3eba928ec54899f90368b0235b6737fda3b0"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/7e1d3eba928ec54899f90368b0235b6737fda3b0"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[],"submit_requirements":[]}
